)]}'
{"id":"openstack%2Foctavia-tempest-plugin~859789","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I8edb50e13d17e30d0eb3d51dc39590f473a84fff","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"gitignore-dotorig","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-03-07 16:21:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8edb50e13d17e30d0eb3d51dc39590f473a84fff","subject":"Add *.orig to .gitignore","status":"MERGED","created":"2022-09-29 08:17:09.000000000","updated":"2024-03-07 16:23:01.000000000","submitted":"2024-03-07 16:21:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859789-gitignore-dotorig","meta_rev_id":"21ce6bc6cd14ece9ee5532910d2bb3215ad47808","_number":859789,"virtual_id_number":859789,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-07 16:21:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2022-10-18 15:12:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2022-10-26 11:34:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-09-29 17:33:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-07 16:58:33.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-07 16:58:33.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-29 10:58:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-29 17:33:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-10-18 15:12:19.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-10-26 11:34:52.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-03-07 08:53:44.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"}],"messages":[{"id":"0e969bf564db2b29e061cecf72569c7eaa31e51c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-29 08:17:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09ac2cd33da64f3383fef0d100da86b50c06aa12","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-29 08:21:28.000000000","message":"Topic set to gitignore-dotorig","accounts_in_message":[],"_revision_number":1},{"id":"065d055748a5e1ea953e65795492ff05dd0f8ba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 10:58:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b3afc02b9ec4c01926c3693af313844\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbde73072fbd4933bdc6dca39aeeb5de : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54b1b7baffe940c494d1979c5db9b0d6 : SUCCESS in 3m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ce6e7496ace844398ace3ca2b599f735 : SUCCESS in 1h 26m 46s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/a9fdd74bf4fe485e8072df245a6ab6e9 : SUCCESS in 1h 31m 37s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/c5c92945c4d54d07a1e8015653742fb9 : SUCCESS in 1h 18m 31s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/825eb6b067034f909d6c35007a87e457 : SUCCESS in 1h 08m 05s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/8d69b04cf3af442c89ee495d998f115c : SUCCESS in 1h 37m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86a2d8c0a4a947aaa22fdeb283688995 : SUCCESS in 1h 49m 29s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/ba0d13a185584d2d8626e965a549b502 : SUCCESS in 1h 51m 26s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/01be077444bf4fb1a5f9dcf751ea6781 : SUCCESS in 1h 40m 26s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/422756d781c0447c9778360e30e4e31e : SUCCESS in 1h 30m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b8d372620b843ad9aa9bbf97e7f054f : SUCCESS in 38m 50s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/03df098e93c94d23b302e1eb6f283ba2 : SUCCESS in 45m 48s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/c8e861fe986a4b8199e7eccf5ef60f67 : SUCCESS in 41m 07s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/40356fe20be34dd8bd2d3167db17d95d : SUCCESS in 39m 17s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1e3fa04ce3544eaf823e4e763e4a0159 : SUCCESS in 1h 55m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a16930420be4825ab99c9e6be1a535c : FAILURE in 13m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ae836e701004977856a03c3bcfc2247 : TIMED_OUT in 2h 37m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6f9827c560d14516b84f63d0b7f82dfb : SUCCESS in 2h 12m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94c7e241daf840b39991cab59707c3e8 : SUCCESS in 43m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/1c892b23b99e4cf29bbab1d706c5f739 : SUCCESS in 39m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/77d7f8d03b6d46b88bc46473c1afead0 : SUCCESS in 39m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/cfe93c5be6b640f29e8ab732c4d2b3a5 : SUCCESS in 41m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/0b1427eeb8e94eda94a9872ce376554f : SUCCESS in 33m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/25b0104cf0b54bef955d3bcd1f838a64 : SUCCESS in 1h 53m 23s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/8294b30080a64ff5a24b6eb2f636e190 : FAILURE in 45m 08s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7ff5d573b3594f8e905040ffa18b996d : SUCCESS in 54m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e784cf76a2c25fac52d8ce82f79186d566416a9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-29 17:33:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"964ce89c20d5684061979ade0519805adaae7178","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-18 15:12:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2df5f6cf769f6712c1ff3fe628b91597abbd7bcb","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-10-26 11:34:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"569d97fd69184a8bede8622d66365c00b160944c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-07 16:58:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f826302ba63acee059b7b54df97f0e3f81846f28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 16:59:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0d01a0917d6fae977686a9e375605eba279d81e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 17:17:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b40211b046a14d11b4cd3dfb1ba5255f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b3e6d40c9f24a53882f23001481cbb5 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f071fd65c364f6aa06d5e07a6c2386c : SUCCESS in 3m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07aca37390574cfd9553516bc2bfbcd0 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/8facc75a3f2047bab9746a606f6f9c8f : CANCELED\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/3532a8771c564ef298600fb336fec46c : CANCELED\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/6e87d87294014d8b943520396cc478cd : CANCELED\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/48fcac5f925147d1a67f74ac05dc7d4b : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bf7fdc8923344ad5befee7ec21529c1f : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/12ef3af996a94ee6b507f712e719e792 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/4033458cde23433681ea203ff642da10 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/4a00b35d61bb4e019f474c28e7da3c25 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/810e8d1cea444743a5bfbce768606af5 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/c12241538f214725ab84a27755730bef : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/7584c472bbf540448fc4b59e4792031c : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/5942714df3784453a9e6e0a5094300ed : FAILURE in 15m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd934c21d9744415981f489d9fb3b7d1 : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3c21604e026b4c0fb93d1b4c4ea69b98 : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/acd9c29f3f884640b14b61c81e1616ea : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/5e6d8ffd0beb4e61a267daf50caa4ed4 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"5500063a896115b2b8645c831faebee7b78a6882","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-03-07 08:53:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"687829436c2ea66cd62e07df24ce32b8f233581a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 11:58:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d57210c04aff4cdb8225e1b08fbf7585\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b439ef2bf484e14bb30e5989094ba08 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e4827fb553f411dbbdd412bcd4f5a13 : SUCCESS in 2m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3b64349035a3466a835ffbe391aa4487 : SUCCESS in 1h 33m 48s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2bde5f6749ec45d987fed029ac32b6e3 : SUCCESS in 1h 05m 35s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/f718247f22004e1f80da5aed41236f9d : SUCCESS in 1h 32m 08s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/2c67e3ffd4f545e4ad4c3aa1f0aa5d24 : SUCCESS in 1h 48m 50s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/919c3a748cbe46bc87cf671e61c95e2e : SUCCESS in 1h 35m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7f92938dcf3d40009bf9441cacb9f7d9 : SUCCESS in 2h 17m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2777e58b49fb4e1999f462a0c143a32a : SUCCESS in 1h 25m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b02d6c691a1f4578aa566cd88fc7646a : SUCCESS in 2h 02m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/43c7fd48994b4b22891f25a5d03e1498 : SUCCESS in 1h 01m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/817d7ea6a0ca4e22a8bb67b9e3363ecc : SUCCESS in 1h 18m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/2aa2a01cd4394d4a9fc05641d1db1da0 : SUCCESS in 1h 30m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/c8ec178a4add424788ff377f8e14a050 : SUCCESS in 1h 47m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/38f2aa0c25b14428a1ebed11d0bddd47 : SUCCESS in 58m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1607a48790ad497ca2c31846bb3f3994 : SUCCESS in 59m 47s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/138289abb2b64591851cad3a5fdf33e0 : SUCCESS in 54m 36s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/dfbf226638e641dbaac9d1e0dc59c2ae : SUCCESS in 56m 54s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/3c0e343d2f3c4cedb0d78cf33465e340 : SUCCESS in 59m 20s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f0881b0cde454fa28992a8973ad37929 : FAILURE in 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/d71785798b954172ad036e677b771711 : SUCCESS in 2h 17m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/95b39ffa51444199bacb3dd99501ecd6 : SUCCESS in 1h 24m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff963817cdbc4c078843cdf44896cb63 : TIMED_OUT in 3h 03m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c5b686107c1404baa8b286e9486dbeb : SUCCESS in 42m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/93a95fb171f94f90a4e11dc22789e450 : SUCCESS in 39m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/86e5deb180734898b97dd3fb5631c9cd : SUCCESS in 42m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/e927369fe5e14a27a669a75a7d4e477e : SUCCESS in 40m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e3f7a35df554432ea158d5ba4fd566ca : SUCCESS in 2h 23m 43s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/95ddcff081f344019de7f671583ea459 : SUCCESS in 46m 07s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/cb93695705294cc4986a31e429cbf65f : SUCCESS in 50m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e209bf47d5ab133e479ca0a161d731c757c41402","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 11:59:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5638d1ffbf58d29bf597e1695362d4b86b7e1d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 16:21:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cedf82af6a784ab7a10c38515f789f07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d242d950e8a4677b2d96a1c2cd08230 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79431ff2d0c343db93f686de8c077644 : SUCCESS in 3m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/837a3f2d9b7f4d779bde1cc23613244f : SUCCESS in 1h 32m 00s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e79fdba0db7147ad90629ac4d8b6f85e : SUCCESS in 1h 00m 58s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/d1cf06a87b634edba7e8c4e1d9298fbf : SUCCESS in 1h 07m 05s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/97df571a0ada4424bb7ddc2f520300e8 : SUCCESS in 1h 20m 58s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/01fd7a45608546c39c6d8e3183ffba94 : SUCCESS in 1h 24m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2f6741fd4f5648f8b04def6a389b925b : SUCCESS in 1h 44m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/556e5a9819564b95b173ff41ac91d805 : SUCCESS in 54m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/f6a132c9f3cc4efcabbb69500d314a1f : SUCCESS in 1h 19m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4abbe488e91f4669a36ee65abde559ad : SUCCESS in 1h 01m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/8b68d7fb185846feb072e882cb705924 : SUCCESS in 1h 20m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/318963b52ac64517b3d0a4798df6a2aa : SUCCESS in 1h 02m 13s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/b4453c446eea406f85ab1937eda069d1 : SUCCESS in 1h 19m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/fcd51aac77c94dbdb7c434acc75df3e6 : SUCCESS in 1h 00m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/271673e5359a4298932cf4db375c5774 : SUCCESS in 38m 56s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/566f76586c604da9b885e02a365bfeb8 : SUCCESS in 39m 19s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/4ae0808e85c64a07bf88fb8683c23f33 : SUCCESS in 33m 07s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/b5a88b108f02460cbb4c1cb07ef2fe60 : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0f932ed59bf7af11348c99cb1fc306ede37e888a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 16:21:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"21ce6bc6cd14ece9ee5532910d2bb3215ad47808","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 16:23:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43c738109fb34cc9ae6df82e1d622d4e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7421883300fc4cb8844955ad8debb3db : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b2da54bdff9c36c0d8f9f74b75af96f8ca390b4","revisions":{"4b2da54bdff9c36c0d8f9f74b75af96f8ca390b4":{"kind":"REWORK","_number":1,"created":"2022-09-29 08:17:09.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/89/859789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/89/859789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/859789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/859789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/859789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/859789/1"}}},"commit":{"parents":[{"commit":"17902190afeb2fcf1986e20b86a8fb46a4194764","subject":"Merge \"Add ALPN support to the pools service client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/17902190afeb2fcf1986e20b86a8fb46a4194764"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-29 08:17:05.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-29 08:17:05.000000000","tz":120},"subject":"Add *.orig to .gitignore","message":"Add *.orig to .gitignore\n\nChange-Id: I8edb50e13d17e30d0eb3d51dc39590f473a84fff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4b2da54bdff9c36c0d8f9f74b75af96f8ca390b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4b2da54bdff9c36c0d8f9f74b75af96f8ca390b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
