)]}'
{"id":"openstack%2Foctavia-tempest-plugin~861743","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I848e08295580709fa739df2d216dad0efd19a795","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"add_missing_cleanups","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-02-26 18:54:42.000000000","reason":"Change was submitted"},"13995":{"account":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"last_update":"2024-02-26 18:54:42.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-02-26 18:54:42.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-02-26 18:54:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I848e08295580709fa739df2d216dad0efd19a795","subject":"Add missing cleanups after octavia resources creation","status":"MERGED","created":"2022-10-18 11:52:47.000000000","updated":"2024-02-26 18:55:55.000000000","submitted":"2024-02-26 18:54:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861743-add_missing_cleanups","meta_rev_id":"a217134f759624bf8cceb0bac519df07e1808dc0","_number":861743,"virtual_id_number":861743,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-26 18:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-26 16:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-02-26 16:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2024-02-26 16:26:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 14:46:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 12:38:25.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-10-19 12:38:25.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-02-21 18:08:51.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2024-02-26 16:26:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"5dd2d248c69945c953c313a9649b0b4d6a900a94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-18 11:52:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7417334c5fb5632452193c8cce30c578cfd9b794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-18 11:57:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1d844e052113d1dc73d5f76205c02030e0cc7b33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-18 12:30:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"25b975dd1f88c3862283b73b485f671d187bff65","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-18 14:12:18.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"ec29b45d7f5fec9dafee048e9abfbf3435c8d854","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 14:46:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/644ade1e313541c0a56fb5e4bb8fd39f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a339b68e3774cf9a6dbd941aeb61ee1 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bea350e39d7d453bb587f42b26d7d835 : SUCCESS in 2m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7281fca98e814eb8951e8f8eaf0d6cb6 : SUCCESS in 1h 43m 24s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/279422f3ccb14143ab959f939628398e : SUCCESS in 1h 49m 06s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/62a905fa35aa4e7ba0531c6fe7f42344 : SUCCESS in 1h 31m 41s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/6b3a9c8ca14f483c880b6babd127f68c : SUCCESS in 1h 11m 09s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/0d2d2691d58643d0ae2bf5242b039cd4 : SUCCESS in 2h 09m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac7aecfe7ae14e31b0f29a89447e82d3 : FAILURE in 1h 28m 42s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/dc60791524f14cfd81f40b2b9ab420ab : FAILURE in 1h 26m 36s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/1e82f6160c504c0588abaceba9fdd951 : FAILURE in 1h 27m 13s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/096f9189237343d099f2067ab1e73d47 : FAILURE in 1h 17m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c15bfc6101064527a59ae2d0cce8b9bf : SUCCESS in 44m 41s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/b46c7f2d4c0148008fb6f55154af6eaa : SUCCESS in 41m 18s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/3716d6d5af2241ee85df000b72087574 : SUCCESS in 40m 58s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/8367d333f137440b97ad1a79e39ae9a5 : SUCCESS in 43m 38s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6e99df39089a4f17b1245c91ce5a1140 : FAILURE in 1h 32m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/f900e88e786644a28e789ba94631888c : FAILURE in 11m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4330cae57ba24d84bbacec512850ca87 : FAILURE in 1h 58m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f71fe96f9fb04fc4aba63ee3eaed2246 : FAILURE in 2h 08m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4e7b03387ef47c19c41cc0c4313fb0f : SUCCESS in 40m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/a7733179332345d59ce3555dfbd2fd2a : SUCCESS in 41m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/5f9c4a43b4e446ef9c3fe7b57b495d85 : SUCCESS in 39m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/91b22f82281d4f0dbffb7e097039c785 : SUCCESS in 42m 58s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/a7177e7ca5d146adb2d5a8cd508832a5 : SUCCESS in 40m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/992e9feb88904e33a59d0575ba20f8e6 : FAILURE in 1h 28m 15s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/e5c792156cb24392a7a0581e6714a8ef : SUCCESS in 41m 31s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/44cc5712944a40a49d06c9105fbc58a4 : FAILURE in 59m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8e6b4c1238e3e4bdb21f7e880dfd687b8c82decb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-19 09:23:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a10637b66ec4d8d2ac1e778bb9c9f3fbc35e73f8","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-19 09:23:38.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fb326554f6979c68c6da812c3c9a5b5b6c98f7ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 11:46:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0fe742125c24ea881e95e28a3f52ea5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ce909b2f4a4a74bc4198b0f95d1389 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0245527f62243c69ab0ba3a2097e836 : SUCCESS in 2m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b977788730644f39846aab0ed7ab4154 : SUCCESS in 1h 58m 13s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/619a74bfb1d54210904883dbc9628f1b : SUCCESS in 1h 34m 01s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/086bd57d5eb146f99e51cea21d9925f4 : SUCCESS in 1h 26m 27s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/4484385a782144a0b335285270bdb9c6 : SUCCESS in 1h 17m 34s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/7bdd0e5ecf8b42ab832218aa96de4590 : SUCCESS in 2h 18m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/263e8f67fdff43e2a17eed8eb9949043 : SUCCESS in 1h 38m 17s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/26695a2b0a1245a296eeda259aafa999 : SUCCESS in 1h 17m 24s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/70c1e1afbfc44019b7e230e79fe51553 : SUCCESS in 1h 33m 58s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/27ed08446ff6496287342188d968eea2 : SUCCESS in 1h 28m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a12c2ff030d9400ca69a74f972d38dcd : SUCCESS in 39m 46s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/8ce7078cf5344b9cac6ba818c75e75d9 : SUCCESS in 32m 42s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/9e4013c18aff43e2a2718aec1ba026f3 : SUCCESS in 35m 20s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/aacbfe0b014942b18fdbcf7b7592c7ee : SUCCESS in 33m 59s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3f7b664a5d5844d48d00ba427975da05 : SUCCESS in 2h 08m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc44e24fc9de4ef5a527991db30ea4a8 : FAILURE in 9m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8200d236dc594a4ca102deae7a5dbb0c : SUCCESS in 1h 51m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cc9ca62a05144ceabc1404346473a95e : SUCCESS in 2h 11m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5bc6c98e11b9460cbb9f4e7c44f5d24e : SUCCESS in 36m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/aeb52303e07b48e6ba5f8b35591cf77a : SUCCESS in 36m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/4d3f1cdc12ff47918255d47d8a89fde3 : SUCCESS in 34m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/8ef0e94cc6c34eefb6fbbba0de27aa55 : SUCCESS in 34m 03s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/140ac09199f04591b25efcefffaca858 : SUCCESS in 32m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/860154cc95274fa2baaa28c080e000df : SUCCESS in 1h 34m 25s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/4e55ac54b1ef4c4ca680013d3759c94f : SUCCESS in 38m 38s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/6bc26738446446d1ab2bd0040b4f6365 : SUCCESS in 47m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fbb2bcf54859ddab2063e0e20cfcb181af9be02b","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-19 12:37:43.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"ef106e34c4b93afc1d00b7ca5c2f3dda7ddcab93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-02-06 14:50:01.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"784cc3122cf887e3fd8f82d501f564b977fc2a85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:26:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd5d4f2fad5c4f9c8356e10eaf64eed1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488a19f5074b4b8e90ff289f87ff593e : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/515d3f9fc7d54c56aad4f8e9805503f5 : SUCCESS in 2m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fd635c26485b4fa091d1758756f61c51 : SUCCESS in 1h 45m 41s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/8557410fefe84d53ab6cce9e4ab74e8d : SUCCESS in 1h 29m 59s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/3467c4ef2fcb4309a24ade752f8f32ae : SUCCESS in 1h 30m 26s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/b1706485d1dc40d0b4676a5b481e25ae : FAILURE in 14m 14s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2707ee74e06c49dc9f9b2b2c05483cd5 : FAILURE in 1h 23m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9c208d55b55d47dfb5a3adf837333ec3 : SUCCESS in 2h 13m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2d71364d2b5d417ebecf2b4a56325b6b : SUCCESS in 58m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/bef6c7d095a643f0acad6140f2f5ea15 : SUCCESS in 1h 40m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/a7dc443f56944b4c90cb9a4a739cf0f5 : SUCCESS in 1h 26m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/d2382978e01244bfacadc50590491f86 : SUCCESS in 1h 17m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/7e52df136c2f466e934236d1420bc684 : SUCCESS in 1h 25m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/02c14b8599c44c6985edf2362ba492ec : FAILURE in 13m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/c4ecc64ac66842e8a885d1361e8b58af : FAILURE in 20m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/543b803783c042b0885a8b860b26fc4b : SUCCESS in 36m 00s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/cc01b26547e549eaae1e1f49ebbd257c : SUCCESS in 53m 02s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/f35c49ae90fc4166b132a6fc2278189b : SUCCESS in 37m 55s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/d1883a6a1ef6423ebc698d676595e138 : FAILURE in 14m 11s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4039a7e9c66247d08f3cff677422c1ad : FAILURE in 37m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/3198b74567854218ae61d9284c6cab32 : SUCCESS in 2h 12m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ffbd796c167d496abc1682edac3e0eba : SUCCESS in 1h 15m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/14ff569ae13145db9160ff231f0c07a7 : FAILURE in 2h 13m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43fb732542ac45439c64d2fec611452c : SUCCESS in 35m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/742028ac882c497e814c716f541d1a5d : SUCCESS in 34m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/9f573ce3900f43ec8301331b2feb8d75 : SUCCESS in 33m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/049f45dc2b09487293fefb596f9003cd : FAILURE in 20m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3d9d93ba460b4c38818191129906c97c : TIMED_OUT in 2h 33m 35s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/513df82f4a3f458bb566c0cc9e5c0e70 : SUCCESS in 39m 18s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/c24e06f167bf471aa5599a6b55740a24 : SUCCESS in 42m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f7e8c6fe68052f205d0e66291a9b33414ee9861c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-07 11:04:57.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"873d262172c7028a196f7561b964ee62249465c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 14:10:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08ac56f92ff84447ae8652633bcc580d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30581c88fcfd4263baba171b434ca752 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74deef6c608c4daeb8df99924af9e7ce : SUCCESS in 3m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6b504d81a474455f8f77a23d4e6162a9 : SUCCESS in 1h 32m 41s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/9524e2673f384525b337adc44c28b914 : FAILURE in 1h 36m 47s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/a6775f9a91964b2cb51978b96a260a7e : SUCCESS in 1h 37m 50s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5092ff5fbce94137a5200d5c56768c98 : SUCCESS in 1h 59m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bc599548a9f148ffa80412b99c4119d5 : FAILURE in 2h 07m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/82dc12c7487e46d9877c8addfacd4f9a : SUCCESS in 1h 12m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/a7637c8ad648456fb17386ad74262260 : SUCCESS in 1h 56m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3bd03495bc9f4318a1db814203cda321 : SUCCESS in 1h 23m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/639fbc2a344c42f1aab5da13d5ce8880 : SUCCESS in 1h 51m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/16827d291c664996b1071ae55613c378 : SUCCESS in 1h 27m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e88239b78129400da8207b9678c58db9 : SUCCESS in 34m 52s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7cfc3ee4a6ae4451ac339b820cab1ae3 : SUCCESS in 35m 21s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/fd40bd80d44d48648bca7e17ca08e001 : SUCCESS in 35m 38s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fe46dc8bd406477ba1c9b3397e5cfc36 : FAILURE in 23m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/6e47c5ac5d6740379e2ac318bcdc8f5d : FAILURE in 43m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1928529fad7643ea8fef8d286fa514ae : FAILURE in 42m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ac1f8bf5a1441ae85115e4031c1b0f0 : TIMED_OUT in 3h 03m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69d20e1a05514b39ad4aeabba4e44d4d : SUCCESS in 35m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/29c26fba0bef431696dee596484553f3 : SUCCESS in 32m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/746c38c1213e4bbd9813f2df2489b9c5 : SUCCESS in 32m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eaa77561634f49bfa4f181e9745000b3 : SUCCESS in 1h 57m 11s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/9ce0d49f11124741a62aa75b50f052df : SUCCESS in 51m 13s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/391d654b5507463491dad271c22f6b00 : SUCCESS in 57m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"57ed1ca56dfe3c0aaa5f2373416ea1ca95006358","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-08 14:46:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5083faf5cb74611b8f04d5c3beec473a786eea2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 17:51:20.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fda43678f4724def854c03e380fc3d79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97863b0f43c34b79a3d1d965725f401f : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba479c64b8e8456c9442deb9eb36556c : SUCCESS in 2m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/629b1221408748478b0271d55a471c1c : FAILURE in 1h 38m 27s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/9d444aea97fc48a98b6172ec6c656f67 : SUCCESS in 1h 55m 31s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/c39ad2da8abc44cc8f0c132260a9afbc : FAILURE in 1h 29m 10s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c33deb7a772c4993bd60799fce158b4a : SUCCESS in 1h 38m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/48838c4fc14e47258f2b46a34d0845f0 : SUCCESS in 1h 43m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/81c23039c15a469b93537ae6aca599ae : SUCCESS in 1h 08m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/a5c425073ccf426695aee364cb0e98dd : SUCCESS in 1h 20m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/2342891a664b4a979b15c104953d0b5f : SUCCESS in 1h 29m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/80e1cdc2aa524e1b9284816a053ea92f : SUCCESS in 1h 21m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/055e0e71e2314113b4113b508729f981 : SUCCESS in 54m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f5de856d781441bcb69fb66fc907a0d5 : SUCCESS in 1h 00m 15s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/769e4f39b37f4874ababb72dde944c86 : SUCCESS in 38m 38s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/8c2a76aa62464ca59d140d6eb5353c5b : SUCCESS in 52m 01s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ed6cc901844f4d248c42c15a82a001ab : FAILURE in 40m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/fc424ff0cef44b0b81b89b68bad07f71 : FAILURE in 51m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e5d128bb9b45407c94ba06c146a71c26 : FAILURE in 43m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/527d6fb622b74ed7bf6435971ccfa935 : TIMED_OUT in 3h 03m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a390a93776aa494c8d6c1f2d8b3d0464 : SUCCESS in 33m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/4ce847f1158a40e7b9fb2252cb605506 : SUCCESS in 52m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/bbf57e63f90c4e6e98c8418df47f5a63 : SUCCESS in 36m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1d6a4ab136cf40738b8a336587b876d3 : SUCCESS in 2h 05m 32s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5a06f8825ea7497c8621a7123f8fffdf : SUCCESS in 52m 24s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/8aed6223b8d647b2a78b3c258f9db739 : SUCCESS in 53m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9783cb65d0fa05c3db7dff9625d1e2fb41eba75e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-16 16:37:37.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"227cb1d5be1fe31b6b3a490fe04d33bd07cd722f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:42:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7700868bd9a94237958c55e453960d6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fccb8d50b5bf4fc3903b3cd1a99862bf : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54a897a2f6e348838659afd34af7d181 : SUCCESS in 2m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3a1937d655ad471baf7c4ad9d8cd058c : SUCCESS in 1h 26m 04s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/91ea2dc2323e48bfb2bf3ba340c53949 : SUCCESS in 1h 19m 20s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/48b1e020dfbd4d1387c2a315b7eea7b1 : SUCCESS in 1h 51m 47s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/18d02b0189994462a1d8f017f656ce0b : FAILURE in 1h 33m 03s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2b0c1ff767a146628660f9da590cbafe : SUCCESS in 1h 51m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7f9b6888f61140f7bd7b7eeacd0ebcab : TIMED_OUT in 2h 34m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3dda86c97a894cfe8215ff41dc80831b : SUCCESS in 1h 26m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/985b455cc9c34101a16a7ce23bb4f3a8 : SUCCESS in 2h 10m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/48b37bc43ed04a96a86f434951db587b : SUCCESS in 1h 15m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/12ab26476eb045caae16d7d4d55f365d : SUCCESS in 1h 59m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/5dfb19cd0a524b56ab09f2d5077cbec5 : SUCCESS in 1h 23m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/516b0d7352ba43259712633bab9f4b25 : SUCCESS in 1h 57m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/75e4858aee514d5eae18b405a7caf7c6 : SUCCESS in 1h 21m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/52c8498f341d463c8eb86a586b841aa9 : SUCCESS in 54m 20s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e7b11ae000934acba7700ebf71304107 : SUCCESS in 33m 09s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/df9522a1fa9a4c23939356c77c40bb84 : SUCCESS in 33m 37s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/cbac687a534d4516bdc1c21352fcc88f : FAILURE in 44m 46s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6da803d864074848a137897683caf218 : FAILURE in 22m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/2eb4c2f5003444aaaf395462ba53b586 : FAILURE in 2h 11m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/711cd4da44c74f8eb7ebb5785e4e1258 : SUCCESS in 1h 20m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6c9a233575dc4566b13b002050434e96 : TIMED_OUT in 3h 03m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e812096e74549649341ee693bc647d3 : SUCCESS in 32m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/354999cf9d55493c9f1f6baa322d846b : SUCCESS in 34m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ec079ad7b5654d6cad44721fd8d81918 : SUCCESS in 38m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/102b3709d6c747bdb61b01a93cbb9821 : SUCCESS in 53m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/73e97deeabbc4cf2a549b39dde15ac85 : SUCCESS in 2h 32m 45s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/776a1967d3994b2ab7dec761a95aadf0 : SUCCESS in 49m 03s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/de219998cc134defbc074d3e1342a0aa : SUCCESS in 52m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"60ca7cacf8fe07a634eb9cb87417124ba1f1183d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-21 09:23:26.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5750b82088fb0ee2b56b1cc4257bcf581e34767d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-22 08:26:21.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"12fb4c4cdef9faf5e5d009eb805ab8012d9aec8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 11:32:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb9cf4d2f93b4ca6a8b4d49981b1574c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eda406b304214dda9eb042a9299c50e7 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00a421a6bd77401f9bff382aa13a2a2e : SUCCESS in 2m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6b8e88c93fbe4756b86e5c02f32cf80e : SUCCESS in 1h 28m 04s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e7e7fcb0fe6449abae6ed252d3519c60 : SUCCESS in 1h 35m 05s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/d9cd8083ea1b4f73baf2f47ecd88eaa2 : SUCCESS in 1h 29m 22s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/19e48280c5f04bed898bfc6453d0af7f : SUCCESS in 1h 30m 16s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/85b486f7f87a42dcbd73f866be967abf : SUCCESS in 1h 13m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c68a96acafe547f78e54dc3d50ff9ba2 : SUCCESS in 1h 39m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/595e314c659d4829ba067a2356d5ba7c : SUCCESS in 1h 16m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/80907927e04547799b68256ca9289f40 : SUCCESS in 1h 21m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2e462c68b2dd4f27bfe2e3a519c01024 : SUCCESS in 1h 25m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/905a15ec61b043ccb7a85851637ffaa8 : SUCCESS in 1h 22m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/bd5b84f9ddc844cc95086734e8d17a4d : SUCCESS in 1h 23m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/3cc8747e2dc24eb08c2e073329c74fca : SUCCESS in 1h 23m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/3b16fb17fa51487aabd2ce7047a21f5d : SUCCESS in 55m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bc6b484595bb4054a85603af32c71f67 : SUCCESS in 46m 07s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/570b95b97f05425eb1c2762307a0d9cc : SUCCESS in 49m 35s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/b389ecf14fe340a7b81b7e0d96b0c6b9 : SUCCESS in 44m 05s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/a23870be8ab642c5b954ea1d61083c76 : SUCCESS in 39m 21s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/063850da95c0495d89c91bd06016f68b : FAILURE in 24m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/7c574998f2a14833bfbf29b777e73813 : SUCCESS in 2h 20m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/00fcdb3835af48adb0bd2692e313e1f5 : SUCCESS in 1h 20m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/688a75a3488b4e8fb9688385a818db6e : TIMED_OUT in 3h 03m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/789659ba22a64005b4ba19ce3df27af1 : SUCCESS in 37m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9cd9f00460424c0396b7a8f58e8d8114 : SUCCESS in 34m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/21c900054a0543cf93823b27e2518c5c : SUCCESS in 35m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/c5824b13d73a4c0c93291f3a98800a5d : SUCCESS in 35m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/29aa15d8503f4c83bbce6059789df706 : SUCCESS in 2h 17m 36s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/8eca072364754266b299905ec8220a00 : SUCCESS in 52m 38s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b7376f9a6cab464aa7fa4572ccd09011 : SUCCESS in 55m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c740e074d9e818f8c7fc62d2bb64ed265c2f6f5e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-22 11:59:44.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4789fcfe1725fca2ccdb7e8098e7aaa6ec130c29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 15:04:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d181c868d08418eb127f3f5dba6b64f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a251e45495564471858e0a5b5f69b877 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/068136e525924ced8ce64ba2c1feed7a : SUCCESS in 2m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/15e6d6f20bb440d8812b59048dc61251 : SUCCESS in 1h 41m 14s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/afbbdec96435492e9845149cdfa42d84 : SUCCESS in 1h 32m 39s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3b2164d32c9646adb033e5bbeca742d4 : SUCCESS in 1h 38m 42s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/10311a832f73440cbfa253a6a38c55ef : SUCCESS in 1h 36m 50s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/edae0327f3954da39c22f9ec1f10d8fc : SUCCESS in 1h 34m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dfe5d86cd7874a99806d024dce1fd5ca : SUCCESS in 1h 37m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dda3037d86a14091a96db6e4eb18eecb : SUCCESS in 1h 29m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4ef38b0ddcff4a9da48431388409aa75 : SUCCESS in 1h 19m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6aa7ef568c5f4c19b98d808a3a6a6e4e : SUCCESS in 1h 30m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/e9c275d057d649c89ad5b057b8cf15b0 : SUCCESS in 1h 20m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/656594b88e0a4353a5bb66e1ebc67e09 : SUCCESS in 1h 17m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/f0b157210e7a441fa79e283499b63576 : SUCCESS in 1h 50m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/633a392c8f014ecbb251bdc873d324d2 : SUCCESS in 1h 29m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ca16e012d03f464b80ca421f6a54eb21 : SUCCESS in 35m 40s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/fe91e14a4b674651b14f4fda9973d56f : SUCCESS in 42m 16s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/1dd5690da47542d0806d480e2834df6f : SUCCESS in 51m 50s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/b3da8c9a66714dbb811a68f22c4e8b83 : SUCCESS in 36m 03s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/831f2df375eb47a984c8d22c60cee95f : FAILURE in 22m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/8c79a96fead8409eb984c032e5d8dbe2 : SUCCESS in 2h 09m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c0fb80b7252941ada30fbff4260a9845 : SUCCESS in 1h 23m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/395d81c512244f5dad1c73c77e81d051 : TIMED_OUT in 3h 03m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3da5a007e8ee48dc8a08bc83ffa1a225 : SUCCESS in 47m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/078ca955c5ab4e739c4d5308062bd475 : SUCCESS in 30m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/6310e5f645e849249d5090f7a07d841b : SUCCESS in 46m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/1ccdcac326da4003950f7cd1708e0bf9 : SUCCESS in 38m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ccd2d455a15843c8a426a2cff4a4fa50 : TIMED_OUT in 2h 32m 58s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/036fc89118ac43f180402d487fec06b8 : SUCCESS in 50m 52s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/404d06089e3540cd9123d515f4003989 : SUCCESS in 1h 01m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"857da1835afee70740c96fe62a93ceb3a73833d1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-25 10:39:02.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c3baa8a04f9efe733de857477f22ae6aa8b5050c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 14:11:23.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64b931326ff94802a479325175ff37f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/675d90d16cce4eea93f1ca4846e29738 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdecb0e889004430b15740ca71d52b75 : SUCCESS in 5m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/810b6425652642288ba923586a7a8f52 : SUCCESS in 1h 13m 55s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/62a4020d94144b57bce525b324699862 : SUCCESS in 1h 01m 28s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/4119a3808169463ba79211f09191aef9 : SUCCESS in 1h 11m 16s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/bbadc28274a6468793539d0df94b5df4 : SUCCESS in 1h 17m 43s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/126d214b9c844d21bb2fbee493113d50 : SUCCESS in 1h 50m 09s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1c7169b71e344c46b8ac0ad0d20e7253 : SUCCESS in 1h 43m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8cdb73e98b5140dc9cc5507365b4428d : SUCCESS in 1h 05m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b40ce98dcb6c43689566da93eceedaed : SUCCESS in 1h 25m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e54c136c057a48afb94372c968fd4ae0 : SUCCESS in 1h 07m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/75d27aeff41a41289904e55d979c6d06 : SUCCESS in 1h 50m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/37a9bf67837341ac8be0673f9e1984a3 : SUCCESS in 1h 01m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/5d09f14ba2d7431c9a77c19ddbcaebe4 : SUCCESS in 1h 38m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/285f4f1d98194b6eb4894d285816891b : SUCCESS in 56m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f195a0e61fc14b9da8ec51e09eaf24ec : SUCCESS in 42m 16s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/5527a96e38324580a6fd1cb5f752b9a2 : SUCCESS in 39m 34s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/fa20855081cc45f0af88ba6f2b4ab733 : SUCCESS in 48m 41s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/ea5605f0c5364d21a4d19164e7078aad : SUCCESS in 42m 00s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/69db5588a69c44aab2f1051b4b36f4a0 : FAILURE in 28m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/230ccc42456e48bb81774f9a0ed44810 : SUCCESS in 1h 49m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a19d49a6ec0343cc988b6b61bbd42534 : SUCCESS in 1h 09m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c95ebcf629ff48e38e6720c5ec355ace : TIMED_OUT in 3h 02m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e553d8087c2420794c11c5edfa7aa6b : FAILURE in 47m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/5438e8339bb24b5b8d4708f351321baf : SUCCESS in 27m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/21e9fb05b4214f4cbde0be1ce97ab91b : SUCCESS in 43m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/eb52f485952648d3a87c81cd7cebc4f6 : SUCCESS in 30m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0c5798623e74431ebf11782a68e3a1be : TIMED_OUT in 2h 34m 04s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/365748dfc3c04252a281ec650cb3edf9 : SUCCESS in 54m 54s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/6cb4585545c94455ad8456e40b9025e1 : SUCCESS in 33m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cd8da4e608269ecd113bd5d6b87f1dbe90636e62","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-26 08:57:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9b777db997a2f90f7549296a593c872f82aed529","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-26 08:57:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b025bab476a26f18289a5bcf65698c7e61c28ca4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 12:02:02.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef3b72a5b49a4aea9a156b57d7e752f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a0e9b17635045a19cf3c859fa0e3450 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bd1789b5e6446019769327ed670f7dc : SUCCESS in 3m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ef7a24a1a6434d03b979cec28f859899 : SUCCESS in 1h 52m 04s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/bc0f78cbfd454f6bb872545720b1b07c : SUCCESS in 1h 00m 23s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/711414f593ef4ebc8007cc320e93522e : SUCCESS in 1h 10m 54s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/df2a2b3a01f54763aa56d7a52a25550c : SUCCESS in 1h 10m 23s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/81f9141dd2de442ca5b284ea29d84b8e : SUCCESS in 1h 35m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d2a8c016dd7b4e7a9222b4f047b93af3 : SUCCESS in 1h 39m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4017024aa3604972a5fd61d3be610e49 : SUCCESS in 59m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/c5c70c701b02403ebf1628d33bccd115 : SUCCESS in 1h 22m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e808266e73b3482481093355ce0dc3c9 : SUCCESS in 1h 05m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/89bef2e03c36441fa93663c82e1363f1 : SUCCESS in 1h 21m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3380c3c098364bfaa5213f80d4f5a7dd : SUCCESS in 1h 10m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/c3ad8383271e41e3a2b233612b06412f : SUCCESS in 1h 46m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/efd04a16d3b54c3eabfda7a7f85d6a82 : SUCCESS in 1h 21m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/92a3da2c75274a1fbcf6708bd280169a : SUCCESS in 48m 46s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9637eb0e7e354aae89b18966d056ac14 : SUCCESS in 54m 59s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/9ddcee3691684434bea6dccc2d3aaba3 : SUCCESS in 46m 18s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/e2b0374c3a1e433386c9c9a97e96ff76 : SUCCESS in 46m 55s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5118fb756dfc4e04a992bb744f07d33f : FAILURE in 24m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/59b6414a8d5c471aa1e3c506a5d24659 : SUCCESS in 1h 57m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/762d11d447bf4312a3cd35a62d789f29 : SUCCESS in 1h 23m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b57a1121123c4e209a09e8a32ffd47f2 : TIMED_OUT in 3h 03m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a01ee1d4a43a41178346d0331183d255 : SUCCESS in 35m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/55180e3d6ece4c4c9f50fc0794981782 : SUCCESS in 42m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7f834f43636b435ab591c221731e6b63 : SUCCESS in 30m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/49c4e3d2b0d4467ea14005b6191f2d3c : SUCCESS in 43m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/85931a87711440e991184ce95a118f5b : SUCCESS in 2h 20m 25s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5b8fe1e3810340d6a92c3ff2489cd36f : SUCCESS in 50m 43s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/0997a22ab7f64a3d9f4a3468080fd779 : SUCCESS in 56m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8dbb111f580294107cc9f5230edc8b17fcc7b26b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-26 16:21:31.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"09e6817c42bd5719a49e7c0ded9afff793c68472","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-26 16:26:01.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4eb0534a2ad131dda86eb285b7d82f623df63532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 16:26:36.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"54a3fe8addddec8c83969e5a1ee1c254f333e65a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 18:54:41.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/293a15b12bca43ff8450fb6494528c7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2eec9378729455e89077c701caffb9e : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa902b7d683f4330a180d97a4a330220 : SUCCESS in 2m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a1b805a03294eb7a01202f4f026d49f : SUCCESS in 1h 30m 04s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/34e1ef3472d64f7ebe85a30349e2b471 : SUCCESS in 58m 37s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/9af2a5cd813d463aa2a4a2ceea8c13fb : SUCCESS in 1h 04m 29s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/3d3aed3bf78f4aee9dd8f6776e21d221 : SUCCESS in 2h 24m 04s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/58cc073a04f94eb18c61f64b1b4344a7 : SUCCESS in 1h 53m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6b58c580dec54df3931c4ffd03e7e3be : SUCCESS in 2h 15m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0ffcb3946f824a0f89d01c288b3435db : SUCCESS in 1h 27m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/1856c1b5730346eb8e9aa4a9de1a67f1 : SUCCESS in 2h 04m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b5970ea2b1e445dd9dc14d7b748b4ec4 : SUCCESS in 1h 26m 04s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/29ca6d2d23884dd8bcbfbabedc59d7b6 : SUCCESS in 1h 55m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/6cb938364b9c4bf897bde872e38bd50e : SUCCESS in 1h 13m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/8400189f7f4a47b8b90dbbcd3f524d65 : SUCCESS in 2h 05m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/af38f144a2c4410793d4d751e31c9384 : SUCCESS in 1h 22m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/652116751d574570835eec48622c5d62 : SUCCESS in 43m 59s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/027d84b4776342be9944b5496521c3fb : SUCCESS in 53m 23s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/0ed321fb768f457d96955287c1115be4 : SUCCESS in 33m 04s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/88d3e11f397947098628df711272152a : SUCCESS in 48m 07s","accounts_in_message":[],"_revision_number":8},{"id":"85f2cc731415dd146bb59d3a298337d8d7d75321","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 18:54:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"a217134f759624bf8cceb0bac519df07e1808dc0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 18:55:55.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a4580cf0ac64dde96bd34b0701185e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6574254f30d946538e94d35ae62d2081 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"427701e77666955407ebe615437c368ed5f2232c","revisions":{"572b5d24721af472ce507607c18a2643b92d29da":{"kind":"REWORK","_number":1,"created":"2022-10-18 11:52:47.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/43/861743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/43/861743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/1"}}},"commit":{"parents":[{"commit":"a66952e9628c1234fb549065863739e2fb4951e5","subject":"Fix generate_pkcs12_bundle for FIPS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a66952e9628c1234fb549065863739e2fb4951e5"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-18 11:49:31.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-18 11:52:45.000000000","tz":120},"subject":"Add missing cleanups after octavia resources creation","message":"Add missing cleanups after octavia resources creation\n\nIn some of the tests, cleanups were missing after creating Octavia\nresources.\n\nThis patch adds the missing cleanups.\n\nChange-Id: I848e08295580709fa739df2d216dad0efd19a795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/572b5d24721af472ce507607c18a2643b92d29da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/572b5d24721af472ce507607c18a2643b92d29da"}]},"branch":"refs/heads/master"},"1dc0ef0a9640e60ad5ac4b5a815a935a33d48046":{"kind":"REWORK","_number":2,"created":"2022-10-18 11:57:04.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/43/861743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/43/861743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/2"}}},"commit":{"parents":[{"commit":"a66952e9628c1234fb549065863739e2fb4951e5","subject":"Fix generate_pkcs12_bundle for FIPS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a66952e9628c1234fb549065863739e2fb4951e5"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-18 11:49:31.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-18 11:56:57.000000000","tz":120},"subject":"Add missing cleanups after octavia resources creation","message":"Add missing cleanups after octavia resources creation\n\nIn some of the tests, cleanups were missing after creating Octavia\nresources.\n\nThis patch adds the missing cleanups.\n\nChange-Id: I848e08295580709fa739df2d216dad0efd19a795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1dc0ef0a9640e60ad5ac4b5a815a935a33d48046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1dc0ef0a9640e60ad5ac4b5a815a935a33d48046"}]},"branch":"refs/heads/master"},"ac81936055a53cb06d9d380a7bed6de65353c919":{"kind":"REWORK","_number":3,"created":"2022-10-18 12:30:29.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/43/861743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/43/861743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/3"}}},"commit":{"parents":[{"commit":"031ecca5132c8491e55d93a9ff88c670fccd7983","subject":"Adds a pool client authentication scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/031ecca5132c8491e55d93a9ff88c670fccd7983"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-18 11:49:31.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-18 12:29:47.000000000","tz":120},"subject":"Add missing cleanups after octavia resources creation","message":"Add missing cleanups after octavia resources creation\n\nIn some of the tests, cleanups were missing after creating Octavia\nresources.\n\nThis patch adds the missing cleanups.\n\nChange-Id: I848e08295580709fa739df2d216dad0efd19a795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ac81936055a53cb06d9d380a7bed6de65353c919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ac81936055a53cb06d9d380a7bed6de65353c919"}]},"branch":"refs/heads/master"},"e66d043b2169beaa544d13bdefcef7802df8ae5a":{"kind":"REWORK","_number":4,"created":"2022-10-19 09:23:06.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/43/861743/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/43/861743/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/4"}}},"commit":{"parents":[{"commit":"031ecca5132c8491e55d93a9ff88c670fccd7983","subject":"Adds a pool client authentication scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/031ecca5132c8491e55d93a9ff88c670fccd7983"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-18 11:49:31.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-19 09:22:46.000000000","tz":120},"subject":"Add missing cleanups after octavia resources creation","message":"Add missing cleanups after octavia resources creation\n\nIn some of the tests, cleanups were missing after creating Octavia\nresources.\n\nThis patch adds the missing cleanups.\n\nChange-Id: I848e08295580709fa739df2d216dad0efd19a795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e66d043b2169beaa544d13bdefcef7802df8ae5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e66d043b2169beaa544d13bdefcef7802df8ae5a"}]},"branch":"refs/heads/master"},"c388747f402c74c6436068aae18fa20dd895a30a":{"kind":"REWORK","_number":5,"created":"2024-02-06 14:50:01.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/43/861743/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/43/861743/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/5"}}},"commit":{"parents":[{"commit":"ffc51f76b2a9c0e6b148d360c3f8b4c712fb84ca","subject":"Merge \"Drop implementation to use pyOpenSSL to manage pkcs12 certs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ffc51f76b2a9c0e6b148d360c3f8b4c712fb84ca"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-18 11:49:31.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2024-02-06 14:49:38.000000000","tz":60},"subject":"Add missing cleanups after octavia resources creation","message":"Add missing cleanups after octavia resources creation\n\nIn some of the tests, cleanups were missing after creating Octavia\nresources.\n\nThis patch adds the missing cleanups.\n\nChange-Id: I848e08295580709fa739df2d216dad0efd19a795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c388747f402c74c6436068aae18fa20dd895a30a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c388747f402c74c6436068aae18fa20dd895a30a"}]},"branch":"refs/heads/master"},"e2617323bbcbe6d69dea8ce52fc97f1194bc8b29":{"kind":"REWORK","_number":6,"created":"2024-02-07 11:04:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/43/861743/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/43/861743/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/6"}}},"commit":{"parents":[{"commit":"ffc51f76b2a9c0e6b148d360c3f8b4c712fb84ca","subject":"Merge \"Drop implementation to use pyOpenSSL to manage pkcs12 certs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ffc51f76b2a9c0e6b148d360c3f8b4c712fb84ca"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-18 11:49:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-07 11:04:40.000000000","tz":60},"subject":"Add missing cleanups after octavia resources creation","message":"Add missing cleanups after octavia resources creation\n\nIn some of the tests, cleanups were missing after creating Octavia\nresources.\n\nThis patch adds the missing cleanups.\n\nAlso remove the yoga jobs that are now unmaintained\n\nChange-Id: I848e08295580709fa739df2d216dad0efd19a795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e2617323bbcbe6d69dea8ce52fc97f1194bc8b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e2617323bbcbe6d69dea8ce52fc97f1194bc8b29"}]},"branch":"refs/heads/master"},"fe93c4ec29c8349d851a3ab1689a1317ff01d32c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-02-16 16:37:37.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/43/861743/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/43/861743/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/7"}}},"commit":{"parents":[{"commit":"d7a446897f9e98c98821d9e176b88f0aaa46bbe2","subject":"Replace stable/yoga job by stable/2023.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d7a446897f9e98c98821d9e176b88f0aaa46bbe2"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-18 11:49:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-16 16:37:27.000000000","tz":60},"subject":"Add missing cleanups after octavia resources creation","message":"Add missing cleanups after octavia resources creation\n\nIn some of the tests, cleanups were missing after creating Octavia\nresources.\n\nThis patch adds the missing cleanups.\n\nAlso remove the yoga jobs that are now unmaintained\n\nChange-Id: I848e08295580709fa739df2d216dad0efd19a795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fe93c4ec29c8349d851a3ab1689a1317ff01d32c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fe93c4ec29c8349d851a3ab1689a1317ff01d32c"}]},"branch":"refs/heads/master"},"427701e77666955407ebe615437c368ed5f2232c":{"kind":"REWORK","_number":8,"created":"2024-02-22 08:26:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/43/861743/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/43/861743/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/43/861743/8"}}},"commit":{"parents":[{"commit":"d7a446897f9e98c98821d9e176b88f0aaa46bbe2","subject":"Replace stable/yoga job by stable/2023.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d7a446897f9e98c98821d9e176b88f0aaa46bbe2"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-10-18 11:49:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 08:17:47.000000000","tz":-300},"subject":"Add missing cleanups after octavia resources creation","message":"Add missing cleanups after octavia resources creation\n\nIn some of the tests, cleanups were missing after creating Octavia\nresources.\n\nThis patch adds the missing cleanups.\n\nChange-Id: I848e08295580709fa739df2d216dad0efd19a795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/427701e77666955407ebe615437c368ed5f2232c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/427701e77666955407ebe615437c368ed5f2232c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
