)]}'
{"id":"openstack%2Foctavia-tempest-plugin~864886","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I3da638eb97939186b923a5854f2ea260b61953e8","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"ussuri-pin-barbican-tempest","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-11-21 09:38:15.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-11-23 18:28:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3da638eb97939186b923a5854f2ea260b61953e8","subject":"Pin barbican-tempest-plugin to version 1.6.0","status":"MERGED","created":"2022-11-17 12:05:23.000000000","updated":"2022-11-23 18:30:51.000000000","submitted":"2022-11-23 18:28:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"864886-ussuri-pin-barbican-tempest","meta_rev_id":"4c6ce3c8c8c82e111202351b63f89ef39e3b72e6","_number":864886,"virtual_id_number":864886,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-23 18:28:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-21 09:38:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-11-23 16:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-11-23 16:33:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-17 12:19:14.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-11-17 15:35:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 16:30:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"696423b9b9ac64376fc6dbfc37ff2c9687e45658","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 12:05:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d20013ee88a550e51514281cdb4065978318ca2b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-17 12:19:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4daf8c7154953808c1494235d7b2721be6e07b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 12:26:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f52d7a3de8472420aa12d88437ffc9441dc34fd8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-17 13:03:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5194782788930557ff29b86d1d03cd29733b1ab5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 14:16:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04e57e8f037678c19d5ca4d02da1a90d92061e82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 15:35:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80fd72080c25481ea2d20ccafc14cc33\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a34f35ecb304ef99b3a5133dddceb79 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17b1917751b24340867bef925391dab9 : SUCCESS in 2m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df22a9b0781a4a189ab0e18a054b32ac : SUCCESS in 2h 03m 35s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/095a6079acb64463af544eeb8283451e : SUCCESS in 1h 20m 19s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/3270a3e7a9914e6da5173b09647a0879 : SUCCESS in 1h 42m 31s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/f1a1731850804bed85e666d897cd3bf3 : SUCCESS in 1h 08m 24s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/d4d922cea5614a82a1c9961a2c72acd8 : SUCCESS in 1h 26m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb37638bceb14562a99433043ce8ecf0 : SUCCESS in 1h 48m 11s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/192a984d3d7c4869a4f5f97c199d538f : SUCCESS in 1h 58m 20s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/e60206c2d17e4caa8c8651cb534e5163 : SUCCESS in 1h 50m 27s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/bff9fae29b18495e881cf1c4f1598f51 : SUCCESS in 1h 57m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e986218cf314c03951deb007759fa73 : SUCCESS in 44m 12s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/aaf0483d3de441cfaf67587cc0dfc0aa : SUCCESS in 40m 50s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/4a8d346a9c00440e8fa41e921d901763 : SUCCESS in 37m 50s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/138ce4d9894b4e7484e559ba3c1183df : SUCCESS in 46m 45s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/71a7196aa6b64637a9a58a2375974c0d : SUCCESS in 1h 59m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/f490c6f373a941cea9e089e3bd009d68 : FAILURE in 11m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f4a88f6449b43a6811e5c3d09ae3996 : SUCCESS in 2h 16m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a9bdb5c1d1b04cacb8a07c5e2afaf807 : TIMED_OUT in 3h 04m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80d9bb2f2d05467ba8859ffee1c67fd9 : FAILURE in 18m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/5f8dfb965a244849bc9347f61c9c3b42 : SUCCESS in 46m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/53dde6a7ed394bdeb493d85f81b4b375 : SUCCESS in 37m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/93b053adca9745aebd53304a033f8228 : SUCCESS in 30m 54s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/dca42592ddfb4147a23baff0cb4d002a : SUCCESS in 28m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a29ca691c86946d4b120fbf4355dc14b : SUCCESS in 1h 28m 26s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/7d6c156b753b47b7b06cd7d740c6635e : SUCCESS in 36m 08s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/9cd997a7ff57496f9fd76b49134ea5b5 : SUCCESS in 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad8a20591eb12837457b3be75173b0b989c848eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 16:24:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fa46ca19fa1f29f08c6eacc95fb635a6c2b8e427","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 16:28:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6ae2b926bf23e966b12b9a45fa9f7dd62305c56c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 19:36:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49017a03b8824b2d8f98efb36431d83f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2203f8c9b7ae46c38e406e64b10427f9 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5cdc961b3244daf9d98c562f0e73925 : SUCCESS in 3m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d728801926574595a5e16988f0af0e26 : FAILURE in 20m 33s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/88a347217846464693ab0ddd6700c506 : SUCCESS in 1h 16m 45s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/7a42d794984d4a7ab1948702a3267e3c : SUCCESS in 1h 14m 59s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/3019f7d94fad4c3fabfa1fc24193b4ae : SUCCESS in 56m 54s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/b075ecf7bf414fe8805618a090b09fd2 : SUCCESS in 1h 30m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67a81975c3c947a494ec6524f4054600 : SUCCESS in 2h 32m 39s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/ce0bd56d4ccc47b89fbf7a390737739e : SUCCESS in 2h 04m 07s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/bca2c75589644946b6ce516caed3ec26 : SUCCESS in 2h 01m 50s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/057caa394c71463eb9afc7ef4c8cfafa : SUCCESS in 1h 41m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/00f2949ba7d44fd5979af32cdb27d68b : SUCCESS in 44m 02s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/202b00ce01184895b2322d8c84ace823 : SUCCESS in 43m 53s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/e842ee11d35442178ac8f39f9da8b43e : SUCCESS in 42m 29s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/db861639acc84d8ea0bd26b584863383 : SUCCESS in 43m 19s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b5923af9634f4929998c400656b4e680 : SUCCESS in 2h 28m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4c5e1f1cba94dbeb301eb868eab7d50 : FAILURE in 11m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad5ff3894fdf41159f39a56f5245b335 : SUCCESS in 2h 22m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0ef4fe1a73f6496a80b787e4fb765b59 : TIMED_OUT in 3h 04m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/334ac09e6ff94a0496ee582188296bf2 : SUCCESS in 34m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/8a4506f183ce460e9d2c3aa7427aa79c : SUCCESS in 39m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/d30267f2c67c4393be69f79ec29b0d28 : SUCCESS in 34m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/cc7db9868f4b404a9f3a79a5b982ab3b : SUCCESS in 32m 32s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/115744bd58b64a0cb1d74804b1d271f9 : SUCCESS in 31m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1dbbca34f5454790bfe6554d123e9283 : TIMED_OUT in 2h 34m 37s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/b8525792f2ba40e1a15ce4ba4576ef12 : SUCCESS in 38m 03s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/2706628e700b48e682d60d463ea96974 : SUCCESS in 43m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2d1e7077e4236fa1aaf15847deb96b78309956eb","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 21:07:33.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"e0ba286fb459ada0a8b2d44245e9df5489e5c615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 23:48:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d80a0ede51424e119514bea9e3e32f02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e13d59599599447cb21914b8ca524972 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb6223ec6b3a4008859b563a65521ae9 : SUCCESS in 2m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dbe8e9a42cae49e0b05d37f19abfada3 : SUCCESS in 1h 24m 09s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/83b7c3918c6c4c91a380e8a2708b325b : SUCCESS in 1h 45m 16s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/24e908da9a1d4b4bbffd1e3b2e91e590 : SUCCESS in 1h 24m 10s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/7f0d7c70d00946af864bb5a02bcadf46 : SUCCESS in 1h 10m 08s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/c26ab064fa8248ab98707f51144fadea : SUCCESS in 1h 32m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6aaf215da662469fb190f949d182d204 : SUCCESS in 1h 53m 59s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/95919bd51bb74cc9beb7d40bf4eeb273 : SUCCESS in 2h 02m 18s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/b2615199909e45278c3e473f73bfa69a : SUCCESS in 2h 16m 16s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/f1b3944bbbf940299c2dc502b9b83031 : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/31b0b1417def472ba53e9e808c1e91d6 : SUCCESS in 55m 49s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/5a6aee7282744d7294edb6ed95652468 : SUCCESS in 43m 40s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/73faf32d4b184b15893422b7f20c3fae : SUCCESS in 40m 33s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/834ea6ef1a0049dcaae4450a94c19030 : SUCCESS in 44m 38s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/68202e94be3c4c608ec814213f74f539 : SUCCESS in 2h 29m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6dd60d8b22242a386633540f005eba1 : FAILURE in 12m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4900f02a534f44b59b2cd4301c7fcebd : SUCCESS in 2h 23m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bd044eb4f1fc4a16b85bfacb3ff69463 : SUCCESS in 2h 30m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d227462bdda547499b140621a586f208 : SUCCESS in 33m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/0d10395b119e42a5aef631011ac243ed : SUCCESS in 37m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/8e235ecc633f45d8bc1621b5a6221c93 : SUCCESS in 39m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/4868e9ee611545eda16bbd1cacecf63c : SUCCESS in 34m 51s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/6c222681b2004461acac92099e2a2186 : SUCCESS in 41m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/41b1b140ce77402c9057af8cb1e8cb01 : SUCCESS in 2h 11m 33s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/bb8771d4d21f47a29cf8620f36284f1e : SUCCESS in 29m 45s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/1724d652fef6442c82a1ed325c3e91c5 : SUCCESS in 53m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2804098475ab73a40a49a6aa2d0220eff9e7ff62","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-18 10:48:06.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"d15d3488dda153da470784cfaf72e64e3b15a0f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 13:47:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b350e6bd5c4440039298e7f1fb469b12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce1b30e04e343d6b4575e528f37d146 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cde80e57cab4c6788369337b8c5cb74 : SUCCESS in 2m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/720d1980e1d445c09741e116efe960d5 : SUCCESS in 1h 35m 35s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/96ca7c2c197e424bbe256c92a1648f8e : SUCCESS in 1h 20m 25s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/54ea712f062443b78eb36b54db3c7c85 : SUCCESS in 1h 27m 43s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/8ee5cf454f944077ae2b47a51b36405e : SUCCESS in 1h 14m 20s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/d50ec0c896194db986813cf913d8e4ad : SUCCESS in 1h 53m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/992c6309633944bda1d1cbb5914278f6 : SUCCESS in 1h 42m 20s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/bb41de672b7340b0a3e1d11b30a7dc1a : SUCCESS in 1h 32m 10s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/3110f961dbd84ed483eec6474782b3b4 : SUCCESS in 1h 45m 48s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/5a2962a7372b45cb8288bf7416345c27 : SUCCESS in 1h 21m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/347b6195df4c436a8388af0e87119118 : SUCCESS in 32m 54s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/b4a823dfc93245458f508e93a3039dfd : SUCCESS in 34m 00s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/f3171ba7cd1749b69494ee4bbf5ac85f : SUCCESS in 35m 33s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/b7c9a7a8a3b04230bd713bc9058b2f5d : SUCCESS in 35m 07s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d3f76a8eb0824c8d960edaedbc8d1b4a : SUCCESS in 1h 57m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/547e0a443585483492fac902bcba8c7b : FAILURE in 9m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/db67974a5df04b3aa9e971664892a4d0 : SUCCESS in 2h 12m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ad572451b1dc47a0abc53f2cb9185bbd : SUCCESS in 2h 57m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55c98d86fb7b4a5b9addd38f341de544 : SUCCESS in 31m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/5272c0dbb7ff476ca0715f6d3e447b75 : SUCCESS in 31m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/62397482f99e4c7db76ab9c90a7416a0 : SUCCESS in 31m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/9a02f133ec1949e594a08fe5d121e7e4 : SUCCESS in 31m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/baef6e8f6f3d4caa950fe9c65e090dea : SUCCESS in 31m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f6560fd41608466db8962826f54eab71 : SUCCESS in 1h 42m 58s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/d4fe544f22b54962b152fca1a9ef7129 : SUCCESS in 50m 48s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/f8ba3584ee1b47c7914f3124d824ff32 : SUCCESS in 55m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e5f92d01109312067f4fa521dae4529f83d620a6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-21 09:38:15.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e5b44e52716255a973e536f37c1ce17cc4d77738","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-21 09:38:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9b43542be97420e382d696b222b8cfe875b6d40a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-21 09:39:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"157a09dbed31f1c68c7a409913f9ee80d7ffc960","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-23 16:30:59.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d3efe772f6bfb25dd910f48747d45b7283761a48","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-23 16:33:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"58af5d4faeb02b89368c039ebdf240f40690d1ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 16:34:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d712a0584637451bb932b307d72083d94d4f22a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 18:28:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67975fead6da45acb123428a462c8e82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5813d646cfdd45f0ab67c2d42d2408b8 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b0674e029494d9fbad0c8d8746d7fe9 : SUCCESS in 2m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e7d6db7688449f3a9978f319c9a3351 : SUCCESS in 1h 33m 41s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/ede4d019b869420685c923cd4d4337c6 : SUCCESS in 1h 31m 05s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/3807b836d0ed414aaf0a4bf4af16da9f : SUCCESS in 1h 26m 25s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/2830c3cd6d9e48d6bbb7223c3b657441 : SUCCESS in 1h 06m 34s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/cb86777c24d748b98fcd7394dfb8bb77 : SUCCESS in 1h 54m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30715e6790e84f2e9ee9b17c3754d199 : SUCCESS in 1h 33m 49s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/fd3c9462d92745b293ef24597c34e9d0 : SUCCESS in 1h 42m 35s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/5cd2796817434659bc0cab04239cce55 : SUCCESS in 1h 30m 30s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/bab0d17f4ca74f58aeaa11153ac2cffd : SUCCESS in 1h 30m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b06c3893ac24aff820ed7d049b0a807 : SUCCESS in 35m 39s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/5cacde7e745b44198da8502601aa74ed : SUCCESS in 39m 19s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/a87a9a94235e43c3b4cb1f53df922e14 : SUCCESS in 35m 42s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/164b997bd0c345a5b1266cf0c079d5b4 : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":4},{"id":"75fba549fb5055e49e9aa0f5768bc5eb6140c8aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 18:28:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"4c6ce3c8c8c82e111202351b63f89ef39e3b72e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 18:30:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca93208296e24ef882123feb4a3b8681\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19bc555f32b24c0c99414a5217566278 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"295576db0467bcf738843b44a1a541a9fa1b0591","revisions":{"c682a5e488ded9c8fd985bb9351bf6ce2ee8caf5":{"kind":"REWORK","_number":1,"created":"2022-11-17 12:05:23.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/86/864886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/86/864886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/1"}}},"commit":{"parents":[{"commit":"031ecca5132c8491e55d93a9ff88c670fccd7983","subject":"Adds a pool client authentication scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/031ecca5132c8491e55d93a9ff88c670fccd7983"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 12:03:45.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 12:05:11.000000000","tz":60},"subject":"Pin barbican-tempest-plugin to version 1.6.0","message":"Pin barbican-tempest-plugin to version 1.6.0\n\nThis version is the last that one that supports Python 3.6, which is the\nversion Ussuri uses.\n\nChange-Id: I3da638eb97939186b923a5854f2ea260b61953e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c682a5e488ded9c8fd985bb9351bf6ce2ee8caf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c682a5e488ded9c8fd985bb9351bf6ce2ee8caf5"}]},"branch":"refs/heads/master"},"7b7d91a3acb30f4f612ff5ed10844dc8983e3118":{"kind":"REWORK","_number":2,"created":"2022-11-17 12:26:40.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/86/864886/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/86/864886/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/2"}}},"commit":{"parents":[{"commit":"031ecca5132c8491e55d93a9ff88c670fccd7983","subject":"Adds a pool client authentication scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/031ecca5132c8491e55d93a9ff88c670fccd7983"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 12:03:45.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 12:26:29.000000000","tz":60},"subject":"Pin barbican-tempest-plugin to version 1.6.0","message":"Pin barbican-tempest-plugin to version 1.6.0\n\nThis version is the last that one that supports Python 3.6, which is the\nversion Ussuri uses.\n\nChange-Id: I3da638eb97939186b923a5854f2ea260b61953e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7b7d91a3acb30f4f612ff5ed10844dc8983e3118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7b7d91a3acb30f4f612ff5ed10844dc8983e3118"}]},"branch":"refs/heads/master"},"3444d6529df1fd13cc7a12aa5e5be197ea77bc48":{"kind":"REWORK","_number":3,"created":"2022-11-17 16:24:15.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/86/864886/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/86/864886/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/3"}}},"commit":{"parents":[{"commit":"031ecca5132c8491e55d93a9ff88c670fccd7983","subject":"Adds a pool client authentication scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/031ecca5132c8491e55d93a9ff88c670fccd7983"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 12:03:45.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 16:24:05.000000000","tz":60},"subject":"Pin barbican-tempest-plugin to version 1.6.0","message":"Pin barbican-tempest-plugin to version 1.6.0\n\nThis version is the last that one that supports Python 3.6, which is the\nversion Ussuri uses.\n\nChange-Id: I3da638eb97939186b923a5854f2ea260b61953e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3444d6529df1fd13cc7a12aa5e5be197ea77bc48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3444d6529df1fd13cc7a12aa5e5be197ea77bc48"}]},"branch":"refs/heads/master"},"295576db0467bcf738843b44a1a541a9fa1b0591":{"kind":"REWORK","_number":4,"created":"2022-11-17 16:28:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/86/864886/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/86/864886/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/86/864886/4"}}},"commit":{"parents":[{"commit":"031ecca5132c8491e55d93a9ff88c670fccd7983","subject":"Adds a pool client authentication scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/031ecca5132c8491e55d93a9ff88c670fccd7983"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 12:03:45.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-17 16:28:25.000000000","tz":60},"subject":"Pin barbican-tempest-plugin to version 1.6.0","message":"Pin barbican-tempest-plugin to version 1.6.0\n\nThis version is the last that one that supports Python 3.6, which is the\nversion Ussuri uses.\n\nChange-Id: I3da638eb97939186b923a5854f2ea260b61953e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/295576db0467bcf738843b44a1a541a9fa1b0591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/295576db0467bcf738843b44a1a541a9fa1b0591"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[]}
