)]}'
{"id":"openstack%2Foctavia-tempest-plugin~867810","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I8aea2b597b9dd9bbdc5a1527fae03e86364aab4c","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-01-09 11:13:05.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-09 11:13:05.000000000","reason":"Change was submitted"},"35087":{"account":{"_account_id":35087,"name":"Bas de Bruijne","email":"bas.debruijne@canonical.com","username":"basdbruijne"},"last_update":"2023-01-09 11:13:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8aea2b597b9dd9bbdc5a1527fae03e86364aab4c","subject":"Make user role logging optional","status":"MERGED","created":"2022-12-15 15:15:32.000000000","updated":"2023-01-09 11:14:08.000000000","submitted":"2023-01-09 11:13:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867810","meta_rev_id":"27e235830fc75f0a07063301c5784d0fb8c0ba5b","_number":867810,"virtual_id_number":867810,"owner":{"_account_id":35087,"name":"Bas de Bruijne","email":"bas.debruijne@canonical.com","username":"basdbruijne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-09 11:13:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-01-09 09:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-09 09:13:03.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2023-01-09 09:15:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-15 17:53:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-28 13:29:30.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-01-09 09:13:03.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-01-09 09:13:03.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1d3c5920ab16f9fef3c443574b5c2ff7fd026d46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35087,"name":"Bas de Bruijne","email":"bas.debruijne@canonical.com","username":"basdbruijne"},"date":"2022-12-15 15:15:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9643a8680557151a5432a8a9826ae1ee0485061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 17:53:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1916c5f9216649c6bdf4e1b189b061dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c23a1d1e75df4101af7dc4c66f616d91 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe02b6368ede478b9475379b36375628 : SUCCESS in 3m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/74e34fa04bc24fc89cc535ce24446e5d : SUCCESS in 1h 56m 30s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/16ec79a5eba94b299006a893c1e55a6d : SUCCESS in 1h 37m 31s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/a3fe0b8b2a9043e29513c6a016abe7fb : SUCCESS in 1h 52m 04s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/8ae58a340f3540d9845c9820a9da8fe8 : SUCCESS in 1h 15m 26s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/18b1b46e70e240baa6790bb356268ae0 : SUCCESS in 1h 58m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a953c116cf4b4955be5e65ed12b5227e : SUCCESS in 1h 41m 40s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/4e1dfe1ef4a14832bb1bfa18eff0fe00 : SUCCESS in 1h 37m 30s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/c9a7ff9221eb48ab933444d26ce1e35e : SUCCESS in 1h 54m 30s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/9423f4a28ec0411a9568445e4c2d9787 : SUCCESS in 1h 39m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/70e5332669294a91861767f30f547e1b : SUCCESS in 35m 57s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/0abb4cd253724aa6bf0e528c12f1f6ad : SUCCESS in 36m 48s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/c88024d576d747ff853dd3cbd20b67a1 : SUCCESS in 35m 35s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/c6f074ec21f940d7b6c7588938c3bb9d : SUCCESS in 36m 45s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b192d688eee747dc96c068bf6c1b3af1 : SUCCESS in 1h 42m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/eddb25eff1ce48199b62094887071e19 : FAILURE in 11m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0d5280b64624dedabea33e2c1607deb : SUCCESS in 2h 27m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cf58d579a2374fbd91e657f81343df15 : SUCCESS in 2h 22m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/611d3662cae84b2eacc0a64f5ecc9312 : SUCCESS in 34m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/42f6410d291349929ae4c010681d9efe : SUCCESS in 35m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/7f7caaa77aba49ad9766108c5e614b28 : SUCCESS in 33m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/4d35e634b3944274af0a5cb5fabe70ca : SUCCESS in 34m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/ba86911d38d84143bb3eef90d3df6e80 : SUCCESS in 31m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d10013361e40491ea67f38295a02a9df : SUCCESS in 1h 35m 46s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/3ccf8063d4214a9191926f184d7372ed : SUCCESS in 35m 47s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/d66e61c619c042fc8095cd18025eefa2 : SUCCESS in 1h 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ccb6a57ab3d729f1c828b8ef80b276ef485cdec","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-28 13:29:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05effb6e649e5f08faa68da8672de38d97aec1bd","author":{"_account_id":35087,"name":"Bas de Bruijne","email":"bas.debruijne@canonical.com","username":"basdbruijne"},"date":"2023-01-03 12:44:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55527a585ae060baa9d809f3c36340f12e8a65d3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-03 12:50:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86d20e9c554bcb594c8486dc71d4b60e8cc1d7ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35087,"name":"Bas de Bruijne","email":"bas.debruijne@canonical.com","username":"basdbruijne"},"date":"2023-01-03 17:41:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"69b4c31fedb2f6eaf459ba6f65f460c449498914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 20:06:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/309dd99da03b4dc79dacc1bc7a0f26cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf87aeb890e146289d7f34a59a609924 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf80525a36b4fdfad1dbd6e75c65428 : SUCCESS in 2m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ccb1ac75c0f541f397efb120cf4362ac : SUCCESS in 1h 10m 29s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/90286fb9877548968879daa83ae549ca : SUCCESS in 1h 30m 21s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/a98e5ab165ab4f2e95746de1c51d88ad : SUCCESS in 1h 30m 01s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/ce0a1911346242ea97f2a9456b47c330 : SUCCESS in 54m 05s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/1521f6ba0f6248bbaed176ea97a34d38 : SUCCESS in 1h 17m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8332650d5a774f47b9ceaffd150fa923 : SUCCESS in 1h 35m 44s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/fdd47a31c9684579b17f8b0a25dde2e1 : SUCCESS in 2h 03m 50s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/341f36abd7be468092c75d6fed6b50d7 : SUCCESS in 1h 51m 48s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/68c352814d7f41d58abbf20575e4270a : SUCCESS in 1h 33m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f7331949a82c4649bc77ae317c5ed504 : SUCCESS in 33m 07s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/cc4864f3329e43ff8159f3ebb97468a7 : SUCCESS in 35m 24s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/d2386e34613a4c728ff63b0ce98f413b : SUCCESS in 32m 05s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/faed2cbbd5804cd395e4021c3d7be078 : SUCCESS in 33m 02s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/01b76f9d0cc745ec8440f03f3071ba03 : SUCCESS in 1h 47m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/763aba6ba5df493892469f2d6b748919 : FAILURE in 8m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/152595b50e9d4a76a08c7b670f683010 : SUCCESS in 2h 15m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/946906c93c934596a311ba1046a29930 : SUCCESS in 2h 23m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a71df60a582401b8c3aef37cabfbf85 : SUCCESS in 31m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/27defc6241c1414c8830f90bc92f5d59 : SUCCESS in 36m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/5256dd00ca4c48bba591ad8d76a85811 : SUCCESS in 35m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/18b0149464014100a72374fc767b2781 : SUCCESS in 33m 25s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/c0c9c6759d9d48979fd753ea26cbf9b5 : SUCCESS in 32m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/401143a077fe4c9094441216861de6e0 : FAILURE in 1h 08m 17s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/e9ea45c104ec4f00af556daacbb7d3ba : SUCCESS in 50m 51s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/97232ce9c27b45088137e9e4b6c6b722 : SUCCESS in 57m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d2d08f937184cb8115fe51af80155d7414337cbd","author":{"_account_id":35087,"name":"Bas de Bruijne","email":"bas.debruijne@canonical.com","username":"basdbruijne"},"date":"2023-01-06 13:03:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eabd47a5a14358a2d8351fbcc6c76797b9723f8b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-09 09:13:03.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"45f4797bca147153f99cd6414f19fa6c5eace68f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-09 09:15:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1a81b107d3465abc8617e9a8916da0b6f39e317c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 09:16:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"70c085f8edd1d5d1985af58410e8d740ca250889","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 11:13:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef58da779e13454fa220b9ba019678e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38ea340ceec2412d8d12f66e2cfdbfec : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd45aa4db9d4fbeaf1191f8ddcf8b1b : SUCCESS in 3m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/69d759ae8a9241dca1b0930dab1deabe : SUCCESS in 1h 43m 07s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/2b07181f34c44631b6f2b3462b626593 : SUCCESS in 1h 34m 02s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/0def89c8d1f044a3b8331d5f2dd51d89 : SUCCESS in 1h 35m 20s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/9c2a363c54b64db389fb3058623d1afe : SUCCESS in 1h 18m 58s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/0b0520f4768749b5830f5a4ad68417be : SUCCESS in 1h 21m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/154838c260854f68a2817d01a5f8cf27 : SUCCESS in 1h 56m 06s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/ae06f776764a47b28a27a6a51b7cc1c6 : SUCCESS in 1h 34m 29s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/aebaeb3ea84a49bbbf7f6614d6747b61 : SUCCESS in 1h 42m 07s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/a1ce11d1789c4c709ddfd4261586ca69 : SUCCESS in 1h 24m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/44cabf38ec1145f7bbb3ccc12be53c34 : SUCCESS in 34m 32s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/276bf52c9d91484b8f137dd48338e65b : SUCCESS in 36m 30s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/562ef97a918c4a5fb6b88d8ff95e533e : SUCCESS in 37m 37s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/92257878612e4a6abba44b29609e6229 : SUCCESS in 35m 58s","accounts_in_message":[],"_revision_number":2},{"id":"30cde786acdd59ef56e2351f5775446d4b308dd5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 11:13:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"27e235830fc75f0a07063301c5784d0fb8c0ba5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 11:14:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e81b3430296e4be6adc087dd728ad372\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebd6abd8f00c4ec5bead4b774ae4fd06 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"530a88a434400c2255e43b0993c36ea1a6778e17","revisions":{"6df7ccd87d4695f6aaca440198a1b7abf4a0be8e":{"kind":"REWORK","_number":1,"created":"2022-12-15 15:15:32.000000000","uploader":{"_account_id":35087,"name":"Bas de Bruijne","email":"bas.debruijne@canonical.com","username":"basdbruijne"},"ref":"refs/changes/10/867810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/10/867810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/867810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/867810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/867810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/867810/1"}}},"commit":{"parents":[{"commit":"dddf6515ab4cec0d02255f17bb6a41660eca6289","subject":"Merge \"Add scenario test for Prometheus over TLS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/dddf6515ab4cec0d02255f17bb6a41660eca6289"}]}],"author":{"name":"Bas de Bruijne","email":"bas.debruijne@canonical.com","date":"2022-12-15 15:12:45.000000000","tz":-240},"committer":{"name":"Bas de Bruijne","email":"bas.debruijne@canonical.com","date":"2022-12-15 15:15:32.000000000","tz":-240},"subject":"Make user role logging optional","message":"Make user role logging optional\n\nNot in every cloud does tempest have permissions to list the\nrole assignments. Since it is not critial for running the tests,\nthe logging of the user roles should be configurable.\n\nChange-Id: I8aea2b597b9dd9bbdc5a1527fae03e86364aab4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6df7ccd87d4695f6aaca440198a1b7abf4a0be8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6df7ccd87d4695f6aaca440198a1b7abf4a0be8e"}]},"branch":"refs/heads/master"},"530a88a434400c2255e43b0993c36ea1a6778e17":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-03 17:41:29.000000000","uploader":{"_account_id":35087,"name":"Bas de Bruijne","email":"bas.debruijne@canonical.com","username":"basdbruijne"},"ref":"refs/changes/10/867810/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/10/867810/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/867810/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/867810/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/867810/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/867810/2"}}},"commit":{"parents":[{"commit":"577953114e2fe3750c93f05fd5be0602bb06e267","subject":"tox skipsdist seems to be problematic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/577953114e2fe3750c93f05fd5be0602bb06e267"}]}],"author":{"name":"Bas de Bruijne","email":"bas.debruijne@canonical.com","date":"2022-12-15 15:12:45.000000000","tz":-240},"committer":{"name":"Bas de Bruijne","email":"bas.debruijne@canonical.com","date":"2023-01-03 17:41:29.000000000","tz":0},"subject":"Make user role logging optional","message":"Make user role logging optional\n\nNot in every cloud does tempest have permissions to list the\nrole assignments. Since it is not critial for running the tests,\nthe logging of the user roles should be configurable.\n\nChange-Id: I8aea2b597b9dd9bbdc5a1527fae03e86364aab4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/530a88a434400c2255e43b0993c36ea1a6778e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/530a88a434400c2255e43b0993c36ea1a6778e17"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[]}
