)]}'
{"id":"openstack%2Foctavia-tempest-plugin~868870","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I293a52fe12972594690970b7d3bd91782b642669","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"tox4","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-02 16:39:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I293a52fe12972594690970b7d3bd91782b642669","subject":"tox skipsdist seems to be problematic","status":"MERGED","created":"2022-12-29 15:16:19.000000000","updated":"2023-01-02 16:40:43.000000000","submitted":"2023-01-02 16:39:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868870-tox4","meta_rev_id":"ae312a249ef723896998147b73d470ab3a4fe91d","_number":868870,"virtual_id_number":868870,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-02 16:39:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-01-02 14:11:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-02 13:52:47.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2023-01-02 14:11:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-29 18:18:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-02 14:11:07.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"0c91249d413740d71b12b50e87d69e14b826e202","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-29 15:16:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b799bc477815f72f8c4bfe6294139c841c92d950","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-29 15:20:38.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f7c69de429110d79f335b151bf7c512e3b1a416e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 18:18:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74bfa02d2060497ea10bbf7bf83d1e26\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ae5ec8d2b94adca52c27571e2e6037 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d04f323a1f9640ada0fbb547593ecda8 : SUCCESS in 2m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aebd470e85404dd7be4535542c1ad510 : SUCCESS in 2m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b8579cc905c249479ef3f8d9da5df3bc : SUCCESS in 1h 26m 12s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/3c6f9bf057d34f6aa578466f04e2f5cc : SUCCESS in 1h 20m 33s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/91dbe3217134494a932e79946ae09867 : SUCCESS in 1h 23m 32s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/7a6c1bbc3a9f4609ae8190f7ffdd662c : SUCCESS in 1h 03m 05s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/e7d542bef1744838a27abbbc858b25b8 : SUCCESS in 1h 37m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74fc61121ec34d5d95b5e14632bfe04f : SUCCESS in 1h 36m 53s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/6a5b0bb51f1f4193920dd7335c6509f6 : SUCCESS in 2h 09m 34s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/f0839c7cf70b4735a4fe57a0ac9ec20d : SUCCESS in 1h 33m 47s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/97f0c721ffd640aeb1448a5d9fdd628c : SUCCESS in 1h 20m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/44daed723f2246448b03ff63cde260de : SUCCESS in 37m 59s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/9b5d2a2dc6ae4e1e9a6b530aa35adf81 : SUCCESS in 33m 51s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/50aa2ecdb57a44d58f2d73127f548842 : SUCCESS in 35m 14s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/565f4c9a046c48be8df86b239cd0a45d : SUCCESS in 37m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1058940e3a99461d9eb49d826a40b9cc : SUCCESS in 1h 36m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5d6c52c23544ad4bb7dc70af8a7ba45 : FAILURE in 9m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/54d259d737d6401ea1e028e00095946d : SUCCESS in 2h 23m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0478ac733b5449ccbca6862b8bfdb92a : SUCCESS in 2h 55m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79edc90c42ac4158a9c2844818961123 : SUCCESS in 34m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/6f6bde7c0def49c187616dfe300ecddd : SUCCESS in 30m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/91b1753f6ad74755b6069fd81ee834af : SUCCESS in 33m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/79bc972211134f7f805adbbf13b75ddb : SUCCESS in 35m 04s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/bd0a1f34679d4c7296fb7d9d9eb1f22e : SUCCESS in 30m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3546136a68014b54a7922f3047daa662 : SUCCESS in 1h 50m 14s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/e9b7a9158e00440db498967b73f28d3a : SUCCESS in 35m 07s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b05b5c991b454df38ac02c88657c113f : FAILURE in 55m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d4f6e03286406d597a404732379bcc602c80776b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-02 13:52:47.000000000","message":"Patch Set 2: Code-Review+2\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"87b53dab5bca054b30b4585c6be2121be53452bd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-02 13:59:19.000000000","message":"Topic tox4-fix removed","accounts_in_message":[],"_revision_number":2},{"id":"2a5409cb77f7c97b229a694ad6da7adb45a7fec0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-02 13:59:26.000000000","message":"Topic set to tox4","accounts_in_message":[],"_revision_number":2},{"id":"6a8f212bd7f2d557e9022b09f0f3121a7d13f64d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-02 14:11:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f0a35f3317ee468add0b536ebc6c9b8333f5d213","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 14:11:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dd1dca07903e701ff12f0d15f5c19f331d88884a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 16:39:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5a4edea2bb649f39ebafcdba182f861\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0defd6b2ec43c2ac0126d84f05524d : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e485d6a239b42e2b976b7985dbd8214 : SUCCESS in 2m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3e167ce5bf148f2bdaa21f7f201c581 : SUCCESS in 2m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/797eb45c75a54f49a739a8beb2686202 : SUCCESS in 1h 48m 21s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/36b300a0d5c449dc8340d642187cba58 : SUCCESS in 1h 20m 41s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/1e94cb304d304df3b2ae6cdf08a76640 : SUCCESS in 1h 03m 35s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/b42972ce84cc4cf4b808e74bf5af445b : SUCCESS in 47m 12s\n- octavia-v2-dsvm-noop-api-scoped-tokens https://zuul.opendev.org/t/openstack/build/9c3f6b96348c406e9b560bd89ccc20a9 : SUCCESS in 1h 39m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a540b788d744fc4ae45cab88401507d : SUCCESS in 2h 17m 23s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/83d81bab432f4587a47e8a9466721c3c : SUCCESS in 2h 05m 15s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/23203f485bdc4d71827ce46976156a00 : SUCCESS in 2h 25m 20s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/333087c2cb6642f69e80bd943c8ebcdf : SUCCESS in 2h 00m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/111809174b7e48708c58a8675b83f054 : SUCCESS in 36m 12s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/da8f69f2f46d4b4dbf224e9934addf69 : SUCCESS in 34m 23s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/472372af328e4fe48e14e69848d9a04f : SUCCESS in 32m 29s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/ad4b17b2aa0e4e17844af11ff0280d85 : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":2},{"id":"68a211ee46ee01fa4daf409b7141cc376c4f7f7a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 16:39:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ae312a249ef723896998147b73d470ab3a4fe91d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 16:40:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ade433558d945cd83a3ab2fc5da66c5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44103bb07097403786a4ae72fa4163e3 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b05fdb0a6e24f6fb2a200151ff2c602 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"577953114e2fe3750c93f05fd5be0602bb06e267","revisions":{"f883601a9e7fd070d893a7ec900520f8bd5a3553":{"kind":"REWORK","_number":1,"created":"2022-12-29 15:16:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/70/868870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/70/868870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/70/868870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/70/868870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/70/868870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/70/868870/1"}}},"commit":{"parents":[{"commit":"dddf6515ab4cec0d02255f17bb6a41660eca6289","subject":"Merge \"Add scenario test for Prometheus over TLS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/dddf6515ab4cec0d02255f17bb6a41660eca6289"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-29 15:16:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-29 15:16:12.000000000","tz":60},"subject":"tox skipsdist seems to be problematic","message":"tox skipsdist seems to be problematic\n\nChange-Id: I293a52fe12972594690970b7d3bd91782b642669\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f883601a9e7fd070d893a7ec900520f8bd5a3553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f883601a9e7fd070d893a7ec900520f8bd5a3553"}]},"branch":"refs/heads/master"},"577953114e2fe3750c93f05fd5be0602bb06e267":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-29 15:20:38.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/70/868870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/70/868870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/70/868870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/70/868870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/70/868870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/70/868870/2"}}},"commit":{"parents":[{"commit":"dddf6515ab4cec0d02255f17bb6a41660eca6289","subject":"Merge \"Add scenario test for Prometheus over TLS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/dddf6515ab4cec0d02255f17bb6a41660eca6289"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-29 15:16:12.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-29 15:20:27.000000000","tz":60},"subject":"tox skipsdist seems to be problematic","message":"tox skipsdist seems to be problematic\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/868527\nChange-Id: I293a52fe12972594690970b7d3bd91782b642669\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/577953114e2fe3750c93f05fd5be0602bb06e267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/577953114e2fe3750c93f05fd5be0602bb06e267"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[]}
