)]}'
{"id":"openstack%2Foctavia-tempest-plugin~878919","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I6fe0e62b0dcd2f139bbb9d8c1aa2743e035ec8b4","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"py-spy","hashtags":[],"change_id":"I6fe0e62b0dcd2f139bbb9d8c1aa2743e035ec8b4","subject":"DNM profile o-* using py-spy","status":"NEW","created":"2023-03-29 15:24:30.000000000","updated":"2024-04-23 11:23:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"debdcac5033934bc0c5e786f36bdacc1e187365f","_number":878919,"virtual_id_number":878919,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-23 11:23:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-03-29 18:35:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"334341771e053c72e969f22aa72e860ef72d39c1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-29 15:24:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72db5e2ad347a7a0783789e86bb14ad73d470dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 18:35:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa5ba29132434d8888c4fcf6eb808acd\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2b65d1eb7fcc43ee924573e7f9fb8f3e : FAILURE in 24m 07s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/9c39b2354cc14cdfb1b3419e3abaff4a : SUCCESS in 1h 26m 41s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/6e1f7ba2b9cc4deb88d27bc53cbd031b : SUCCESS in 1h 42m 35s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/fb64d9736e434f5ebcac36470f45dd67 : SUCCESS in 58m 52s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a63f2125229845bfa18618b447eedc34 : FAILURE in 35m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1743d31499a1537e806a60308192cc805a3b529d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-06 08:08:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5aade687f7e64246e5df52d785ed9f07a2b9222","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 08:08:33.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error was:\n\n  while parsing a block collection\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 501, column 7\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 507, column 7","accounts_in_message":[],"_revision_number":2},{"id":"53bb6ecc248a80dd337f6e1f6ac72d0e1fc440cb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-06 08:17:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6a0bcecb8528a1cd243de495e24d35c104b592f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 08:17:57.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia-tempest-plugin on branch master.  The error\nwas:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).\n  \n  The incorrect values are around:\n  \n  job:\n    name: octavia-v2-dsvm-noop-api\n    parent: octavia-dsvm-noop-base\n    timeout: 10800\n  ...","accounts_in_message":[],"_revision_number":3},{"id":"782443a71dfff83811d2e73c7affe06f11d02c1e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-06 08:27:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ded328367a8b7a88b74959fa6258d20fa07074fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 10:51:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df9654397beb4529b082c23ae7f550b1\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7dd720c041ce4689962ec3d01473f709 : POST_FAILURE in 2h 04m 48s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/f43102c3c0da41faae36a5f9f95b7ecb : POST_FAILURE in 2h 16m 49s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/8da01f6e813344afa7f67ac23e3ad07a : POST_FAILURE in 2h 03m 32s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/c5e507ed7643459488acbb9f4b9134c2 : POST_FAILURE in 2h 07m 41s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9d0491f843964746b57b53c5f74634cb : POST_FAILURE in 2h 16m 54s","accounts_in_message":[],"_revision_number":4},{"id":"8312293bd6e4dca27ed9ec1c89a122dd8f2e083e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-06 11:19:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1751ab12c8f708edfd32c8d9555f2ed8368ec783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 13:51:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ac7979884d94074b9e8417fd8c702b2\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/00f46552c07a46a3bae76430f53ca88d : POST_FAILURE in 2h 13m 58s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/d87957a937ca4e7ab3048f9fe125edf1 : POST_FAILURE in 2h 08m 06s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/77cb125e6b84485791634f464c6af3dc : POST_FAILURE in 2h 01m 19s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/195b00f2742f4f5da441db8b766d2cb8 : POST_FAILURE in 2h 16m 02s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/398f67aaeae14463b9b803dff28efd5c : POST_FAILURE in 2h 08m 23s","accounts_in_message":[],"_revision_number":5},{"id":"07fa2596523c85a9160ce839c16e38661ff4cd97","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-03 13:41:47.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"9f4cab5d0d2bf331edaae1eddc8827df5920873d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 16:12:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1537409a48184d95b4133e78b254fb42\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53f622188d6b4a96b4f5b1f60bf0f625 : POST_FAILURE in 2h 22m 19s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/a2152b76631b4453bf2e0f982881fdad : POST_FAILURE in 2h 11m 58s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/cec3c461bb294371b954a75bfda1c63c : POST_FAILURE in 2h 10m 04s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/d6870f6685a64ff1a91f3f3c15e2fd81 : POST_FAILURE in 2h 07m 59s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d7831b99ed394e87af9b20f777c79b21 : POST_FAILURE in 2h 08m 34s","accounts_in_message":[],"_revision_number":6},{"id":"cab1ccac02cfc3dfe6614e089916535c8a76460d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-06-20 10:46:46.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d839013a1fdbdc4587910d1b2b317f3acbc314b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 12:58:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36c1a07541fd41a690ddca5fa5709ea8\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a7ab8f39edea422488197bfb1624e614 : POST_FAILURE in 2h 07m 04s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/0842b94ae8ba4a41808a01778bf63b44 : POST_FAILURE in 2h 04m 45s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/16d71006e2e540d6ba48d1ecbea68cb3 : POST_FAILURE in 1h 59m 28s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/6fda19cd6576487581fad2dc402c6eb6 : POST_FAILURE in 2h 02m 25s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/77c90a8c040e464183bf4d15c756007d : POST_FAILURE in 2h 10m 36s","accounts_in_message":[],"_revision_number":7},{"id":"6cb6c3950d5a594f8537ea556122bfda2cc05718","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-30 08:56:41.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"3ae1ab062742a7a14c55b0a3532971e617d1b3ad","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-30 11:26:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8487b4d77916462ae9b62b95fd22a4d68a303103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 13:30:57.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fe6efd742084fa98beea151674e87ca\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/154822dcb19c44b28a1e3827fa0ff39f : POST_FAILURE in 1h 59m 25s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ad5ba082489a44108f34ac7e4f2c9972 : POST_FAILURE in 1h 59m 58s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/a034c6242a0e439a99d1770a4c2db556 : POST_FAILURE in 2h 00m 11s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/ff09466d00c546998190cf09ad7fffa8 : POST_FAILURE in 2h 02m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dd5cfb6947a248e09badef6ffc7e7a15 : SUCCESS in 1h 35m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/b795aaa40e95448ea449041ee07583ba : SUCCESS in 1h 28m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/1fbe9a3b251348ddbf6f117bb9e9c26b : SUCCESS in 1h 15m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/42c9251ebc3d47dfbc65ea4381a374a0 : SUCCESS in 1h 15m 10s","accounts_in_message":[],"_revision_number":9},{"id":"3d0422ec83c106c6baf7097a5b1afec7ec54116e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-23 09:08:32.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"debdcac5033934bc0c5e786f36bdacc1e187365f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 11:23:29.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20fb747922f748589acb09d822893509\n\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8016ec01ab364ea3abc19c50f7f57147 : POST_FAILURE in 2h 00m 25s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/d2072672ef9b4154abdce09a17116f9b : POST_FAILURE in 2h 07m 27s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/abc81448b0d44b4782b57a342d9777c8 : POST_FAILURE in 2h 00m 00s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/bddca4c928aa47eeb4644a4b738a20c8 : POST_FAILURE in 2h 03m 30s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b41541f5e15841c6b7d26f2556109f53 : POST_FAILURE in 2h 09m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1fddd275ee164542a607b3f2a9050d2e : SUCCESS in 1h 37m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/41753857f61245da97fd7b871a2f8de6 : FAILURE in 1h 32m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2e3c169bcaa941c29decc2d0cf420a86 : SUCCESS in 1h 25m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/0fbdd81d4ac44203a28fc64b115dc696 : SUCCESS in 1h 24m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/2e22543efd5e469683ba5770b4a741ca : SUCCESS in 2h 01m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/8bcdbdd5846f470b857476a03483a2a7 : SUCCESS in 1h 06m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/03391a701e3e4e56a45305787ff90e34 : SUCCESS in 1h 17m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/346edbdfe8b145a5b0955ad7b4b4313a : SUCCESS in 1h 18m 45s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e412fddab11bf35c3a5c0e4d474d5d2a0dc10f3d","revisions":{"eb27551d3345538079c736a555b08dce52917e48":{"kind":"REWORK","_number":1,"created":"2023-03-29 15:24:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/19/878919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/19/878919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/1"}}},"commit":{"parents":[{"commit":"cbda95e7279e3da03ced8abff040ad8e0c2a9685","subject":"Merge \"Spliting centos scenario job into 2 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cbda95e7279e3da03ced8abff040ad8e0c2a9685"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-29 15:24:01.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-29 15:24:01.000000000","tz":120},"subject":"DNM profile o-* using py-spy","message":"DNM profile o-* using py-spy\n\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/878917\nChange-Id: I6fe0e62b0dcd2f139bbb9d8c1aa2743e035ec8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/eb27551d3345538079c736a555b08dce52917e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/eb27551d3345538079c736a555b08dce52917e48"}]},"branch":"refs/heads/master"},"9fb70e7876c177cfb97db2e5c14157e06a851711":{"kind":"REWORK","_number":2,"created":"2023-04-06 08:08:05.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/19/878919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/19/878919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/2"}}},"commit":{"parents":[{"commit":"cbda95e7279e3da03ced8abff040ad8e0c2a9685","subject":"Merge \"Spliting centos scenario job into 2 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cbda95e7279e3da03ced8abff040ad8e0c2a9685"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-29 15:24:01.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-06 08:07:35.000000000","tz":120},"subject":"DNM profile o-* using py-spy","message":"DNM profile o-* using py-spy\n\nChange-Id: I6fe0e62b0dcd2f139bbb9d8c1aa2743e035ec8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9fb70e7876c177cfb97db2e5c14157e06a851711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9fb70e7876c177cfb97db2e5c14157e06a851711"}]},"branch":"refs/heads/master"},"d96b77f78f9b50507ee72a1e46b502251bc44809":{"kind":"REWORK","_number":3,"created":"2023-04-06 08:17:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/19/878919/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/19/878919/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/3"}}},"commit":{"parents":[{"commit":"cbda95e7279e3da03ced8abff040ad8e0c2a9685","subject":"Merge \"Spliting centos scenario job into 2 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cbda95e7279e3da03ced8abff040ad8e0c2a9685"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-29 15:24:01.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-06 08:17:22.000000000","tz":120},"subject":"DNM profile o-* using py-spy","message":"DNM profile o-* using py-spy\n\nChange-Id: I6fe0e62b0dcd2f139bbb9d8c1aa2743e035ec8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d96b77f78f9b50507ee72a1e46b502251bc44809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d96b77f78f9b50507ee72a1e46b502251bc44809"}]},"branch":"refs/heads/master"},"a302372cdfb554a496cb18e5cf55524448ded9b9":{"kind":"REWORK","_number":4,"created":"2023-04-06 08:27:29.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/19/878919/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/19/878919/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/4"}}},"commit":{"parents":[{"commit":"cbda95e7279e3da03ced8abff040ad8e0c2a9685","subject":"Merge \"Spliting centos scenario job into 2 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cbda95e7279e3da03ced8abff040ad8e0c2a9685"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-29 15:24:01.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-06 08:27:19.000000000","tz":120},"subject":"DNM profile o-* using py-spy","message":"DNM profile o-* using py-spy\n\nChange-Id: I6fe0e62b0dcd2f139bbb9d8c1aa2743e035ec8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a302372cdfb554a496cb18e5cf55524448ded9b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a302372cdfb554a496cb18e5cf55524448ded9b9"}]},"branch":"refs/heads/master"},"307d68824e3bf95f58fe9432871e6dbc2a91bcdf":{"kind":"REWORK","_number":5,"created":"2023-04-06 11:19:52.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/19/878919/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/19/878919/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/5"}}},"commit":{"parents":[{"commit":"cbda95e7279e3da03ced8abff040ad8e0c2a9685","subject":"Merge \"Spliting centos scenario job into 2 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cbda95e7279e3da03ced8abff040ad8e0c2a9685"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-29 15:24:01.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-06 11:19:44.000000000","tz":120},"subject":"DNM profile o-* using py-spy","message":"DNM profile o-* using py-spy\n\nChange-Id: I6fe0e62b0dcd2f139bbb9d8c1aa2743e035ec8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/307d68824e3bf95f58fe9432871e6dbc2a91bcdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/307d68824e3bf95f58fe9432871e6dbc2a91bcdf"}]},"branch":"refs/heads/master"},"285104df328a3ba45b4689090a76ee6906d41fa2":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-05-03 13:41:47.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/19/878919/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/19/878919/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/6"}}},"commit":{"parents":[{"commit":"e7f794e685c083b6b87832f6801f57d78095344b","subject":"Merge \"Add h2 traffic/scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e7f794e685c083b6b87832f6801f57d78095344b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-29 15:24:01.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-03 13:41:47.000000000","tz":0},"subject":"DNM profile o-* using py-spy","message":"DNM profile o-* using py-spy\n\nChange-Id: I6fe0e62b0dcd2f139bbb9d8c1aa2743e035ec8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/285104df328a3ba45b4689090a76ee6906d41fa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/285104df328a3ba45b4689090a76ee6906d41fa2"}]},"branch":"refs/heads/master","description":"Rebase"},"7e78efcfc09d5605d81a0f9c4313d1992d7cae91":{"kind":"REWORK","_number":7,"created":"2023-06-20 10:46:46.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/19/878919/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/19/878919/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/7"}}},"commit":{"parents":[{"commit":"b9af9163cb7a5277e4c72dc1becb563148f7fa1b","subject":"Splitting scenario jobs in 2 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b9af9163cb7a5277e4c72dc1becb563148f7fa1b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-29 15:24:01.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-06-20 10:46:37.000000000","tz":120},"subject":"DNM profile o-* using py-spy","message":"DNM profile o-* using py-spy\n\nChange-Id: I6fe0e62b0dcd2f139bbb9d8c1aa2743e035ec8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7e78efcfc09d5605d81a0f9c4313d1992d7cae91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7e78efcfc09d5605d81a0f9c4313d1992d7cae91"}]},"branch":"refs/heads/master"},"ae4e26a7cbda58d473c6082a713da2a0c2d3e6e6":{"kind":"REWORK","_number":8,"created":"2024-01-30 08:56:41.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/19/878919/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/19/878919/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/8"}}},"commit":{"parents":[{"commit":"ffc51f76b2a9c0e6b148d360c3f8b4c712fb84ca","subject":"Merge \"Drop implementation to use pyOpenSSL to manage pkcs12 certs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ffc51f76b2a9c0e6b148d360c3f8b4c712fb84ca"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-29 15:24:01.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-30 08:56:31.000000000","tz":60},"subject":"DNM profile o-* using py-spy","message":"DNM profile o-* using py-spy\n\nChange-Id: I6fe0e62b0dcd2f139bbb9d8c1aa2743e035ec8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ae4e26a7cbda58d473c6082a713da2a0c2d3e6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ae4e26a7cbda58d473c6082a713da2a0c2d3e6e6"}]},"branch":"refs/heads/master"},"40e55fe24516e105b84f7e8348f8cd74ae332ea9":{"kind":"REWORK","_number":9,"created":"2024-01-30 11:26:26.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/19/878919/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/19/878919/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/9"}}},"commit":{"parents":[{"commit":"ffc51f76b2a9c0e6b148d360c3f8b4c712fb84ca","subject":"Merge \"Drop implementation to use pyOpenSSL to manage pkcs12 certs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ffc51f76b2a9c0e6b148d360c3f8b4c712fb84ca"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-29 15:24:01.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-30 11:26:16.000000000","tz":60},"subject":"DNM profile o-* using py-spy","message":"DNM profile o-* using py-spy\n\nChange-Id: I6fe0e62b0dcd2f139bbb9d8c1aa2743e035ec8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/40e55fe24516e105b84f7e8348f8cd74ae332ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/40e55fe24516e105b84f7e8348f8cd74ae332ea9"}]},"branch":"refs/heads/master"},"e412fddab11bf35c3a5c0e4d474d5d2a0dc10f3d":{"kind":"REWORK","_number":10,"created":"2024-04-23 09:08:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/19/878919/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/19/878919/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/19/878919/10"}}},"commit":{"parents":[{"commit":"a1a2bed19ffe387212364018c61f4a2ae1cff78b","subject":"Merge \"Add *.orig to .gitignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a1a2bed19ffe387212364018c61f4a2ae1cff78b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-29 15:24:01.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-23 09:08:24.000000000","tz":120},"subject":"DNM profile o-* using py-spy","message":"DNM profile o-* using py-spy\n\nChange-Id: I6fe0e62b0dcd2f139bbb9d8c1aa2743e035ec8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e412fddab11bf35c3a5c0e4d474d5d2a0dc10f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e412fddab11bf35c3a5c0e4d474d5d2a0dc10f3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
