)]}'
{"id":"openstack%2Foctavia-tempest-plugin~881546","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ie54714015e943fd1cb75ca95f8cf241fbc99268c","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"add_tls_terminated_api_tests-patch16","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-04-30 11:40:06.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-04-30 08:29:53.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-04-24 17:05:29.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-04-30 11:40:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie54714015e943fd1cb75ca95f8cf241fbc99268c","subject":"Test new HSTS feature","status":"MERGED","created":"2023-04-26 10:33:16.000000000","updated":"2024-04-30 11:41:16.000000000","submitted":"2024-04-30 11:40:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":36,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"881546-add_tls_terminated_api_tests-patch16","meta_rev_id":"7f3214134b5c66e84b2d344645c8c585d88925b9","_number":881546,"virtual_id_number":881546,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-30 11:40:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2024-04-24 17:05:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-04-30 08:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-04-30 08:29:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-26 11:07:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-23 20:46:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-26 14:45:46.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-10-30 11:52:51.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-02-06 07:57:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-04-23 11:56:18.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"7d0d25c917b3fa6e6abe392b296124e618e13268","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-26 10:33:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1caaa6d16d0ee1471dd4c47974424029f4ab430a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 11:07:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a51e499d9e6340f1a1aaad4636e088d0\n\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/62689f28533a454c917771618f2cdd06 : FAILURE in 33m 14s","accounts_in_message":[],"_revision_number":1},{"id":"fb3c242bbba439329279e32782233f89f8cafa2c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-26 13:02:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ac292a80b0e27871b8cdc049bb036ab45b553342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 13:50:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/281776cda22a45c89dc9e7dca66b955a\n\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ea623071c014ee9b8bc63656c6bb27b : FAILURE in 45m 59s","accounts_in_message":[],"_revision_number":2},{"id":"0d4535d724df6f296f765b13c5747ff48521edc2","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-27 10:44:57.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1739a5868dd6ec0adf204ab0e651d3bbd7581661","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 12:18:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8fe4ccf537f4991883503ffc1d43019\n\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4017e3602ae44ee29f0bb08b69af10e1 : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":2},{"id":"911002c136097d67cc44a6b9ae4d54ff95b5697f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-03 13:21:59.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"f7a14afbf01277bcc06ef3ef48a2e3ee0d1a8627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 13:22:26.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/octavia for 879749,27","accounts_in_message":[],"_revision_number":2},{"id":"de22d9e32d0894984ceead51d10bc8fd0036d9eb","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-03 13:22:35.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"939657bb43ebd63d276eaeb38df0e2f03f003103","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-03 15:04:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"477b8a042b8dee1abf96c2ca5cf1953d0e208603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 15:46:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03a734bac3bf4d2db349f7422381dc3f\n\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4824eccc407c44d285199c37dc043b6f : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":2},{"id":"fc4801035a81b1f8ce16a1dabc547efe711ebf3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-04 13:44:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4fe0764f8f65dedcdd1d981242207d000baa348d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 14:30:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/768179abd8dc4b63bea7fe3425835f2e\n\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8becc8aff05848909fbd49ebc1aeac50 : FAILURE in 42m 17s","accounts_in_message":[],"_revision_number":3},{"id":"633b1998b0a81c6bc4fe47264046fe2f09f60cfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-04 14:44:36.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3b50a769081fa8b240a75cbe7419de258e42485f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-04 14:54:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8f3e667369c0fe9554df329468ad63cc9c625828","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 15:46:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e8451db2dd74ee7aa0c331103d7a5c2\n\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/397cd4b6a6364766b18ac0beb8160030 : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":5},{"id":"c08a773c548d516c032b5e48c3bf1e2c52b41b86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-04 16:16:48.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"6869be04de9ad841c45007aeb867794652a8ccc9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-04 16:26:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"74177ef64ad81b3317252c42e17a401f5fb03dfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 19:43:16.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/513ae4b480df4af9a1a9cbc07508e81d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3eb714936324c9b8f033deb6675972b : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2fb0890a5ca4f1daf0a2887cdf6d464 : FAILURE in 3m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dfb87cbdb9844fd7a752dfbe0271306c : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/b1888dc6770e4c7aa2b758b515e264c5 : SUCCESS in 1h 43m 50s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/bbe9a331a7d74ab58de73832ea722608 : SUCCESS in 1h 27m 52s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/2c8ab7f24acf4ebcbd6164d2b8d4719d : SUCCESS in 1h 32m 51s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0db85169adf744beaab98e0eecda1ce6 : SUCCESS in 1h 39m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b71c4346a9c4236b6539df3dead3cde : TIMED_OUT in 2h 33m 58s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/f0219339ec35493f9aaa4787d6f76cf7 : SUCCESS in 2h 23m 46s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/8b4c1c8b35f541d4a84611ada35bc999 : SUCCESS in 2h 13m 39s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/80f08a64cf704bf0af008f5898928ad4 : SUCCESS in 2h 22m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a42d7ec77950494d906a9d6147911fa2 : SUCCESS in 49m 51s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/fd21431110114134b53e608feccf6e6f : FAILURE in 45m 09s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/ecf6892567e442749c676945e78581ea : FAILURE in 45m 36s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/205fb1ae4d504348b15ccef10c2e803d : FAILURE in 45m 56s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8f200be878ed4714b88938af9e38e1f2 : TIMED_OUT in 2h 33m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/c793c753e29647238b4fe151e91bca76 : FAILURE in 14m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/0e9fa0d5d08e4b1cb091a2431cf67a1c : SUCCESS in 2h 04m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c11c8e98a04d455782e0d4b8c72491d1 : SUCCESS in 1h 32m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/66abbe7f96e646979e0082c6af52b12f : TIMED_OUT in 3h 04m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/196ee95fba7c45f4b7310a9a03fc524c : SUCCESS in 37m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/dd4223a2182a49d8a51112610cb43f72 : SUCCESS in 48m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/6153b9fe0e4e49c99f3f2ad78e3d1f8f : SUCCESS in 43m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/87f32b2c754b402fb4aefc059f1e00ad : SUCCESS in 42m 51s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/6148c8066c4549abb8720f1d221a1fdc : SUCCESS in 39m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c0b5a191cd724d11aa14c33f602c7420 : TIMED_OUT in 2h 33m 44s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2122681dada44d4c929c2ed57340da04 : FAILURE in 36m 41s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/ec8bd3df1a0b4ec2820019a5966b855d : FAILURE in 44m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7a4d29d7e88696ddc1247eb802fabe7c92c623b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-05 06:33:32.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"df8a1b54eaa2929c2154e2aa826fb58893c13eae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 06:33:54.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/octavia for 879749,28","accounts_in_message":[],"_revision_number":7},{"id":"7f42c23c6216a4f206ea8b978d0951deceb3d561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-05 06:36:57.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"bc0eedc3c06eab76a38c554e92bd733cc51f97e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 06:37:18.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/octavia for 879749,28","accounts_in_message":[],"_revision_number":8},{"id":"b3609ef770c48f7f2bcff405954ce59ccb1f7c54","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-05 06:49:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a8c78b259d46e649950465cb107c88e2c4adebf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 06:50:10.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/octavia for 879749,29","accounts_in_message":[],"_revision_number":8},{"id":"116eb4e2e8d4f67130ae94768b0a7dca7bea00ba","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-05 11:20:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6b3139214ef93ed240c2ed5d2d53bc54ceece226","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 14:30:42.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b9e193b243b4a18820187e0fd9920d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0103f7973327431893d182e155e93219 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5860439336b4e07ae937fd9d76bd65d : SUCCESS in 2m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/074053b6593342bba7038ff7db711abe : SUCCESS in 1h 43m 34s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/a3d3131b02a843589e54f74d59eb3c9b : SUCCESS in 1h 23m 15s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/360669663dc24a0681878f973d9cae15 : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/90c2681b5f7d467996baaeb275f9de85 : SUCCESS in 1h 09m 26s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/27e8c534eff14003be23cd60a8ad078e : SUCCESS in 1h 35m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e3d6e33b68140d4951b5c489a047bc3 : SUCCESS in 2h 15m 22s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/81a71060f72a494594f078d0a44d27e1 : SUCCESS in 2h 08m 59s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/1245d562b22c416e98bc1a96118d7620 : TIMED_OUT in 2h 34m 09s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/c23625b5569b44b3b82e02aea5c96d0c : SUCCESS in 2h 18m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/da468f2f0ab9429b9fa397be7ce23688 : SUCCESS in 47m 45s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/3f8d0ea6369944faa54d056ead7ef579 : SUCCESS in 46m 17s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/c540615c1c404144ac8ce4588458b249 : SUCCESS in 42m 19s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/df4c3b0da57743f794aa87e8c915b8bf : SUCCESS in 38m 00s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/508336ccb04349719dc5aecf536c795d : SUCCESS in 1h 54m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/859d41ff47364718bbc1a1690503d702 : FAILURE in 9m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/596b320044e048beb69481d5152feef8 : SUCCESS in 1h 44m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0c294b6394f14e55ba3546b13a6fedc7 : SUCCESS in 1h 25m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/34edeb798c0047c0848c729132bf1430 : TIMED_OUT in 3h 04m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0372d3c72d14d96b2f18e66fe6b5ca8 : SUCCESS in 38m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/61f413dc837448a2b220996059907510 : SUCCESS in 36m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/1deec687537345c2a3d46cb3562e6d83 : SUCCESS in 35m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/2a7a0f7bb93b42b096f5aa56a1f42a22 : SUCCESS in 36m 36s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/8bce6cf70a2e450d8d9b7061e9b6435e : SUCCESS in 34m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9d3d3bed19eb4ba393aa54aecd6c4b21 : SUCCESS in 2h 27m 59s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/fd2e5ef3ecc1489db86bc1dc0c66ec9f : FAILURE in 41m 00s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/381e04d314034647b2cbfd038d482fcc : FAILURE in 42m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1b4f353bfd80c99788b2068f9bc40d88f204fbd1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 06:37:26.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3282354d9b975c5ef06fd1852171bd92a2ce5386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 09:58:08.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b239efe26bfa441ca6d9d8b6563bd8c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2413c206b2f4d0b93defd748a633e59 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9401d1c607e48d7b8ad60fb7e454a13 : SUCCESS in 3m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0249861013ae435685faac0aa102d7ed : SUCCESS in 1h 35m 15s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/c50dac934da04968950ef82808407a57 : SUCCESS in 1h 07m 33s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/90b619c2a0e442bfbe457941bcda9acf : SUCCESS in 1h 22m 38s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/84cff1bf571f44229734a86c5cca9982 : SUCCESS in 58m 09s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/29e9ec3fc69444058daf05325db524f5 : SUCCESS in 1h 33m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/637eddd7d1924851bb175492578008d1 : NODE_FAILURE Node request 300-0021159460 failed in 0s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/7f4e98b7ed074376b0523068a9b7ce5a : SUCCESS in 2h 32m 35s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/d4d8291659b043118bf3ea6eb86795b2 : SUCCESS in 1h 54m 39s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/c70b485e810040818db41d8ae95cd574 : SUCCESS in 2h 11m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f742f5b4d6334c58a4dfc42e30e9e718 : NODE_FAILURE Node request 300-0021159464 failed in 0s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/69f9db19a1d943a499c56df1217907d0 : SUCCESS in 39m 21s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/27983dd22deb4efdac2e68f22afdcfe3 : SUCCESS in 37m 05s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/445173cbd0644eb58de3fa38f52c31aa : SUCCESS in 36m 16s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5c1895104c8c4a95ab47a0e3129efb16 : SUCCESS in 2h 21m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/3542b6b510584e11a5e24d635941a70c : NODE_FAILURE Node request 300-0021159469 failed in 0s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/9d46f8f7a9e448798bd7860d23006fb5 : SUCCESS in 1h 46m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ec1ed2093fba478c82700f774480ae75 : NODE_FAILURE Node request 300-0021159471 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57639bc642c547918c5a547ecb8cd68b : TIMED_OUT in 3h 03m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eccf309471284c089b8bbfec03c4496d : SUCCESS in 36m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/2cf551eca2c1499bab197077679ed6e4 : SUCCESS in 34m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/41355660d4e948dbb699855dbcdca595 : SUCCESS in 34m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/2f09a6cb1be442789d58e32866cd2ee8 : SUCCESS in 34m 55s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/b20c7306706d47e6bf72142a67e42ec4 : SUCCESS in 34m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e49833bfbcf8415aae3d21ff52fd15fe : SUCCESS in 2h 29m 38s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/ddd8c874deba412fbab79cc0ec800bb3 : FAILURE in 45m 20s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/4e282da1180648ccae543d98eb177613 : FAILURE in 37m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3ddacf7b9c5f4dafdf82204014e6aaaf9e6745d3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 11:48:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"aadd80bc510cb0898285203f6b10e62af4b8ba46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 14:57:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7d9509871f94998aa43a3fa8c63c112\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74d3ffa7a5f545cda4d97f8a01ea3941 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28b8b18d1e084eab8856f1023bd9c492 : SUCCESS in 3m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f0aac8a15784c558c887e93001fe885 : SUCCESS in 1h 18m 08s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/3b4abafe3b27406f9b8ff21a8caed609 : SUCCESS in 1h 14m 46s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/ce95d1e1893a4977bee424c55ad89636 : SUCCESS in 1h 15m 37s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/b8bb2de7619547c2b684e1b4d568254e : SUCCESS in 1h 06m 40s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b41f8e6c7687497da2cf44eddc7f57cd : SUCCESS in 1h 21m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dfb06614b0ca4883b40f0e7edc9f8e63 : SUCCESS in 2h 27m 21s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/8233ad6eeb7349bd80d17f912796a509 : SUCCESS in 2h 07m 05s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/5be058226ab642e39d0a9d5a15ad6efe : SUCCESS in 2h 08m 43s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/2d75397c2aae420d95c8a1dae8791435 : SUCCESS in 2h 05m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/97119d6bb70e4d35889361e755601006 : SUCCESS in 52m 57s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/077fc0f458494a5b95a7b8855702d103 : SUCCESS in 49m 10s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/2ccde4d1507342fab21ffc0e69c2adf3 : SUCCESS in 46m 20s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/5949984ec0a94a7f9cc66a1fe676643a : SUCCESS in 42m 28s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/accb04352f164a2ca2b8d2b8fce4c2c7 : TIMED_OUT in 2h 34m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b5ad531b6554618a1338549d23561d0 : FAILURE in 9m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/9743b0385b1540af9d514efddadfb1f7 : SUCCESS in 2h 05m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e94aeb05a36440dfb39987c5534ff53c : SUCCESS in 1h 16m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/df213792877146bda6228dab50f5ee36 : TIMED_OUT in 3h 03m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d81981e8e82c4614add64ba95ff6b394 : SUCCESS in 41m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/a5552a4211a14af6886f8869c3ff23b5 : SUCCESS in 39m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/1159a2394529486fbfc0997596e3a158 : SUCCESS in 39m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/59d27b96dc554aa98eb6bab9e513fa11 : SUCCESS in 45m 11s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/de0717e6629f4a3fa1c50c7bef3a353b : SUCCESS in 41m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/aee2eb79365645f58bcce4b025621f9d : TIMED_OUT in 2h 34m 14s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/abc82bd42b854d389e2ecb3112c87a12 : FAILURE in 39m 02s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/9693e2afb1944779afa647685e01560f : FAILURE in 41m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"81c58508ca4479dd9ad17d3678788f6e61e35767","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-07-19 15:53:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2571fb6e1385a04d349c0e8cf850be4cf3b8f62a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-07-19 15:56:46.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"96c72dcd99ae69d77744db2b14712baa7c86a62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 19:02:42.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e932c31fb2c4b13a3225af2c499e191\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a229b7377984b38a79a265817296406 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ff0b8047d747ff9f42287fa3b5f328 : SUCCESS in 3m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/601864d431e142a7acc89c9070d158ca : SUCCESS in 1h 18m 33s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/74ea75aac98c454a85a76612fe0d7984 : SUCCESS in 1h 09m 54s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/4926ba2d23844ef58e5d8638b0cec51d : SUCCESS in 56m 56s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/b0b687963d9b46d28089811636739b47 : SUCCESS in 1h 18m 17s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7f9cd35a538d4e8c8c966638474ac693 : SUCCESS in 1h 53m 09s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/81df8e357b194ddf9527f0184e9fcd25 : SUCCESS in 2h 04m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fdae7aab79fc468e998f55c8c43907cb : SUCCESS in 1h 14m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/b70e0acf398a45b392dc66c95d3e783c : SUCCESS in 2h 20m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/af321ebf01c6412c84ea3f7674eee935 : SUCCESS in 1h 47m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/c7174620d8214101bd927a1ce0cf11fa : SUCCESS in 1h 32m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/77a944e55ae340bf8e764859627962df : SUCCESS in 55m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/bd3847dc6fad42e6a853a745916170bc : SUCCESS in 1h 13m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/72723f65727b44e8b8a65e33d3d8616c : SUCCESS in 55m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7d871f4b6f294ad2abbdb2d710cb20cc : FAILURE in 51m 18s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/6642c2c37add45c48bf4c8072ac930f2 : FAILURE in 44m 36s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/0423e0b8e0cb48efab05b529dcaf27e5 : SUCCESS in 36m 46s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/24bf66ae4d6b4a75bcc6f6b0e640d65e : SUCCESS in 42m 24s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/de12f5f949904921af3414864dd421d5 : TIMED_OUT in 2h 34m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/42f6befee5f54eb890b08a90e48d04c0 : SUCCESS in 1h 48m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b7f3cacee43a46b6b3dd01f1b0dba265 : SUCCESS in 1h 36m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9550e1ee3979433a8207cf20a6b5af39 : TIMED_OUT in 3h 03m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db363acfe4c94e89b412514281bc4385 : SUCCESS in 42m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/46d4c2b3b6b24738bb63bef703394ba3 : SUCCESS in 46m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/983e996d1cf540399ad4249f1970d1af : SUCCESS in 45m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/3d8b000be7b9495aac205fb3a52622dd : SUCCESS in 35m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4df9db23ed6540eabfe5bdb0b5114aab : TIMED_OUT in 2h 33m 06s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/6d01269e3b384bb2bc411207207fbc45 : SUCCESS in 1h 01m 23s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/085da70b4af74c68afb60b96da042ee3 : SUCCESS in 1h 03m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ee860df0903bc8d2a7a7d76f447e873f8145e2e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-07-24 09:29:04.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"275b9d49142fdd7db1a4413eef48dc51276334c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 12:34:37.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00d471ff262b4f8d91343ee6874e138d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44b12e912f9b4855b46c7e02bdc5d590 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/392ec19d140647939337d61dd4ff586f : SUCCESS in 3m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/94d0fdee359b4746976bc5ddcb8b2589 : SUCCESS in 1h 09m 28s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/b35ba9594055451d9938db345c2ae756 : SUCCESS in 1h 06m 49s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/3fb7b98416b048749f6325f51d9a109d : SUCCESS in 1h 27m 26s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/6a0fb2333987432b8e9fb9777a9f4d10 : SUCCESS in 1h 13m 55s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f75fe3159eba471eb02572530f035a17 : SUCCESS in 1h 32m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b2f12183a84549dbbee229ba20e173f2 : SUCCESS in 1h 48m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/017e34fc1d0f4f68981bbf86e9b756c2 : SUCCESS in 1h 21m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/f13ff2f8cd1e4b3497172a0585aa64bb : SUCCESS in 1h 26m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ace491532ce24c838dbbcfa86bd8e936 : SUCCESS in 1h 15m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/f7abe304bd0641af8a75caa4ec0a65a9 : SUCCESS in 1h 16m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/21008fd9de7f4bc49fde4066913053f8 : SUCCESS in 56m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/b2c77d4037e2486f8063ace8c6b1d192 : SUCCESS in 1h 38m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/31801ff584b749ef8448f7ac40cb820b : SUCCESS in 56m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d26e25ece2f45e9a08d4c9969a041f6 : SUCCESS in 37m 12s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/d743055f5eb34992b5b48dcb3f55bcc2 : SUCCESS in 46m 23s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/422b0084d1764ea58ffb28a58367eb24 : SUCCESS in 38m 04s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/d49f02cadfe547d3a6c3f14cdc7bbeff : SUCCESS in 43m 01s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/91ff9f07207d476b9cda7e9c2cda2276 : TIMED_OUT in 2h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/43e5761928bf4656af44bd6bb73aff5b : SUCCESS in 1h 40m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/08fa24ab32c54ffbafc6704137d1a9c3 : SUCCESS in 1h 22m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6d32285c7bf944e2b1351123015c5307 : TIMED_OUT in 3h 03m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99a55be310e34da688fc17e40c4b33bc : SUCCESS in 32m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/452b33a9ca934882bc55ff006c2847f4 : SUCCESS in 40m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/83c55f74f26a4891ae32ab6fdd2497c7 : SUCCESS in 33m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/e1f3bba2cdde40479fdd2af65a700dd0 : SUCCESS in 39m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6145910ce34943ccac01ebf45f5e6634 : SUCCESS in 2h 28m 42s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2297407e943d488e8ce41a0f12ff68d5 : SUCCESS in 51m 57s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/41344b2d4e75463aa1e5d749d9f1ac2c : SUCCESS in 49m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a11efc2af37a189e1a5bf7659fdc5e76f2322c83","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-23 12:02:10.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"cf2f813088cbda034de0b05a7fe29ac66b927651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 15:10:38.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79b1d6c3a48e42089e97868ff009ab2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68dceae7cb91430693be26c6cfac703b : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbeadabd9270491e9aaf3c386eecab67 : SUCCESS in 3m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4de86d1fa2164568b054aabd6ebfd23f : SUCCESS in 1h 24m 49s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/2195cb662a6c4fa4bcfc4c3a2b05553b : SUCCESS in 1h 09m 26s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/d9ef8e1c3c314d8aad84fb454a7609f9 : SUCCESS in 1h 02m 35s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/b88d8c318c214356b8c42defdec6932d : SUCCESS in 1h 11m 41s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/24a27ee7ad9844d8bcba6bb546b2b74b : SUCCESS in 1h 20m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d4e8ccde119648538474f8993538ef1d : SUCCESS in 1h 59m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/23b338e4f3c0468d90df881249c5ac03 : SUCCESS in 1h 13m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/9b8f40b79d3c4a9282e47c608f73d1bc : SUCCESS in 1h 57m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ad9ef80143904a6e980d625f538901ee : SUCCESS in 1h 21m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/bdd2df9a41f9464e885228d3d5bece4c : SUCCESS in 1h 15m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/41b267bb3cd44a98922bf3f90eb8a3af : SUCCESS in 1h 20m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/79e212fd98394692bc0151a89eae0e05 : SUCCESS in 1h 39m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/aeaa2b9337824041a565526b6e38dbae : SUCCESS in 1h 15m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b43d8a66f61e4d6189d584073b554f4d : SUCCESS in 41m 09s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ce7eb13ec79c4c0880748f4427a0ce00 : SUCCESS in 43m 44s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/2784712d213b4da09e3b6115af2af722 : SUCCESS in 36m 42s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/68a028e569824c67af8e8247253d014a : SUCCESS in 44m 36s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0cc0a1642747451785fd9e52e644e36b : TIMED_OUT in 2h 33m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/564437c405b04659b5b8589f557ce236 : FAILURE in 18m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5556f70466424b3b9f40e6c99d57530c : FAILURE in 17m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/76e0c22425fe4d3d8a191ba5ff970e99 : TIMED_OUT in 3h 03m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/804a4911947b43168adfa31438d642ec : SUCCESS in 39m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/12b980df6f974ad5930fe84ae3b04179 : SUCCESS in 34m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/d001524c3bab4877a614153c58226a01 : SUCCESS in 34m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/69bc4078373140d69b21ff57ebea26ba : SUCCESS in 35m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f7d6d5c594704c8a96e4b7d636e726e7 : SUCCESS in 2h 08m 46s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/0b1c71eeba944759a20b487861ee6558 : SUCCESS in 50m 23s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/12a0a0cf47db48fe8c2521d828d0f689 : SUCCESS in 58m 23s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"957af75953758fbd723958feaa3c7312fb3df85d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-23 20:46:58.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"33f6e7b259e3ecb4438cce67600756558af65140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-29 15:37:21.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"07d73e31834aec5761c0aebff90e94931a0ca18a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-08-29 15:46:33.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":12},{"id":"fc649a6249cf2938755982a3433c68453d0a590b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:51:46.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/069f1e4c9d714983b4fab12e65c2ddb9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/542cdd909a814c349a2b8487437d8541 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60f80dfe5bf2422ead18f53576671442 : SUCCESS in 2m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6d52977170244245a355cd1a5f27a2f3 : FAILURE in 1h 20m 23s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/5f28c74055cd4ba199cdc12bf10039a8 : FAILURE in 1h 27m 41s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/b9b9178b4ee8451b8979f83cc6648b76 : FAILURE in 1h 31m 36s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/2de60b811cb54ab1873bf49374c994ae : FAILURE in 1h 28m 45s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c7b4a8aa73b24ff98633762f0a08a665 : FAILURE in 1h 25m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f0c72a7fc15440fbbbed67d5cc6471cb : SUCCESS in 1h 39m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e0d00b2008464d0588e88799ab512ba2 : SUCCESS in 1h 26m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/f2504f50c39e46eea5b16d85681a4f4a : SUCCESS in 2h 00m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/a387fedb0fa0420d9e5b33f8d9fe5d0d : SUCCESS in 1h 18m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/e4ab485e0cbe419ab467ff817a2b1eb4 : SUCCESS in 1h 18m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/fccd90ec2be14fb693e6124a4d9fa36c : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/d5b895c1aed24af68402ef643224ae69 : SUCCESS in 1h 17m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/694dc972e9e64009aff43a7d35ec9097 : SUCCESS in 1h 00m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/43d0505d4bc748289e11d5b7edde272c : SUCCESS in 40m 01s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/f7af6b1ccee84abaac11e74c5bbcd196 : SUCCESS in 42m 26s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/6a74da0b2cc143a3b54f15eea595c61a : SUCCESS in 48m 05s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/3b7a5b503e924ae7b60aab8e5df5f33c : SUCCESS in 41m 33s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5e9eb24848744c1bbb09fae2b551ab32 : SUCCESS in 2h 33m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/32bd87b357a94e1a99bd2cdcd3c43aea : SUCCESS in 1h 42m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4afc3e6910a14c1c9a0ad16f5745ba45 : SUCCESS in 1h 20m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7a00987c5eec492fa4e1325423a9cefb : TIMED_OUT in 3h 03m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f57beef420645b281b52cd792f9e4de : SUCCESS in 41m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/cd77cdc4dd534df58bb609f6f838fed7 : SUCCESS in 36m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/c121d249192745f9acbedef4058575ac : SUCCESS in 36m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/b3c5ed3a59d04937aa5f1df7c847778a : SUCCESS in 35m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8f988c5331fb48968d7e2ec6bbf5e1a5 : TIMED_OUT in 2h 33m 23s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/1af946d95fbe42daaa2c1a761c7e0f7b : FAILURE in 58m 52s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e77e6da8187a4d70b66c0e0d2b7941e3 : FAILURE in 55m 56s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b1a3a0e53de54c095393f8345ad957459512706f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-30 16:37:57.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"329c737cef134637bf615fd6a5c47f53586b9485","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-30 16:40:00.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"3e02adb5b7b2352ed271a1b5a97dd43474f591ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-30 17:03:18.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ea57a8440bf020d7d047c34c683da285b62ba14f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 20:08:21.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94e0d634b9aa48a087c066d6ba3bfab1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/318b9eae77834eb58b04f42050dcbbdd : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2353e0ff76ff4e95b3955637a641591b : SUCCESS in 3m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/66258c3dce3141cea40c7d2668c12eae : FAILURE in 1h 25m 33s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/1b68a313acc14255a5b351a457d0f6da : FAILURE in 1h 13m 03s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/d0176e6f2e07445685ac36afc5a8b126 : FAILURE in 59m 03s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/c9ca060b924641eb8cc6083a211bd769 : FAILURE in 1h 19m 41s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9139f8134a7f42f5be1165c56f495c58 : FAILURE in 1h 22m 54s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3e44f01951d8447f9210be5354a66997 : SUCCESS in 1h 41m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4693cb4534334b2fa964cec77b6a0d92 : SUCCESS in 1h 07m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7aae3edc6e1440bdbc84ec081192ddfb : SUCCESS in 1h 46m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ca53a384f8024a6bb970151df3ba9091 : SUCCESS in 1h 19m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/a988e725b85e469f9c91858436ba0021 : FAILURE in 1h 01m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/7009685c7e5c4f7fab0cd7b7c335cabc : SUCCESS in 57m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/367e2a5bbd274e69938a35ac4d2d0748 : SUCCESS in 1h 24m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/c02f29ebd37b4dfe83c40bd08755778c : SUCCESS in 1h 14m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4a049f30a8dd49329503e23af703218d : SUCCESS in 40m 09s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/dec3de5a330a4e8c84a4b94233c998ec : SUCCESS in 40m 06s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/b99faa1a7914470e8ed4918d84c879d5 : SUCCESS in 47m 43s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/786238f4c7e340b394e5fee0e9a9c4ef : SUCCESS in 39m 43s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/53788b161c7345c2bcbbd1d9b024a7ee : TIMED_OUT in 2h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/e8c852973b30472f98d41f87f84c0bf8 : SUCCESS in 1h 51m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e6f23a2bf38c4076b15e38f6f7575d5f : SUCCESS in 1h 07m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6d14894c594047dbad5bc0dbf39b1ab7 : TIMED_OUT in 3h 02m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7dcae077a6f4cf59488f04acb0ce303 : SUCCESS in 35m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7e2aea5c477a4dc18edada563f0e7f26 : SUCCESS in 45m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/799e490bed7e497e90386106732f7bb1 : SUCCESS in 35m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/20a9a800bcbb4098ae513928c4e30bcd : SUCCESS in 36m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/568feef7825349b089fbe2c44c4c9eaf : TIMED_OUT in 2h 32m 51s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/60a99b1aa5ce43c3b851706a1eae1144 : FAILURE in 47m 32s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b8a7919df66b4ccc9566ae07cf769c67 : FAILURE in 53m 55s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"4dd24cc4a80b3b118e185313ccd1265ac38805f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-11 16:21:57.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"7757e033d9ae6339b8cf7538fd85b7a5a4831f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 19:29:06.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd1cea021c804813b8faf32a092d0fee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a063ac2a618142cab756854e077084eb : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d19f6fd7c445dd905fcd8aa1241a7c : SUCCESS in 3m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c06f0db888e4ba3b51bec170052b865 : SUCCESS in 1h 26m 43s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/4800c47810a54b63815bfbc037e18463 : SUCCESS in 1h 10m 30s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/583c8441a90a4e35ac713c0894b3584d : SUCCESS in 1h 13m 23s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/fb53dd6b72bb46a5a522df59e2598d1f : SUCCESS in 1h 17m 01s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e25981b952e14713babb18536e3b1f6e : SUCCESS in 1h 14m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/250383f3e90f4c36a9a9ae9f24744a97 : SUCCESS in 1h 54m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/db6124b426c24a0f93ec8d8adce4d118 : SUCCESS in 1h 03m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/581adb856f5c44e1a442675672f3767e : SUCCESS in 1h 20m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/f0228324de594093a8f10dbeef62aa0b : SUCCESS in 1h 02m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/7a9bc0aed90c4bbea2f7f5289fd05eea : SUCCESS in 1h 41m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/4d09960106de4e409b2813b273121baa : SUCCESS in 1h 05m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/9b305b844b384185aa4de7a63f4d3832 : FAILURE in 1h 47m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/e24ae4bc1c014a59a25a4468980cdcd3 : SUCCESS in 1h 23m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e95962368314e3a8c638ba1d1128ed8 : SUCCESS in 47m 03s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/b87e2392784c47e5942119ec1f36a29b : SUCCESS in 55m 39s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/7679bbf7fac148078606c5b8093e2541 : SUCCESS in 48m 23s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/f53dc385ad4e45afa3b837cdc33faba9 : SUCCESS in 46m 28s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/58608fff9b884559b3de413ef8984778 : SUCCESS in 2h 12m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/d1359035d0014816b0434988ee593093 : SUCCESS in 1h 54m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e36ba99c8cf3421e85d3b86207563593 : FAILURE in 1h 08m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aa7f6f8a916744b085e1bb1eeb2bad01 : TIMED_OUT in 3h 04m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7bdc8860d2b749018da15761a872266a : SUCCESS in 45m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/50ded3965e794556b2c0e16bf7e16978 : SUCCESS in 43m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/6eb65b413f644302b32a6d11dcf0d7c7 : SUCCESS in 44m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/51d2c2f9b69643e196bf17800c959db6 : SUCCESS in 44m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/134cceef39fa476e8a636f5d9a6e339b : SUCCESS in 2h 11m 49s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/6f57ee4701fe4b4aa393d7a7b455f798 : FAILURE in 48m 43s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/f80e558428c44057a56fba0f3bdc721c : FAILURE in 57m 01s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"819d627b084b64a988e8f5874d41ad58e2d9107d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-12 13:48:17.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"45a49ba83549fab6997fe1a8e143a9c9e885de83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-12 13:52:43.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"fca97a33c8230282e2620bf43d4103b8d3c3fed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 16:58:47.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab63b06e45443de8fd8745686cb2336\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ccd62f177704f38a1d4904faf4fe666 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0caf911bdef4b64b032b6c163599bf4 : SUCCESS in 3m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd71f92ba7b440e0aeaf3f08f90b56bf : SUCCESS in 1h 07m 45s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/a4213d62a7704bcca27bd89699cff832 : SUCCESS in 1h 34m 11s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/99a2028c9e37407da721ef8ba48ca69e : SUCCESS in 1h 12m 44s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/4662e5a933ad4962a3d6eee40b791995 : SUCCESS in 58m 18s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/abdd43198d104b8c9f5b39b97bc9b629 : SUCCESS in 1h 28m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cfac918df2cd45869c16096f76b9cdf0 : SUCCESS in 1h 37m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/643b58ce8a0f4e979abfdfd7a59b9404 : SUCCESS in 57m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/c438c8d490544a249f4d59f910260a01 : SUCCESS in 1h 16m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/eac8b37789b54f15b0419f6d7bc1ab00 : SUCCESS in 1h 04m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/5d2030b01fe04720bd90579bdb39bd03 : SUCCESS in 1h 33m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/b8ba52b4f0124546836492392dd51f61 : SUCCESS in 1h 04m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/6c83cdc41d2e4a5ba2062922f2949cc2 : SUCCESS in 1h 30m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/bfb388bb4c204a21aa15e73efe468220 : SUCCESS in 1h 16m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7fb4bc1be72a4f80be382c86fc9d145a : SUCCESS in 40m 07s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/b8a22e5ea766426196bd0a21d4e1d8ac : SUCCESS in 46m 41s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/6db92d753dd0490a9ac7e0b75b09227d : SUCCESS in 42m 37s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/ae00ab65368247dba1a8732d512f46f0 : SUCCESS in 41m 35s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/76946ea60dbe4f9bb1fc2ff0525c95bd : SUCCESS in 2h 01m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/0a410996dd504046a33edaf60846c39f : SUCCESS in 2h 13m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8c5c10a4fecd4894ba4e1bc266aa5f7c : SUCCESS in 1h 12m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d522dfd38c4e4849a2dce0dc8ab6832a : TIMED_OUT in 3h 03m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1391db820ca24fa69f92a309389b70cd : SUCCESS in 33m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/d1f4eab58ee24b76ba98453253aa0d8d : SUCCESS in 30m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/f0e1b685a50e4f2d9d253979adab8eb2 : SUCCESS in 33m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/65069b284b314323b89f577e3da29d88 : SUCCESS in 31m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/31b684b1969146e6ac8cdc439fe68fe9 : SUCCESS in 2h 14m 07s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/aadb61f801ec461dad3d1fd762150996 : FAILURE in 48m 57s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/dd8393abe8e34f0eb91e38dd510d2b07 : FAILURE in 1h 02m 39s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"fe392725a747530e73f6c422324e58884f777ea2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-24 15:49:35.000000000","message":"Patch Set 17: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"d76e6e0fa9085df47886569662782651928f64f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-24 15:52:03.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"6c25ff18dfd8eb951f3f867f7f8d765bab22bc64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 19:10:37.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da98e7a393b049af9944e485b7e4b345\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f89efc955974650a4c6ffb94b749534 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d291ed5e30a453a83c846dafd19e9b8 : SUCCESS in 2m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/360f509ad69a46fdadbd6781a9c75fc9 : SUCCESS in 57m 38s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/4c5b16d111824b3a9cc04eba81801a77 : SUCCESS in 1h 13m 59s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/a9368e0437bf4a4ab3ba3aad87939bb8 : SUCCESS in 1h 14m 19s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/1da88f6d73ad4e0a9fd0d0f657cee43f : SUCCESS in 1h 16m 38s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1eac0a3cb1434521a6e4ef7a0fda19f4 : SUCCESS in 1h 36m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e402f29e53af4628ba7856abd41fbd28 : SUCCESS in 2h 01m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a9aacb3714674b858d46fcf47d254089 : SUCCESS in 1h 05m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/bb08764049024e568af231b9061e3cd1 : SUCCESS in 1h 36m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/f865210953d04052b5936b4c60b6b97c : SUCCESS in 1h 00m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/77bb76ddb1974e12a6f244c789dcac48 : SUCCESS in 1h 25m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/0fef4dcfd56f49adabd3b7ee7c96a291 : SUCCESS in 1h 08m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/eb5195cab7344bb89aca618d497940bf : SUCCESS in 1h 17m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/e02cd3e5f7b74f0d8a6a09d10ff8fc14 : SUCCESS in 1h 14m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eac22d986a464fc38419b1c36575dcca : SUCCESS in 40m 20s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/6958e03fe3cb42ee93431dc3bee56ab5 : SUCCESS in 40m 34s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/928c6bc023524c7a9e55ccdf9f641a20 : SUCCESS in 48m 53s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/97a9f41ae32342b7b2f65872e189c498 : SUCCESS in 39m 36s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fc39cc57d7024458a4649af959aef1bc : TIMED_OUT in 2h 33m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/f8f7ad6334604e79b654d7e49c123b71 : SUCCESS in 2h 06m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/943b99ebb9b7433a8245db2449750f90 : SUCCESS in 1h 11m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e6b51318c8de4e029e98c7b254658cb1 : TIMED_OUT in 3h 04m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/461058ad9bd34c339c3283af1222ddc2 : SUCCESS in 35m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/75e1898120c94adcb45d29cc385ae21a : SUCCESS in 32m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/2ede8071d83f423283ce2cb68c35bd0e : SUCCESS in 34m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/a74ceb36a1e64db7beab280525c43526 : SUCCESS in 35m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/34fe3eeb35334d458ccabe6d93d85660 : TIMED_OUT in 2h 34m 29s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/cdeebb8b96354c81adbcff37a30b00a6 : SUCCESS in 46m 46s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/4b259b06c90d4b5a83d21ae1048f1dab : SUCCESS in 56m 20s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"fae732f1d843495da2f36c41cfd21057e4f695ab","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-26 14:45:46.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"5e43be737db0d1764646d25dc53eb0b25299e4aa","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-30 11:30:58.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"b5d3d0fd66aac8a8b75adb238ff0d27c166bbd33","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-30 11:52:43.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"397ea2671e08830a34d100c14abf997e21132397","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-30 11:52:51.000000000","message":"Patch Set 18: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"e0b85e837174fea166cd05dc9903a5847ecc6b93","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-30 12:08:34.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"fcd6e9477f7a3d4e9e0476b5cdbafefaf949e6ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-11-06 16:38:30.000000000","message":"Patch Set 19: Patch Set 18 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"16d61f17b90cfc7ab4a21195f538d0730d108d7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 19:48:18.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2a602657b324e7dbd33a0999789e8fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e424abf5da394903ac7c565d3b84bc57 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc3518a892045bca66b3946eedaee81 : SUCCESS in 2m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ef5cd6a0e5464dd884c087dd72588703 : SUCCESS in 1h 19m 50s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3470042aa6fc4298aa68d80f7cf25de1 : SUCCESS in 1h 22m 42s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/b9d39908ba4249b59650ed2fc5d2b9d1 : SUCCESS in 1h 13m 50s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/e96ea1c72c5e4ccd824adbefbda7c6d4 : SUCCESS in 1h 31m 00s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/df976c0e37234e168540194210eccb34 : SUCCESS in 1h 35m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d33290f766424d89b2d27c85ccb63176 : SUCCESS in 1h 45m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/09d417ae4e224e27a6c29052c283a837 : SUCCESS in 1h 16m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/958cc1f3d218400aab10457d7b28a82f : SUCCESS in 1h 40m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/c51cef752c59406eaf5a1c9ac7edbb26 : SUCCESS in 1h 24m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/db9c10eeab374c66a8ad05757d236a04 : SUCCESS in 1h 38m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/1e650c04518b402dbe526caa73d6af6c : SUCCESS in 1h 15m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/61ce31e41e1946a9b3c3dbb08036637c : SUCCESS in 1h 38m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/a16fa6ab1ca44cdfba289b0b313e9c81 : SUCCESS in 1h 15m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/842842c83e3b47189d69dbb7e9ef9bbd : SUCCESS in 40m 41s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ce593247cdca482ca11d34f120db7e3c : SUCCESS in 40m 36s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/34efa7b62a1046a8a1caa4e0f47d4db7 : SUCCESS in 43m 21s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/046ff5b7516b46d7a43431d47f2c2563 : SUCCESS in 39m 13s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/78cc981a20374deba8ae8cfab764d311 : SUCCESS in 2h 05m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/c0ccb94233dd45ef9d3b1739177e473a : SUCCESS in 2h 06m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6a744a0b00034da4a69748690fbbd93d : SUCCESS in 1h 17m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/967d5c73fad14e05ac9c9ed519cc4a22 : TIMED_OUT in 3h 07m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78e12780c3b04657b49b48ae129d027b : SUCCESS in 39m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/8d6a0db0de7d40c5a4e9347459d44ce5 : SUCCESS in 37m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/a811c0e2713b4fe5a2c8efb8a726c2d3 : SUCCESS in 38m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/508684e45be54d85b56c15d64671c6e7 : SUCCESS in 40m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5a3f9eaad30841259c500a4a0281f2c3 : SUCCESS in 2h 04m 58s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/e1d0a171873e443e8553455c28fee444 : SUCCESS in 49m 04s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b92e0aa3fc554bc799b49ae937d75b39 : SUCCESS in 57m 22s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"288e70fa7838b8bdefdf88ed0057e86af4ea1050","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-30 08:48:34.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"d847683386f8606ffb64a51276bea1597f65f2db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 11:25:55.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/117c5bc053874b15a227c6fb92627807\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/212b8eac6f604866984f3352932ea3e2 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b06dac569e0947b09807a1fd418e61b9 : SUCCESS in 3m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e57769308e7a4fd9a740374d7c167c25 : SUCCESS in 1h 23m 21s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7cd6c444987c4e178aea72298a01c673 : SUCCESS in 1h 20m 50s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/9f58f946c70042e5babf76315ab01e82 : SUCCESS in 1h 39m 22s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/5bc0aa5bfd46466da9bb6d7363473d99 : SUCCESS in 1h 20m 21s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3ac5ccc1b2784c7baeedbc44bb8e8f86 : SUCCESS in 1h 30m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/89e5fdaf4a2e40498a9ce1b8db7df2cc : SUCCESS in 1h 35m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6fc15296c0cc41ca9deaa2465cf66fb5 : SUCCESS in 1h 01m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/591e527a5a5d4cd998d9980c99bd4e4c : SUCCESS in 1h 43m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7b96bf82e3884a2f817f8963a42cfeaf : SUCCESS in 1h 26m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/5cb9d719a1b4451cb1ccabb1ba8e7848 : SUCCESS in 1h 15m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/3e1d6a1eec464b7cb7aa9215f9f2d0d5 : SUCCESS in 56m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/4055d9f6edb4498e912c4e4a68897e01 : SUCCESS in 1h 15m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/4768022ac6bd430cadd361f0fb1eb09c : SUCCESS in 57m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9c2d61fca8c743ac87bcf7c18e78bd01 : SUCCESS in 36m 00s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/6b706627e55f4ffcbb77ba77072a477d : SUCCESS in 35m 19s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/22b1ce93c7e94167bc77ad625d3dd874 : SUCCESS in 38m 37s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/a13b85a2f4d7462abbb204f98358b3d9 : SUCCESS in 37m 38s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0787e08502234f89bbd2b65355ffd174 : FAILURE in 31m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/b0fc6ea59e38450ea1acb20a6eab9f1c : SUCCESS in 2h 11m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/410bb9aa5afe444dbba960b9af5beeca : SUCCESS in 1h 19m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea56b602ff8f40f98395d9d001fd5465 : FAILURE in 1h 55m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/009b0c7b624a458f8097e99835a7c15c : SUCCESS in 45m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/bb08576e377f4695aefa0379a4d44964 : SUCCESS in 41m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/9fb3c8f2580e431484ea75f4b01d155a : SUCCESS in 49m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/38c93198af2746d1b4cb5c70879253ff : SUCCESS in 43m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2181759b82c34bf1a850cb52d784cff4 : TIMED_OUT in 2h 34m 29s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/4e27eaf7bb8647f0920964153199b52e : SUCCESS in 51m 25s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/3387316fdc5b494a83ba55aa548b2563 : SUCCESS in 1h 11m 08s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"80030ea08ffcafefabd6c9206c5e46722f397478","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-05 17:14:08.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"67f9e0170898ea09d60095c2b9e4324a0d41d62d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 21:11:33.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c08468fb95024649af247fb9f69d397d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9201033c4f354b558a67ce251c640d00 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c084ad10decd48239046837db090ece6 : SUCCESS in 2m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ce1911a3c0e7403e963cc02d2c21b88b : SUCCESS in 1h 09m 23s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/cf893533370b4fc193eb3d409de20d32 : FAILURE in 1h 06m 05s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/0bf0c41a9c84421aadf33e13fb731e04 : FAILURE in 1h 08m 18s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/64f67d0cb3d74f859d66adfb27d02a8b : FAILURE in 1h 00m 37s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5c210f6852dc4f51902e0836b5d024d8 : SUCCESS in 1h 13m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ed224bef7bbe4e63b8a61f3364ff9547 : SUCCESS in 1h 29m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ee69ff31e99f46f284ecbecdb9b0bfd3 : FAILURE in 1h 26m 13s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/9ffc2d47505d48aababeba51660dd84b : SUCCESS in 1h 17m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3d8792d6281c4060bc7fb29fc4cac86a : SUCCESS in 1h 26m 55s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/97d484906d854911ad0942c4e8e2edbe : SUCCESS in 1h 16m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/6c8cf40d05a94781bb0e3596e9c9b41b : SUCCESS in 59m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/f29104e1d96b4b769f1f392740b07922 : FAILURE in 19m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/e20f410e6e8b4598a8d70b70cd62e6dd : SUCCESS in 1h 00m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b71a71f11ffd408781df827f42cf6b9a : SUCCESS in 48m 03s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/710d41db135c4f2b9dfd8dd6ba895f13 : SUCCESS in 36m 25s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/71982ae322ae4bdaa8901506bfa46514 : SUCCESS in 36m 34s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/de91009df2d849ad85a76f21f9bbce98 : FAILURE in 34m 45s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2088b88f46004a5594a767b444afde2f : FAILURE in 32m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/4c5d859a0c5c41b5af36a3a7e9938162 : SUCCESS in 2h 00m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/93d8da4cc66a4fb58f02625b2b85fc23 : SUCCESS in 1h 11m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a5ad5c6467e3477f95e652c541980800 : TIMED_OUT in 3h 03m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4eed372617634d9ca39de0e3bb58c388 : SUCCESS in 44m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/695ddb2d1fb14390a015ce65c97979ba : SUCCESS in 42m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/d56dcd242b9448eab5c62b79beda5dc3 : SUCCESS in 30m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/4c119caf61c0419ba3541603093330b3 : SUCCESS in 45m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2b498f03070c439793748002e12fb48a : SUCCESS in 2h 06m 06s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/bd2ef20f76f0400aba80d3176bb29d32 : SUCCESS in 37m 57s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/54034c1902334301b78ce148feffd452 : SUCCESS in 50m 24s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"66d919535f162fc78109ce63c317a84c60aaded0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-06 07:03:35.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"95113d5cf52badea153b2ab5bb2e9be349f860c9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 07:57:15.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"ffdd5820493fe4b29c2d63ee4ad1ec7aeaf6b8d3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 08:03:27.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"56e00c2cd98f663872615e67c8eabca2cfb20662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 10:11:49.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/91bffd17ea06422e8ee2044696843cbe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1faafb682b543c29dc8f3fd3503e27c : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d33cc656b64d15a95d7c157aac2f61 : SUCCESS in 2m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a9259e1015f5407b811ae62c438facff : SUCCESS in 1h 32m 13s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/dbd1c15e8dda4c0ab7c8eeb190f14c1c : SUCCESS in 1h 56m 41s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/7cf44b5b4e6d4ad7ae00097c974cb4c3 : SUCCESS in 1h 17m 34s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/1e4c600757d343fa86a32f944fd8dae0 : SUCCESS in 1h 41m 04s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/297a58a177d040a2a08aa09557d6b862 : SUCCESS in 1h 19m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a4beb56b398044f69f818cfbacb75b87 : SUCCESS in 2h 02m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ebd19b687ccb4ac28d9004531d80376c : FAILURE in 1h 18m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/64376186100c4f92b825f3e3a15ac4b2 : SUCCESS in 1h 56m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/aec245eadea645e6a6bd8a7a6ae54474 : SUCCESS in 1h 33m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/590e391b59684fb7adb20f9216b3f2fb : SUCCESS in 1h 54m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/f1713bf359f547feb4b6d71e38fd9367 : SUCCESS in 1h 19m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/8ef2bea4220543538da533c0232ee08e : SUCCESS in 1h 40m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/f428faa07d3a4f79adee0f9246ca18e5 : SUCCESS in 1h 38m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b5cf4b32b91c4735b4872fb8a8e9bb54 : SUCCESS in 57m 14s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/b2ae44634c1f4d64bc1691ef8e8b4538 : SUCCESS in 54m 24s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/5c450a097f4645378a18daf3d2fa48a2 : SUCCESS in 55m 11s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/408264627ae5488298e52173b3ca2347 : FAILURE in 53m 10s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/84df8ce4c67945449fae1b66fd34ce73 : FAILURE in 29m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/30c5694939a8456592e3de751de3000f : SUCCESS in 2h 01m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8c1f7284b11447f89663015d72b1decd : SUCCESS in 1h 30m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/adf979ddc16f4579a96c80cacb521f16 : TIMED_OUT in 3h 04m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa5cb11aabdb46c0a43db52f7b796cc6 : SUCCESS in 41m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/1f9a01231e0f41d286e3940ceffe9ba9 : SUCCESS in 43m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/37659626a33e4c4d86c251949e5dab68 : SUCCESS in 43m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/df761365929e4c8ca0891081e4a2eb70 : SUCCESS in 48m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2054a36019474a9c8c28f5f9f29846a6 : TIMED_OUT in 2h 33m 08s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/46192505d42546dc9560939260754e64 : SUCCESS in 48m 38s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/bd6604ad455043eba602a6fe88458576 : SUCCESS in 57m 19s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"90e9f0070860ffc9943c4a7a23cb9ed63b7d61ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-06 11:33:58.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"c788a3089e44d029c8edf2c22c963bafe62d2f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 14:41:30.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1da29a9bc38491aa5ca5f36a126161e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e9ae1501d7e4e0e85a85b5103022c9e : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32223280732a4a1697f4364e2e932468 : SUCCESS in 2m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7034384f13604df99bad0e8612e960d6 : SUCCESS in 1h 14m 06s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/f0e10cad8bdf4aa1940146025f4ed146 : SUCCESS in 1h 04m 53s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/e0c899f1569f44f1b3dba0fbd2f86f1f : SUCCESS in 1h 01m 49s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/3482dff003104fe28e11f0ee444a7582 : FAILURE in 15m 53s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/52157a8f9ea14bfb8f3f4118c7832641 : SUCCESS in 1h 08m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3ba89a5ad6784b1982b8ec9bc1666f25 : SUCCESS in 1h 33m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/aee8b3646ae2420d9d0f99923ff865cd : SUCCESS in 56m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/5247011583054e14afc577394d4f36b3 : SUCCESS in 1h 42m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/fd8eb5b528a64c9c8b91e14a5dc773fd : SUCCESS in 1h 03m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/27b28a448c5e47638a2c8f947c374a59 : SUCCESS in 1h 50m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/578353a3175646fd88787959c0716f8a : SUCCESS in 1h 14m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/cf424bf0983043b88343cf959f28691e : FAILURE in 15m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/587e2599b3854683a004c0a93ac12ed7 : FAILURE in 17m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2628fee6392b48b681ca567f18fd75dc : FAILURE in 48m 11s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/5adbd8b3dff94466ae740061edba0f53 : FAILURE in 52m 55s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/4510f217e51d4430a2db87632c9323e3 : FAILURE in 48m 30s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/d57037c9335a4109a542e64402e254f6 : FAILURE in 20m 06s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/446a9ba7f8ee48f69a9643d05f6b4744 : FAILURE in 34m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/4d46e4cb142d4956abadd50d22c16a82 : SUCCESS in 2h 06m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/456325e8a4bd45a1b3ea6e5a62a892eb : SUCCESS in 1h 13m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cf4e45abe9144f95be9d856ae0b49b8e : TIMED_OUT in 3h 04m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f893016b69ef433cbf97637b63ccfb47 : SUCCESS in 32m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/1f60ffd606904d12a2bc8c8baa63be4e : SUCCESS in 31m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/3b8e4d69cad54e4b91e982c238871c10 : SUCCESS in 33m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/6520cc1a4fef489889ce9eb4b611bb4b : FAILURE in 16m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ab7d18556b354fc1b4a7dae7c7919584 : SUCCESS in 2h 07m 28s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5bfdb1559c904035ae2d653614087009 : SUCCESS in 37m 18s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/d81ae9284ae7498391c4aec9f19fa344 : SUCCESS in 56m 30s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"dd71a515574601d4a76b19c935f42bf5c0ed3168","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-12 09:27:41.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"2617c9c8518f16e8027b84d9e9153d436820c8ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-12 12:12:44.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"7b116fa0973ac296bbc15a650c4207c7207dbe8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-12 12:42:02.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"01cda230bf9439c55c08be2bc3e4b9e26501bc59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-12 12:50:41.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"f0bc6a2f5c86d4bdd1b8895c49097ae8ad1db626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-12 13:08:30.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"9e4cc73292ba6b9a1d678fdffac46c62b83557ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-12 14:57:58.000000000","message":"Uploaded patch set 28: New patch set was added with same tree, parent tree, and commit message as Patch Set 27.","accounts_in_message":[],"_revision_number":28},{"id":"e58421a9a1458b23ef239dca80307cddaac728fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 18:11:22.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fcb9622d64142049f6127d72a4d901e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c24ec0144642e08e18ff7fdca212bf : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f47c4f84ec84f1a955ec51ff1b641ab : SUCCESS in 4m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/605fa35befbf4be280cc77278c40c9f6 : SUCCESS in 1h 06m 04s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/21a936d6cd284d03936a112f5f320513 : SUCCESS in 1h 17m 38s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/73b15384071f47339b231c47b1c072a8 : SUCCESS in 1h 24m 28s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/3e52eaf575b0442d830c5054f705aa32 : SUCCESS in 1h 32m 06s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/daffcd6b79fa45ad8d689d2ee56bcb8e : SUCCESS in 1h 20m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/41cb007628084b6197fadd9b7e1c432e : SUCCESS in 1h 31m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/320530b3485145cdaab75b9af5f37035 : FAILURE in 1h 16m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/1aec6f24923241e7a25c851de4e5e551 : SUCCESS in 1h 39m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/abc110b598b743afbd03aea7cee17459 : SUCCESS in 1h 22m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/578713cced98460f9ee3dac93537d546 : SUCCESS in 1h 54m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ea0afa80d776454d9bd9efb277496cea : SUCCESS in 1h 24m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/aa1cd7a97c754b6d94d56f12606cef92 : SUCCESS in 1h 53m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/f81e3ccc160e4ad0890c918b69e95433 : SUCCESS in 1h 17m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9378e4db2a5b48c3a2d5c8429cb92d51 : FAILURE in 1h 00m 16s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/ce5b17af18274795a77aa119ef944630 : FAILURE in 51m 29s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/10e656ba5734446ea0a839a7982d678b : FAILURE in 51m 12s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/3cc8abdbb3054c4b9734566066520588 : FAILURE in 51m 16s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/47ff7b6363de40c08cb8967052d149d3 : FAILURE in 37m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/a23412f076f44e5d922a708fb7b455a8 : FAILURE in 50m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/13f3a084cb034e4aa96719f2068c3b60 : FAILURE in 36m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/149f7970d210438ea7a4d8ee4d540b12 : TIMED_OUT in 3h 05m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46e84c72412e4c31b9803bdd9daf83e2 : SUCCESS in 32m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/58e0c89ab6d04dd29c1f715fc4a8557a : SUCCESS in 48m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/d10b24254c0844d9be0b2eee7e712b7f : SUCCESS in 47m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/5691c85c03474f9b86e8b27be476ae92 : SUCCESS in 41m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5f8d432a04d34c4f8abbbaf79e4a0be5 : TIMED_OUT in 2h 33m 02s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/e0b0fcc3d16d493faab1f32751a78692 : SUCCESS in 43m 19s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/069a7ab2a39d4e9dadf24d3325a60cf6 : FAILURE in 50m 50s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"7789d671241acc11837b1acdafb638e8cb4d613c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-12 15:47:46.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"5a3e6196c738eaeed14a784ce0039debfd4b0229","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 18:52:34.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa1c09518c75455ca8f2542127b5c22a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e123e040ddb4970a07521dcf93833e2 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbf106efcb1e4a5c89dafc86d10ce14d : SUCCESS in 3m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/381c710fc8b04a15bcabc267b7827ab6 : SUCCESS in 1h 52m 42s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b8b34e11d9b7464a956e975973c6d616 : SUCCESS in 1h 45m 59s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/39eec17abb574a95bc0b8b7d23624d1d : SUCCESS in 1h 20m 31s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/6e5b17a87cbf480cb9fa4537305275bc : SUCCESS in 2h 38m 56s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ee555d04310c44d5ba1245fe5dc8ec63 : SUCCESS in 1h 19m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e72c3cdeaa79465b8fa3a96f60db2b40 : SUCCESS in 2h 18m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3a89564a378e474cb2628b4d5b0340fb : SUCCESS in 1h 25m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/a61086ccac0740fb9f7f28ad57c49144 : SUCCESS in 1h 51m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/a064dc3750c443e9b7ef4c45179599a9 : SUCCESS in 1h 25m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/cbe7f069a44549649ab86ed1ed78891c : SUCCESS in 1h 56m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/58ae79335c8c49aba66299405e5a361e : SUCCESS in 1h 30m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/65c879f0aaa24e349a3d71c254d050ce : SUCCESS in 1h 36m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/35f609e0bcd043bf8080d4ce19d92aea : SUCCESS in 1h 27m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f17a60f2a55247578a25881727c658f6 : SUCCESS in 52m 35s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/baea0525834049ab874751b7f20eede6 : SUCCESS in 34m 07s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/d6f8358245874f838cc43978c50f3a98 : SUCCESS in 33m 48s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/b1dc3024bbfc4bf7b97dd5c82e3dabd1 : SUCCESS in 57m 42s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/da471bbfdb2e43b3b695152c30f4790d : FAILURE in 33m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/fb9e3b1f39044526941d3b3f2030c2fe : TIMED_OUT in 2h 36m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/419a31ca13064c389b4e8a214bc80eb9 : SUCCESS in 1h 25m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1ffa1e63fe26455084d3eec777913271 : TIMED_OUT in 3h 03m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ad3a8b699504596b5fea3b23d8112f7 : SUCCESS in 33m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/509a8f486d774d50867473ecc475fccb : SUCCESS in 30m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3a3186a7ba3e438ca4073403bbb41c4c : SUCCESS in 31m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/f6e47ef2c2f143b0b184331b2b614804 : SUCCESS in 33m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0f68635fa51f4ee9a405a1e73089d508 : SUCCESS in 2h 05m 40s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/684d5868f34e42f7a7f846933c637f65 : SUCCESS in 37m 28s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/a6aa83d9223a48e4a48b2abbabc4b419 : SUCCESS in 1h 00m 11s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"9e87afb806ad500288ad25a8f79b2652cf2937be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-15 10:55:35.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"84d42dfb25c6a7c063eeec13b3a982319881b600","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-15 10:57:18.000000000","message":"Uploaded patch set 31: New patch set was added with same tree, parent tree, and commit message as Patch Set 30.","accounts_in_message":[],"_revision_number":31},{"id":"133649c7a563187c694f00c3b0bc1fd9703cda1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 14:12:54.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce887b5c48a74cbb93c002a1e5399e62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4047d836fcde4b7d91b706ad156ff9c9 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a3a4287c74496f93c76bfb46b3227d : SUCCESS in 2m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/155a1caa8d8047748673b20dd1dc81b5 : SUCCESS in 1h 42m 27s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/579bbd2ab42d42acb974430043dc58e5 : SUCCESS in 1h 26m 17s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/fc1e82c063f744bd8530112ef0e9e596 : SUCCESS in 1h 41m 35s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/595b62f66c1844b496a0258bdf0cbdbf : SUCCESS in 1h 44m 39s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a9507e08ebd94cdca75aac42c30adabb : SUCCESS in 1h 57m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/aeb3edf45c8a444aaee16a416e01335c : SUCCESS in 2h 08m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cd31ddaf84844d8187c0788ec40bc38c : SUCCESS in 1h 12m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/7e4cbc3b243a44c0aaf89cd745f87b09 : SUCCESS in 2h 03m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/28ca38df0fbe449bbcfb7306e5cfa015 : SUCCESS in 1h 30m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/c5ae01837a974958a4ca65a4932903aa : SUCCESS in 2h 03m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/cd1e8438e773411db163e602dcb27cf6 : SUCCESS in 1h 30m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/e68f9bc10afa4e6e80b1bbd586b9bf8d : SUCCESS in 1h 43m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/b7cc7e3d43724889ae02250a0f73c707 : SUCCESS in 1h 18m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e67ba34b1e0a445882344ec773217caf : SUCCESS in 53m 09s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/ea4557a81c9147c6be18d23b04c204b7 : SUCCESS in 45m 31s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/27146bf9a6d747f095de0327ae99cfc1 : SUCCESS in 47m 20s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/a59aed657a4048ee806b7f4f0525ae83 : SUCCESS in 49m 57s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/227e1de4fb344edb89d88963fdc9248d : FAILURE in 30m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/918c5e35bf7b40868584237fdc5143b8 : SUCCESS in 2h 05m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cab763b666b94178a119ff1671ffe7a0 : SUCCESS in 1h 28m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/587414e177d844d8a4857c225d1041ec : TIMED_OUT in 3h 03m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/09a74a67a7964aeb898d431e748482fe : SUCCESS in 49m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4cd88f51c66440739393e1e66bbb113c : SUCCESS in 45m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/0313f50e23ba4d0db166cfcc99620c1d : SUCCESS in 49m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/a45b925dd948434494d3ee2769a304d2 : SUCCESS in 43m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1e30625ced854611bb7985e786b0230a : TIMED_OUT in 2h 34m 00s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/4a1f763c57204f96804cc4dd366b4493 : SUCCESS in 40m 16s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7d879865cad246fb8fe73a5788ff7bf8 : SUCCESS in 54m 10s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"e682eddc1eaa7861370666e3d5b095a0b5eb8f73","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-04-16 13:32:33.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"f0ebe0424adb0323b21e34b5926aec9a5b22041c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-16 15:22:33.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"2a7020bcc7fee6dfba05bc01f4e4314bbc23fc47","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-04-16 16:02:56.000000000","message":"Patch Set 31: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":31},{"id":"10e08bedd07f181090b767105968fb714d05e858","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-23 10:08:01.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"2d661db7ed9d42fc3d9a4d8ccee3a1e3ef1e62ac","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-23 11:56:18.000000000","message":"Patch Set 32: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":32},{"id":"0195a9649b511a213260bc449bbe16003c8eb6a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-23 12:25:57.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":33},{"id":"f7b9f374b480eae57b2b02af6f50f5ce8cf53ae8","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-23 12:27:35.000000000","message":"Patch Set 32:\n\n(3 comments)","accounts_in_message":[],"_revision_number":32},{"id":"b19c1c1d11455da777716a6ce24c6817aaee4b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 15:31:03.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ed692b7b18c461c8b06d84d49de1181\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d04dfcd59de4861be757042c2cc6492 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/922719c779c149559b375d01d2aaecf7 : SUCCESS in 2m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e098c699833843208b55bfb03cf785e9 : SUCCESS in 1h 24m 00s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9382f356c13e42b29b88d5f104ef30c6 : SUCCESS in 1h 09m 09s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/b7e0521d23c94f3989a2cf7b03ab41b4 : SUCCESS in 1h 10m 14s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/a82ee1c2e9a24879a6ba36af1512e78f : SUCCESS in 1h 16m 01s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a496859a63e74d30a4d38952265ca040 : SUCCESS in 1h 16m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d396991b659c4262a8e35d176f834c82 : SUCCESS in 1h 40m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f33d355abd1549b4a1c2a00cd01d47be : SUCCESS in 1h 02m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2019050288f54a49a8a642b94805c5dc : SUCCESS in 1h 25m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/bfdef35e77c045b69d9e3a23bf16f2e4 : SUCCESS in 59m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/61df21d1873d4e76b636bdd1626919d3 : SUCCESS in 1h 19m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/191ded587df748919f55253818d0f360 : SUCCESS in 1h 03m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/bd8d6c6499e340cebf44d7035203ae53 : SUCCESS in 1h 48m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/3211a9bbc9f6499c82b42d3ad432e802 : SUCCESS in 1h 11m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1e56d3447e05499f8afa5cd3dfd00d29 : SUCCESS in 38m 56s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/f75f6b239027426fb7d040c5fef35435 : SUCCESS in 48m 36s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/618634a06ea04100a4d3639c3e79ddd4 : SUCCESS in 51m 04s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/572f11da5fa441c9a6f2bcb09ebdea56 : SUCCESS in 48m 03s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d500cad246304c2ea39e89699d1cdea6 : FAILURE in 24m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/48121d062437497aa08e9f83d716259e : SUCCESS in 2h 07m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/be8f4ff457d145d3a2bc9710a989672a : SUCCESS in 1h 26m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/07815ec3bebd4a4aacde5f25eec95a1f : TIMED_OUT in 3h 03m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22375bde1c6a466894ee46d1e675a50f : SUCCESS in 42m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/985fe87446ec4dcf8c75629e9c3ba2b2 : SUCCESS in 31m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/348af75ab04f43bdac83651b9c968cb2 : SUCCESS in 44m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/3475d2a39e934c829e9da64ef6f824bc : SUCCESS in 43m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d62ea2f0012a4a9ba8c04b43b56ae81e : TIMED_OUT in 2h 32m 51s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/02234915f182480e8caa01d13dbc73db : SUCCESS in 46m 47s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7fd17a9f6bf84511aa15278cc5df304a : SUCCESS in 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"129ad4a18495ecf775c4797a3ec8d85386df5ff3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-23 16:34:52.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"82677fa73c72ee2e09a394dc211c4f00bb46ce6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-23 16:56:52.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"ea590dd1f80b3db7080042c7fa1af60a4da8dc2e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-23 16:57:48.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"d41adb14f9e50929804d2c317e31bcf7812f19f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 20:16:01.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e39a3dc2a84246a7a861e16af4adf7d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25227d542e1408590c9295b4ab4bd9b : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e33b54646e20412daa09564098bd7c6d : SUCCESS in 4m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b4bcfed124eb44099747828ac0a9444f : SUCCESS in 1h 34m 02s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/be6c190e24294ca0b184c788c064d5eb : SUCCESS in 58m 27s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/d315b53525dc47beb35ef0fd2beb1504 : SUCCESS in 1h 03m 09s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/cd9b03072e704a568bed9975e29a56c0 : SUCCESS in 1h 09m 05s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f8f6802fa148436996dab895f8ef0f03 : SUCCESS in 1h 18m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f3573b7a291e43cd95ef3c2ae217ec8f : SUCCESS in 1h 37m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5870513abfce4f94959e870930e220bb : SUCCESS in 1h 18m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/327e616e3aa14b7daba3cda60072fabc : SUCCESS in 1h 21m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/aaa5729349004d848b6faa325d222361 : SUCCESS in 1h 05m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ac45cde8dd0a45f8b3da286c58829563 : SUCCESS in 1h 21m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/f8bd9f8dbc10489dac7c3a1ef7b699a9 : SUCCESS in 1h 06m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/13ad4025dad34f22903d264732b83ca4 : SUCCESS in 1h 56m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/5b0854965cc946408f0940662cfdb00d : SUCCESS in 1h 09m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dede45735e9243728ea40ce6d51fa90a : SUCCESS in 58m 52s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/c0ac2a0794bd4deaae73cf4caf250885 : SUCCESS in 45m 12s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/62a8b94a92cd4246a2a285c652632853 : SUCCESS in 52m 17s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/a2a78b106be24f2fa0a660862a9a5b5b : SUCCESS in 49m 13s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/919ffc15e88446b2b612f6a7b9f54ac8 : FAILURE in 36m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/156c38f9b3ca41b790b865cad4a6af9b : SUCCESS in 1h 56m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/03f816145f6047f8ac2de5d6d9ca3885 : SUCCESS in 1h 25m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/99e2e10fb6b6469093286f4fd27b2c2e : TIMED_OUT in 3h 03m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94c729b6278941dbb49731a4642fbd96 : SUCCESS in 30m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/482c70d2d2eb438cb9a769ce57f758ad : SUCCESS in 38m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/0285933cb6624e49b7cd75a54c1808f1 : SUCCESS in 48m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/807cb0b884f948dd9e5e65bf4fde70f3 : SUCCESS in 29m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/52f3f02916fe4e64987a3e3e8da2f8bf : SUCCESS in 2h 00m 09s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/8783d5d8ae374b21ab977228f1471ae3 : SUCCESS in 57m 23s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/31d85757542f40a3b9ab263e3fa7dd23 : SUCCESS in 40m 34s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"7353dd764b1853219501338c6b9ed8c1f10cf614","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-24 17:05:29.000000000","message":"Patch Set 35: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"43cd8d28c861d931e781b76c36e960234a3881e7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-30 08:29:53.000000000","message":"Patch Set 35: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":35},{"id":"4b68275e8a523c8b2d6be3eee5d82f6c953fb192","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 09:20:09.000000000","message":"Patch Set 35: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":35},{"id":"119f65905aeafaec18e8843456696ab653611fb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:40:06.000000000","message":"Patch Set 35: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d96da302cf0d4733a690636958cd79b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b12a61eccf44e9195e52230f6a49b56 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/663136dab4d74f46888eac3a789b45bb : SUCCESS in 2m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cbbc4d7ef23d4e6d8054a7d86ebde0fd : SUCCESS in 1h 24m 30s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4dc8117bebc54da09659be2d7fcfa73d : SUCCESS in 1h 35m 38s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/f10f914f78574a5391a1e017f00b4c78 : SUCCESS in 1h 23m 31s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/70a911b7577949e2a97a90f45c7a78be : SUCCESS in 1h 35m 19s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9287b00139ab4d8dbe8a603b13af6665 : SUCCESS in 1h 39m 23s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cc86532ec93647719330c194300d8a35 : SUCCESS in 2h 18m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5629a34a1a68468da0720a6f195d559c : SUCCESS in 1h 11m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/a799d56d4df744ada63606d942e196dc : SUCCESS in 1h 43m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/cf9217389682499cb5e70364d1304a8d : SUCCESS in 1h 18m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/27c59ba2962a4c94b1b5a48ced998bbf : SUCCESS in 1h 21m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/1dcdeaa1bf95407b8af7849164630c22 : SUCCESS in 1h 37m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/7ef3a686b3c1491ebad02bf058711f79 : SUCCESS in 2h 02m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/d75106ad555f4b778542a8fbc1797655 : SUCCESS in 1h 30m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f3c03afe2c584e00b1ac377bea7f1fd3 : SUCCESS in 34m 53s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/db1ead7d18ee4f54972a66f234a51c60 : SUCCESS in 54m 32s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/8284cbd48b1c4335bbf252c7681a37d4 : SUCCESS in 33m 37s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/121b6c05b9064d719d45b89dba954b76 : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":35},{"id":"ff5a0975dcd0e0faed4557c71dcc3ddcfcaae326","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:40:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":35},{"id":"7f3214134b5c66e84b2d344645c8c585d88925b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:41:16.000000000","message":"Patch Set 35:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcd1d9b76c424ce68617fa7a146fdec8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a9be100eadd4a97a105d0e6c9ce547d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":35}],"current_revision_number":35,"current_revision":"c03e9c39d24feffae9ac246f947e0655b7224547","revisions":{"1d202d9977bf222cfdbc19a0e226680a575cc068":{"kind":"REWORK","_number":1,"created":"2023-04-26 10:33:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/1"}}},"commit":{"parents":[{"commit":"8a6e50424aa0482ec287367c3be596f633431a1d","subject":"Merge \"Fix the pool client auth test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8a6e50424aa0482ec287367c3be596f633431a1d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:33:08.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1d202d9977bf222cfdbc19a0e226680a575cc068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1d202d9977bf222cfdbc19a0e226680a575cc068"}]},"branch":"refs/heads/master"},"c801a69482b9c7f9340a110fe651e4ac96e30398":{"kind":"REWORK","_number":2,"created":"2023-04-26 13:02:39.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/2"}}},"commit":{"parents":[{"commit":"8a6e50424aa0482ec287367c3be596f633431a1d","subject":"Merge \"Fix the pool client auth test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8a6e50424aa0482ec287367c3be596f633431a1d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 13:02:25.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c801a69482b9c7f9340a110fe651e4ac96e30398"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c801a69482b9c7f9340a110fe651e4ac96e30398"}]},"branch":"refs/heads/master"},"4451024a240fed53b155ca147b6829285066a8cd":{"kind":"REWORK","_number":3,"created":"2023-05-04 13:44:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/3"}}},"commit":{"parents":[{"commit":"8a6e50424aa0482ec287367c3be596f633431a1d","subject":"Merge \"Fix the pool client auth test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8a6e50424aa0482ec287367c3be596f633431a1d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-04 13:44:07.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4451024a240fed53b155ca147b6829285066a8cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4451024a240fed53b155ca147b6829285066a8cd"}]},"branch":"refs/heads/master"},"49630109c8325e9bfc5acf3a2a042013d9b3b4f0":{"kind":"REWORK","_number":4,"created":"2023-05-04 14:44:36.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/4"}}},"commit":{"parents":[{"commit":"8a6e50424aa0482ec287367c3be596f633431a1d","subject":"Merge \"Fix the pool client auth test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8a6e50424aa0482ec287367c3be596f633431a1d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-04 14:44:26.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/49630109c8325e9bfc5acf3a2a042013d9b3b4f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/49630109c8325e9bfc5acf3a2a042013d9b3b4f0"}]},"branch":"refs/heads/master"},"6cc9492ae3d1f3915cd5255f668c841bb78be428":{"kind":"REWORK","_number":5,"created":"2023-05-04 14:54:47.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/5"}}},"commit":{"parents":[{"commit":"8a6e50424aa0482ec287367c3be596f633431a1d","subject":"Merge \"Fix the pool client auth test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8a6e50424aa0482ec287367c3be596f633431a1d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-04 14:54:39.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6cc9492ae3d1f3915cd5255f668c841bb78be428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6cc9492ae3d1f3915cd5255f668c841bb78be428"}]},"branch":"refs/heads/master"},"e7472b59363c3f749d10803624d8da04e5673bcd":{"kind":"REWORK","_number":6,"created":"2023-05-04 16:16:48.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/6"}}},"commit":{"parents":[{"commit":"8a6e50424aa0482ec287367c3be596f633431a1d","subject":"Merge \"Fix the pool client auth test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8a6e50424aa0482ec287367c3be596f633431a1d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-04 16:16:32.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e7472b59363c3f749d10803624d8da04e5673bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e7472b59363c3f749d10803624d8da04e5673bcd"}]},"branch":"refs/heads/master"},"ef22d7d9b645715169e3b7dd55d6bfa01c9f7dd5":{"kind":"REWORK","_number":7,"created":"2023-05-05 06:33:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/7"}}},"commit":{"parents":[{"commit":"8a6e50424aa0482ec287367c3be596f633431a1d","subject":"Merge \"Fix the pool client auth test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8a6e50424aa0482ec287367c3be596f633431a1d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-05 06:28:52.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ef22d7d9b645715169e3b7dd55d6bfa01c9f7dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ef22d7d9b645715169e3b7dd55d6bfa01c9f7dd5"}]},"branch":"refs/heads/master"},"e645f12db457bb5b23cd9c4c68416a05308d33ea":{"kind":"REWORK","_number":8,"created":"2023-05-05 06:36:57.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/8"}}},"commit":{"parents":[{"commit":"8a6e50424aa0482ec287367c3be596f633431a1d","subject":"Merge \"Fix the pool client auth test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8a6e50424aa0482ec287367c3be596f633431a1d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-05 06:36:45.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e645f12db457bb5b23cd9c4c68416a05308d33ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e645f12db457bb5b23cd9c4c68416a05308d33ea"}]},"branch":"refs/heads/master"},"f93144702adc61d88dbfaa1ca7d08bb4f8eda6bb":{"kind":"REWORK","_number":9,"created":"2023-07-19 15:56:46.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/9"}}},"commit":{"parents":[{"commit":"8a6e50424aa0482ec287367c3be596f633431a1d","subject":"Merge \"Fix the pool client auth test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8a6e50424aa0482ec287367c3be596f633431a1d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-07-19 15:56:36.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f93144702adc61d88dbfaa1ca7d08bb4f8eda6bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f93144702adc61d88dbfaa1ca7d08bb4f8eda6bb"}]},"branch":"refs/heads/master"},"e65bcc62015796f2a45b96de561c010bd0663c51":{"kind":"REWORK","_number":10,"created":"2023-07-24 09:29:04.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/10"}}},"commit":{"parents":[{"commit":"8a6e50424aa0482ec287367c3be596f633431a1d","subject":"Merge \"Fix the pool client auth test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8a6e50424aa0482ec287367c3be596f633431a1d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-07-24 09:28:54.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e65bcc62015796f2a45b96de561c010bd0663c51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e65bcc62015796f2a45b96de561c010bd0663c51"}]},"branch":"refs/heads/master"},"94c29e8c91ce361d9e13a146de681df5222717a9":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-08-29 15:37:21.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/11"}}},"commit":{"parents":[{"commit":"bc918d489887d56aebd6eb8d3300151e3076d81f","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bc918d489887d56aebd6eb8d3300151e3076d81f"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-08-29 15:36:19.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/94c29e8c91ce361d9e13a146de681df5222717a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/94c29e8c91ce361d9e13a146de681df5222717a9"}]},"branch":"refs/heads/master"},"6c668894f9f36376047342365be3a2d437964ed4":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2023-08-29 15:46:33.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/46/881546/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/12"}}},"commit":{"parents":[{"commit":"5ba4f3b351ca40f119ebc7c79831136ec7da508d","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5ba4f3b351ca40f119ebc7c79831136ec7da508d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2023-08-29 15:46:33.000000000","tz":0},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6c668894f9f36376047342365be3a2d437964ed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6c668894f9f36376047342365be3a2d437964ed4"}]},"branch":"refs/heads/master","description":"Rebase"},"74878040927d17a1f34ed012e26007e972ef298f":{"kind":"REWORK","_number":13,"created":"2023-08-30 16:37:57.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/13"}}},"commit":{"parents":[{"commit":"5ba4f3b351ca40f119ebc7c79831136ec7da508d","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5ba4f3b351ca40f119ebc7c79831136ec7da508d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-08-30 16:37:41.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/74878040927d17a1f34ed012e26007e972ef298f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/74878040927d17a1f34ed012e26007e972ef298f"}]},"branch":"refs/heads/master"},"b125fe8a3abcf4ef5474fc7f6781146df252ba8e":{"kind":"REWORK","_number":14,"created":"2023-08-30 17:03:18.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/14"}}},"commit":{"parents":[{"commit":"5ba4f3b351ca40f119ebc7c79831136ec7da508d","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5ba4f3b351ca40f119ebc7c79831136ec7da508d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-08-30 17:03:08.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b125fe8a3abcf4ef5474fc7f6781146df252ba8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b125fe8a3abcf4ef5474fc7f6781146df252ba8e"}]},"branch":"refs/heads/master"},"dccfc3aa9386a266fcab573abbd5f5fb88247078":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2023-10-11 16:21:57.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/46/881546/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/15"}}},"commit":{"parents":[{"commit":"2ac5e8169fd1ce38998ac2e66ae6c216e0b9f584","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2ac5e8169fd1ce38998ac2e66ae6c216e0b9f584"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2023-10-11 16:21:57.000000000","tz":0},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/dccfc3aa9386a266fcab573abbd5f5fb88247078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/dccfc3aa9386a266fcab573abbd5f5fb88247078"}]},"branch":"refs/heads/master","description":"Rebase"},"c05f655b013218214fa29bf8d655580424d5497a":{"kind":"REWORK","_number":16,"created":"2023-10-12 13:52:43.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/16"}}},"commit":{"parents":[{"commit":"2ac5e8169fd1ce38998ac2e66ae6c216e0b9f584","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2ac5e8169fd1ce38998ac2e66ae6c216e0b9f584"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-10-12 13:52:06.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/880806\nDepends-on: https://review.opendev.org/c/openstack/octavia-lib/+/880821\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c05f655b013218214fa29bf8d655580424d5497a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c05f655b013218214fa29bf8d655580424d5497a"}]},"branch":"refs/heads/master"},"d3eb97624a9cf9ea47f61f16d2df69cfbca3b57c":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2023-10-24 15:49:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/17"}}},"commit":{"parents":[{"commit":"2ac5e8169fd1ce38998ac2e66ae6c216e0b9f584","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2ac5e8169fd1ce38998ac2e66ae6c216e0b9f584"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-10-24 15:49:35.000000000","tz":0},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d3eb97624a9cf9ea47f61f16d2df69cfbca3b57c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d3eb97624a9cf9ea47f61f16d2df69cfbca3b57c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5bc677c4beb4a626ac906440b18221dd687cdad5":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2023-10-24 15:52:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/18"}}},"commit":{"parents":[{"commit":"515f76190b64ead5e2a100a27a752ebb5d29654b","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/515f76190b64ead5e2a100a27a752ebb5d29654b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-10-24 15:51:22.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5bc677c4beb4a626ac906440b18221dd687cdad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5bc677c4beb4a626ac906440b18221dd687cdad5"}]},"branch":"refs/heads/master"},"99b4b4326c670c8c566d38da6f4f359bc1c3d809":{"kind":"NO_CHANGE","_number":19,"created":"2023-11-06 16:38:30.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/46/881546/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/19"}}},"commit":{"parents":[{"commit":"7b43812a2c15c5a75d605aee8ec7ba4d20610d88","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7b43812a2c15c5a75d605aee8ec7ba4d20610d88"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2023-11-06 16:38:30.000000000","tz":0},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/99b4b4326c670c8c566d38da6f4f359bc1c3d809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/99b4b4326c670c8c566d38da6f4f359bc1c3d809"}]},"branch":"refs/heads/master","description":"Rebase"},"af68af87b5a721eaec4afc5beb5b824a028cc47f":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2024-01-30 08:48:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/20"}}},"commit":{"parents":[{"commit":"fe75926d18e5123010b92cb71f4cb9939e3c21b4","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fe75926d18e5123010b92cb71f4cb9939e3c21b4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-30 08:48:08.000000000","tz":60},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/af68af87b5a721eaec4afc5beb5b824a028cc47f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/af68af87b5a721eaec4afc5beb5b824a028cc47f"}]},"branch":"refs/heads/master"},"376708ca622be1065acac2cb255a60ffddf0c29a":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2024-02-05 17:14:08.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/21"}}},"commit":{"parents":[{"commit":"fa820adf265486e33c42d0f53d0e36ec15883610","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fa820adf265486e33c42d0f53d0e36ec15883610"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-02-05 17:13:49.000000000","tz":60},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/376708ca622be1065acac2cb255a60ffddf0c29a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/376708ca622be1065acac2cb255a60ffddf0c29a"}]},"branch":"refs/heads/master"},"dd14c051ea93d92aadcd28a5b4930f0b35544c2b":{"kind":"REWORK","_number":22,"created":"2024-02-06 07:03:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/22"}}},"commit":{"parents":[{"commit":"fa820adf265486e33c42d0f53d0e36ec15883610","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fa820adf265486e33c42d0f53d0e36ec15883610"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-02-06 07:03:22.000000000","tz":60},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/dd14c051ea93d92aadcd28a5b4930f0b35544c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/dd14c051ea93d92aadcd28a5b4930f0b35544c2b"}]},"branch":"refs/heads/master"},"dafc1e7072239c201522df55b6723c9ac1e6fc8d":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2024-02-06 11:33:58.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/23"}}},"commit":{"parents":[{"commit":"664ada9bdbc134b57d6882e4548425b27efe2bd8","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/664ada9bdbc134b57d6882e4548425b27efe2bd8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-02-06 11:33:47.000000000","tz":60},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/dafc1e7072239c201522df55b6723c9ac1e6fc8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/dafc1e7072239c201522df55b6723c9ac1e6fc8d"}]},"branch":"refs/heads/master"},"a51075a42e082d97cef80dce0fb7640d467388fe":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2024-02-12 12:12:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/24"}}},"commit":{"parents":[{"commit":"c39d0872a8b275520ee1e48434ba19c42ebff7c6","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c39d0872a8b275520ee1e48434ba19c42ebff7c6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-02-12 12:12:33.000000000","tz":60},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a51075a42e082d97cef80dce0fb7640d467388fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a51075a42e082d97cef80dce0fb7640d467388fe"}]},"branch":"refs/heads/master"},"4678f68e184c474dfe529f927e665a5b6d64151a":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2024-02-12 12:42:02.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/25"}}},"commit":{"parents":[{"commit":"ceac0f54dc01cc8d8150557b8ebc612bb947dbe8","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ceac0f54dc01cc8d8150557b8ebc612bb947dbe8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-02-12 12:41:54.000000000","tz":60},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4678f68e184c474dfe529f927e665a5b6d64151a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4678f68e184c474dfe529f927e665a5b6d64151a"}]},"branch":"refs/heads/master"},"1557ef0786359e8a29865d88090d867c66e375ba":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2024-02-12 12:50:41.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/26"}}},"commit":{"parents":[{"commit":"86bbb79857cd6677cf1e7a895796509aa743b02e","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/86bbb79857cd6677cf1e7a895796509aa743b02e"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-02-12 12:50:34.000000000","tz":60},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1557ef0786359e8a29865d88090d867c66e375ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1557ef0786359e8a29865d88090d867c66e375ba"}]},"branch":"refs/heads/master"},"9ff107cc88fad71bb563f3d0cf2d59027704824a":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2024-02-12 13:08:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/27"}}},"commit":{"parents":[{"commit":"3f18fadc35e0716c8fc255f1b680a1989a34f4a9","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3f18fadc35e0716c8fc255f1b680a1989a34f4a9"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-02-12 13:08:24.000000000","tz":60},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9ff107cc88fad71bb563f3d0cf2d59027704824a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9ff107cc88fad71bb563f3d0cf2d59027704824a"}]},"branch":"refs/heads/master"},"96f4191a85a33a80a95049f0f824e33029b92795":{"kind":"NO_CHANGE","_number":28,"created":"2024-02-12 14:57:58.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/28"}}},"commit":{"parents":[{"commit":"c32683be6793df8b7488bc1102a5e247d1dd73df","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c32683be6793df8b7488bc1102a5e247d1dd73df"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-02-12 14:57:45.000000000","tz":60},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/96f4191a85a33a80a95049f0f824e33029b92795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/96f4191a85a33a80a95049f0f824e33029b92795"}]},"branch":"refs/heads/master"},"1d95bc687eef4449b0aab8b6934611bf78394b09":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2024-04-12 15:47:46.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/29"}}},"commit":{"parents":[{"commit":"17c3652647f20fd41519fccc9e35dcab18e1c569","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/17c3652647f20fd41519fccc9e35dcab18e1c569"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-12 15:47:28.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1d95bc687eef4449b0aab8b6934611bf78394b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1d95bc687eef4449b0aab8b6934611bf78394b09"}]},"branch":"refs/heads/master"},"3f3458889f9e0e495a95221c7a34b89f14074e8a":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2024-04-15 10:55:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/30"}}},"commit":{"parents":[{"commit":"68151934e32e4fda59a4907049948d56f0199f14","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/68151934e32e4fda59a4907049948d56f0199f14"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-15 10:55:21.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3f3458889f9e0e495a95221c7a34b89f14074e8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3f3458889f9e0e495a95221c7a34b89f14074e8a"}]},"branch":"refs/heads/master"},"e4728b52ebb084e3dc826002903aff60f8d84abf":{"kind":"NO_CHANGE","_number":31,"created":"2024-04-15 10:57:18.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/31"}}},"commit":{"parents":[{"commit":"5bb75afc3ff8a84e9cbe95d763501683c8d0a923","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5bb75afc3ff8a84e9cbe95d763501683c8d0a923"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-15 10:57:06.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e4728b52ebb084e3dc826002903aff60f8d84abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e4728b52ebb084e3dc826002903aff60f8d84abf"}]},"branch":"refs/heads/master"},"4d6c7caaafbd99ec28b488facbfe122b78a52c6a":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2024-04-23 10:08:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/32"}}},"commit":{"parents":[{"commit":"094f6b56f045e94c7e40366764625b510baff83a","subject":"Add TERMINATED_HTTPS listener API tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/094f6b56f045e94c7e40366764625b510baff83a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-26 10:30:08.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-23 10:07:43.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4d6c7caaafbd99ec28b488facbfe122b78a52c6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4d6c7caaafbd99ec28b488facbfe122b78a52c6a"}]},"branch":"refs/heads/master"},"e1516b320f3d598704e2e0d5676c85ab246235ff":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2024-04-23 12:25:57.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/33"}}},"commit":{"parents":[{"commit":"6de2bc33816550e90f48f1863851b44cb76728aa","subject":"Annotate mem_listener_client attribute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6de2bc33816550e90f48f1863851b44cb76728aa"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-23 12:07:04.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-23 12:07:04.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e1516b320f3d598704e2e0d5676c85ab246235ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e1516b320f3d598704e2e0d5676c85ab246235ff"}]},"branch":"refs/heads/master"},"7451162244de8945d0a904472ea114c4778e827c":{"kind":"REWORK","_number":34,"created":"2024-04-23 16:34:52.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/34"}}},"commit":{"parents":[{"commit":"6de2bc33816550e90f48f1863851b44cb76728aa","subject":"Annotate mem_listener_client attribute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6de2bc33816550e90f48f1863851b44cb76728aa"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-23 12:07:04.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-23 16:34:36.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7451162244de8945d0a904472ea114c4778e827c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7451162244de8945d0a904472ea114c4778e827c"}]},"branch":"refs/heads/master"},"c03e9c39d24feffae9ac246f947e0655b7224547":{"kind":"REWORK","_number":35,"created":"2024-04-23 16:56:52.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/46/881546/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/46/881546/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/46/881546/35"}}},"commit":{"parents":[{"commit":"6de2bc33816550e90f48f1863851b44cb76728aa","subject":"Annotate mem_listener_client attribute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6de2bc33816550e90f48f1863851b44cb76728aa"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-23 12:07:04.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-23 16:53:25.000000000","tz":120},"subject":"Test new HSTS feature","message":"Test new HSTS feature\n\nTest HTTP Strict Transport Security with Octavia.\n\nPartial-Bug: #2017972\nChange-Id: Ie54714015e943fd1cb75ca95f8cf241fbc99268c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c03e9c39d24feffae9ac246f947e0655b7224547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c03e9c39d24feffae9ac246f947e0655b7224547"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
