)]}'
{"id":"openstack%2Foctavia-tempest-plugin~882567","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I68418293798dc37785ce97b98577c365739d4b85","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"stable-focal","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-05-09 10:32:40.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I68418293798dc37785ce97b98577c365739d4b85","subject":"Fix jobs running on Jammy vs. Focal nodes","status":"ABANDONED","created":"2023-05-08 15:32:19.000000000","updated":"2023-05-10 09:13:14.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f9a4f34c014a9ee4255ce174fa7c0e1f7598804d","_number":882567,"virtual_id_number":882567,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-09 13:38:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-08 15:32:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dad55321d08b1fcfa7d6a184f53dc31a3f883c6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 15:32:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67ce9d8d5fce44a4c125e3bc7d4a68b1b795e676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 15:32:49.000000000","message":"Patch Set 1: Verified-1\n\n(13 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  Job octavia-v2-dsvm-scenario not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: octavia-v2-dsvm-scenario-amphora-v2\n      parent: octavia-v2-dsvm-scenario\n      vars:\n        devstack_localrc:\n          OCTAVIA_ENABLE_AMPHORAV2_JOBBOARD: True\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n                default_provider_driver: amphorav2\n                enabled_provider_drivers: amphorav2:The v2 amphora driver.\n              task_flow:\n                jobboard_expiration_time: 100\n          test-config:\n            \"$TEMPEST_CONFIG\":\n              load_balancer:\n                enabled_provider_drivers: amphorav2:The v2 amphora driver.\n                provider: amphorav2\n\n  in \"openstack/octavia/zuul.d/amphorav2-jobs.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"2ffa3c5e1385020c64921ee7a33d7e6d851a58a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 15:37:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dc34673d18a55ca8bf2e0390f7ba5f4f073938cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 15:38:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3cbbea4ee584b4b1527b72ff9c156064c3c76b5a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 15:39:40.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"875b0904c76ae47c51ca043d59834046e5dd026d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 16:06:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c240c60e46a834861b020c4cecf87d91560d814f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 16:08:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8fac1299d582daa4a78be29a51de93d6b0fb2877","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 16:15:57.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"eb0e4a75b237a29d36a3db3f7b3f675fc9a3f2f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 19:14:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3008162c1ce445feaf8a7a3853e279d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d2e006bc544c44945f1279b6b08f7f : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206ee31298c94c879968cd468c8b43b3 : SUCCESS in 2m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/779eea7d1703452195067593e7953411 : SUCCESS in 1h 29m 22s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/4dd257dbbd614320b59122cd7772719a : SUCCESS in 1h 14m 14s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/048db18cd5e14a7ba11b1e79eb4d8d94 : SUCCESS in 1h 07m 46s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/540fb5d00fb34ba6b558cf7ce770f4b4 : SUCCESS in 43m 08s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4e0ad27e30ec4f449f4a290df982018b : SUCCESS in 1h 27m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc56b3a238f042ffbd11977e035dca7f : SUCCESS in 1h 59m 18s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/a91243efa5014d0fabe12f76d84683da : SUCCESS in 2h 14m 21s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/3e4e3631a10b4b1ca70afb50c98f0628 : SUCCESS in 2h 16m 39s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/c683dda337414475a36db11252f1450e : SUCCESS in 1h 53m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9ad8a5a171fa4fcea48ef109db7bf1ae : SUCCESS in 41m 08s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/50eda99eae624f6aaa28fa5a295b9d62 : SUCCESS in 44m 56s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/c1f0071f913248e7a609f413766561e6 : SUCCESS in 35m 36s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/e4a35fc57b9b49bdb4cc3e39667831e3 : SUCCESS in 38m 31s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6fdb0fe755ec4931816f41446c44a572 : SUCCESS in 2h 09m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/369be58a881041bd960add094a203405 : FAILURE in 12m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/578a53b28d8f4d1991e16fdbd9c53eb2 : SUCCESS in 1h 42m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/542ea44046aa4a7ca45d2452d384fea4 : SUCCESS in 1h 21m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f5e8064b7daa403182567cf458baef3c : TIMED_OUT in 3h 03m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d43770440ad94a27aafd677eb4224bd7 : SUCCESS in 36m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/85169b01e7414a06a9b1e26ffc39ecc6 : SUCCESS in 37m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/7e56e10ff6ad44a9bf8007871eadb380 : SUCCESS in 34m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/5601003c90fe45988eff5985c9908377 : SUCCESS in 36m 34s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/579a6cb4cd1a4856b6c22aeec7e05025 : SUCCESS in 35m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/729b2944be1f473383efec5b9453f51e : SUCCESS in 2h 18m 30s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/ea44ad89531b4750bc25f6c38a96471c : FAILURE in 37m 34s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/fbf927f208d64842ab292e06323a2a5a : FAILURE in 29m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"286c548d1fd513868cf326e73451ef28975c2a06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-09 07:07:11.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"9ef0bf31dec685fc9e1c000338fa76ef8913dd42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 07:07:32.000000000","message":"Patch Set 6: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch stable/train.  The error was:\n\n  Job octavia-v2-dsvm-noop-py2-api not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-python-jobs\n        - openstack-python3-train-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - octavia-tox-tips\n      check:\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-noop-py2-api:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-py2-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n              voting: false\n          - octavia-v2-dsvm-py2-scenario-centos-7:\n              voting: false\n          - octavia-v2-dsvm-scenario-two-node:\n              nodeset: octavia-single-node-ubuntu-bionic\n              voting: false\n          - octavia-v2-dsvm-py2-scenario-two-node:\n              nodeset: octavia-single-node-ubuntu-bionic\n              voting: false\n          - octavia-v2-dsvm-tls-barbican:\n              nodeset: octavia-single-node-ubuntu-bionic\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n              voting: false\n          - octavia-v2-dsvm-scenario-amphora-v2:\n              nodeset: octavia-single-node-ubuntu-bionic\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n              voting: false\n          - octavia-v2-act-stdby-iptables-dsvm-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n              voting: false\n          - octavia-v2-act-stdby-iptables-dsvm-py2-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n              voting: false\n          - octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7:\n              voting: false\n      queue: octavia\n      gate:\n        fail-fast: true\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-noop-py2-api:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-xenial:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-bionic:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos7:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-fedora28:\n              branches: ^(?!stable/.*).*$\n          - octavia-amphora-image-build\n\n  in \"openstack/octavia/zuul.d/projects.yaml@stable/train\", line 3, column 3","accounts_in_message":[],"_revision_number":6},{"id":"d350083f4e03cc8925747c02b96fbaad8d38ddb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-09 07:15:38.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"72c4ef479cd54b981af5a0967538e443a93fd565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 07:15:59.000000000","message":"Patch Set 7: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch stable/train.  The error was:\n\n  Job octavia-v2-dsvm-py2-scenario not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-python-jobs\n        - openstack-python3-train-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - octavia-tox-tips\n      check:\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-noop-py2-api:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-py2-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n              voting: false\n          - octavia-v2-dsvm-py2-scenario-centos-7:\n              voting: false\n          - octavia-v2-dsvm-scenario-two-node:\n              nodeset: octavia-single-node-ubuntu-bionic\n              voting: false\n          - octavia-v2-dsvm-py2-scenario-two-node:\n              nodeset: octavia-single-node-ubuntu-bionic\n              voting: false\n          - octavia-v2-dsvm-tls-barbican:\n              nodeset: octavia-single-node-ubuntu-bionic\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n              voting: false\n          - octavia-v2-dsvm-scenario-amphora-v2:\n              nodeset: octavia-single-node-ubuntu-bionic\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n              voting: false\n          - octavia-v2-act-stdby-iptables-dsvm-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n              voting: false\n          - octavia-v2-act-stdby-iptables-dsvm-py2-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n              voting: false\n          - octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7:\n              voting: false\n      queue: octavia\n      gate:\n        fail-fast: true\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-noop-py2-api:\n              nodeset: octavia-single-node-ubuntu-bionic\n          - octavia-v2-dsvm-scenario:\n              nodeset: octavia-single-node-ubuntu-bionic\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-xenial:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-bionic:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos7:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-fedora28:\n              branches: ^(?!stable/.*).*$\n          - octavia-amphora-image-build\n\n  in \"openstack/octavia/zuul.d/projects.yaml@stable/train\", line 3, column 3","accounts_in_message":[],"_revision_number":7},{"id":"4eeaff0aed676d3b434b41cfb862316c7ca0d879","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-09 07:24:25.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"e60a8747f9a916637d86c3fee6ddb18f7646f719","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-09 07:25:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0ccb57dc097b7d73f544dbdf79e72d2a5019c16d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 10:31:01.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/838be881d34a48c7bd3021db0c497f97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/563e5bd131a34d85bc1fa99589ef8e62 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dfe4d38f8be45399b33caf3f8aaec5d : SUCCESS in 2m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b6656eaa45e4067a6f0718679b1a9ef : SUCCESS in 1h 15m 04s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/ca03363c240c43c693cbe8d187d19742 : SUCCESS in 1h 15m 58s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/e4b701d757a1418eab8f83d37e2cd93f : NODE_FAILURE Node request 300-0021170636 failed in 0s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/54e92af131384c7fbf225ce9b9e93faa : SUCCESS in 54m 32s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/573f473631ba4275963c32a55fd960e4 : SUCCESS in 1h 26m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/16b2bb07b52a433daa14024831dc1a39 : SUCCESS in 2h 13m 54s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/c8fd0302af83450eb3d04fe8f8aa0b28 : SUCCESS in 2h 10m 03s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/04fe183c8b7d4de0bda8323f3c33893a : SUCCESS in 2h 29m 58s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/da95af9c32a849359450d39ede51a297 : SUCCESS in 1h 47m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/48a9157d7645414fb8f598a8bb07cfbb : SUCCESS in 48m 00s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/8857c28899bb495eacd0910386be0b28 : NODE_FAILURE Node request 300-0021170644 failed in 0s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/1458384e8f8244b5a77aec9893d4499f : SUCCESS in 41m 47s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/36a341e1ef3c4b84838eb141378bfeaf : SUCCESS in 44m 40s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0bda73bae73d4d16b64765a7bf3d9e33 : TIMED_OUT in 2h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8f728a6523f406194c8e3638c2487eb : NODE_FAILURE Node request 300-0021170648 failed in 0s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/a1abac8cb55f45a3bb4e752c789b2c10 : SUCCESS in 1h 48m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8dfeb6b19255448f9607a25c1d7d2a84 : SUCCESS in 1h 29m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/def4ef8ed18d48f79a6cc863e7a48e9f : TIMED_OUT in 3h 04m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33d28069fef0499885c23dbbc34608ce : SUCCESS in 40m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/8d6a2f593b3a41c9bceb7c602152c448 : SUCCESS in 41m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/b53961bf47984ea19c117b26c2b9b8b0 : NODE_FAILURE Node request 300-0021170654 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/a121b7cb04a94e9a8f0415c9789931f7 : SUCCESS in 37m 20s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/61196b17e17c4e9cb5af9d8dade049c9 : NODE_FAILURE Node request 300-0021170656 failed in 0s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/34d71c7fe2dd417295569351de41ec07 : TIMED_OUT in 2h 34m 50s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5b69a17cb3264083bf196e9d98c75dbf : FAILURE in 24m 59s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/2326b9405bf54147aef9dbc0575ff390 : FAILURE in 25m 47s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"51262235233d103c8b2d155b8fc0890b20fb2a27","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-09 10:32:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fe81d218f414603ba5b021d4d03ab89b8cda51d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 13:38:55.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7b8f99deeed4c7d914c36753bc0ae8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76021d1d37f84c378e2711d4354db2d6 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b853149dbe4d57b17abb10b57683a8 : SUCCESS in 4m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0db0bb8f8bf44cdfa18cdfa33614339b : SUCCESS in 1h 11m 14s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/d22fc006569949ae87e94d54440483b4 : SUCCESS in 1h 23m 55s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/1eb0dddefc524332ba9f88da6d9d8cfa : SUCCESS in 1h 27m 10s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/b426b373131a40ee9b0c916aafc83aa4 : SUCCESS in 56m 51s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b745720c5b094dd7b356215f109fbd5e : SUCCESS in 1h 33m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d6a3589b51445ffb4f0de63f8b40b20 : SUCCESS in 2h 14m 55s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/df5c21c22f73439c841990643434c7f0 : SUCCESS in 2h 33m 38s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/b75435e557e6408a9b3cc8a22b6f6615 : SUCCESS in 1h 55m 18s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/6f5c14ba228848239e4cd5d8575c034c : SUCCESS in 2h 16m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f7b77c92eb3145418956acc1534ebab0 : SUCCESS in 39m 04s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/cde7206503ad4d5bbcd6e8bcb5955583 : SUCCESS in 43m 21s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/f65b514c699140009f0e12f16db5aeb2 : SUCCESS in 37m 10s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/964c9024ddb04a33ac594a499835421d : SUCCESS in 44m 19s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/209cb033e672424b8083b23ef9aa579f : SUCCESS in 2h 05m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec69fb603a6441dda1e05cc4aff7e0bb : FAILURE in 9m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/ef28faea1bd0491f9186dfdf3a554984 : SUCCESS in 1h 59m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/88ccc025126440e6a710f3e08577cd84 : SUCCESS in 1h 41m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/61e7ef141e3b4d82975b5b0a92147e85 : TIMED_OUT in 3h 03m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ee75139340f4c4eb3e5c20528a846ea : SUCCESS in 43m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/def1aabd076c46739c119a0497015d1d : SUCCESS in 37m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/a833a947472146f2962ace6ab5df8224 : SUCCESS in 40m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/379aedbc15514c9cab483a9dda8c82a4 : SUCCESS in 38m 05s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/8f05ee966f7941feaa833145fcdfb75d : SUCCESS in 33m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8ec2060265c04d428ee8055a3282ac4f : SUCCESS in 2h 13m 12s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/057aad80688944189f881f470a3e0329 : FAILURE in 36m 02s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/fd10f394059e4e958f2c684de681c02c : FAILURE in 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f9a4f34c014a9ee4255ce174fa7c0e1f7598804d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-10 09:13:14.000000000","message":"Abandoned\n\nsuperceded by https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/882722","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"e17262ca40f774b4ca0f3610c7c6575f3a8bfc57","revisions":{"f1fecfbbf419fe03f1c236a49bfeadc95f9800fd":{"kind":"REWORK","_number":1,"created":"2023-05-08 15:32:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/67/882567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/67/882567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/1"}}},"commit":{"parents":[{"commit":"e82bb6741ff82fa8b975990a7f5dbe8d4a295c66","subject":"Merge \"Updating Octavia tempest jobs on Ubuntu jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e82bb6741ff82fa8b975990a7f5dbe8d4a295c66"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 15:31:42.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 15:31:42.000000000","tz":120},"subject":"Fix jobs running on Jammy even on stable branches","message":"Fix jobs running on Jammy even on stable branches\n\nChange-Id: I68418293798dc37785ce97b98577c365739d4b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f1fecfbbf419fe03f1c236a49bfeadc95f9800fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f1fecfbbf419fe03f1c236a49bfeadc95f9800fd"}]},"branch":"refs/heads/master"},"275426bc45185b8988db4723db9540c40991f0ab":{"kind":"REWORK","_number":2,"created":"2023-05-08 15:37:05.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/67/882567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/67/882567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/2"}}},"commit":{"parents":[{"commit":"e82bb6741ff82fa8b975990a7f5dbe8d4a295c66","subject":"Merge \"Updating Octavia tempest jobs on Ubuntu jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e82bb6741ff82fa8b975990a7f5dbe8d4a295c66"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 15:31:42.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 15:36:56.000000000","tz":120},"subject":"Fix jobs running on Jammy even on stable branches","message":"Fix jobs running on Jammy even on stable branches\n\nChange-Id: I68418293798dc37785ce97b98577c365739d4b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/275426bc45185b8988db4723db9540c40991f0ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/275426bc45185b8988db4723db9540c40991f0ab"}]},"branch":"refs/heads/master"},"edb7028a19b70988df7773ac3ffde1c9ed2c938d":{"kind":"REWORK","_number":3,"created":"2023-05-08 15:38:54.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/67/882567/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/67/882567/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/3"}}},"commit":{"parents":[{"commit":"e82bb6741ff82fa8b975990a7f5dbe8d4a295c66","subject":"Merge \"Updating Octavia tempest jobs on Ubuntu jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e82bb6741ff82fa8b975990a7f5dbe8d4a295c66"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 15:31:42.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 15:38:46.000000000","tz":120},"subject":"Fix jobs running on Jammy even on stable branches","message":"Fix jobs running on Jammy even on stable branches\n\nChange-Id: I68418293798dc37785ce97b98577c365739d4b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/edb7028a19b70988df7773ac3ffde1c9ed2c938d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/edb7028a19b70988df7773ac3ffde1c9ed2c938d"}]},"branch":"refs/heads/master"},"51d3efdef34f723867722b12be4e4e24e808a1dd":{"kind":"REWORK","_number":4,"created":"2023-05-08 16:06:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/67/882567/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/67/882567/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/4"}}},"commit":{"parents":[{"commit":"e82bb6741ff82fa8b975990a7f5dbe8d4a295c66","subject":"Merge \"Updating Octavia tempest jobs on Ubuntu jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e82bb6741ff82fa8b975990a7f5dbe8d4a295c66"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 15:31:42.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 16:05:18.000000000","tz":120},"subject":"Fix jobs running on Jammy even on stable branches","message":"Fix jobs running on Jammy even on stable branches\n\nBecause octavia-tempest-plugin is branchless, job definitions need to\nwork for all octavia branches.\n\nChange-Id: I68418293798dc37785ce97b98577c365739d4b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/51d3efdef34f723867722b12be4e4e24e808a1dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/51d3efdef34f723867722b12be4e4e24e808a1dd"}]},"branch":"refs/heads/master"},"9f35010e7e0dec1cd39f1c5412a2f6cb864806b2":{"kind":"REWORK","_number":5,"created":"2023-05-08 16:08:56.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/67/882567/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/67/882567/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/5"}}},"commit":{"parents":[{"commit":"e82bb6741ff82fa8b975990a7f5dbe8d4a295c66","subject":"Merge \"Updating Octavia tempest jobs on Ubuntu jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e82bb6741ff82fa8b975990a7f5dbe8d4a295c66"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 15:31:42.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 16:08:48.000000000","tz":120},"subject":"Fix jobs running on Jammy even on stable branches","message":"Fix jobs running on Jammy even on stable branches\n\nBecause octavia-tempest-plugin is branchless, job definitions need to\nwork for all octavia branches.\n\nChange-Id: I68418293798dc37785ce97b98577c365739d4b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9f35010e7e0dec1cd39f1c5412a2f6cb864806b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9f35010e7e0dec1cd39f1c5412a2f6cb864806b2"}]},"branch":"refs/heads/master"},"0633325f323a81a4a08f2b1b72576030946d10e7":{"kind":"REWORK","_number":6,"created":"2023-05-09 07:07:11.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/67/882567/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/67/882567/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/6"}}},"commit":{"parents":[{"commit":"e82bb6741ff82fa8b975990a7f5dbe8d4a295c66","subject":"Merge \"Updating Octavia tempest jobs on Ubuntu jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e82bb6741ff82fa8b975990a7f5dbe8d4a295c66"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 15:31:42.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-09 07:04:27.000000000","tz":120},"subject":"Fix jobs running on Jammy vs. Focal and cleanup","message":"Fix jobs running on Jammy vs. Focal and cleanup\n\nBecause octavia-tempest-plugin is branchless, job definitions need to\nwork for all octavia branches \u003e\u003dwallaby. Older branches should not use\nthe definitions from current master.\n\nRemoved Python2 jobs because they are not supported anymore since Ussuri.\n\nChange-Id: I68418293798dc37785ce97b98577c365739d4b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0633325f323a81a4a08f2b1b72576030946d10e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0633325f323a81a4a08f2b1b72576030946d10e7"}]},"branch":"refs/heads/master"},"094519390e7b21984a417f1c560999025ae133e2":{"kind":"REWORK","_number":7,"created":"2023-05-09 07:15:38.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/67/882567/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/67/882567/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/7"}}},"commit":{"parents":[{"commit":"e82bb6741ff82fa8b975990a7f5dbe8d4a295c66","subject":"Merge \"Updating Octavia tempest jobs on Ubuntu jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e82bb6741ff82fa8b975990a7f5dbe8d4a295c66"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 15:31:42.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-09 07:15:10.000000000","tz":120},"subject":"Fix jobs running on Jammy vs. Focal and cleanup","message":"Fix jobs running on Jammy vs. Focal and cleanup\n\nBecause octavia-tempest-plugin is branchless, job definitions need to\nwork for all octavia branches \u003e\u003dwallaby. Older branches should not use\nthe definitions from current master.\n\nRemoved certain Python2 jobs because they are not supported anymore\nsince Ussuri.\n\nChange-Id: I68418293798dc37785ce97b98577c365739d4b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/094519390e7b21984a417f1c560999025ae133e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/094519390e7b21984a417f1c560999025ae133e2"}]},"branch":"refs/heads/master"},"ca10a110b01e2888005815c0ac49c2032a5c13b7":{"kind":"REWORK","_number":8,"created":"2023-05-09 07:24:25.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/67/882567/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/67/882567/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/8"}}},"commit":{"parents":[{"commit":"e82bb6741ff82fa8b975990a7f5dbe8d4a295c66","subject":"Merge \"Updating Octavia tempest jobs on Ubuntu jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e82bb6741ff82fa8b975990a7f5dbe8d4a295c66"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 15:31:42.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-09 07:24:03.000000000","tz":120},"subject":"Fix jobs running on Jammy vs. Focal nodes","message":"Fix jobs running on Jammy vs. Focal nodes\n\nBecause octavia-tempest-plugin is branchless, job definitions need to\nwork for all octavia branches \u003e\u003dwallaby. Older branches should not use\nthe definitions from current master.\n\nChange-Id: I68418293798dc37785ce97b98577c365739d4b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ca10a110b01e2888005815c0ac49c2032a5c13b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ca10a110b01e2888005815c0ac49c2032a5c13b7"}]},"branch":"refs/heads/master"},"e17262ca40f774b4ca0f3610c7c6575f3a8bfc57":{"kind":"REWORK","_number":9,"created":"2023-05-09 07:25:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/67/882567/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/67/882567/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/67/882567/9"}}},"commit":{"parents":[{"commit":"e82bb6741ff82fa8b975990a7f5dbe8d4a295c66","subject":"Merge \"Updating Octavia tempest jobs on Ubuntu jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e82bb6741ff82fa8b975990a7f5dbe8d4a295c66"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-08 15:31:42.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-09 07:25:46.000000000","tz":120},"subject":"Fix jobs running on Jammy vs. Focal nodes","message":"Fix jobs running on Jammy vs. Focal nodes\n\nBecause octavia-tempest-plugin is branchless, job definitions need to\nwork for all octavia branches \u003e\u003dwallaby. Older branches should not use\nthe definitions from current master.\n\nChange-Id: I68418293798dc37785ce97b98577c365739d4b85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e17262ca40f774b4ca0f3610c7c6575f3a8bfc57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e17262ca40f774b4ca0f3610c7c6575f3a8bfc57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
