)]}'
{"id":"openstack%2Foctavia-tempest-plugin~882722","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ic76288560086d225fa59aee399cd75ae4c459559","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-05-10 18:44:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic76288560086d225fa59aee399cd75ae4c459559","subject":"Fix jobs running on Jammy vs. Focal vs. Bionic nodes","status":"MERGED","created":"2023-05-09 14:51:57.000000000","updated":"2023-05-10 18:46:15.000000000","submitted":"2023-05-10 18:44:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882722","meta_rev_id":"b5b624530edd9a8faf66b5598a1d72a86e59f913","_number":882722,"virtual_id_number":882722,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 18:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-05-10 16:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-10 09:11:37.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-05-10 16:14:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 18:30:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-10 09:11:37.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-05-10 16:14:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"2af09621ba1dc41c3d232d11cd708bc3cc119445","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-09 14:51:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74138b9a110abae2f433e30484c5c2269bf74dee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-09 15:17:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b129115de6cf4c58005641181d5f2d2c79dc2801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 18:30:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b954ddc420414890a8a3f37ae1fff027\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3993e0d839b54017afcaef12ffc7a237 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ae4c32ba154fb3b403302ed46b4913 : SUCCESS in 3m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7a70270c08b442989c91b9011e69e571 : SUCCESS in 1h 26m 52s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/bd2d5180475041eeb78e9b8006e44999 : SUCCESS in 1h 19m 00s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/284b47ba22f6474b99173a85a6bf603d : SUCCESS in 1h 30m 43s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/52aaf12779b7420c97fd17cadd79e923 : SUCCESS in 51m 31s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cf48be57d4c24a1ca2bb20133e7df606 : SUCCESS in 1h 20m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2f529ee2b5d4c299c9f83414844d306 : SUCCESS in 2h 08m 11s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/8500b0df66484d70a688358b93e22d16 : SUCCESS in 2h 22m 50s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/e22bd6a06e9e45208e3d7cde8ba969af : SUCCESS in 1h 52m 50s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/7244fb1b6d5e4669b81bf53c95061a49 : SUCCESS in 2h 01m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/822ca8e72f854d4f89064c838f73bd59 : SUCCESS in 44m 12s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/6e4408b48fc1420fb2384977d134ef61 : SUCCESS in 42m 17s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/1cb05eea341b4b1389213833a769db75 : SUCCESS in 42m 27s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/4a5f4716150c42dea57b7e83ed02ca78 : SUCCESS in 43m 55s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d98959e59476425284b73f5711d48bf2 : SUCCESS in 2h 13m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/c51ca8df7d374aeaaa3ae2105a2c0d92 : FAILURE in 9m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/61ab3400dbda45658bf44f6a7ec87230 : SUCCESS in 1h 58m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4493529d60c94dbe8a8efcb3d57a33d6 : SUCCESS in 1h 18m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8a3e03b56a4f4203bc628ffe6ff35375 : TIMED_OUT in 3h 04m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/642c518f3a8f496f99afc8546cf250a1 : SUCCESS in 42m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/8c1a397f02a9469c993a28aa61e6c9fd : SUCCESS in 34m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/7e22e16f94754e518130687c573ee837 : SUCCESS in 37m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/6dedd681303241e7b73cf19d724c8b8e : SUCCESS in 41m 26s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/79c1177046144773930e2651530a05ec : SUCCESS in 38m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/010c65e1dbcf4f859fa57f5b15283163 : SUCCESS in 2h 20m 32s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/6956104a543f4743939879282a6b21f1 : FAILURE in 43m 39s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/87ed4b6309a34fcc82ea6a66b64b83ea : FAILURE in 40m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60fd22bb9ca668533e1883b296825a8c44606c34","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-10 09:11:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1617843128c43f630aab73d8f99bba58cb80f98c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-05-10 16:14:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"221cd718492ba772d81fec0b8a0f7d7dfe26425c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 16:15:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b0bfee097fc1051c64b582d55cb5df4eab8d200b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 18:44:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc410374d6aa42d6b3cc00c82a312242\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e590c3b43a48c3af6a84503db151a6 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ccd774727ad4ca79868964a193e953d : SUCCESS in 2m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b35a7b0fdc654f4e8ed8887fb280dc25 : SUCCESS in 1h 18m 51s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/79ec6d161d824c48be392a5e6531fc6f : SUCCESS in 1h 16m 11s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/1f8bae40a50d4b2f9b62ad181074d843 : SUCCESS in 1h 13m 34s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/1a9c559f00fa413c9cb777834e1b0b01 : SUCCESS in 58m 07s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/256fa4cb5efa4256b22edc8208062c98 : SUCCESS in 2h 01m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8dfa22605584441bafd067fe4bba8732 : SUCCESS in 2h 27m 01s\n- octavia-v2-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/bba7dad82ad6418c97ab4c1b26977d97 : SUCCESS in 2h 17m 23s\n- octavia-v2-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/70af3c8c1f63430bb76340dbe658ab27 : SUCCESS in 2h 08m 57s\n- octavia-v2-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/6c24bbbb381e43b0b6b9a51a68004edd : SUCCESS in 1h 47m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/db7ca938cd3743aaae926ab6c241e71c : SUCCESS in 46m 15s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/a649e92a8f534c368df0f68f9d26b692 : SUCCESS in 42m 26s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/d7d668ab2cb549889495915c2e5d6579 : SUCCESS in 43m 04s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/a81e93ec58574510b939d4dafbcd83f5 : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":2},{"id":"8b04a24f281e8c8e2eef479665f77d74924de535","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 18:44:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b5b624530edd9a8faf66b5598a1d72a86e59f913","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 18:46:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45c916d585354824965ceaa1bc90b81f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e64c02c89a5f4106bc40beb2992b350e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"07909503b014ad5f62b3d3075d5b5f919674224c","revisions":{"e6235bd018108025b2f4092ebf1dcd78a36f0bf3":{"kind":"REWORK","_number":1,"created":"2023-05-09 14:51:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/882722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/22/882722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/22/882722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/22/882722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/22/882722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/22/882722/1"}}},"commit":{"parents":[{"commit":"e82bb6741ff82fa8b975990a7f5dbe8d4a295c66","subject":"Merge \"Updating Octavia tempest jobs on Ubuntu jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e82bb6741ff82fa8b975990a7f5dbe8d4a295c66"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-09 14:51:42.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-09 14:51:42.000000000","tz":120},"subject":"Fix jobs running on Jammy vs. Focal nodes","message":"Fix jobs running on Jammy vs. Focal nodes\n\nBecause octavia-tempest-plugin is branchless, job definitions need to\nwork for all octavia branches \u003e\u003dwallaby. Older branches should not use\nthe definitions from current master.\n\nChange-Id: Ic76288560086d225fa59aee399cd75ae4c459559\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e6235bd018108025b2f4092ebf1dcd78a36f0bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e6235bd018108025b2f4092ebf1dcd78a36f0bf3"}]},"branch":"refs/heads/master"},"07909503b014ad5f62b3d3075d5b5f919674224c":{"kind":"REWORK","_number":2,"created":"2023-05-09 15:17:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/882722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/22/882722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/22/882722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/22/882722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/22/882722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/22/882722/2"}}},"commit":{"parents":[{"commit":"e82bb6741ff82fa8b975990a7f5dbe8d4a295c66","subject":"Merge \"Updating Octavia tempest jobs on Ubuntu jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e82bb6741ff82fa8b975990a7f5dbe8d4a295c66"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-09 14:51:42.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-09 15:17:23.000000000","tz":120},"subject":"Fix jobs running on Jammy vs. Focal vs. Bionic nodes","message":"Fix jobs running on Jammy vs. Focal vs. Bionic nodes\n\nBecause octavia-tempest-plugin is branchless, job definitions need to\nwork for all octavia branches \u003e\u003dwallaby. Older branches should not use\nthe definitions from current master.\n\nChange-Id: Ic76288560086d225fa59aee399cd75ae4c459559\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/07909503b014ad5f62b3d3075d5b5f919674224c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/07909503b014ad5f62b3d3075d5b5f919674224c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
