)]}'
{"id":"openstack%2Foctavia-tempest-plugin~886120","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ibf465d4e187dcd09381832521a8a0d72ad09c5cb","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-06-21 11:23:54.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ibf465d4e187dcd09381832521a8a0d72ad09c5cb","subject":"Fixing nested-virt jobs","status":"NEW","created":"2023-06-14 18:35:06.000000000","updated":"2023-06-21 11:23:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"908c624e743b64adc4540bcb0e7346ace6e3a8a5","_number":886120,"virtual_id_number":886120,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-06-15 08:34:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-21 11:12:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":-1,"date":"2023-06-21 11:23:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-14 21:39:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-21 11:12:19.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"3888a8350f925330b162123aa0f6bcf888298326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-14 18:35:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"384c342d8646971e93d6f2f150005026f81a54b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 21:39:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e65f24a813a24d97a6e356576a3a9482\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e71ebc2f274a429f8fbe566094f256 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ccf4bbecc844c5c9b7f0c30465391cb : SUCCESS in 2m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1aedc5f117c4bd98db6d516e9e2f002 : SUCCESS in 1h 50m 33s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/21980b1f2c364576bbe799886d153bd7 : SUCCESS in 1h 24m 34s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/f64550b840f34409b2277d45d5b719a8 : SUCCESS in 1h 20m 25s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/f429d7abec6e421787cff32a9931a36b : SUCCESS in 49m 10s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6b5b87aed8cf416bb6ab2657091ae314 : SUCCESS in 1h 39m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/007a87c89c4b44ef97cf5961354835b0 : SUCCESS in 1h 40m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d8322e116a774e289190f9c899f56e41 : SUCCESS in 1h 23m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/95c011142b0845a2a93227033af5dca6 : SUCCESS in 1h 21m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/3db5fa71005c4ba899e4f7621f621f22 : SUCCESS in 1h 04m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-xena https://zuul.opendev.org/t/openstack/build/d7a7a35d1f774fcf8831866085b3c801 : SUCCESS in 1h 20m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-xena https://zuul.opendev.org/t/openstack/build/7508db468b0744e6b493f45488a7111e : SUCCESS in 1h 02m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-wallaby https://zuul.opendev.org/t/openstack/build/999dc2b349f94ad6b68052e218f60583 : SUCCESS in 1h 15m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-wallaby https://zuul.opendev.org/t/openstack/build/1183775c436543a48086bbf373b09b1f : SUCCESS in 1h 08m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/83fd2cbb03f5486ba48c5b6359952b68 : SUCCESS in 39m 52s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/9abe839497c7480eb23e84ad9e8c07f0 : SUCCESS in 40m 59s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/be27e19370ad4962872dbbb6d1102e5f : SUCCESS in 37m 14s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/71bac5b6cfdd496780e1bd3522ae0371 : SUCCESS in 38m 26s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a270496a1971428c82b4750623727b9c : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c7db84f16fa4cd7a0a6b761d7be6b15 : FAILURE in 9m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/4280c6ec753e4404a616b805182fe0c0 : SUCCESS in 1h 47m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/12774f1b2adf477eac81413abf68cb9d : SUCCESS in 1h 19m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eb1ea47969df4585bd11226c5ff9cb93 : TIMED_OUT in 3h 03m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b6e3bbab19e64a4d9106defc4e4f716d : SUCCESS in 33m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/2f2758f43f84417f94ff259842a84b4f : SUCCESS in 32m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/8965ab4d06e346ee817e1f97d178f803 : SUCCESS in 30m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/eb1dbd0b15a748fc9d560b3fb3db7163 : SUCCESS in 41m 39s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/f1369d49c816472a93440031217be36d : SUCCESS in 35m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/82dc4710f66f49eb866f5c37727afe03 : TIMED_OUT in 2h 33m 46s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/70d4d5253f0d450d8f369c43080072f9 : SUCCESS in 45m 58s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7224e9e9cb1e476a96d161399c9dc6b0 : SUCCESS in 46m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a074bdf50d16e5c52d5c9a0fe3aad5ae4c1aaf90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-15 05:26:54.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3a69782eb2fa4ffb501dfa17dcc17b3eca8e6c57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-15 05:27:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b9dab20f09f7e71f87bfb218240eea608400b6e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 08:34:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23a30dcc276b47b996bb6f830d158d0e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/355dc81da0a447ea849c5d5977440504 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20991121f158406fb8c48ee3a432bf55 : SUCCESS in 5m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ac7ab8c554f2427f86a423f0d60483a5 : SUCCESS in 1h 02m 49s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/4633186d56224b70acf9b9ffd1977ba7 : SUCCESS in 1h 04m 32s\n- octavia-v2-dsvm-noop-api-stable-xena https://zuul.opendev.org/t/openstack/build/31eb6bbc3e694d3db9b159176320c848 : SUCCESS in 1h 10m 24s\n- octavia-v2-dsvm-noop-api-stable-wallaby https://zuul.opendev.org/t/openstack/build/df9acd1b907d44c891a7aa0e5e0913ec : SUCCESS in 54m 56s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6a8159f818734685b6eb69a1e425f4cb : SUCCESS in 1h 15m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7cf2434f12a745e89c0b1c103fc13509 : SUCCESS in 2h 00m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8d7877e2fe4640808e20a5d5f5e316c8 : SUCCESS in 1h 20m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/3e17d15abaed4391bf8be6a92d00dd8f : SUCCESS in 1h 23m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/07ac89389fe4403b8063e2ec0e31c32d : SUCCESS in 56m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-xena https://zuul.opendev.org/t/openstack/build/10a89e3cf5b34870aae15ead52ff90a2 : SUCCESS in 1h 24m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-xena https://zuul.opendev.org/t/openstack/build/3bdba612fcc74d8193fcef5fe3d97ea1 : SUCCESS in 1h 22m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-wallaby https://zuul.opendev.org/t/openstack/build/a9c62719b21443a2b39b52afcb8fede0 : SUCCESS in 1h 28m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-wallaby https://zuul.opendev.org/t/openstack/build/8776efa27ea24a9bb063c918b02c9a8a : SUCCESS in 52m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6556bdb12400403e8d9aec59ef4a83d3 : SUCCESS in 39m 36s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/f40566e511ce43519b5acf78f78998cb : SUCCESS in 35m 18s\n- octavia-v2-dsvm-tls-barbican-stable-xena https://zuul.opendev.org/t/openstack/build/a83ae4359f4643ae8719e6276d3cd9ab : SUCCESS in 45m 08s\n- octavia-v2-dsvm-tls-barbican-stable-wallaby https://zuul.opendev.org/t/openstack/build/a86f1ff49e0c420b8dc73e7a03c9f228 : SUCCESS in 47m 00s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/41f1594832a74651aaa38ce3bb17223c : TIMED_OUT in 2h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/13d8dd3b293f48caa63a4859d0ec7ddc : FAILURE in 8m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/6dae7c87c54f430d8251875fa663233b : SUCCESS in 1h 42m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/40116d618d4945be86555b6ee07ce53b : SUCCESS in 1h 08m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fcf0076927404a44a7479ff2df1c3fe0 : TIMED_OUT in 3h 03m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d40a2eee16c6421da8d9868e4353b4e5 : SUCCESS in 31m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/21a29ed9e9544ec19a168ae66a9d8c9f : SUCCESS in 42m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-xena https://zuul.opendev.org/t/openstack/build/c456ba1330ed4bd08ae42ce881c3140f : SUCCESS in 42m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-wallaby https://zuul.opendev.org/t/openstack/build/b4eec630d039444394aef7710ded5bb4 : SUCCESS in 29m 06s (non-voting)\n- octavia-v2-dsvm-spare-pool-stable-wallaby https://zuul.opendev.org/t/openstack/build/36470ab91c214d45b73f65e4de353c9d : SUCCESS in 39m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/80af099b12b348ab88b867457b0ef2d9 : TIMED_OUT in 2h 32m 38s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/db657bd080e34890a637d4ff5116a5f2 : SUCCESS in 47m 50s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/4da42b4c167f48478c85c6a3f550cb74 : SUCCESS in 52m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bbd0958b289bf39158c3810a2e42f97ffcec5d8c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-15 15:09:38.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"40806e068ea5cd34435cd0ba13952620ae4256c8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-21 07:42:57.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"ac00d49d87f4320f296116b523a36a79901cbccc","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-06-21 11:12:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"908c624e743b64adc4540bcb0e7346ace6e3a8a5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-21 11:23:54.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c4739db483c331d306ec7b2aaa8786c3d90f77b8","revisions":{"9768a576390ba7497f5f3a9ae5f7d23fd0db1616":{"kind":"REWORK","_number":1,"created":"2023-06-14 18:35:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/20/886120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/20/886120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/20/886120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/20/886120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/20/886120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/20/886120/1"}}},"commit":{"parents":[{"commit":"b9af9163cb7a5277e4c72dc1becb563148f7fa1b","subject":"Splitting scenario jobs in 2 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b9af9163cb7a5277e4c72dc1becb563148f7fa1b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-06-14 18:22:40.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-06-14 18:26:15.000000000","tz":120},"subject":"Fixing nested-virt jobs","message":"Fixing nested-virt jobs\n\nCirros 0.6.1 crashes on OVH nested-virt nodes, using 0.6.1 seems to fix\nthe issue.\nThis patch may be reverted when devstack defaults to 0.6.2\n\nChange-Id: Ibf465d4e187dcd09381832521a8a0d72ad09c5cb\nCloses-Bug: #2023559\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9768a576390ba7497f5f3a9ae5f7d23fd0db1616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9768a576390ba7497f5f3a9ae5f7d23fd0db1616"}]},"branch":"refs/heads/master"},"ee0ffc6d82f4f766cdd9e440f38c3eed8d3b793c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-06-15 05:26:54.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/20/886120/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/20/886120/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/20/886120/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/20/886120/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/20/886120/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/20/886120/2"}}},"commit":{"parents":[{"commit":"b9af9163cb7a5277e4c72dc1becb563148f7fa1b","subject":"Splitting scenario jobs in 2 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b9af9163cb7a5277e4c72dc1becb563148f7fa1b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-06-14 18:22:40.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-06-15 05:26:41.000000000","tz":120},"subject":"Fixing nested-virt jobs","message":"Fixing nested-virt jobs\n\nCirros 0.6.1 crashes on OVH nested-virt nodes, using 0.6.2 seems to fix\nthe issue.\nThis patch may be reverted when devstack defaults to 0.6.2\n\nChange-Id: Ibf465d4e187dcd09381832521a8a0d72ad09c5cb\nCloses-Bug: #2023559\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ee0ffc6d82f4f766cdd9e440f38c3eed8d3b793c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ee0ffc6d82f4f766cdd9e440f38c3eed8d3b793c"}]},"branch":"refs/heads/master"},"c4739db483c331d306ec7b2aaa8786c3d90f77b8":{"kind":"REWORK","_number":3,"created":"2023-06-15 05:27:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/20/886120/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/20/886120/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/20/886120/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/20/886120/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/20/886120/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/20/886120/3"}}},"commit":{"parents":[{"commit":"b9af9163cb7a5277e4c72dc1becb563148f7fa1b","subject":"Splitting scenario jobs in 2 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b9af9163cb7a5277e4c72dc1becb563148f7fa1b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-06-14 18:22:40.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-06-15 05:27:44.000000000","tz":120},"subject":"Fixing nested-virt jobs","message":"Fixing nested-virt jobs\n\nCirros 0.6.1 crashes on some nested-virt nodes, using 0.6.2 seems to fix\nthe issue.\nThis patch may be reverted when devstack defaults to 0.6.2\n\nChange-Id: Ibf465d4e187dcd09381832521a8a0d72ad09c5cb\nCloses-Bug: #2023559\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c4739db483c331d306ec7b2aaa8786c3d90f77b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c4739db483c331d306ec7b2aaa8786c3d90f77b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
