)]}'
{"id":"openstack%2Foctavia-tempest-plugin~900171","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I1a463e320b94b0e99e92541581e1ee5feffd356a","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-19 18:34:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a463e320b94b0e99e92541581e1ee5feffd356a","subject":"Drop implementation to use pyOpenSSL to manage pkcs12 certs","status":"MERGED","created":"2023-11-06 13:04:51.000000000","updated":"2024-01-19 18:35:25.000000000","submitted":"2024-01-19 18:34:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900171","meta_rev_id":"7ff31838ec028716565abc3b14553631ce57c3ba","_number":900171,"virtual_id_number":900171,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-19 18:34:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-11-08 00:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-01-19 16:36:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-01-19 16:36:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-06 14:08:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-08 00:53:23.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-01-19 16:36:55.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"34a94a43ea39c79c96371d0bb4de47587ee212f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 13:04:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8504fba103da3c3fb5b505d4034fcdb6bfaf640","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 13:05:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b54da26583dd21b81f48814f172c35a76e7e92f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 14:08:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8546e3c95e1452da57d5cceead4bec2\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0b2e18307527439bb02f6f7fc4706997 : SUCCESS in 1m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d34a8f1b666476e9b6e4aa3d44fce57 : FAILURE in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ba7bf0c0f9469f9fb932fff808bb04 : FAILURE in 2m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/201491d39bd343618c5834cd423db4c8 : FAILURE in 25m 17s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/bb93387d240e4e14aa231a23692ca940 : FAILURE in 19m 08s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/93792832db2845a09c316e5573c7b8fa : FAILURE in 20m 16s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/d659524ef1674197afcc578e105fd82a : FAILURE in 26m 06s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8bf571e12cd84e63af45a28e57f2d7ea : FAILURE in 24m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/df671d976f834e609deb44eb1d4e1f13 : FAILURE in 24m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e171f1b23ec64bb9a99b8aa7e1515847 : FAILURE in 25m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/c3daa6a389d2438b8b674940d360ce06 : FAILURE in 25m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/a1f4e7382743419d812b9e93dca536d5 : FAILURE in 24m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/6efe9cc5fde54ba09b534809900e0208 : FAILURE in 30m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/8ce26d924f6d4522b5f5478eb201f3c5 : FAILURE in 31m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/a84f56ecd47f4b8faf8139d973b9e273 : FAILURE in 30m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/9d8b46e6ef304fcaada7304c45ddf0dc : FAILURE in 27m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/37e1a257047846bd867dc4d5ff2d4a55 : FAILURE in 25m 18s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/17b44a879d8540ee8a8d7c4564b8c77c : FAILURE in 34m 04s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/fa3da33d6f784eba9b8b0aecdff9c9e1 : FAILURE in 37m 55s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/e0539fa23246465a9ef9fafb0147dbce : FAILURE in 33m 20s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d778226fb39a4974850aa7de625dc6be : FAILURE in 30m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/fd61c6c8cf5148cbb6fda2c919013b61 : FAILURE in 41m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0b48d05112e442f9a60059296023ff56 : FAILURE in 40m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4055c9dbf7d84d32b7baf0ecb617eef7 : FAILURE in 36m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fc0b35240aa424a92732ee3b96cfec7 : FAILURE in 24m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/9794f172023b47408b6d401c69fdc641 : FAILURE in 24m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/095c2fe76f764c1393fe020aac9af633 : FAILURE in 30m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/f31825718f8d4f97bed3c2905ca898ec : FAILURE in 25m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d1bacfafd0ae490ab766fc9da13447ed : FAILURE in 31m 20s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2831d56dc84549bb9556d3dc3c384b66 : FAILURE in 37m 55s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e7b7c54887154f18b45480e8774de99b : FAILURE in 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"872aba0568e8e6151f614296aa2163eecf933713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 14:26:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"63dbcc74a2ad2c3d7d386328d0c2728d6051b65c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 17:30:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50347393eb5b465d8fb9e593644ba14d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/734d9eba2b0f46199bd9ce6153f1792e : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1722901ab73c4795ad7f05131ce1dfb4 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6cf7f5043e94310bbd45bdf46a760d0 : SUCCESS in 2m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e8c0c47902064bbb8dc4fd296b839bfc : SUCCESS in 1h 31m 42s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/e07464e86c7e4093af851f53b30eb088 : SUCCESS in 1h 14m 33s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/6455a6c42722436ab21fc3d0cc575b69 : SUCCESS in 1h 31m 07s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/987c758ecd6848c58d82d41e66185ba0 : SUCCESS in 1h 28m 35s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/892506d06e834cad8990ba6b4017f416 : SUCCESS in 1h 23m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9c8d883ee6bb4e83ac907a8ed098a9d2 : SUCCESS in 1h 35m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a7f7ac8dcba84b5184f82bdd87607d13 : SUCCESS in 53m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7f4c89aac3b24f7d836e1bc60d9957d7 : SUCCESS in 1h 44m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/0b4dbda02c5f4439ac7de3d425ff3672 : SUCCESS in 1h 33m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/833e2724c33849bebe06dfa2fccd3430 : SUCCESS in 1h 15m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/0a262a90952a4dcf9b8bbb107fbedb35 : SUCCESS in 56m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/c5f9fe3395644dd98963f33ba7140a86 : SUCCESS in 1h 21m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/58975226c155466291a9bc7116580347 : SUCCESS in 1h 24m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4cb7e89e0a4046f6a6074f02f073b1ac : SUCCESS in 38m 13s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7dfc2c9ae40f4d0d92c3405a45b561a1 : SUCCESS in 31m 54s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/5ce5f778c95b4e769a89f7097a5470d6 : SUCCESS in 49m 01s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/909956f8aae340319527f6660f31da33 : SUCCESS in 36m 22s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2333e0e2c7af4c50baec583cfe67d3ec : SUCCESS in 2h 33m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/052310820b6447cd83f7e2d1970145ea : SUCCESS in 1h 54m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1716be7e12aa4b369be77cb8cdec06a2 : SUCCESS in 1h 21m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd173bd93fc04ce180c8aeeed9d4c616 : TIMED_OUT in 3h 03m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4b982c35cd0482690b5a508eac9f459 : SUCCESS in 43m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/a985ba17f4ed4f87a187c65fe76c37b8 : SUCCESS in 34m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/4dd2a7020b024c0791c2811bc90162ab : SUCCESS in 33m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/ada544e8baf043a99b4939cdf01eb7af : SUCCESS in 32m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/39af813f8cad47428626f0b833391a22 : SUCCESS in 2h 18m 09s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/5ec651641178493aae33831af2496b96 : SUCCESS in 34m 28s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/4865c6af83664302bf45561a350a93d8 : SUCCESS in 55m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b44abed589849d342257257663fbe1028b0fefa4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-08 00:53:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0d9b7338c621bac469cd7f12a06814b1d8b15eae","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-19 16:36:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f7b65f5965b602dca77bb8fc907874b0c771f76c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 16:37:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4c94d7e471c5119393601fcf35167f98b5456de5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 18:34:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f90ee1b5b24537a124b5da2d3ec2bb\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3b9e58c01af840e4bd30a27cde0477e9 : SUCCESS in 1m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89c07dd8cfab4e608abf772b3a233e7e : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cee5a9d94014b6480c76ed931c85a24 : SUCCESS in 3m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5b69dc0061bd4af08eaf6ca87263b155 : SUCCESS in 1h 19m 11s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/438fd3bcf8494e48a0bc4d54b897e4f0 : SUCCESS in 1h 03m 30s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/3ba9b96ac5e641bfb83e51ce019b8530 : SUCCESS in 59m 04s\n- octavia-v2-dsvm-noop-api-stable-yoga https://zuul.opendev.org/t/openstack/build/4669a524488b4766ae802dd8b853abe7 : SUCCESS in 1h 25m 48s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/735e4cee770a4bd3903c0c4ae8ffd878 : SUCCESS in 1h 09m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/05e022f134b947bebada00a1f07b7bab : SUCCESS in 1h 53m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ae0ef8ed2de1419ca6201c5b74e70c5e : SUCCESS in 1h 30m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/90ee2d518dfc4412b983582a42550583 : SUCCESS in 1h 30m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/9dc060a6fc8240e88d824d1053786594 : SUCCESS in 1h 26m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/9b3e8473dd344468acd21ae9d874601d : SUCCESS in 1h 30m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/f249d22e87c84678a3fc03f34cf9351c : SUCCESS in 1h 20m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/0783cfd28c6648f9ba4f2e3bb8c411c2 : SUCCESS in 1h 32m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/e3c0ab6cdfa148789674bb560bc86cf9 : SUCCESS in 1h 19m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2918832387964ce69b47c018a9ccf41e : SUCCESS in 44m 36s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ab85bf2ebbab41f4a5f1514707b4e81c : SUCCESS in 53m 22s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/9ceaa3b59c40497badf87223ff6b79b4 : SUCCESS in 46m 32s\n- octavia-v2-dsvm-tls-barbican-stable-yoga https://zuul.opendev.org/t/openstack/build/6ada9e86104247dfb748fc11a421b92c : SUCCESS in 44m 19s","accounts_in_message":[],"_revision_number":3},{"id":"64292094de31d33ecf4a1ffde7e50b4ad9069fce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 18:34:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7ff31838ec028716565abc3b14553631ce57c3ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 18:35:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c91e44f6d2924de9b44f8a8e7eaae0cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7c24ebd0ac49a799494816d2a8445c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"25872b36de18a857b1ef36450980f52c8e08e97d","revisions":{"87ab3399ab42655619194c209f949694cf5ab888":{"kind":"REWORK","_number":1,"created":"2023-11-06 13:04:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/900171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/71/900171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/900171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/900171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/900171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/900171/1"}}},"commit":{"parents":[{"commit":"bacca6692696a79aa57f15c2815f8db41ff6adf3","subject":"Fixed typo in override-checkout branch for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bacca6692696a79aa57f15c2815f8db41ff6adf3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 13:03:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 13:04:48.000000000","tz":540},"subject":"Drop implementation to use pyOpenSSL to manage pkcs12 certs","message":"Drop implementation to use pyOpenSSL to manage pkcs12 certs\n\nThe cryptography library has been bumped to 3.1 in upper-constraints\nfile during Ussuri, which is quite old. So we no longer have to\nmaintain logic for cryptography \u003c 3.0.\n\nChange-Id: I1a463e320b94b0e99e92541581e1ee5feffd356a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/87ab3399ab42655619194c209f949694cf5ab888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/87ab3399ab42655619194c209f949694cf5ab888"}]},"branch":"refs/heads/master"},"5b6a8dc49e05fc40068f935b3914dcd91c153ce6":{"kind":"REWORK","_number":2,"created":"2023-11-06 13:05:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/900171/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/71/900171/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/900171/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/900171/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/900171/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/900171/2"}}},"commit":{"parents":[{"commit":"bacca6692696a79aa57f15c2815f8db41ff6adf3","subject":"Fixed typo in override-checkout branch for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bacca6692696a79aa57f15c2815f8db41ff6adf3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 13:03:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 13:05:41.000000000","tz":540},"subject":"Drop implementation to use pyOpenSSL to manage pkcs12 certs","message":"Drop implementation to use pyOpenSSL to manage pkcs12 certs\n\nThe cryptography library has been bumped to 3.1 in upper-constraints\nfile during Ussuri, which is quite old. So we no longer have to\nmaintain logic for cryptography \u003c 3.0.\n\nChange-Id: I1a463e320b94b0e99e92541581e1ee5feffd356a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5b6a8dc49e05fc40068f935b3914dcd91c153ce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5b6a8dc49e05fc40068f935b3914dcd91c153ce6"}]},"branch":"refs/heads/master"},"25872b36de18a857b1ef36450980f52c8e08e97d":{"kind":"REWORK","_number":3,"created":"2023-11-06 14:26:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/900171/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/71/900171/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/900171/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/900171/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/900171/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/900171/3"}}},"commit":{"parents":[{"commit":"bacca6692696a79aa57f15c2815f8db41ff6adf3","subject":"Fixed typo in override-checkout branch for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bacca6692696a79aa57f15c2815f8db41ff6adf3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 13:03:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 14:26:15.000000000","tz":540},"subject":"Drop implementation to use pyOpenSSL to manage pkcs12 certs","message":"Drop implementation to use pyOpenSSL to manage pkcs12 certs\n\nThe cryptography library has been bumped to 3.1 in upper-constraints\nfile during Ussuri, which is quite old. So we no longer have to\nmaintain logic for cryptography \u003c 3.0.\n\nChange-Id: I1a463e320b94b0e99e92541581e1ee5feffd356a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/25872b36de18a857b1ef36450980f52c8e08e97d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/25872b36de18a857b1ef36450980f52c8e08e97d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
