)]}'
{"id":"openstack%2Foctavia-tempest-plugin~907181","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I26db31be06253c5941e3d011e1d974053b6cf381","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-02-06 08:27:12.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I26db31be06253c5941e3d011e1d974053b6cf381","subject":"WIP Delete listeners after each test in PoolScenarioTest","status":"NEW","created":"2024-01-30 09:01:17.000000000","updated":"2024-02-06 11:46:43.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"57b2bf20240254f19a9a7d064b1ad06f6e9f9439","_number":907181,"virtual_id_number":907181,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-06 11:46:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-30 14:01:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-30 11:33:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-30 14:01:47.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"ed643e10b4025c3f254b54b66a352ee6758e7903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-30 09:01:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4012deae1afd763547a2b85f640f0684e5d84cbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 11:33:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9f8bbb03e6544079566a654f4d68b2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9172b607c88a496493dff02a51353904 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d27434d9ee0246e7be4e78ba36324b13 : SUCCESS in 2m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bb5fb6e276ba408d98cacd43e6005f1f : SUCCESS in 1h 32m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b8c7ff68c51b4a338212b23b79c46651 : SUCCESS in 58m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/cefa30a01c54415abc7a48edc5ceb429 : SUCCESS in 1h 21m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/55b133e36a8445f5824f5b21ad392a3c : SUCCESS in 54m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/dc1d4fe003334b928a54ff1bf26e8d6c : SUCCESS in 1h 18m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/077c61a5e84a486b938513da64a51c6a : SUCCESS in 1h 03m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/dd51e8d20d5e47fcbd9acd397953e5a5 : SUCCESS in 1h 12m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/3fd2ea845da644f19b6ec26d1f2b9132 : SUCCESS in 1h 05m 29s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/787598c34ecd42d1ad55ff15df9ac92d : FAILURE in 20m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/0359906ba1b041a89bef2738f30b17c5 : SUCCESS in 1h 59m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bb7a66a587044f73b97197392ac25a38 : SUCCESS in 1h 30m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bf429028883a47babc49a6e8b4d70413 : FAILURE in 1h 35m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/384e036085184944aa73f000a052d175 : SUCCESS in 34m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3f202327035347639d955ea9e78b22a2 : SUCCESS in 28m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/4f2bb8389c76400395f2d56ba2e7335c : SUCCESS in 33m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/4ee5dd9aa0e64016a37f8635fe2618ad : SUCCESS in 40m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ca4ff79835b247edb19e7144e11e8271 : SUCCESS in 2h 31m 32s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/c57e952517e8472a8757f959c4d04242 : SUCCESS in 52m 01s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7eb0458c682d4a22b8290ceeee7caefa : SUCCESS in 43m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0aa588ae26871eaa1afa1289ba531ef2bd0c0b6","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-30 14:01:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cd96fb6ab6c35a42d2686315334c301671cfa714","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-30 14:18:55.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d7e5c299cf571154ff985d615649d4731bda7728","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-06 08:27:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57b2bf20240254f19a9a7d064b1ad06f6e9f9439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 11:46:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee7851b63e71475fbfebe415f70add89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0903647ead149bb98dcbebd09a2c165 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25eac830560346998db8c5254db05e5d : SUCCESS in 3m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6c117b875f464ea29f20c6ec205d285b : FAILURE in 2h 14m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/43e7349397884640ac6645f8f3745959 : FAILURE in 1h 32m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/952e48a75d1843359109698f18b5c445 : SUCCESS in 1h 46m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/651e930f28274393b6a770cf0c5679ea : SUCCESS in 1h 21m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/6fbd4c00fe84450888dcdc9ff9b15804 : SUCCESS in 1h 48m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/7d3673da9ed046c1974d5eded55db3bc : SUCCESS in 1h 16m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/4f62610cadd94f8f831369937d6b2bec : SUCCESS in 1h 57m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-yoga https://zuul.opendev.org/t/openstack/build/dc8bfeb8d97448448a0db4d60521b301 : SUCCESS in 1h 27m 46s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8191241361ab44f4b100ba79ef00ab30 : FAILURE in 31m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/274b5ba384fb46aa91692926c693a2d4 : SUCCESS in 1h 59m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6e1190cc88fd4314882f7de7a030a423 : SUCCESS in 1h 27m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/683cba6257684add80af307bf12b228f : TIMED_OUT in 3h 03m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3bc22628bf2f4c959245c4ea1d63196c : SUCCESS in 48m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ca7fcba10ea141f685eaf38db7b5abcd : SUCCESS in 49m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/8a2d11c1e78542a58c677bc4bef4a5a9 : SUCCESS in 50m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-yoga https://zuul.opendev.org/t/openstack/build/19d3a619cb9d429589436551cc7e21ea : SUCCESS in 48m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/df3a1f99c3ed465d96134062891387b8 : TIMED_OUT in 2h 34m 06s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/98f04c4c03d04dbc98b040822366bc99 : SUCCESS in 35m 48s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e7152849a5c745debff6c73d3aae17f7 : SUCCESS in 53m 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8c5df1d4326589520acf0d23282dc4697585a3a","revisions":{"c8c5df1d4326589520acf0d23282dc4697585a3a":{"kind":"REWORK","_number":1,"created":"2024-01-30 09:01:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/907181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/81/907181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/907181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/907181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/907181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/81/907181/1"}}},"commit":{"parents":[{"commit":"ffc51f76b2a9c0e6b148d360c3f8b4c712fb84ca","subject":"Merge \"Drop implementation to use pyOpenSSL to manage pkcs12 certs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ffc51f76b2a9c0e6b148d360c3f8b4c712fb84ca"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-01-30 09:00:39.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-01-30 09:00:51.000000000","tz":60},"subject":"WIP Delete listeners after each test in PoolScenarioTest","message":"WIP Delete listeners after each test in PoolScenarioTest\n\nThe test functions of this class create at most 24 listeners in the same\nload balancer, it can trigger memory allocation issues in haproxy with\nthe default connection_limit value.\n\nChange-Id: I26db31be06253c5941e3d011e1d974053b6cf381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c8c5df1d4326589520acf0d23282dc4697585a3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c8c5df1d4326589520acf0d23282dc4697585a3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
