)]}'
{"id":"openstack%2Foctavia-tempest-plugin~909845","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I22375b6772466e90f0ebebf0af96f80a73e42741","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-03-10 11:26:19.000000000","reason":"\u003cGERRIT_ACCOUNT_28609\u003e replied on the change","reason_account":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I22375b6772466e90f0ebebf0af96f80a73e42741","subject":"DNM Testing lower connection_limit in scenario.v2.test_pool","status":"NEW","created":"2024-02-22 08:32:30.000000000","updated":"2024-03-10 11:26:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f779f79182a044578b324b5fd4d0df5d0b4c51c0","_number":909845,"virtual_id_number":909845,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-22 11:49:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 11:49:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-10 11:26:19.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"state":"CC"}],"messages":[{"id":"9567028d0d11ba7f708e463cc5e4de547b531bb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-22 08:32:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c38785cec2e4189a169119d82a0a78f7902eec13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 11:49:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/025e6d4c82b1492cab90f6151f32651b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1be30cf4661400ba4e57e21c21f13db : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8df9800a3044f70b8884f14ec2ed3e9 : SUCCESS in 3m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/78d6b8add8fa4c5d9da4b1f02f7b4d0b : SUCCESS in 1h 31m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6956f59a4f134296822b82abe52d481a : SUCCESS in 57m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9a041a6ab8364596a2848d51548c20ea : SUCCESS in 1h 19m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b15a12ef01024c44bd19fba71d7bd787 : SUCCESS in 1h 02m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ec7b4e8d9713429dbf296afa85e8dbcd : SUCCESS in 1h 15m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7c7494b1ef1a4434bb54bb6daa46707b : SUCCESS in 1h 03m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/2c0131115f2749718530a85959148f02 : SUCCESS in 1h 17m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/fdfe099e09014e9fabcd093aba6dc2c5 : SUCCESS in 58m 31s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8a3707afc73e4ab29310f410200e55f2 : FAILURE in 27m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/7a16259098974eed9038e7ebc408391d : SUCCESS in 2h 07m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4b1c89380fcf4ac582bacd57651eee34 : SUCCESS in 1h 15m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/592c29d6304b485ab1ed282d14f2fcfe : TIMED_OUT in 3h 03m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73af54518b37423890a1b4230aa929ad : SUCCESS in 53m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/61ea5b8e022c4b699e7c8d0dbbf451fd : SUCCESS in 44m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/9bdfe515591c4be3b3486f6a27ed7421 : SUCCESS in 43m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/be7251d5f73641efae653882e8c22cc9 : SUCCESS in 44m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/61ed439e318743a2bfe7b4672e401bcb : TIMED_OUT in 2h 33m 30s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/247b4ccf4a274e0eb9145c9a70eb557b : SUCCESS in 50m 38s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/3ecfa38fcd0b4c2b96e2a4e8a28f43e2 : SUCCESS in 54m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40b7901ad4db6c12bf5e7a090d5bfc98a56eb8d4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-22 11:59:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1e9730b6ad1cccc2969d899cc4d29d1623399e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 15:04:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2a03226ff9948e58c279e1148095994\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80438ab689ee4c3786aef515e1065eda : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60e6044b0b014adb9e73bf66fa2fc556 : SUCCESS in 2m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/60d9d6d84e264a5199e846da17cb4fa9 : SUCCESS in 1h 41m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0b89f67863794ead9958c7ddaf9aa4b2 : SUCCESS in 1h 03m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2d174f71871342c19a2d7dd6ecaac352 : SUCCESS in 1h 19m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/3f7cc1a68a34461ba5a3c3a3ff25b45c : SUCCESS in 1h 05m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/c5469e1925f2488ab0d90ad247396b86 : SUCCESS in 1h 17m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/d655a1beb60b47459b548bcac7f50e8e : SUCCESS in 1h 03m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/8dbee31181ca4edc922376a4baafbc54 : SUCCESS in 1h 16m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/8ae52884e87c4c47afa7c46910a87e08 : SUCCESS in 1h 27m 11s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9037f6813451429cb085a76e33e178df : FAILURE in 21m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/29bba9efab4a424a9bf9c8f5a63edd71 : SUCCESS in 2h 10m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/63d5fd4e5ce94cefa02b5148180b3aff : SUCCESS in 1h 27m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f12c9ec9c36343dfa69a827045e09c31 : TIMED_OUT in 3h 03m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c6a4bcca17646c0874264207be3029e : SUCCESS in 47m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/df0346b5062b4599abf4399f6d4cce42 : SUCCESS in 37m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/282d980e6a374e46b4519b992f9ab9cb : SUCCESS in 43m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/31a4b46fa8014e8196aab968b6ed03ae : SUCCESS in 30m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8ad140c6801f4e8ea1885c2c64d7400f : TIMED_OUT in 2h 32m 41s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/3b01f30d2d3648bfa7ef258c4bb1c25b : SUCCESS in 46m 26s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/6dfc0b490d2f4d2e9221f49ceaf91e3e : SUCCESS in 56m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f779f79182a044578b324b5fd4d0df5d0b4c51c0","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2024-03-10 11:26:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4788a6a570a05cf126baeb6e825072c8eb0c4da2","revisions":{"4788a6a570a05cf126baeb6e825072c8eb0c4da2":{"kind":"REWORK","_number":1,"created":"2024-02-22 08:32:30.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/45/909845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/45/909845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/45/909845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/45/909845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/45/909845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/45/909845/1"}}},"commit":{"parents":[{"commit":"d7a446897f9e98c98821d9e176b88f0aaa46bbe2","subject":"Replace stable/yoga job by stable/2023.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d7a446897f9e98c98821d9e176b88f0aaa46bbe2"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 08:24:17.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 08:24:17.000000000","tz":-300},"subject":"DNM Testing lower connection_limit in scenario.v2.test_pool","message":"DNM Testing lower connection_limit in scenario.v2.test_pool\n\nChange-Id: I22375b6772466e90f0ebebf0af96f80a73e42741\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4788a6a570a05cf126baeb6e825072c8eb0c4da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4788a6a570a05cf126baeb6e825072c8eb0c4da2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
