)]}'
{"id":"openstack%2Foctavia-tempest-plugin~915851","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I39fa2347774704d77a83e7a44175e60366a02e84","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-05 00:14:35.000000000","reason":"Change was abandoned"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-09-05 00:14:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I39fa2347774704d77a83e7a44175e60366a02e84","subject":"Remove reference to train and ussuri","status":"ABANDONED","created":"2024-04-15 12:19:07.000000000","updated":"2024-09-05 00:14:35.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"99a340029859c47f7b59cd04cebb3eabe6c16dce","_number":915851,"virtual_id_number":915851,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-05 18:13:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-05 16:16:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-15 15:28:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-16 06:46:40.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-06-05 16:16:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"00892e152b5ee020fb76c794f948be0ab629c79e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 12:19:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccc3a28bcbc4a7cbaa9b7815ac5d794f859e8352","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 12:22:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b08047acb7325fc625730fc8f300bcccab4520bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 15:28:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5cdf473915a4113996c30c3d626e3fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/094054b50a5948ad9f1777601b278a7a : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff1e7669b37498b9fdbdd54dc034878 : SUCCESS in 3m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8133eab2a5f8419687050bb8b4af8cd4 : SUCCESS in 1h 17m 37s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/1785e08632184b219d38be7f28f1518b : SUCCESS in 1h 03m 12s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/1d5684de4d9749a1b14e83f9f24ed428 : SUCCESS in 1h 19m 50s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/d5e9a3c38de4453dad3c7473522ce7c6 : SUCCESS in 1h 20m 39s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f843ba5e2dce4b93a388d1b73d459e3d : SUCCESS in 1h 53m 38s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b1eaf904d8bb434fb093f4075e0c4c92 : SUCCESS in 1h 43m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1e9b48e9c4bb4e3894af74cd0afb616c : SUCCESS in 1h 27m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/fdf1dada74194de49f2dd910676011be : SUCCESS in 1h 24m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2121ca5113ed44a49499733290c4cc75 : SUCCESS in 1h 00m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/42a67f69fb2e4e188b56cac83c08523d : SUCCESS in 1h 23m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/fc1214e8fd0c44baafaf700c6afcd8f9 : SUCCESS in 1h 07m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/5369dfbfdc8144b89025af892cb9d613 : SUCCESS in 1h 21m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/8168bdb8e3094713bb5f34a62d781d59 : SUCCESS in 1h 02m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd13505d5e8c4b7dbebad911fd29e5db : SUCCESS in 49m 57s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/f47d023a47e54d11b75355905ad8091b : SUCCESS in 47m 06s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/379ad17e166c4eb48ecbb670ff206d18 : SUCCESS in 39m 14s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/02c4a625dbc24054b5c03990cf1ae9fa : SUCCESS in 38m 52s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0e3d8377ca4640b9bce79c74eb3b68d1 : FAILURE in 24m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/a7cc9acb0c2c427b876fe4328e253b73 : SUCCESS in 2h 05m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f7daaea885f94a91b6f7f5eb958fa306 : SUCCESS in 1h 34m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/42167193d17f4ea3bb3946eed3b8eaba : TIMED_OUT in 3h 03m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63488916725e4aaf8060f513807be2a9 : SUCCESS in 35m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/05ea5c03c96c4d55b65e64223ef0a6d8 : SUCCESS in 33m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/9a8ec9a435374027951fddb214075049 : SUCCESS in 42m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/8dbec627548b49aca01d4e4969f101b0 : SUCCESS in 34m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8214bdc8329e48d1aac063e2de7319d1 : TIMED_OUT in 2h 33m 15s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/c72104d9197149159340937cc3955c1f : FAILURE in 52m 14s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/2c60dfdc24464c8bbd401133b4254c3b : SUCCESS in 45m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f7295146d4f7a426f0d2ea0410c87992b03f143","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-16 06:46:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"98a6d4084506565ecb39d45ef20be8939f6062f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-05 15:08:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2b97bc652ce58ee8b1574cc3ceb70c95a3396da8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-05 16:16:18.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1f03fcb175e3de994878b37e28635e55bfe1b82a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 18:13:24.000000000","message":"Patch Set 3: Verified+1\n\n(5 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06bb4a7243e947f091ecac66613dc732\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52daa3d58163488998236b3ef559d6ad : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/917b30e6c5f14a87ae05224986589574 : SUCCESS in 2m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc5ea9f5387740ba83e0d3a3aedcaa4b : SUCCESS in 1h 36m 30s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/83a13fa461044fa899114a1d614581c6 : SUCCESS in 1h 06m 29s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7da35d5f92f34e45affd68b325925d21 : SUCCESS in 1h 14m 13s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0470252c46234714a1fb1d9a544783a2 : SUCCESS in 1h 56m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a052b328cff948af9e784537ecd69a8b : SUCCESS in 2h 23m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/24942b7ad888455192c60271b9af3783 : SUCCESS in 1h 04m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/09acbaefc5384464902c435531f8d46f : SUCCESS in 2h 04m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/53e97ec26c1d4ad9a0ac0031d92b6096 : SUCCESS in 1h 04m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/6651117fb13a46279a3d87c8ea434b3d : SUCCESS in 1h 55m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3918ae1df00b4f8ba373ed388287ab27 : SUCCESS in 55m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dc797526b53f4f34bb42d619019585b4 : SUCCESS in 49m 19s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/050e6243c60341cdb1d840cca98686a7 : SUCCESS in 35m 42s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/14af24381e314ae59d313215479c94ae : SUCCESS in 37m 42s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ad31594014e34aac821d742d3bfd1bfb : FAILURE in 24m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/428812e7f34047478434fb6d692ef5ff : SUCCESS in 2h 22m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4466a027057d44969b9bf4387db9f6db : SUCCESS in 1h 32m 41s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/fde33d22946040d58a395224dd178a85 : FAILURE in 26m 46s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6d836ce0625a458b9f098aafe602b169 : FAILURE in 25m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/10a71685ce78492bbbf597cc1a1b02f6 : TIMED_OUT in 3h 03m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3544946dd7f3419c8f99191997b6eb84 : SUCCESS in 34m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/90500892fb294b21a95ae2c42d85bdd9 : SUCCESS in 34m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/14d16b720a1348028166da9882b002d0 : SUCCESS in 39m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f58f6432cef743788d1bc418c52ed650 : TIMED_OUT in 2h 32m 48s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/fc1275f50ebc4e73a6f8c153e4ed31ab : SUCCESS in 55m 51s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/ebb8f3c0ff664b14b2be74a136607fe8 : SUCCESS in 50m 46s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 504","accounts_in_message":[],"_revision_number":3},{"id":"99a340029859c47f7b59cd04cebb3eabe6c16dce","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-05 00:14:35.000000000","message":"Abandoned\n\nthis was done by 0dbd292c8542202dbbc21c3f55fc30eb7d67df48","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"94df05da2d145a9e126ea74c6fd0371498ee15ed","revisions":{"39522b77e205f542ffd22dec2ec9e85dd228afeb":{"kind":"REWORK","_number":1,"created":"2024-04-15 12:19:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/915851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/915851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/915851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/915851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/915851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/915851/1"}}},"commit":{"parents":[{"commit":"a1a2bed19ffe387212364018c61f4a2ae1cff78b","subject":"Merge \"Add *.orig to .gitignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a1a2bed19ffe387212364018c61f4a2ae1cff78b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:15:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:15:06.000000000","tz":540},"subject":"Remove reference to train and ussir","message":"Remove reference to train and ussir\n\n... because these two branches are already EOL.\n\nChange-Id: I39fa2347774704d77a83e7a44175e60366a02e84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/39522b77e205f542ffd22dec2ec9e85dd228afeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/39522b77e205f542ffd22dec2ec9e85dd228afeb"}]},"branch":"refs/heads/master"},"31678bbc29ba0978f92a78a18fd64a7c40dffb58":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-04-15 12:22:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/915851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/915851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/915851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/915851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/915851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/915851/2"}}},"commit":{"parents":[{"commit":"a1a2bed19ffe387212364018c61f4a2ae1cff78b","subject":"Merge \"Add *.orig to .gitignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a1a2bed19ffe387212364018c61f4a2ae1cff78b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:15:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:22:09.000000000","tz":0},"subject":"Remove reference to train and ussuri","message":"Remove reference to train and ussuri\n\n... because these two branches are already EOL.\n\nChange-Id: I39fa2347774704d77a83e7a44175e60366a02e84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/31678bbc29ba0978f92a78a18fd64a7c40dffb58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/31678bbc29ba0978f92a78a18fd64a7c40dffb58"}]},"branch":"refs/heads/master","description":"Edit commit message"},"94df05da2d145a9e126ea74c6fd0371498ee15ed":{"kind":"REWORK","_number":3,"created":"2024-06-05 15:08:04.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/51/915851/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/51/915851/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/915851/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/915851/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/915851/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/51/915851/3"}}},"commit":{"parents":[{"commit":"cb2c9586cf0e97c5bff7ab1e3c8b5c6abad30dc0","subject":"Merge \"Use only keystone roles in keystone RBAC tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cb2c9586cf0e97c5bff7ab1e3c8b5c6abad30dc0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:15:06.000000000","tz":540},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-05 15:07:46.000000000","tz":120},"subject":"Remove reference to train and ussuri","message":"Remove reference to train and ussuri\n\n... because these two branches are already EOL.\n\nChange-Id: I39fa2347774704d77a83e7a44175e60366a02e84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/94df05da2d145a9e126ea74c6fd0371498ee15ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/94df05da2d145a9e126ea74c6fd0371498ee15ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
