)]}'
{"id":"openstack%2Foctavia-tempest-plugin~915853","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I8d42a55c723a07c7c366bd06f406b6a2e55f768e","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-05 00:15:09.000000000","reason":"Change was abandoned"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-09-05 00:15:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8d42a55c723a07c7c366bd06f406b6a2e55f768e","subject":"Add stable/2024.1 jobs","status":"ABANDONED","created":"2024-04-15 12:19:07.000000000","updated":"2024-09-05 00:15:09.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"982fbc8c5e3eb93547b23f7f9e9f86810f1b0268","_number":915853,"virtual_id_number":915853,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-05 18:14:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-05 16:15:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-15 12:20:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-16 06:52:49.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-06-05 16:15:19.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"c7bca469e35c6aa0f8f6cd8ab8287fe83d17138b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 12:19:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"952ec5a744931cd590addc1e2a6f611b9089648e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 12:20:36.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  Job octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 not defined\n\nThe problem appears in the the \"opendev.org/openstack/octavia-tempest-plugin\" project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - publish-openstack-docs-pti\n        - tempest-plugin-jobs\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - octavia-v2-dsvm-noop-api\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/projects.yaml@master\", line 3","accounts_in_message":[],"_revision_number":1},{"id":"7db692d980d770b457af8ff98eaf502c1217ac96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 12:22:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c1f928f77c6eade6a41a4bc5cdac55a601098cb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 12:22:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1f7b441006d52a1f924725eeedb94a2c05389ec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 15:50:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8505e77f948e41779b9d24908d5d5da1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37af8d2cc3e4dd8b7e2303d92821cbc : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/302894ce0f56413885482394c60eee66 : SUCCESS in 3m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4bae8e5be5a048dc995fb7ce3d3ed596 : SUCCESS in 1h 45m 30s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/862e35dc2fc14171ae3b2d3945adfada : SUCCESS in 1h 29m 05s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b3e187504fda410e882aca73df28b994 : SUCCESS in 1h 19m 15s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/c9366fb3b9914d2797d9a95221f7ae9d : SUCCESS in 1h 06m 17s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/fb60623f6cfc4edbb36071e376c1e085 : SUCCESS in 1h 21m 06s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ef71fe2beee64fedac9a44f6dff78435 : SUCCESS in 1h 33m 03s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/22576a92f63a450a836ed643d26567fb : SUCCESS in 1h 36m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fb0f94e479e54480b44fca8fde8b2d1e : SUCCESS in 1h 26m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/95bf17cf4c4d48478c59176bb50bf412 : SUCCESS in 2h 03m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/811073a7d6e449e388ef2f549b44b360 : SUCCESS in 52m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4e78d4f7c1cd4e0887a41b57a3556c9c : SUCCESS in 1h 50m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4face1b073084814a161ae59e75ff926 : FAILURE in 31m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/d0cd3d4a053a4a13b6d869a4830e1f9d : SUCCESS in 1h 46m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/fbc98d21ecad4d7c9dc3eb2e18d6e2fe : SUCCESS in 1h 03m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/0fc275dc134147b7b791e45b8ab25104 : SUCCESS in 1h 16m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/46fe25782138475985d66b764996e1c5 : SUCCESS in 52m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/864b696803364b23b9e4e9f81bbd9b72 : SUCCESS in 36m 41s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/c109c8b4c21e413ca96c11406896402f : SUCCESS in 52m 17s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/27fe4604c0224277a8eb912b1ff7228f : SUCCESS in 46m 15s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/0adfaa91c0a44a278b4c4e49c63d9ceb : SUCCESS in 53m 43s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/2d4a942553d444329c4393fbc4f05944 : SUCCESS in 56m 49s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/323bd934d66744d39e0f24efdeebe5a0 : FAILURE in 24m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/4116d020e28342e7adf6cfa66e079782 : SUCCESS in 2h 21m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a7d8ef0362b54a0082f549225073d5e4 : SUCCESS in 1h 28m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57351b89920b4418a8b2b5abf803d02d : TIMED_OUT in 3h 03m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b517efa46ba487fb62d991e8b9a8485 : SUCCESS in 42m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/28f260e5d13e47abb61d21928b925719 : SUCCESS in 47m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2cbf3455f8704066be4a091d17dc4b4a : SUCCESS in 49m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/6e2013adf91e4cd3b5ef625ef0c698fa : SUCCESS in 48m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/4cae268d95d54f348e09aba528e12ea5 : SUCCESS in 44m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/da3b450ec4ff41278e33e9e8aea894c6 : FAILURE in 30m 17s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/bbbff499e49a43488913c1eab9070961 : SUCCESS in 50m 11s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/655ad72a7a114422ac41f301694df19c : SUCCESS in 58m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0af36e3aaa01f3087ff5ec1ef6a31d9c49dde5fc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-16 04:57:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8e46d8e85d098f3828d1fa9a2ffdc4fe62bed3c7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-16 06:52:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"331d1877f0a0fb262aec0427b99a5dab1a48e0ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-16 07:02:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"55574281183f7d6f5096564b487b48957f0bbb5e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-16 07:02:09.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"35f811d1de6511ac0d4e80865c40dd12195eebdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-16 07:02:33.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"a8c392f030252e30b77e7b813357df114034a25d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 07:02:41.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/915852 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"46e9741fd44357c07970ef12068ef5d09da1f590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 10:11:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df7e71f8f0454a9b962eec2f09c2bcf7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bbf15a42e654840a66bbfec49d28405 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ec028d985449cc8d4f2d0385b11ad9 : SUCCESS in 4m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4d6468240a624d44a8c7af5f80da7e69 : SUCCESS in 1h 35m 24s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/36c1e2984fd94d3ca1b60dcce78a69d8 : SUCCESS in 1h 26m 53s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/3f63f362c31e44f2bb7a39713b47d599 : SUCCESS in 1h 30m 54s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/c3b431febe8846ccabbcb9a1009b8a80 : SUCCESS in 1h 53m 58s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/0022a3840c4643b1894b4d7c987b8e42 : SUCCESS in 1h 25m 25s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3fbe4fb4299b47a6ac36116a5a508f96 : SUCCESS in 1h 42m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/943b4ebdd3aa4f1389192ade15b082f3 : SUCCESS in 1h 59m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c1a34922270c4b52a2b85f14eaaaf606 : SUCCESS in 1h 12m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/497ecd5463da47408b5710c2eceec531 : SUCCESS in 2h 10m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/afee61d2731e45cda1dfa1a8e8798d33 : SUCCESS in 54m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/1276906da35b4352be2e017b4f0c9ee2 : SUCCESS in 1h 41m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/50a1ff9c17bb42d3ad2c9ed74d3facbb : SUCCESS in 53m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/65f55b4d27d5485693f1818523c1d5ad : SUCCESS in 1h 49m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/d14b510df0ec48acb8a5cb3312bd3c72 : SUCCESS in 56m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/11addb12fd8443dfa1db3f66b2ede307 : SUCCESS in 1h 37m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/c8ea059335e24bac97f5be467a92beb5 : SUCCESS in 1h 29m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b376934634b4b3a8da6aad19fc85d1b : SUCCESS in 1h 01m 32s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/f20b731934b848ec95a701fd9e19d5b4 : SUCCESS in 56m 18s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2723703747574ce291da0f6b93612d74 : SUCCESS in 54m 46s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3d8b4e5cf28140ebabae4bae0a9cbd83 : SUCCESS in 34m 10s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/8ef60805e0654157949ce6d087df0596 : SUCCESS in 37m 03s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5416cc26abd34e6cbba5b76d367155a3 : FAILURE in 27m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/5d658be16dac4c5bb3d9676db105d2f2 : SUCCESS in 2h 23m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/db8b157d86dd48daa835798dde0fc52b : SUCCESS in 1h 18m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7a62996d315f4c03b8d7ec44d95a2938 : TIMED_OUT in 3h 03m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04a5760f8fbd4c7a87402bdbbaf31380 : SUCCESS in 32m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/96c412d68c0343998a66b69b89930e11 : SUCCESS in 51m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4f96bf2b36e64c25a13c216b98fbd3e3 : SUCCESS in 31m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/c57e08381bfd4d3c86f348dc83c7c798 : SUCCESS in 28m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/eb819cb7016e4c12b6300c5fdc798dee : SUCCESS in 31m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f5ebfe4e03944bcebcf0364293fca4ed : SUCCESS in 2h 08m 03s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/f4d15665bf774f1ebfc7a76f0cc8d060 : SUCCESS in 37m 49s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/5b3670f5c1e9483e89fe1eb7de1bdc98 : SUCCESS in 37m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"67d17200005cea803cee2986ae13a2e5261b51ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-05 15:08:04.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"1b2d60cb5a1e29bd889706bfac33f851d3b0affa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-05 16:15:19.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5f63957b4892a71a23e2879ce7ddcf6543d04efc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 18:14:05.000000000","message":"Patch Set 6: Verified+1\n\n(5 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65cec48d4c684a19817bc5d89b66311e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13e64a8536ac46e590b781e536a12ce8 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3af4ecb7593944c89b1730409a27e170 : SUCCESS in 4m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af99878754b14b208b8baa253682efcf : SUCCESS in 1h 11m 07s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/395670d4582647539a2591a5107b56c9 : SUCCESS in 1h 01m 43s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/a521b33d02c64896a70cb3d2da1bb4bf : SUCCESS in 1h 03m 05s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/982220c26dbc4131b0ee9d7c662e7dfb : SUCCESS in 1h 09m 53s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0af3277b56c046d58d3fa31fb6924d3d : SUCCESS in 1h 39m 54s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/db2ebc9204da47f4a29fee06f8f6a436 : SUCCESS in 1h 51m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c8301c8ca02e453d9196f437265cf942 : SUCCESS in 1h 01m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/487f13a290994dd3b8aa8cdc9ff5fe9f : SUCCESS in 1h 41m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/162218a45ed343e0baebb80c2b5aa6c7 : SUCCESS in 57m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/341ce88f6b0f4b9e99ec1dccad086622 : SUCCESS in 1h 49m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/5c53e6dd1f904dbebde7a5fa0bde88c6 : SUCCESS in 1h 24m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/9045405f6c5748ba8875cd72ceed85c2 : SUCCESS in 1h 53m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/217f657dde1b47568c0de7f3c76e8c27 : SUCCESS in 1h 27m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7af7eb1a3bbb4e7d89ce6efa455067e9 : SUCCESS in 51m 52s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/11d5e3edc8e84f2889f9ac75abac4f0a : SUCCESS in 1h 02m 26s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/c1397e592361404fb7cc587335a08618 : SUCCESS in 54m 11s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7dd888753c054d2a8501d350bf3fdabd : SUCCESS in 35m 08s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cf8ffd8ca87747019eefa8c687b5d8dc : FAILURE in 24m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/8a96597e87ca4910b265d2bd5803a22a : SUCCESS in 2h 13m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ea4ef8f11dc645f09a540565d18a8bec : SUCCESS in 1h 17m 54s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/fdcf534755c646e0b35ab82732aabac1 : FAILURE in 33m 27s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c2be5e0f5b9b486188fc6cacf7c6aea3 : FAILURE in 26m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2de248263c45474ca65dabbc9f03d734 : TIMED_OUT in 3h 03m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64fcc7c0181249b89bbb5ced7c133b63 : SUCCESS in 36m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/92e44b33a6c84e28aec6a956c7b719ce : SUCCESS in 54m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/a610f8c602a949ab83442bd32c5ce02e : SUCCESS in 43m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/bdb42f518800465697f02e3c105df5c6 : SUCCESS in 53m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d4a67e3232074e2281a7890d89e9cbb6 : TIMED_OUT in 2h 34m 16s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/207d7ad514984c9ca85c6b701f6172fa : SUCCESS in 48m 50s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/dd74966790b9445884839a3ef3f4a256 : SUCCESS in 53m 52s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 504","accounts_in_message":[],"_revision_number":6},{"id":"982fbc8c5e3eb93547b23f7f9e9f86810f1b0268","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-05 00:15:09.000000000","message":"Abandoned\n\nthis was done by 0dbd292c8542202dbbc21c3f55fc30eb7d67df48","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"62fde90912f8bf67d21e2a08fe61fb002a38b9c2","revisions":{"5ce4ee15930a947c3a4b3868b0dbef1dd79155fb":{"kind":"REWORK","_number":1,"created":"2024-04-15 12:19:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/915853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/53/915853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/1"}}},"commit":{"parents":[{"commit":"2451531ac1e5d163ceb9f40261bcfd03cc516f61","subject":"Update branch regex to adapt to unmaintained branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2451531ac1e5d163ceb9f40261bcfd03cc516f61"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:18:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:18:52.000000000","tz":540},"subject":"Add stable/2024.1 jobs","message":"Add stable/2024.1 jobs\n\nChange-Id: I8d42a55c723a07c7c366bd06f406b6a2e55f768e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5ce4ee15930a947c3a4b3868b0dbef1dd79155fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5ce4ee15930a947c3a4b3868b0dbef1dd79155fb"}]},"branch":"refs/heads/master"},"88a2762a88e83b7b1a311da913dc08dad7d4e2eb":{"kind":"REWORK","_number":2,"created":"2024-04-15 12:22:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/915853/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/53/915853/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/2"}}},"commit":{"parents":[{"commit":"2451531ac1e5d163ceb9f40261bcfd03cc516f61","subject":"Update branch regex to adapt to unmaintained branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2451531ac1e5d163ceb9f40261bcfd03cc516f61"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:18:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:21:51.000000000","tz":540},"subject":"Add stable/2024.1 jobs","message":"Add stable/2024.1 jobs\n\nChange-Id: I8d42a55c723a07c7c366bd06f406b6a2e55f768e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/88a2762a88e83b7b1a311da913dc08dad7d4e2eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/88a2762a88e83b7b1a311da913dc08dad7d4e2eb"}]},"branch":"refs/heads/master"},"5a4cad57773d83a00f21df94dfed43d5e4bbcc58":{"kind":"NO_CHANGE","_number":3,"created":"2024-04-15 12:22:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/915853/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/53/915853/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/3"}}},"commit":{"parents":[{"commit":"493e79986167f151edd3f40fd050813e3443526e","subject":"Update branch regex to adapt to unmaintained branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/493e79986167f151edd3f40fd050813e3443526e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:18:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:22:23.000000000","tz":0},"subject":"Add stable/2024.1 jobs","message":"Add stable/2024.1 jobs\n\nChange-Id: I8d42a55c723a07c7c366bd06f406b6a2e55f768e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5a4cad57773d83a00f21df94dfed43d5e4bbcc58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5a4cad57773d83a00f21df94dfed43d5e4bbcc58"}]},"branch":"refs/heads/master","description":"Rebase"},"c0248a2c161b02689ab86d95e7788ce4d07ebae9":{"kind":"REWORK","_number":4,"created":"2024-04-16 07:02:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/915853/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/53/915853/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/4"}}},"commit":{"parents":[{"commit":"2451531ac1e5d163ceb9f40261bcfd03cc516f61","subject":"Update branch regex to adapt to unmaintained branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2451531ac1e5d163ceb9f40261bcfd03cc516f61"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:18:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-16 07:01:56.000000000","tz":540},"subject":"Add stable/2024.1 jobs","message":"Add stable/2024.1 jobs\n\nChange-Id: I8d42a55c723a07c7c366bd06f406b6a2e55f768e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c0248a2c161b02689ab86d95e7788ce4d07ebae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c0248a2c161b02689ab86d95e7788ce4d07ebae9"}]},"branch":"refs/heads/master"},"714712ac683387395ea0717a0c6142dcb1ebc7b1":{"kind":"NO_CHANGE","_number":5,"created":"2024-04-16 07:02:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/915853/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/53/915853/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/5"}}},"commit":{"parents":[{"commit":"493e79986167f151edd3f40fd050813e3443526e","subject":"Update branch regex to adapt to unmaintained branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/493e79986167f151edd3f40fd050813e3443526e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:18:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-16 07:02:33.000000000","tz":0},"subject":"Add stable/2024.1 jobs","message":"Add stable/2024.1 jobs\n\nChange-Id: I8d42a55c723a07c7c366bd06f406b6a2e55f768e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714712ac683387395ea0717a0c6142dcb1ebc7b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/714712ac683387395ea0717a0c6142dcb1ebc7b1"}]},"branch":"refs/heads/master","description":"Rebase"},"62fde90912f8bf67d21e2a08fe61fb002a38b9c2":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-06-05 15:08:04.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/53/915853/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/53/915853/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/53/915853/6"}}},"commit":{"parents":[{"commit":"4cf9d22686ca5bb2c141d80c581ed5d8edea6b85","subject":"Update branch regex to adapt to unmaintained branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4cf9d22686ca5bb2c141d80c581ed5d8edea6b85"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 12:18:52.000000000","tz":540},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-05 15:07:50.000000000","tz":120},"subject":"Add stable/2024.1 jobs","message":"Add stable/2024.1 jobs\n\nChange-Id: I8d42a55c723a07c7c366bd06f406b6a2e55f768e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/62fde90912f8bf67d21e2a08fe61fb002a38b9c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/62fde90912f8bf67d21e2a08fe61fb002a38b9c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
