)]}'
{"id":"openstack%2Foctavia-tempest-plugin~917076","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I13d977eba7c324df3036e0b0dc31a3943bd8549e","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"2024-1-stable-job","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-06-27 20:48:35.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-06-27 20:48:35.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-06-27 18:59:49.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I13d977eba7c324df3036e0b0dc31a3943bd8549e","subject":"Add stable/2024.1 jobs on master gate","status":"MERGED","created":"2024-04-26 00:13:53.000000000","updated":"2024-06-27 20:50:35.000000000","submitted":"2024-06-27 20:48:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917076-2024-1-stable-job","meta_rev_id":"5c79e7ef33cd4a3a5a41437413890f1dfcecd91e","_number":917076,"virtual_id_number":917076,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-27 20:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-05-06 15:43:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2024-06-27 18:59:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-30 10:25:58.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2024-06-27 18:59:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-26 03:18:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-30 10:25:58.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-05-02 04:14:41.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-05-06 15:43:32.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-05-22 21:12:17.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"5e19ee2dd2bda5fd7f0e660b614477f440649b67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-26 00:13:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6778a5fbd909ce4931495e01bb7bd2c0b46bf768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 03:18:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbbad9029b01401983e824680991fdfe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c9a20eee4c64f7d9a64e95876771afb : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b773431e9146cabe489ac762ebd44b : SUCCESS in 2m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/145f9d87e3bf48e78bab550b160207d7 : SUCCESS in 1h 33m 15s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/864f3fc1493847dcbde83d208e54825c : SUCCESS in 1h 30m 47s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/0e90ad24670b4da6821a000c57d2bed8 : SUCCESS in 1h 26m 26s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3ce6d7b1a898430899f88b12fd767e06 : SUCCESS in 1h 48m 56s\n- octavia-v2-dsvm-noop-api-stable-zed https://zuul.opendev.org/t/openstack/build/9e336a7fa3274202977ee26e1840a633 : SUCCESS in 2h 03m 34s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/debc878ff29145b5ba471ae6dc9f0df4 : SUCCESS in 1h 48m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f557e6cf63e34111b8ce7b8f03177d0d : SUCCESS in 1h 35m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cd45de2364b14ab69aa4468531fd28f2 : SUCCESS in 1h 25m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/495b5a10972744dc8f519a8437e2e7d7 : SUCCESS in 1h 39m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/859f39ddc02a4b0da25b0dbf21c27e3d : SUCCESS in 1h 20m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/bdc18541e946409d9a71cd7c8c78554b : SUCCESS in 1h 52m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/05901800c9ff4d5799efd917e7e97594 : SUCCESS in 1h 23m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/075aa4f94b9f454a9ac74ce2e961509b : SUCCESS in 1h 37m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-zed https://zuul.opendev.org/t/openstack/build/6f0a158403e247b29c1a092f46f7c75f : SUCCESS in 1h 08m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/499d499a91ac44bc97f54c7c78ae92d1 : SUCCESS in 37m 16s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e2492670ca0847129475100e16db5d9b : SUCCESS in 49m 09s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/ecd1e8739517470a90f55f8f65ffac22 : SUCCESS in 47m 34s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/c397bc2f794844fd8ccc3659676d2c10 : SUCCESS in 48m 41s\n- octavia-v2-dsvm-tls-barbican-stable-zed https://zuul.opendev.org/t/openstack/build/455247ecf89345d591833791460a281b : SUCCESS in 46m 58s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8c68b9f27b854005ad23356a9ca6dcf5 : FAILURE in 31m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/192c9ef3850649ab863fa033c2171013 : SUCCESS in 2h 20m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4e46d78960da4d2dbca2467a812787c4 : SUCCESS in 1h 31m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/44b83f5d876241e69662f8d90d400a75 : TIMED_OUT in 3h 03m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9648e1b07174951946e6fb35e68ef82 : SUCCESS in 36m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3c906130b4774b5da45ab43cc0e1c914 : SUCCESS in 32m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/26e0e763876c4ce486719d17b9e09872 : SUCCESS in 30m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/2459693eeb0b4294865e1fb0cf771e2e : SUCCESS in 30m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-zed https://zuul.opendev.org/t/openstack/build/3085dd4cc1294879bab213ca50c27aaf : SUCCESS in 40m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5ecf4cc4ee7f4d5f82eb311a8ec21e5c : SUCCESS in 2h 10m 17s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/93f32fd99ec44b09b770e6b4a980551a : SUCCESS in 46m 51s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/83362d1e2de544c7a95008ff09b68561 : SUCCESS in 56m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43f2c4a2c60b0929c51e8d4f71e80a1f133b507a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-30 10:25:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"11ad1194e89d8ce6e97fb12d731fb3a5333d7560","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-06 15:43:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7435f0393267a5ae524822e5e546ccf86a1aa9f8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-27 18:59:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90ed54d610284b94140e8041e1440c5f3f64d6c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 19:00:33.000000000","message":"Patch Set 1: -Verified\n\n(5 comments)\n\nStarting gate jobs.\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 504","accounts_in_message":[],"_revision_number":1},{"id":"6a2b35886e933123009779ff77812ecac56510e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 20:48:34.000000000","message":"Patch Set 1: Verified+2\n\n(5 comments)\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b17f1a7b7e441b998401e6305e743e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e62ec0f9881b4cc3a89fca7f3fb66a87 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c76f5366a7f416cadbe8ddbc0b5e4b4 : SUCCESS in 3m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b4862dac68b477489781347fcd2cfd9 : SUCCESS in 1h 31m 31s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/2d48aabdf0924b298184cac8b8bad161 : SUCCESS in 1h 18m 17s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/87e3414ec2e74cf7baa73c442946b246 : SUCCESS in 1h 00m 50s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/973c2b2facee4e3cacf6c78754b968ec : SUCCESS in 1h 46m 22s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0d39b362a7e944138898d882f632a32f : SUCCESS in 1h 25m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/606b21a09804497387dd5d5509f103f9 : SUCCESS in 1h 46m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/090695c7b9b948959f96323d16d0f5fe : SUCCESS in 1h 05m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/01d5797983ce4d98bc6d7f2e72b8ddf1 : SUCCESS in 1h 42m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8ed75cee93ab42f696b9b7ad920299e9 : SUCCESS in 1h 38m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/d53488eb0e0940738d0bfec9ab6e68ed : SUCCESS in 1h 18m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6c48edea37d24fd89b1d4be60bbf88cf : SUCCESS in 55m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/4dfb78250048455b99fbce17de89ba7e : SUCCESS in 1h 45m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/67a9f10f904f4a0dada2a9fe4e3db2e1 : SUCCESS in 1h 16m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/66260834c55448ebbf053a07f93c4d12 : SUCCESS in 40m 55s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/9ef046a25edb4ec58635e65cc0302309 : SUCCESS in 49m 38s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4c0e26bae8bc4861a7c140311a545e50 : SUCCESS in 46m 48s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/f5764560c4b54c0ca6dd5926cb8b951a : SUCCESS in 47m 38s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 504","accounts_in_message":[],"_revision_number":1},{"id":"e65b219dbcc67594e98ff78cc874818dfe321f4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 20:48:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5c79e7ef33cd4a3a5a41437413890f1dfcecd91e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 20:50:35.000000000","message":"Patch Set 1:\n\n(5 comments)\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54c0c45e8c9e49db83f42b87a6b43fb0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eba9a5aaf9d2438f824ca3add8b98152 : SUCCESS in 46s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 504","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4eb494d2790e3c1abddadc3c3cbd0b9673aa7ec","revisions":{"c4eb494d2790e3c1abddadc3c3cbd0b9673aa7ec":{"kind":"REWORK","_number":1,"created":"2024-04-26 00:13:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/76/917076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/76/917076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/76/917076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/76/917076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/76/917076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/76/917076/1"}}},"commit":{"parents":[{"commit":"a1a2bed19ffe387212364018c61f4a2ae1cff78b","subject":"Merge \"Add *.orig to .gitignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a1a2bed19ffe387212364018c61f4a2ae1cff78b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-25 23:53:00.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-26 00:13:52.000000000","tz":-420},"subject":"Add stable/2024.1 jobs on master gate","message":"Add stable/2024.1 jobs on master gate\n\nAs 2024.1 is released, we should add its job on master\ngate to keep branchless tempest plugins compatible to stable branch.\n\nAlso, fixing a few of the stable/2023.2 jobs to override-checkout 2023.2.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I13d977eba7c324df3036e0b0dc31a3943bd8549e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c4eb494d2790e3c1abddadc3c3cbd0b9673aa7ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/c4eb494d2790e3c1abddadc3c3cbd0b9673aa7ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
