)]}'
{"id":"openstack%2Foctavia-tempest-plugin~919789","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I9fc26307f64a2eccbbaf2d979e44fa668328565f","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-05-31 01:21:41.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-05-30 17:07:43.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-05-30 12:16:20.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I9fc26307f64a2eccbbaf2d979e44fa668328565f","subject":"Use only keystone roles in keystone RBAC tests","status":"MERGED","created":"2024-05-16 07:00:05.000000000","updated":"2024-05-31 01:22:42.000000000","submitted":"2024-05-31 01:21:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"919789","meta_rev_id":"c5bf81ca951c88d5aff963102d81426d3e70c37c","_number":919789,"virtual_id_number":919789,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-31 01:21:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-05-30 12:42:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-30 17:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-05-30 12:16:20.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-30 17:07:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-16 10:07:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-22 11:08:50.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"CC"},{"updated":"2024-05-29 13:36:53.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-05-30 00:52:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-30 12:42:02.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"7c95a62d7dad957c2673c829c197628a726feb23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-16 07:00:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46eaf2cabb14ba4ba45ff66b1f566a21a76cba23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 10:07:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a69c7f33f63041a3b82390ee8ab3f126\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520c8e5ffab24a6bad5d0da5a3ee205f : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c0d5a5238a8426db92f0103756f187f : SUCCESS in 2m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc3893ec0f6b4441907987c0f65f6c35 : SUCCESS in 1h 38m 02s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6a98aff1352e4830b1946b408b33a0cd : SUCCESS in 59m 21s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/88b0aa093c274d60a374fd9eac9af929 : SUCCESS in 1h 12m 00s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/936676857b684407a88afe3b1b61a2d0 : SUCCESS in 1h 40m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a3f29c37ab0342a48f40acbbe734982d : SUCCESS in 1h 47m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/39b2e6b1438440569c15f874216b5b51 : SUCCESS in 1h 40m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9ea0fbcebb8b476db92f13137932c187 : SUCCESS in 1h 30m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/115dc13c25054ec7b6b1109177ab11cd : SUCCESS in 1h 29m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/99bef92f4af04ae18a8088ae448f17b4 : SUCCESS in 1h 29m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/65955dffcb0f4aff838b9f0769278939 : SUCCESS in 1h 21m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/62f1fe66d3b84bfba46d94e90ed9d486 : SUCCESS in 54m 48s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/afdd0054420442c5a210a78eb0cf6f5b : SUCCESS in 52m 29s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/415c0d1f5df140eeaf3fe09e6150b418 : SUCCESS in 52m 03s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8d81d12a78e74d5e852f5b7890ea7dce : FAILURE in 31m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/857d6e38886b493993922618c02debcd : SUCCESS in 2h 24m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0ac5b332d94c40a4a26a2ac82b3d459f : SUCCESS in 1h 36m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c34f937367dc458b9c76927b4de8a45d : TIMED_OUT in 3h 03m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d85216ada6714e1791461d8f2de711eb : SUCCESS in 42m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4c301d65092c40b8ab627a0848e8e4e1 : SUCCESS in 47m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/8f78e7d5f9754b42a994934bda5ea780 : SUCCESS in 50m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b6806d821599457887da1bdbe124e248 : TIMED_OUT in 2h 33m 30s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2c41526e435f486c908df8bbefdb5126 : SUCCESS in 42m 46s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/449bd8f5563448a0be463f7077011e0d : SUCCESS in 51m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c025c3fa8eb6904d13d9024f0883b7ff73f37ae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-29 13:24:46.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"281be3f0a1bcba5898afd7b63dd9105c036d028b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-29 13:36:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5204c6ad48fbc6f255c6a1a8e39d3fc43c38e0ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-29 13:40:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"8cf3dcda70c980f82eb98507c4bdfc7c269d1780","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-29 13:41:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2418b93322d8c838d37751c5ddaaddb5a0f9538c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-29 13:44:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"add904712e990ec14276049a582c22bb24cf4eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 16:44:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd9bbe67581a48419fe81e078a365b56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c70b9ca860984dab97361b7f18b26862 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea28fcd01c6343428c9c151d02e5c70c : SUCCESS in 3m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de876e6eacf544a58b48cded0bd656c6 : SUCCESS in 1h 34m 26s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6e6085d1239d46319f050474b269abb5 : SUCCESS in 1h 00m 11s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7c53271a6dca4393a2eee57d1a95a7c2 : SUCCESS in 1h 14m 44s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/929a4cf272994e0aa89173c4905bcd59 : SUCCESS in 1h 39m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b8143bf1b4814c8f841fde543f7e1616 : TIMED_OUT in 2h 34m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5bbbac933a9049fcbc00dce69123f220 : SUCCESS in 1h 19m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/3ec5caf132f64af7844ef35cd4e63da4 : SUCCESS in 2h 08m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/839aae8c7e99461bb3ff3592ec18638b : SUCCESS in 1h 25m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/910d64d3fe8a4fb3b2ebfed9d648326f : SUCCESS in 1h 58m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/155738fba7284b5f899a82b32bf3e32b : SUCCESS in 1h 26m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0c418ed2dcb2407eb28878c25ab3749d : SUCCESS in 50m 03s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/a7dbec1f0c0148159cc63481c0ab4c9b : SUCCESS in 51m 38s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/e9a97ca4e36446ebb86dae5e4327272f : SUCCESS in 43m 44s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6aeb71d9f4624bd8bbe947673de8e502 : FAILURE in 37m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/6abc0299d56541f8aa43c8fb8160153c : SUCCESS in 2h 23m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2a91f8b91aed460c8477f64581e5cee1 : SUCCESS in 1h 37m 44s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/11145546074141b8a700cb67e55b263d : FAILURE in 26m 54s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/19dbc5d1be8d44c09e2331bcf95879df : FAILURE in 32m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c9e57a82ce243769de2da397096444b : TIMED_OUT in 3h 03m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98b95d006c6a482694f01f51e3f0d3fd : SUCCESS in 40m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/8677aa9046c34633a401376038fab53b : SUCCESS in 37m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/0dbccdf9c657450592e853119b33dd8b : SUCCESS in 37m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4b4e5cc4c3814fe9b38ca957ecd78fed : TIMED_OUT in 2h 33m 34s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2d3d4f7004b04a4fa272d926f11bcd00 : SUCCESS in 47m 06s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/02d2a8e36bf84e6aabac1921f37f22ba : SUCCESS in 56m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6dbfcd9a30893c4edb42f979126a3b36f4c91763","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-29 17:43:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"acc987a17411d990623bb0b90d34f73cef7b2ed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 20:48:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef75ad0288eb429b958fe25a015d4373\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/809e3e4111ee4b8bb530474076fbe730 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e1cfdf37cb4115972d5f4be46a38ca : SUCCESS in 2m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f7e8dc792b14546a659f05512b36fda : SUCCESS in 1h 29m 52s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/dca2efe324814454bf8086593e4ed86d : SUCCESS in 1h 17m 18s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/667ca1519c434bc3bfcae0668723a455 : SUCCESS in 1h 28m 25s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bc40bc3127504bd1aa48f191b057bbc5 : SUCCESS in 1h 56m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/19334fd17051424b95938b91b5bb3009 : SUCCESS in 2h 04m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6ef281519dbf4e3cae6e79b4bad5394b : SUCCESS in 58m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2d0e520a5dc54e1687fa3eb5135491d0 : SUCCESS in 1h 26m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/5ac59c01fbb74ff1944a396d1b00907b : SUCCESS in 1h 10m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/2b90604dc6c24c32a808fc72768f524a : SUCCESS in 1h 24m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/9aa22ef3961f494487f00fa71ee99edd : SUCCESS in 1h 16m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9627c7643e5e40a08c1630b910a9bbcd : SUCCESS in 52m 34s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/c5ce4309c4534feab560673c7e19fac5 : SUCCESS in 51m 53s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/b700f583e98f44399135518730a28b42 : SUCCESS in 43m 24s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/02984a3f671f45ceac135875f657fcea : FAILURE in 34m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/3ff4662364344b178f8e6ee1856e90ce : SUCCESS in 2h 06m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/68ec5bda2b674868834fd704c61f58fe : SUCCESS in 1h 24m 34s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/907cdd6d0a8441aeb95d576a362f638a : FAILURE in 24m 08s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/37302697421d4948920add11a0e89aa6 : FAILURE in 28m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ac18e28c6df43c580640987bd794f04 : TIMED_OUT in 3h 03m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/00e231476c414900954846afad14cd63 : SUCCESS in 33m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9447f555cfef48999577a76932d97604 : SUCCESS in 47m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/1ae7603bbfc0405dad61e0a062469362 : SUCCESS in 30m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/207bfc2a54d0447f9845d924c8f917d4 : SUCCESS in 2h 16m 58s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/31bad8360d1d4c009e74ec8caffecbf6 : SUCCESS in 45m 35s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/f8db1e5b84744f44b042a802eb4cc1c0 : RETRY_LIMIT in 2m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4aa496de50cb95b4c9b83cd8388990eef9f9907c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-30 00:52:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2e203bcb2030be0d3fb12c0510e43eb27dfb4eec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-30 07:21:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"937cedf388b52ba05401c3a41b5e6dee211d8ec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 10:27:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f42a0ce7dc44b1eb1e678fadccc4840\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce526be8e83f47f8baf71df189ca685d : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e6cac342ef041a19aa7f33b73a643fb : SUCCESS in 2m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd9ff536e7ab45c9812990b10f3ea701 : SUCCESS in 2m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f72f89394bb4309af4bac022a097bf3 : SUCCESS in 1h 22m 25s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b56d9f80420b42cd9c79ac7b9debd378 : SUCCESS in 1h 00m 58s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3dcd575a46414152984d535bf3da448e : SUCCESS in 1h 09m 41s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bbb0f5eb882e4b36bdb9c5988abea79d : SUCCESS in 1h 27m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e93b8c2119c04a7c97b707e8bfa30706 : SUCCESS in 1h 53m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/eb6811e832884118aea03c18687359c3 : SUCCESS in 1h 10m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b6ecf4a3fe45470fa25353e312d2d0b8 : SUCCESS in 1h 47m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b8ed66d3227d4557a1dd2a5a43b6b940 : SUCCESS in 1h 03m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/833557dae5c74783a424f9c5a348c283 : SUCCESS in 1h 55m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/1cb4516d874a4c9099f5fac833c9d6e4 : SUCCESS in 1h 04m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bf39507083fb4dc38b64ef880f6db08b : SUCCESS in 44m 54s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/92b32f0730f6445b9c4601fd59f4b46c : SUCCESS in 43m 41s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/d592a6dcb3ff4ca0bcb590a1bf9f30ea : SUCCESS in 35m 29s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/de57d8de03ac44ffaa09f64392700ca5 : FAILURE in 28m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/2d688e2357ae4516ab5c361307425a6e : SUCCESS in 2h 05m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/af902cb06eed4b6ebe7c7e6ecb1c563a : SUCCESS in 1h 25m 46s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/6e2a2ba747a34709a2b1f6aebd3dea6f : FAILURE in 24m 05s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a7266e14222248b2a1717c012a67f491 : FAILURE in 24m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e95c64369e8d4bfa9fd241373145c938 : TIMED_OUT in 3h 02m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b5eafd8a2a84dab816581b90d041027 : SUCCESS in 44m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/47aa87de6a4a454a861017b90c65f576 : SUCCESS in 30m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/b3e588a7f45146f0a756be6c53b9aeff : SUCCESS in 28m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7b1e2d505e2f45fd85f1ee16e95cc9d5 : SUCCESS in 2h 17m 55s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/d92ba97f3d4348b191589b6098be8723 : SUCCESS in 45m 41s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e429d3b2f2274cffb0ef1f8666797323 : SUCCESS in 1h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f40b4a6103deac07fab332c1e387eb9a186ca8c4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-30 12:16:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0d0e3222a7b9d19803059a2f44f3b0f724e1fbb7","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-30 12:42:02.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"85c6efe340d2926bf38f06acb2ed3f34c567bdaf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-30 17:07:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d4201a6658837edb226dc4fd0bcc1e8826b04b42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 17:08:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2992a05cbdee7720a44a80a970638358d27fbec2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 19:43:34.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b6a2b34170d46039dcb305b25700fad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81d0b69dc7445d2ac055c12d54f010e : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3c7063f4e9a47b190ebcd041d8bce6f : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d65f5a6c60643288e52ff056a1e8e57 : SUCCESS in 2m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d3324989f9c74f2f9b649865d85876a5 : SUCCESS in 1h 30m 53s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/af143ad22033465b95db04b34fa81b36 : SUCCESS in 1h 23m 45s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/96b0e8df5521445ca21193234d372221 : SUCCESS in 1h 46m 54s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0c11bbabb90c465aa2d345907f65d21c : SUCCESS in 1h 20m 25s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/eb39ee6f02824f84854b6eaba833ede7 : TIMED_OUT in 2h 33m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e29962d88cb74e1a9611ffe26e56c787 : SUCCESS in 1h 42m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/1408209415c847788c8849da28548b35 : SUCCESS in 2h 01m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/824f064ca0b843dab5e677ba58df2678 : SUCCESS in 1h 21m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7704c6ff72ba496cb7de521cb1da9b13 : SUCCESS in 1h 30m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/8c4941f5918042db8a642feb822de8d6 : SUCCESS in 1h 13m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4095925b0cb343b2a9b8d39b4b846a8b : SUCCESS in 41m 43s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/7161b6bcb4aa468fb0d62745839874e6 : SUCCESS in 46m 20s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/b4a4945f595442378b3de145522fcb2f : SUCCESS in 40m 12s","accounts_in_message":[],"_revision_number":4},{"id":"a5ba4b9016881cc87b3f337f8a47a65e6793c937","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2024-05-30 20:27:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e3bd09a9f9b4347a5340c75d4d867076c498de19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 23:32:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a467f99263f4f7aafd9ff62384d73ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e854b5c2112341b9b80aad5a1e55f7f0 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cf51d17a87842d2b232424857007733 : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f96a9a41eb4942c7868cf6888abbee4d : SUCCESS in 3m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24a8c871024f4b9d983f8131dc8c6757 : SUCCESS in 1h 27m 21s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/a7d66b9a32764fd59aae289d7c332fda : SUCCESS in 1h 11m 21s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/bdb91f9e34784e538ff18de81fbc35cc : SUCCESS in 1h 17m 21s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8015a8aa947345dca6c4957994c11c2d : SUCCESS in 1h 05m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/888e70f4543f46208a69510d7d0b98e0 : SUCCESS in 1h 58m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/481372a3f3044d6880d95f0c98e16cf1 : SUCCESS in 1h 05m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/216a6c62a468483e8b8fc85f52df5d55 : SUCCESS in 1h 22m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/035913c1109f4ddaaf950acd5c81d0d4 : SUCCESS in 1h 07m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/e8873b9393b94d7db3b9cd148796144d : SUCCESS in 1h 30m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/a50c9452270f48c098c55efb81923851 : SUCCESS in 1h 05m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cdd0a9e848294f8d94774eaddcaa8c26 : SUCCESS in 47m 02s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/f014d822d12d4f298cf4ec5950bf811f : SUCCESS in 44m 35s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/6f37b1012ad740adb9b91bdb24f2f11f : SUCCESS in 35m 59s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cc781c43a33b483f821721c304d79b2b : FAILURE in 33m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/f5ca7c4b5fa54670b356633fe325f980 : SUCCESS in 2h 13m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/92e41d89b07847069200f1a8d7bee5bc : SUCCESS in 1h 14m 35s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/80414bdd7b7947a7b7d3651a4a49e5c2 : FAILURE in 25m 47s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d9a3645b5b0e4e368fe8af84adb2d230 : FAILURE in 27m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/becfc26305244ca6b2eb31f614c1b1fc : TIMED_OUT in 3h 03m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1cb826a1d8d24cc884ea8d35cbe29ab0 : SUCCESS in 45m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/8ed32c2bf5104897b7a305c48bb0a883 : SUCCESS in 42m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/38de4630a38742a7a5acc75c45d1b22f : SUCCESS in 41m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fceeb50d2b3346499b80eb8d779713dd : TIMED_OUT in 2h 32m 36s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/43e336d39a9c45628f583d6c9d3bc6c0 : SUCCESS in 52m 05s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/6c947844e2e54c00a9b7e756e961d6db : SUCCESS in 55m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b611868f1680f63d86c9d9d40a0a4c23b0bb974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 23:32:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"be5571869fcf0a99b97e449eed6e5849c6c3c30a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 01:21:41.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b46a69428e24cba8e0bc36bc054aad6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18faa51290e94c818a0587e5c0413761 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4db7d29ed6448cfa94e8405d035312b : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d7f9cc49fb6479faf49d218507180b2 : SUCCESS in 4m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/844bea03596945afbb7197fa6f3ac456 : SUCCESS in 1h 21m 58s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/3958fbd0635040c7b769cb1ceb1dac8e : SUCCESS in 1h 15m 54s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/a35539a6e8734ad3b2f65856ee4b57e3 : SUCCESS in 1h 24m 59s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/899ab834491e4dd8a5a0dc8b7e523db1 : SUCCESS in 1h 37m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/545a0db50086453e86b7a6e6465c1b12 : SUCCESS in 1h 48m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ed99ef6331e84bec8b4561415684d163 : SUCCESS in 1h 08m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e7d028994ac3450886b6acf483365701 : SUCCESS in 1h 21m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/70411411aa5f4f3f9b3d7b06aaf2dff4 : SUCCESS in 1h 02m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/399d2fd9451b4f20bb33502a3117fe3e : SUCCESS in 1h 29m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/f468b1d749344025bfbcb5e037065f6a : SUCCESS in 1h 15m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ec542151f7c74087853cbcef5e4960f0 : SUCCESS in 35m 38s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e1ca4fe1aba14f188b3aaf7643566bd6 : SUCCESS in 45m 25s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/fa60be9f59414090be7572d9aa690352 : SUCCESS in 51m 21s","accounts_in_message":[],"_revision_number":4},{"id":"b6ba050d499691be655cc1a2b1fc78c01093fa39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 01:21:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c5bf81ca951c88d5aff963102d81426d3e70c37c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 01:22:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b977709e21344ace8d71744bcda7adde\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2412ac453f65439197402fa487986341 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf453b0f34b248e8b10498f0126cc24a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cba3b22f636e02fc07a91317149daf58aa651f28","revisions":{"4b90424b27b66621f91ca8f5284758386c5b57da":{"kind":"REWORK","_number":1,"created":"2024-05-16 07:00:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/919789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/89/919789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/1"}}},"commit":{"parents":[{"commit":"1f42112bd0e8cb7faa50b58d02060f61707e8f3b","subject":"Merge \"Add backup member tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1f42112bd0e8cb7faa50b58d02060f61707e8f3b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-16 06:57:08.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-16 06:57:14.000000000","tz":-240},"subject":"Use only keystone roles in keystone RBAC tests","message":"Use only keystone roles in keystone RBAC tests\n\n- don\u0027t use the load-balancer_* roles, they may not exist (and they don\u0027t\n  do anything) in this configuration\n- create a temporary user with a temporary non-member role\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/919770\n\nChange-Id: I9fc26307f64a2eccbbaf2d979e44fa668328565f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4b90424b27b66621f91ca8f5284758386c5b57da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4b90424b27b66621f91ca8f5284758386c5b57da"}]},"branch":"refs/heads/master"},"4c7b6b388947fe0b8de5be7f9df6a28f23fc4f73":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-05-29 13:24:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/919789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/89/919789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/2"}}},"commit":{"parents":[{"commit":"1f42112bd0e8cb7faa50b58d02060f61707e8f3b","subject":"Merge \"Add backup member tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1f42112bd0e8cb7faa50b58d02060f61707e8f3b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-16 06:57:08.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-29 13:24:31.000000000","tz":120},"subject":"Use only keystone roles in keystone RBAC tests","message":"Use only keystone roles in keystone RBAC tests\n\n- don\u0027t use the load-balancer_* roles, they may not exist (and they don\u0027t\n  do anything) in this configuration\n- create a temporary user with a temporary non-member role\n\nChange-Id: I9fc26307f64a2eccbbaf2d979e44fa668328565f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4c7b6b388947fe0b8de5be7f9df6a28f23fc4f73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4c7b6b388947fe0b8de5be7f9df6a28f23fc4f73"}]},"branch":"refs/heads/master"},"4bc6c489d256297f3c9682c867c2c430bebd6a0c":{"kind":"REWORK","_number":3,"created":"2024-05-29 13:40:30.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/919789/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/89/919789/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/3"}}},"commit":{"parents":[{"commit":"1f42112bd0e8cb7faa50b58d02060f61707e8f3b","subject":"Merge \"Add backup member tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1f42112bd0e8cb7faa50b58d02060f61707e8f3b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-16 06:57:08.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-29 13:40:15.000000000","tz":120},"subject":"Use only keystone roles in keystone RBAC tests","message":"Use only keystone roles in keystone RBAC tests\n\n- don\u0027t use the load-balancer_* roles, they may not exist (and they don\u0027t\n  do anything) in this configuration\n- create a temporary user with a temporary non-member role\n\nChange-Id: I9fc26307f64a2eccbbaf2d979e44fa668328565f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4bc6c489d256297f3c9682c867c2c430bebd6a0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4bc6c489d256297f3c9682c867c2c430bebd6a0c"}]},"branch":"refs/heads/master"},"cba3b22f636e02fc07a91317149daf58aa651f28":{"kind":"REWORK","_number":4,"created":"2024-05-30 07:21:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/919789/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/89/919789/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/89/919789/4"}}},"commit":{"parents":[{"commit":"1f42112bd0e8cb7faa50b58d02060f61707e8f3b","subject":"Merge \"Add backup member tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1f42112bd0e8cb7faa50b58d02060f61707e8f3b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-16 06:57:08.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-30 07:16:52.000000000","tz":120},"subject":"Use only keystone roles in keystone RBAC tests","message":"Use only keystone roles in keystone RBAC tests\n\n- don\u0027t use the load-balancer_* roles, they may not exist (and they don\u0027t\n  do anything) in this configuration\n- create a temporary user with a temporary non-member role\n\nChange-Id: I9fc26307f64a2eccbbaf2d979e44fa668328565f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cba3b22f636e02fc07a91317149daf58aa651f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cba3b22f636e02fc07a91317149daf58aa651f28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
