)]}'
{"id":"openstack%2Foctavia-tempest-plugin~926456","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Ife8a314033a2c9d7c692516fd21930ef1623af32","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-08-30 18:21:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ife8a314033a2c9d7c692516fd21930ef1623af32","subject":"DNM: testing RBAC new defaults","status":"ABANDONED","created":"2024-08-16 02:07:33.000000000","updated":"2024-08-30 18:21:53.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"08bb496aa1a734bbe060c275df163e8e3decd22a","_number":926456,"virtual_id_number":926456,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-22 22:14:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-16 05:01:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1780addee582d1e849386849711221191689f7fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-16 02:07:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4994ea42fa873eb2449feb740253fd5999c635fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 05:01:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/424339528fb04048bfedeb0a396876e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c77b9453e444d49d5aac1286226d5f : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a4c61155d7466187363e8994e44ad9 : SUCCESS in 3m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c6a3f2c5dd4741ff81879f3f5ddb11da : SUCCESS in 1h 44m 04s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/297fd7ef7cd94ab59a2d437b611c1c14 : SUCCESS in 1h 44m 43s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/1ea6334328174a5ca3059ae1b673f8f3 : SUCCESS in 1h 09m 02s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/36ee8971481849d9b7b27757bc2efc03 : SUCCESS in 1h 34m 28s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/aa9c8b3907ec4c5b91bf0f435f54110f : SUCCESS in 1h 36m 11s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/601a8175a89245d4abe7aeb9f94e6ade : FAILURE in 27m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5e37e3b702664f7ca9bc4d06d82d0895 : FAILURE in 29m 13s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/c13ca5d0f6f040a98fdfb29e35b0ba74 : SUCCESS in 2h 22m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/08ce7ac536274c94ba5ef08d832abdbb : SUCCESS in 1h 33m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/29b8a32f656242209c7777f634b39fbc : SUCCESS in 1h 48m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/1c8b5d8b742f4adfb57f8250cff9ee6a : SUCCESS in 1h 17m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/306a33ef7e64408bb65f0cc593ef5e10 : SUCCESS in 2h 04m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/cc58451be6524311827efc4161b35281 : SUCCESS in 1h 21m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d67e04b7f0804d9f852bbe6ba77e7729 : FAILURE in 32m 23s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/c7c8d8fcda7842f886efc89635a9320f : SUCCESS in 57m 17s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/22e35b1ab4424a02b0dac2d2b96f7b2f : SUCCESS in 51m 53s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/66badcfbc95c46c6ae11af25dd1f4099 : SUCCESS in 54m 47s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4cb1c5b55a2c4ddf80fe0c923212aa5a : FAILURE in 27m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/ab52b00d048941c5b1ce7c98af9ad52b : FAILURE in 33m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/21ddc09d11fe4b8fbd32d859085230db : FAILURE in 34m 16s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/98bf37336eab461ca8e288943fcd50c3 : FAILURE in 37m 31s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d809ccaf6296453ca0c5e7968923611c : FAILURE in 38m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/79575e3da64a431baf8c4cea012b52b4 : FAILURE in 28m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f15c60832b0c4d11bcc0b334e247a7d3 : FAILURE in 29m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/59432ec9fd784d25baed159b83379033 : SUCCESS in 48m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/bddcb8e8f7804865b20e5736724b3e0c : SUCCESS in 46m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/d636019acad04932b5d254b8a359c2dc : SUCCESS in 49m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/53294fe9340a4711ad29ff00d555d296 : FAILURE in 28m 36s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/1cb23555548a4bcebe225c63064cca31 : SUCCESS in 58m 23s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/bb220fa04a924c1eabd71ec440a0c527 : SUCCESS in 1h 02m 48s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 504","accounts_in_message":[],"_revision_number":1},{"id":"c7c86af98738c94cf92bbcfb9d6107fd44f52cd6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-21 03:35:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7dc1f1e5c899b98f6b70c39f2834ee925d26f94c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 05:51:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ea76c8a48e54a089bc93be83d096895\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaa5c88ca3694c23a9e542609a297aee : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af27f1e903d44e380a28e91bb75a805 : SUCCESS in 3m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/85799142999345cfb5909e2d9633b040 : SUCCESS in 1h 23m 00s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1307df48a61a4d43911073b2c53b69c2 : SUCCESS in 1h 35m 31s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b3e0c5a5254c49bab22cc47a1002a151 : SUCCESS in 1h 40m 17s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ae809edd69254334b2dbf175717786d5 : SUCCESS in 1h 07m 28s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0aa2edd1eee94168bc805e1b50e3295b : SUCCESS in 1h 26m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8bc4ce5c3c2247ba9d3671da29c4d0d1 : FAILURE in 30m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1c8c984daa6545c18b3d024ae2cc8685 : FAILURE in 29m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/0a84df07c93d41b38d65fd3e5b649701 : SUCCESS in 2h 13m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/7f1fc7e4d8ae4a0483122152f99537c3 : SUCCESS in 1h 19m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/8329d3cfde0e4ef185bb6a4261dc98f3 : SUCCESS in 1h 57m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/cf0373a9c69345dcb2270e474d3bc6da : SUCCESS in 1h 16m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/95b42e1e177240fe9f3982fecd16adcd : SUCCESS in 1h 21m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/04a6c694eeb54621a06796ee9afdbaa6 : SUCCESS in 1h 06m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2f6410f239084d058f21f2a36124b42c : FAILURE in 18m 40s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3aa3438db2324fc0bfa2b2d76f23cb4c : SUCCESS in 36m 48s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4ff1b709deda495686ec566e8e35570e : SUCCESS in 35m 26s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/6097dc133b594f7e9bede81b7e7d5a57 : SUCCESS in 53m 43s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/87ed5ff72086468c903c4dc84bbac63c : FAILURE in 18m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/852352f37b314857a1f8bc175a46ce78 : FAILURE in 38m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bc9a8f0d9e0045c9b3daede28a443f86 : FAILURE in 40m 26s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/f6eb1b2d406c4b4ea00c83e08f277efe : FAILURE in 37m 23s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/192868659b1f4f35a13455ef2d6784da : FAILURE in 37m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/17327e958fee43479b5272444bb03c27 : FAILURE in 29m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f59b070328a4bffb484cd3e1a8c12df : FAILURE in 18m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/af098adc896144089754f45c58fd8318 : SUCCESS in 49m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/7ce5f76167da4539bb29aaff8f92a20c : SUCCESS in 28m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/771fbc6bffdb4000a266b483f39f086a : SUCCESS in 29m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8490e598da3a4ff29b5f3c54ee09c38b : FAILURE in 20m 14s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/a0d4b28ddb32489da801f1ecc4f4245b : SUCCESS in 44m 30s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/abd9baa6fc214db68065a5d984ad121e : SUCCESS in 51m 13s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 504","accounts_in_message":[],"_revision_number":1},{"id":"b03a086dd3cd795dd8bc0226d443f2d2b069afd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-22 02:32:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c05732695968dc8779895e1057d6943354737918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 02:36:06.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/925625 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"1482bb17edf3f602b1b09d52e8f9398c379edc58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-22 20:08:31.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"deb3f2607af8ce58cf86b4075c843536d151ccc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 22:14:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e998491f657c436691d21b7699aad9ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1652fbf46bb4aab91e0ec9ac6af53b8 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4478a37e8c9464397d559630fe94bfd : SUCCESS in 3m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e4a04c205baf4dadaa357fa26f62c4a2 : SUCCESS in 1h 41m 08s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/f09e74b164fe4abb86a55fbe3a8d9030 : SUCCESS in 1h 11m 38s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/aa8f9aa5ebcd41e09a396b49d7d4376a : SUCCESS in 1h 36m 44s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/5d23f7a7fcc54ce5b26ac7348a69cf95 : SUCCESS in 1h 10m 46s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b79664e2e5954381bc7602b1ac64e569 : SUCCESS in 1h 33m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5e24ee0def2043b587694eb292dc4147 : FAILURE in 26m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8e955f9f707d4a14800f321efe7f02a1 : FAILURE in 17m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/02a083a1822d4494ad3bf6f342e4ee97 : SUCCESS in 1h 42m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/586c70f3399246ad9b609570e6629c54 : SUCCESS in 1h 06m 55s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/970cdc200e954430bd3678eb4edcc1f7 : SUCCESS in 1h 26m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2ac9064b27f247a1ad22453940c9639c : SUCCESS in 1h 29m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/a8987a7796a644a894e6b7668ea741c2 : SUCCESS in 1h 48m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/068d690092b94ffeb9d56d3486062927 : SUCCESS in 1h 15m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f84824732b5c4caf8edb5040cba274bd : FAILURE in 26m 16s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1302d188890a449280241d09b88aa2d1 : SUCCESS in 57m 15s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/610ee01ca03e40eebf3a057ba367661f : SUCCESS in 48m 19s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/11b8a1d851794abbb829cef04fbe9159 : SUCCESS in 55m 17s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/512177971b8f40db97da1b5bd89fd5e1 : FAILURE in 23m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/0f71f3d01ef847b5b3be104e0e487e92 : FAILURE in 34m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/531bf7b8e1154aceb940876d7f78c6a2 : FAILURE in 35m 45s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/6aa599b9297e4b1eaf1d4c69e4525460 : FAILURE in 34m 07s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b92994d941a24853b985b04beb9fa90d : FAILURE in 32m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6b5fb831a9944717910ebc96015c53d9 : FAILURE in 25m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78703207329849b29b8eac032fdb4bea : FAILURE in 21m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/ee60386f1e4e4c7d947a2b9b11094a4b : SUCCESS in 35m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b0b7cfe428964ffe9bc332721f05593f : SUCCESS in 27m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/e81c954b27d1462cbe64e5d9f90eb74e : SUCCESS in 33m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/205f95a87b5b46c9a4eead0e0bd67cd9 : FAILURE in 22m 20s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/48a6872c2a634ad2831916ca394a62d3 : SUCCESS in 45m 45s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/52c513b9d69543a9b358c5dc374598ac : SUCCESS in 46m 07s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 504","accounts_in_message":[],"_revision_number":3},{"id":"08bb496aa1a734bbe060c275df163e8e3decd22a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 18:21:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"735b8b25b8dd4b1610f93fab66dbab14339ea140","revisions":{"b2d1b3aa6d5bddf0b314ff6a092aec7df2e8ca5e":{"kind":"REWORK","_number":1,"created":"2024-08-16 02:07:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/56/926456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/56/926456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/56/926456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/56/926456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/56/926456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/56/926456/1"}}},"commit":{"parents":[{"commit":"b85b33e2b9ce12e659c66ee8f8449180d7a9d532","subject":"Merge \"Updates of jobs config for 2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b85b33e2b9ce12e659c66ee8f8449180d7a9d532"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-16 02:06:27.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-16 02:07:32.000000000","tz":-420},"subject":"DNM: testing RBAC new defaults","message":"DNM: testing RBAC new defaults\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/925625\nChange-Id: Ife8a314033a2c9d7c692516fd21930ef1623af32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b2d1b3aa6d5bddf0b314ff6a092aec7df2e8ca5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b2d1b3aa6d5bddf0b314ff6a092aec7df2e8ca5e"}]},"branch":"refs/heads/master"},"f1154846d41c9bec982a74a112763a5780c7d758":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-22 02:32:55.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/56/926456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/56/926456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/56/926456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/56/926456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/56/926456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/56/926456/2"}}},"commit":{"parents":[{"commit":"1aaaac6fc0d7a479862b98eb654fe23036a65580","subject":"Remove system scope token usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1aaaac6fc0d7a479862b98eb654fe23036a65580"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-16 02:06:27.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-22 02:32:55.000000000","tz":0},"subject":"DNM: testing RBAC new defaults","message":"DNM: testing RBAC new defaults\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/925625\nChange-Id: Ife8a314033a2c9d7c692516fd21930ef1623af32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f1154846d41c9bec982a74a112763a5780c7d758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f1154846d41c9bec982a74a112763a5780c7d758"}]},"branch":"refs/heads/master","description":"Rebase"},"735b8b25b8dd4b1610f93fab66dbab14339ea140":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-22 20:08:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/56/926456/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/56/926456/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/56/926456/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/56/926456/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/56/926456/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/56/926456/3"}}},"commit":{"parents":[{"commit":"d4352b26a93c3fe939e4dcd4b5bdbd33110bc967","subject":"Remove system scope token usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d4352b26a93c3fe939e4dcd4b5bdbd33110bc967"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-16 02:06:27.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-22 20:08:31.000000000","tz":0},"subject":"DNM: testing RBAC new defaults","message":"DNM: testing RBAC new defaults\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/925625\nChange-Id: Ife8a314033a2c9d7c692516fd21930ef1623af32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/735b8b25b8dd4b1610f93fab66dbab14339ea140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/735b8b25b8dd4b1610f93fab66dbab14339ea140"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
