)]}'
{"id":"openstack%2Foctavia-tempest-plugin~932061","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I9de892ee2de89675703c52185c11acd47a424606","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"task/30837","attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-10-28 13:45:49.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9de892ee2de89675703c52185c11acd47a424606","subject":"DNM: test with different failover_threshold","status":"NEW","created":"2024-10-10 12:12:43.000000000","updated":"2024-10-31 16:02:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ac7b9d04149d1d94d70d440d964fcb03948e7d50","_number":932061,"virtual_id_number":932061,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-31 16:02:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-10 13:00:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"798fefa7877087bc4e040d2796a4413de93a938e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-10 12:12:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24df7f637ebf239f7a2191a87ba701e27819fda9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-10 12:15:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5903252b9c54134c93a003b2fdc1675dfe12ecc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-10 12:21:34.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"54fe0327de377c338f8e1351df360f29e52f91c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 13:00:47.000000000","message":"Patch Set 3: Verified+1\n\n(5 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aabd2fe34014a68a5df86a4cf9ca64a\n\n- octavia-v2-failover-cb-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50e7f0f245b945dea849cdda75987c60 : FAILURE in 34m 55s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 494","accounts_in_message":[],"_revision_number":3},{"id":"4cb843d338e7e26eeacb56c488365c0d0287d0ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-10 14:44:51.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2882182e257e542501ca62541dea3690b42eca96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 15:36:58.000000000","message":"Patch Set 4: Verified+1\n\n(5 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0e7b30a79f8436fb2d587ffce6b8a5a\n\n- octavia-v2-failover-cb-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66da57d9a1bf495d96e4387876a4068a : FAILURE in 48m 58s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 494","accounts_in_message":[],"_revision_number":4},{"id":"2589cf6498a0fb414acc9f3a985b9adab60be2bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-10 16:12:16.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1fd739ef4866c48aecf2d330a751da6350bf8509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 17:08:12.000000000","message":"Patch Set 5: Verified+1\n\n(5 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e035d25669d1490bab8e22998af80401\n\n- octavia-v2-failover-cb-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e80773c6d6c5484592df319b0270e81c : FAILURE in 54m 01s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 494","accounts_in_message":[],"_revision_number":5},{"id":"a5cfe5f1836ce9b8d5610ea56bf970513051f9c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-14 12:06:35.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"7124c055df2dca970f968c2dd0befe2ce879f4cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 12:58:04.000000000","message":"Patch Set 6: Verified+1\n\n(5 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bf5a457852d4edc83cdddb18241cf6a\n\n- octavia-v2-failover-cb-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2765847fd779410289c939133f1fd703 : FAILURE in 50m 05s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 494","accounts_in_message":[],"_revision_number":6},{"id":"6c9ac9ad5d15d0eda926a01097ceeaae6699a53f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-14 13:58:19.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"1186b69c15959f14befcaf23293051c445245c1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 14:52:03.000000000","message":"Patch Set 7: Verified+1\n\n(5 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecadfbffecc6411ba04d8b5ec0b93999\n\n- octavia-v2-failover-cb-dsvm-scenario https://zuul.opendev.org/t/openstack/build/85944a9f01f044bc867ead10f7f17fb0 : FAILURE in 51m 01s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 494","accounts_in_message":[],"_revision_number":7},{"id":"fd825e163165c4114b5c3a86c9f2f8238b7d0ef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-15 07:43:40.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"f790506b0c6f68cd92089d1b1e92f396085829ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 08:36:37.000000000","message":"Patch Set 8: Verified+1\n\n(5 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dedc02f712764ebcb609f7e60164cf68\n\n- octavia-v2-failover-cb-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45b10019ff5c4ef4840d66a9dde81b99 : FAILURE in 51m 20s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/octavia-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"octavia-v2-dsvm-noop-api\" job stanza:\n\n  job:\n      name: octavia-v2-dsvm-noop-api\n      parent: octavia-dsvm-noop-base\n      timeout: 10800\n      vars:\n        devstack_local_conf:\n          post-config:\n            $OCTAVIA_CONF:\n              api_settings:\n  ...\n\n  in \"openstack/octavia-tempest-plugin/zuul.d/jobs.yaml@master\", line 494","accounts_in_message":[],"_revision_number":8},{"id":"60019529ef9bae7614caf9beeb7f605ee7648d2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 13:15:33.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"9200f5940dd37e012996458b6494669c183f4cb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 13:45:49.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20eb09a3d8a84a9481859fc8f656ec45\n\n- octavia-v2-failover-cb-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5d957f17e1045d9a7f37a05da01fea9 : FAILURE in 27m 29s","accounts_in_message":[],"_revision_number":9},{"id":"a8bc0715666c394cda788a1a2ecc2e1580b998a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 14:16:32.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"fdd8f0e3a808f5617be41449fd84779327898162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 14:51:27.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2f442fca8174049a1808f0fba30de73\n\n- octavia-v2-failover-cb-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f112d144510e41b8a81f883c5d318073 : FAILURE in 33m 25s","accounts_in_message":[],"_revision_number":10},{"id":"979fcb48d25d6425fccc3529a1725dba907e8231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-31 15:33:20.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"ac7b9d04149d1d94d70d440d964fcb03948e7d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 16:02:28.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef7fc6eb325744848f73cbe488ae8abd\n\n- octavia-v2-failover-cb-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7292e816111848059ee696914e66fad5 : FAILURE in 27m 22s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"1794f9a1d1609575fdd098fc6367a1c4066c5737","revisions":{"b8fa35c6e6dba789e7042ebb99ee0cc03057d319":{"kind":"REWORK","_number":1,"created":"2024-10-10 12:12:43.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/61/932061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/932061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/1"}}},"commit":{"parents":[{"commit":"2dd8659366995c8ab6a3bdda7d8d5c961aa4f43a","subject":"Add scenario tests for failover circuit breaker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2dd8659366995c8ab6a3bdda7d8d5c961aa4f43a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:12:16.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:12:16.000000000","tz":120},"subject":"DNM: test with different failover_threshold","message":"DNM: test with different failover_threshold\n\nChange-Id: I9de892ee2de89675703c52185c11acd47a424606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b8fa35c6e6dba789e7042ebb99ee0cc03057d319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b8fa35c6e6dba789e7042ebb99ee0cc03057d319"}]},"branch":"refs/heads/master"},"4f00a93dc88acc3d4525529213ba51072b422654":{"kind":"REWORK","_number":2,"created":"2024-10-10 12:15:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/61/932061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/932061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/2"}}},"commit":{"parents":[{"commit":"a544c1718f1f6c360ac40319a02b64e213bc8557","subject":"Add scenario tests for failover circuit breaker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/a544c1718f1f6c360ac40319a02b64e213bc8557"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:12:16.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:15:41.000000000","tz":120},"subject":"DNM: test with different failover_threshold","message":"DNM: test with different failover_threshold\n\nChange-Id: I9de892ee2de89675703c52185c11acd47a424606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4f00a93dc88acc3d4525529213ba51072b422654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4f00a93dc88acc3d4525529213ba51072b422654"}]},"branch":"refs/heads/master"},"5100d1a255780a8aa80ed2619c66eb49f5a59ed8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-10 12:21:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/61/932061/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/932061/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/3"}}},"commit":{"parents":[{"commit":"29140480f9d2ff67643d792398e9294258b4a752","subject":"Add scenario tests for failover circuit breaker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/29140480f9d2ff67643d792398e9294258b4a752"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:12:16.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:21:23.000000000","tz":120},"subject":"DNM: test with different failover_threshold","message":"DNM: test with different failover_threshold\n\nChange-Id: I9de892ee2de89675703c52185c11acd47a424606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5100d1a255780a8aa80ed2619c66eb49f5a59ed8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5100d1a255780a8aa80ed2619c66eb49f5a59ed8"}]},"branch":"refs/heads/master"},"fad0f56f806ed70babd65882c8a309e26402b89d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-10 14:44:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/61/932061/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/932061/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/4"}}},"commit":{"parents":[{"commit":"26e855cd9bece755280ddfdecd99a5f77afa9c47","subject":"Add scenario tests for failover circuit breaker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/26e855cd9bece755280ddfdecd99a5f77afa9c47"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:12:16.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 14:44:41.000000000","tz":120},"subject":"DNM: test with different failover_threshold","message":"DNM: test with different failover_threshold\n\nChange-Id: I9de892ee2de89675703c52185c11acd47a424606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fad0f56f806ed70babd65882c8a309e26402b89d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fad0f56f806ed70babd65882c8a309e26402b89d"}]},"branch":"refs/heads/master"},"7878e5962278ec9b4dc78b75cd4ebba3153ae62a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-10-10 16:12:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/61/932061/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/932061/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/5"}}},"commit":{"parents":[{"commit":"98a85047f08c776dd042c84d7116b6041fefb815","subject":"Add scenario tests for failover circuit breaker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/98a85047f08c776dd042c84d7116b6041fefb815"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:12:16.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 16:12:03.000000000","tz":120},"subject":"DNM: test with different failover_threshold","message":"DNM: test with different failover_threshold\n\nChange-Id: I9de892ee2de89675703c52185c11acd47a424606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7878e5962278ec9b4dc78b75cd4ebba3153ae62a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7878e5962278ec9b4dc78b75cd4ebba3153ae62a"}]},"branch":"refs/heads/master"},"5adc0d4f7141911eaa0ecc35c5f127dd0569463b":{"kind":"REWORK","_number":6,"created":"2024-10-14 12:06:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/61/932061/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/932061/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/6"}}},"commit":{"parents":[{"commit":"98a85047f08c776dd042c84d7116b6041fefb815","subject":"Add scenario tests for failover circuit breaker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/98a85047f08c776dd042c84d7116b6041fefb815"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:12:16.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-14 12:06:17.000000000","tz":120},"subject":"DNM: test with different failover_threshold","message":"DNM: test with different failover_threshold\n\nChange-Id: I9de892ee2de89675703c52185c11acd47a424606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5adc0d4f7141911eaa0ecc35c5f127dd0569463b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5adc0d4f7141911eaa0ecc35c5f127dd0569463b"}]},"branch":"refs/heads/master"},"8226bb01d6b037faa72cd30c413999a3aa398830":{"kind":"REWORK","_number":7,"created":"2024-10-14 13:58:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/61/932061/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/932061/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/7"}}},"commit":{"parents":[{"commit":"3bc1a0b5c95f03da821c83a71c76336d6e4a4fcb","subject":"Add scenario tests for failover circuit breaker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3bc1a0b5c95f03da821c83a71c76336d6e4a4fcb"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:12:16.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-14 13:58:10.000000000","tz":120},"subject":"DNM: test with different failover_threshold","message":"DNM: test with different failover_threshold\n\nChange-Id: I9de892ee2de89675703c52185c11acd47a424606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8226bb01d6b037faa72cd30c413999a3aa398830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8226bb01d6b037faa72cd30c413999a3aa398830"}]},"branch":"refs/heads/master"},"e9d0dd429437e16c98bf20ac5df34796c042994e":{"kind":"REWORK","_number":8,"created":"2024-10-15 07:43:40.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/61/932061/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/932061/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/8"}}},"commit":{"parents":[{"commit":"8ea2efe0ac5a8a705d0e70f033dcd1b5f8537885","subject":"Add scenario tests for failover circuit breaker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8ea2efe0ac5a8a705d0e70f033dcd1b5f8537885"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:12:16.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-15 07:43:29.000000000","tz":120},"subject":"DNM: test with different failover_threshold","message":"DNM: test with different failover_threshold\n\nChange-Id: I9de892ee2de89675703c52185c11acd47a424606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e9d0dd429437e16c98bf20ac5df34796c042994e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e9d0dd429437e16c98bf20ac5df34796c042994e"}]},"branch":"refs/heads/master"},"d59694f62175f0f19ffdb9bb97bedc5e5698247f":{"kind":"REWORK","_number":9,"created":"2024-10-28 13:15:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/61/932061/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/932061/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/9"}}},"commit":{"parents":[{"commit":"2b87b25aff755453d6e69fd84cbcba7dad3fabd5","subject":"Add scenario tests for failover circuit breaker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2b87b25aff755453d6e69fd84cbcba7dad3fabd5"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:12:16.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 13:15:04.000000000","tz":60},"subject":"DNM: test with different failover_threshold","message":"DNM: test with different failover_threshold\n\nChange-Id: I9de892ee2de89675703c52185c11acd47a424606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d59694f62175f0f19ffdb9bb97bedc5e5698247f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d59694f62175f0f19ffdb9bb97bedc5e5698247f"}]},"branch":"refs/heads/master"},"6382f941c3f97ccdcc1e59c51f92d9a369aa58f8":{"kind":"REWORK","_number":10,"created":"2024-10-28 14:16:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/61/932061/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/932061/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/10"}}},"commit":{"parents":[{"commit":"2b87b25aff755453d6e69fd84cbcba7dad3fabd5","subject":"Add scenario tests for failover circuit breaker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2b87b25aff755453d6e69fd84cbcba7dad3fabd5"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:12:16.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 14:16:18.000000000","tz":60},"subject":"DNM: test with different failover_threshold","message":"DNM: test with different failover_threshold\n\nChange-Id: I9de892ee2de89675703c52185c11acd47a424606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6382f941c3f97ccdcc1e59c51f92d9a369aa58f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/6382f941c3f97ccdcc1e59c51f92d9a369aa58f8"}]},"branch":"refs/heads/master"},"1794f9a1d1609575fdd098fc6367a1c4066c5737":{"kind":"REWORK","_number":11,"created":"2024-10-31 15:33:20.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/61/932061/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/61/932061/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/61/932061/11"}}},"commit":{"parents":[{"commit":"2b87b25aff755453d6e69fd84cbcba7dad3fabd5","subject":"Add scenario tests for failover circuit breaker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2b87b25aff755453d6e69fd84cbcba7dad3fabd5"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-10 12:12:16.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-31 15:32:58.000000000","tz":60},"subject":"DNM: test with different failover_threshold","message":"DNM: test with different failover_threshold\n\nChange-Id: I9de892ee2de89675703c52185c11acd47a424606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1794f9a1d1609575fdd098fc6367a1c4066c5737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1794f9a1d1609575fdd098fc6367a1c4066c5737"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
