)]}'
{"id":"openstack%2Foctavia-tempest-plugin~934390","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I8e49fc19faff59b2811a7ee09b7ccaa286f73d82","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"bump-hacking","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-10 10:26:40.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-12-10 13:45:42.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-12-10 10:26:40.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-11-25 13:21:43.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I8e49fc19faff59b2811a7ee09b7ccaa286f73d82","subject":"Update hacking","status":"MERGED","created":"2024-11-08 01:41:32.000000000","updated":"2024-12-10 13:49:07.000000000","submitted":"2024-12-10 13:45:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934390-bump-hacking","meta_rev_id":"4019269ab1ba3cc1420088cd1087a8e1a2512689","_number":934390,"virtual_id_number":934390,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 13:45:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2024-12-10 10:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-10 08:15:05.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2024-12-10 10:26:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-08 03:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-14 14:34:04.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-11-20 16:10:02.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"cf80e86599629939cefc3274473cdda575b3556f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-08 01:41:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8fccf4c4040a9394c9f308e400c7259ddd1eeac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-08 01:54:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d3f6566ddd4a2ddfce7e14e265b93c1091fccc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-08 02:11:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6a639b5b306e23cea0ac2d0b2d4f387ff1dfc0eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 03:43:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/642c39610ac8436ca18037e423eccc20\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/85f361603bc6452584249fb1d8f9eb4f : SUCCESS in 2m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5d33fd8c6654d63aca04b2423db39c3 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d6e81d11c8149ab85c25b6d5537871f : SUCCESS in 3m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff2f71840e5a4997861cea1a4c2ae222 : FAILURE in 27m 35s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/d8fdc335c83443d8b2a41364b569c7f3 : FAILURE in 20m 21s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1471e7c0589f4834b5ea793f0a0150fd : FAILURE in 28m 30s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/97d61f2814d849c8be539eca70ff53ea : FAILURE in 20m 33s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/ee89d2d277f749e38737d6e9f3f159b0 : FAILURE in 26m 00s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/86cee5b3a292411d97b41b3339fecc0e : FAILURE in 39m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2bb0ed2068df4481bb19be0bef3f396e : FAILURE in 22m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/49260da7bf794f0cb413a81784784b17 : FAILURE in 38m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ad24ddebbba54aec9b4b56208acc6adc : FAILURE in 33m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/0cc9caddcf1f4fd38a5d7004857c495c : FAILURE in 38m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d02cca3e16214041b131835fc5625834 : FAILURE in 22m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/5b49d311f99b445aa143037ce77bf52a : FAILURE in 43m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/30885518e5a74accb266108eeaae3ff4 : FAILURE in 34m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/d3e400deb08542d7b3806089acf37b5b : FAILURE in 42m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/fe9535eba08c4666990c0fd3a0ce8c27 : FAILURE in 28m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/cba1f5d45e89487b880864e7547441b8 : FAILURE in 42m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3041ccd1ae704ca1be7a3f7c54389383 : FAILURE in 35m 21s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/12a550cd402d433583a59d74243cafc6 : FAILURE in 35m 54s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/b55192828e04490db907c8214ca3f894 : FAILURE in 36m 26s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4c0e51e7c1a14cf59864ae5bff45c1fb : FAILURE in 44m 35s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/4248898860d7417eb8dc9473cd039e7e : FAILURE in 36m 49s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e4e8666235564e08831bc0bc332a90dd : FAILURE in 39m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/7df8dbf7d73e468796bd6f7073ecd9dd : FAILURE in 42m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a34dee48371b4227a88d014ab120004e : FAILURE in 46m 29s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/d66e197a95e14fc6b0a8db22068e5e98 : FAILURE in 33m 30s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cd40dbdd2078422d8a1e5f088f86cd47 : FAILURE in 46m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c8141b50b4314632a99a2921368b41e8 : FAILURE in 44m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e077630ada2c4a62896c7f25d3bfaa13 : FAILURE in 38m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f8ed739c0d1f42c584fecaa16acffa7a : FAILURE in 40m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/15313e899b7c413287e47afc3374e4c8 : FAILURE in 33m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/d27da76e5ece45f6a6f61eea4e1d3a63 : FAILURE in 38m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/eecd532051274439ba4bd2a802cf9db8 : FAILURE in 40m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/78599103076f46f6b53d1681981c137a : FAILURE in 40m 42s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/a74b84f2a04448068eeaaf1b191d2ad4 : FAILURE in 43m 09s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/cc97a787672348699d465b6b56fb6036 : FAILURE in 46m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6d38fdf79d1f4210ec3530ba10d5d7a192d3bddf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-08 18:23:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"55573bad471c12763cee9ad5fdd16c681acfe57d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-08 18:47:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"da6898b70ef193c67906be8832b790b6201c8a31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 21:54:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0b17cd2950f46589a8c6a46aff489a6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cd12964e01914ac79ecc7639c4979e5c : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/254d98835a5c47d3b37119543db0cd95 : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d1d556c15394bc89e45fcc50c7c1260 : SUCCESS in 2m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fd07017d009a4505b8d7b02cdef635af : SUCCESS in 1h 29m 13s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/41bbbe558d5542848478e754861db1bd : SUCCESS in 1h 01m 06s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8602340a35294a82b15429119b1e165b : SUCCESS in 1h 31m 58s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/249aa2ebbfd240d3969dd52803a61460 : SUCCESS in 1h 18m 32s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/829a5c87e33e4f9eb4f16e8778f4e535 : SUCCESS in 1h 16m 27s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3a473cc9684342bdbe0510c5706d3c44 : SUCCESS in 1h 28m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/17f368ceeb81454399f6b97614018f24 : SUCCESS in 2h 12m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1cfd331a90f94943b5f88cdc5cab4eb2 : SUCCESS in 1h 15m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/fe5204403d2142bc8fea22fd96a31134 : SUCCESS in 1h 43m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ddcbf9a6d17b40cba58d1905bd819467 : SUCCESS in 1h 25m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e2d49cd066ff460b840c8a3f65677692 : SUCCESS in 2h 00m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/a47e337729354078a359a72c6cd21c34 : SUCCESS in 1h 15m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/f8c0be297ad144449cdaace7f59a1bd2 : SUCCESS in 1h 51m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e3dc8285a7b5430da1460029bc3f1ddb : SUCCESS in 1h 15m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/744aea814bd0430fbb96ed1550adf7dd : SUCCESS in 1h 34m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/577a682651384c8fafeb27042c1f8115 : SUCCESS in 1h 17m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4bfbccab1dad476a8e0914a4aaacec59 : SUCCESS in 43m 27s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/bddc1a00ddf14f2b8e244838c1956ba7 : SUCCESS in 40m 50s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3ee0fd9830b64e1d83d8506e5dd11e9a : SUCCESS in 40m 20s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6a61d5a3dc4d42c99c01e34ec9b1154c : SUCCESS in 38m 42s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/20cd5f1303f74d79a2719d8f26eb7540 : SUCCESS in 47m 21s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/44608e6510d740188268c6548a662537 : TIMED_OUT in 2h 32m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/d355a0294ebc430e8fbb7d82ba083969 : SUCCESS in 1h 52m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ee9d1d64523744fe9c009b745170e8ff : SUCCESS in 1h 03m 42s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/bb7bfc779a174a2085f2b085767896a5 : SUCCESS in 2h 21m 09s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/415ffe93c78a459db795f5ee60c05eac : SUCCESS in 1h 29m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/be01a628a5e647a9b6703e2471ccd780 : TIMED_OUT in 3h 03m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fab0b714671f491e8ad2c497dc769074 : SUCCESS in 31m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1a6044f5084041d3bae9a546d12d90ac : SUCCESS in 37m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/416cb828c3de49af91f6f21411cae364 : SUCCESS in 37m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/830c41150c024abba6e9cc429fdbe167 : SUCCESS in 34m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/6434a5feffe8436a9e34f6c85dd35c2c : SUCCESS in 34m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2fcca351bf3f4dbc9571f9753e398401 : TIMED_OUT in 2h 32m 53s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/17c7d61b4aa148e5aadc3733c48b9e98 : SUCCESS in 51m 02s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/857a1fb19690459b833992ac29ba851d : SUCCESS in 57m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"586e25afdf5bf2d34c4d8ce7c6a5c73b2ab71b3b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-14 14:34:04.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"53221b3e68cd125dc048464257efe2344e9afd9c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-20 16:10:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"64b5b7d9f741487e694848fb4338d4f936b75485","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-20 16:11:29.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0ca006fb0a4631ec434bde06e2a58adde4a3b43b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 16:12:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fbc4156968a91a51cc1d37e238b5eb158161afb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 16:22:59.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/213c7aaeabb64414a6d40c3d3f1e28ad\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/45f3792b1636455fabe1ce126864d38c : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ca70f613a574ac4a1eb34572f836ed5 : SUCCESS in 2m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d9c5b7f2b1640e99743a95317f0b5db : SUCCESS in 1m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3464616fd2674ef391163f7c0b068b24 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/923131f1614b4ad2a1eb9b4f02f147f2 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/63d5b86913814b1885a02d67465abff0 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9ad6d867ae924d4b8f398f54cac69b55 : CANCELED\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/cb88ba96afa5495781b50ffff3c48f02 : FAILURE in 9m 03s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d63c06f932c84a3eb792256ba57d211d : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/98fe4def77c647d1b6fc9d1054c33a3f : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/728da7c7748a46c79e1df40c496b9a10 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/6a714a2e0ef741a482dffccb73159c88 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/0fbaf2728bd74c24bfb39eec6836f238 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/2480e74f52034d3a8475c9d3e1067dfd : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e25dedde758d41518200cf0f3ff07f92 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/291a451a84f8495c9e42acef195174af : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/d0a9462912c644339a24d0c145447c98 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/f9eb5ac9bc634446a003e0268808a00d : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/7db41cd0017743b8a0aca6e9b468f7e1 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d62ae285a89b4af7a7183eb80c79dd18 : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/5c2249a9d1724e3e9d0c818034899c40 : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/715aadd3aeda47f6a094962a987868fc : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/ead654efb2bf4f0a97a1efa6f2094964 : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/5ab3c96602704ec4bcc80489e6dec99a : CANCELED","accounts_in_message":[],"_revision_number":5},{"id":"e99a27037942f9d5c1b69421abcd6f0bf22c04bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 16:52:54.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"55fecae676aebfd93d4c6568623befa843cb0c4c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 16:55:53.000000000","message":"Topic 2023-1-last removed","accounts_in_message":[],"_revision_number":6},{"id":"fbdc289ea2b2f7d66fbfb97d080dc73b26915fb5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 16:56:01.000000000","message":"Topic set to bump-hacking","accounts_in_message":[],"_revision_number":6},{"id":"9f3826a75aa164195112c0598f6afbca16302fbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 20:50:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b644de6f18f4fbba11e4d197378dae8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c53e3c3eebb0440588c523b4133322fc : SUCCESS in 1m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0897c91365164c769dadf0e42b87be24 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2781cb4cc3f64a0e8e74f2032cfdb4b3 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8142de3432645dcb9ee4c969a72abd3 : SUCCESS in 3m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6b8f352a4e374cc3a81b12ce06d3b91d : SUCCESS in 1h 47m 10s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/fc4c03733e3f48d49f3a8310f88f5a09 : SUCCESS in 58m 05s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/0a143d8bcbbb4903a6341ef3a9f48afd : SUCCESS in 1h 12m 58s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2dffe972af9745e6834365a637f52686 : SUCCESS in 1h 14m 28s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ce2f251ee2304a068934450dd41d9914 : SUCCESS in 1h 28m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9db3e22e26b84ee7a0b83850234324a2 : SUCCESS in 1h 41m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/de97f6d1890a4a91918a11c956f3de8d : SUCCESS in 59m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4748664cb6fb4483a525ed75803c1672 : SUCCESS in 1h 43m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/21038f1fb08d4c4da4e4a6f0a553ed46 : SUCCESS in 1h 02m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/10c6cd071cb94de2a3b26c7c81645655 : SUCCESS in 2h 28m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/50e82023f8714de0ba83195a8b3f22f6 : SUCCESS in 1h 30m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/f2738c510fa24ef0b2e7bc6c18f122ad : SUCCESS in 2h 16m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/642a21efd3794f6798216ce85f4d8637 : SUCCESS in 1h 06m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8cb3bac61cc14c88a8a86fbefd98bef2 : SUCCESS in 32m 47s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/836a55fcf65d4ea689ee74534eab4a5c : SUCCESS in 31m 46s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/6d4af65f9c244933b61c55da8b151dd5 : SUCCESS in 34m 41s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9199a5cf19d24b79b89ec058c7dcc507 : SUCCESS in 34m 37s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e35b8cfab03a4c5897e73f632c927f63 : TIMED_OUT in 2h 32m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/be07209550944ce19d7aefc3660c2180 : SUCCESS in 2h 15m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3b4a4fe56cc0400b846d24f898756c45 : SUCCESS in 53m 41s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/0b83ce5bb9374fb89bfa7d079434142d : SUCCESS in 2h 27m 10s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4a80749e03264b9a8dc8691eb42580db : SUCCESS in 1h 27m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6c11b4d3025647a69cab8b38ffa0529e : TIMED_OUT in 3h 03m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54006ce062824e39bfbd8454395880d5 : SUCCESS in 30m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/6714c678fdf645abac7020fc7246e350 : SUCCESS in 49m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/a0f6ff4870314ef7b077465b36087282 : SUCCESS in 33m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/5ce1f31ef05a45c6a8826cd680373da3 : SUCCESS in 28m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/377b84ea168d4755b894a2b4fa7b739d : TIMED_OUT in 2h 33m 10s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/d7475615ac9946e5b299bbe0cc38c3cd : SUCCESS in 47m 31s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/edda35943d3144b7aa1ddb2025089f48 : SUCCESS in 1h 02m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6c7824b300a7022fa438b6e1aece5e6313b3a4cf","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-25 13:21:43.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"427de4dd7590a5ef1c26b66e9e35742517e1ea74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-12-10 08:15:05.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"e187e6cea677050d31b044113221f976441f617a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-12-10 10:26:40.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1bf76ad05b70da08e59d73a92b052e2af1eb19fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 11:19:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d36912a8d29490ebe3054eb99a82c17\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9287c31559f74fdead46759993f99202 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844a607faaba4870b7f44fca39bbcc05 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b6d02b02686454583df4a490f1a6c03 : SUCCESS in 2m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7da6b71895ab4507b69667ff427fcdab : SUCCESS in 1h 13m 49s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/9f1bcfffb0c0429784dd5095c5ba31a8 : SUCCESS in 1h 40m 51s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/f3c800046ac64903923f76cc4b37db96 : SUCCESS in 1h 34m 29s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/c5941661c7cc4670bdf95cb5c04c17a2 : SUCCESS in 1h 33m 16s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7cc99e8818a644a8b350e2838a4fa7d5 : SUCCESS in 1h 29m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9b5d31523ae641c0a57eb61e3d6e707c : SUCCESS in 2h 23m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/42139d54582e4f23bb12e1afdc06716d : SUCCESS in 1h 30m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/46ec94c956b84250a66c6c3954ac1d02 : SUCCESS in 2h 27m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1781878472de40a6bd24540c6a581123 : SUCCESS in 1h 03m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e44366e5225e4d5a9197e335db87eebb : SUCCESS in 2h 27m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/737456083d4548e8a23e108229704623 : SUCCESS in 51m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9031d00b329149cda28f66713c8a13e3 : SUCCESS in 2h 00m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9e8a5066414b43acab1075c0b5361f7a : SUCCESS in 57m 42s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bf5e6819a2334dd78f79a3663d5ea2cf : TIMED_OUT in 2h 34m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/49b5bd555ee149fcb62dcaadc665ac11 : SUCCESS in 2h 29m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ccbfa1d026c44fe5bce201bd4cb43be2 : SUCCESS in 1h 21m 41s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/2da50ca04ca241d0842aa51db31ede3d : SUCCESS in 2h 14m 42s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6bf8f6d3e69348e19978fbb846af862b : SUCCESS in 1h 27m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/945f0fa533c1420ca7feef00efaa94e2 : TIMED_OUT in 3h 02m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ca8b966111d4035878da6fff9e10ea2 : SUCCESS in 28m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ce0c4ab78a36412daf0aa249d872555b : SUCCESS in 34m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/508b69a20ed549e98b6162919d9e3ebc : SUCCESS in 29m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/23677d8ee1ce4ac1a8b92a6b2efeb5e5 : SUCCESS in 40m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6a0cfe192140453595c61abd5b26ff97 : SUCCESS in 2h 13m 36s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/59a5f992ee72450e8cde33b11ca6021f : FAILURE in 23m 36s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/69f1c2fe2d9b47bc81c66ef8c79a0363 : FAILURE in 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"13ce27a2a8cc2eab154eb3ff386be3a33ffd4f88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 11:20:30.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2490770ba4b1110e7fa29c83ca371adae41a0470","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 13:45:42.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f80b890baee45a19c498a848f059587\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/dff5827dbd2e4f1caf0b663ffff6735d : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d4da3625be424c8dff73c828b26626 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f741fc6daea46b591bad1dff4c2b24b : SUCCESS in 2m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/01b5a26911ef4c5f8bee697ad11e79ee : SUCCESS in 1h 47m 37s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/fee192e9f5fc4d2ca5ff60af2a9804d9 : SUCCESS in 57m 52s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/78adb4d892a44538b4a8888cd5a87be0 : SUCCESS in 1h 01m 06s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/df755559eedf4dc7a79b59543fe5d301 : SUCCESS in 1h 25m 30s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5c4e1d6f2f5644f2a440b8bc75f07161 : SUCCESS in 1h 42m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/716f17171cde42938a7e3ed79cb5c772 : SUCCESS in 2h 15m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8859664076f248cd9c730c86cdf3c1dd : SUCCESS in 50m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/6860ebede3374260bf72c0bb8f4a0d0f : SUCCESS in 2h 14m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ef79a5aa9eb545a28d8607a262903ddd : SUCCESS in 1h 10m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/f6e648ef87664268940428813e7f2648 : SUCCESS in 1h 36m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/bbad4af538c4479191be93dc13a4256d : SUCCESS in 1h 18m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/d22ebb6638bb4c6ea7f3faa87638ecb2 : SUCCESS in 1h 16m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4d1b6ce91af949db9ab0f80ad3e2d114 : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":7},{"id":"99d088768fea796a6e9f4c5b6e5c88a925acef4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 13:45:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"4019269ab1ba3cc1420088cd1087a8e1a2512689","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 13:49:07.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97c76624573542f5a5f5e311b5b5ba23\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/879513aa8d0645bf9c754cca908cebbf : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa871aaacac247b888b3b9ec9798e4c2 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"77b8baea5d0562914096b9aac6daf9fe48e0da64","revisions":{"75eb6cc7f314c8c3cfc70095121b2fa96a8085f6":{"kind":"REWORK","_number":1,"created":"2024-11-08 01:41:32.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/90/934390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/90/934390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/1"}}},"commit":{"parents":[{"commit":"4af9bae253bc7e8a1fde4f84d45aecbb825f10df","subject":"Remove support for Pyton 3.8 and older","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4af9bae253bc7e8a1fde4f84d45aecbb825f10df"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-08 01:39:29.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-08 01:41:30.000000000","tz":0},"subject":"Update hacking","message":"Update hacking\n\nThis patch updates the version of hacking to be consistent with octavia.\nIt is also required for flake8 to understand f-strings correctly.\nAlong with this version update, the patch fixes issues in the code found\nby the new versions.\n\nChange-Id: I8e49fc19faff59b2811a7ee09b7ccaa286f73d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/75eb6cc7f314c8c3cfc70095121b2fa96a8085f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/75eb6cc7f314c8c3cfc70095121b2fa96a8085f6"}]},"branch":"refs/heads/master"},"1b407cf89e3d9b9679f65340e804b3b38a7487a7":{"kind":"REWORK","_number":2,"created":"2024-11-08 01:54:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/90/934390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/90/934390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/2"}}},"commit":{"parents":[{"commit":"4af9bae253bc7e8a1fde4f84d45aecbb825f10df","subject":"Remove support for Pyton 3.8 and older","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4af9bae253bc7e8a1fde4f84d45aecbb825f10df"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-08 01:39:29.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-08 01:54:19.000000000","tz":0},"subject":"Update hacking","message":"Update hacking\n\nThis patch updates the version of hacking to be consistent with octavia.\nIt is also required for flake8 to understand f-strings correctly.\nAlong with this version update, the patch fixes issues in the code found\nby the new versions.\n\nIt also disables the docs job, which will be re-enabled in the next\npatch.\n\nChange-Id: I8e49fc19faff59b2811a7ee09b7ccaa286f73d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1b407cf89e3d9b9679f65340e804b3b38a7487a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/1b407cf89e3d9b9679f65340e804b3b38a7487a7"}]},"branch":"refs/heads/master"},"e7a9138843de32b228f747203d861defd5f04191":{"kind":"REWORK","_number":3,"created":"2024-11-08 02:11:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/90/934390/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/90/934390/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/3"}}},"commit":{"parents":[{"commit":"4af9bae253bc7e8a1fde4f84d45aecbb825f10df","subject":"Remove support for Pyton 3.8 and older","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4af9bae253bc7e8a1fde4f84d45aecbb825f10df"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-08 01:39:29.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-08 02:10:57.000000000","tz":0},"subject":"Update hacking","message":"Update hacking\n\nThis patch updates the version of hacking to be consistent with octavia.\nIt is also required for flake8 to understand f-strings correctly.\nAlong with this version update, the patch fixes issues in the code found\nby the new versions.\n\nThis patch also removes O349 check for importing mock since flake8 has\nadded H216 which does the same check.\n\nIt also disables the docs job, which will be re-enabled in the next\npatch.\n\nChange-Id: I8e49fc19faff59b2811a7ee09b7ccaa286f73d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e7a9138843de32b228f747203d861defd5f04191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e7a9138843de32b228f747203d861defd5f04191"}]},"branch":"refs/heads/master"},"5dfb082ed5b9ca06da1fb063a319403c3225934d":{"kind":"REWORK","_number":4,"created":"2024-11-08 18:23:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/90/934390/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/90/934390/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/4"}}},"commit":{"parents":[{"commit":"4af9bae253bc7e8a1fde4f84d45aecbb825f10df","subject":"Remove support for Pyton 3.8 and older","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4af9bae253bc7e8a1fde4f84d45aecbb825f10df"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-08 01:39:29.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-08 18:23:44.000000000","tz":0},"subject":"Update hacking","message":"Update hacking\n\nThis patch updates the version of hacking to be consistent with octavia.\nIt is also required for flake8 to understand f-strings correctly.\nAlong with this version update, the patch fixes issues in the code found\nby the new versions.\n\nThis patch also removes O349 check for importing mock since flake8 has\nadded H216 which does the same check.\n\nIt also disables the docs job, which will be re-enabled in the next\npatch.\n\nChange-Id: I8e49fc19faff59b2811a7ee09b7ccaa286f73d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5dfb082ed5b9ca06da1fb063a319403c3225934d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5dfb082ed5b9ca06da1fb063a319403c3225934d"}]},"branch":"refs/heads/master"},"69bbf0693f45d47ed14439268523e020e600bdb6":{"kind":"REWORK","_number":5,"created":"2024-11-08 18:47:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/90/934390/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/90/934390/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/5"}}},"commit":{"parents":[{"commit":"4af9bae253bc7e8a1fde4f84d45aecbb825f10df","subject":"Remove support for Pyton 3.8 and older","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4af9bae253bc7e8a1fde4f84d45aecbb825f10df"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-08 01:39:29.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-08 18:47:26.000000000","tz":0},"subject":"Update hacking","message":"Update hacking\n\nThis patch updates the version of hacking to be consistent with octavia.\nIt is also required for flake8 to understand f-strings correctly.\nAlong with this version update, the patch fixes issues in the code found\nby the new versions.\n\nThis patch also removes O349 check for importing mock since flake8 has\nadded H216 which does the same check.\n\nIt also disables the docs job, which will be re-enabled in the next\npatch.\n\nChange-Id: I8e49fc19faff59b2811a7ee09b7ccaa286f73d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/69bbf0693f45d47ed14439268523e020e600bdb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/69bbf0693f45d47ed14439268523e020e600bdb6"}]},"branch":"refs/heads/master"},"0ba927bceda1c9eb0ef12f41b2b81ca53f972e87":{"kind":"REWORK","_number":6,"created":"2024-11-22 16:52:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/934390/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/90/934390/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/6"}}},"commit":{"parents":[{"commit":"f5c810709825d69a95196691b5cb38f8d59a68f9","subject":"Update octavia-tempest-plugin configref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f5c810709825d69a95196691b5cb38f8d59a68f9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-08 01:39:29.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 16:52:39.000000000","tz":540},"subject":"Update hacking","message":"Update hacking\n\nThis patch updates the version of hacking to be consistent with octavia.\nIt is also required for flake8 to understand f-strings correctly.\nAlong with this version update, the patch fixes issues in the code found\nby the new versions.\n\nThis patch also removes O349 check for importing mock since flake8 has\nadded H216 which does the same check.\n\nChange-Id: I8e49fc19faff59b2811a7ee09b7ccaa286f73d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0ba927bceda1c9eb0ef12f41b2b81ca53f972e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0ba927bceda1c9eb0ef12f41b2b81ca53f972e87"}]},"branch":"refs/heads/master"},"77b8baea5d0562914096b9aac6daf9fe48e0da64":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-12-10 08:15:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/90/934390/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/90/934390/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/90/934390/7"}}},"commit":{"parents":[{"commit":"b139fe94ae75765e6d80f7ffb5de8c62aa3cb9f6","subject":"Update octavia-tempest-plugin configref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b139fe94ae75765e6d80f7ffb5de8c62aa3cb9f6"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-08 01:39:29.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-12-10 08:14:56.000000000","tz":60},"subject":"Update hacking","message":"Update hacking\n\nThis patch updates the version of hacking to be consistent with octavia.\nIt is also required for flake8 to understand f-strings correctly.\nAlong with this version update, the patch fixes issues in the code found\nby the new versions.\n\nThis patch also removes O349 check for importing mock since flake8 has\nadded H216 which does the same check.\n\nChange-Id: I8e49fc19faff59b2811a7ee09b7ccaa286f73d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/77b8baea5d0562914096b9aac6daf9fe48e0da64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/77b8baea5d0562914096b9aac6daf9fe48e0da64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
