)]}'
{"id":"openstack%2Foctavia-tempest-plugin~935595","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I8219106ec26ddc2edfaf39a69c64a9887fdfb54c","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"migrate-to-noble","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-13 18:12:37.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-01-13 18:12:37.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-01-13 18:12:37.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-01-13 15:58:19.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2025-01-13 18:12:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8219106ec26ddc2edfaf39a69c64a9887fdfb54c","subject":"Update gate jobs as per the 2025.1 cycle testing runtime","status":"MERGED","created":"2024-11-19 03:07:02.000000000","updated":"2025-01-13 18:14:55.000000000","submitted":"2025-01-13 18:12:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"935595-migrate-to-noble","meta_rev_id":"348fa5c9c7e9afb4eb4004a23a828a0470363b23","_number":935595,"virtual_id_number":935595,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-13 18:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-13 15:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-01-09 18:58:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2025-01-10 19:55:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-13 15:58:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 06:14:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-19 09:06:43.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-12-03 18:31:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-12-21 13:36:34.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-01-09 07:30:56.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2025-01-10 19:55:10.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"069181867a017dfaaef632fb144864b96f12af51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-19 03:07:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6646363ac8a22bac6d15c2d40397e78e695f8b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 06:14:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c394ecd813bc4736876b8f3324b92973\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52586217630849d4b3bae080803f6c36 : FAILURE in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/956bf10fa2534333aa2b06d89baee1ad : FAILURE in 4m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d8f600bdb3124810a734e99876756719 : SUCCESS in 1h 20m 00s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/559430fff0ca40f7944079b600b8e795 : SUCCESS in 1h 21m 36s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/9d0b9e46fc6748d480e850f0ec310bb3 : SUCCESS in 1h 16m 53s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/3d9b36f1ee4d409a81c6ee6c27527fce : SUCCESS in 1h 12m 31s\n- octavia-v2-dsvm-noop-api-stable-2023-1 https://zuul.opendev.org/t/openstack/build/deece25db1f9476f84dcc1666f664b93 : FAILURE in 9m 46s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ba729516b29f42ca895c76f359fe5230 : SUCCESS in 1h 08m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/55dd8120d65c4b5aadea632e7be75f80 : SUCCESS in 2h 00m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/96539136e6c64dfdbc7e9f53144fbeba : SUCCESS in 1h 22m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/26fc2a137f3948a68184e494b0fb1ba1 : SUCCESS in 1h 59m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/eec56f6246204f828203cdee2531d413 : SUCCESS in 1h 17m 13s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/978280ba9db540cbbc0d9660830d3e1b : SUCCESS in 2h 03m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/bc9d6b2835ed473c8b4b14c19d267d5f : SUCCESS in 1h 16m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/84655e19049041f080cc8d6f92ba7db0 : SUCCESS in 1h 30m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/fe00b079b28446e9bae877d6aab8026f : SUCCESS in 1h 04m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/1248704613fc4a9ca7f70bbca1ae95b9 : FAILURE in 8m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-1 https://zuul.opendev.org/t/openstack/build/c70e3fc753a745a3a251c3619fd0336e : FAILURE in 10m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a2faccf6e6414556b87ca5443946f1cb : SUCCESS in 38m 03s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/70dce59b10024efcb5e8201d4cfcbd54 : SUCCESS in 32m 56s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/154589d012e14d6b96245d9465577fb9 : SUCCESS in 33m 26s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/ad86a528d8874d3eb85a02c4d377dd14 : SUCCESS in 33m 50s\n- octavia-v2-dsvm-tls-barbican-stable-2023-1 https://zuul.opendev.org/t/openstack/build/3e57c4d052ef4321bd0f4d4b4d15ecde : FAILURE in 8m 05s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f9c9e1348c464b0d910376d58ceb04b8 : SUCCESS in 2h 09m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/80c6c7f1c3f045db9c3834216f20dce4 : SUCCESS in 2h 00m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cebdb913d37f4ab1b1fcf65ebbf461b3 : SUCCESS in 1h 25m 31s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/a52ed08666294e688d2934ce1a57f257 : SUCCESS in 1h 57m 16s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/974c36c76413496383faa3f1599721df : SUCCESS in 1h 14m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f8b5439c004406da9922c2c92651566 : TIMED_OUT in 3h 03m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43d8a03a959e401d9cbf22f2edcaa7eb : SUCCESS in 36m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ef3440707bc544e189f99a3035e8efa3 : SUCCESS in 31m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/cd2e77b3a068419f94147256e41f7073 : SUCCESS in 31m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/df340e18b1884004a4cedcbd18197a17 : SUCCESS in 30m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-1 https://zuul.opendev.org/t/openstack/build/6ce69d8953674d83b00426578f0e6425 : FAILURE in 7m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2c1aecf5a6a44fadbbcef227e36c8850 : SUCCESS in 2h 15m 21s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/f03030eeaf0e46c3be26ad55b14301cd : FAILURE in 22m 23s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/432d0e5af7c5410193b627477a1ccb53 : FAILURE in 26m 10s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia_tempest_plugin/tests/scenario/v2/test_traffic_ops.py","accounts_in_message":[],"_revision_number":1},{"id":"89ca968c97303fd9a8350f2764b4b2badd444e29","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-19 09:06:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3ce8e102c196c8a2ded97680b65a82d6a3d4c81a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-27 01:53:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b8b63214f62d22e8748fac715324485fa84b6c36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 05:00:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2ee7bbb60a44613af8bfee1904d5638\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/623b61f6a1d0478fa048c97a38d93f05 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45ea18c6f35f40cdb8cb857a17ae8c76 : SUCCESS in 2m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/406ae5ce771247b59f53553a80d224df : SUCCESS in 1h 37m 59s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/c2eaec25a2b544ad9641091d83e65c3d : SUCCESS in 43m 21s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/0f4eb975a4cb44e5adb53d05957002b5 : SUCCESS in 44m 32s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6c7f1a3bd20146108825c7dab4cd304b : SUCCESS in 1h 21m 44s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d5c012b622bc4be3be3cec165b6bd45a : SUCCESS in 1h 40m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0fd64a5690254f6aa8b87e268c006543 : SUCCESS in 1h 34m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1d8a38ac75374cfcadffc0870b0b371f : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a2a1b646a06e4ba1ada17d011994b3bb : SUCCESS in 1h 30m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/eca5d9be65e3477981a4c743dc6e6be0 : SUCCESS in 1h 27m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/7ce3dd5fb4344818ae6ea1fc51ffa4f6 : SUCCESS in 1h 25m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8480b3370a0a48b183fe6c6174186630 : FAILURE in 16m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/8289d5caac4544e5921e698443e2b791 : SUCCESS in 1h 04m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/54377de2515e457a9b570c40c4654047 : SUCCESS in 1h 22m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c9795ca231b84d8d82519f47b80e2d01 : SUCCESS in 45m 39s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/37e7f9c83b8348d0858d9839244b6cf4 : SUCCESS in 40m 02s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/9b88a42e771f4140979ba5d3a9d37b95 : SUCCESS in 42m 01s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/944c2c19482f4839bd9119307b94f9ef : SUCCESS in 41m 11s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e892d4c2c77a45418ac3796bddd5e30b : TIMED_OUT in 2h 33m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/e1aef6ea77464ab2bae21c022be7b4cf : SUCCESS in 1h 37m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1c2dbf9f2aa848e2a353ae607f71b59c : SUCCESS in 59m 52s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/41e262dc565b4f378fe740b5ddb8346b : SUCCESS in 2h 08m 59s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e5b03b1ad69742d183aac75eefe43281 : SUCCESS in 1h 21m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/554347514c75454fa73d31a678bd5931 : TIMED_OUT in 3h 03m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8677b4c65a3f4b9fb2ab01f2e20dfe46 : SUCCESS in 54m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e8221dade2504f588a31b760430930a2 : SUCCESS in 34m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3261146009e34d05a6249dcbc4907db4 : SUCCESS in 37m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/90fa6a9f7bdc4d3ead7a4da44d4d0ce2 : SUCCESS in 33m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/87f8d495dc3247e7add89b792468f539 : TIMED_OUT in 2h 33m 02s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/fc9323d52f21413db2d99879c76f59f6 : FAILURE in 21m 19s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/50d30515ccb3406d929faa6e3428262b : FAILURE in 28m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3081aa55a7d9490b500b4e3369bb21e67e5a2ab9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-27 19:54:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13e9f58e41b0096ff1a7c8a71f3100dcad0aeb3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 23:03:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a658d5be3764c5a902e4481bba333d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7da6f4ba3bb437b8d5f5cfab8f08acc : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffb24b49234a400f9fb13c4746145307 : SUCCESS in 2m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/01e918fa7a194b2fbeeae80839f6af3f : SUCCESS in 1h 28m 37s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ca37ca38cc324f28a8265de9d5a39b45 : SUCCESS in 1h 34m 39s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d62e0f979f2a46b9886c66f1522d071f : SUCCESS in 1h 49m 17s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/23ad777d45d74581b9d26162f327a0fd : SUCCESS in 1h 37m 52s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/45262ecc8d0e4681906ddfbf224cf7ad : SUCCESS in 1h 14m 57s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/37f70868e5724a23824dbeda16f3f857 : SUCCESS in 2h 00m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b874fb7bf2af4e7492cf454685a9c7e8 : SUCCESS in 1h 10m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f92bc083e93d47b79de8f0302c598090 : SUCCESS in 1h 56m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/213c63d0bd724add84f52199890b8c7a : SUCCESS in 1h 12m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1466904f027e459081de8dd5ad7d812f : SUCCESS in 1h 55m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/41a9a95eae5641e5b46469a56936e6a9 : SUCCESS in 1h 18m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/dc90567dfccd4ddbb238d396020f1934 : SUCCESS in 1h 29m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/dadc91e3bff34a6cbd5ca7e969bb7421 : SUCCESS in 1h 09m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e87243e231af4311bb6f35fc1fe4329c : SUCCESS in 41m 32s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a5d2803b25724df5aa48d612d6553d56 : SUCCESS in 43m 23s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/de0ed9c10d2346f297520258504a8365 : SUCCESS in 44m 48s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/d702cc1184074dc2a7c3febf7e9166e4 : SUCCESS in 35m 33s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/98d7b8a3396b4526a89768f09e7eeffd : TIMED_OUT in 2h 33m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/4ce0e9e8c73742cd94ae136befa1f567 : SUCCESS in 2h 00m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1f2a7f9200194dea8595d87134ef3385 : SUCCESS in 1h 11m 26s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/8ccf0d1fd9594923ac1cd5e3bc7cc228 : SUCCESS in 2h 06m 17s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c7831fa477db46bebf3e46345c45a7ae : SUCCESS in 1h 21m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b1cda4fe99264d19ac1f016c2f645873 : TIMED_OUT in 3h 05m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0fa2b98487c47d9af2a2dbf59b05df1 : SUCCESS in 37m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/81b876ca6a7a4d83bc4ba32996240c49 : SUCCESS in 31m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1d171fa216154a39af833d44e792407b : SUCCESS in 41m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/5be89761f3dc472ea1a67d43b315e249 : SUCCESS in 37m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e33c59ed8a7440918a929539a7165c52 : TIMED_OUT in 2h 33m 08s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/f4494ccf66ff41dc8698a21b7bbc8552 : FAILURE in 20m 10s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/cd579f8d37c34cb6aff16ec4df4e73c7 : FAILURE in 22m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b43c0500214b151d092d4caf7c20cdafc22498fe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-12-03 18:31:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"859547e5b078779746e5ae7cf10092e061db741e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-20 19:43:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9cc63876146fc50428c9116c1f56860c0d8bb5b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 22:48:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31736dd17ca54a06afba076915e8828f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7c465e541e4af5998cb710a982a638 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fe4d2a46102465abfcafd00a00bc6ea : SUCCESS in 2m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1798f6ccd74483e952d06fbf2b6a555 : SUCCESS in 1h 34m 50s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/69d0b11158734bfc88aa9d620af56c3d : SUCCESS in 59m 09s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/911e09a7f4d94b658292c936ec26437a : SUCCESS in 1h 25m 06s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2df22a9799f844b0a0b0d07941a819b2 : SUCCESS in 52m 15s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/96ec251b3a384c26a6e8d68191ad969d : SUCCESS in 1h 19m 38s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3a18e935008d4d9f8e6d4d93ac96e440 : SUCCESS in 1h 38m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/39cf3c488bce4acc9a0b13375e3975f8 : SUCCESS in 58m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/6a574b29592140bc8203a128c0b53fc2 : SUCCESS in 2h 02m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ac021b1449704e399b139079295a1db2 : SUCCESS in 1h 07m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/5472b49b93b348519d554c3626c00a86 : SUCCESS in 1h 59m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e7fd0f6e7173400a8554fcd8cc3d65ca : SUCCESS in 58m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/24bd7c8c76fd4706ad9ed0ac6f05b29a : SUCCESS in 1h 11m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/bc112088c9674c1797f9a4fb1abdd80c : SUCCESS in 52m 09s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fe466b2bf4624fa9acc1524be211ddd9 : SUCCESS in 1h 56m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/6873701d6f2a40bf85aa9b00261ed6b0 : SUCCESS in 1h 54m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2c07f92ecd5945b5a50f961fbd2258f5 : SUCCESS in 1h 15m 02s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/d5b3e3000d37411b884daac1a76ebca0 : SUCCESS in 1h 57m 07s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5db4e476b8684cf2bd483432d1d05f0f : SUCCESS in 1h 19m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1b7bbf46caa24dae8a52b5b84f4dad89 : TIMED_OUT in 3h 03m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71554365221944d4b54057aae1d36e39 : SUCCESS in 33m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/578a326819ab44bc961e9f3fc609cf2c : SUCCESS in 32m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/08d0cc32207045269c2b6fb9eb94654c : SUCCESS in 32m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/630e82a176f54a0b83a6982d65c33981 : SUCCESS in 30m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6b54050b87dd47ffb33dde5facff3205 : SUCCESS in 2h 05m 54s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/4f1b13377a7a4fa4832a34f7edbf5c9e : SUCCESS in 52m 05s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/1c9eea88b6cf4d4484eb1c2f3ea03cd2 : SUCCESS in 58m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"58f029950f8d050a56021862d14cba1c1f6313af","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-21 13:36:34.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e24a3a9308346c6cbb160284c22c428e65c1d4dc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-01-09 07:30:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7e76c31e4baa8dc8e030f87a18155a54b4beee48","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-01-09 07:31:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dc84b31cd666f15ad064fcddac13b813e2ef0628","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-01-09 10:25:17.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ead8da3d92838bae04014ca9a463a7582fb80446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 13:30:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dccc11fec4743dca15dc6d8adb562a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f20188ef33f4fffa335cd443605e669 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c7b43c37eaf40df98a7bcc8faa28968 : SUCCESS in 2m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b81b184b615a4a41841c1131ca095396 : SUCCESS in 1h 14m 38s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/098af062c81640e6844a7828e15a2f9f : SUCCESS in 1h 20m 24s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/c91f58e1d1db47de94a5fecf6939fa04 : SUCCESS in 1h 25m 06s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/bbfdba1a348f4197bf4a88b2dbac3aed : SUCCESS in 1h 39m 16s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/273eb0ee2bc649a4b538d2760951dc13 : SUCCESS in 1h 23m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0370f53b83494c37bc056f7dca4c2011 : SUCCESS in 2h 15m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8a7c13ab82b146cdaf6b45efd9b92148 : SUCCESS in 1h 13m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/d2b3d819287747a5832ed3140e4da9aa : SUCCESS in 2h 31m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/8550cf1840a54aae8d1442707e3cc855 : SUCCESS in 1h 20m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/5e183ddffa2045519199ceef04f542f4 : SUCCESS in 2h 03m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8e516c116b9142d7a2b1e8ff1ab5d4c0 : SUCCESS in 1h 20m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2f532bb8d7b0455d838169e4c7b0fe1d : SUCCESS in 1h 43m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4c71183cb1d84fcd87de4ecc926f5d9e : SUCCESS in 50m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5f5abb9de2fb44df8e52c652bd4a28f0 : SUCCESS in 34m 44s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/d3a860c974de4c62b9c97db219dc20fa : SUCCESS in 37m 35s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/ac3a33d79ad3477daf24225ab2255376 : SUCCESS in 32m 51s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6bc732c74a104befa341101bc5a22cc6 : SUCCESS in 44m 57s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/205c0c0818354a8dba6cae3e04004383 : SUCCESS in 2h 10m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/c80c446ccbb44fbfa3d6e977c4a291cf : SUCCESS in 2h 11m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f738f6b74a9b4af0af252f8d0fbb115e : SUCCESS in 56m 23s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/73b9941f0a9047ecac60a2f0e19ad6a2 : SUCCESS in 1h 43m 59s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1bb1cea9fd054692a52cd881d90bac97 : SUCCESS in 1h 26m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/83dd26feb9cf4120b07a471799a15254 : TIMED_OUT in 3h 03m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c38da043f7f14654b36f82a4977569f8 : SUCCESS in 39m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e21afadece034c8796850689ae5f5f14 : SUCCESS in 33m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/64a6fbab2825494dbe4117ac0b1b7abe : SUCCESS in 32m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/eafafe2485ed4ab6bc9dea7b5713defb : SUCCESS in 34m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c7718ebc36f0477e94a604e96b539998 : SUCCESS in 2h 14m 03s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/4466193a369d4de0a8be50a4b48f5497 : SUCCESS in 50m 29s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/ec6720af38d4438586df7487c685ec58 : SUCCESS in 56m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c27895a2350ee340128c3ec46141216968c09ec9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-09 18:58:49.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ef96ab6d9cff09fa171a5d8661c657754ef0e076","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-01-10 19:55:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c5df6ab7f55af86fc5f0b26dedef1e32fc2aeac7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-13 15:58:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0576a5178ac81e7d3445bffc77bc49a97f6dc13d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 15:59:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c24e7bf2e2efa3406cd90430b2c71de40528c5e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:12:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddbd0ab4f225446aaa1c0274e1bd3670\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa98f3741dd4f1e88b976ccb3c4edeb : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f2818cfdcf1428ea6d0bca277f45cd8 : SUCCESS in 2m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/88853d9c28c041b29c58eb576d5c6e93 : SUCCESS in 1h 13m 45s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/bf143ed6320d432db6a0096f76c7b558 : SUCCESS in 1h 22m 38s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/dd72fc71c89a4c03b95ae5741e7d750a : SUCCESS in 1h 29m 25s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/30e01947399e4022ac1ec72047431535 : SUCCESS in 1h 15m 03s\n- octavia-v2-dsvm-noop-api-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9cf29f10cb45402e9929c91129e2f302 : SUCCESS in 1h 24m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b7617716b3944425a9f990236c719669 : SUCCESS in 1h 49m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d02693586d924e61b64442e5bfbf54e2 : SUCCESS in 1h 09m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/2941968575114b369eee16da93ddcc12 : SUCCESS in 1h 43m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/fad1b1242ba24b35b5d792c81dfa8d56 : SUCCESS in 1h 00m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/98ff00adbb31476d9368335d3f4c0349 : SUCCESS in 2h 07m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/b1387ed72e524c05b9b1bc77ccf42980 : SUCCESS in 1h 37m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/5316887228b24bc1ab81ae7cdee2318b : SUCCESS in 1h 54m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2046d48b9d854548b1855d3ed043e46c : SUCCESS in 1h 14m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8e6f6a6940984edca9793b83e19e135c : SUCCESS in 54m 34s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/90ab74007c584f3f8c3ac91fd4e0b911 : SUCCESS in 31m 15s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/7d7478e083614f3787ef86a1ab8a3c5c : SUCCESS in 34m 36s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b64de77a19584bc6b7c33180b25788da : SUCCESS in 44m 33s","accounts_in_message":[],"_revision_number":4},{"id":"46ae0e70719276c2c20e3a6beb0252c42cbb4758","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:12:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"348fa5c9c7e9afb4eb4004a23a828a0470363b23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:14:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b92479d5d4b4cf790befc44c7a17ba0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/208ba5089e554651bb8ac0ff68624e41 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e8441d7c95fef7feb94a25376fd8c86dfaae31f3","revisions":{"8b772313df376c6d58dc4ff8ad50399d2157a7e8":{"kind":"REWORK","_number":1,"created":"2024-11-19 03:07:02.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/95/935595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/935595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/1"}}},"commit":{"parents":[{"commit":"4af9bae253bc7e8a1fde4f84d45aecbb825f10df","subject":"Remove support for Pyton 3.8 and older","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4af9bae253bc7e8a1fde4f84d45aecbb825f10df"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-19 03:05:58.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-19 03:06:59.000000000","tz":-480},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbutnu Jammy (for smooth upgrade from previous releases).\n\nThis commit keep a scenario job to run on Jammy which can be removed\nin the next cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: I8219106ec26ddc2edfaf39a69c64a9887fdfb54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8b772313df376c6d58dc4ff8ad50399d2157a7e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8b772313df376c6d58dc4ff8ad50399d2157a7e8"}]},"branch":"refs/heads/master"},"cff8df730d86212c48b6135316221dfd21955d41":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-27 01:53:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/95/935595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/935595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/2"}}},"commit":{"parents":[{"commit":"0ba927bceda1c9eb0ef12f41b2b81ca53f972e87","subject":"Update hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/0ba927bceda1c9eb0ef12f41b2b81ca53f972e87"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-19 03:05:58.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-11-27 01:53:10.000000000","tz":0},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbutnu Jammy (for smooth upgrade from previous releases).\n\nThis commit keep a scenario job to run on Jammy which can be removed\nin the next cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: I8219106ec26ddc2edfaf39a69c64a9887fdfb54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cff8df730d86212c48b6135316221dfd21955d41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cff8df730d86212c48b6135316221dfd21955d41"}]},"branch":"refs/heads/master","description":"Rebase"},"cba398c9d3584e0c6dc3b9499f2436508e9e407d":{"kind":"REWORK","_number":3,"created":"2024-12-20 19:43:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/95/935595/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/935595/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/3"}}},"commit":{"parents":[{"commit":"035e8b12d48e6b600b43dd65a66d8092041d97fb","subject":"Re-enabling tempest-plugin-jobs (pep8)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/035e8b12d48e6b600b43dd65a66d8092041d97fb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-20 19:43:27.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-20 19:43:27.000000000","tz":-480},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbutnu Jammy (for smooth upgrade from previous releases).\n\nThis commit keep a scenario job to run on Jammy which can be removed\nin the next cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: I8219106ec26ddc2edfaf39a69c64a9887fdfb54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cba398c9d3584e0c6dc3b9499f2436508e9e407d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cba398c9d3584e0c6dc3b9499f2436508e9e407d"}]},"branch":"refs/heads/master"},"e8441d7c95fef7feb94a25376fd8c86dfaae31f3":{"kind":"REWORK","_number":4,"created":"2025-01-09 10:25:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/95/935595/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/95/935595/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/95/935595/4"}}},"commit":{"parents":[{"commit":"035e8b12d48e6b600b43dd65a66d8092041d97fb","subject":"Re-enabling tempest-plugin-jobs (pep8)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/035e8b12d48e6b600b43dd65a66d8092041d97fb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-12-20 19:43:27.000000000","tz":-480},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-01-09 10:24:39.000000000","tz":60},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbutnu Jammy (for smooth upgrade from previous releases).\n\nThis commit keep a scenario job to run on Jammy which can be removed\nin the next cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: I8219106ec26ddc2edfaf39a69c64a9887fdfb54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e8441d7c95fef7feb94a25376fd8c86dfaae31f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/e8441d7c95fef7feb94a25376fd8c86dfaae31f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
