)]}'
{"id":"openstack%2Foctavia-tempest-plugin~945622","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I055cb7ae58bdc4eaf2c1eb073760b1eef725f8e0","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"34451":{"account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"last_update":"2025-04-02 19:19:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I055cb7ae58bdc4eaf2c1eb073760b1eef725f8e0","subject":"Tunning ovn-provider jobs","status":"MERGED","created":"2025-03-26 15:26:08.000000000","updated":"2025-04-02 19:21:24.000000000","submitted":"2025-04-02 19:19:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945622","meta_rev_id":"d8b5098593c37154c88b266d4fecd44f6dd806db","_number":945622,"virtual_id_number":945622,"owner":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-02 19:19:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2025-04-01 14:14:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2025-04-02 16:14:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2025-04-02 16:14:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 19:58:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-01 14:14:28.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-04-02 16:14:00.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"e896e97710775186116d4257a7e4fbd78428bc83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-03-26 15:26:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8e499e61f65456014f31e44ad0da58a5845569f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 19:58:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3c6bba8fe5f4ad3bd48850119d492fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe13fea39ab043d29cc79d1dadafe7bc : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5573729144824e00af230595b6bc0d06 : SUCCESS in 2m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3b616579d037465ab0de6c26c52f1107 : SUCCESS in 1h 28m 36s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f7a54a7fddc34c3184a1a5866ba7a8fc : SUCCESS in 1h 09m 05s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e4c33789132140eb96a0a8c7750f6d68 : SUCCESS in 1h 07m 03s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/239a7cba22ab47d48adfe1c67f3bfafd : SUCCESS in 1h 02m 34s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/5e273b3222014db78e8a1787df3748bf : SUCCESS in 1h 18m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/894f6c43f0364ec18ce25111c0a5bee5 : SUCCESS in 1h 53m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b2339cb3cc7b4c4d84e22a7a1091bb50 : SUCCESS in 1h 15m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f898a3227209435790ea43a2cce0e9ef : SUCCESS in 1h 40m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ec14aefe22c24ec88a6632aa80bb41e8 : SUCCESS in 54m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/4c0989adb235452f87c38af0747d0310 : SUCCESS in 1h 42m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/59e8751c14254f1ea384f3d80ec161ad : SUCCESS in 1h 12m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/c655e2c170484317b2549f82f25adf4f : SUCCESS in 1h 19m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/ab15c29d9f394f0b941f129be36341dc : SUCCESS in 57m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0c144f9a0adb4ed7ba3ff010c8736019 : SUCCESS in 36m 25s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a29345c8ec054c8686881cbe6d60bfb9 : SUCCESS in 32m 12s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/ce79333c305c48d184f929f9d81c8056 : SUCCESS in 34m 04s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/580a3156d71145f689428edba6aa5398 : SUCCESS in 32m 46s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6a3c3a516aff4dabbb91c7579db5b1c4 : SUCCESS in 2h 22m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/0cae1a765e7148d2946b0d1a3bbe8746 : SUCCESS in 2h 18m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c4c24815d210455f851164fdfb31371f : SUCCESS in 51m 31s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/d6f2432e5b9f426ab23894182410296d : SUCCESS in 2h 15m 35s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/536f9cf2bb284e11b26d3b43a91d3bbd : SUCCESS in 1h 25m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7f28d71121464c7ab65f5f8d799ad25f : TIMED_OUT in 3h 02m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ee8924103e3469b81b2ebf683120c47 : SUCCESS in 34m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/fc8433d194334cd9a0ebecc3361a612b : SUCCESS in 38m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8d06fd17810246fa9f9ef15ef9e7b9f6 : SUCCESS in 26m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/16e3dced881842ae9bb79257456ea403 : SUCCESS in 26m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2ce99f9f9c9c45268797bedc61364d13 : TIMED_OUT in 2h 34m 45s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/3f990cba08be495ca265a32d54c9ae5b : SUCCESS in 1h 05m 12s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/32b4c9fe85ef4674a0b6ecf694db5421 : SUCCESS in 41m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27290ca9d1f94f190aef83a37d3e881a63fb6288","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-04-01 14:14:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4409ccdbb72f0db5f8525476b2c61ed4fa93f566","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-02 16:14:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0165b014298f3e309ce2703c1a02d4b6cf8947da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 16:14:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"470eac5750095a3929de4cf9267e41f52d39db36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 19:19:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6457775920a844b2815cf18081e05d60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44ba31b700c646a88cd0fb154c41c6f3 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54f2b0fa3ff74157b4ef9a676a19869a : SUCCESS in 4m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/20db66792dc7421d960a43af7d8598ee : SUCCESS in 1h 33m 33s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/d91ee64369324a1db3c4551464e12034 : SUCCESS in 1h 04m 36s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/88f8bf786a65432daecc0c02c8140ddf : SUCCESS in 1h 18m 53s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4f8cc8caf1424b9da91e0307663e6b37 : SUCCESS in 1h 26m 56s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/0e0b4ae741a64ef5ac2efb97095df96e : SUCCESS in 1h 23m 54s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7973aa1df73844e78f005ed3c45d9c7a : SUCCESS in 1h 48m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d027d006343b432bb9afbbbae6da5d93 : SUCCESS in 1h 01m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/836fb067d6bc489ab05e7fe9f2d9003e : SUCCESS in 2h 24m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/c34f004098374effa6cdba07ad1e28e0 : SUCCESS in 1h 33m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/aac0cf8fbc2a4162b27a34f33d4fbe5a : SUCCESS in 1h 39m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/45fa9807d7004c3397ff536c746d7100 : SUCCESS in 1h 11m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/332d3d52a80349fd860c134abc8214c3 : SUCCESS in 1h 54m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/15e3405a775b4457baabcad77e9a8aa6 : SUCCESS in 58m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c9e68ad0a455425a9a00d90fef78db34 : SUCCESS in 47m 26s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/da69ea86054840bc8330f5630c514a5e : SUCCESS in 36m 28s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/30f2dac43d444f889033e4a06f105e19 : SUCCESS in 28m 06s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/8d7764364bdb4c6aba902ef0df1e0b85 : SUCCESS in 21m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4bf391c22844908ec81ad29dc7a9b49c243afe24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 19:19:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d8b5098593c37154c88b266d4fecd44f6dd806db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 19:21:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/002d3a6d8cb94c16913f77d8dbfc4bb8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b1fb55f1d5f42bea770eb465d858bb9 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2425df49882dc4f5b3aba18adb91e9d1ab4714b9","revisions":{"2425df49882dc4f5b3aba18adb91e9d1ab4714b9":{"kind":"REWORK","_number":1,"created":"2025-03-26 15:26:08.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/22/945622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/22/945622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/22/945622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/22/945622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/22/945622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/22/945622/1"}}},"commit":{"parents":[{"commit":"875c38f0355ecd31ce3a4ddf654d3adb8ea2370d","subject":"Merge \"Remove system scope token usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/875c38f0355ecd31ce3a4ddf654d3adb8ea2370d"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-03-26 15:24:19.000000000","tz":60},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-03-26 15:24:19.000000000","tz":60},"subject":"Tunning ovn-provider jobs","message":"Tunning ovn-provider jobs\n\nFrom last executions we can see many TIMEOUTs. To minimize this\nbehaviour we increase a little the timeout for the ovn-povider\njobs and also add a regex to the scenario one to ensure only\nrelated tests are executed.\n\nChange-Id: I055cb7ae58bdc4eaf2c1eb073760b1eef725f8e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2425df49882dc4f5b3aba18adb91e9d1ab4714b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/2425df49882dc4f5b3aba18adb91e9d1ab4714b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
