)]}'
{"id":"openstack%2Foctavia-tempest-plugin~946803","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Icc89010f7be92639bbc23056236c5e422f6406a0","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"multi-vip","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-09-25 13:27:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icc89010f7be92639bbc23056236c5e422f6406a0","subject":"DNM/WIP Testing ovn-provider job with multivip","status":"ABANDONED","created":"2025-04-09 18:09:47.000000000","updated":"2025-09-25 13:27:19.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6f9e78aee15b20494c9628a8e9059f475bdb8e17","_number":946803,"virtual_id_number":946803,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-23 16:18:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-09 18:55:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-10 11:34:45.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"CC"}],"messages":[{"id":"217d2dfaf0989f22c51898c7862e64c3b3138f12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-04-09 18:09:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60f51bdaf9d182390500f2408528b7657c05ef65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 18:55:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc1a2253c18f45fbb44a46ece67f0cac\n\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1fdce44dc53f41e0bf5ed2576ccca71f : FAILURE in 32m 16s\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/dd0bcdf7918441d4a0805f0f582732cd : FAILURE in 44m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea72ae203d13d6bff09a6e7b264e62d364258268","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-04-09 19:26:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"47f9972b23c0c71f9fcf1e93338182d7d78f80b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-04-09 20:03:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"65b008d90b657be6c513a6bf0c304f4ed3e85947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 21:06:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/503f701ce91442099ec1fad551c9e92f\n\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a610ef93358148cea3fb8cec07409939 : SUCCESS in 1h 00m 10s\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b43d5c621d934fbc9c608e82f43305b1 : FAILURE in 50m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7b9fd74e146752a946b8d3619f820a00da09f6a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-04-10 06:08:22.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"0d14fe9f44bab00b0a4a2e605d8deb4e8e8320ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 07:09:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f607241635b475caf18d4d593e472b6\n\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/29953c67cbae4b13babc4eadf28b99b1 : SUCCESS in 58m 17s\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/560914bbd1a64a8ba87e26df12a5f461 : FAILURE in 54m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3c2adeae6b3057231dfdd0e82a2646f7ae75b1b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-04-10 08:41:37.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"a62c79cf293fa16da84b85de2b255e7a84a73719","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-04-10 08:47:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fa3c1ba3a731df571daa06990ecc29e399e1cef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 09:24:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36815469e79f4f7ba2b9be869c179b15\n\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/254bc30c776544209f353283fd764823 : SUCCESS in 25m 43s\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7abe3153ef92403aa33c8e779aef87c5 : FAILURE in 35m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4f0c5789f5e73699b7564de6199000dcb71ed736","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-04-10 11:34:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"22108a69e43b101d0fec3be371c510fd26b132fd","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-04-22 14:53:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0833c49172f6d878ab61f66e487305636cfd3292","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 15:49:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73c9c84d830e44d697c5c14c3b2b42fe\n\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1ef88066cc7544f89ed7cd294b73783b : SUCCESS in 39m 58s\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e6d6f733eb414aedaf23d4c72750ef15 : FAILURE in 55m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c1d328d76f374c00a281fb2dc5c4d19900f1b664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-04-23 15:14:57.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7bb50e5103c6da5af7acf3d56da3afa0c79971c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 16:18:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6952be66caa04d74b3009509be76f3a6\n\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5842165724344a6c9bc188ae81464350 : SUCCESS in 25m 57s\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/6854d40f95c24cf3be4738896d2952f6 : FAILURE in 58m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6f9e78aee15b20494c9628a8e9059f475bdb8e17","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-25 13:27:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"bd173678304e9ca2d6326e4af7490af3f7aeae59","revisions":{"fa0c1a5a8b668bc0245b1a469bff5868f350928e":{"kind":"REWORK","_number":1,"created":"2025-04-09 18:09:47.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/03/946803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/03/946803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/1"}}},"commit":{"parents":[{"commit":"b18ba59731675da25dc987cc9bb5ca9939d6a500","subject":"Add tests for additional VIPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b18ba59731675da25dc987cc9bb5ca9939d6a500"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-09 18:01:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-09 18:09:30.000000000","tz":120},"subject":"DNM/WIP Testing ovn-provider job with multivip","message":"DNM/WIP Testing ovn-provider job with multivip\n\nChange-Id: Icc89010f7be92639bbc23056236c5e422f6406a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fa0c1a5a8b668bc0245b1a469bff5868f350928e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fa0c1a5a8b668bc0245b1a469bff5868f350928e"}]},"branch":"refs/heads/master"},"aa628054f1da6c511e7b44e6c23a8e240247a8b2":{"kind":"REWORK","_number":2,"created":"2025-04-09 19:26:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/03/946803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/03/946803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/2"}}},"commit":{"parents":[{"commit":"b18ba59731675da25dc987cc9bb5ca9939d6a500","subject":"Add tests for additional VIPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b18ba59731675da25dc987cc9bb5ca9939d6a500"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-09 18:01:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-09 19:25:56.000000000","tz":120},"subject":"DNM/WIP Testing ovn-provider job with multivip","message":"DNM/WIP Testing ovn-provider job with multivip\n\nChange-Id: Icc89010f7be92639bbc23056236c5e422f6406a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/aa628054f1da6c511e7b44e6c23a8e240247a8b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/aa628054f1da6c511e7b44e6c23a8e240247a8b2"}]},"branch":"refs/heads/master"},"d79f08bf5e315c267ffa5c8e9c97b1b0607793a5":{"kind":"REWORK","_number":3,"created":"2025-04-09 20:03:59.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/03/946803/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/03/946803/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/3"}}},"commit":{"parents":[{"commit":"b18ba59731675da25dc987cc9bb5ca9939d6a500","subject":"Add tests for additional VIPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b18ba59731675da25dc987cc9bb5ca9939d6a500"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-09 18:01:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-09 20:03:52.000000000","tz":120},"subject":"DNM/WIP Testing ovn-provider job with multivip","message":"DNM/WIP Testing ovn-provider job with multivip\n\nChange-Id: Icc89010f7be92639bbc23056236c5e422f6406a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d79f08bf5e315c267ffa5c8e9c97b1b0607793a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d79f08bf5e315c267ffa5c8e9c97b1b0607793a5"}]},"branch":"refs/heads/master"},"7417520ec851e8d86d5132e89f8e8cad8e3f4540":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-04-10 06:08:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/03/946803/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/03/946803/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/4"}}},"commit":{"parents":[{"commit":"b18ba59731675da25dc987cc9bb5ca9939d6a500","subject":"Add tests for additional VIPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b18ba59731675da25dc987cc9bb5ca9939d6a500"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-09 18:01:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-10 06:08:05.000000000","tz":120},"subject":"DNM/WIP Testing ovn-provider job with multivip","message":"DNM/WIP Testing ovn-provider job with multivip\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/946866\n\nChange-Id: Icc89010f7be92639bbc23056236c5e422f6406a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7417520ec851e8d86d5132e89f8e8cad8e3f4540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7417520ec851e8d86d5132e89f8e8cad8e3f4540"}]},"branch":"refs/heads/master"},"f8f92e2cc5ef60d03f01c50999c7e1c36b0f3ffc":{"kind":"REWORK","_number":5,"created":"2025-04-10 08:41:37.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/03/946803/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/03/946803/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/5"}}},"commit":{"parents":[{"commit":"b18ba59731675da25dc987cc9bb5ca9939d6a500","subject":"Add tests for additional VIPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b18ba59731675da25dc987cc9bb5ca9939d6a500"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-09 18:01:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-10 08:41:30.000000000","tz":120},"subject":"DNM/WIP Testing ovn-provider job with multivip","message":"DNM/WIP Testing ovn-provider job with multivip\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/946866\n\nChange-Id: Icc89010f7be92639bbc23056236c5e422f6406a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f8f92e2cc5ef60d03f01c50999c7e1c36b0f3ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f8f92e2cc5ef60d03f01c50999c7e1c36b0f3ffc"}]},"branch":"refs/heads/master"},"ee1a8fe3d46dba9d3e7afa9f8bafd82d650b60d7":{"kind":"REWORK","_number":6,"created":"2025-04-10 08:47:37.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/03/946803/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/03/946803/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/6"}}},"commit":{"parents":[{"commit":"b18ba59731675da25dc987cc9bb5ca9939d6a500","subject":"Add tests for additional VIPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b18ba59731675da25dc987cc9bb5ca9939d6a500"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-09 18:01:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-10 08:47:30.000000000","tz":120},"subject":"DNM/WIP Testing ovn-provider job with multivip","message":"DNM/WIP Testing ovn-provider job with multivip\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/946866\n\nChange-Id: Icc89010f7be92639bbc23056236c5e422f6406a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ee1a8fe3d46dba9d3e7afa9f8bafd82d650b60d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/ee1a8fe3d46dba9d3e7afa9f8bafd82d650b60d7"}]},"branch":"refs/heads/master"},"bd173678304e9ca2d6326e4af7490af3f7aeae59":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-04-23 15:14:57.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/03/946803/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/03/946803/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/03/946803/7"}}},"commit":{"parents":[{"commit":"9d6078e8e6b9e42299cdd0ee5677aca949718a16","subject":"Add tests for additional VIPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9d6078e8e6b9e42299cdd0ee5677aca949718a16"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-09 18:01:56.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-04-23 15:14:36.000000000","tz":120},"subject":"DNM/WIP Testing ovn-provider job with multivip","message":"DNM/WIP Testing ovn-provider job with multivip\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/946866\n\nChange-Id: Icc89010f7be92639bbc23056236c5e422f6406a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bd173678304e9ca2d6326e4af7490af3f7aeae59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bd173678304e9ca2d6326e4af7490af3f7aeae59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
