)]}'
{"id":"openstack%2Foctavia-tempest-plugin~947110","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I0f91080b86a9193ad6d5c4c73e3981ab3777737f","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-16 09:28:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f91080b86a9193ad6d5c4c73e3981ab3777737f","subject":"Add stable/2025.1 job","status":"MERGED","created":"2025-04-14 12:34:12.000000000","updated":"2025-09-16 09:30:26.000000000","submitted":"2025-09-16 09:28:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947110","meta_rev_id":"0bac4b88c00acaa6d5ccf0b34d6333512046e4d4","_number":947110,"virtual_id_number":947110,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-16 09:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-22 21:35:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-09-16 07:34:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-09-16 07:34:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-14 15:42:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-22 21:35:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-09-16 07:34:52.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"ab63209809c4dde6908d3801a5454c1a8f60d402","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-14 12:34:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63f3e23ab4175ec3095ad3224d0d08567f22e2c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 15:42:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6e12fbd72a9469bafc4211c8799ddef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/838d8b7205fe47328c356e10d7a25c1d : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e4da6dd8d24cb4839e6456ed5ca8dc : SUCCESS in 2m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/553f3ec9d27145edb000b57f8c6bc04e : SUCCESS in 1h 21m 44s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/9784acb9534349d39613bce2267384e8 : FAILURE in 1h 02m 40s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b48d64535335444db33680ff85006a83 : SUCCESS in 1h 08m 38s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/46eb507090224f9389c7bf76528bdb08 : SUCCESS in 58m 54s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6feb9b0656094535886cabb3ebb4cce5 : SUCCESS in 1h 02m 17s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/f3137741669841cba0b31f04a3b0fda2 : SUCCESS in 1h 13m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/073d2dd1367948b789d114ebdd37def3 : SUCCESS in 1h 45m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cf0cbdc73f874656abc538cca7efc35a : SUCCESS in 1h 22m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/469b52b501054cb7aad12f2eacfff456 : SUCCESS in 2h 24m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/e1871787a990424ca90dee1cefd413e0 : FAILURE in 56m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b17a08c1edd0436b886b432fddd1989d : SUCCESS in 2h 20m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1be0c19a9cd94af8a5afdc70d69d1349 : SUCCESS in 1h 28m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/379ec29834df44a7a2897499d6e122dd : SUCCESS in 1h 37m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/4924a557a52b49fe9b18a901f6addee9 : SUCCESS in 48m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/42d079606ce443eba411ab386ee60214 : SUCCESS in 1h 41m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/a2bb546719b446e1afd7fae8bfbae03d : SUCCESS in 1h 13m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b0b7ed0fe872417b8fc89e2c00fc9a04 : SUCCESS in 45m 04s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/901a25e25b8449c480b322dfd35263e2 : SUCCESS in 34m 41s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/93d0ab20ac924aafb6316245a87ab226 : SUCCESS in 50m 02s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/4a7cb437fbd14b299379722b0059e01c : SUCCESS in 48m 54s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6a899fa64a954ee7878c24fabd103ed1 : SUCCESS in 43m 10s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ed2260b3674a4fa08fae5e359e0716d8 : SUCCESS in 2h 20m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/1ccae7f673a74cbd992e09dccb850043 : SUCCESS in 2h 19m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cd2491fd67214376865ac8b8e44054be : SUCCESS in 1h 22m 49s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/45021c48d24e48ba90ab417b43e779b3 : SUCCESS in 2h 26m 07s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6edf2c9117fa469bb1104ff79841a882 : SUCCESS in 1h 27m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9568b2a99cff447486f9a511c270a3b5 : TIMED_OUT in 3h 03m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecf5b70f751049a0995a5dd8b6d43c71 : SUCCESS in 31m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/135ef313791c4b1196278c692059ad4a : SUCCESS in 34m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/5fbaceae59ca493492189eba142b7e08 : SUCCESS in 30m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e12d3ece0e2d41f08b8690e1dfb82e3e : SUCCESS in 31m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b6489fde5734430d9accd0d9c9f3d534 : SUCCESS in 29m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/20873ec423c543aebe89efedd8f23b2e : SUCCESS in 2h 30m 59s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/a218dab64f8d4676a9b8357db163691b : SUCCESS in 1h 07m 38s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e60e1344b6dc4bdea36a5032f85ef2dc : SUCCESS in 36m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36cbfabec03d088cf4397e113d5824397202e9dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-15 01:17:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"de5e446130091e0b855748ec736278c3485de235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 04:21:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ef9e5a1e2ee4288a836935ebc83e15c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc8bdb9e13ab41e98d2d266b759b106d : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41b3abb9913477c832431969a3dcf30 : SUCCESS in 2m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b4665972d97b4339bb735d406ce431eb : SUCCESS in 1h 07m 02s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/96d9afe4b6684d2e838a7f4b4eef45af : SUCCESS in 1h 40m 39s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/9503ea0b63e54794b2c145a3bcd6dd68 : SUCCESS in 1h 11m 58s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/7cffdd7f50694dc4bf3c844215939dc3 : SUCCESS in 1h 23m 56s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6ef3dc1bd6994c6fba24947abc860901 : SUCCESS in 1h 17m 05s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/c05d7bb747ef48fc94ca7c02e9792b46 : SUCCESS in 1h 04m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/10eac66fce4940be97ab47e4d244c763 : SUCCESS in 1h 58m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8c3cf588a11944398802aed561640ad4 : SUCCESS in 1h 15m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/e856fe96bef7492f9e0f522c7a57094a : SUCCESS in 1h 54m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/169e9cc8d37e48f9aa4c596fc21a8e41 : SUCCESS in 1h 07m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/9eb3887d8b204d5db5d1a6c2d13d8ea1 : SUCCESS in 1h 56m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/86b2570039bb4365b6bec01b04b1db47 : SUCCESS in 1h 24m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/a7c9573654874394b2e67ce16a4beae1 : SUCCESS in 1h 16m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3c46f754509243c281fdef89a8ea460e : SUCCESS in 1h 10m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/f8e8d275ab744fe684842932cd72621f : SUCCESS in 56m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/33d807201ffc4bd2993afe45cfc94f94 : SUCCESS in 42m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6dece13a2f134d8fb65455f9167561c8 : SUCCESS in 25m 11s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/6987106472994cc1986403c4017011f6 : SUCCESS in 38m 52s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/03c30b30e45444c3ae6d76be34eeb010 : SUCCESS in 25m 29s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/6a45c7cca1784000939aa3a51b1773e5 : SUCCESS in 24m 25s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6dcd3c717eb549ee975b3787edd3aada : SUCCESS in 32m 57s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/30c50135580d4e5b98d7f6e409cff8f9 : SUCCESS in 2h 28m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/00ad6887d8ab4acc9e5649f82224148d : SUCCESS in 1h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9fc6f2774f084755bbf98d3761e6706e : SUCCESS in 54m 39s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/239e8be9b25449ffa375c1f57346b932 : SUCCESS in 1h 32m 00s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7b99de39daa748d2a230fa033dd33987 : SUCCESS in 50m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/072e7a015bae417bbd0a30799e918030 : TIMED_OUT in 3h 03m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d415f13bc4943568ee49879abc4fe04 : SUCCESS in 36m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/eb1031b8946c4403a959cb3c1eab069e : SUCCESS in 33m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/6e59a54d8ced4424a60bf1c1bb18367e : SUCCESS in 34m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/0f6f3654aa8e4211b6b572b456f17d27 : SUCCESS in 21m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/31b39124c80a49e0ac16462474e0c307 : SUCCESS in 28m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1b715b75255e4bc4bf6c2af9079ef7c5 : TIMED_OUT in 2h 33m 11s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/19c9dc138e404a59a42de666396b4c26 : SUCCESS in 1h 16m 46s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/f966cb0d33974997a5a726469173c7f2 : SUCCESS in 40m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"301c54dc1a4fca4a1a0571c0c8cdf3fc5a6b6b95","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-22 21:35:04.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9b5b1a2f711d8d1dac7d3fbc67eaf005c95031e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-16 07:34:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dfa5f1b71c90bd8d5d9736a12cd3f14da7d1474d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 07:35:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e108bea2b1d6f8d5353d20864770c18d177a83a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 09:28:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78cb9571bd70499a838cf97db5bb1c1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/247ddf6ea805417a930842ce20894607 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22097f46b1364f5db71a7e1407ce1277 : SUCCESS in 4m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98f9ac14950d468aa1d069909de51085 : SUCCESS in 1h 38m 00s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/e39ba6edaf1646a298629abbdc1f2a87 : SUCCESS in 58m 54s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/099be7a37faa463cb2059fca703de983 : SUCCESS in 58m 44s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d33af35d40994ee7acd8683cd31cb8db : SUCCESS in 1h 10m 05s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/95de63ecffce4b2782503c035ca34dc0 : SUCCESS in 1h 32m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3eec866b14b5403f96e86b84be183d5d : SUCCESS in 1h 53m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c92b7084a5b34531bcb7052619d26913 : SUCCESS in 1h 07m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/297944f7953a458e8a8f475b6154653f : SUCCESS in 1h 43m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/cef29dcc6c814c479d1e36eda5bb221a : SUCCESS in 1h 06m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f4bcfdda297b45cd8459fd10c2e995db : SUCCESS in 1h 39m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4334b48bc088498e8ea80af96d7b23c6 : SUCCESS in 57m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/c0fb69730b8f46a286d78b7148500169 : SUCCESS in 1h 38m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/cb0e81f203204672af12cc6d699fc70c : SUCCESS in 54m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b3944ca75be7440b83ac4bcaf9dd8f55 : SUCCESS in 34m 29s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/291e53c587eb4dbfa5ed9e77031c9e38 : SUCCESS in 35m 59s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/be75c16017204da583d6ad56aa0c43c8 : SUCCESS in 30m 29s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/81786dd7fad443d1bb67a740a45d5731 : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":2},{"id":"29d0a439b84265e1b62b3fd60ee18c5dad1ec54e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 09:28:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0bac4b88c00acaa6d5ccf0b34d6333512046e4d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 09:30:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ec768ef467243f9a954473cfec3e160\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa7222da63745a0b90179478a3bcaa2 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"744a7a827366490e958157dd8f8f7af5334e8eda","revisions":{"661fe4a9b0635025f0cfae25e866e316eb7cb226":{"kind":"REWORK","_number":1,"created":"2025-04-14 12:34:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/947110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/10/947110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/947110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/947110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/947110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/947110/1"}}},"commit":{"parents":[{"commit":"875c38f0355ecd31ce3a4ddf654d3adb8ea2370d","subject":"Merge \"Remove system scope token usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/875c38f0355ecd31ce3a4ddf654d3adb8ea2370d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-14 12:34:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-14 12:34:00.000000000","tz":540},"subject":"Add stable/2025.1 job","message":"Add stable/2025.1 job\n\nChange-Id: I0f91080b86a9193ad6d5c4c73e3981ab3777737f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/661fe4a9b0635025f0cfae25e866e316eb7cb226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/661fe4a9b0635025f0cfae25e866e316eb7cb226"}]},"branch":"refs/heads/master"},"744a7a827366490e958157dd8f8f7af5334e8eda":{"kind":"REWORK","_number":2,"created":"2025-04-15 01:17:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/947110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/10/947110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/947110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/947110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/947110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/10/947110/2"}}},"commit":{"parents":[{"commit":"875c38f0355ecd31ce3a4ddf654d3adb8ea2370d","subject":"Merge \"Remove system scope token usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/875c38f0355ecd31ce3a4ddf654d3adb8ea2370d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-14 12:34:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-15 01:17:05.000000000","tz":540},"subject":"Add stable/2025.1 job","message":"Add stable/2025.1 job\n\nChange-Id: I0f91080b86a9193ad6d5c4c73e3981ab3777737f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/744a7a827366490e958157dd8f8f7af5334e8eda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/744a7a827366490e958157dd8f8f7af5334e8eda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
