)]}'
{"id":"openstack%2Foctavia-tempest-plugin~947959","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I700248094c7b81ba9cc28e8507d4f5dd2b3fa1c2","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"multi-vip","attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-09-18 10:58:03.000000000","reason":"\u003cGERRIT_ACCOUNT_34451\u003e replied on the change","reason_account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}},"34067":{"account":{"_account_id":34067,"name":"Lilach Avraham","email":"lavraham@redhat.com","username":"lavraham"},"last_update":"2026-09-18 10:58:03.000000000","reason":"\u003cGERRIT_ACCOUNT_34451\u003e replied on the change","reason_account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-09-09 13:59:28.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"34451":{"account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"last_update":"2026-09-18 10:58:03.000000000","reason":"\u003cGERRIT_ACCOUNT_34451\u003e replied on the change","reason_account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}}},"hashtags":[],"change_id":"I700248094c7b81ba9cc28e8507d4f5dd2b3fa1c2","subject":"Improve check member balanced for traffic not a multiple of members","status":"NEW","created":"2025-04-23 15:14:57.000000000","updated":"2026-09-18 10:58:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"95d3ced61b5f5958e925a80be7c255246f601ab8","_number":947959,"virtual_id_number":947959,"owner":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":34067,"name":"Lilach Avraham","email":"lavraham@redhat.com","username":"lavraham"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-09 18:34:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34067,"name":"Lilach Avraham","email":"lavraham@redhat.com","username":"lavraham"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":34067,"name":"Lilach Avraham","email":"lavraham@redhat.com","username":"lavraham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34067,"name":"Lilach Avraham","email":"lavraham@redhat.com","username":"lavraham"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-23 18:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-25 13:28:15.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-09-18 10:58:03.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34067,"name":"Lilach Avraham","email":"lavraham@redhat.com","username":"lavraham"},"state":"REVIEWER"},{"updated":"2026-09-18 10:58:03.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"7e948d3519688a95e70eb47cd900400d93bd063c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-04-23 15:14:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fc36e1bcedaebde3aa47a47d60035c80b986e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 18:26:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2017224a58af4ba9a48973943bed0d9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c00d2e5d77a14f5d9a8ffbbe495fadce : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a48bb360bc844a85a0cbd7a70cce9179 : SUCCESS in 3m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/477926b454204035b3055f11c74d4dad : SUCCESS in 1h 05m 39s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/c45cebb53dfe49e9abe2e694c5149cbd : SUCCESS in 1h 12m 44s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/668f18836d87419a8f0148829b82873f : SUCCESS in 1h 12m 29s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/d5229648cd2b4a998401415be12e2186 : SUCCESS in 1h 16m 31s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/51d1205272a24ae7beda812e6da0cefd : SUCCESS in 1h 09m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8ff994b0f69c4844b0934f28d26d2568 : SUCCESS in 1h 53m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7756b63598ef4cbd859e77cdc3b031b6 : SUCCESS in 1h 05m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/17dc371c92dc4e679c87367c027fbd53 : SUCCESS in 1h 43m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4a3a2c72c5c74f788a7463a17afc6d37 : SUCCESS in 53m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/00418ac07294413b9c4da5970b1ad413 : SUCCESS in 1h 37m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/f4dd29a17e554718aee6c37742408e15 : SUCCESS in 1h 13m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/bf5960d7be0841c18ed686ec69f5d605 : SUCCESS in 1h 18m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4f32aaf9f3a84743be009d0a8f2794b9 : SUCCESS in 1h 03m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fa7dc4bb5041456a909ec40d000d28b7 : SUCCESS in 46m 26s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e86c4beb01c5411a9f883c26fb647f23 : SUCCESS in 33m 01s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8430a09fc1744cafb5a3d1498ccb2466 : SUCCESS in 39m 25s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/11c19f9be5684b448de8954991a9908e : SUCCESS in 34m 25s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/854b6c440b0a43e38ab32337e30d4033 : TIMED_OUT in 2h 33m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/c60f1c07d6a9427b959b5cb26039aa54 : SUCCESS in 2h 01m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fd94e6883c7f4ae7983d9517f5905996 : SUCCESS in 1h 14m 38s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/9bbe1f98e4e24f7798c84b8351ebc8a6 : SUCCESS in 2h 08m 19s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1fc65fa5079541bc99acbf8abe39f7bc : SUCCESS in 1h 14m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/29b67c9be83849c8b053f50547be9ff5 : TIMED_OUT in 3h 03m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9a055a6cbec462682b32d76eee4005d : SUCCESS in 22m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/da0057c10224471caae107c22f8d67ca : SUCCESS in 29m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/692f3342523d42f48cc5f854d6f2da03 : SUCCESS in 29m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/53ab80f84da5459995512d8d031b60af : SUCCESS in 26m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c672ae6329c34925955937387c258be6 : TIMED_OUT in 2h 33m 09s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/c6e2af12c53e48eb9d0f9e311cdb119c : SUCCESS in 1h 03m 21s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/f01e3bb873b64508bbd542f07dc49be5 : SUCCESS in 46m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bea1563e3e2e948703f4f6d9abab84a829419b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-25 13:28:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6f2ac47a5f63d25058155448ffde06c2de51ed99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 16:36:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20ec11bfe62449e8a0a5efd3465f5a9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85533209fce34f68870bb88368def3e9 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/598760ccae534a4ebc2e6e370fe6c73c : SUCCESS in 3m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bee7777e1c454b64bb3fbd2da269585b : SUCCESS in 44m 25s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/dc0c70b1e7db47cb9157283ab1280397 : SUCCESS in 59m 50s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/9b039c2ef8df4f3690a31b722d603245 : SUCCESS in 1h 00m 09s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/ad95717d92a046f29de626c403a00822 : SUCCESS in 1h 06m 49s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/01f9f4e8612848b792f7ddaee6eaec70 : SUCCESS in 1h 11m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/86a8555d27b344509f8750c7e05c4c74 : SUCCESS in 1h 58m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/24ca66e04cf24ee487712e96896c673b : SUCCESS in 1h 06m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/66c5f7a9970647b7bc6f744d8a1ab871 : SUCCESS in 1h 44m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/3d661ada6fb34bd0b400b141174c42f0 : SUCCESS in 1h 23m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/6ca01c91bc754b6895837e2b7ddbec65 : SUCCESS in 2h 00m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/90b5734a918d4e458cce95b668701726 : SUCCESS in 1h 13m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/19bf6e24a48c491ca0e5e522c473a87f : SUCCESS in 1h 38m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8c7a1ff37be7404e831770076492862a : SUCCESS in 1h 14m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a2509c21f7004e85aef98aff8f529622 : SUCCESS in 43m 57s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/f93f1f4fe42e4da48168eda3f9322135 : SUCCESS in 59m 04s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ff78c0828e6342b2aad2a2aeb66f7ff8 : SUCCESS in 31m 45s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1957513c403d48489bf59e9516035221 : SUCCESS in 48m 50s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2ad772f38d3e432cba2b5426aceeae87 : SUCCESS in 1h 54m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/3fa5ab72e95e48b1a531b0d1df20af4c : FAILURE in 14m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1038684c0b9c4160ac40fb6c34141b48 : FAILURE in 12m 50s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/0417c78b69784efe963099530c79d84a : FAILURE in 8m 17s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7c971e0fc6df4b4386e2f3e7986476fd : FAILURE in 17m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c4d14d1fef6f4b8792ac978ccb1bebf8 : TIMED_OUT in 3h 05m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a233e23395cf48c980e2122fb5acad5c : SUCCESS in 33m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/743cdd08c9174edf917b2938efed8cc7 : SUCCESS in 33m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/bdcf1fd1ca6c4107b52c7555e5077f05 : SUCCESS in 30m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/10fb5bac8de144f9a46d7d726e2b8953 : SUCCESS in 29m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d404d52b5c99478d9eba812b2349fd5a : TIMED_OUT in 2h 33m 15s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/26b43640f2774e73b5a79a8e749beee7 : SUCCESS in 1h 03m 34s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/cefaf2dbfce34762bd3314a4d4c72cbb : SUCCESS in 48m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08c773663f24f0c4fe49cacdbf6d0ca44cfaa3e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-10-03 16:15:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"df49b9b3b0c9355c66d45a8ec95de222edfe4f1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-10-03 17:00:18.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"bfccfb238a8cff1529110d2fc137645589928c00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 19:36:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bae87ea15f445e48fe02b3d5d1eab96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17883d6224294d1bae3ae2eacfec00ab : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f943a8c4d50549d3af4b26a964a108bb : SUCCESS in 4m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d623211f13254217901ce24dd266e3dc : SUCCESS in 1h 22m 30s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/d4c66ccbcf454c888149a26d2262f3ab : SUCCESS in 1h 05m 53s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/db6585c63b5f4f9ab67dad0c3f5c2fd6 : SUCCESS in 58m 46s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/c8a6be61ce4348e6bfe712e077f663be : SUCCESS in 1h 12m 26s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/3f87646d228c4cc092276302b483b894 : SUCCESS in 45m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a776dcb0b7524ad99faa0b3ad8c10292 : FAILURE in 2h 02m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/934f725907784fcbbd12d35d606045a4 : SUCCESS in 1h 15m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/4635907a4c8d4a2097fd1f210b5f742d : FAILURE in 1h 06m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/68e090a80c06430293cc9886e051d194 : SUCCESS in 1h 15m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/16bbd372e8bb4373bef7fedb50425999 : FAILURE in 1h 48m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/195e80689a904ff1b2b2c19720781a3d : SUCCESS in 1h 19m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3ef6ef1704364a36a93da7d2e2c6fcf9 : FAILURE in 1h 50m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/a9c94f316fd34dd3824dd6ad317910d0 : SUCCESS in 58m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b14955a41d334718ac935465aef57fb2 : SUCCESS in 45m 06s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/7b7e0b45af384d6c9fbd3ef328817144 : SUCCESS in 38m 21s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/51edcf4255a647719e9a3b68b95fe64e : SUCCESS in 34m 11s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/4aca3eed34794f38b08576dc18d7b84f : SUCCESS in 42m 06s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2706907db46e4b898d13e673d77be07e : SUCCESS in 2h 33m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/b4a8f6ee81e044a8ad56760fde5c62cc : FAILURE in 13m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/023683d67aed4e6884ab563bc77a828a : FAILURE in 10m 50s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/1242343128b943d7aeedca085c7edc51 : FAILURE in 9m 27s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e1e687d09afc409799c1dae9c27bc60a : FAILURE in 10m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/600d8e8d8d1d4d28bb43399d9f929048 : FAILURE in 2h 23m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d618ff79de3f471e96ec69942906eb20 : SUCCESS in 37m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/745c18f425dc4d388ec24eed73b6ad50 : SUCCESS in 42m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ea315a3e42224aeeaeee767f75708df3 : SUCCESS in 34m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/a1ee69badbf24cde91a9a61768200798 : SUCCESS in 33m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4c711c2f446545c7b3f47fbb0ee46fc8 : TIMED_OUT in 2h 33m 15s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/50265dc3e0534121a264493a0ee4038a : SUCCESS in 1h 03m 49s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/20afab20a17b49b0bb13f1be2da04bb5 : SUCCESS in 24m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b725662973258678a72e7e7ea0c8ccf247b79d16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-10-06 12:21:37.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a4388f392df0df81c5cb4955d9701aa6ad77a6f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 15:28:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6af9f02d13cb4c8f813deab401409272\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8e9bdf3148e4c42abb98e0e2974ca47 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0b3bd12dee24006b78e8bb554fb7966 : SUCCESS in 2m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f1d183eb82de43ee8702928009985d7d : SUCCESS in 1h 08m 08s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/f31e3fcda7154dec89033b321acdbecc : SUCCESS in 1h 16m 00s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/afc2c64a4b9140b292b5db62f4576d79 : SUCCESS in 56m 55s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/b670d341d9e9481a927d286cb15d37d6 : SUCCESS in 1h 10m 03s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/0fad9d77b2384ea89266e7f8eec17d9b : SUCCESS in 1h 27m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cfc3d15d78dc402baa40f5851da234c5 : FAILURE in 2h 11m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4090195f8d1a4d64bd9c4207e8ed90e4 : SUCCESS in 1h 12m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/110d0f4e313d44229b07aa7288a5a765 : FAILURE in 2h 11m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/5f4bfbbc2cb649b0a4a8a16802f35685 : SUCCESS in 1h 06m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/5e4c98f4ab804ba8b754024ba4874655 : FAILURE in 2h 01m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/61119f36a0b84bf4887ff150ede712af : SUCCESS in 1h 03m 13s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3c4914f17f90488a892068e82bb16ff4 : FAILURE in 1h 59m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/dfde1e5cb3ff4d758e8212b5f9348580 : SUCCESS in 1h 03m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/be74a46ceda74ce6965b5fae1b5ff9e6 : SUCCESS in 43m 31s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/28626c2e00e9434f846e4aa57c5e4696 : SUCCESS in 36m 46s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f0760daeda364779bb1b01d03bd4f648 : SUCCESS in 33m 23s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/008e16575c1d402a88089e2bb1a04b19 : SUCCESS in 34m 29s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/97beee4c5cc94ce487692a8b3a489f5a : TIMED_OUT in 2h 33m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/a745da4b3fd743249e8018377ae92825 : FAILURE in 9m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/255621b59cff43e29da89e1fb96d8234 : FAILURE in 10m 07s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/63ff91f2803444c2919833b106b82fbf : FAILURE in 8m 00s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d36b83c5f0c6456581fa2d06ec4fa772 : FAILURE in 14m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dce79615f1c3428e839f7837edc174a4 : TIMED_OUT in 3h 03m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/303bce9b53584969b7bde219b3212586 : SUCCESS in 36m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/b35b461d9ea84acbac0123e18c1ff4dd : SUCCESS in 42m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/3132b6719c1249289b48b830e6170702 : SUCCESS in 38m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1878f6c46fcc4582b65057f43bcc5432 : SUCCESS in 31m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/faf7f58a5ea44b9c82e697227ecf27eb : TIMED_OUT in 2h 33m 18s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2e286ab8c63b4ea19080a7bb7972dc52 : SUCCESS in 44m 25s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/50b4bc0a1dfe43d79964447d99363d8b : SUCCESS in 39m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"35f6aaeeb449eb8bf0b610346408819189169fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-10-06 15:35:55.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"abb7ef5e23e9075f1b56fdf6b543fb1da9d85b2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 18:42:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ba35363735942039b8873a4f5d55a77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cce63a20c6249faa1b928aae4559fe6 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88b627e5e0a74b6da5f50427f2d1d76c : SUCCESS in 2m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b3501749966a4cfc8e961282c7f87bfe : SUCCESS in 42m 04s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/29a77ea7d28f4786963dfb5bab969c8a : SUCCESS in 1h 10m 50s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/53eadf2948b248a3b6d54fc21bc13d4f : SUCCESS in 1h 08m 52s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/39136820088d41629154f364e4f5f7e6 : SUCCESS in 59m 57s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/ec5fc8a921f5454492e75903fa049324 : SUCCESS in 44m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f861976b3c1543719dad59482010ac23 : SUCCESS in 1h 59m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3584a59e2ea64172a7d26bd6122948f8 : SUCCESS in 1h 19m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/be3a2e7cd7cf4f37b53d5a43a5c12d19 : SUCCESS in 1h 48m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/7d0a1c5c8dc14d71bb0dc3f51f55f3ea : SUCCESS in 1h 01m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/d2b4d14f160c41c686d03d5e474f5c6e : SUCCESS in 1h 44m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/181277527e804f8fb385af3fac7e6195 : SUCCESS in 1h 07m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d30525bc88fc4789af3d5aadd837543b : SUCCESS in 1h 46m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1906cd86291945c184fda7177a9ea248 : SUCCESS in 1h 02m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/08e853cd0890435c981122db1117354e : SUCCESS in 37m 51s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/4726cf113ed8446b89a4e7ebf0c7b58a : SUCCESS in 42m 15s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/311a336ac03a40379b1777595a332824 : SUCCESS in 31m 27s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/a78fd4b513a94c29b053e4042c209c4d : SUCCESS in 38m 25s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f918e045b0624715b60d7145ff1045d7 : TIMED_OUT in 2h 33m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/5ac84ef3be8143fbb1a4097b68ed5d3d : FAILURE in 11m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dc45968e2b5b4e92ac2261d640d4a49d : FAILURE in 8m 36s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/64487e07beb84fafa51d604b9fabf398 : FAILURE in 16m 46s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7669f8df1a5c4532b8e119b75c227b7f : FAILURE in 13m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8d39a5b7168d4261b3073172040c6d8e : TIMED_OUT in 3h 03m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e45e4bfdc944de6b0b040e99eac50ca : SUCCESS in 32m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/343ceb88835d4123b07903c0aaa33d60 : SUCCESS in 33m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/0f503422fb2d4220bd2d79b7dce7de59 : SUCCESS in 33m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/538d38676e7242c09a2b24ec3482b59a : SUCCESS in 38m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c6356d1903484eab829c19268e882e2a : TIMED_OUT in 2h 33m 12s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/f0a8a4155363453d804ddb1c33122702 : SUCCESS in 54m 12s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/21472c22c47c448faca27d212ed00e8b : SUCCESS in 48m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c6a874aae1c34a804a5390e03bd1ac0a05f99903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-10-08 11:23:34.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"c304bb9e2e2e765af3941dad617ead0b068a05fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 14:32:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7babb9ad79d4472fa134fd2220484ed9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6beb9f2dcb41da9a667a6f2715f2de : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/904a07bb7092471a9b400a2b7d785eca : SUCCESS in 3m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/026fd6d89cca4b59ab3d3f1fd62878bd : SUCCESS in 1h 37m 04s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/ffb9eb01cb9f408aac348a7f407964c7 : SUCCESS in 34m 42s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/10f72f4a73214269b529cb9d77c407a6 : SUCCESS in 1h 33m 36s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d852da9b3d3f4d02a4557ce4f9324b88 : SUCCESS in 40m 42s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/ad619b790c844f128ab181acdd92dc68 : SUCCESS in 42m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0486ba98eae34a66a5f9946e86afe229 : FAILURE in 1h 58m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bf72ab0221294ce3b2205b5d3e1e5d57 : SUCCESS in 1h 10m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/ad9eaefdd1324f65afe5ab46f6f749b5 : FAILURE in 2h 07m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/8137bd0abd504f6cadcbe0def2bab16a : SUCCESS in 47m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/76f37431018149c0821799d95e662b47 : FAILURE in 2h 02m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/79e5d0e597014533964d57ceae0dc47e : SUCCESS in 56m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/0b986c9c7db5425b802839dacf3c818b : FAILURE in 2h 02m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/faf63caa38aa4d2b901ac972caddaa3b : SUCCESS in 1h 18m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aae74b8eb8d14b9f8dc18bba31dd8081 : SUCCESS in 57m 32s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/f2f5884a478f4202a2750c3fd598d3df : SUCCESS in 43m 40s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ceed80277580419695d7881b58879731 : SUCCESS in 52m 00s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/fb645eda9277480aada7a87b05c6a1a1 : SUCCESS in 44m 43s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6bb6abbab0c74e0093dd302929e15986 : TIMED_OUT in 2h 33m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/7aac110612d947919fcaf8a532888ecc : FAILURE in 13m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8399f1ae89b24df38b301bf8980d35ad : FAILURE in 12m 30s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/0d45cb3301754421bf09a7bf4ed33a80 : FAILURE in 7m 06s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bee220d96e674a4293cb8e3585314cba : FAILURE in 8m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cfd436e9e8b34f97b5090f78a9652740 : TIMED_OUT in 3h 04m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9deb486c6c6492ab467c1debad6f11c : SUCCESS in 53m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/b112c6bc653f47e5a4f1b8185a05abc0 : SUCCESS in 47m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/cff85fa680a44e1bb071c45da64d89fb : SUCCESS in 26m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/fda06b1a9b3743f8b7a31713592e62f5 : SUCCESS in 21m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/268f323a3c7e485b97fda3539ec38749 : FAILURE in 2h 31m 35s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/4c8eb1a349354fae91bbc508e50b5530 : SUCCESS in 46m 27s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/40935628d80a4c87acfe269dd15a35d2 : SUCCESS in 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"15780a352f7dd64a15ace04f13a61cbe4162eb35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-10-09 09:03:42.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"1d5b622083a27c309f08d54f5cf18fe837be9bcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 12:31:45.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f26155a3a1794d5ba2cbb274e6b43172\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d9476cd323ce484daec62be5d6ada0e9 : SUCCESS in 2m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b407ab590a054d00a3da03c7969e7c42 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b1b188a697945e087d8488e072195a5 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ea02c83d4ab4af4830b8677644975da : SUCCESS in 2m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cce9020a76e42e1acb1b193e75bde96 : SUCCESS in 2m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba218a8371494a039af598bc46b4f12a : SUCCESS in 1h 16m 58s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/cfc92c4e01194563910dd5aab6e95123 : SUCCESS in 1h 15m 29s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/0c125c259eba40d99cc634255f416ac4 : SUCCESS in 1h 28m 35s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/74bcf5e3b213483ca67814c3e06016d0 : SUCCESS in 1h 07m 16s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/5dfcc9e71b434a9daf86f52231e6e231 : SUCCESS in 1h 20m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/37184ec957b04f628ead45f1b05ec025 : SUCCESS in 1h 56m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/043baeb5c54d4ffb88cc7cb80dc87df3 : SUCCESS in 1h 27m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/efa264a096a94589b04223489f2296e8 : FAILURE in 1h 59m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/5a7207b8f2004c01b39e3b2d6e897305 : SUCCESS in 1h 20m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1cff15dd96bc4d30a5e0f35cbcadfc8b : FAILURE in 1h 47m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/fd2fe57482894822ac20500d0acba8db : SUCCESS in 1h 12m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/231ab36704634ae0bff33e0938025dcd : FAILURE in 1h 19m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/de9b68d90a124e4c8d879999f9b09e8b : SUCCESS in 1h 05m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bcb0297c7a2c4db9909ef73a4ed3873b : SUCCESS in 41m 57s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/5974a89bf9c64741837d1bb63a997cd9 : SUCCESS in 46m 09s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/93ca3fae88ec4cf5a35a8ffc4307584c : SUCCESS in 34m 03s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/b86f4d5a94c547d6b8ea149a4ddb9437 : SUCCESS in 43m 53s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a5e2942e526f45c887b453c39bfc77b8 : TIMED_OUT in 2h 33m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/4fc23010a9574a82a35ab4143ff0cc2b : FAILURE in 10m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8af96b3d0ccd49fbb99f1152dd5d0fe8 : FAILURE in 18m 00s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/408f0f47e97d456184652646a0e12f92 : FAILURE in 16m 57s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2b81068938a7412cb83877bdbda5932c : FAILURE in 10m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b179db34b704d36ab8ef6dfd8a23b18 : TIMED_OUT in 3h 04m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63937cc87e2641ba8fa78e29da6cf974 : SUCCESS in 34m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/e9b9b6ed2a9d4601b50fe8ac0b81ea7f : SUCCESS in 24m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1c86545750f645afa0a500ad7fa07210 : SUCCESS in 31m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/93ccb2786410435f9f5beed69b5e4cdd : SUCCESS in 45m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/124d2810a9354293991d13539691bc92 : SUCCESS in 2h 18m 43s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/9d3bca2bb4554bb6aaf3cc86c2a4374b : SUCCESS in 1h 11m 40s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/502cf1c22c674e619a54b5876cddb87f : SUCCESS in 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5e4cd5b62f485320b0637aa0b1e4beda1c91e7c2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-09 14:20:48.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"cb9f16db7721a35bd4785c5346c6cc612803742c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2026-09-09 13:59:28.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"20af844a24d3d21144740be203d34a16f55dafb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:35:55.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7942676c392e45bb861d077a0f646697\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adbd4cf120f54877a5b17a779ad58c6a : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/583395de8d8549c29519b296b8a80104 : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/152a287981fc4501a3e2a103795613d1 : SUCCESS in 4m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee9023cb4eae442f912543f1ea286e5e : SUCCESS in 1h 18m 15s\n- octavia-v2-dsvm-noop-api-stable-2026-1 https://zuul.opendev.org/t/openstack/build/486261014d104a00b08488316be35580 : SUCCESS in 1h 43m 37s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/9f3c32db03014fd6a8148f2da4f22086 : SUCCESS in 1h 33m 15s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/4559c256119e4bfb925219652ab232bb : SUCCESS in 1h 19m 06s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/897d109d55f04318a83e414f2ea29930 : SUCCESS in 1h 19m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ce6f85cf66ed4b1383c56e403815c78d : SUCCESS in 1h 43m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d06b311199c242d9be60e1316b0eb7ed : FAILURE in 1h 17m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/67bd3ab830a1487fa912c8fa4ca78834 : SUCCESS in 2h 02m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/c058a8209cea40bc85a87899bce63e4c : SUCCESS in 1h 43m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/3739c15fba4544f19798f9c8e2ad2d78 : SUCCESS in 1h 54m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/d21fc1bf5b3841d3acac6228f8e56d73 : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/9d5d78ed122443f788953d5018a46283 : SUCCESS in 2h 06m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/2891b9c80282481dac1b6090dbaeeaf2 : SUCCESS in 1h 29m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/03f87824ff364d0fa463ae872e464cef : SUCCESS in 47m 15s\n- octavia-v2-dsvm-tls-barbican-stable-2026-1 https://zuul.opendev.org/t/openstack/build/f820c02e1f8f47f7bcdcecb3122e1687 : SUCCESS in 58m 56s\n- octavia-v2-dsvm-tls-barbican-stable-2025-2 https://zuul.opendev.org/t/openstack/build/514f8278382d4ff7912373bcf397ee2b : SUCCESS in 46m 54s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/1f1e23cfed5043139953ca39fc16407d : SUCCESS in 38m 01s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1659d2dd5c554fdcb31b814569125f5e : SUCCESS in 1h 57m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/f4ca927a9e524e5498162a04346ef569 : SUCCESS in 1h 56m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fa329b37cb544dbbb1a4a04317f2b793 : SUCCESS in 1h 16m 15s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-traffic-ops https://zuul.opendev.org/t/openstack/build/e5af379ee4d644cc9044bbccaf696714 : FAILURE in 26m 23s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-non-traffic-ops https://zuul.opendev.org/t/openstack/build/72ba00e3486f48b49951102bb25775c8 : FAILURE in 35m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node-traffic-ops https://zuul.opendev.org/t/openstack/build/8bda81ef028c40c2b5a6b48367f82493 : SUCCESS in 1h 02m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f0a9e29e7764e9d94a35aa03caa3741 : SUCCESS in 42m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2026-1 https://zuul.opendev.org/t/openstack/build/16270fe20ce147a5a439d648954ede0a : SUCCESS in 43m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/d0b3720661d54c0194de4aa0c5d30301 : SUCCESS in 42m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/7f56120f8e694b61b4498b769458fb00 : SUCCESS in 34m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d4216f617ee048358404de4a521ed7c2 : SUCCESS in 46m 17s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/339f9541109f42058efce9f22991548d : FAILURE in 1h 38m 21s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/fa576051030145f4b23f51b3e948fd15 : SUCCESS in 51m 53s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9eb19ba92baeeacf1ca42386557b00447b9765c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2026-09-09 16:43:28.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"3af80563dabe9455f047ff8aaa3f9dc104d1afcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:34:33.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7f0225fdbfa432a9e5e13a5d6617058\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8d34a55235f4b1dbb1ee5e5e2c49df0 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4debfc58e474f60a1521e87534639c5 : SUCCESS in 2m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4f4772580234f9f83b81f4fdb623dd1 : SUCCESS in 5m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/749ceca2d6c64cf0aa6682d5b44a24b4 : SUCCESS in 1h 01m 35s\n- octavia-v2-dsvm-noop-api-stable-2026-1 https://zuul.opendev.org/t/openstack/build/d1c4fb712aef443980465334b6191cc4 : SUCCESS in 1h 13m 42s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/1e0e5fc4feb1468e9bdd02c2b16ac13d : SUCCESS in 1h 16m 58s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/185d4f05f8fc4472831318eb1d0ae2d0 : SUCCESS in 1h 19m 02s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/c5eb91d7856640a58b934618eec26b0b : SUCCESS in 1h 01m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d938c1f3ee344000aa0a058906e7ff1d : SUCCESS in 1h 23m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c6cfd3d2ece14c0fb88c6762c18b3352 : SUCCESS in 1h 00m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/bbfc1a4b34ec449aae2ba6c26bcb65b1 : SUCCESS in 1h 37m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/025d66522d8544558e1f88e4c8c1f039 : SUCCESS in 1h 09m 04s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/9938cb80618f45a79e0019c8c3446063 : SUCCESS in 1h 35m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/761bb9668a514f1598c22271bd5cc94e : SUCCESS in 1h 12m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/c082e382081340ffb52c946e4a941903 : SUCCESS in 1h 34m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/c10ce22b643f425a9af13ac354c0f913 : SUCCESS in 1h 29m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/77725982437f4e7ebcca232c9c817ad5 : SUCCESS in 37m 40s\n- octavia-v2-dsvm-tls-barbican-stable-2026-1 https://zuul.opendev.org/t/openstack/build/0529cb19a44d45df8e4c4da6190d7987 : SUCCESS in 41m 10s\n- octavia-v2-dsvm-tls-barbican-stable-2025-2 https://zuul.opendev.org/t/openstack/build/71e7d42f67624b46ad5259d4360021dc : SUCCESS in 42m 15s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/e867ea3dbb614baa8a6ea8bb139a7701 : SUCCESS in 38m 44s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f9fe598903014adc8f61ca01a661f4ef : SUCCESS in 1h 35m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/e2340ff308b5467cb21ceb99d86e5769 : SUCCESS in 1h 30m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cf9048df893c442d81bc1d9d1b2e6e79 : SUCCESS in 1h 29m 43s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-traffic-ops https://zuul.opendev.org/t/openstack/build/1484ed7494cd404c8d7485fd887d258a : FAILURE in 25m 01s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8bdbd16aba0b46f0bbf757dbe6e1676a : FAILURE in 34m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node-traffic-ops https://zuul.opendev.org/t/openstack/build/5ad86622cbc7408bb2ae0e93077273ff : SUCCESS in 56m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7566f517054e4559a45225d4548b63fd : SUCCESS in 42m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2026-1 https://zuul.opendev.org/t/openstack/build/b88a049e77e041dfb7f8d116468ba3e3 : SUCCESS in 48m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/44377b3aefb94ae99cac60e6e2ab2970 : SUCCESS in 36m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/b042e591cbde40558b902778eeecf13b : SUCCESS in 40m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fc349812907644b1a4112baacd31020c : SUCCESS in 54m 07s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/468d7f88fa2c48968fef7d655d31bb9a : SUCCESS in 1h 19m 22s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/0feecc3b37d647e28a233542e300d345 : SUCCESS in 45m 07s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"f18254f27ff5a09a9154d6fedbc17523808bfe43","revisions":{"8d12677db0a0d83a1968386ec2d87b2bd52748d1":{"kind":"REWORK","_number":1,"created":"2025-04-23 15:14:57.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/59/947959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/59/947959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/1"}}},"commit":{"parents":[{"commit":"40563b56461596380255db1986bf5943568e7401","subject":"Add member test with multiple subnets on same network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/40563b56461596380255db1986bf5943568e7401"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-04-23 15:09:17.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-04-23 15:09:17.000000000","tz":120},"subject":"Improve check member balanced for traffic not a multiple of members","message":"Improve check member balanced for traffic not a multiple of members\n\nWhen ROUND_ROBIN is used as the load balancing algorithm but the number\nof requests is not a multiple of the total number of members, we cannot\nguarantee that all members will receive the same number of requests.\n\nThis fix addresses that corner case. It also ensures that if a different\nload balancing algorithm is used, the previous assertion requiring an\nequal number of requests per member is no longer mandatory.\n\nChange-Id: I700248094c7b81ba9cc28e8507d4f5dd2b3fa1c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8d12677db0a0d83a1968386ec2d87b2bd52748d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/8d12677db0a0d83a1968386ec2d87b2bd52748d1"}]},"branch":"refs/heads/master"},"5878305b48a821ef4aadb0ad43b8fd1289080a3d":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-09-25 13:28:15.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/59/947959/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/59/947959/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/2"}}},"commit":{"parents":[{"commit":"84c77e787169d6d7d682378a3d4a82e0f2f0fb80","subject":"Add member test with multiple subnets on same network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/84c77e787169d6d7d682378a3d4a82e0f2f0fb80"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-04-23 15:09:17.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-09-25 13:27:54.000000000","tz":120},"subject":"Improve check member balanced for traffic not a multiple of members","message":"Improve check member balanced for traffic not a multiple of members\n\nWhen ROUND_ROBIN is used as the load balancing algorithm but the number\nof requests is not a multiple of the total number of members, we cannot\nguarantee that all members will receive the same number of requests.\n\nThis fix addresses that corner case. It also ensures that if a different\nload balancing algorithm is used, the previous assertion requiring an\nequal number of requests per member is no longer mandatory.\n\nChange-Id: I700248094c7b81ba9cc28e8507d4f5dd2b3fa1c2\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5878305b48a821ef4aadb0ad43b8fd1289080a3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5878305b48a821ef4aadb0ad43b8fd1289080a3d"}]},"branch":"refs/heads/master"},"693e11e2553869a5a2c1a8a38e2369ff6c5dd752":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2025-10-03 16:15:36.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/59/947959/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/59/947959/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/3"}}},"commit":{"parents":[{"commit":"b83da9786b4206c373e59e4217e32ca362c87121","subject":"Use default router for IPv6 tests and custom otherwise","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/b83da9786b4206c373e59e4217e32ca362c87121"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-04-23 15:09:17.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-10-03 16:14:49.000000000","tz":120},"subject":"Improve check member balanced for traffic not a multiple of members","message":"Improve check member balanced for traffic not a multiple of members\n\nWhen ROUND_ROBIN is used as the load balancing algorithm but the number\nof requests is not a multiple of the total number of members, we cannot\nguarantee that all members will receive the same number of requests.\n\nThis fix addresses that corner case. It also ensures that if a different\nload balancing algorithm is used, the previous assertion requiring an\nequal number of requests per member is no longer mandatory.\n\nChange-Id: I700248094c7b81ba9cc28e8507d4f5dd2b3fa1c2\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nSigned-off-by: Fernando Royo \u003cfroyo@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/693e11e2553869a5a2c1a8a38e2369ff6c5dd752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/693e11e2553869a5a2c1a8a38e2369ff6c5dd752"}]},"branch":"refs/heads/master"},"9d42caf804c57417dc86535be41b22b5c493320d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-10-03 17:00:18.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/59/947959/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/59/947959/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/4"}}},"commit":{"parents":[{"commit":"588af49c29bba7a2aeef9802c5d26382362bdeea","subject":"Use default router for IPv6 tests and custom otherwise","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/588af49c29bba7a2aeef9802c5d26382362bdeea"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-04-23 15:09:17.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-10-03 17:00:09.000000000","tz":120},"subject":"Improve check member balanced for traffic not a multiple of members","message":"Improve check member balanced for traffic not a multiple of members\n\nWhen ROUND_ROBIN is used as the load balancing algorithm but the number\nof requests is not a multiple of the total number of members, we cannot\nguarantee that all members will receive the same number of requests.\n\nThis fix addresses that corner case. It also ensures that if a different\nload balancing algorithm is used, the previous assertion requiring an\nequal number of requests per member is no longer mandatory.\n\nChange-Id: I700248094c7b81ba9cc28e8507d4f5dd2b3fa1c2\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nSigned-off-by: Fernando Royo \u003cfroyo@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9d42caf804c57417dc86535be41b22b5c493320d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/9d42caf804c57417dc86535be41b22b5c493320d"}]},"branch":"refs/heads/master"},"3cff934caff6e896a907f428151526c3a7135511":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-10-06 12:21:37.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/59/947959/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/59/947959/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/5"}}},"commit":{"parents":[{"commit":"78abce5e7c09fe8358a008306cd872f29ca713c2","subject":"Use same router for IPV4/IPv6 tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/78abce5e7c09fe8358a008306cd872f29ca713c2"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-04-23 15:09:17.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-10-06 12:21:24.000000000","tz":120},"subject":"Improve check member balanced for traffic not a multiple of members","message":"Improve check member balanced for traffic not a multiple of members\n\nWhen ROUND_ROBIN is used as the load balancing algorithm but the number\nof requests is not a multiple of the total number of members, we cannot\nguarantee that all members will receive the same number of requests.\n\nThis fix addresses that corner case. It also ensures that if a different\nload balancing algorithm is used, the previous assertion requiring an\nequal number of requests per member is no longer mandatory.\n\nChange-Id: I700248094c7b81ba9cc28e8507d4f5dd2b3fa1c2\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nSigned-off-by: Fernando Royo \u003cfroyo@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3cff934caff6e896a907f428151526c3a7135511"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/3cff934caff6e896a907f428151526c3a7135511"}]},"branch":"refs/heads/master"},"bf4de6a4a7974eb5044b3c6a684585282eaf826f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-10-06 15:35:55.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/59/947959/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/59/947959/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/6"}}},"commit":{"parents":[{"commit":"84c77e787169d6d7d682378a3d4a82e0f2f0fb80","subject":"Add member test with multiple subnets on same network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/84c77e787169d6d7d682378a3d4a82e0f2f0fb80"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-04-23 15:09:17.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-10-06 15:34:33.000000000","tz":120},"subject":"Improve check member balanced for traffic not a multiple of members","message":"Improve check member balanced for traffic not a multiple of members\n\nWhen ROUND_ROBIN is used as the load balancing algorithm but the number\nof requests is not a multiple of the total number of members, we cannot\nguarantee that all members will receive the same number of requests.\n\nThis fix addresses that corner case. It also ensures that if a different\nload balancing algorithm is used, the previous assertion requiring an\nequal number of requests per member is no longer mandatory.\n\nChange-Id: I700248094c7b81ba9cc28e8507d4f5dd2b3fa1c2\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nSigned-off-by: Fernando Royo \u003cfroyo@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bf4de6a4a7974eb5044b3c6a684585282eaf826f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bf4de6a4a7974eb5044b3c6a684585282eaf826f"}]},"branch":"refs/heads/master"},"fc2282e0bc40dff3b2aa0dc477502ef26fe4e8f5":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-08 11:23:34.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/59/947959/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/59/947959/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/7"}}},"commit":{"parents":[{"commit":"15c17b825268ad8f205787af7db086d32cf1318d","subject":"Adding IPv6 subnets to router and additional config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/15c17b825268ad8f205787af7db086d32cf1318d"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-04-23 15:09:17.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-10-08 11:22:49.000000000","tz":120},"subject":"Improve check member balanced for traffic not a multiple of members","message":"Improve check member balanced for traffic not a multiple of members\n\nWhen ROUND_ROBIN is used as the load balancing algorithm but the number\nof requests is not a multiple of the total number of members, we cannot\nguarantee that all members will receive the same number of requests.\n\nThis fix addresses that corner case. It also ensures that if a different\nload balancing algorithm is used, the previous assertion requiring an\nequal number of requests per member is no longer mandatory.\n\nChange-Id: I700248094c7b81ba9cc28e8507d4f5dd2b3fa1c2\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nSigned-off-by: Fernando Royo \u003cfroyo@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fc2282e0bc40dff3b2aa0dc477502ef26fe4e8f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fc2282e0bc40dff3b2aa0dc477502ef26fe4e8f5"}]},"branch":"refs/heads/master"},"5c07aab502edc994d78ec0ff9ef37d9ec786749b":{"kind":"REWORK","_number":8,"created":"2025-10-09 09:03:42.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/59/947959/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/59/947959/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/8"}}},"commit":{"parents":[{"commit":"83eee989c66047045d5fc626acaf483f5dbad5b8","subject":"Fix IPv6 traffic tests and improve network setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/83eee989c66047045d5fc626acaf483f5dbad5b8"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-04-23 15:09:17.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-10-09 09:01:48.000000000","tz":120},"subject":"Improve check member balanced for traffic not a multiple of members","message":"Improve check member balanced for traffic not a multiple of members\n\nWhen ROUND_ROBIN is used as the load balancing algorithm but the number\nof requests is not a multiple of the total number of members, we cannot\nguarantee that all members will receive the same number of requests.\n\nThis fix addresses that corner case. It also ensures that if a different\nload balancing algorithm is used, the previous assertion requiring an\nequal number of requests per member is no longer mandatory.\n\nChange-Id: I700248094c7b81ba9cc28e8507d4f5dd2b3fa1c2\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nSigned-off-by: Fernando Royo \u003cfroyo@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5c07aab502edc994d78ec0ff9ef37d9ec786749b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/5c07aab502edc994d78ec0ff9ef37d9ec786749b"}]},"branch":"refs/heads/master"},"f992d8787c736c19be3c498640491f0607e5af9b":{"kind":"REWORK","_number":9,"created":"2026-09-09 13:59:28.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/59/947959/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/59/947959/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/9"}}},"commit":{"parents":[{"commit":"4bed0a6223e507dcf4430160d250fb477b0031a3","subject":"Fix IPv6 traffic tests and improve network setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/4bed0a6223e507dcf4430160d250fb477b0031a3"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-04-23 15:09:17.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2026-09-09 13:48:44.000000000","tz":120},"subject":"Improve check member balanced for traffic not a multiple of members","message":"Improve check member balanced for traffic not a multiple of members\n\nWhen ROUND_ROBIN is used as the load balancing algorithm but the number\nof requests is not a multiple of the total number of members, we cannot\nguarantee that all members will receive the same number of requests.\n\nThis fix addresses that corner case. It also ensures that if a different\nload balancing algorithm is used, the previous assertion requiring an\nequal number of requests per member is no longer mandatory.\n\nChange-Id: I700248094c7b81ba9cc28e8507d4f5dd2b3fa1c2\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nSigned-off-by: Fernando Royo \u003cfroyo@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f992d8787c736c19be3c498640491f0607e5af9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f992d8787c736c19be3c498640491f0607e5af9b"}]},"branch":"refs/heads/master"},"f18254f27ff5a09a9154d6fedbc17523808bfe43":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-09-09 16:43:28.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/59/947959/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/59/947959/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/59/947959/10"}}},"commit":{"parents":[{"commit":"f505f26627e76d7631aa4004249eb73dde4a7d1e","subject":"Fix IPv6 traffic tests and improve network setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f505f26627e76d7631aa4004249eb73dde4a7d1e"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2025-04-23 15:09:17.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2026-09-09 16:33:26.000000000","tz":120},"subject":"Improve check member balanced for traffic not a multiple of members","message":"Improve check member balanced for traffic not a multiple of members\n\nWhen ROUND_ROBIN is used as the load balancing algorithm but the number\nof requests is not a multiple of the total number of members, we cannot\nguarantee that all members will receive the same number of requests.\n\nThis fix addresses that corner case. It also ensures that if a different\nload balancing algorithm is used, the previous assertion requiring an\nequal number of requests per member is no longer mandatory.\n\nChange-Id: I700248094c7b81ba9cc28e8507d4f5dd2b3fa1c2\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nSigned-off-by: Fernando Royo \u003cfroyo@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f18254f27ff5a09a9154d6fedbc17523808bfe43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/f18254f27ff5a09a9154d6fedbc17523808bfe43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
