)]}'
{"id":"openstack%2Foctavia-tempest-plugin~953509","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I4dc15dc3140f8c29fb7b2649239925dd3037136a","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"remove-USE_PYTHON3","hashtags":[],"change_id":"I4dc15dc3140f8c29fb7b2649239925dd3037136a","subject":"Zuul: do not use USE_PYTHON3","status":"NEW","created":"2025-06-27 00:48:26.000000000","updated":"2025-06-27 03:54:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cdcc767288f05326723cf6f0e9d81b984873cd2a","_number":953509,"virtual_id_number":953509,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-27 03:54:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-27 03:54:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48b39578cf5aa46a4e90ce878e65d45577f35666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-27 00:48:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdcc767288f05326723cf6f0e9d81b984873cd2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 03:54:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37ab86dde7cd4f3cab2c8d7ee20f91dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/968aa1de8e574b928d0ad9027d9cd30c : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d217b0459d59444683784e122e20f670 : SUCCESS in 2m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cdba671b0b8d4e958936c193b01051c7 : SUCCESS in 1h 05m 12s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/3dbbf571a5234491a70395f32486883b : SUCCESS in 1h 18m 28s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/eaf8ad5b97d744d18476e13d69ca0f0d : SUCCESS in 42m 04s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/f37d8f9682234b46b751f790e9e1e24b : SUCCESS in 1h 19m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b4d1384418f24622ae849e0a5b94caa9 : SUCCESS in 1h 27m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fcb475ff41f64afb81f027996b6db8c4 : SUCCESS in 56m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/9460f67a4f3144c2afb7c6a3f7d93c43 : SUCCESS in 1h 24m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/611909aa4829446182103b77ee05777d : SUCCESS in 48m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/5f8e94f56e3a42fb9c454718354dc78f : SUCCESS in 1h 17m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/aadac9e2bb214717918919fbbed19636 : SUCCESS in 56m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4106e028a08e4e8e81772f089f6f4714 : SUCCESS in 26m 16s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/0c674a40acc44616b218b14f3618af17 : SUCCESS in 36m 29s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8121622f98b54b1d859420309f629f4c : SUCCESS in 25m 15s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/859749c7e95540d584ee7b0232d1c30f : TIMED_OUT in 2h 33m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/d6a9488822db458e9ed5af8e3f0ce627 : FAILURE in 9m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/628c4c6095bf4715a0ff2d52e8bd3e18 : FAILURE in 10m 19s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/551441d8083a4884a2ab17d5a569c539 : FAILURE in 9m 50s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ef13fcff931143b280b85d0b52d3f668 : FAILURE in 9m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/29a1e5281f6f46129f30cb9940aeb4fd : TIMED_OUT in 3h 03m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e7df044767d4317b5eeecb1d1a189c7 : SUCCESS in 33m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e1162546655b4d58b8567d25c58c0f2b : SUCCESS in 29m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d2a55877eb814cb081b8c0ffc7ebeae5 : SUCCESS in 35m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c7707fad3734e768c46879d9bc6e2b7 : FAILURE in 12m 02s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/bf41f27788964f8b8941be3df2d37108 : SUCCESS in 1h 03m 54s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/f96e91aa4216433f8b14a2a4abc14aad : SUCCESS in 43m 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"719209f59bdbc2b46dbfc029ee71fcd205197205","revisions":{"719209f59bdbc2b46dbfc029ee71fcd205197205":{"kind":"REWORK","_number":1,"created":"2025-06-27 00:48:26.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/09/953509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/09/953509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/09/953509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/09/953509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/09/953509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/09/953509/1"}}},"commit":{"parents":[{"commit":"13cb2ec83d8457189030ab1be209c28186d85c69","subject":"Add retry decorator for tests failing due to traffic not passing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/13cb2ec83d8457189030ab1be209c28186d85c69"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-06-27 00:48:11.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-06-27 00:48:11.000000000","tz":120},"subject":"Zuul: do not use USE_PYTHON3","message":"Zuul: do not use USE_PYTHON3\n\nDevstack has removed the USE_PYTHON3 variable[1][2] and now always uses\nPython 3.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/920658\n[2] Commit 5412dbfe7b797149f1f68100de8003b1876398fe\n\nChange-Id: I4dc15dc3140f8c29fb7b2649239925dd3037136a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/719209f59bdbc2b46dbfc029ee71fcd205197205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/719209f59bdbc2b46dbfc029ee71fcd205197205"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
