)]}'
{"id":"openstack%2Foctavia-tempest-plugin~966371","triplet_id":"openstack%2Foctavia-tempest-plugin~master~Idc2b28982ad114851829fd4ba0c73952df2d25da","project":"openstack/octavia-tempest-plugin","branch":"master","topic":"2024-1-last","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-11-11 19:00:31.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-11-11 17:02:57.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"Idc2b28982ad114851829fd4ba0c73952df2d25da","subject":"Update stable branch jobs","status":"MERGED","created":"2025-11-06 22:16:02.000000000","updated":"2025-11-11 19:03:13.000000000","submitted":"2025-11-11 19:00:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966371-2024-1-last","meta_rev_id":"bf4c428d31bd4420bb07c0e6e1b3380ef3a8c587","_number":966371,"virtual_id_number":966371,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-11 19:00:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-11-11 17:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-11-11 17:02:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-06 22:21:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-07 00:36:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"367c568574fab77e66f1f961704521073748ea32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-06 22:16:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aacddffd71ff8ae852ed457fee98b9b85801ec4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 22:21:38.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/octavia on branch unmaintained/2024.1.  The\nproblem was:\n\n  Job octavia-v2-dsvm-noop-api-stable-2024-1 not defined\n\nThe problem appears in the the \"opendev.org/openstack/octavia\" project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n  ...\n\n  in \"openstack/octavia/zuul.d/projects.yaml@unmaintained/2024.1\", line 3","accounts_in_message":[],"_revision_number":1},{"id":"ae4bfa5c0be8defb569c35800f0b12c3c4ef1138","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-06 22:30:09.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ea1ac75e36af4864fa9eb2ffd7313c7b7963c883","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-11-07 00:36:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ba3d5a338099803aefba23fc276bc42bf518500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 01:36:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d8680a489cd40aaa0edc2bcce88c53b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b07376e9e88d4e8a8ebd0b0c8f998612 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1a3f7bab16c4725a6fd12e24d583cad : SUCCESS in 2m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4b141fd18b447c391a7bcc11f763743 : SUCCESS in 1h 35m 46s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/e275c62b597741cf991b4b6a944a6c01 : SUCCESS in 1h 13m 03s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/ef92c7ed06de4810a9e5eeeb4a6786ed : FAILURE in 18m 19s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/790f8b9fa99a4031b9bf633b4163aca8 : FAILURE in 28m 33s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/89569b4f540a43eb8a7823d94e60bdf6 : SUCCESS in 59m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ce9b36a8ab174c8bb2515f2339610d9e : SUCCESS in 1h 52m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/02adb0ecf94948dda53f045275f850bd : SUCCESS in 1h 08m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/1a61de78dfba494182e11767a67ee979 : SUCCESS in 1h 51m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/99686edd8a2440ffa98c50447d9d6859 : SUCCESS in 1h 02m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/55956dac25f548499bf6f2214edd7732 : FAILURE in 29m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/aa55b979fb3644e0bfae761774617996 : FAILURE in 22m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/21d2974f8c2f4f1fb6be5897a811ef16 : FAILURE in 20m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/91385b6eea58471081267bb8b2fcc91a : FAILURE in 32m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/b8aa21c9e91543ca8e8d7ce7d70c546e : FAILURE in 36m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/41e9b78cc87e4995bc05c31140ca1003 : FAILURE in 26m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ce7fd8a7354246a0852f1af159eccae0 : SUCCESS in 48m 42s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/e39eec66ddbb4fbbabaa6d2b55d1a323 : FAILURE in 38m 15s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/58f6a1e6625248fe94540d81159b0263 : FAILURE in 34m 47s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f98f0125b9a64b7f8c73ddc128557778 : SUCCESS in 2h 21m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/b72bb21571484d49a58ba5b253e73d0a : FAILURE in 9m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3fc1a7b0701f43049b71706529ea634d : FAILURE in 12m 16s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/c2dd94fe39d54a9e8375eaee3d93d9af : FAILURE in 14m 00s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ee18b41515324f679f11ad4a01711b2d : FAILURE in 7m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f8bcbf024e4f456b85f2466ceb3770f0 : TIMED_OUT in 3h 04m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/150cb6abc891405ea6405133abdee2bb : SUCCESS in 37m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/67acef1e0dd84d208512f9f74362188b : SUCCESS in 38m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/4239c701024f48ce9d8e5682bf477fad : FAILURE in 28m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/22ef406f4bb146639912ff6774bfb895 : FAILURE in 24m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/57f899cd6009466ea36c255083f5c12d : SUCCESS in 2h 34m 24s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/8e65ecc71e3b4c11918904ae36bfca47 : SUCCESS in 52m 19s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/73115897890244ad9139bf3553d5a942 : SUCCESS in 47m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a9ed7e87379f81eb46438912513f63d38da6fa8b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-07 02:50:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bdf93eb088a971a66a63c99ad7614696fe4a5538","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-07 03:43:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81e38ea58baa862bccde8ccab7a0d9dd57e6233a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 06:53:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/770ea933650944c5a7e2337e4de7f36c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd5599b802b4bd0908365e09c955127 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3e38a3e33f5410fbadca6a0d25a4641 : SUCCESS in 3m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/82c2fe39a66f4b9e8701237f19981a58 : SUCCESS in 1h 31m 14s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/204291502dcb4bccab754f3158b1dc4e : SUCCESS in 1h 07m 35s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/2e786962ef164fb7b7926205b0c45370 : SUCCESS in 1h 02m 45s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/87dd5d68544944c6aad19a7def9032a4 : SUCCESS in 1h 01m 22s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/5d141bf003af40c196e5b50895d662f0 : SUCCESS in 1h 42m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a80e6276e379473f850f4aae92c1af3a : SUCCESS in 2h 01m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/344e747251a142dc88b50781d32e5aa3 : SUCCESS in 1h 07m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/645905e124f844269863314d30b14109 : SUCCESS in 1h 59m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/d501896f389f4ac485621b1fd4207d00 : SUCCESS in 1h 07m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/2065e267ee9c45fa90b17623d63240a5 : SUCCESS in 1h 56m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/d403308bb48948c0ba7fcf01a4c7dacb : SUCCESS in 1h 00m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/66856d7c969141a29903163130f77797 : SUCCESS in 1h 45m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/53fd4d0154f24d8297ed798ace8ffaac : SUCCESS in 1h 05m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/6de722fd618a48b6b8494ed4d200533c : FAILURE in 22m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/305d8f8c548144d986df0ae2fbb57a36 : FAILURE in 25m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/132fa1e20198435981b9bd9b5a0fd68b : SUCCESS in 38m 58s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/ba28e9aeebd64838bbf5e8ce89251b0b : SUCCESS in 41m 53s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/26abbdaa431e4c5d8bd1875bb4e6ba24 : SUCCESS in 37m 54s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/23a121089c0142cc9226b9ec82e95a62 : TIMED_OUT in 2h 33m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/72cb3bfe58a343469f8fd1966eaef013 : FAILURE in 9m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fa556fc7dcb446eebed66ed1cad55dfa : FAILURE in 8m 39s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/07b26d2be51b408a8b5dd23468171236 : FAILURE in 7m 19s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bfa019199977414fac68a848e9192749 : FAILURE in 7m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0bddc5d3286a4bf1a4d781cfeffe43d9 : TIMED_OUT in 3h 02m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9b20e06e4bf4bd4ab04b431b3af8710 : SUCCESS in 34m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/a828b23ad12f4eb4b7552a652e086e2d : SUCCESS in 34m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/51402d0ddcf3415985e5e01db3c8a676 : SUCCESS in 37m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e3c7cf603e1b451d9083a97f46445055 : SUCCESS in 31m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/64604f61439f4d919dca74370b4c1369 : SUCCESS in 2h 27m 37s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/b3b8a923c8b849a587669d645ddd53de : SUCCESS in 1h 06m 36s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/b557fee88e2f45d7a45dc0c33c43fc30 : SUCCESS in 52m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"650e881124482b83f008368e2796b7078227b733","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-11-07 22:56:14.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3244b895601dfe8c6d85fb1a3f5660497b7a0359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-08 02:00:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7242e7e9a724a2987e315bd6dd804b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa1bb1a9bbf44788ed842edf99faa87 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c7dd46355774b808dfce34feb60b967 : SUCCESS in 2m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/87fca20f19d34f7980583a00f28f9f51 : SUCCESS in 1h 36m 41s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/a432584f345041c398e8e87add8703f6 : SUCCESS in 59m 27s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/19e4cd427b71483781d1ec58af5d6131 : SUCCESS in 59m 33s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/812db1d44ebe4e0dbd790753df13928b : SUCCESS in 58m 31s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/9a2ce693ca724646ad30ea1b936bbb2f : SUCCESS in 1h 31m 19s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/48ced57a8d0f47ba88f9a456aedbb920 : SUCCESS in 1h 51m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/445d11ca476e474c93ef54e6fc6e51b4 : SUCCESS in 1h 03m 13s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/6264eb6f50ea4b648652e99af7cafc04 : SUCCESS in 1h 52m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/35e28e66a8264bd99a7d07a15a6a1be1 : SUCCESS in 1h 11m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/0a407eef434c4710a1b54ee630e3b7b4 : SUCCESS in 1h 46m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/50e5d61067d8453880e00862d094e030 : SUCCESS in 1h 09m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/68ae66e74cce40989648b7f7f85dd209 : SUCCESS in 1h 44m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/bd494f82f1f9461099bfb1bfd712b83e : SUCCESS in 1h 02m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8a1a95845af146d2914361563c6ddd90 : SUCCESS in 37m 27s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/a4b4e49688fc49c2a766b39323d4c58a : SUCCESS in 38m 26s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/55154457b6924ab0aac4fb637ba911eb : SUCCESS in 32m 25s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6c83e0317e2c4e9c94bfe9bbecc51b32 : TIMED_OUT in 2h 32m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/3a949e639410428182883f9fbfccfa52 : FAILURE in 9m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6560e04440ae46e4b9a175e0e733228a : FAILURE in 9m 33s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-traffic-ops https://zuul.opendev.org/t/openstack/build/54ab7adc41a442b2bc0a4eedcb09aba2 : FAILURE in 13m 58s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-9-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dc9168ab4a1247e3abbd9da169776557 : FAILURE in 8m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/43b3f44be7374cb1b7e19caa966589a8 : TIMED_OUT in 3h 03m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7209b4981555405e9fdb6ba4502a96bd : SUCCESS in 36m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/950bacf5189b45b5bca8017e6a13bd15 : SUCCESS in 34m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/baa00d2c8852475f9f877d77e3127283 : SUCCESS in 33m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ae2e9879ff1249baa71842fd2cd233e2 : SUCCESS in 31m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e325e37ce07d4baba2977262020397f6 : TIMED_OUT in 2h 32m 55s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/6e7d1b6170f9487aae0162bb5d77cdde : SUCCESS in 1h 06m 26s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/43ae668a8f804deaaff0e9c0e35880f4 : SUCCESS in 48m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d37d42b69639663ae84b2f9f6f4ad2d6220c6fe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-11-11 17:02:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"be2274893ae928ee15925f8b3057d4d5de869f47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 17:04:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1da0cb982138cd66be02daf2f6f86364e8ae9a80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 19:00:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dd132f041ba406681e2eae519ae174e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7474187dd2df40cdbd1a1c744a6537c9 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef49ce7c9bab45b4a27ae6c903b634da : SUCCESS in 3m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/09b86676c18348c0b12c2bd4ec3eb23f : SUCCESS in 1h 26m 15s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/a30d696daecf46799a6baca1abd46fb7 : SUCCESS in 1h 04m 08s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/f2aeb78899384177917b12d474abffdf : SUCCESS in 1h 10m 15s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7f1133aab628422c8d16a07df988a6eb : SUCCESS in 1h 06m 28s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/0264fc5ce1664f83928e46f2edf80b6c : SUCCESS in 1h 29m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4fdb0a82c7f14e729cbd7ae4a32044e6 : SUCCESS in 1h 53m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/413d56e605bb4d388ff14261375e52ef : SUCCESS in 1h 13m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/1ff8249da55540f3ab04f05e8e6808e6 : SUCCESS in 1h 54m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/d29f1d2dd866406bad2de09e2ee4151c : SUCCESS in 1h 05m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/ce2ac332fa9e407fa4c3dbba3eb34b8c : SUCCESS in 1h 48m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/2f8e1309a7944018a76f4997c6bc6e4c : SUCCESS in 1h 03m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/5b26956062bc46ebaa74ce34df94a873 : SUCCESS in 1h 43m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/51facbc8e8ff4d2eaae04022db6d1a89 : SUCCESS in 57m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/47d261211c0f4ae69797b48ddb4e5f1f : SUCCESS in 36m 00s\n- octavia-v2-dsvm-tls-barbican-stable-2025-2 https://zuul.opendev.org/t/openstack/build/1309bc3713f046e3b0ba17538aa76a1e : SUCCESS in 34m 59s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/abe1b7e91d14435db1f401029ce8b66c : SUCCESS in 40m 02s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/befcde3aea2e4aa0b9faf379439551c9 : SUCCESS in 31m 02s","accounts_in_message":[],"_revision_number":3},{"id":"aa33d7cabef5ea3d08ef5d1084e0692a21c7f9ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 19:00:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bf4c428d31bd4420bb07c0e6e1b3380ef3a8c587","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 19:03:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2814bfba3c9e4a4e99feeeccfb8e5181\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83463d03f90b404e809f97e586f7fe7a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"27e3c8e262ba53498f83b1b2b7bfaa3ab93819d4","revisions":{"54834c159db8276b2e11634787ab4c5a3587c098":{"kind":"REWORK","_number":1,"created":"2025-11-06 22:16:02.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/966371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/71/966371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/966371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/966371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/966371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/966371/1"}}},"commit":{"parents":[{"commit":"fb464221b31ce0dee2e2f2af0c95d5b07379132d","subject":"Merge \"Add stable/2025.2 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fb464221b31ce0dee2e2f2af0c95d5b07379132d"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 21:52:12.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 22:16:01.000000000","tz":0},"subject":"Update stable branch jobs","message":"Update stable branch jobs\n\nRemove unmaintained/2024.1 jobs.\n\nChange-Id: Idc2b28982ad114851829fd4ba0c73952df2d25da\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/54834c159db8276b2e11634787ab4c5a3587c098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/54834c159db8276b2e11634787ab4c5a3587c098"}]},"branch":"refs/heads/master"},"bb850eea9b1c0457544a4d187f5b1b99a7e9560e":{"kind":"REWORK","_number":2,"created":"2025-11-06 22:30:09.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/966371/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/71/966371/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/966371/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/966371/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/966371/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/966371/2"}}},"commit":{"parents":[{"commit":"fb464221b31ce0dee2e2f2af0c95d5b07379132d","subject":"Merge \"Add stable/2025.2 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fb464221b31ce0dee2e2f2af0c95d5b07379132d"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 21:52:12.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-11-06 22:30:09.000000000","tz":0},"subject":"Update stable branch jobs","message":"Update stable branch jobs\n\nRemove unmaintained/2024.1 jobs from gate\nbut keeping the jobs definition as they are\nneeded by unmaintained/octavia gate.\n\nChange-Id: Idc2b28982ad114851829fd4ba0c73952df2d25da\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bb850eea9b1c0457544a4d187f5b1b99a7e9560e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/bb850eea9b1c0457544a4d187f5b1b99a7e9560e"}]},"branch":"refs/heads/master"},"27e3c8e262ba53498f83b1b2b7bfaa3ab93819d4":{"kind":"REWORK","_number":3,"created":"2025-11-07 22:56:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/71/966371/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/71/966371/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/966371/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/966371/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/966371/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/71/966371/3"}}},"commit":{"parents":[{"commit":"fb464221b31ce0dee2e2f2af0c95d5b07379132d","subject":"Merge \"Add stable/2025.2 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/fb464221b31ce0dee2e2f2af0c95d5b07379132d"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 21:52:12.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-11-07 22:56:14.000000000","tz":0},"subject":"Update stable branch jobs","message":"Update stable branch jobs\n\nRemove unmaintained/2024.1 jobs from gate\nbut keeping the jobs definition as they are\nneeded by unmaintained/octavia gate.\n\nChange-Id: Idc2b28982ad114851829fd4ba0c73952df2d25da\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/27e3c8e262ba53498f83b1b2b7bfaa3ab93819d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/27e3c8e262ba53498f83b1b2b7bfaa3ab93819d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
