)]}'
{"id":"openstack%2Foctavia-tempest-plugin~984602","triplet_id":"openstack%2Foctavia-tempest-plugin~master~I0d7285862bc1810fa44b7b845380c6f2f7f7fec5","project":"openstack/octavia-tempest-plugin","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-18 13:58:55.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0d7285862bc1810fa44b7b845380c6f2f7f7fec5","subject":"Use consistent name for service_available options","status":"NEW","created":"2026-04-14 16:58:10.000000000","updated":"2026-04-18 13:58:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"22aedac6af094ebd5d819f3c6808b4b9fd403937","_number":984602,"virtual_id_number":984602,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-15 09:38:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-14 17:57:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61d1431b77277202361add5cd5e9de7e7632e000","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-14 16:58:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36b93309625b4b3e739a548e006f0ec457164f01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-14 17:01:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3f62fedfb3eaad1917726483dc27d39c7bbd55cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 17:57:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2474c090fb3943b5bfcafa2c70f9ddff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e3630a3334b4aa986e9297ddc145dc4 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f35b567b5e742d2b7a4f1b9b96290cf : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/706f542a4d9f4276a7da136a07b3ca1b : SUCCESS in 2m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4fef89857be141ddbb904921b5dffc6a : FAILURE in 24m 22s\n- octavia-v2-dsvm-noop-api-stable-2026-1 https://zuul.opendev.org/t/openstack/build/7916a25a7083415ea87defefd5b000e5 : FAILURE in 32m 08s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/b5744752cd084b81ac8d0bec1b7dea5e : FAILURE in 32m 20s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/45fbf101120c49bba011b1211ca28f97 : FAILURE in 32m 40s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/414b079db3594a8eb71a58d050f36dfb : FAILURE in 30m 42s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/10abbe086a924c718f0360cbb51ab32f : FAILURE in 33m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b3da8aab84d9478080ba20a74a0f7119 : FAILURE in 54m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/073a86fe02664a7ca0b5bb86d9f88e3b : FAILURE in 55m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/3d3fb44c93074c509de168411cc3ba4c : FAILURE in 54m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/1892e6f2dfe24b82989280f8e2dae5e4 : FAILURE in 31m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/8f3c8d0dccfe4a0da0e41270def74f50 : FAILURE in 31m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/94b0811326d642ef918dc10ae73783f2 : FAILURE in 32m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/611a8b42da1c47bc9675601e794c4dfa : FAILURE in 17m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/2ccee66fcd1943ee859ed0b0d739b447 : FAILURE in 35m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4349454d138e47e7bdf4d8496388b072 : FAILURE in 25m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ac77232871d645b3a678a13a1762d0b6 : FAILURE in 30m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e70bcf4a5a324ab2896bb3bafb54d0e9 : FAILURE in 54m 53s\n- octavia-v2-dsvm-tls-barbican-stable-2026-1 https://zuul.opendev.org/t/openstack/build/b478049c39304136b186e1f18f35d463 : FAILURE in 36m 36s\n- octavia-v2-dsvm-tls-barbican-stable-2025-2 https://zuul.opendev.org/t/openstack/build/7f3cf6f157884f47aac72e5a0858410d : FAILURE in 54m 00s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/c22a6544144f4d68a76ad335cefb3bed : FAILURE in 35m 39s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b035c5b9e9a54cd9b7099596e4adb6c8 : FAILURE in 31m 43s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4300904eccc64bbebae3b77cf45d482d : FAILURE in 32m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/6029e2f50af449a98f79d88bb6a63879 : FAILURE in 30m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/361c1b63416c426cbd2a8eff3b3725d1 : FAILURE in 27m 30s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-traffic-ops https://zuul.opendev.org/t/openstack/build/3ef8326618244f07a2ed59c8ef0a571a : FAILURE in 52m 20s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7188b4fea11846c9939aeffe808480dd : FAILURE in 51m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node-traffic-ops https://zuul.opendev.org/t/openstack/build/81c940a35a214eae8074ee8c979ab70a : FAILURE in 44m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0fd4c1e3e3e24f9eadc723f33fe311ed : FAILURE in 32m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2026-1 https://zuul.opendev.org/t/openstack/build/8c22167fbd8f4b36b162b395237c6d4d : FAILURE in 14m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/18299f5218af400ea746c79e552ddfbc : FAILURE in 21m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/92922f7875754c8cb598b689eff3f68a : FAILURE in 35m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/8049c90b0b81417aa43f9cd4576228fa : FAILURE in 17m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5b8a081fb54b46248f0b0e558312cfb0 : FAILURE in 14m 43s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/6989cafadf604033a7e1b806c0599f65 : FAILURE in 24m 51s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/e99155576fd24728947274bd8a2d7324 : FAILURE in 18m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b1fcaedf15bddbb11ef95e3220d51b81172937a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-15 07:55:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"09d86c973e0ab77c029149a81db8344d0c14cea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 09:38:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a142809adff94d259ccff15ef3fe8265\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/515f2abd816242698f758a5ae4972544 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d578739c7a874df6aa3eb15250ff3306 : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a97d4ce1f77d4238b6781ed85bd25a0c : SUCCESS in 2m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bfd4d509218844af9665897e82a8c968 : SUCCESS in 1h 11m 06s\n- octavia-v2-dsvm-noop-api-stable-2026-1 https://zuul.opendev.org/t/openstack/build/c80bf46e6d014b6883fdca09f02b120b : SUCCESS in 1h 06m 32s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/139193bbba144d71bbe31b63f695634e : SUCCESS in 1h 00m 46s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/503b9cc391b643b88842568989080be2 : SUCCESS in 1h 06m 57s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/c546976a4a124c25bf319e092637b7dd : SUCCESS in 1h 10m 29s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/d1f1169360e24d4ab68826dcace59c7e : SUCCESS in 1h 15m 57s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5b53ee7bc1d24db4a557ea590a15038b : SUCCESS in 1h 39m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/05d53a776ea745e180eaf551a62b897a : SUCCESS in 1h 08m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/88576bc360f547218f0529b8341c4a6f : SUCCESS in 1h 36m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/4dc2235d7f5f4cf4b3135fe7164575b0 : SUCCESS in 1h 22m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/34409e3a3912431ba67307b13f0cf222 : SUCCESS in 1h 34m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/aa8587d83b0a4e7eafc464d5efa56717 : SUCCESS in 1h 13m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/c042b3a15d214e2f901bb9cc6ed405ff : SUCCESS in 1h 37m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/d2fad4bcf94f4eb581e2c126a1f1cbcf : SUCCESS in 1h 07m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/64db815ef2c7439bb29139423390cd5d : FAILURE in 1h 24m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/5bcb52aba0964292b445e43559b88864 : SUCCESS in 1h 08m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a43b4b7b57234916a11a31edfeb5b0a8 : SUCCESS in 35m 39s\n- octavia-v2-dsvm-tls-barbican-stable-2026-1 https://zuul.opendev.org/t/openstack/build/60ee0aa932c94607a537e1dc20c6f336 : SUCCESS in 44m 25s\n- octavia-v2-dsvm-tls-barbican-stable-2025-2 https://zuul.opendev.org/t/openstack/build/721478a04e4d4707becd9d3919725024 : SUCCESS in 37m 41s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/2a14d267a3a1464fafd81907599bcd79 : SUCCESS in 40m 37s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f43ef2ba9f9e4653bd0e66489c182a14 : SUCCESS in 32m 04s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9c3ad805dfc2446886b1e866190eba4a : SUCCESS in 1h 31m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/4048d8ff40e24815824b82a1413a9bd8 : SUCCESS in 1h 38m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6a868d15b0804714b5c39c965d21ad80 : SUCCESS in 1h 12m 22s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-traffic-ops https://zuul.opendev.org/t/openstack/build/3a15cd1d9e744351baff68389954b8c1 : SUCCESS in 1h 41m 17s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-non-traffic-ops https://zuul.opendev.org/t/openstack/build/aa28e8fc57f74a3f82eea67b5a27d03b : SUCCESS in 1h 11m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node-traffic-ops https://zuul.opendev.org/t/openstack/build/c53962b1667147918aaa61b1e7231053 : SUCCESS in 53m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7a74093c2bb4b4fbbf3c5acb3c43549 : SUCCESS in 33m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2026-1 https://zuul.opendev.org/t/openstack/build/26d4600c85c54653a9e08b0def03c23c : FAILURE in 49m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/e32077fa1d2c4abd89364d26030cde7b : SUCCESS in 34m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/1f8cd68400b844038a2760d42e351305 : SUCCESS in 32m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/9ae9763e8bbd4a14b5eb3fce4d04c544 : SUCCESS in 30m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/915f3d9e79974e0b8df12edd31a9e23e : SUCCESS in 38m 21s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/7710ba09b2de408388faac1fdeb48e0d : SUCCESS in 1h 11m 38s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/caca03875e3f4f4b8c5d529cf9fa2940 : SUCCESS in 41m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d4cac595809df3760f8224a04df8b5c0b8ecaa3d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-16 12:26:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4188db69ed9662f92cabce34f0bb7185cd97b8e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 15:33:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e98564e70bc14ddcaa4dc7e72d5498e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90df8660591547b381db2dcaaa904a78 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5adc804b22ca4fa7a363f707527cd5a2 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ad98c6095b648f7bcf9bcb5e471cadf : SUCCESS in 2m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e621792c9b7a41b9b747dc4db65594f1 : SUCCESS in 2h 03m 34s\n- octavia-v2-dsvm-noop-api-stable-2026-1 https://zuul.opendev.org/t/openstack/build/c2cf61319d074fb9b2ab3659e62b5561 : SUCCESS in 1h 22m 41s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/84c0ce7ad3174afba4cc2b7f870692d6 : SUCCESS in 1h 38m 29s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/caf6f0ad46d9447e9774b2890fa10c22 : SUCCESS in 1h 05m 23s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/bd488945a3484acfbe2e65c30d478dad : SUCCESS in 1h 19m 34s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/a165c3cfb6e04da79853de8189f86a59 : SUCCESS in 1h 17m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/20fa10803eed4ea69102ccba645928b0 : SUCCESS in 1h 55m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d2580b733a7d4935a9a64ba6ce5b38a6 : SUCCESS in 1h 14m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/fb447855513644a6840d8f0fee580a83 : POST_FAILURE in 1h 54m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/c3fee7a14b864c7fa515170e31b85eb7 : FAILURE in 1h 45m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/12202c9e26684653a36f2ecabb02beee : SUCCESS in 1h 52m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/d5743d115d974fbf8d9466f729ba389b : SUCCESS in 1h 01m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/7e016cb24e13424297600cd14d76c21b : SUCCESS in 1h 53m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/aad91777a2f348168845c7e316058ba4 : SUCCESS in 1h 21m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b7a7e05df3424c72ab9efbb6f135b7ff : FAILURE in 1h 43m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/816133fc3d6b458c96c3ba1b84bd3bb7 : SUCCESS in 1h 08m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/417460da3ae54ab4944b7fc0615320ba : SUCCESS in 55m 33s\n- octavia-v2-dsvm-tls-barbican-stable-2026-1 https://zuul.opendev.org/t/openstack/build/83628f24e70b413d9f61a2e50e3a04f2 : SUCCESS in 38m 40s\n- octavia-v2-dsvm-tls-barbican-stable-2025-2 https://zuul.opendev.org/t/openstack/build/df991a23988b4f4f89c8a1080b352ce3 : SUCCESS in 47m 03s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/f320079bb04849788f6e2d58f3c81f48 : SUCCESS in 43m 58s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/c70a328854f44a7e9ed938c41aac231c : SUCCESS in 35m 13s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3b429c6b6bd048f3997b576a14907815 : SUCCESS in 1h 44m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/76bb02577ced4273b6b60411efad6cb5 : SUCCESS in 1h 40m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4725df09280f4b7296502c2fe7b9e79a : SUCCESS in 1h 16m 37s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-traffic-ops https://zuul.opendev.org/t/openstack/build/10f05f59094e40619c4c56b107dda79b : SUCCESS in 1h 50m 37s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-non-traffic-ops https://zuul.opendev.org/t/openstack/build/06dcf71369594533935c6edce097e03d : SUCCESS in 1h 14m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node-traffic-ops https://zuul.opendev.org/t/openstack/build/19d80684f0c54c35a4460687107958d8 : TIMED_OUT in 3h 04m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69037e6929d84b4b82485c117a36f845 : SUCCESS in 32m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2026-1 https://zuul.opendev.org/t/openstack/build/87d7bed5c88247c385743b6034ab13ab : FAILURE in 1h 00m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/f7efe66e7e6a4951869dda760eb1168a : SUCCESS in 44m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/9c7c0598d48c4269a2ae90283544642f : SUCCESS in 43m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1f576918f076462d9ce4e2696d656e2d : SUCCESS in 38m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b3bbe841ba6d4aa584988b1add41968f : SUCCESS in 46m 19s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/2813d1ba1bdd49d7b5a85690b310678e : SUCCESS in 57m 02s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/6502cfc6a63f4e1c87612a5ae2baecdf : SUCCESS in 49m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"19529c6d69cf6013b8ee39fd50272b53a4d69032","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-18 11:24:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22aedac6af094ebd5d819f3c6808b4b9fd403937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 13:58:55.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f6a907e978c49c2a05e877714c695d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf453cc8fa9e44209d0e625c530d3643 : POST_FAILURE in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa4316169f964f129286838eb738f321 : POST_FAILURE in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07b01be3d99547eea0ef297c0ebd9c47 : POST_FAILURE in 2m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8677c8fda4b441a7b037b36f737aba98 : POST_FAILURE in 1h 37m 22s\n- octavia-v2-dsvm-noop-api-stable-2026-1 https://zuul.opendev.org/t/openstack/build/b9217014fbad445294e7362023264d35 : POST_FAILURE in 59m 24s\n- octavia-v2-dsvm-noop-api-stable-2025-2 https://zuul.opendev.org/t/openstack/build/3d820465c1814815b2e6600e7bf5bc67 : POST_FAILURE in 1h 03m 38s\n- octavia-v2-dsvm-noop-api-stable-2025-1 https://zuul.opendev.org/t/openstack/build/0bde1c590c924bbd8694b306a12c23e3 : POST_FAILURE in 1h 19m 59s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7ce60bbf3f1a4931ac4923b2b298a1a0 : POST_FAILURE in 1h 27m 16s\n- octavia-v2-dsvm-noop-api-advanced-rbac https://zuul.opendev.org/t/openstack/build/2cd0af9b2c5a457c8e246365f67d4f43 : POST_FAILURE in 1h 25m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/894212ad666541fa911c55af114da558 : POST_FAILURE in 1h 34m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3cd8d15e732d4ec7acbf43024de9cb60 : POST_FAILURE in 1h 12m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/1d6d8152b32e4c3f99f5f960e6c00bf6 : POST_FAILURE in 1h 45m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2026-1 https://zuul.opendev.org/t/openstack/build/86d5d8f3ff53476c9efe662dd504d95f : POST_FAILURE in 1h 24m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/1761a65e938f403f803dbef0fc70bf1c : POST_FAILURE in 1h 35m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-2 https://zuul.opendev.org/t/openstack/build/3f4ee66021ee454ebb9507a2b4146b6f : POST_FAILURE in 1h 34m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/97cce06c3d4245c2b1d1d340326c8b29 : POST_FAILURE in 2h 06m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2025-1 https://zuul.opendev.org/t/openstack/build/f2329ae57161491f8af9728c95a3fa16 : POST_FAILURE in 1h 13m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a734ecf84bbe4665bc16e31f80353d75 : TIMED_OUT in 2h 32m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/58d93bbdebf64dd0ad72b5417d1b51db : POST_FAILURE in 1h 29m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dc6ca5ab67c3427fa78172e117687993 : POST_FAILURE in 47m 03s\n- octavia-v2-dsvm-tls-barbican-stable-2026-1 https://zuul.opendev.org/t/openstack/build/7f0e7fac353c4d9f8c16e1f5ac90135b : POST_FAILURE in 57m 02s\n- octavia-v2-dsvm-tls-barbican-stable-2025-2 https://zuul.opendev.org/t/openstack/build/8c1c9c401031488d87d9053879e04e2a : POST_FAILURE in 42m 27s\n- octavia-v2-dsvm-tls-barbican-stable-2025-1 https://zuul.opendev.org/t/openstack/build/d07e4c19d1e34a5398b39e72793e0d99 : POST_FAILURE in 42m 18s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7ca98685cb714880926e8fb946f3676d : POST_FAILURE in 42m 33s\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eaf8e1a1f51940d6848b8ec4377f3fa7 : POST_FAILURE in 1h 34m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/cd644b0ee8374f959e0a6f4a25a1ed01 : POST_FAILURE in 27m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/497c4d8f913b4da29d428eeedd9bbf41 : POST_FAILURE in 1h 09m 26s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-traffic-ops https://zuul.opendev.org/t/openstack/build/c011efc383f24071bdda6134a36f936f : POST_FAILURE in 34m 08s (non-voting)\n- octavia-v2-dsvm-scenario-rockylinux-10-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3c77ab94429e450a8926ecdeff21e481 : POST_FAILURE in 1h 24m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-two-node-traffic-ops https://zuul.opendev.org/t/openstack/build/0ed20ff2ae0f4a3896b465037b9687fc : POST_FAILURE in 1h 09m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/040fd9d12b7f45a496e4528bd3b82fd9 : POST_FAILURE in 53m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2026-1 https://zuul.opendev.org/t/openstack/build/6a96ceca1f18482290cab4f9a98b705e : POST_FAILURE in 58m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-2 https://zuul.opendev.org/t/openstack/build/4f230277705a4d748b86a79659fa3f83 : POST_FAILURE in 43m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2025-1 https://zuul.opendev.org/t/openstack/build/4c12fc4fdbac4645b0835e01e26c1e8d : POST_FAILURE in 36m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/33dd4ce2a70741d7b8474385bcd228e2 : POST_FAILURE in 39m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3e19be1c2f334faa88757b99c82b0337 : POST_FAILURE in 38m 15s (non-voting)\n- neutron-ovn-provider-v2-api https://zuul.opendev.org/t/openstack/build/8a93e770ca4542bd9d0e8912bba15be0 : POST_FAILURE in 1h 08m 54s (non-voting)\n- neutron-ovn-provider-v2-scenario https://zuul.opendev.org/t/openstack/build/7dc471440f024c9eb5def7a7fc2a6ffb : POST_FAILURE in 53m 08s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"13d417ff4ecf67077ef094cfc1fb4307e923448a","revisions":{"7aea4a9053955cc1c7aaf55597889f35163c9e26":{"kind":"REWORK","_number":1,"created":"2026-04-14 16:58:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/984602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/02/984602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/984602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/984602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/984602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/984602/1"}}},"commit":{"parents":[{"commit":"d58111c58d8511034d857c42a3e5e45e7b055ee8","subject":"Add stable/2026.1 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d58111c58d8511034d857c42a3e5e45e7b055ee8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-14 16:56:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-14 16:57:06.000000000","tz":540},"subject":"Use consistent name for service_available options","message":"Use consistent name for service_available options\n\nCode-name is more commonly used than the service type.\n\nChange-Id: I0d7285862bc1810fa44b7b845380c6f2f7f7fec5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7aea4a9053955cc1c7aaf55597889f35163c9e26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/7aea4a9053955cc1c7aaf55597889f35163c9e26"}]},"branch":"refs/heads/master"},"cf96668307d1d345a4e72483a6682d027545384f":{"kind":"REWORK","_number":2,"created":"2026-04-14 17:01:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/984602/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/02/984602/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/984602/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/984602/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/984602/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/984602/2"}}},"commit":{"parents":[{"commit":"d58111c58d8511034d857c42a3e5e45e7b055ee8","subject":"Add stable/2026.1 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d58111c58d8511034d857c42a3e5e45e7b055ee8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-14 16:56:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-14 17:01:00.000000000","tz":540},"subject":"Use consistent name for service_available options","message":"Use consistent name for service_available options\n\nCode-name is more commonly used than the service type.\n\nChange-Id: I0d7285862bc1810fa44b7b845380c6f2f7f7fec5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cf96668307d1d345a4e72483a6682d027545384f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/cf96668307d1d345a4e72483a6682d027545384f"}]},"branch":"refs/heads/master"},"13d417ff4ecf67077ef094cfc1fb4307e923448a":{"kind":"REWORK","_number":3,"created":"2026-04-15 07:55:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/984602/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-tempest-plugin","ref":"refs/changes/02/984602/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/984602/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/984602/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/984602/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-tempest-plugin refs/changes/02/984602/3"}}},"commit":{"parents":[{"commit":"d58111c58d8511034d857c42a3e5e45e7b055ee8","subject":"Add stable/2026.1 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/d58111c58d8511034d857c42a3e5e45e7b055ee8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-14 16:56:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-15 07:54:59.000000000","tz":540},"subject":"Use consistent name for service_available options","message":"Use consistent name for service_available options\n\nCode-name is more commonly used than the service type.\n\nChange-Id: I0d7285862bc1810fa44b7b845380c6f2f7f7fec5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/13d417ff4ecf67077ef094cfc1fb4307e923448a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-tempest-plugin/commit/13d417ff4ecf67077ef094cfc1fb4307e923448a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
