)]}'
{"id":"openstack%2Foctavia~1002187","triplet_id":"openstack%2Foctavia~master~I6072a626cfc3cf8d2417e00420290581e78e4290","project":"openstack/octavia","branch":"master","attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-08-24 21:20:00.000000000","reason":"\u003cGERRIT_ACCOUNT_38360\u003e replied on the change","reason_account":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6072a626cfc3cf8d2417e00420290581e78e4290","subject":"doc: note ML2/Linux Bridge is deprecated in octavia-interface.service","status":"NEW","created":"2026-08-24 18:32:38.000000000","updated":"2026-08-24 21:20:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2af9e8289689d3e14a3dcf46d87c51d4a92f2227","_number":1002187,"virtual_id_number":1002187,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-24 19:59:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-24 19:59:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-24 21:20:00.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"state":"CC"}],"messages":[{"id":"b62aa7dcd1e10978c5439ce05c6295de86aa51fe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-08-24 18:32:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53e5d6049471d5b367ae0f1c0d4437ab88cae9f1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-08-24 19:40:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0b3a90130a4218ecef947aa3b4cb0b65054143c2","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-08-24 19:40:09.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"233f50785adbc99a366afe81231243bbf6b0fa9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 19:59:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2dde606fb7b49128a20ce907efba993\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/521aaf409b134b1bac88e689ebda5e0f : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd42471eb5694f9cbc8ff0fb5825408c : SUCCESS in 13m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cc30c2c3e2341d0a8edc229e82a7f40 : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":2},{"id":"2af9e8289689d3e14a3dcf46d87c51d4a92f2227","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-08-24 21:20:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9ea359c08dbe062295b5440a7fe62a480e48a26a","revisions":{"66837e240f87fe15e4bf2998d1d6aabe5951d581":{"kind":"REWORK","_number":1,"created":"2026-08-24 18:32:38.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/87/1002187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/1002187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/1002187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/1002187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/1002187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/1002187/1"}}},"commit":{"parents":[{"commit":"9a0e0a765ce0602cb38110655cd0cb73d8133e9c","subject":"Merge \"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a0e0a765ce0602cb38110655cd0cb73d8133e9c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-08-24 18:32:38.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-08-24 18:32:38.000000000","tz":0},"subject":"docs:","message":"docs:\n\nChange-Id: I6072a626cfc3cf8d2417e00420290581e78e4290\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66837e240f87fe15e4bf2998d1d6aabe5951d581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66837e240f87fe15e4bf2998d1d6aabe5951d581"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"9ea359c08dbe062295b5440a7fe62a480e48a26a":{"kind":"REWORK","_number":2,"created":"2026-08-24 19:40:07.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/87/1002187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/1002187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/1002187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/1002187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/1002187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/1002187/2"}}},"commit":{"parents":[{"commit":"9a0e0a765ce0602cb38110655cd0cb73d8133e9c","subject":"Merge \"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a0e0a765ce0602cb38110655cd0cb73d8133e9c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-08-24 18:32:38.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-08-24 19:39:11.000000000","tz":0},"subject":"doc: note ML2/Linux Bridge is deprecated in octavia-interface.service","message":"doc: note ML2/Linux Bridge is deprecated in octavia-interface.service\n\nFlag that the veth/brctl steps and the octavia-interface.service\nRequires\u003d/After\u003d dependency assume ML2/Linux Bridge, which is\ndeprecated in favor of OVS/OVN.\n\nCloses-Bug: #2163755\nChange-Id: I6072a626cfc3cf8d2417e00420290581e78e4290\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ea359c08dbe062295b5440a7fe62a480e48a26a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ea359c08dbe062295b5440a7fe62a480e48a26a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
