)]}'
{"id":"openstack%2Foctavia~1002621","triplet_id":"openstack%2Foctavia~master~Iac5922d313f5405c69f22267be689a61ec9de33f","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-09-01 09:37:36.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Iac5922d313f5405c69f22267be689a61ec9de33f","subject":"DNM/WIP Testing disabling redis maintenance notifications","status":"NEW","created":"2026-08-27 10:06:22.000000000","updated":"2026-09-01 12:10:39.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d3aefb32a4e04b05a97e2df7897ee1f9aee6fdb1","_number":1002621,"virtual_id_number":1002621,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-01 12:10:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-27 11:35:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"896c6ab7524f3dba3a9cae6e0c9b5aee0f4a59da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-27 10:06:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af1717a7479245e053d89a5f4129fd4f3c22f55f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 11:35:19.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/taskflow/+/1002620 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"727a44f263d698f7b0f22c99aafe6026283d4c79","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-01 09:37:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3aefb32a4e04b05a97e2df7897ee1f9aee6fdb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 12:10:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d02ee5d19af4f63b2b86482006e6678\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d14bea47388f4d06a0019163a900cd80 : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/679b7185683645c89546abb395d080cd : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c884f82ee39405e89b7374567628861 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/172d12686f59444fb17d14d3c56c9d92 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d7c644bc1554744a4b403eb45cdb480 : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/07b049745abe40cbbf947d5fc4fcb2f4 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00b428f55c144364b3d38d40c037088f : SUCCESS in 13m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/337674e8c4e34f3cb9f4c31915854f72 : SUCCESS in 9m 38s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/c53e9918c55441ada03c59333d0d5978 : SUCCESS in 4m 07s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/186cc34af449408a9c5636260f06cc2c : SUCCESS in 5m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/21c9ec9ddf134741b6ddaa00ea24b97a : SUCCESS in 4m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cd5156664d264b5c9f8ef3546166dc89 : SUCCESS in 3m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ab800b156ac24f0ca9fe80bfe7d913ff : SUCCESS in 42m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/37cce2025aeb432f89857a96f7cac762 : SUCCESS in 1h 24m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e7027229980d4c8fb7a61095dd32817a : SUCCESS in 58m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/914224f02b2b426f9721f908cb848dac : SUCCESS in 1h 31m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3f165fcd4d184ecfb81922a6e1295774 : SUCCESS in 1h 05m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d0e5d6cd0bc4e89b58981d52b38604f : SUCCESS in 37m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8596309af1c421688c06c31f30de88d : SUCCESS in 34m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/61ccf0025bb94ad29c0ce90340b41fe7 : SUCCESS in 37m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c6a02602214d407da83f2da6000ed453 : SUCCESS in 2h 31m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a3235814fbbb43a081a5a943472f8abb : SUCCESS in 1h 25m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/55569a57007c4e2a82bcd9eb79b23598 : FAILURE in 7m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/acb6dcb73a7d4e888bf93f8498caa08e : FAILURE in 9m 03s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3723054c2128b4268f8879a17ec6f93a10084175","revisions":{"3723054c2128b4268f8879a17ec6f93a10084175":{"kind":"REWORK","_number":1,"created":"2026-08-27 10:06:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/21/1002621/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/21/1002621/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/1002621/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/1002621/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/1002621/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/21/1002621/1"}}},"commit":{"parents":[{"commit":"fbc4af60a9dfa922343e0f86f0541b4971c1b908","subject":"Merge \"Fix VRRP split-brain on amphora failover in ACTIVE_STANDBY\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbc4af60a9dfa922343e0f86f0541b4971c1b908"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-08-27 10:06:02.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-08-27 10:06:10.000000000","tz":120},"subject":"DNM/WIP Testing disabling redis maintenance notifications","message":"DNM/WIP Testing disabling redis maintenance notifications\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/1002620\nChange-Id: Iac5922d313f5405c69f22267be689a61ec9de33f\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3723054c2128b4268f8879a17ec6f93a10084175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3723054c2128b4268f8879a17ec6f93a10084175"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
