)]}'
{"id":"openstack%2Foctavia~1003426","triplet_id":"openstack%2Foctavia~master~Ie0f5275b667c03882589286135e7a92e9451b7dd","project":"openstack/octavia","branch":"master","topic":"bug/2166163","attention_set":{},"removed_from_attention_set":{"5490":{"account":{"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},"last_update":"2026-09-08 15:18:07.000000000","reason":"Change was submitted"},"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-09-08 13:27:13.000000000","reason":"\u003cGERRIT_ACCOUNT_38562\u003e replied on the change","reason_account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},"31976":{"account":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"last_update":"2026-09-08 15:18:07.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2026-09-08 15:18:07.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-09-07 12:39:07.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ie0f5275b667c03882589286135e7a92e9451b7dd","subject":"Avoid leaking neutron client sockets","status":"MERGED","created":"2026-09-02 10:07:59.000000000","updated":"2026-09-15 16:09:33.000000000","submitted":"2026-09-08 15:18:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":35,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1003426-bug/2166163","meta_rev_id":"329df03c68f18c4bb3087b8fb1a60d4677acc086","_number":1003426,"virtual_id_number":1003426,"owner":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 15:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},"all":[{"value":1,"date":"2026-09-08 13:27:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-09-15 16:09:33.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-07 12:39:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-08 13:51:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},"all":[{"value":1,"date":"2026-09-08 13:27:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-09-15 16:09:33.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-07 12:39:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}],"CC":[{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 13:00:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 13:41:04.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-09-03 05:52:57.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-09-03 15:21:58.000000000","updated_by":{"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},"reviewer":{"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},"state":"CC"},{"updated":"2026-09-03 16:26:56.000000000","updated_by":{"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},"reviewer":{"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},"state":"REVIEWER"},{"updated":"2026-09-03 16:30:20.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"CC"}],"messages":[{"id":"7b39112974def94c0374ab0c34b14dbb8778a69a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-02 10:07:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee3ad33f2b02edf61fa73c13950834afeaf0e942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-02 10:24:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b1181c8775f0c52da4995ba5ff387388fc97725e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 13:00:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df7fcd38b8b3492d9164fe35fb04e618\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2d4ca88d37146d6b57f3136383a934a : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/755a1b16f60f4fe3935719320314530f : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff06df65baca40e79da4bc6dc14b9e6d : SUCCESS in 6m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cb81e23d10b4e5ebf7422bf2ae44740 : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f537922502e24661a872e5054b22a185 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5da616b3d5d74efea7ccbf6684a2ce37 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25ce36df00934dc18ce5bbc8a397e2cf : SUCCESS in 5m 22s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/f516c510eb9c43a68fe1d4d0b7224b6e : SUCCESS in 3m 50s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/b3e39083237641a59d063308df0b8c3c : SUCCESS in 7m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/02b412a355574df197b31148d2862ec2 : SUCCESS in 2m 43s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f4565acd347d45dd80c6af2745224169 : SUCCESS in 4m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/441e261749c446e58f2e42e38d7e1bf0 : SUCCESS in 36m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/709257b36e4d447e81a019bd38e35f29 : SUCCESS in 1h 24m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/66b06b99e5aa434bb039373b3d8a3653 : SUCCESS in 1h 03m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/72dda3a218a3475aaca2324536569b5f : SUCCESS in 1h 27m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ee1aac9dd2784eaeb96ebb9fbd2ff74c : SUCCESS in 1h 07m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a88e4780c0034d728461dce4b16a9445 : SUCCESS in 37m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c244e9fde28b418fa228ed1308ef1ddb : SUCCESS in 46m 36s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2a87fe60e43342c1a921ef94db6b79b8 : SUCCESS in 45m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c66189785bdc4c049643ee095c2d570b : SUCCESS in 34m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fcf816ab4ec84afdb3c4f6aec874d680 : SUCCESS in 36m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d0f4fd5544b54d4d9ad18ed030171ccd : SUCCESS in 2h 34m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9d1499026b5b4679833c0a3d5fe4b2dc : SUCCESS in 1h 23m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/1593677b30ad44209424988ec070397e : FAILURE in 7m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/13e998b91c4d45e19857291f3138cbd5 : FAILURE in 7m 24s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/fips/enable-fips.yaml for job octavia-v2-dsvm-scenario-two-node.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/multinode/enable-multinode.yaml for job octavia-v2-dsvm-scenario-two-node.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/pre.yaml for job octavia-v2-dsvm-scenario-two-node.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/octavia-tempest-plugin/playbooks/prepare-ovn-multinode.yaml for job octavia-v2-dsvm-scenario-two-node.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/tempest/playbooks/devstack-tempest.yaml for job octavia-v2-dsvm-scenario-two-node.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"7f8d8bc974c7a40ec55a2f1f43aef4e5288b8de5","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-09-02 13:41:04.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"427bd30c32fc4d408fe7d21fd30b9b9ef7289b6d","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-09-02 13:42:11.000000000","message":"Patch Set 2: -Backport-Candidate Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"548e16b9cc8feef8e5e7fbe2bc82dcde475c6db9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-02 14:29:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"6fd5f688f4c24150991d92c69088b51408e1c264","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-02 14:29:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5c6820535b6b9a1de1672f8dfded6ebf44cd7e2d","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-02 14:35:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ea99b649f54ee8f8ff573fb0313c814b739332ab","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-09-02 15:25:52.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8f12271153f2b90c6cced87a1143b2672ed27020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 19:28:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84fc4aba64ed40bd8f181183e27cb0bb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32351072f7a64923a75bc8f8dff0e3f8 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efdb091134b64406b47fbc8ebfd8b2e9 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9533b7ef24f421aaaf705e2e27ace9a : SUCCESS in 7m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f16d7796f9f4c84b668fcc562f57866 : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c6efc6a0354642fda3786b8f9c9599d4 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e22c594d0ab7448d8d214094d43aa704 : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1b34050d6884d7aa08090974c2ef2fc : SUCCESS in 4m 48s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/cd822a74ee8948bc8fac4ca5ff11bf39 : SUCCESS in 5m 32s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/48173ad92d634acab437604f8a5c6ba9 : SUCCESS in 6m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/093cef1bd55442f1996b563057718b52 : SUCCESS in 4m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0864a1f1e01a4e199778ab1ba9b1878f : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72c7a6776ac74f5d96c4a8a09c9f20d5 : SUCCESS in 1h 06m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9a04445be06a41d683c53fb339018aa6 : SUCCESS in 1h 23m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/09cb005f695343adb00eda74dc7aaaf0 : SUCCESS in 1h 00m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a580af7575c448cda711ab72baf73677 : SUCCESS in 1h 31m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/afb8a7be8b774e54b84c75b7a668adce : SUCCESS in 1h 11m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9abd2ba6ca3425ab2cbebd4a7b36c20 : SUCCESS in 35m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6799ec0611d046bbaaf551cc07bbf8ec : FAILURE in 23m 22s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a9a8dac59eb94579a7878b3b3e3d2dbd : SUCCESS in 48m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be5a9ad8371a4da09f14ea1b4785f264 : SUCCESS in 33m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d13c57c7f1e8490382b49f275bf8da0f : SUCCESS in 35m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6bb4ae2f2a9d47c9b3ae3ecb087b0850 : SUCCESS in 2h 26m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0c99a46a4c5442dea8381cd1b825d107 : SUCCESS in 1h 24m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/5d6539fdd7b84491855292bbfdaef57d : FAILURE in 8m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e6c936a2d915496cb5a7539727da98b7 : FAILURE in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cc7158d01a88c874d2d2f6f9baa800121be97ebf","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-03 05:52:57.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb6e6d920b2315b11dd367def5548dbb2e55cfc2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-03 07:00:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0b5a4e17e66750f9777a2722fc5f0f1e2398436d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-03 07:04:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"003865fa82732acfa475613194062e7a31997e51","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-09-03 08:15:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"97612f8fbaa902e665842ab4310169ac3a475eaa","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-03 08:30:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"21da757d9f462ffae422f36b727d2d939005cd7c","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-03 08:41:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0ef37ff5eaf3192faa1f0a985d5ad6035ce4ae42","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-03 11:59:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e8446631641e61c2be296b92f596f402791da451","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-03 12:24:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7dac7d9ebc9a845e83a584ace36b6c516ace1d5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-03 12:37:42.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"a23b399ff032a787b926b0e1337796fa67224552","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-03 12:38:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"553b42fc524c614251a73ca7bd50a998fdf09b32","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-03 12:52:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ed1ef2536e1a838150fdf34a5163439ae8176066","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-03 13:04:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5a935522c688be28f80ec890b2dc76f294e85943","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-03 13:06:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9240844bae3edd26b5185f4c12faa6ca2733e323","author":{"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},"date":"2026-09-03 15:21:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"063e0c568f909e23bf0e0ed14087e663b8a7290e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-03 15:30:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5387e2a04e92240cd4239992a59c385f026115c3","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-03 15:33:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9db2d95b5fcf368692a84c5083d6ef48f945118d","author":{"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},"date":"2026-09-03 16:26:56.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"de5c988e2914e937041a6fdcae54e5fc8d34be98","author":{"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},"date":"2026-09-03 16:27:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8fd8a248ad7ec011496f2e6b6eedf6bdaa876452","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-03 16:30:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"04f32c68a7492c260e4e165190b04280d4ee57da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 19:04:38.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84357384d322430ba36bf88ac6c41988\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61da67f0b5434aba86d537d6703026b6 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2361c2906b3945d792493dc59ab21c3c : FAILURE in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/917e258c5c8b41a28887ef2069e25d13 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b10a80dcbe294e6a8177f4686c3e1cb3 : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fbb054a44e6a4bc7a601140c47df8d35 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01d4d29dd3c54712b95c78adbbbd8915 : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f9d0a8f71ff450da7f13fdb5cf73259 : SUCCESS in 9m 18s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/457c9a8c211846de8092d5a8ce696371 : SUCCESS in 3m 25s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/bd194218f9964fe5a95a93eb391cab86 : SUCCESS in 5m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/288e30c1317347298e9c359610443d72 : SUCCESS in 2m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cba452781a11404dad4a0389876a7080 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/059494ba16a24fa3ad58c814434152ce : SUCCESS in 30m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/98ed744b140f475292801b6b08f481a7 : SUCCESS in 1h 25m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d5e5c13d88514762993922fac3d60fe8 : SUCCESS in 56m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b3d5e87b6f61475cbc93432c602bdbb5 : SUCCESS in 1h 29m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ae986d56ba4f4682a75a63e9f088364e : SUCCESS in 1h 06m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/53fb3893ca0b4671a81d292d8d986542 : SUCCESS in 36m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ef6c5541784d461383d0f60501b69e1d : SUCCESS in 46m 30s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2ab1868432d646f4999f71400cb22e74 : SUCCESS in 45m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/deaaaf446c9f48329434d51c5fa0b5a3 : SUCCESS in 33m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/51a04acc0ead41ce95fcea253582104a : SUCCESS in 34m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cf1ccfa71756404e9c60929c7fb77712 : SUCCESS in 2h 26m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2d36eb3c5da340d3a3a3689f14a62fe4 : SUCCESS in 1h 23m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/c063c0e438b9433db5c7889985c44588 : SUCCESS in 1h 14m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d5def549b2ee48a4885261a3a09a0ee9 : SUCCESS in 59m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d6ba9ac393a40f8f5d141d71e34612c1fd9e4968","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-04 06:43:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ab44ab102f31bcf838850bd3ee24d540aa9ad657","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-04 07:14:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e1b10b9d46a5cd82287df6dff8053c22eeec5eb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-04 08:27:01.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"20249256a14af2abff94cfcbfe004f7c9d965fb1","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-04 08:33:59.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"beff943983b64a0e4c4be6bfc3bbd373a75a5b93","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-04 08:38:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ca1709a982a8fa2b9dbecbff4d118c28dab94317","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2026-09-04 08:43:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c3f453bc5920812fc328a6189f4a9f3ae6b29664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 10:30:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9513acf1c214b82ab92de29f6bfe851\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15ddcc8fd5004977af17770ba66032fc : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fa2ee8062bb4e59bdd7ef39eb050bf8 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/616309fa9bc5460db463e55c6ae2841d : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/696822f3efe84eba91837bcc9bdf65a8 : SUCCESS in 6m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dd257292c68a485eb9c86c92419c6455 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48c807119e97425cb9b91b5312b06194 : SUCCESS in 13m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/736f92f2c01c4eb28039ebbc4b0ec309 : SUCCESS in 9m 42s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/2674ad73cfc74bb3bd91178b961bd57a : SUCCESS in 5m 26s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/381500d9f7cd4c0ba3cf5e41f7990d35 : SUCCESS in 6m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c137e54b53e04dc0b7ba5fde65aa0ab4 : SUCCESS in 2m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/21db745ac2d349ec9f4328a7172c67b3 : SUCCESS in 5m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aa3344b8e52b4624a85562175216fa5a : SUCCESS in 29m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fb5603e7234547cc929c8b80931011a0 : SUCCESS in 1h 49m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/af3315a436aa47e0b5fbf8a199bcbcc8 : SUCCESS in 1h 21m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f5cbdb44a94f4c9586eb199a0dfcad7b : SUCCESS in 1h 55m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9086335ffd4c4a5b90b2221f67c37e7f : SUCCESS in 1h 23m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4534eeb957f41968b19466a65716553 : SUCCESS in 38m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8319509410d14a25962ea0936810b9e9 : SUCCESS in 45m 27s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7f4799ce3b224523b76a2e25269c300e : SUCCESS in 45m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45ca9b289d6c481d8efa5d38c99381e6 : SUCCESS in 34m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a1a259778a0b40b585e38549a9d89044 : SUCCESS in 39m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fb69c77101f34537842e547071ef535c : FAILURE in 1h 22m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/98fb54335af04c1c82aa44757d051a95 : SUCCESS in 1h 48m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/e4f7b63f4c004c9685d63b74e2e1c1e3 : POST_FAILURE in 2h 00m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/85970b3f092a48b0b90dc19668e58838 : SUCCESS in 1h 00m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a2768da9eaf08345aa1386d1e0706887ba215536","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-07 12:39:07.000000000","message":"Patch Set 6: Backport-Candidate+2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ef9399479f8d9cd4d9b2e60bd82b5c29c5db8349","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-09-08 13:27:13.000000000","message":"Patch Set 6: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7388f5d870eb573d9664cd488bc1cb169bb9944e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-08 13:51:33.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d6f1c8e6ea9bba645251978a01525b1b6381d25f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 13:52:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"dcda9d42a389cfd555789a967b833babb3337d60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 15:18:07.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f36f962b244f4a45877cf5980e08d0ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2b04c4c27494186a4314e9aeb68c349 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4de6cedac6b347d99bce0c0206c876dc : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe208cfd80a74712adbbe41928e5a50d : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3028e680c263441abab2ebf94a123fb3 : SUCCESS in 16m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcd8628bf27b4ce3a53ac8642d8cccef : SUCCESS in 9m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6aadc5d90e4a4bc38d0ae72456c7d3fa : SUCCESS in 3m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b543331a409f42699a5ac618193f621b : SUCCESS in 5m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d3d42a09a7cd495d90f715ecf3905733 : SUCCESS in 1h 14m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cda26c8bb6dd43a99e9bd077488004ad : SUCCESS in 1h 25m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/62fc8e2952cd4df5aa46ed76c9eac395 : SUCCESS in 1h 14m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9f27d5abb4ab42b5af1a734ea0647f5d : SUCCESS in 43m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dae89c4d378445c1a69d1ac45cc8db92 : SUCCESS in 59m 37s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e3b5715bfbef48baaaffa207b4159367 : SUCCESS in 48m 58s","accounts_in_message":[],"_revision_number":6},{"id":"603425a83aa75c82a1c46d961e0f43c2769135b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 15:18:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"e65a8db79806063436d6089e2dd061a05c677e80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 15:19:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16db667d633c4d74b2156b525dfd6569\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f05fd9da1b4a9c8b05237a1d118cf8 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c6df1000d7040ecab21a79f0e4d5ffe : SUCCESS in 53s","accounts_in_message":[],"_revision_number":6},{"id":"329df03c68f18c4bb3087b8fb1a60d4677acc086","author":{"_account_id":5490,"name":"Kurt Garloff","email":"kurt@garloff.de","username":"kgarloff"},"date":"2026-09-15 16:09:33.000000000","message":"Patch Set 6: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e3613a16d2398b0384ff1cb04c26ea19d69ac87b","revisions":{"85e0aec830d075c421ea6a516320e8fa494d39cb":{"kind":"REWORK","_number":1,"created":"2026-09-02 10:07:59.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/26/1003426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/1003426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/1003426/1"}}},"commit":{"parents":[{"commit":"e2c18db97f3f6706461e53c3f38496513c5a2ed1","subject":"Merge \"Disable pypi when using local mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2c18db97f3f6706461e53c3f38496513c5a2ed1"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2026-09-02 09:52:07.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2026-09-02 10:07:56.000000000","tz":120},"subject":"Avoid leaking neutron client sockets","message":"Avoid leaking neutron client sockets\n\nFixes issues of \"Too many open files\".\nFor every neutron API query using the user neutron client a new\nopenstack SDK connection is opened, but never closed.\nAn explicit close on the connection is required to allow the\npython garbage collector to clean up the connection.\nOtherwise the connection stays referenced (memory leak) and leaves\nsocket descriptors open (socket leak).\n\nCloses-Bug: #2166163\nChange-Id: Ie0f5275b667c03882589286135e7a92e9451b7dd\nSigned-off-by: Bodo Petermann \u003cb.petermann@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85e0aec830d075c421ea6a516320e8fa494d39cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85e0aec830d075c421ea6a516320e8fa494d39cb"}]},"branch":"refs/heads/master"},"9ee44b5b93f1b1214db44244f7e721e2f0cc3f3d":{"kind":"REWORK","_number":2,"created":"2026-09-02 10:24:21.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/26/1003426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/1003426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/1003426/2"}}},"commit":{"parents":[{"commit":"e2c18db97f3f6706461e53c3f38496513c5a2ed1","subject":"Merge \"Disable pypi when using local mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2c18db97f3f6706461e53c3f38496513c5a2ed1"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2026-09-02 09:52:07.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2026-09-02 10:24:01.000000000","tz":120},"subject":"Avoid leaking neutron client sockets","message":"Avoid leaking neutron client sockets\n\nFixes issues of \"Too many open files\".\nFor every neutron API query using the user neutron client a new\nopenstack SDK connection is opened, but never closed.\nAn explicit close on the connection is required to allow the\npython garbage collector to clean up the connection.\nOtherwise the connection stays referenced (memory leak) and leaves\nsocket descriptors open (socket leak).\n\nCloses-Bug: #2166163\nChange-Id: Ie0f5275b667c03882589286135e7a92e9451b7dd\nSigned-off-by: Bodo Petermann \u003cb.petermann@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ee44b5b93f1b1214db44244f7e721e2f0cc3f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ee44b5b93f1b1214db44244f7e721e2f0cc3f3d"}]},"branch":"refs/heads/master"},"ef1d91510919692ca8187ee4931753f856ab643a":{"kind":"REWORK","_number":3,"created":"2026-09-02 14:29:10.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/26/1003426/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/1003426/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/1003426/3"}}},"commit":{"parents":[{"commit":"e2c18db97f3f6706461e53c3f38496513c5a2ed1","subject":"Merge \"Disable pypi when using local mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2c18db97f3f6706461e53c3f38496513c5a2ed1"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2026-09-02 09:52:07.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2026-09-02 14:29:00.000000000","tz":120},"subject":"Avoid leaking neutron client sockets","message":"Avoid leaking neutron client sockets\n\nFixes issues of \"Too many open files\".\nFor every neutron API query using the user neutron client a new\nopenstack SDK connection is opened, but never closed.\nAn explicit close on the connection is required to allow the\npython garbage collector to clean up the connection.\nOtherwise the connection stays referenced (memory leak) and leaves\nsocket descriptors open (socket leak).\n\nCloses-Bug: #2166163\nChange-Id: Ie0f5275b667c03882589286135e7a92e9451b7dd\nSigned-off-by: Bodo Petermann \u003cb.petermann@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef1d91510919692ca8187ee4931753f856ab643a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef1d91510919692ca8187ee4931753f856ab643a"}]},"branch":"refs/heads/master"},"304dfff2e3c01c2377620b522feb37547895e1f5":{"kind":"REWORK","_number":4,"created":"2026-09-03 12:37:42.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/26/1003426/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/1003426/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/1003426/4"}}},"commit":{"parents":[{"commit":"e2c18db97f3f6706461e53c3f38496513c5a2ed1","subject":"Merge \"Disable pypi when using local mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2c18db97f3f6706461e53c3f38496513c5a2ed1"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2026-09-02 09:52:07.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2026-09-03 12:37:22.000000000","tz":120},"subject":"Avoid leaking neutron client sockets","message":"Avoid leaking neutron client sockets\n\nFixes issues of \"Too many open files\".\nFor every neutron API query using the user neutron client a new\nopenstack SDK connection is opened, but never closed.\nAn explicit close on the connection is required to allow the\npython garbage collector to clean up the connection.\nOtherwise the connection stays referenced (memory leak) and leaves\nsocket descriptors open (socket leak).\n\nCloses-Bug: #2166163\nChange-Id: Ie0f5275b667c03882589286135e7a92e9451b7dd\nSigned-off-by: Bodo Petermann \u003cb.petermann@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/304dfff2e3c01c2377620b522feb37547895e1f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/304dfff2e3c01c2377620b522feb37547895e1f5"}]},"branch":"refs/heads/master"},"669cff3608a89d79df35879f856cf7829942a346":{"kind":"REWORK","_number":5,"created":"2026-09-03 15:30:58.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/26/1003426/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/1003426/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/1003426/5"}}},"commit":{"parents":[{"commit":"e2c18db97f3f6706461e53c3f38496513c5a2ed1","subject":"Merge \"Disable pypi when using local mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2c18db97f3f6706461e53c3f38496513c5a2ed1"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2026-09-02 09:52:07.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2026-09-03 15:30:46.000000000","tz":120},"subject":"Avoid leaking neutron client sockets","message":"Avoid leaking neutron client sockets\n\nFixes issues of \"Too many open files\".\nFor every neutron API query using the user neutron client a new\nopenstack SDK connection is opened, but never closed.\nAn explicit close on the connection is required to allow the\npython garbage collector to clean up the connection.\nOtherwise the connection stays referenced (memory leak) and leaves\nsocket descriptors open (socket leak).\n\nCloses-Bug: #2166163\nChange-Id: Ie0f5275b667c03882589286135e7a92e9451b7dd\nSigned-off-by: Bodo Petermann \u003cb.petermann@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/669cff3608a89d79df35879f856cf7829942a346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/669cff3608a89d79df35879f856cf7829942a346"}]},"branch":"refs/heads/master"},"e3613a16d2398b0384ff1cb04c26ea19d69ac87b":{"kind":"REWORK","_number":6,"created":"2026-09-04 08:27:01.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/26/1003426/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/1003426/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/1003426/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/1003426/6"}}},"commit":{"parents":[{"commit":"e2c18db97f3f6706461e53c3f38496513c5a2ed1","subject":"Merge \"Disable pypi when using local mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2c18db97f3f6706461e53c3f38496513c5a2ed1"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2026-09-02 09:52:07.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2026-09-04 08:26:38.000000000","tz":120},"subject":"Avoid leaking neutron client sockets","message":"Avoid leaking neutron client sockets\n\nFixes issues of \"Too many open files\".\nFor every neutron API query using the user neutron client a new\nopenstack SDK connection is opened, but never closed.\nAn explicit close on the connection is required to allow the\npython garbage collector to clean up the connection.\nOtherwise the connection stays referenced (memory leak) and leaves\nsocket descriptors open (socket leak).\n\nCloses-Bug: #2166163\nChange-Id: Ie0f5275b667c03882589286135e7a92e9451b7dd\nSigned-off-by: Bodo Petermann \u003cb.petermann@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3613a16d2398b0384ff1cb04c26ea19d69ac87b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3613a16d2398b0384ff1cb04c26ea19d69ac87b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
