)]}'
{"id":"openstack%2Foctavia~1004617","triplet_id":"openstack%2Foctavia~stable%2F2026.1~I079380d551bdae06feb5a482b029c87cc86b6d61","project":"openstack/octavia","branch":"stable/2026.1","topic":"tlsv13-ciphers-stable/2026.1","attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-09-09 12:52:10.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-09 13:44:01.000000000","reason":"\u003cGERRIT_ACCOUNT_38562\u003e replied on the change","reason_account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}}},"removed_from_attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-09-09 13:44:01.000000000","reason":"\u003cGERRIT_ACCOUNT_38562\u003e replied on the change","reason_account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}}},"hashtags":[],"change_id":"I079380d551bdae06feb5a482b029c87cc86b6d61","subject":"Fix cipher configuration for TLSv1.3","status":"NEW","created":"2026-09-08 13:34:59.000000000","updated":"2026-09-09 15:29:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":919846,"cherry_pick_of_patch_set":21,"meta_rev_id":"d47eecea11a4f5ddbb1e79220d97b807982b9447","_number":1004617,"virtual_id_number":1004617,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-09 15:29:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-09-09 13:44:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 13:34:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2026-09-08 13:36:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-09-08 13:36:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-09-08 16:44:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10c7b2489b8c4609aaa29b88be049b846ff393d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 13:34:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"012dc71424c01c7d4c6e6b1bfba99d2192f14581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:44:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d26d854b1ac4826b970176e51e8c20f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63fb9b21736c4a5c9532561e99022398 : SUCCESS in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/977e932ff54f4d30b4a56826cfc00c5d : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e918df83b569411ebf43194cf607bfbc : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63c2807f204144a98faf4358a8d162cb : SUCCESS in 6m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4894c3c5650242f38f8a3b5a9677f507 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ea86ad260b4e60a9b6563d8e05db63 : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76b0bd9ea4574c6f956b04a725ae1ab7 : SUCCESS in 10m 21s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/bfcf83b265d44f49b68d765329cc5890 : SUCCESS in 7m 03s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/e1ead132e3f349038fea6a5e806694f5 : SUCCESS in 7m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/007a379b337646c898064c335e7caed9 : SUCCESS in 6m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6f2c9ef00b684e809b2c33ed445f2131 : SUCCESS in 9m 00s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cd8dd3a952e845a983ca2ee4fb07a8ce : SUCCESS in 8m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0795d2715d734cc08e478e110d30697a : SUCCESS in 47m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/67e7be1c825c4b85a966e2a83452497c : SUCCESS in 1h 40m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/25c6a9251ff34428b38977833a65087a : SUCCESS in 1h 11m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/10625e11299b4afda5c7719b630db021 : SUCCESS in 1h 55m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7a35888bfa2a445588050bc8a0f63f65 : SUCCESS in 1h 13m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c64c9b6c7584462f91cadab7f185bc54 : SUCCESS in 42m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/88c90488bd934efda5f0ce4fcf0cc995 : SUCCESS in 54m 24s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/412dea03bd684314b01c647c1a2cfed1 : SUCCESS in 59m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e039cd2f5eb4e469cbbb4c0f8359cb6 : SUCCESS in 43m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0016a4b3fc9846fc879bd8f9ed9b88fa : SUCCESS in 39m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d26773e1b935467cab6b0f4304e2a922 : TIMED_OUT in 3h 04m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/652c372760f64b6188db0d8b7d9a13bf : SUCCESS in 1h 32m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/95b947932ce54b6e8a1c8a3a1fb8cd83 : FAILURE in 11m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d83086ece15b495c8106d8ffb91f3ff5 : FAILURE in 11m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5d1760a5a09b959694f86b38e9f0b3cd447dbf2","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-09-09 08:52:49.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a80d7451cabe504629fd3f20c76a3676dd5ce9d8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-09 11:27:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2621185353f0264cdc28222db09ad807960726c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-09 12:04:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c5e691a6907ae809d83d6e044ab71c040445391","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-09 12:05:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89f80ea20df0f3b5dae2bbee669cf92aa3d5e3de","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-09 12:49:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25d8731c258e2c40e8924c976f34f822052b9ad0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-09 12:52:10.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"888cc8b59f5424cf71b684aec1473707d3eaea95","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-09 12:52:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e2754a55d4632ac82829b2bc74dad9d75fc30663","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-09-09 13:44:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d47eecea11a4f5ddbb1e79220d97b807982b9447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 15:29:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c594a905e2c4f378bbb4cb79d4e4fbb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/409c2632000549b5a8d011f38193e55f : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d207f629b584084b108de7ed4d4f35b : SUCCESS in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/910f6bd198614842a1fce013ae684262 : SUCCESS in 6m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac2f5374040f4b7bb7ba88410fc9936d : SUCCESS in 4m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/31b2c5e8adf94441813940f48fc6b0e1 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9724462ac4204902a908746e62665da9 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78fcb42dea32494cb4fc6d516a11748d : SUCCESS in 9m 26s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/ef29bcab0a3b427bab65b4d6b599f597 : SUCCESS in 5m 10s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/3cf2aed2a29a4d02aee4e6bd496b9798 : SUCCESS in 10m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/614f26929e25463aa2aa463bbfe3fd27 : SUCCESS in 3m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d38052060c6140e58e6aa57522544fc8 : SUCCESS in 4m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1ed82b0137004f68bad1f872eef66eac : SUCCESS in 6m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba57e640f4534d5385a320e519aa0963 : SUCCESS in 1h 31m 11s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e73e8468a3de45ffa81d06c7ad1e46bb : SUCCESS in 1h 37m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/161fa50ff0f24128a37bff75294686de : SUCCESS in 1h 03m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/abae9a7669c347c5946311ebebadc3d4 : SUCCESS in 1h 44m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/677639f789434ff1bb25e8fffe98a902 : SUCCESS in 1h 16m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2bd4486f87084d9faeff10b75b32fc23 : SUCCESS in 40m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/742c163f3dde4544998280aa71f3015c : SUCCESS in 45m 34s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5c1672572d53457e9bf0450b41764c2e : SUCCESS in 44m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/086b2cc3ee5444ce9f8a16d014e1cb84 : SUCCESS in 35m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dbd24c6ae6ac4fad9fa289dc364fcba7 : SUCCESS in 39m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cb22193e1c8c4b46850aa71640d446d6 : SUCCESS in 2h 34m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3c6fce3d6beb43a6a7531dd4b7911141 : SUCCESS in 1h 36m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/e6afaa19ee0e4e7fafd6588fc3fb0291 : FAILURE in 10m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e065cf8f099c428fa47cf73849de59a4 : FAILURE in 9m 50s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"87b4581f8b033785f89aa6a6c2050d7aa9a0a1f2","revisions":{"8dc91b6703635c17be3243c2c050fba74b56ca23":{"kind":"REWORK","_number":1,"created":"2026-09-08 13:34:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/17/1004617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/1004617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/1004617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/1004617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/1004617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/1004617/1"}}},"commit":{"parents":[{"commit":"fa00c4a799a527db8c157db60862f4fc75bb404a","subject":"Fixing CI mirrors on stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa00c4a799a527db8c157db60862f4fc75bb404a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-05-16 14:09:23.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:34:59.000000000","tz":0},"subject":"Fix cipher configuration for TLSv1.3","message":"Fix cipher configuration for TLSv1.3\n\nFor TLSv1.3 ciphers the ciphersuites[1] bind option has to be used.\n\n[1]: https://www.haproxy.com/documentation/haproxy-configuration-manual/latest/#ciphersuites%20(Bind%20options)\n\nCloses-Bug: #2065798\nChange-Id: I079380d551bdae06feb5a482b029c87cc86b6d61\nSigned-off-by: Tom Weininger \u003ctweining@redhat.com\u003e\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit d691a2eeefea77384311d9be884a497ddd491c72)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8dc91b6703635c17be3243c2c050fba74b56ca23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8dc91b6703635c17be3243c2c050fba74b56ca23"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"1dfb009d3fe30db7c0fb43c73120f98a420ddf1b","ours":"fa00c4a799a527db8c157db60862f4fc75bb404a","theirs":"d691a2eeefea77384311d9be884a497ddd491c72","merge_strategy":"recursive","contains_conflicts":false}},"87b4581f8b033785f89aa6a6c2050d7aa9a0a1f2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-09 12:52:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/17/1004617/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/1004617/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/1004617/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/1004617/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/1004617/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/1004617/2"}}},"commit":{"parents":[{"commit":"fa00c4a799a527db8c157db60862f4fc75bb404a","subject":"Fixing CI mirrors on stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa00c4a799a527db8c157db60862f4fc75bb404a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-05-16 14:09:23.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-09 12:52:10.000000000","tz":0},"subject":"Fix cipher configuration for TLSv1.3","message":"Fix cipher configuration for TLSv1.3\n\nFor TLSv1.3 ciphers the ciphersuites[1] bind option has to be used.\n\n[1]: https://www.haproxy.com/documentation/haproxy-configuration-manual/latest/#ciphersuites%20(Bind%20options)\n\nCloses-Bug: #2065798\nChange-Id: I079380d551bdae06feb5a482b029c87cc86b6d61\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit d691a2eeefea77384311d9be884a497ddd491c72)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87b4581f8b033785f89aa6a6c2050d7aa9a0a1f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87b4581f8b033785f89aa6a6c2050d7aa9a0a1f2"}]},"branch":"refs/heads/stable/2026.1","description":"Edit commit message","conflicts":{"base":"1dfb009d3fe30db7c0fb43c73120f98a420ddf1b","ours":"fa00c4a799a527db8c157db60862f4fc75bb404a","theirs":"d691a2eeefea77384311d9be884a497ddd491c72","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
