)]}'
{"id":"openstack%2Foctavia~1007633","triplet_id":"openstack%2Foctavia~master~Id1a93ed67f3c12168b33e3bdbfc6e8e1bd209bca","project":"openstack/octavia","branch":"master","topic":"bug/2168669","attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-10-09 12:11:13.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-10-07 07:40:56.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"removed_from_attention_set":{"34451":{"account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"last_update":"2026-10-09 12:16:04.000000000","reason":"\u003cGERRIT_ACCOUNT_34451\u003e replied on the change","reason_account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-10-09 12:11:13.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Id1a93ed67f3c12168b33e3bdbfc6e8e1bd209bca","subject":"Replace deprecated SQLAlchemy noload and return get_all_orm for L7 policies","status":"NEW","created":"2026-09-27 14:35:43.000000000","updated":"2026-10-09 13:21:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"0cf0de8337ea9d067857631295863bced5e433cf","_number":1007633,"virtual_id_number":1007633,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-07 12:14:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-10-09 12:16:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":1,"date":"2026-10-09 13:20:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-10-09 12:11:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-27 17:03:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-28 09:48:37.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-09-28 15:21:14.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2026-10-06 08:46:50.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-10-09 13:20:52.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"2b749eeedbaa6af13edad18ed4b1eeace2402f0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-27 14:35:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa860c2c51d500555dc800b3cbc85ecb2bf41d7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-27 17:03:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad2b798419bf46298704be18053b4312\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e1d60e907db4643824130a0aced94f1 : FAILURE in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac6524bcbafa4a299302b1600f820c9a : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c73f1caf8b64410d90a675657431c977 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3b60395465d448b68c2262883e43c39d : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8840b2b4f9424f0694c678ef2874c278 : SUCCESS in 13m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95c415ebb5ab4b49ad1789d433c95e22 : SUCCESS in 9m 50s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/501ba210ce844fc2be2f3facb08b960a : SUCCESS in 4m 57s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/271d2b86678c4528954e68fa9d31729c : FAILURE in 5m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5d108aa480e0445591c69d3e9677ded9 : SUCCESS in 3m 29s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6be157c7f2af4bbdae5bf837006f560b : FAILURE in 5m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c377f91bc01349e89488ae0dd6cbde6b : FAILURE in 1h 14m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/053d4c07f0e74d4f84d7e16050040ee0 : SUCCESS in 1h 21m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cdefc1894b0341a4a537e7724d51397b : SUCCESS in 57m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/deda3549ad6d4203b00ab110fc3ff8df : SUCCESS in 1h 25m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9d8bc1ba9ae746559c0ed7977b19fb78 : SUCCESS in 1h 07m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/868ecb08716642b39e17ca6fd54de6f6 : SUCCESS in 34m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a322ad6a0ac64003834d7b5bdd84334c : SUCCESS in 41m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ac0651e8ea141528802cbb9776a29ae : SUCCESS in 32m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/257dc934c0f3451ba86252345008416a : SUCCESS in 32m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d35439bdf5e34ae1a462c77f2186a260 : SUCCESS in 2h 24m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c7940047ba15430780a68226e549ed8a : SUCCESS in 1h 20m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/87de81872fdd4d24bc2cf64adb6f6e36 : SUCCESS in 1h 11m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/35c53ece70be402b8b122f8c931185b3 : SUCCESS in 54m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5a220c36cedffa46d572e0d8a065da43e1d9ebe","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-09-28 09:48:37.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78e6dcaced99673431c54b1118593f19cbc66a97","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2026-09-28 15:21:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fb650f99b00404a2e824098f2f3eb020727cdaec","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-10-06 08:46:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9cb5f6fcb982e6430c8ca2ce5995de93bca4be4","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-10-06 09:26:12.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d4b7e66856df91b76360d63240f12d0b8dd92ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-10-06 14:25:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20f223bf30f666523102a6b8c37c0281dbe7e16f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-10-06 14:26:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c35ea2e83b17d98263d2211480868e61025e81fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 17:05:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/026680f73114404da008097cf676ec38\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfc8e8a358f04e1a82f44b6ee8e81e4e : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2f65cfe64b347908525957bb8cb9ca1 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7994b78860f3443dbad2cf081a919ae8 : SUCCESS in 5m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/921963fb218d43e6bd8b97dc97ac2242 : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f1db8a5506e45c88f1725fc5acfdabe : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84c73b239ea748b5864f191dc0a0b555 : SUCCESS in 10m 00s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/767416e54e3149dfb93f1950126007c7 : SUCCESS in 5m 07s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/3d09351e4d6e41949557fd369d4fc943 : SUCCESS in 5m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/54c2bbfda74748c4b7d033a185d3d49c : SUCCESS in 4m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7eec833de8484cc7916605d25bab9692 : SUCCESS in 6m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e9272c47c17346ceb71c1ae2fb02923b : SUCCESS in 1h 01m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6878d1605aa34be99ab74fcdfdbf872b : SUCCESS in 1h 20m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/662b35e3c14c405dbc9692723084710d : SUCCESS in 1h 10m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6ec0b6da7d124600afe4ee23dadb5a00 : SUCCESS in 1h 31m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a92d795c3d2748c6bcd8f439e8339338 : SUCCESS in 1h 03m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d30501590cd74d2082512220b4c11fd3 : SUCCESS in 38m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7908f03171934342b55a101cbb63fd51 : SUCCESS in 41m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6c55115329a4bf9908e001b3075293d : SUCCESS in 32m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f1ab5be6ee5d4ddeba5ae20eb531943d : SUCCESS in 35m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3d691c80089e48d3b0e67e615b70f2b4 : SUCCESS in 2h 34m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d1ac3d08a67444c1a767e0477b534631 : SUCCESS in 1h 19m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/96df53b68a1741f38f4bdcebaf755d80 : SUCCESS in 1h 08m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/da3042b505844f2dba7bd7e70f23c417 : SUCCESS in 1h 01m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08fafedb9ef72d44222b13cdad1682649b6ee9bf","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-10-07 07:40:56.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0497f758019feb28a50821a60252117f318f575","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-10-07 09:03:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"594ff7e1d154247781fbcaefc03da358fee534e2","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-10-07 09:05:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ebc8e09291d6a760d2d87baf66e71f5950e0aa83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-07 12:14:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3251858e645a4d4ea3dbd4cd2ceb5ff4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5508be8ad29c47e7991a32e102e1e474 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b650e781ddbe4c1eba3bf4d18c3b9de2 : SUCCESS in 7m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95de0e955b914c8f95b11925027abadb : SUCCESS in 5m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/69b88cb34b1d49a29ed23fde3ece9e74 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3879409d58814927bef7dc0005ba24a4 : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/507b7ad76a8544939045b887e3cbdef0 : SUCCESS in 10m 28s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/ce615a85e32546f48c733bd4ec5040c1 : SUCCESS in 6m 45s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/0bdf010930b844a6b45860f83364d50b : SUCCESS in 5m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0c89240362514e299e2190c752d707da : SUCCESS in 4m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c5a55bb79e5f4d9bb2044c2ae8bf33c0 : SUCCESS in 5m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2569009991e446bcba05a0091841756e : SUCCESS in 1h 15m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fd616eab4dec433a8c18b488b401ca8a : SUCCESS in 1h 23m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/27b89d91a0504ffb81c55cd35139e92e : SUCCESS in 1h 03m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/75bd76fb08c64cae808a68304a1b1f13 : SUCCESS in 1h 27m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c263227c952f45868397f3568052dcb1 : SUCCESS in 1h 04m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11ba5298842f4f1983abc76a9db6fde5 : SUCCESS in 36m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fc1a0a1f1d09498684c1af1e817a0248 : SUCCESS in 40m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7e3e2724b4242e38f1693a5c2f021fa : SUCCESS in 32m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e7c43f8e93d241acb2a4a675f12502ae : SUCCESS in 34m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2314ebc514cf489eb12ab0e746ee74f7 : SUCCESS in 2h 28m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7afbeac90c6945ada3d40bd2135906fe : SUCCESS in 1h 21m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/8de7f39f20c34e328eadfd4ecde0d795 : SUCCESS in 1h 14m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/67479e26877948d19f77ced00faf2409 : SUCCESS in 1h 04m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"af30f82fbf3abdb6ebff8e26ff1822b061f825f2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-10-09 12:11:13.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"55d5477db61be3377d2a93ef4500903f8911b7c6","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2026-10-09 12:16:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"36e0a1efbd4834d07fb5bf1e3afc21a5eafb33df","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-09 13:20:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0cf0de8337ea9d067857631295863bced5e433cf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-09 13:21:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"64a5187af0ff55803157759b11a28623c72f02e8","revisions":{"5b8de83baf404c791c132dc5df1b399582e9721c":{"kind":"REWORK","_number":1,"created":"2026-09-27 14:35:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/1007633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/1007633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/1007633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/1007633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/1007633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/1007633/1"}}},"commit":{"parents":[{"commit":"57460f16a2864b889b4c89e79d978a45cdd84bf6","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57460f16a2864b889b4c89e79d978a45cdd84bf6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:35:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:35:04.000000000","tz":540},"subject":"Replace deprecated sqlalchemy.orm.noload","message":"Replace deprecated sqlalchemy.orm.noload\n\nThis behavior was deprecated in SQLAlchemy 2.1 due to confusing\nbehavior caused by this option[1].\n\nReplace it by raiseload so that we ensure that these non-loaded\nattributes are not read.\n\n[1] https://github.com/sqlalchemy/sqlalchemy/issues/11045\n\nCloses-Bug: #2168669\nChange-Id: Id1a93ed67f3c12168b33e3bdbfc6e8e1bd209bca\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b8de83baf404c791c132dc5df1b399582e9721c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b8de83baf404c791c132dc5df1b399582e9721c"}]},"branch":"refs/heads/master"},"ac01addff830e82fd7b4fe13aeb679ef28a1dce0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-10-06 14:26:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/1007633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/1007633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/1007633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/1007633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/1007633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/1007633/2"}}},"commit":{"parents":[{"commit":"e028a787c532ccd28fa29bea929d0c000075be67","subject":"Change L7 policy get all API to directly return ORM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e028a787c532ccd28fa29bea929d0c000075be67"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:35:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-10-06 14:20:13.000000000","tz":540},"subject":"Replace deprecated sqlalchemy.orm.noload","message":"Replace deprecated sqlalchemy.orm.noload\n\nThis behavior was deprecated in SQLAlchemy 2.1 due to confusing\nbehavior caused by this option[1].\n\nReplace it by raiseload so that we ensure that these non-loaded\nattributes are not read.\n\n[1] https://github.com/sqlalchemy/sqlalchemy/issues/11045\n\nCloses-Bug: #2168669\nChange-Id: Id1a93ed67f3c12168b33e3bdbfc6e8e1bd209bca\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac01addff830e82fd7b4fe13aeb679ef28a1dce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac01addff830e82fd7b4fe13aeb679ef28a1dce0"}]},"branch":"refs/heads/master"},"64a5187af0ff55803157759b11a28623c72f02e8":{"kind":"REWORK","_number":3,"created":"2026-10-07 09:03:17.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/33/1007633/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/1007633/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/1007633/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/1007633/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/1007633/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/1007633/3"}}},"commit":{"parents":[{"commit":"0aa355608cdd75f166901f42c383e82378f903c4","subject":"Merge \"Fix creation of TLS-enabled pools in the LB graph API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0aa355608cdd75f166901f42c383e82378f903c4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:35:04.000000000","tz":540},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-10-07 09:02:58.000000000","tz":60},"subject":"Replace deprecated SQLAlchemy noload and return get_all_orm for L7 policies","message":"Replace deprecated SQLAlchemy noload and return get_all_orm for L7 policies\n\nnoload was deprecated in SQLAlchemy 2.1 due to confusing\nbehavior caused by this option[1]. Replace it by raiseload so\nthat we ensure that these non-loaded attributes are not read.\n\nThe model.to_date_model() call within the API attempts to load all\nrelationships. However this prevents us from replacing noload, which\nis deprecated in SQLAlchemy 2.1, due to lack of explicit join calls.\n\nReturn the full ORM directly instead, to leave loading to the later\nlogics. Now matches all the other get_all_API methods.\n\n[1] https://github.com/sqlalchemy/sqlalchemy/issues/11045\n\nAssisted-by: Opencode with GPT 5.5\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: Id1a93ed67f3c12168b33e3bdbfc6e8e1bd209bca\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64a5187af0ff55803157759b11a28623c72f02e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64a5187af0ff55803157759b11a28623c72f02e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
