)]}'
{"id":"openstack%2Foctavia~435612","triplet_id":"openstack%2Foctavia~master~I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef","project":"openstack/octavia","branch":"master","topic":"548989","hashtags":[],"change_id":"I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef","subject":"WIP: Floating IP Network Driver (spans L3s)","status":"NEW","created":"2017-02-17 21:56:57.000000000","updated":"2019-11-25 08:43:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e44df7672e3c4985e961db59f61aa4b8fe70ad61","_number":435612,"virtual_id_number":435612,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-18 22:25:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-04-11 19:38:17.000000000","updated_by":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-08-22 04:08:40.000000000","updated_by":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"reviewer":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"state":"REVIEWER"},{"updated":"2017-08-25 00:53:12.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2017-08-25 14:39:05.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2017-10-13 20:00:36.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2018-08-24 22:26:55.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2019-01-29 05:43:38.000000000","updated_by":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"reviewer":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"state":"REVIEWER"},{"updated":"2019-11-18 22:25:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a69674e09f536f09fd8b239b195d89a9cdcfa80","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-17 21:56:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ac2d0f8c4135041a2d0d2019638fb0e797b8591","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-17 22:01:05.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)\n\nDoesn\u0027t yet support failover, I think? Haven\u0027t tested, but I believe there\u0027s things I will probably have to tweak?\nAlso, definitely won\u0027t work for ACTIVE/STANDBY yet, as the vrrp configs probably need to change a bit (and that might be specific to the FLIP implementation?).","accounts_in_message":[],"_revision_number":1},{"id":"cd3ed050a5cd32b54bd6948d5a0f6e0e0828ba31","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-02-17 22:02:22.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/8aafc4bbbc909953afc56dab491bcca1f93d9069/3843 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"9c60613e8ebc19cec01f533c041d62f065ad1597","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-17 23:10:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/1/check/gate-octavia-docs-ubuntu-xenial/9ca6ead//doc/build/html/ : SUCCESS in 2m 28s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/1/check/gate-octavia-pep8-ubuntu-xenial/a5ada53/ : SUCCESS in 5m 40s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/1/check/gate-octavia-python27-ubuntu-xenial/3808d1d/ : SUCCESS in 2m 32s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/1/check/gate-octavia-python35/aae4545/ : SUCCESS in 4m 08s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/1/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/1d68603/ : SUCCESS in 1h 06m 08s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/1/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/67545cd/ : SUCCESS in 1h 12m 57s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/1/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/48c8e35/ : FAILURE in 44m 26s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/1/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/b9b0797/ : FAILURE in 44m 20s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/1/check/gate-octavia-tox-functional-ubuntu-xenial/40da4b1/ : SUCCESS in 2m 49s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/1/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/a38621b/ : FAILURE in 37m 39s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/1/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/42630f0/ : FAILURE in 29m 26s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/1/check/gate-octavia-v1-dsvm-scenario-multinode-nv/bf642be/ : FAILURE in 50m 53s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/1/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/a483812/ : FAILURE in 39m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e07fb862f71ff8ddf0e80fec4bff4fcbcc904ac9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-18 00:32:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1060efa09a9a92844251a66fd70514e9a983be13","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-18 00:32:36.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"bb7ceeeb94c0b628510f19f81fa2b2bb66e01553","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-02-18 00:37:54.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/a1cc049f3b2867136eb18560db78e47ad30d4514/3846 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"613dc3a4b9236d104431c949b10ad2543f3a6739","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-18 01:15:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"91786e501993a5db214495cf5e626fe9220da0c6","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-02-18 01:21:14.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/f086a7aa13c30217cbff9c314c18137a698a627a/3847 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"3449fde5df92600fe8548f7b56a109e6a91d0512","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-18 02:21:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/3/check/gate-octavia-docs-ubuntu-xenial/6e95568//doc/build/html/ : SUCCESS in 2m 19s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/3/check/gate-octavia-pep8-ubuntu-xenial/d73091b/ : SUCCESS in 3m 45s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/3/check/gate-octavia-python27-ubuntu-xenial/28d75b1/ : SUCCESS in 6m 41s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/3/check/gate-octavia-python35/c6554a8/ : SUCCESS in 3m 58s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/3/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/aa2ed9f/ : SUCCESS in 1h 00m 15s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/3/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/a05b948/ : SUCCESS in 46m 52s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/3/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/2a62f78/ : FAILURE in 1h 05m 54s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/3/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/d1bb0e5/ : FAILURE in 44m 10s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/3/check/gate-octavia-tox-functional-ubuntu-xenial/399027c/ : SUCCESS in 2m 38s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/3/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/da86e4c/ : FAILURE in 27m 04s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/3/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/d8e8762/ : FAILURE in 27m 19s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/3/check/gate-octavia-v1-dsvm-scenario-multinode-nv/8ca520a/ : FAILURE in 30m 40s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/3/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/284c38e/ : FAILURE in 59m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8777551699f41372f844ed6807fa7f9ff6b53691","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-20 18:56:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c366a633fd77ac325ddb6a6d3a4182da28f75e79","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-02-20 19:01:52.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/f08986aa141073e71a04851fa88fabc2720bec59/3868 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"b160832da7a97c3c37a4aa26ba5759e5ace44025","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-20 19:51:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c345499ecc046c547a17bca0633c6d7960d7a6b6","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-02-20 19:56:56.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/3deacf80f1f1ecd26ecd78be150c8dc1b8437075/3871 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"2b73d00d09342d5660b46f43ae6b77a6b7000eeb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-20 21:01:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/5/check/gate-octavia-docs-ubuntu-xenial/e4311db//doc/build/html/ : SUCCESS in 3m 01s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/5/check/gate-octavia-pep8-ubuntu-xenial/2216a2b/ : SUCCESS in 3m 30s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/5/check/gate-octavia-python27-ubuntu-xenial/ad7f519/ : SUCCESS in 3m 55s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/5/check/gate-octavia-python35/f8f3fdc/ : SUCCESS in 3m 59s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/5/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/3e5cdfd/ : SUCCESS in 46m 56s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/5/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/aeb753d/ : SUCCESS in 48m 40s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/5/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/7fb251b/ : FAILURE in 1h 06m 31s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/5/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/a696936/ : FAILURE in 44m 10s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/5/check/gate-octavia-tox-functional-ubuntu-xenial/73661ad/ : SUCCESS in 3m 09s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/5/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/f07cd0f/ : FAILURE in 44m 01s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/5/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/b25db71/ : FAILURE in 29m 56s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/5/check/gate-octavia-v1-dsvm-scenario-multinode-nv/b70f05b/ : FAILURE in 50m 12s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/5/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/a2a2952/ : FAILURE in 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a9c5da6d6cede2ffe66ab231c36f233c2ff7e9a7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-20 21:39:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"19bca8c4cc5056471a7109138bdf47321fe5ec8b","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-02-20 21:44:46.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/876e3189a2b00b0ede88ae551c255a5a0c811337/3872 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"040c955e72de684c2d4353a49ce04bf2fec7ee82","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-20 22:51:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/6/check/gate-octavia-docs-ubuntu-xenial/46ae3b6//doc/build/html/ : SUCCESS in 2m 29s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/6/check/gate-octavia-pep8-ubuntu-xenial/a44b2ab/ : SUCCESS in 8m 07s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/6/check/gate-octavia-python27-ubuntu-xenial/2a429ea/ : SUCCESS in 4m 08s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/6/check/gate-octavia-python35/00c2d34/ : SUCCESS in 2m 22s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/6/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/66b7bc2/ : SUCCESS in 45m 42s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/6/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/8a4f181/ : SUCCESS in 1h 02m 43s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/6/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/be7d1e3/ : FAILURE in 46m 51s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/6/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/8e3ae8f/ : FAILURE in 1h 12m 11s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/6/check/gate-octavia-tox-functional-ubuntu-xenial/9ea0eb4/ : SUCCESS in 3m 00s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/6/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/629fc45/ : FAILURE in 29m 27s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/6/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/cf8c25e/ : FAILURE in 47m 38s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/6/check/gate-octavia-v1-dsvm-scenario-multinode-nv/f406bf9/ : FAILURE in 58m 54s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/6/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/4e8b689/ : FAILURE in 35m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"54fd6bd4398f4e9454fab87fbc947090512af6ed","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-21 00:54:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"669dc6d6785d99928ca3fcd7c3551e0082daa7ab","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-02-21 00:59:32.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/ffa22b59317a3f42ea7b47ca56dc805646b6b4fc/3873 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"06e308768323c43cf5eb245bec182415e5ff0179","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-21 01:40:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/7/check/gate-octavia-docs-ubuntu-xenial/95984a8//doc/build/html/ : SUCCESS in 4m 54s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/7/check/gate-octavia-pep8-ubuntu-xenial/9f0f1f8/ : SUCCESS in 6m 13s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/7/check/gate-octavia-python27-ubuntu-xenial/253ae27/ : SUCCESS in 2m 53s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/7/check/gate-octavia-python35/68a7aa4/ : SUCCESS in 2m 13s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/7/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/46c82d2/ : SUCCESS in 45m 16s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/7/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/4c2c2a6/ : SUCCESS in 45m 46s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/7/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/44f5ffb/ : FAILURE in 43m 54s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/7/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/cd02df1/ : FAILURE in 44m 49s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/7/check/gate-octavia-tox-functional-ubuntu-xenial/caea452/ : SUCCESS in 2m 35s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/7/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/b6df156/ : FAILURE in 26m 57s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/7/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/e77e052/ : FAILURE in 27m 56s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/7/check/gate-octavia-v1-dsvm-scenario-multinode-nv/cedb1c0/ : FAILURE in 30m 21s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/7/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/c64f049/ : FAILURE in 31m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3c9b813166321dab1c12f9cbf1704e2a75b33c80","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-23 00:25:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"23a7d4566ff8481dff0ba7b7331248360f9c1fd2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-23 00:25:59.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"7efc520557c431b6c23200362cfbca8c5a49ee39","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-23 00:28:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"92502439ba3df54817ee0afa136d08159fd7b9ab","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-02-23 00:32:07.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/04a755a5976b079b4e95449a1fb5f472c538af9a/3877 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"4f9e539ec0af65c9f47a95cbbd0d344251f891d5","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-02-23 00:34:37.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/a8b374a970197d21ae54e7166ad0a8c7f9afed89/3878 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"f10eb0f708505f91653d9670525ca0f63bf86e6c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-23 01:30:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/9/check/gate-octavia-docs-ubuntu-xenial/6aeed13//doc/build/html/ : SUCCESS in 5m 03s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/9/check/gate-octavia-pep8-ubuntu-xenial/d082924/ : SUCCESS in 3m 14s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/9/check/gate-octavia-python27-ubuntu-xenial/cd83950/ : SUCCESS in 2m 51s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/9/check/gate-octavia-python35/e1a5677/ : SUCCESS in 2m 29s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/9/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/02206c0/ : SUCCESS in 44m 22s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/9/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/7f29edd/ : SUCCESS in 45m 38s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/9/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/f35e097/ : FAILURE in 45m 26s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/9/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/631ce0f/ : FAILURE in 55m 10s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/9/check/gate-octavia-tox-functional-ubuntu-xenial/9584f0b/ : SUCCESS in 2m 34s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/9/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/96d724c/ : FAILURE in 26m 59s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/9/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/d003a88/ : FAILURE in 28m 08s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/9/check/gate-octavia-v1-dsvm-scenario-multinode-nv/bb9f7c0/ : FAILURE in 33m 41s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/9/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/4cd547b/ : FAILURE in 59m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8da9d86a46b44a35384c73a39cd7f1bf1783cc58","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-23 10:50:45.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e41b4de8100fa4dcd98a039ae7fc1f6b63c56fb5","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-02-23 10:56:09.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/517abae8855df1aa54dbe5401e035be72fdb38db/3880 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"18f4d5a0d244518112e863cd75aaaf671445ea9a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-23 11:08:18.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/10/check/gate-octavia-docs-ubuntu-xenial/cf09d8b//doc/build/html/ : SUCCESS in 2m 15s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/10/check/gate-octavia-pep8-ubuntu-xenial/15bd264/ : FAILURE in 3m 12s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/10/check/gate-octavia-python27-ubuntu-xenial/0208c11/ : SUCCESS in 2m 13s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/10/check/gate-octavia-python35/4192cdf/ : SUCCESS in 4m 41s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/10/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/b864d4f/ : FAILURE in 5m 37s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/10/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/0eb1ac5/ : FAILURE in 5m 17s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/10/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/b1e60aa/ : FAILURE in 14m 28s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/10/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/d799ea4/ : FAILURE in 5m 18s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/10/check/gate-octavia-tox-functional-ubuntu-xenial/a0ce0c0/ : SUCCESS in 5m 00s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/10/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/2820853/ : FAILURE in 13m 55s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/10/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/1cc2401/ : FAILURE in 13m 49s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/10/check/gate-octavia-v1-dsvm-scenario-multinode-nv/b1f2962/ : FAILURE in 6m 40s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/10/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/ab3408e/ : FAILURE in 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"628e6238bd0c08502f0ddd460a8ccffc5db4dd15","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-02-28 01:06:38.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"51a714f4800b65ddae86d2927537a70aba1d6e9e","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-02-28 01:16:25.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/9aa499c253108115575c0bdffd047b7e0b7da095/3913 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"90f11753f4be0fdf6140a92b4783b9a0d694de5d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-28 02:14:45.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/11/check/gate-octavia-docs-ubuntu-xenial/b0d833a//doc/build/html/ : SUCCESS in 4m 22s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/11/check/gate-octavia-pep8-ubuntu-xenial/cde8f24/ : FAILURE in 5m 24s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/11/check/gate-octavia-python27-ubuntu-xenial/9ca016d/ : SUCCESS in 5m 07s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/11/check/gate-octavia-python35/1d83f6d/ : SUCCESS in 4m 08s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/11/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/3f386e3/ : SUCCESS in 1h 01m 18s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/11/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/9d12a4e/ : SUCCESS in 1h 03m 47s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/11/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/057987f/ : FAILURE in 54m 10s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/11/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/6dd4d4d/ : FAILURE in 27m 11s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/11/check/gate-octavia-tox-functional-ubuntu-xenial/dc7b643/ : SUCCESS in 2m 37s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/11/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/b4b8e22/ : SUCCESS in 45m 38s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/11/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/2d5cb1c/ : SUCCESS in 45m 39s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/11/check/gate-octavia-v1-dsvm-scenario-multinode-nv/600d7b5/ : SUCCESS in 31m 51s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/11/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/3ab24e8/ : SUCCESS in 1h 02m 18s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"763d17ec463fb683c25512689945cc022baaa7fa","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-03-01 01:53:23.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"986f3ae07f72b439c8d5c04cf7bd0a0275f365c1","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-03-01 01:58:50.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/0934d250d65439febe2fa57a25af04a6096123f3/3927 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"46c001e9a264b851354ee8dd5cf2cfe293dac2c4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-01 03:12:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/12/check/gate-octavia-docs-ubuntu-xenial/c73eeb7//doc/build/html/ : SUCCESS in 4m 49s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/12/check/gate-octavia-pep8-ubuntu-xenial/bff7c0e/ : SUCCESS in 4m 01s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/12/check/gate-octavia-python27-ubuntu-xenial/772e9bf/ : SUCCESS in 3m 33s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/12/check/gate-octavia-python35/73db5b8/ : SUCCESS in 5m 52s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/12/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/b805c95/ : SUCCESS in 1h 01m 57s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/12/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/35364d6/ : SUCCESS in 1h 03m 31s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/12/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/6938d68/ : FAILURE in 1h 17m 23s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/12/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/0b0209d/ : FAILURE in 29m 31s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/12/check/gate-octavia-tox-functional-ubuntu-xenial/1352123/ : SUCCESS in 3m 54s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/12/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/8d7a6ff/ : SUCCESS in 44m 47s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/12/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/b937b5f/ : SUCCESS in 29m 31s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/12/check/gate-octavia-v1-dsvm-scenario-multinode-nv/7b9342e/ : SUCCESS in 52m 19s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/12/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/088663c/ : SUCCESS in 35m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"119a7fe68329e0d6525f656b4b71ae15cc8d2d7a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-03 02:43:53.000000000","message":"Patch Set 12: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"318a48f81fd0d536fe729466add4bc2067d2e0a6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-03-07 13:28:11.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d415cb4d58fb1437c31effe55a304162562ad43b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-03-07 13:28:44.000000000","message":"Topic changed from bug/1659488 to floatingip_driver","accounts_in_message":[],"_revision_number":13},{"id":"ad5bbf2a62a92de3960c5a18ef85ea432da74cd0","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-03-07 13:33:40.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/83ac0365dea04109eff2677effa5f3802dd0fbc4/4021 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"c733af92457056bcb17905074198a4184f1035d8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-07 15:01:33.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/13/check/gate-octavia-docs-ubuntu-xenial/56b08a5//doc/build/html/ : SUCCESS in 2m 55s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/13/check/gate-octavia-pep8-ubuntu-xenial/e14e2d9/ : SUCCESS in 3m 40s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/13/check/gate-octavia-python27-ubuntu-xenial/e738281/ : SUCCESS in 4m 46s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/13/check/gate-octavia-python35/3cf743d/ : SUCCESS in 4m 19s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/13/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/b95d227/ : SUCCESS in 1h 03m 51s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/13/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/82e3f87/ : SUCCESS in 1h 00m 20s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/13/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/f43e26f/ : FAILURE in 1h 31m 40s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/13/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/78aced9/ : FAILURE in 54m 56s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/13/check/gate-octavia-tox-functional-ubuntu-xenial/1118dab/ : SUCCESS in 4m 44s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/13/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/64ed7e1/ : SUCCESS in 46m 18s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/13/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/3ff0d53/ : SUCCESS in 49m 11s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/13/check/gate-octavia-v1-dsvm-scenario-multinode-nv/828d26a/ : SUCCESS in 35m 42s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/13/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/b9f9a7d/ : SUCCESS in 1h 01m 50s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/13/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/a29b9c7/ : SUCCESS in 48m 16s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"7a4975129ca2b1342432cde3fc09d11933ffcaeb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-03-17 02:09:18.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"86c67ecede00c68d53ee6fded140bdb0d1f022e5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-03-17 02:10:31.000000000","message":"Patch Set 14: Workflow-1","accounts_in_message":[],"_revision_number":14},{"id":"f86565f0e0b5684cfa5759db33377ed341f233ed","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-03-17 02:14:32.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/fc67b5d778e5b04e194bb4fea404e2883f3cb461/4125 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"701196592f4001b932e8b0fc78e94ad66ab39478","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-17 03:42:25.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/14/check/gate-octavia-docs-ubuntu-xenial/f76db26//doc/build/html/ : SUCCESS in 4m 50s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/14/check/gate-octavia-pep8-ubuntu-xenial/281bca7/ : SUCCESS in 5m 37s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/14/check/gate-octavia-python27-ubuntu-xenial/b2825d9/ : SUCCESS in 4m 54s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/14/check/gate-octavia-python35/30820bb/ : SUCCESS in 5m 03s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/14/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/7bbe358/ : SUCCESS in 1h 02m 03s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/14/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/51d63c6/ : SUCCESS in 47m 45s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/14/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/8dae9d5/ : FAILURE in 1h 32m 07s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/14/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/858e737/ : FAILURE in 31m 50s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/14/check/gate-octavia-tox-functional-ubuntu-xenial/eeaa3ca/ : FAILURE in 4m 21s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/14/check/gate-octavia-tox-functional-py35-ubuntu-xenial/9c61faf/ : FAILURE in 2m 57s (non-voting)\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/14/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/d710ff0/ : SUCCESS in 29m 46s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/14/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/d7c4ede/ : FAILURE in 16m 59s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/14/check/gate-octavia-v1-dsvm-scenario-multinode-nv/484c552/ : SUCCESS in 32m 23s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/14/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/8d4a9fe/ : FAILURE in 31m 04s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/14/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/78a0d1f/ : SUCCESS in 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c077d7e977c9fab0721d4b5656601f1b093f7e6d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-03-18 06:17:41.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"22c4868d76601139048d2809ddbbf3fe7a2c17fd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-03-18 06:17:49.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"ec1ea4fc82bb148a74be91dfaa9e25ba8151148f","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-03-18 06:23:17.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/cb27feea1bf65c5b865083c2cb24d71badae0d1c/4130 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"597fd8d364d9df532c5654d9092cf4edd47b5c19","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-18 07:35:27.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/15/check/gate-octavia-docs-ubuntu-xenial/ed8ae29//doc/build/html/ : SUCCESS in 4m 04s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/15/check/gate-octavia-pep8-ubuntu-xenial/d134ec9/ : SUCCESS in 3m 55s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/15/check/gate-octavia-python27-ubuntu-xenial/8b0aac9/ : SUCCESS in 4m 04s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/15/check/gate-octavia-python35/d084b0b/ : SUCCESS in 2m 27s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/15/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/90410c3/ : SUCCESS in 1h 12m 07s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/15/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/bbc05fd/ : SUCCESS in 1h 03m 11s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/15/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/e9efceb/ : FAILURE in 1h 16m 44s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/15/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/08d96ce/ : FAILURE in 30m 28s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/15/check/gate-octavia-tox-functional-ubuntu-xenial/3738246/ : SUCCESS in 3m 00s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/15/check/gate-octavia-tox-functional-py35-ubuntu-xenial/22d21b0/ : SUCCESS in 2m 58s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/15/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/86c6340/ : SUCCESS in 28m 57s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/15/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/8c5e4e6/ : FAILURE in 17m 27s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/15/check/gate-octavia-v1-dsvm-scenario-multinode-nv/6dc6765/ : SUCCESS in 38m 19s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/15/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/a2fad44/ : FAILURE in 32m 46s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/15/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/cf28a5a/ : SUCCESS in 34m 47s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"877ed16f23e063fafe11dcf84c6ee59854acba0a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-11 09:00:13.000000000","message":"Patch Set 15: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"84038e7101348e74894fe400e230d0c40c8d5b93","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-04-11 19:30:33.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"9fbd3f0cc19bbbb1887163522633a53efc9a7ce5","author":{"_account_id":16923,"name":"HP Octavia Sonar CI","email":"octavia.sonar@gmail.com","username":"octaviasonar","tags":["SERVICE_USER"]},"date":"2017-04-11 19:38:17.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* Rackspace-Octavia-Sonar-CI http://sonar.octavia.cloud/archive/octavia/35eb352d0e0228a4b405001fcf5b739fb1d3fc7d/4417 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"af6fd7351dec2d0f408280dba2f538514585abcd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-11 20:48:51.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/16/check/gate-octavia-docs-ubuntu-xenial/5a4bb96//doc/build/html/ : SUCCESS in 3m 47s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/16/check/gate-octavia-pep8-ubuntu-xenial/2b471b2/ : SUCCESS in 5m 23s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/16/check/gate-octavia-python27-ubuntu-xenial/2642bbd/ : SUCCESS in 3m 44s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/16/check/gate-octavia-python35/2c592f5/ : SUCCESS in 5m 04s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/16/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/4728817/ : SUCCESS in 50m 52s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/16/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/a2a9d6e/ : SUCCESS in 42m 15s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/16/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/87b1db2/ : FAILURE in 1h 16m 51s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/16/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/a26d6dc/ : FAILURE in 16m 53s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/16/check/gate-octavia-tox-functional-ubuntu-xenial/a7303c5/ : SUCCESS in 4m 04s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/16/check/gate-octavia-tox-functional-py35-ubuntu-xenial/6961311/ : SUCCESS in 4m 12s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/16/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial-nv/3a6ec60/ : SUCCESS in 48m 28s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/16/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial-nv/f63f001/ : FAILURE in 26m 27s (non-voting)\n- gate-octavia-v1-dsvm-scenario-multinode-nv http://logs.openstack.org/12/435612/16/check/gate-octavia-v1-dsvm-scenario-multinode-nv/e315699/ : SUCCESS in 34m 00s (non-voting)\n- gate-octavia-v1-dsvm-py3x-scenario-multinode-nv http://logs.openstack.org/12/435612/16/check/gate-octavia-v1-dsvm-py3x-scenario-multinode-nv/68c3999/ : FAILURE in 26m 46s (non-voting)\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/16/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/2867359/ : FAILURE in 44m 51s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/16/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/cfd8dca/ : FAILURE in 33m 52s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"9f71c3977a06807229793af85e2136b5702e8b10","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-02 09:16:13.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"1388cbade48201c6a91913204754c771bf3802ed","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-02 09:16:19.000000000","message":"Patch Set 17: Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"3e9d976a3b4124e6514ae552dd169ed458072f11","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-02 10:53:16.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/17/check/gate-octavia-docs-ubuntu-xenial/0a04ee4//doc/build/html/ : SUCCESS in 5m 29s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/17/check/gate-octavia-pep8-ubuntu-xenial/b6670c1/ : SUCCESS in 6m 19s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/17/check/gate-octavia-python27-ubuntu-xenial/6fa5ba9/ : SUCCESS in 4m 20s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/17/check/gate-octavia-python35/2c2136f/ : SUCCESS in 2m 46s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/17/check/octavia-coverage-ubuntu-xenial/a8f2aea/ : SUCCESS in 5m 41s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/17/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/181677d/ : SUCCESS in 1h 07m 57s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/17/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/ea76e6d/ : SUCCESS in 50m 18s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/17/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/3f3e5b6/ : FAILURE in 1h 29m 06s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/17/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/e3aea6e/ : FAILURE in 1h 34m 21s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/17/check/gate-octavia-tox-functional-ubuntu-xenial/9a95264/ : SUCCESS in 4m 43s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/17/check/gate-octavia-tox-functional-py35-ubuntu-xenial/c305dd3/ : SUCCESS in 5m 41s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/17/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/5c8060c/ : SUCCESS in 42m 23s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/17/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/2816acc/ : SUCCESS in 27m 56s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/17/check/gate-octavia-v1-dsvm-scenario-multinode/1334c35/ : SUCCESS in 35m 05s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/17/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/48e8e73/ : SUCCESS in 1h 01m 13s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/17/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/ab91372/ : SUCCESS in 44m 33s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/17/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/a4d7608/ : FAILURE in 36m 30s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0a00cbdf3f47af428902c73c54fa22eed5e55f10","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-02 20:00:25.000000000","message":"Patch Set 17: Workflow-1\n\numm ok, somehow I tried to put -A and it put +A lol, bad misclick :P","accounts_in_message":[],"_revision_number":17},{"id":"35b9469d077267162aba53d8cd31bccb60870a3a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-05 04:52:31.000000000","message":"Patch Set 17: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":17},{"id":"d01fd9db6fe34ef43e23eeecd9e393e2b9a6671c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-05 07:25:29.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"ae16e62c0aab395d45003207bae9bb2c35d90145","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-05 07:25:41.000000000","message":"Patch Set 18: Workflow-1","accounts_in_message":[],"_revision_number":18},{"id":"6a679a9c7c239815cbe4702e267823bc5edfaceb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-05 08:46:46.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/18/check/gate-octavia-docs-ubuntu-xenial/54f1058//doc/build/html/ : SUCCESS in 3m 19s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/18/check/gate-octavia-pep8-ubuntu-xenial/4191170/ : SUCCESS in 6m 19s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/18/check/gate-octavia-python27-ubuntu-xenial/0b34bf5/ : SUCCESS in 3m 54s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/18/check/gate-octavia-python35/f4bc2c7/ : SUCCESS in 4m 43s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/18/check/octavia-coverage-ubuntu-xenial/1fe9a4c/ : SUCCESS in 6m 09s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/18/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/89abc50/ : SUCCESS in 54m 21s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/18/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/328fa7e/ : SUCCESS in 42m 49s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/18/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/cbd692d/ : FAILURE in 1h 20m 10s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/18/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/3c199d0/ : FAILURE in 1h 19m 37s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/18/check/gate-octavia-tox-functional-ubuntu-xenial/794dc5a/ : SUCCESS in 4m 20s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/18/check/gate-octavia-tox-functional-py35-ubuntu-xenial/4465661/ : SUCCESS in 4m 07s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/18/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/440feb6/ : SUCCESS in 29m 54s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/18/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/71ff4be/ : SUCCESS in 39m 13s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/18/check/gate-octavia-v1-dsvm-scenario-multinode/8521cd0/ : SUCCESS in 34m 33s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/18/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/180e380/ : SUCCESS in 36m 05s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/18/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/751bb2c/ : SUCCESS in 45m 27s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/18/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/135e8ca/ : FAILURE in 35m 55s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"912e1e86b3c050287163e887145aa25bcb3939f4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-05 21:35:06.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"eaae17f7ba0c686ada4c689ec1c92a89a9141da2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-05 21:35:13.000000000","message":"Patch Set 19: Workflow-1","accounts_in_message":[],"_revision_number":19},{"id":"de885dfdf93077ff4dc311b58e23861902619bc6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-05 23:08:08.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/19/check/gate-octavia-docs-ubuntu-xenial/1a53542//doc/build/html/ : SUCCESS in 2m 39s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/19/check/gate-octavia-pep8-ubuntu-xenial/4f1a833/ : SUCCESS in 5m 44s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/19/check/gate-octavia-python27-ubuntu-xenial/52946ff/ : SUCCESS in 3m 49s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/19/check/gate-octavia-python35/d1ed464/ : SUCCESS in 3m 46s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/19/check/octavia-coverage-ubuntu-xenial/4f71138/ : SUCCESS in 7m 55s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/19/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/5d9d8ca/ : SUCCESS in 44m 56s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/19/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/c374f27/ : SUCCESS in 40m 51s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/19/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/2f5c066/ : FAILURE in 1h 30m 58s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/19/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/7e5fd8a/ : FAILURE in 1h 21m 14s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/19/check/gate-octavia-tox-functional-ubuntu-xenial/9af0036/ : SUCCESS in 7m 57s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/19/check/gate-octavia-tox-functional-py35-ubuntu-xenial/d390085/ : SUCCESS in 5m 35s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/19/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/d94be64/ : SUCCESS in 42m 04s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/19/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/d7d53a7/ : SUCCESS in 39m 41s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/19/check/gate-octavia-v1-dsvm-scenario-multinode/99d7d9c/ : SUCCESS in 34m 34s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/19/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/53c8574/ : SUCCESS in 34m 38s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/19/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/d8bb195/ : SUCCESS in 45m 30s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/19/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/6cd0302/ : FAILURE in 36m 53s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"95cc09f19286346a65fca614d68fda72add8c309","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-10 04:51:13.000000000","message":"Patch Set 19: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":19},{"id":"79dc47b31f3dc5d175d08a7b0a03567962d13460","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-10 05:10:41.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"3bfdeec662f2492957c9ceab1e9f75e0a52d67ac","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-10 05:10:51.000000000","message":"Patch Set 20: Workflow-1","accounts_in_message":[],"_revision_number":20},{"id":"2484069220bb9129a11bf17830cbb374686ccb57","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-10 06:08:40.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"4b67ca20777066a8f70686c9d9d9c0608210441d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-10 06:10:32.000000000","message":"Patch Set 21: Workflow-1","accounts_in_message":[],"_revision_number":21},{"id":"05c102687d81913c0cf2dcd2d9cb2e2b0b0fe28b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-10 07:49:32.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/21/check/gate-octavia-docs-ubuntu-xenial/26c61b4//doc/build/html/ : SUCCESS in 4m 20s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-octavia-pep8-ubuntu-xenial/0ae4290/ : SUCCESS in 6m 39s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-octavia-python27-ubuntu-xenial/6b721e8/ : SUCCESS in 5m 15s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/21/check/gate-octavia-python35/64d6d45/ : SUCCESS in 4m 22s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/octavia-coverage-ubuntu-xenial/7f059e8/ : SUCCESS in 7m 13s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/d32095a/ : SUCCESS in 59m 35s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/21/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/28d97ed/ : SUCCESS in 55m 39s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/21/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/76dbbd0/ : SUCCESS in 1h 40m 38s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/21/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/08d41ab/ : SUCCESS in 1h 40m 00s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-octavia-tox-functional-ubuntu-xenial/0738a4c/ : SUCCESS in 7m 01s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-octavia-tox-functional-py35-ubuntu-xenial/c07de88/ : SUCCESS in 6m 01s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/784ab1f/ : SUCCESS in 1h 01m 20s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/ba88772/ : FAILURE in 1h 02m 00s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/21/check/gate-octavia-v1-dsvm-scenario-multinode/d89ebaa/ : SUCCESS in 1h 10m 30s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/21/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/bedaba7/ : SUCCESS in 42m 28s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/21/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/512405b/ : SUCCESS in 57m 36s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/21/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/f3a8ac6/ : FAILURE in 23m 26s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"b37f23dce7a84f9415b08372cb5ce1e4ad0f963b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-10 11:46:25.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"2d762a8bcd1ab567b7a196a8cd5a551f832d1efb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-10 12:55:46.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/21/check/gate-octavia-docs-ubuntu-xenial/88dc6cd//doc/build/html/ : SUCCESS in 4m 36s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-octavia-pep8-ubuntu-xenial/1cab1d0/ : SUCCESS in 6m 30s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-octavia-python27-ubuntu-xenial/1892cf5/ : SUCCESS in 4m 55s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/21/check/gate-octavia-python35/8697914/ : SUCCESS in 3m 17s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/octavia-coverage-ubuntu-xenial/319c474/ : SUCCESS in 7m 21s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/17788d6/ : SUCCESS in 57m 28s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/21/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/4b297d4/ : SUCCESS in 1h 05m 36s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/21/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/a0331bd/ : SUCCESS in 50m 02s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/21/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/8a39e1f/ : SUCCESS in 52m 29s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-octavia-tox-functional-ubuntu-xenial/5d5c458/ : SUCCESS in 5m 17s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-octavia-tox-functional-py35-ubuntu-xenial/ff7bfe1/ : SUCCESS in 7m 02s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/3f7a017/ : SUCCESS in 32m 48s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/21/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/4c2d750/ : SUCCESS in 34m 41s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/21/check/gate-octavia-v1-dsvm-scenario-multinode/088c5ef/ : SUCCESS in 1h 03m 58s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/21/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/004b56b/ : SUCCESS in 38m 12s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/21/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/9a002b6/ : SUCCESS in 45m 54s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/21/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/569b18f/ : FAILURE in 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"80e5caacaa153c1d8673b79ea126e06cdb75a821","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-10 13:55:03.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"8c5e16ecb3e180869d79d4b6cf62e1a7cf548517","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-10 13:55:09.000000000","message":"Patch Set 22: Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"0b24641644bd17137acbf558da167b61486ef403","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-10 14:43:53.000000000","message":"Patch Set 22: Workflow-1\n\nWHY DOES MOBILE KEEP DOING +A INSTEAD OF -A \u003e_\u003c","accounts_in_message":[],"_revision_number":22},{"id":"798a30b26848f7e8e38a8b94ccb6907f51911581","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-10 14:51:06.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/22/check/gate-octavia-docs-ubuntu-xenial/c5d41f1//doc/build/html/ : SUCCESS in 2m 57s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/22/check/gate-octavia-pep8-ubuntu-xenial/aa14c9d/ : SUCCESS in 3m 44s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/22/check/gate-octavia-python27-ubuntu-xenial/a5658c9/ : SUCCESS in 3m 02s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/22/check/gate-octavia-python35/ff77d4c/ : SUCCESS in 3m 10s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/22/check/octavia-coverage-ubuntu-xenial/d76ec73/ : SUCCESS in 9m 19s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/22/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/7b5636b/ : SUCCESS in 44m 58s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/22/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/56ef70a/ : SUCCESS in 43m 11s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/22/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/4fc034a/ : SUCCESS in 52m 34s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/22/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/42a9405/ : SUCCESS in 38m 47s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/22/check/gate-octavia-tox-functional-ubuntu-xenial/3b71a1c/ : SUCCESS in 11m 08s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/22/check/gate-octavia-tox-functional-py35-ubuntu-xenial/26ca908/ : SUCCESS in 5m 57s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/22/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/0a60a49/ : SUCCESS in 39m 51s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/22/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/448ea1b/ : SUCCESS in 41m 24s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/22/check/gate-octavia-v1-dsvm-scenario-multinode/14c1d47/ : SUCCESS in 48m 07s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/22/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/bfc1369/ : SUCCESS in 34m 44s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/22/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/b552a41/ : SUCCESS in 53m 57s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/22/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/174dd2a/ : FAILURE in 36m 22s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"e1cf132f1b1c53437b70715e24dbeb4f911ab0b1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-11 17:54:58.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"ca60027c1006db3aa912aec0368e46cd45b12085","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-05-11 17:55:02.000000000","message":"Patch Set 23: Workflow-1","accounts_in_message":[],"_revision_number":23},{"id":"c1c137449016ae075cb66ea3d5e56b7622687f4d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-11 19:09:38.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/23/check/gate-octavia-docs-ubuntu-xenial/713000c//doc/build/html/ : SUCCESS in 2m 25s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/23/check/gate-octavia-pep8-ubuntu-xenial/4748c6a/ : SUCCESS in 3m 34s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/23/check/gate-octavia-python27-ubuntu-xenial/12ffa58/ : SUCCESS in 2m 25s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/23/check/gate-octavia-python35/57d06c9/ : SUCCESS in 2m 25s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/23/check/octavia-coverage-ubuntu-xenial/2b92fea/ : SUCCESS in 7m 31s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/23/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/9372dab/ : SUCCESS in 57m 13s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/23/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/32a8885/ : SUCCESS in 54m 43s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/23/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/211cbf2/ : SUCCESS in 1h 13m 41s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/23/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/f91072f/ : SUCCESS in 39m 26s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/23/check/gate-octavia-tox-functional-ubuntu-xenial/6c2b59a/ : SUCCESS in 4m 51s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/23/check/gate-octavia-tox-functional-py35-ubuntu-xenial/1533efd/ : SUCCESS in 4m 50s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/23/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/5eaea47/ : SUCCESS in 30m 07s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/23/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/ac3309b/ : SUCCESS in 28m 06s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/23/check/gate-octavia-v1-dsvm-scenario-multinode/cd5faf8/ : SUCCESS in 1h 02m 44s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/23/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/42fe507/ : SUCCESS in 34m 48s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/23/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/36d483d/ : SUCCESS in 29m 58s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/23/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/7d0bc63/ : FAILURE in 33m 43s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"11071385ff157177904442bb9e4545c6f9e0bf27","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-06-02 19:14:30.000000000","message":"Patch Set 24: Patch Set 23 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"1a127e8658e9f53829b44627d198daa9ca04c7e4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-06-02 19:14:36.000000000","message":"Patch Set 24: Workflow-1","accounts_in_message":[],"_revision_number":24},{"id":"2f7681957bbf62381172d8ffeab17614e7bac6db","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-02 20:51:39.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/24/check/gate-octavia-docs-ubuntu-xenial/7d71721//doc/build/html/ : SUCCESS in 4m 08s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/24/check/gate-octavia-pep8-ubuntu-xenial/6c6bc91/ : SUCCESS in 5m 41s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/24/check/gate-octavia-python27-ubuntu-xenial/c464abd/ : SUCCESS in 4m 10s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/24/check/gate-octavia-python35/dd6d1d8/ : SUCCESS in 4m 29s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/24/check/octavia-coverage-ubuntu-xenial/5ce82dc/ : SUCCESS in 7m 49s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/24/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/b4ddbaa/ : SUCCESS in 1h 12m 38s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/24/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/9099f17/ : SUCCESS in 58m 42s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/24/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/4db4947/ : SUCCESS in 1h 26m 51s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/24/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/3a0c062/ : SUCCESS in 42m 25s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/24/check/gate-octavia-tox-functional-ubuntu-xenial/81ed973/ : SUCCESS in 6m 28s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/24/check/gate-octavia-tox-functional-py35-ubuntu-xenial/23523b7/ : SUCCESS in 4m 37s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/24/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/2959804/ : SUCCESS in 46m 54s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/24/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/3a695ab/ : SUCCESS in 46m 19s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/24/check/gate-octavia-v1-dsvm-scenario-multinode/457ce2f/ : FAILURE in 1h 03m 25s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/24/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/af007fe/ : SUCCESS in 1h 04m 13s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/24/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/4c7fbbc/ : SUCCESS in 36m 02s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/24/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/7d18ac6/ : FAILURE in 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"051eb3ca3d62e163d6e8c014712c6b28211d5f7a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-06-06 00:53:48.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":24},{"id":"f063e699e2f2307920df323ac03326926fe87ae6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-06-06 00:53:54.000000000","message":"Patch Set 25: Workflow-1","accounts_in_message":[],"_revision_number":25},{"id":"0372f720d3a312912fb4b581f9048fad102aae52","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-06 02:27:03.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/25/check/gate-octavia-docs-ubuntu-xenial/08b3a34//doc/build/html/ : SUCCESS in 4m 47s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/25/check/gate-octavia-pep8-ubuntu-xenial/9e026a1/ : SUCCESS in 6m 22s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/25/check/gate-octavia-python27-ubuntu-xenial/821d125/ : SUCCESS in 4m 45s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/25/check/gate-octavia-python35/19a1aa7/ : SUCCESS in 4m 44s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/25/check/octavia-coverage-ubuntu-xenial/ead45de/ : SUCCESS in 7m 46s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/25/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/bfc4878/ : SUCCESS in 1h 07m 38s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/25/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/60865b6/ : SUCCESS in 1h 02m 30s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/25/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/fa1e930/ : SUCCESS in 59m 47s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/25/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/b30b550/ : SUCCESS in 1h 31m 34s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/25/check/gate-octavia-tox-functional-ubuntu-xenial/d12c874/ : SUCCESS in 4m 22s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/25/check/gate-octavia-tox-functional-py35-ubuntu-xenial/1c24b80/ : SUCCESS in 6m 49s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/25/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/39d0535/ : SUCCESS in 1h 01m 39s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/25/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/02ea461/ : SUCCESS in 50m 44s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/25/check/gate-octavia-v1-dsvm-scenario-multinode/f98bf00/ : SUCCESS in 37m 58s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/25/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/81bdf87/ : SUCCESS in 37m 07s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/25/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/21c7263/ : SUCCESS in 1h 09m 27s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/25/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/31721d0/ : FAILURE in 45m 06s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"b57f5eaac47b8ea8f90e0f8a70e915b52bc9ddf4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-06-12 23:53:23.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"79a86830061ebdaedfe5c089344e123bd89176ec","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-06-12 23:53:30.000000000","message":"Patch Set 26: Workflow-1","accounts_in_message":[],"_revision_number":26},{"id":"81d3666d84265d3fb9ac0506279a95a578d8e006","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-13 01:14:55.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/26/check/gate-octavia-docs-ubuntu-xenial/c03578e//doc/build/html/ : SUCCESS in 2m 42s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/26/check/gate-octavia-pep8-ubuntu-xenial/83b613b/ : SUCCESS in 7m 44s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/26/check/gate-octavia-python27-ubuntu-xenial/b009d17/ : SUCCESS in 3m 51s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/26/check/gate-octavia-python35/8618fb6/ : SUCCESS in 3m 51s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/26/check/octavia-coverage-ubuntu-xenial/598b4e3/ : SUCCESS in 7m 59s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/26/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/3e0ce66/ : FAILURE in 8m 54s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/26/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/9795b55/ : SUCCESS in 42m 04s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/26/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/a078870/ : FAILURE in 7m 59s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/26/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/ac10674/ : SUCCESS in 1h 20m 13s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/26/check/gate-octavia-tox-functional-ubuntu-xenial/31a99dd/ : SUCCESS in 8m 19s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/26/check/gate-octavia-tox-functional-py35-ubuntu-xenial/15fca90/ : SUCCESS in 7m 19s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/26/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/f5652e8/ : FAILURE in 22m 13s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/26/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/1f86c39/ : SUCCESS in 50m 07s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/26/check/gate-octavia-v1-dsvm-scenario-multinode/9ede266/ : FAILURE in 10m 15s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/26/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/8baf618/ : SUCCESS in 38m 54s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/26/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/04e413d/ : SUCCESS in 53m 41s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/26/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/7a6d5de/ : FAILURE in 6m 42s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"19578847164565e9ae4b645d52e5866315d071c8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-06-29 01:49:47.000000000","message":"Patch Set 27: Patch Set 26 was rebased","accounts_in_message":[],"_revision_number":26},{"id":"9a6984f35e965e6053c07d3786b7f00440117a1b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-06-29 01:49:52.000000000","message":"Patch Set 27: Workflow-1","accounts_in_message":[],"_revision_number":27},{"id":"d77baaf7d61bce6bd4969cdb585973f2aa4c8223","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-29 02:59:04.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/27/check/gate-octavia-docs-ubuntu-xenial/2274c8b//doc/build/html/ : SUCCESS in 3m 06s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/27/check/gate-octavia-pep8-ubuntu-xenial/3ad2c58/ : SUCCESS in 5m 32s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/27/check/gate-octavia-python27-ubuntu-xenial/44c07a6/ : SUCCESS in 7m 49s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/27/check/gate-octavia-python35/3e76e64/ : SUCCESS in 7m 48s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/27/check/octavia-coverage-ubuntu-xenial/1433ae6/ : SUCCESS in 10m 39s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/27/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/fd21b27/ : SUCCESS in 1h 06m 34s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/27/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/38ba208/ : SUCCESS in 54m 26s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/27/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/d095438/ : SUCCESS in 45m 36s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/27/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/5867f35/ : SUCCESS in 49m 17s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/27/check/gate-octavia-tox-functional-ubuntu-xenial/bcd34c7/ : SUCCESS in 6m 59s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/27/check/gate-octavia-tox-functional-py35-ubuntu-xenial/744151c/ : SUCCESS in 7m 07s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/27/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/03ca6df/ : SUCCESS in 38m 59s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/27/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/391d128/ : SUCCESS in 37m 45s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/27/check/gate-octavia-v1-dsvm-scenario-multinode/d086b83/ : SUCCESS in 46m 23s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/27/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/8a60229/ : SUCCESS in 52m 37s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/27/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/f872350/ : SUCCESS in 42m 09s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/27/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/21a9629/ : FAILURE in 43m 28s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"7b21532d86fb18a782d99d8f0b9df8bd188ac47b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-06-29 21:31:10.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"3b9d8a2a80c35be9785be7621a808e12a2802076","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-06-29 21:31:52.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"087a37a3007271fb65fcc6aafc2ad0ebf2ede60d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-29 22:52:19.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/29/check/gate-octavia-docs-ubuntu-xenial/8f1b084//doc/build/html/ : SUCCESS in 3m 59s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/29/check/gate-octavia-pep8-ubuntu-xenial/56343e5/ : SUCCESS in 5m 35s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/29/check/gate-octavia-python27-ubuntu-xenial/739fb72/ : SUCCESS in 4m 09s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/29/check/gate-octavia-python35/d6de5bc/ : SUCCESS in 2m 43s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/29/check/octavia-coverage-ubuntu-xenial/ee58d6f/ : SUCCESS in 8m 58s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/29/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/2fc027d/ : SUCCESS in 45m 28s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/29/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/d3ee7bd/ : SUCCESS in 45m 08s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/29/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/29b86ac/ : SUCCESS in 36m 12s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/29/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/9cfbed9/ : SUCCESS in 1h 19m 26s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/29/check/gate-octavia-tox-functional-ubuntu-xenial/675e5f0/ : SUCCESS in 9m 01s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/29/check/gate-octavia-tox-functional-py35-ubuntu-xenial/89a4495/ : SUCCESS in 7m 03s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/29/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/d5ae285/ : SUCCESS in 27m 26s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/29/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/ef81fe8/ : SUCCESS in 49m 33s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/29/check/gate-octavia-v1-dsvm-scenario-multinode/9eb2816/ : SUCCESS in 35m 21s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/29/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/8c99af0/ : SUCCESS in 46m 42s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/29/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/56866a0/ : SUCCESS in 41m 57s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/29/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/37c5143/ : FAILURE in 40m 59s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"d45aa3eed356794b58e284efe653c63b2825d310","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-06-30 01:43:34.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"e76bc9d6dbc94812c49e6cb01455e2e94f32f6a8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-30 02:20:08.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/30/check/gate-octavia-docs-ubuntu-xenial/0cd4556//doc/build/html/ : SUCCESS in 2m 33s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-octavia-pep8-ubuntu-xenial/e85b388/ : SUCCESS in 5m 44s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-octavia-python27-ubuntu-xenial/388ed4a/ : SUCCESS in 4m 34s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/30/check/gate-octavia-python35/5ec7e44/ : SUCCESS in 4m 22s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/octavia-coverage-ubuntu-xenial/9518568/ : SUCCESS in 7m 35s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/3f3dafe/ : FAILURE in 23m 30s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/30/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/0a733d3/ : FAILURE in 31m 29s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/30/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/6448f37/ : FAILURE in 30m 34s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/30/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/edb0f95/ : FAILURE in 22m 50s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-octavia-tox-functional-ubuntu-xenial/6ab66d1/ : SUCCESS in 4m 50s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-octavia-tox-functional-py35-ubuntu-xenial/986c246/ : SUCCESS in 5m 04s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/fbc7568/ : FAILURE in 20m 48s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/7e5d816/ : FAILURE in 19m 35s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/30/check/gate-octavia-v1-dsvm-scenario-multinode/dd01fc3/ : FAILURE in 28m 58s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/30/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/1aae0f3/ : FAILURE in 21m 12s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/30/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/352c436/ : FAILURE in 35m 12s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/30/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/3d44dbb/ : FAILURE in 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"3a6f6673de7e795afadb31d9c39d171df6e14612","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-07-01 09:20:48.000000000","message":"Patch Set 30:\n\nrecheck","accounts_in_message":[],"_revision_number":30},{"id":"5ca3191d32c9269ad76289153ccdce7916a429ad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-01 10:19:45.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/30/check/gate-octavia-docs-ubuntu-xenial/cb5328f//doc/build/html/ : SUCCESS in 2m 30s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-octavia-pep8-ubuntu-xenial/7e52131/ : SUCCESS in 5m 17s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-octavia-python27-ubuntu-xenial/fd26061/ : SUCCESS in 2m 29s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/30/check/gate-octavia-python35/ce64f36/ : SUCCESS in 2m 31s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/octavia-coverage-ubuntu-xenial/6d70659/ : SUCCESS in 5m 41s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/e31e2b9/ : SUCCESS in 44m 18s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/30/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/273efe5/ : SUCCESS in 41m 14s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/30/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/1cd4d52/ : SUCCESS in 35m 29s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/30/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/03bda8e/ : SUCCESS in 36m 07s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-octavia-tox-functional-ubuntu-xenial/c8ac3b3/ : SUCCESS in 6m 23s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-octavia-tox-functional-py35-ubuntu-xenial/e58f8c2/ : SUCCESS in 6m 05s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/596c889/ : SUCCESS in 27m 00s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/30/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/eea9fce/ : SUCCESS in 25m 41s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/30/check/gate-octavia-v1-dsvm-scenario-multinode/8d96eba/ : SUCCESS in 58m 43s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/30/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/c193300/ : SUCCESS in 31m 29s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/30/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/9fbeee9/ : SUCCESS in 30m 44s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/30/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/dc4966f/ : FAILURE in 22m 21s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"e604330e33c118360c9f7483daa3b5d398cebedf","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-07-07 22:05:04.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.","accounts_in_message":[],"_revision_number":31},{"id":"edfe051a379ddde48b7101f08fc6da296ffcc666","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-07 23:35:12.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/31/check/gate-octavia-docs-ubuntu-xenial/881918c//doc/build/html/ : SUCCESS in 6m 10s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/31/check/gate-octavia-pep8-ubuntu-xenial/591e850/ : SUCCESS in 6m 11s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/31/check/gate-octavia-python27-ubuntu-xenial/22d4a9e/ : SUCCESS in 6m 42s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/31/check/gate-octavia-python35/587b787/ : SUCCESS in 4m 47s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/31/check/octavia-coverage-ubuntu-xenial/b8fdded/ : SUCCESS in 8m 27s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/31/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/99fa31b/ : SUCCESS in 1h 03m 08s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/31/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/b402388/ : SUCCESS in 57m 12s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/31/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/24f8844/ : SUCCESS in 1h 18m 21s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/31/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/d59fec4/ : SUCCESS in 1h 27m 38s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/31/check/gate-octavia-tox-functional-ubuntu-xenial/f924247/ : SUCCESS in 7m 14s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/31/check/gate-octavia-tox-functional-py35-ubuntu-xenial/c05653f/ : SUCCESS in 7m 40s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/31/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/89b3853/ : SUCCESS in 50m 25s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/31/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/7f844c5/ : SUCCESS in 37m 58s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/31/check/gate-octavia-v1-dsvm-scenario-multinode/51a7397/ : SUCCESS in 57m 41s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/31/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/2be74c4/ : SUCCESS in 43m 18s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/31/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/38ee899/ : SUCCESS in 35m 24s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/31/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/30f3d18/ : FAILURE in 42m 29s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"d49e06a044e14ad88e04a56879b516112763fa73","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-26 20:39:10.000000000","message":"Patch Set 31: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":31},{"id":"ab89160464ce13cb4181ec6f8295102bc5d816e7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-02 09:30:05.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"b1e927cd0ccb1a6c26a8ef3cb0738630609fc24d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-02 09:30:25.000000000","message":"Patch Set 32: Workflow+1","accounts_in_message":[],"_revision_number":32},{"id":"15c70bd6656d6713d2421562caa1a344ae90bdfa","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-02 09:30:28.000000000","message":"Patch Set 32: Workflow-1","accounts_in_message":[],"_revision_number":32},{"id":"31e72f098d435491b9e6ec5872729bbe1e6084d2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-02 11:30:08.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/32/check/gate-octavia-docs-ubuntu-xenial/4f8d051//doc/build/html/ : SUCCESS in 5m 09s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/32/check/gate-octavia-pep8-ubuntu-xenial/fce860d/ : SUCCESS in 7m 36s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/32/check/gate-octavia-python27-ubuntu-xenial/334179d/ : SUCCESS in 5m 23s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/32/check/gate-octavia-python35/902672f/ : SUCCESS in 4m 07s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/32/check/octavia-coverage-ubuntu-xenial/0452a91/ : SUCCESS in 10m 44s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/32/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/1117d20/ : SUCCESS in 57m 08s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/32/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/36a2acf/ : SUCCESS in 56m 02s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/32/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/1a060c3/ : SUCCESS in 1h 28m 00s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/32/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/6e51b39/ : SUCCESS in 54m 56s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/32/check/gate-octavia-tox-functional-ubuntu-xenial/825b188/ : SUCCESS in 8m 32s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/32/check/gate-octavia-tox-functional-py35-ubuntu-xenial/4065539/ : SUCCESS in 8m 15s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/32/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/e7615f6/ : FAILURE in 1h 57m 11s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/32/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/ab144f9/ : SUCCESS in 44m 02s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/32/check/gate-octavia-v1-dsvm-scenario-multinode/1e0c57d/ : SUCCESS in 1h 00m 32s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/32/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/2e89d0e/ : SUCCESS in 58m 32s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/32/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/5f8b388/ : SUCCESS in 41m 23s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/32/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/3fd59fa/ : FAILURE in 34m 47s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"69a40d98d6f4ee4892e3af806be9fd6416f652f4","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-08-02 12:44:58.000000000","message":"Patch Set 32:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial: https://bugs.launchpad.net/bugs/1686542\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":32},{"id":"3707adcc0a4d73c59c6ed6f6b094bf7c39f373c3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-02 21:03:50.000000000","message":"Patch Set 33: Patch Set 32 was rebased","accounts_in_message":[],"_revision_number":32},{"id":"3ac32ed9c28b5d0ef165e31062b5e5e80d607b77","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-02 21:03:55.000000000","message":"Patch Set 33: Workflow-1","accounts_in_message":[],"_revision_number":33},{"id":"ee4167ab5b8321d6fff462b1ea5024cd5b6a8dd4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-02 22:12:59.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/33/check/gate-octavia-docs-ubuntu-xenial/dd253b7//doc/build/html/ : SUCCESS in 5m 43s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/33/check/gate-octavia-pep8-ubuntu-xenial/a0665bf/ : SUCCESS in 3m 23s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/33/check/gate-octavia-python27-ubuntu-xenial/0b67c52/ : SUCCESS in 2m 43s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/33/check/gate-octavia-python35/48d940f/ : SUCCESS in 2m 51s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/33/check/octavia-coverage-ubuntu-xenial/e5c6a0c/ : SUCCESS in 6m 55s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/33/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/c58ae6b/ : SUCCESS in 45m 06s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/33/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/912660d/ : SUCCESS in 43m 00s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/33/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/d0b540b/ : SUCCESS in 35m 13s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/33/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/963f096/ : SUCCESS in 38m 06s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/33/check/gate-octavia-tox-functional-ubuntu-xenial/4d1eeed/ : SUCCESS in 6m 45s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/33/check/gate-octavia-tox-functional-py35-ubuntu-xenial/78fd46b/ : SUCCESS in 7m 27s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/33/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/bb21985/ : SUCCESS in 57m 18s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/33/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/673083f/ : SUCCESS in 27m 33s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/33/check/gate-octavia-v1-dsvm-scenario-multinode/1ded17f/ : SUCCESS in 1h 03m 26s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/33/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/5941844/ : SUCCESS in 57m 54s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/33/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/215272a/ : SUCCESS in 31m 14s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/33/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/b2aa9d7/ : FAILURE in 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"4e2f57adf60755f94c6f4eeba72ca4a1ac359e2f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-08 00:56:53.000000000","message":"Patch Set 34: Patch Set 33 was rebased","accounts_in_message":[],"_revision_number":33},{"id":"2c64a062e1ed6251d88dc8d00631732bd6e635f8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-08 00:56:56.000000000","message":"Patch Set 34: Workflow-1","accounts_in_message":[],"_revision_number":34},{"id":"42efee9487e1aadd6fbb7b30b1d817326d7fb23d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-08 02:28:54.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/34/check/gate-octavia-docs-ubuntu-xenial/c2ed84d//doc/build/html/ : SUCCESS in 6m 05s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/34/check/gate-octavia-pep8-ubuntu-xenial/a4199d0/ : SUCCESS in 5m 05s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/34/check/gate-octavia-python27-ubuntu-xenial/7266e2a/ : SUCCESS in 3m 46s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/34/check/gate-octavia-python35/0bd5159/ : SUCCESS in 3m 43s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/34/check/octavia-coverage-ubuntu-xenial/1e92afd/ : SUCCESS in 9m 27s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/34/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/de89a97/ : SUCCESS in 58m 14s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/34/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/5c0fcb7/ : SUCCESS in 53m 55s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/34/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/c04dd0b/ : SUCCESS in 1h 08m 58s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/34/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/2ea3a4e/ : SUCCESS in 1h 31m 44s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/34/check/gate-octavia-tox-functional-ubuntu-xenial/79369c5/ : SUCCESS in 5m 53s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/34/check/gate-octavia-tox-functional-py35-ubuntu-xenial/ff55460/ : SUCCESS in 6m 26s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/34/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/83766af/ : SUCCESS in 1h 19m 25s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/34/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/2d8aecb/ : SUCCESS in 42m 05s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/34/check/gate-octavia-v1-dsvm-scenario-multinode/1ca3069/ : SUCCESS in 1h 24m 51s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/34/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/b68d7cd/ : SUCCESS in 1h 01m 42s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/34/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/f50c535/ : SUCCESS in 48m 38s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/34/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/1e75e89/ : FAILURE in 33m 39s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"cb6b9b8f34920d5265fc4f0ddabda3d403246e8b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-16 20:17:58.000000000","message":"Patch Set 35: Patch Set 34 was rebased","accounts_in_message":[],"_revision_number":34},{"id":"c47337ff2c18cb92d983d16ebd3110de0a1f93b1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-16 20:18:02.000000000","message":"Patch Set 35: Workflow-1","accounts_in_message":[],"_revision_number":35},{"id":"be92376f5ecfb8d535431d2b8b0ba93c0a9b17fc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-16 21:37:23.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/35/check/gate-octavia-docs-ubuntu-xenial/a6f03cf//doc/build/html/ : SUCCESS in 6m 52s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/35/check/gate-octavia-pep8-ubuntu-xenial/5337c65/ : SUCCESS in 5m 14s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/35/check/gate-octavia-python27-ubuntu-xenial/98c8af2/ : SUCCESS in 4m 34s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/35/check/gate-octavia-python35/24dffa7/ : SUCCESS in 3m 50s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/35/check/octavia-coverage-ubuntu-xenial/f23dfa9/ : SUCCESS in 10m 37s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/35/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/ac07bb6/ : SUCCESS in 58m 46s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/35/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/38f2618/ : SUCCESS in 51m 10s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/35/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/b9f5046/ : SUCCESS in 1h 16m 12s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/35/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/793c791/ : SUCCESS in 1h 17m 07s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/35/check/gate-octavia-tox-functional-ubuntu-xenial/4c705f4/ : SUCCESS in 7m 27s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/35/check/gate-octavia-tox-functional-py35-ubuntu-xenial/4e5c057/ : SUCCESS in 7m 02s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/35/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/f229531/ : SUCCESS in 44m 57s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/35/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/f4e1474/ : SUCCESS in 40m 36s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/35/check/gate-octavia-v1-dsvm-scenario-multinode/16cfb9d/ : SUCCESS in 59m 32s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/35/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/4904012/ : SUCCESS in 1h 01m 01s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/35/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/3b9f511/ : SUCCESS in 45m 58s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/35/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/c35b03d/ : FAILURE in 36m 20s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"ec6fcc9efdfa04607d8901f76990ba8ce644ae3d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-17 23:39:45.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"30b13c7d8274ffdaddc714d9c498406f912e894c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-17 23:39:57.000000000","message":"Patch Set 36: Workflow-1","accounts_in_message":[],"_revision_number":36},{"id":"27a2425389f4f15e4b5aac4c8dc3d74809f52ab8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-18 01:11:46.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/36/check/gate-octavia-docs-ubuntu-xenial/5d6eb97//doc/build/html/ : SUCCESS in 5m 37s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/36/check/gate-octavia-pep8-ubuntu-xenial/ae4d639/ : FAILURE in 5m 06s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/36/check/gate-octavia-python27-ubuntu-xenial/7240d36/ : FAILURE in 4m 17s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/36/check/gate-octavia-python35/daac9a4/ : FAILURE in 4m 16s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/36/check/octavia-coverage-ubuntu-xenial/ce93a92/ : FAILURE in 3m 55s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/36/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/87ce48e/ : SUCCESS in 1h 07m 11s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/36/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/a23e24b/ : SUCCESS in 1h 01m 13s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/36/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/7d2b847/ : SUCCESS in 1h 24m 11s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/36/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/a4d3f54/ : SUCCESS in 1h 30m 40s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/36/check/gate-octavia-tox-functional-ubuntu-xenial/cad7fa7/ : SUCCESS in 7m 28s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/36/check/gate-octavia-tox-functional-py35-ubuntu-xenial/d3b6ac3/ : SUCCESS in 8m 05s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/36/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/442260e/ : SUCCESS in 46m 26s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/36/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/f8db4f6/ : SUCCESS in 51m 50s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/36/check/gate-octavia-v1-dsvm-scenario-multinode/444f3c1/ : SUCCESS in 57m 34s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/36/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/9d34e87/ : SUCCESS in 58m 00s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/36/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/2e6f334/ : SUCCESS in 57m 29s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/36/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/3b1ed14/ : FAILURE in 47m 25s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"4062a42117795d06f179eaf9ee90229a17f258fe","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-18 01:38:59.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"ac4a718c9b316be467c645a0c75aedd237abdcee","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-18 01:39:11.000000000","message":"Patch Set 37: Workflow-1","accounts_in_message":[],"_revision_number":37},{"id":"39c800288230ea0a60423ab4de76bb1f995811e7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-18 03:11:12.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/37/check/gate-octavia-docs-ubuntu-xenial/f9a2320//doc/build/html/ : SUCCESS in 5m 15s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/37/check/gate-octavia-pep8-ubuntu-xenial/31df686/ : SUCCESS in 6m 07s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/37/check/gate-octavia-python27-ubuntu-xenial/c8a4255/ : SUCCESS in 5m 43s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/37/check/gate-octavia-python35/9b6ef31/ : SUCCESS in 4m 19s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/37/check/octavia-coverage-ubuntu-xenial/9d06952/ : SUCCESS in 9m 34s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/37/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/58b5e6e/ : SUCCESS in 1h 01m 06s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/37/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/943a9da/ : SUCCESS in 59m 47s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/37/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/effd70a/ : SUCCESS in 1h 29m 00s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/37/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/2af608f/ : SUCCESS in 1h 24m 12s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/37/check/gate-octavia-tox-functional-ubuntu-xenial/e3367cf/ : SUCCESS in 6m 12s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/37/check/gate-octavia-tox-functional-py35-ubuntu-xenial/681d82e/ : SUCCESS in 6m 59s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/37/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/2e7da55/ : SUCCESS in 50m 08s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/37/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/06ae3f2/ : SUCCESS in 49m 41s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/37/check/gate-octavia-v1-dsvm-scenario-multinode/ade5116/ : SUCCESS in 57m 00s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/37/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/6652b04/ : SUCCESS in 55m 44s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/37/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/9b3de94/ : SUCCESS in 52m 58s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/37/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/23c58ec/ : FAILURE in 37m 23s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"e1a779b3baacd366f091eba7c1bd431d81c17e6a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-19 01:52:14.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"34e7ef43ab58c5c30ddba2480f7103e23245b115","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-19 01:55:22.000000000","message":"Patch Set 38: Workflow-1","accounts_in_message":[],"_revision_number":38},{"id":"c8ec8d40ad03dc7ccb941a0e58e47d23ebc60a80","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-19 03:22:37.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://logs.openstack.org/12/435612/38/check/gate-octavia-docs-ubuntu-xenial/4d15928/ : FAILURE in 4m 50s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/38/check/gate-octavia-pep8-ubuntu-xenial/bd96578/ : SUCCESS in 4m 59s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/38/check/gate-octavia-python27-ubuntu-xenial/5733f92/ : SUCCESS in 5m 57s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/38/check/gate-octavia-python35/c870596/ : SUCCESS in 3m 57s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/38/check/octavia-coverage-ubuntu-xenial/6e156b1/ : SUCCESS in 8m 19s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/38/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/a6fd4d2/ : SUCCESS in 1h 04m 16s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/38/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/be16370/ : SUCCESS in 56m 59s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/38/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/9f71e88/ : SUCCESS in 1h 30m 08s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/38/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/3b403ee/ : SUCCESS in 1h 19m 44s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/38/check/gate-octavia-tox-functional-ubuntu-xenial/09eccb3/ : SUCCESS in 8m 33s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/38/check/gate-octavia-tox-functional-py35-ubuntu-xenial/191ca53/ : SUCCESS in 7m 37s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/38/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/ac0e69f/ : SUCCESS in 45m 04s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/38/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/6693a70/ : SUCCESS in 45m 25s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/38/check/gate-octavia-v1-dsvm-scenario-multinode/3db3573/ : SUCCESS in 1h 09m 38s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/38/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/27073df/ : SUCCESS in 53m 58s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/38/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/69b3533/ : SUCCESS in 51m 45s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/38/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/7e55989/ : FAILURE in 41m 26s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"13e200b6233b6126a383c2aa8829de97c7526127","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-19 03:34:37.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"aad6bf75d67aaa85d18ecbd0354fde7fb3cc80bb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-19 05:05:10.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://logs.openstack.org/12/435612/39/check/gate-octavia-docs-ubuntu-xenial/11493da/ : FAILURE in 4m 12s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/39/check/gate-octavia-pep8-ubuntu-xenial/13bb6b6/ : SUCCESS in 5m 09s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/39/check/gate-octavia-python27-ubuntu-xenial/e8b3fe7/ : SUCCESS in 4m 15s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/39/check/gate-octavia-python35/f333856/ : SUCCESS in 4m 00s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/39/check/octavia-coverage-ubuntu-xenial/39ccfe6/ : SUCCESS in 7m 41s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/39/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/56401b4/ : SUCCESS in 58m 42s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/39/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/47e0dad/ : SUCCESS in 59m 07s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/39/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/93fb3ae/ : SUCCESS in 1h 28m 45s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/39/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/7ac7e22/ : SUCCESS in 1h 23m 07s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/39/check/gate-octavia-tox-functional-ubuntu-xenial/5fe3e29/ : SUCCESS in 6m 46s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/39/check/gate-octavia-tox-functional-py35-ubuntu-xenial/39d1402/ : SUCCESS in 7m 55s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/39/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/06a840b/ : SUCCESS in 50m 17s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/39/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/3b4ae87/ : SUCCESS in 54m 08s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/39/check/gate-octavia-v1-dsvm-scenario-multinode/632def2/ : SUCCESS in 58m 15s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/39/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/c5cbf9f/ : SUCCESS in 54m 48s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/39/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/f061ca9/ : FAILURE in 1h 05m 37s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/39/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/72f7ef3/ : FAILURE in 32m 49s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"242f66fb02567d14dcfeb4bdc8334274dbe80978","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-22 01:34:47.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"429237cf772e34eb019e7dde8ac8b5614769d187","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-22 01:36:57.000000000","message":"Patch Set 40: Workflow-1","accounts_in_message":[],"_revision_number":40},{"id":"79a516539e335a466d905c31a99c3c36af4886c9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-22 03:06:01.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/40/check/gate-octavia-docs-ubuntu-xenial/9cc89a5//doc/build/html/ : SUCCESS in 4m 44s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/40/check/gate-octavia-pep8-ubuntu-xenial/d943b9a/ : SUCCESS in 4m 59s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/40/check/gate-octavia-python27-ubuntu-xenial/6cf4c4b/ : SUCCESS in 4m 19s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/40/check/gate-octavia-python35/33e7e2c/ : SUCCESS in 4m 00s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/40/check/octavia-coverage-ubuntu-xenial/2352f74/ : SUCCESS in 8m 25s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/40/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/9f583da/ : SUCCESS in 1h 01m 47s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/40/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/445f459/ : SUCCESS in 59m 17s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/40/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/ab51f25/ : SUCCESS in 1h 17m 28s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/40/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/4b0c9f4/ : SUCCESS in 1h 29m 37s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/40/check/gate-octavia-tox-functional-ubuntu-xenial/42be486/ : SUCCESS in 6m 45s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/40/check/gate-octavia-tox-functional-py35-ubuntu-xenial/964fe9b/ : SUCCESS in 7m 11s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/40/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/d611266/ : SUCCESS in 43m 54s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/40/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/88f259a/ : SUCCESS in 42m 00s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/40/check/gate-octavia-v1-dsvm-scenario-multinode/443a668/ : SUCCESS in 59m 17s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/40/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/3c4b736/ : SUCCESS in 55m 45s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/40/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/5973115/ : SUCCESS in 56m 32s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/40/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/093e46d/ : FAILURE in 39m 10s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"23bc656418b6127af7d0671fb058735442ddf8a6","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2017-08-22 04:08:40.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"2abd7a1f9412a7b0326ccdd250138440de616d79","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2017-08-22 04:32:20.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"18bd5293e2b3524a92ffed5c36670eba1d753343","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-22 06:03:07.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"42379a635b1194e1f269a7618b974c6de15d3c48","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-22 06:08:08.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.","accounts_in_message":[],"_revision_number":42},{"id":"7271dd0843e9b563c70971ea4c237873baa6ddd6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-22 06:08:13.000000000","message":"Patch Set 42: Workflow-1","accounts_in_message":[],"_revision_number":42},{"id":"314206783287662c4518b50193e5cc23cc78435a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-22 07:26:29.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/42/check/gate-octavia-docs-ubuntu-xenial/60db6af//doc/build/html/ : SUCCESS in 6m 13s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/42/check/gate-octavia-pep8-ubuntu-xenial/46bea3a/ : SUCCESS in 5m 21s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/42/check/gate-octavia-python27-ubuntu-xenial/ad0dff9/ : SUCCESS in 4m 47s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/42/check/gate-octavia-python35/c68ec4b/ : SUCCESS in 4m 17s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/42/check/octavia-coverage-ubuntu-xenial/d65fcb3/ : SUCCESS in 9m 32s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/42/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/077f4ca/ : SUCCESS in 1h 03m 31s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/42/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/760b7a5/ : SUCCESS in 1h 00m 31s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/42/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/1fd1573/ : SUCCESS in 1h 17m 33s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/42/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/a705f25/ : SUCCESS in 1h 16m 44s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/42/check/gate-octavia-tox-functional-ubuntu-xenial/bf411c8/ : SUCCESS in 7m 29s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/42/check/gate-octavia-tox-functional-py35-ubuntu-xenial/0c67461/ : SUCCESS in 7m 34s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/42/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/30f392b/ : SUCCESS in 44m 22s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/42/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/0a5dd75/ : SUCCESS in 43m 59s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/42/check/gate-octavia-v1-dsvm-scenario-multinode/36ab68f/ : SUCCESS in 1h 01m 13s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/42/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/012c7a5/ : SUCCESS in 52m 39s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/42/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/34695cf/ : SUCCESS in 45m 26s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/42/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/0afb508/ : FAILURE in 35m 59s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"b0600bffe61d19ea803a07549b76d001d308b5bf","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-22 21:31:07.000000000","message":"Patch Set 40:\n\n(2 comments)","accounts_in_message":[],"_revision_number":40},{"id":"7e177f0facc3dbfab77934ef5cb0933dedbb3858","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-22 22:21:30.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"5c291cd723f822218ad4d6e074d5625bc6af0213","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-22 22:21:37.000000000","message":"Patch Set 43: Workflow-1","accounts_in_message":[],"_revision_number":43},{"id":"806712963ca56ae8dfde16527c06b95d54aa9e9b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-22 23:43:32.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/43/check/gate-octavia-docs-ubuntu-xenial/1a18a72//doc/build/html/ : SUCCESS in 5m 59s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/43/check/gate-octavia-pep8-ubuntu-xenial/a29e94e/ : SUCCESS in 5m 22s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/43/check/gate-octavia-python27-ubuntu-xenial/0092e6b/ : FAILURE in 4m 48s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/43/check/gate-octavia-python35/0cbed84/ : FAILURE in 4m 47s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/43/check/octavia-coverage-ubuntu-xenial/87743c6/ : FAILURE in 8m 13s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/43/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/f65522d/ : SUCCESS in 1h 01m 02s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/43/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/811c02e/ : SUCCESS in 56m 21s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/43/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/47d7a73/ : SUCCESS in 1h 15m 36s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/43/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/0ceb40e/ : SUCCESS in 1h 14m 38s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/43/check/gate-octavia-tox-functional-ubuntu-xenial/b6c8384/ : SUCCESS in 6m 39s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/43/check/gate-octavia-tox-functional-py35-ubuntu-xenial/efffdda/ : SUCCESS in 7m 44s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/43/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/34cdfa9/ : SUCCESS in 43m 15s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/43/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/82cd60a/ : SUCCESS in 41m 24s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/43/check/gate-octavia-v1-dsvm-scenario-multinode/06ad6de/ : SUCCESS in 1h 02m 18s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/43/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/1703dcc/ : SUCCESS in 55m 49s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/43/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/6e5ba4b/ : SUCCESS in 46m 13s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/43/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/1adc298/ : FAILURE in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"31bdef0dc760c5aca0d64f44551dbb1c1563db8b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-23 00:55:25.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"ae79b66dc127d3ca22e2689e641c4d5dd710ac55","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2017-08-23 01:02:29.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"7a1e993977282beaa11ea7058fe031b80d48c027","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2017-08-23 01:03:21.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"4aae2bd9e733cc2049a8dc47e809bf7f3bbe7eeb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-23 01:18:40.000000000","message":"Patch Set 44: Workflow-1","accounts_in_message":[],"_revision_number":44},{"id":"4698934cdf453f339c7969e40c600f8d5586dbd3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-23 02:27:29.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/44/check/gate-octavia-docs-ubuntu-xenial/438e33e//doc/build/html/ : SUCCESS in 4m 58s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/44/check/gate-octavia-pep8-ubuntu-xenial/f5c893f/ : SUCCESS in 5m 12s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/44/check/gate-octavia-python27-ubuntu-xenial/8f400fe/ : SUCCESS in 4m 01s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/44/check/gate-octavia-python35/0526fff/ : SUCCESS in 3m 49s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/44/check/octavia-coverage-ubuntu-xenial/cd4f6a4/ : SUCCESS in 7m 47s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/44/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/34604a2/ : SUCCESS in 1h 02m 12s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/44/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/a0eaa64/ : SUCCESS in 1h 03m 31s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/44/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/5ae7267/ : SUCCESS in 1h 28m 46s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/44/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/73455ba/ : SUCCESS in 1h 10m 20s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/44/check/gate-octavia-tox-functional-ubuntu-xenial/d5ed6b9/ : SUCCESS in 6m 07s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/44/check/gate-octavia-tox-functional-py35-ubuntu-xenial/5473b29/ : SUCCESS in 6m 46s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/44/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/aa102d7/ : SUCCESS in 48m 44s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/44/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/60c0cd8/ : SUCCESS in 45m 55s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/44/check/gate-octavia-v1-dsvm-scenario-multinode/ca848f1/ : SUCCESS in 1h 00m 32s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/44/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/6158151/ : SUCCESS in 1h 01m 38s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/44/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/e3af1bc/ : SUCCESS in 51m 41s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/44/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/4cea657/ : FAILURE in 29m 41s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"70298e483eca88ca91879613ce1c1d1679548391","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-24 18:52:18.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"b6b441571c533a003a901c8ac434a21f6933670f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-24 23:47:03.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"9ba386b4cdea12333af017355a86df3919bf24fa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2017-08-25 00:53:12.000000000","message":"Uploaded patch set 46.","accounts_in_message":[],"_revision_number":46},{"id":"938f64907843804c513cfc397d3feb3c34c5b7a6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-25 02:35:58.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/46/check/gate-octavia-docs-ubuntu-xenial/7b27666//doc/build/html/ : SUCCESS in 4m 16s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/46/check/gate-octavia-pep8-ubuntu-xenial/c36868f/ : SUCCESS in 4m 46s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/46/check/gate-octavia-python27-ubuntu-xenial/f5d0768/ : SUCCESS in 4m 08s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/46/check/gate-octavia-python35/792303b/ : SUCCESS in 3m 40s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/46/check/octavia-coverage-ubuntu-xenial/7ca0e8b/ : SUCCESS in 8m 46s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/46/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/c97b067/ : SUCCESS in 1h 01m 01s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/46/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/84817fc/ : SUCCESS in 59m 47s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/46/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/13637c7/ : SUCCESS in 1h 42m 29s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/46/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/26683ff/ : SUCCESS in 1h 23m 08s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/46/check/gate-octavia-tox-functional-ubuntu-xenial/667626c/ : SUCCESS in 7m 34s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/46/check/gate-octavia-tox-functional-py35-ubuntu-xenial/2ea1405/ : SUCCESS in 7m 15s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/46/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/1984e02/ : SUCCESS in 49m 21s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/46/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/8b8779e/ : SUCCESS in 42m 35s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/46/check/gate-octavia-v1-dsvm-scenario-multinode/0f75532/ : SUCCESS in 56m 39s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/46/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/072f316/ : SUCCESS in 56m 24s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/46/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/e7b98fe/ : SUCCESS in 49m 24s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/46/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/1be4c1d/ : FAILURE in 50m 53s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"7795a5571e9e594db422dd4548fdc3141a4982b5","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2017-08-25 14:39:05.000000000","message":"Patch Set 46:\n\nI had the discussion with johnsom earlier in the ACTIVE-ACTIVE context that (1) our flows are riddled with if...else.. -- we need to rethink/refactor that since the more special cases we have to support the more difficult it gets to maintain those. (2) The agent should have a way to configure it for different behavior, e.g. some configuration saying we are running in FIP mode. I think we have a similar pattern for different OSes.","accounts_in_message":[],"_revision_number":46},{"id":"69f2db8454e71627bfce2d67c1341e75088a799c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-29 23:03:45.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"4c24bf50b95d47def4c506437c2aee5717531693","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-29 23:18:03.000000000","message":"Uploaded patch set 48.","accounts_in_message":[],"_revision_number":48},{"id":"b54bc9a6dfe1244ebc9c3d4b972025bc6e91aaae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2017-08-30 00:02:12.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"4327cff78c2d34f98593c8fbeca3d81c6e645e7e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 01:19:15.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/49/check/gate-octavia-docs-ubuntu-xenial/88f74b3//doc/build/html/ : SUCCESS in 5m 26s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/49/check/gate-octavia-pep8-ubuntu-xenial/e9e8c48/ : SUCCESS in 6m 30s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/49/check/gate-octavia-python27-ubuntu-xenial/543733b/ : SUCCESS in 3m 41s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/49/check/gate-octavia-python35/8134d64/ : SUCCESS in 3m 39s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/49/check/octavia-coverage-ubuntu-xenial/ac21245/ : SUCCESS in 8m 21s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/49/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/483cfd8/ : SUCCESS in 59m 25s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/49/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/19fd059/ : SUCCESS in 58m 08s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/49/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/27731fe/ : SUCCESS in 1h 11m 13s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/49/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/ce69acc/ : SUCCESS in 1h 16m 48s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/49/check/gate-octavia-tox-functional-ubuntu-xenial/8d63fde/ : SUCCESS in 8m 39s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/49/check/gate-octavia-tox-functional-py35-ubuntu-xenial/7418ad6/ : SUCCESS in 7m 40s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/49/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/0cd2bb6/ : SUCCESS in 49m 05s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/49/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/7f1d01f/ : SUCCESS in 46m 22s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/49/check/gate-octavia-v1-dsvm-scenario-multinode/a563fd2/ : SUCCESS in 1h 02m 34s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/49/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/bf789d4/ : SUCCESS in 59m 23s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/49/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/f803e37/ : SUCCESS in 53m 51s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/49/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/70a906f/ : FAILURE in 42m 48s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"c87a8bbae9b8f3e31cb775ab58b4c1592062dc9f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-08-31 19:21:20.000000000","message":"Patch Set 49: Workflow-1","accounts_in_message":[],"_revision_number":49},{"id":"5bda422bc5dae03d86736d6006a46fe63526d87a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-08 22:35:40.000000000","message":"Uploaded patch set 50.","accounts_in_message":[],"_revision_number":50},{"id":"22f47f57e09bc68ee82fcaa42b21a77b8c877446","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-08 22:36:22.000000000","message":"Patch Set 51: Patch Set 50 was rebased","accounts_in_message":[],"_revision_number":50},{"id":"f39d4795c925237111934e8c3f17db376080ef10","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-08 22:36:52.000000000","message":"Patch Set 51: Workflow-1","accounts_in_message":[],"_revision_number":51},{"id":"ce15e41add9740e97d5c216b0c6204cf9fcf3364","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-09 01:46:37.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/51/check/gate-octavia-docs-ubuntu-xenial/88bbf58//doc/build/html/ : SUCCESS in 5m 29s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/51/check/gate-octavia-pep8-ubuntu-xenial/c82905a/ : FAILURE in 5m 29s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/51/check/gate-octavia-python27-ubuntu-xenial/c3f663a/ : SUCCESS in 4m 15s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/51/check/gate-octavia-python35/d3c8c45/ : SUCCESS in 4m 09s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/51/check/octavia-coverage-ubuntu-xenial/bfe2742/ : SUCCESS in 8m 30s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/51/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/5ef5f3b/ : SUCCESS in 56m 57s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/51/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/e4c78e0/ : SUCCESS in 56m 01s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/51/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/c8fbeae/ : SUCCESS in 1h 20m 33s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/51/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/69e2a9c/ : SUCCESS in 1h 22m 42s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/51/check/gate-octavia-tox-functional-ubuntu-xenial/f6f0b5b/ : SUCCESS in 6m 55s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/51/check/gate-octavia-tox-functional-py35-ubuntu-xenial/07c746a/ : SUCCESS in 6m 45s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/51/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/abb24b9/ : SUCCESS in 46m 17s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/51/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/1158a4d/ : SUCCESS in 50m 35s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/51/check/gate-octavia-v1-dsvm-scenario-multinode/6b2b9e8/ : SUCCESS in 1h 00m 16s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/51/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/722bd6c/ : SUCCESS in 57m 09s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/51/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/e9005aa/ : SUCCESS in 46m 16s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/51/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/f948cb0/ : FAILURE in 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"308c5153857c92ffa9562c9c0be8ab1bfc126aa4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-14 19:58:40.000000000","message":"Uploaded patch set 52.","accounts_in_message":[],"_revision_number":52},{"id":"6053be1bd0c34eca56712ba9c05ac5423ee2a34f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-14 19:58:52.000000000","message":"Patch Set 52: Workflow-1","accounts_in_message":[],"_revision_number":52},{"id":"7b3eb9384ca8160a364044e071e8fd0eb6cc8579","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-14 21:06:17.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/52/check/gate-octavia-docs-ubuntu-xenial/dc32e12//doc/build/html/ : SUCCESS in 6m 05s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/52/check/gate-octavia-pep8-ubuntu-xenial/06a30c6/ : FAILURE in 6m 41s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/52/check/gate-octavia-python27-ubuntu-xenial/5978b6f/ : SUCCESS in 5m 07s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/52/check/gate-octavia-python35/116725b/ : SUCCESS in 6m 04s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/52/check/octavia-coverage-ubuntu-xenial/2bb1b31/ : SUCCESS in 13m 31s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/52/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/08a76fb/ : SUCCESS in 1h 01m 36s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/52/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/df5c83a/ : SUCCESS in 59m 49s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/52/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/9724c01/ : SUCCESS in 53m 18s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/52/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/e135996/ : SUCCESS in 54m 19s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/52/check/gate-octavia-tox-functional-ubuntu-xenial/bf9300c/ : SUCCESS in 11m 04s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/52/check/gate-octavia-tox-functional-py35-ubuntu-xenial/1c6f430/ : SUCCESS in 8m 08s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/52/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/ee49285/ : SUCCESS in 52m 40s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/52/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/f3b353f/ : SUCCESS in 43m 49s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/52/check/gate-octavia-v1-dsvm-scenario-multinode/319ffe1/ : SUCCESS in 1h 01m 39s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/52/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/9e46adb/ : SUCCESS in 58m 02s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/52/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/78f3ced/ : SUCCESS in 46m 09s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/52/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/21d91f2/ : FAILURE in 42m 22s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"34f7df50d1200aa07542ef6a76dceefc7310faa9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-18 23:07:54.000000000","message":"Uploaded patch set 53.","accounts_in_message":[],"_revision_number":53},{"id":"6cbfa49b47e6c312e0d0999c036a0575fc5bf18a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-18 23:09:44.000000000","message":"Patch Set 53: Workflow-1","accounts_in_message":[],"_revision_number":53},{"id":"f8ad5fc73664c27c77ec225ee4cbd67178fcad6b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-19 00:23:27.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/53/check/gate-octavia-docs-ubuntu-xenial/82041a4//doc/build/html/ : SUCCESS in 6m 38s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/53/check/gate-octavia-pep8-ubuntu-xenial/c720e29/ : SUCCESS in 5m 50s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/53/check/gate-octavia-python27-ubuntu-xenial/9fab89f/ : SUCCESS in 4m 52s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/53/check/gate-octavia-python35/ea51302/ : SUCCESS in 4m 10s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/53/check/octavia-coverage-ubuntu-xenial/1666449/ : SUCCESS in 9m 02s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/53/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/9ca311e/ : SUCCESS in 1h 09m 53s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/53/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/46f5f30/ : SUCCESS in 1h 06m 04s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/53/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/4278075/ : SUCCESS in 1h 11m 37s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/53/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/c30feaa/ : SUCCESS in 1h 11m 12s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/53/check/gate-octavia-tox-functional-ubuntu-xenial/3fb24a0/ : SUCCESS in 6m 40s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/53/check/gate-octavia-tox-functional-py35-ubuntu-xenial/b841873/ : SUCCESS in 6m 34s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/53/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/a70fcc4/ : SUCCESS in 54m 06s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/53/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/05a3b85/ : SUCCESS in 38m 42s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/53/check/gate-octavia-v1-dsvm-scenario-multinode/1bca24b/ : SUCCESS in 1h 02m 10s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/53/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/750a861/ : SUCCESS in 1h 09m 48s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/53/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/75885fd/ : SUCCESS in 45m 36s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/53/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/46abf59/ : FAILURE in 41m 44s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"05f965a6d0d149f3c31426fe663a4930a5f78f84","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-21 09:39:37.000000000","message":"Uploaded patch set 54.","accounts_in_message":[],"_revision_number":54},{"id":"4ddb96d4c3017f924997a19c2f2ddb00923c7930","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-21 09:40:16.000000000","message":"Patch Set 54: Workflow-1","accounts_in_message":[],"_revision_number":54},{"id":"9a87159357b3a13734f9e11fe18416cf5f4dc70b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-21 11:10:35.000000000","message":"Patch Set 54: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/54/check/gate-octavia-docs-ubuntu-xenial/686be4d//doc/build/html/ : SUCCESS in 5m 38s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/54/check/gate-octavia-pep8-ubuntu-xenial/04ca9de/ : SUCCESS in 6m 01s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/54/check/gate-octavia-python27-ubuntu-xenial/39b5512/ : SUCCESS in 4m 08s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/54/check/gate-octavia-python35/9550e15/ : SUCCESS in 4m 43s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/54/check/octavia-coverage-ubuntu-xenial/aa01908/ : SUCCESS in 9m 18s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/54/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/d3381cb/ : SUCCESS in 57m 33s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/54/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/6bae6bd/ : SUCCESS in 54m 22s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/54/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/f8080af/ : SUCCESS in 1h 24m 48s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/54/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/bc461e7/ : SUCCESS in 1h 29m 34s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/54/check/gate-octavia-tox-functional-ubuntu-xenial/52721df/ : SUCCESS in 7m 35s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/54/check/gate-octavia-tox-functional-py35-ubuntu-xenial/9d9774d/ : SUCCESS in 10m 14s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/54/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/6fd0c9b/ : FAILURE in 28m 38s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/54/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/de80b9b/ : SUCCESS in 40m 32s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/54/check/gate-octavia-v1-dsvm-scenario-multinode/3ed3b8e/ : FAILURE in 34m 23s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/54/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/64fecb1/ : FAILURE in 35m 02s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/54/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/933f906/ : FAILURE in 34m 31s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/54/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/b518123/ : FAILURE in 35m 23s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"38eaf21feae1aa0aac42aaa6c0e58dfdcacb22a6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-22 22:09:07.000000000","message":"Uploaded patch set 55.","accounts_in_message":[],"_revision_number":55},{"id":"36af6923f0a7428ad1887b0bc1180dd41c4d014b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-22 23:28:52.000000000","message":"Patch Set 55: Workflow-1","accounts_in_message":[],"_revision_number":55},{"id":"6ffcfc689b6d4cc199aa9205c0b846d9a14c5d86","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 23:31:35.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/55/check/gate-octavia-docs-ubuntu-xenial/1d496c1//doc/build/html/ : SUCCESS in 5m 14s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/55/check/gate-octavia-pep8-ubuntu-xenial/bddcec9/ : SUCCESS in 5m 50s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/55/check/gate-octavia-python27-ubuntu-xenial/9839a60/ : SUCCESS in 6m 33s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/55/check/gate-octavia-python35/0af724a/ : SUCCESS in 4m 08s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/55/check/octavia-coverage-ubuntu-xenial/12a6f2e/ : SUCCESS in 12m 11s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/55/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/baec7ed/ : SUCCESS in 58m 03s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/55/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/0b3073e/ : SUCCESS in 56m 34s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/55/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/2fee081/ : SUCCESS in 1h 14m 13s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/55/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/768fd81/ : FAILURE in 1h 21m 04s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/55/check/gate-octavia-tox-functional-ubuntu-xenial/4ee9de7/ : SUCCESS in 11m 51s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/55/check/gate-octavia-tox-functional-py35-ubuntu-xenial/acfaa61/ : SUCCESS in 6m 49s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/55/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/b81d28b/ : SUCCESS in 50m 49s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/55/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/08c860f/ : SUCCESS in 37m 45s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/55/check/gate-octavia-v1-dsvm-scenario-multinode/6b18929/ : SUCCESS in 57m 51s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/55/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/ceba918/ : SUCCESS in 1h 01m 46s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/55/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/cae4915/ : SUCCESS in 53m 32s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/55/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/5aa903b/ : FAILURE in 35m 48s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"b017b13b6de50ffc7c67e5c001b7bd8f71c53bc7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-25 19:00:49.000000000","message":"Uploaded patch set 56.","accounts_in_message":[],"_revision_number":56},{"id":"5f583b8dbc65f389334737c18ca10325631ce842","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-25 20:25:57.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/56/check/gate-octavia-docs-ubuntu-xenial/ba11a31//doc/build/html/ : SUCCESS in 6m 14s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-octavia-pep8-ubuntu-xenial/b23a0db/ : SUCCESS in 10m 41s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-octavia-python27-ubuntu-xenial/78caf9d/ : SUCCESS in 8m 31s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/56/check/gate-octavia-python35/cef185d/ : SUCCESS in 7m 09s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/octavia-coverage-ubuntu-xenial/58a4f9c/ : SUCCESS in 9m 05s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/dabfcef/ : FAILURE in 34m 39s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/56/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/828489e/ : FAILURE in 42m 44s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/56/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/8247a0a/ : FAILURE in 59m 35s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/56/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/57380b6/ : FAILURE in 43m 44s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-octavia-tox-functional-ubuntu-xenial/e1e680f/ : SUCCESS in 7m 10s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-octavia-tox-functional-py35-ubuntu-xenial/15fab4c/ : SUCCESS in 6m 51s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/541b8c7/ : SUCCESS in 56m 47s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/daa5f04/ : SUCCESS in 54m 32s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/56/check/gate-octavia-v1-dsvm-scenario-multinode/80ea82b/ : SUCCESS in 1h 04m 12s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/56/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/cd8cb4f/ : SUCCESS in 1h 08m 02s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/56/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/f28c764/ : SUCCESS in 53m 38s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/56/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/99a8d7a/ : FAILURE in 35m 40s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"35e6b415a1770618eb109bb7ac19da56089ac6fe","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-26 04:58:56.000000000","message":"Patch Set 56:\n\nrecheck","accounts_in_message":[],"_revision_number":56},{"id":"53d325302f072a4c26ddda8e71c7c1a4f0a4cdef","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-26 04:59:03.000000000","message":"Patch Set 56: Workflow-1","accounts_in_message":[],"_revision_number":56},{"id":"f7e648e0214e247a18fbf24344ed4c6754001ab0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-26 06:18:42.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/56/check/gate-octavia-docs-ubuntu-xenial/f02674a//doc/build/html/ : SUCCESS in 5m 22s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-octavia-pep8-ubuntu-xenial/8bbdd48/ : SUCCESS in 5m 05s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-octavia-python27-ubuntu-xenial/4cccd3b/ : SUCCESS in 4m 12s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/56/check/gate-octavia-python35/93d1a8d/ : SUCCESS in 3m 45s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/octavia-coverage-ubuntu-xenial/fd27eea/ : SUCCESS in 8m 34s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/ce66a4c/ : SUCCESS in 59m 07s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/56/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/1ba23c0/ : SUCCESS in 57m 17s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/56/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/25ae236/ : FAILURE in 1h 18m 02s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/56/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/e9b323b/ : SUCCESS in 1h 16m 00s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-octavia-tox-functional-ubuntu-xenial/8fc6db7/ : SUCCESS in 6m 03s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-octavia-tox-functional-py35-ubuntu-xenial/9570c88/ : SUCCESS in 10m 23s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/56e3815/ : SUCCESS in 43m 37s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/56/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/1edfd2d/ : SUCCESS in 42m 33s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/56/check/gate-octavia-v1-dsvm-scenario-multinode/ea5a268/ : SUCCESS in 1h 02m 00s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/56/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/1eb0001/ : SUCCESS in 1h 03m 34s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/56/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/7560e1a/ : SUCCESS in 45m 57s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/56/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/a9ebd4a/ : FAILURE in 39m 04s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"a1667236cf959e8b4b0ccf1632c3eb0664127c3f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-27 20:03:38.000000000","message":"Uploaded patch set 57.","accounts_in_message":[],"_revision_number":57},{"id":"9828877a6c1ec917ae1b4f1b58b62508990f64f4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-27 20:45:37.000000000","message":"Uploaded patch set 58.","accounts_in_message":[],"_revision_number":58},{"id":"be02658f5a9d7a71a30edbda678101dfa224e08d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-27 21:58:54.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-requirements http://logs.openstack.org/12/435612/58/check/gate-octavia-requirements/f0c2850/ : SUCCESS in 3m 04s\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/58/check/gate-octavia-docs-ubuntu-xenial/15ac76d//doc/build/html/ : SUCCESS in 6m 43s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/58/check/gate-octavia-pep8-ubuntu-xenial/6d36285/ : FAILURE in 7m 25s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/58/check/gate-octavia-python27-ubuntu-xenial/9dae0a0/ : SUCCESS in 4m 25s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/58/check/gate-octavia-python35/06c4215/ : SUCCESS in 3m 58s\n- gate-octavia-releasenotes http://docs-draft.openstack.org/12/435612/58/check/gate-octavia-releasenotes/c48b28b//releasenotes/build/html/ : SUCCESS in 4m 21s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/58/check/octavia-coverage-ubuntu-xenial/e0abb65/ : SUCCESS in 12m 31s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/58/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/59ce5c6/ : SUCCESS in 59m 32s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/58/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/cb27303/ : SUCCESS in 1h 04m 08s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/58/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/b1f8e85/ : SUCCESS in 48m 41s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/58/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/9e8a5a1/ : FAILURE in 1h 11m 19s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/58/check/gate-octavia-tox-functional-ubuntu-xenial/caa4583/ : SUCCESS in 9m 42s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/58/check/gate-octavia-tox-functional-py35-ubuntu-xenial/ceecd5f/ : SUCCESS in 7m 30s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/58/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/d524fc9/ : SUCCESS in 46m 46s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/58/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/cd57ee3/ : SUCCESS in 46m 37s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/58/check/gate-octavia-v1-dsvm-scenario-multinode/1498a0d/ : SUCCESS in 1h 07m 28s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/58/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/790ea8b/ : SUCCESS in 51m 05s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/58/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/91fd3e5/ : SUCCESS in 42m 54s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/58/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/0f482ac/ : FAILURE in 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"c5b8b1bbabb0037f95fb875a3ecc3f585d9fff6b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-27 23:11:04.000000000","message":"Uploaded patch set 59.","accounts_in_message":[],"_revision_number":59},{"id":"7544d6f1fe4ca823da1cecfd3f43b111c7d01fec","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-27 23:57:02.000000000","message":"Uploaded patch set 60.","accounts_in_message":[],"_revision_number":60},{"id":"aaf32bc1cb11a9075f285859863b26f65582cfea","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-28 00:10:28.000000000","message":"Uploaded patch set 61.","accounts_in_message":[],"_revision_number":61},{"id":"1255b98c5f12696ba91a26e15a76f4ee291fa9f4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-09-28 00:30:32.000000000","message":"Uploaded patch set 62.","accounts_in_message":[],"_revision_number":62},{"id":"5fe37f1371a7fd70f96f73dca05aabd9403c52c3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-28 01:50:41.000000000","message":"Patch Set 62: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-requirements http://logs.openstack.org/12/435612/62/check/gate-octavia-requirements/6269ed8/ : SUCCESS in 4m 05s\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/62/check/gate-octavia-docs-ubuntu-xenial/3befecc//doc/build/html/ : SUCCESS in 6m 39s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/62/check/gate-octavia-pep8-ubuntu-xenial/54f1a43/ : SUCCESS in 6m 05s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/62/check/gate-octavia-python27-ubuntu-xenial/470f3d9/ : SUCCESS in 5m 58s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/62/check/gate-octavia-python35/843d5dc/ : SUCCESS in 5m 59s\n- gate-octavia-releasenotes http://docs-draft.openstack.org/12/435612/62/check/gate-octavia-releasenotes/b5f7511//releasenotes/build/html/ : SUCCESS in 5m 31s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/62/check/octavia-coverage-ubuntu-xenial/98906c2/ : SUCCESS in 8m 29s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/62/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/b25f52a/ : SUCCESS in 57m 29s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/62/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/6ab6b92/ : SUCCESS in 54m 40s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/62/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/170bff8/ : SUCCESS in 1h 12m 27s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/62/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/e15ac8f/ : SUCCESS in 1h 12m 22s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/62/check/gate-octavia-tox-functional-ubuntu-xenial/e49552c/ : SUCCESS in 6m 48s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/62/check/gate-octavia-tox-functional-py35-ubuntu-xenial/51abb0e/ : SUCCESS in 6m 56s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/62/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/eb9cc8a/ : SUCCESS in 44m 36s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/62/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/69d0471/ : SUCCESS in 40m 44s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/62/check/gate-octavia-v1-dsvm-scenario-multinode/14462ef/ : SUCCESS in 1h 12m 59s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/62/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/456d61d/ : SUCCESS in 1h 03m 37s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/62/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/41db748/ : SUCCESS in 43m 05s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/62/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/cf98380/ : FAILURE in 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"1b10c9ff340c13514031968218084acaae338925","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-07 04:05:38.000000000","message":"Patch Set 62: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":62},{"id":"62922578cfe255bc12ba9434371d725b9204c791","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-10-13 03:03:50.000000000","message":"Uploaded patch set 63.","accounts_in_message":[],"_revision_number":63},{"id":"979725e1bc604484f6044f80f4e5ec7c7d882842","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-10-13 03:03:56.000000000","message":"Patch Set 63: Workflow-1","accounts_in_message":[],"_revision_number":63},{"id":"9fbbb1d9544599ce0c3577e500290e3ac1a4e8bf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-13 04:47:15.000000000","message":"Patch Set 63: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-requirements http://logs.openstack.org/12/435612/63/check/gate-octavia-requirements/7dea3ea/ : SUCCESS in 2m 17s\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/63/check/gate-octavia-docs-ubuntu-xenial/eb298e6//doc/build/html/ : SUCCESS in 5m 24s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/63/check/gate-octavia-pep8-ubuntu-xenial/23945b7/ : FAILURE in 4m 49s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/63/check/gate-octavia-python27-ubuntu-xenial/30e6ea7/ : SUCCESS in 11m 45s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/63/check/gate-octavia-python35/0a80a4d/ : SUCCESS in 5m 00s\n- gate-octavia-releasenotes http://docs-draft.openstack.org/12/435612/63/check/gate-octavia-releasenotes/7260402//releasenotes/build/html/ : SUCCESS in 5m 29s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/63/check/octavia-coverage-ubuntu-xenial/aff7ed4/ : SUCCESS in 9m 26s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/63/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/477cfcb/ : SUCCESS in 1h 04m 38s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/63/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/5a15759/ : SUCCESS in 1h 05m 09s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/63/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/1496176/ : SUCCESS in 1h 13m 05s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/63/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/ed46474/ : FAILURE in 1h 40m 43s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/63/check/gate-octavia-tox-functional-ubuntu-xenial/bb4c2f1/ : SUCCESS in 10m 37s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/63/check/gate-octavia-tox-functional-py35-ubuntu-xenial/07662b4/ : SUCCESS in 11m 23s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/63/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/6276272/ : FAILURE in 49m 06s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/63/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/e55994e/ : SUCCESS in 49m 18s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/63/check/gate-octavia-v1-dsvm-scenario-multinode/421e1c8/ : SUCCESS in 56m 22s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/63/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/426251e/ : SUCCESS in 58m 49s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/63/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/06cb766/ : SUCCESS in 51m 34s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/63/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/7521b3f/ : FAILURE in 37m 05s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"00e4c8aaff9454a2239eac889310e7cf0b79c23a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-10-13 18:35:10.000000000","message":"Patch Set 64: Published edit on patch set 63.","accounts_in_message":[],"_revision_number":64},{"id":"7c81552f824c845a20d14cf5427a9cd5b50b31c1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-10-13 18:35:15.000000000","message":"Patch Set 64: Workflow-1","accounts_in_message":[],"_revision_number":64},{"id":"63aa27e56ac74e0cd9dc8e94ecee15d334e69c0c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-13 20:00:36.000000000","message":"Patch Set 64: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-requirements http://logs.openstack.org/12/435612/64/check/gate-octavia-requirements/88e4575/ : SUCCESS in 2m 02s\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/12/435612/64/check/gate-octavia-docs-ubuntu-xenial/d6cf75f//doc/build/html/ : SUCCESS in 5m 20s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/12/435612/64/check/gate-octavia-pep8-ubuntu-xenial/1444f43/ : SUCCESS in 5m 14s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/12/435612/64/check/gate-octavia-python27-ubuntu-xenial/dac0052/ : SUCCESS in 3m 45s\n- gate-octavia-python35 http://logs.openstack.org/12/435612/64/check/gate-octavia-python35/2e0f13a/ : SUCCESS in 4m 00s\n- gate-octavia-releasenotes http://docs-draft.openstack.org/12/435612/64/check/gate-octavia-releasenotes/0d3233c//releasenotes/build/html/ : SUCCESS in 7m 15s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/12/435612/64/check/octavia-coverage-ubuntu-xenial/142da05/ : SUCCESS in 7m 43s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/12/435612/64/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/2a3de08/ : SUCCESS in 54m 31s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/12/435612/64/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/6202824/ : SUCCESS in 1h 01m 57s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/64/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/f206e7f/ : SUCCESS in 1h 09m 52s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/12/435612/64/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/7db7b8f/ : FAILURE in 1h 24m 02s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/12/435612/64/check/gate-octavia-tox-functional-ubuntu-xenial/d451c81/ : SUCCESS in 8m 11s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/12/435612/64/check/gate-octavia-tox-functional-py35-ubuntu-xenial/aa793cb/ : SUCCESS in 6m 52s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/64/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/8ffccdc/ : SUCCESS in 39m 37s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/12/435612/64/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/5283166/ : SUCCESS in 39m 36s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/64/check/gate-octavia-v1-dsvm-scenario-multinode/c354a5a/ : SUCCESS in 1h 01m 09s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/64/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/f1fa0ba/ : SUCCESS in 1h 01m 53s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/12/435612/64/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/3801317/ : SUCCESS in 42m 35s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/12/435612/64/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/bf675c3/ : FAILURE in 43m 37s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"f4a097dc9b103307cca139804a75d13f81b64cd1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-10-16 17:52:05.000000000","message":"Uploaded patch set 65: Patch Set 64 was rebased.","accounts_in_message":[],"_revision_number":65},{"id":"2d8fb78bcde91dcc3cbf555d0499c16d8a3d7d87","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-10-16 17:52:12.000000000","message":"Patch Set 65: Workflow-1","accounts_in_message":[],"_revision_number":65},{"id":"94828f9f712a56ebc99fb1406052a4b0ef21b193","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-16 19:10:27.000000000","message":"Patch Set 65: Verified+1\n\nBuild succeeded.\n\n- requirements-check http://logs.openstack.org/12/435612/65/check/requirements-check/11e18e2/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/65/check/openstack-tox-pep8/bc870ff/ : SUCCESS in 6m 25s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/65/check/openstack-tox-py27/ca15ea8/ : SUCCESS in 5m 30s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/65/check/openstack-tox-py35/bd06279/ : SUCCESS in 4m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/65/check/build-openstack-sphinx-docs/aa973ea/html/ : SUCCESS in 8m 02s\n- build-openstack-releasenotes http://logs.openstack.org/12/435612/65/check/build-openstack-releasenotes/f7fced0/html/ : SUCCESS in 7m 46s\n- openstack-tox-cover http://logs.openstack.org/12/435612/65/check/openstack-tox-cover/418f826/ : SUCCESS in 11m 04s\n- legacy-neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/65/check/legacy-neutron-lbaasv2-dsvm-api/e176223/ : SUCCESS in 58m 12s\n- legacy-neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/65/check/legacy-neutron-lbaasv2-dsvm-py3x-api/addceea/ : SUCCESS in 1h 05m 00s (non-voting)\n- legacy-neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/65/check/legacy-neutron-lbaasv2-dsvm-scenario/5c53939/ : SUCCESS in 1h 02m 16s (non-voting)\n- legacy-neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/65/check/legacy-neutron-lbaasv2-dsvm-py3x-scenario/a02c49e/ : FAILURE in 1h 16m 11s (non-voting)\n- legacy-octavia-tox-functional http://logs.openstack.org/12/435612/65/check/legacy-octavia-tox-functional/e2fc8d9/ : SUCCESS in 8m 03s\n- legacy-octavia-tox-functional-py35 http://logs.openstack.org/12/435612/65/check/legacy-octavia-tox-functional-py35/7b11722/ : SUCCESS in 9m 10s\n- legacy-octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/65/check/legacy-octavia-v1-dsvm-scenario/7afc0c9/ : SUCCESS in 49m 47s (non-voting)\n- legacy-octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/65/check/legacy-octavia-v1-dsvm-py3x-scenario/b7af478/ : SUCCESS in 53m 41s (non-voting)\n- legacy-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/65/check/legacy-octavia-v1-dsvm-scenario-multinode/bc7d6d6/ : SUCCESS in 1h 01m 56s\n- legacy-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/65/check/legacy-octavia-v1-dsvm-py3x-scenario-multinode/a6a7cdc/ : SUCCESS in 1h 00m 51s\n- legacy-octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/65/check/legacy-octavia-v1-dsvm-scenario-kvm-centos.7/32da98b/ : SUCCESS in 51m 09s (non-voting)\n- legacy-octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/65/check/legacy-octavia-v1-dsvm-scenario-lxd/f2a6d3e/ : FAILURE in 46m 25s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"97b408e43d49570a8d76e657d185a5120f8573c4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-10-31 23:26:16.000000000","message":"Patch Set 66: Patch Set 65 was rebased","accounts_in_message":[],"_revision_number":66},{"id":"4245ef72d2f14b5d8ffc3b0cd3615a0a9fc2070c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-01 00:58:59.000000000","message":"Patch Set 66: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/66/check/requirements-check/170ba2c/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/66/check/openstack-tox-pep8/3a6a930/ : FAILURE in 8m 58s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/66/check/openstack-tox-py27/655feab/ : SUCCESS in 6m 03s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/66/check/openstack-tox-py35/d030508/ : SUCCESS in 7m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/66/check/build-openstack-sphinx-docs/d6c6d80/html/ : SUCCESS in 7m 49s\n- build-openstack-releasenotes http://logs.openstack.org/12/435612/66/check/build-openstack-releasenotes/4307f59/html/ : SUCCESS in 7m 27s\n- openstack-tox-cover http://logs.openstack.org/12/435612/66/check/openstack-tox-cover/c8635d6/cover/ : SUCCESS in 9m 00s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/66/check/neutron-lbaasv2-dsvm-api/344db5f/ : SUCCESS in 1h 02m 56s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/66/check/neutron-lbaasv2-dsvm-py3x-api/4908545/ : SUCCESS in 1h 01m 14s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/66/check/neutron-lbaasv2-dsvm-scenario/cc17ade/ : SUCCESS in 1h 15m 09s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/66/check/neutron-lbaasv2-dsvm-py3x-scenario/c6d9475/ : SUCCESS in 1h 27m 01s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/66/check/octavia-tox-functional/905a8fc/ : SUCCESS in 10m 35s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/66/check/octavia-tox-functional-py35/748df36/ : SUCCESS in 8m 02s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/66/check/octavia-v1-dsvm-scenario/421da0c/ : SUCCESS in 45m 15s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/66/check/octavia-v1-dsvm-py3x-scenario/fddf59e/ : SUCCESS in 51m 47s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/66/check/octavia-v1-dsvm-scenario-multinode/17d526d/ : SUCCESS in 59m 27s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/66/check/octavia-v1-dsvm-py3x-scenario-multinode/ebcf44f/ : SUCCESS in 57m 57s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/66/check/octavia-v1-dsvm-scenario-kvm-centos.7/7e96cc1/ : SUCCESS in 58m 58s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/66/check/octavia-v1-dsvm-scenario-lxd/8e56e62/ : FAILURE in 37m 30s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"55372e03b4d6215875ae0133e3d32d6e6f474ebb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-02 01:21:24.000000000","message":"Patch Set 67: Patch Set 66 was rebased","accounts_in_message":[],"_revision_number":67},{"id":"9d564158ba4a448e8d36bf25e16645e41e992627","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-02 03:44:55.000000000","message":"Patch Set 67: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/67/check/requirements-check/df89845/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/67/check/openstack-tox-pep8/482b96f/ : FAILURE in 7m 33s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/67/check/openstack-tox-py27/d77db60/ : SUCCESS in 6m 17s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/67/check/openstack-tox-py35/d4c8a42/ : SUCCESS in 6m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/67/check/build-openstack-sphinx-docs/a6bba5e/html/ : SUCCESS in 7m 07s\n- build-openstack-releasenotes http://logs.openstack.org/12/435612/67/check/build-openstack-releasenotes/38f6f3b/html/ : SUCCESS in 6m 02s\n- openstack-tox-cover http://logs.openstack.org/12/435612/67/check/openstack-tox-cover/c9ffcc4/cover/ : SUCCESS in 10m 37s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/67/check/neutron-lbaasv2-dsvm-api/0bb4d6c/ : SUCCESS in 59m 01s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/67/check/neutron-lbaasv2-dsvm-py3x-api/4d51537/ : SUCCESS in 1h 01m 10s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/67/check/neutron-lbaasv2-dsvm-scenario/4365d39/ : SUCCESS in 1h 21m 30s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/67/check/neutron-lbaasv2-dsvm-py3x-scenario/5672734/ : SUCCESS in 1h 19m 29s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/67/check/octavia-tox-functional/3ad5f18/ : SUCCESS in 7m 59s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/67/check/octavia-tox-functional-py35/1685001/ : SUCCESS in 7m 53s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/67/check/octavia-v1-dsvm-scenario/4b0beba/ : SUCCESS in 53m 43s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/67/check/octavia-v1-dsvm-py3x-scenario/3553806/ : SUCCESS in 45m 14s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/67/check/octavia-v1-dsvm-scenario-multinode/7b2cf27/ : SUCCESS in 58m 49s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/67/check/octavia-v1-dsvm-py3x-scenario-multinode/967cfa2/ : SUCCESS in 59m 32s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/67/check/octavia-v1-dsvm-scenario-kvm-centos.7/495aac5/ : TIMED_OUT in 2h 18m 40s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/67/check/octavia-v1-dsvm-scenario-lxd/5e8abd4/ : FAILURE in 40m 03s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"8d17685c2f493c3806d5804f3398d515f1e68abc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-02 23:15:36.000000000","message":"Patch Set 67: Workflow-1","accounts_in_message":[],"_revision_number":67},{"id":"97d492c8098456a04ad49e172f0bf54f60aff925","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-11 13:10:32.000000000","message":"Uploaded patch set 68.","accounts_in_message":[],"_revision_number":68},{"id":"2de092014c056a6fb27a35d171c96238ca2ee03f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-11 13:31:36.000000000","message":"Uploaded patch set 69.","accounts_in_message":[],"_revision_number":69},{"id":"033747822080e8757ee6a1433d3b7b58e3759b69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-11 15:36:34.000000000","message":"Patch Set 69: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/69/check/requirements-check/03b6855/ : SUCCESS in 2m 54s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/69/check/openstack-tox-pep8/db4f16a/ : SUCCESS in 6m 55s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/69/check/openstack-tox-py27/dc3ef36/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/69/check/openstack-tox-py35/7457a84/ : SUCCESS in 10m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/69/check/build-openstack-sphinx-docs/628399b/html/ : SUCCESS in 6m 32s\n- build-openstack-releasenotes http://logs.openstack.org/12/435612/69/check/build-openstack-releasenotes/af2fab1/html/ : SUCCESS in 4m 55s\n- openstack-tox-cover http://logs.openstack.org/12/435612/69/check/openstack-tox-cover/50d683d/cover/ : SUCCESS in 8m 43s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/69/check/neutron-lbaasv2-dsvm-api/cca721c/ : FAILURE in 1h 59m 10s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/69/check/neutron-lbaasv2-dsvm-py3x-api/d710632/ : FAILURE in 2h 03m 33s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/69/check/neutron-lbaasv2-dsvm-scenario/65600de/ : SUCCESS in 1h 17m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/69/check/neutron-lbaasv2-dsvm-py3x-scenario/4d00fa0/ : SUCCESS in 1h 11m 38s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/69/check/octavia-tox-functional/4b47593/ : SUCCESS in 7m 54s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/69/check/octavia-tox-functional-py35/c7eba39/ : SUCCESS in 7m 08s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/69/check/octavia-v1-dsvm-scenario/f25cf21/ : SUCCESS in 53m 55s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/69/check/octavia-v1-dsvm-py3x-scenario/5ed9589/ : SUCCESS in 39m 35s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/69/check/octavia-v1-dsvm-scenario-multinode/8fe9ed7/ : SUCCESS in 50m 16s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/69/check/octavia-v1-dsvm-py3x-scenario-multinode/45b0f9e/ : SUCCESS in 52m 42s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/69/check/octavia-v1-dsvm-scenario-kvm-centos.7/5309709/ : SUCCESS in 53m 30s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/69/check/octavia-v1-dsvm-scenario-lxd/999de1e/ : FAILURE in 1h 05m 15s (non-voting)","accounts_in_message":[],"_revision_number":69},{"id":"eae08e8d31a0df8c70bc9e7cc9f680ea6dee7f61","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-16 05:40:32.000000000","message":"Patch Set 69: Workflow-1","accounts_in_message":[],"_revision_number":69},{"id":"def17f18c0fd52b6d94ead68623d5bf1f84e94e3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-16 09:17:43.000000000","message":"Uploaded patch set 70.","accounts_in_message":[],"_revision_number":70},{"id":"2fd9e34e0a73d245470550878f641d897aa3ce4b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-16 09:52:28.000000000","message":"Uploaded patch set 71.","accounts_in_message":[],"_revision_number":71},{"id":"944414f275577dc5b3fa8e986d540dbdc22d53f1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-16 09:52:44.000000000","message":"Patch Set 71: Workflow-1","accounts_in_message":[],"_revision_number":71},{"id":"db88526251fc2ddc166ff3488b1c0348f5a05995","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-16 12:00:15.000000000","message":"Patch Set 71: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/71/check/requirements-check/b80e717/ : SUCCESS in 3m 01s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/71/check/openstack-tox-pep8/d2f6be7/ : SUCCESS in 7m 33s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/71/check/openstack-tox-py27/41a7c48/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/71/check/openstack-tox-py35/61e64b7/ : SUCCESS in 10m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/71/check/build-openstack-sphinx-docs/23f42eb/html/ : SUCCESS in 6m 44s\n- build-openstack-releasenotes http://logs.openstack.org/12/435612/71/check/build-openstack-releasenotes/8d57df1/html/ : SUCCESS in 5m 34s\n- openstack-tox-cover http://logs.openstack.org/12/435612/71/check/openstack-tox-cover/181c721/cover/ : SUCCESS in 9m 36s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/71/check/neutron-lbaasv2-dsvm-api/689cc14/ : FAILURE in 2h 00m 27s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/71/check/neutron-lbaasv2-dsvm-py3x-api/cc7021e/ : FAILURE in 2h 00m 18s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/71/check/neutron-lbaasv2-dsvm-scenario/621e9f6/ : SUCCESS in 1h 28m 09s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/71/check/neutron-lbaasv2-dsvm-py3x-scenario/47e8e33/ : SUCCESS in 1h 30m 46s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/71/check/octavia-tox-functional/dc1c761/ : SUCCESS in 7m 43s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/71/check/octavia-tox-functional-py35/c1649d6/ : SUCCESS in 12m 55s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/71/check/octavia-v1-dsvm-scenario/63fa5f2/ : SUCCESS in 44m 26s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/71/check/octavia-v1-dsvm-py3x-scenario/7d32c15/ : SUCCESS in 47m 34s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/71/check/octavia-v1-dsvm-scenario-multinode/629dfbe/ : SUCCESS in 55m 39s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/71/check/octavia-v1-dsvm-py3x-scenario-multinode/0178fbf/ : SUCCESS in 52m 47s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/71/check/octavia-v1-dsvm-scenario-kvm-centos.7/1717b03/ : SUCCESS in 41m 53s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/71/check/octavia-v1-dsvm-scenario-lxd/86b552c/ : FAILURE in 39m 33s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"372237025c9b09f1102d58f8fc96d81bbc3e00e2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-27 20:50:20.000000000","message":"Uploaded patch set 72.","accounts_in_message":[],"_revision_number":72},{"id":"87565e366a9ac382051e0a54554a01a70b0c615f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-27 22:55:52.000000000","message":"Patch Set 72: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/72/check/requirements-check/44179bf/ : SUCCESS in 2m 58s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/72/check/openstack-tox-pep8/d0df4c6/ : SUCCESS in 6m 20s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/72/check/openstack-tox-py27/caa5187/ : FAILURE in 5m 13s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/72/check/openstack-tox-py35/929e90b/ : FAILURE in 4m 49s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/72/check/build-openstack-sphinx-docs/ed5e278/html/ : SUCCESS in 7m 06s\n- openstack-tox-cover http://logs.openstack.org/12/435612/72/check/openstack-tox-cover/b18cb92/ : POST_FAILURE in 5m 39s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/72/check/neutron-lbaasv2-dsvm-api/a104b51/ : FAILURE in 2h 01m 47s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/72/check/neutron-lbaasv2-dsvm-py3x-api/0e4909e/ : FAILURE in 2h 02m 54s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/72/check/neutron-lbaasv2-dsvm-scenario/24f188d/ : SUCCESS in 1h 30m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/72/check/neutron-lbaasv2-dsvm-py3x-scenario/429a8b4/ : SUCCESS in 1h 30m 43s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/72/check/octavia-tox-functional/13198ff/ : SUCCESS in 7m 35s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/72/check/octavia-tox-functional-py35/424257a/ : SUCCESS in 8m 40s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/72/check/octavia-v1-dsvm-scenario/ab4fc12/ : SUCCESS in 46m 09s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/72/check/octavia-v1-dsvm-py3x-scenario/13f3521/ : SUCCESS in 43m 28s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/72/check/octavia-v1-dsvm-scenario-multinode/cab9c57/ : SUCCESS in 52m 38s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/72/check/octavia-v1-dsvm-py3x-scenario-multinode/b88b792/ : SUCCESS in 56m 54s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/72/check/octavia-v1-dsvm-scenario-kvm-centos.7/308e036/ : SUCCESS in 40m 59s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/72/check/octavia-v1-dsvm-scenario-lxd/156353f/ : FAILURE in 44m 37s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"f9ff1ba54e4378cc8b7d11826c4843ea0de3d656","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-30 00:39:15.000000000","message":"Uploaded patch set 73.","accounts_in_message":[],"_revision_number":73},{"id":"a527ea26f8f39e083465a0f77859a34c276b2ca3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-30 01:24:14.000000000","message":"Uploaded patch set 74.","accounts_in_message":[],"_revision_number":74},{"id":"fc5f649823124a05ddab6d791efb29066f075c07","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-30 02:13:57.000000000","message":"Uploaded patch set 75.","accounts_in_message":[],"_revision_number":75},{"id":"0f5b91c38bd7b1bde6db15b7441596bfd9cb7da1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-30 03:04:13.000000000","message":"Uploaded patch set 76.","accounts_in_message":[],"_revision_number":76},{"id":"5db2c717174838591f557951370e352dc67eafb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 05:05:44.000000000","message":"Patch Set 76: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/76/check/requirements-check/7e04649/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/76/check/openstack-tox-pep8/0c2a6f9/ : SUCCESS in 6m 17s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/76/check/openstack-tox-py27/365e1fa/ : FAILURE in 5m 59s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/76/check/openstack-tox-py35/cf2e088/ : FAILURE in 5m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/76/check/build-openstack-sphinx-docs/1bb826a/html/ : SUCCESS in 7m 59s\n- openstack-tox-cover http://logs.openstack.org/12/435612/76/check/openstack-tox-cover/81d9ca2/ : POST_FAILURE in 5m 24s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/76/check/neutron-lbaasv2-dsvm-api/0a29241/ : FAILURE in 1h 59m 51s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/76/check/neutron-lbaasv2-dsvm-scenario/de3773a/ : SUCCESS in 1h 28m 41s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/76/check/octavia-tox-functional/200cd60/ : SUCCESS in 7m 02s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/76/check/octavia-v1-dsvm-scenario/bc6ece7/ : SUCCESS in 44m 44s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/76/check/octavia-v1-dsvm-scenario-multinode/6f14ce8/ : SUCCESS in 52m 06s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/76/check/octavia-v1-dsvm-scenario-kvm-centos.7/d01a19e/ : SUCCESS in 50m 28s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"ff2f52fd26a49a79564ef12f9660adf595f1771e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-11-30 05:34:02.000000000","message":"Uploaded patch set 77.","accounts_in_message":[],"_revision_number":77},{"id":"339d35c794c2214eeea217a9683f2f3d58f9e48f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 07:39:24.000000000","message":"Patch Set 77: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/77/check/requirements-check/ff13f4c/ : SUCCESS in 2m 30s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/77/check/openstack-tox-pep8/8dd9356/ : SUCCESS in 9m 17s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/77/check/openstack-tox-py27/247ede2/ : SUCCESS in 4m 45s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/77/check/openstack-tox-py35/9700910/ : SUCCESS in 4m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/77/check/build-openstack-sphinx-docs/655cb15/html/ : SUCCESS in 7m 02s\n- openstack-tox-cover http://logs.openstack.org/12/435612/77/check/openstack-tox-cover/3ff1221/cover/ : SUCCESS in 10m 21s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/77/check/neutron-lbaasv2-dsvm-api/13f4917/ : FAILURE in 2h 04m 08s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/77/check/neutron-lbaasv2-dsvm-scenario/fe23c1d/ : SUCCESS in 1h 15m 14s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/77/check/octavia-tox-functional/2d3c117/ : SUCCESS in 8m 33s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/77/check/octavia-v1-dsvm-scenario/ddd146b/ : SUCCESS in 46m 10s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/77/check/octavia-v1-dsvm-scenario-multinode/4e0af27/ : SUCCESS in 1h 12m 51s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/77/check/octavia-v1-dsvm-scenario-kvm-centos.7/54d1642/ : SUCCESS in 44m 18s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"9c364348df897f872d75842b0667c6f2c614c773","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-01 00:13:13.000000000","message":"Uploaded patch set 78.","accounts_in_message":[],"_revision_number":78},{"id":"4e5e03a923b095f262e5dcf5686c1b212c3608bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-01 02:25:21.000000000","message":"Patch Set 78: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/78/check/requirements-check/61cade5/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/78/check/openstack-tox-pep8/8ce605b/ : SUCCESS in 7m 43s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/78/check/openstack-tox-py27/f9bc703/ : SUCCESS in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/78/check/openstack-tox-py35/da3a45d/ : SUCCESS in 5m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/78/check/build-openstack-sphinx-docs/526b326/html/ : SUCCESS in 6m 40s\n- openstack-tox-cover http://logs.openstack.org/12/435612/78/check/openstack-tox-cover/700209b/cover/ : SUCCESS in 9m 32s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/78/check/neutron-lbaasv2-dsvm-api/987b92f/ : FAILURE in 2h 01m 37s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/78/check/neutron-lbaasv2-dsvm-scenario/93294f7/ : SUCCESS in 1h 34m 52s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/78/check/octavia-tox-functional/0e52222/ : SUCCESS in 7m 45s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/78/check/octavia-v1-dsvm-scenario/e56714b/ : SUCCESS in 45m 12s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/78/check/octavia-v1-dsvm-scenario-multinode/3b8edb7/ : SUCCESS in 55m 10s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/78/check/octavia-v1-dsvm-scenario-kvm-centos.7/32a2222/ : SUCCESS in 48m 58s (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"535417cde41f4179129e56a154688e8fbf66dcbe","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-01 19:27:04.000000000","message":"Uploaded patch set 79.","accounts_in_message":[],"_revision_number":79},{"id":"810b9047db448f65e2bd8080877396e6d1a66b71","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-01 21:28:31.000000000","message":"Uploaded patch set 80.","accounts_in_message":[],"_revision_number":80},{"id":"093e156ec3ea9f13898b0e4cb4617a2944fa4df3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-01 22:18:48.000000000","message":"Uploaded patch set 81.","accounts_in_message":[],"_revision_number":81},{"id":"edaa1ea502e05912ae3fca61c3254ee957e9be02","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-01 23:44:26.000000000","message":"Uploaded patch set 82.","accounts_in_message":[],"_revision_number":82},{"id":"d14bff2f59746acb4be78e70e7fd0531516cc4f6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-02 00:04:50.000000000","message":"Uploaded patch set 83.","accounts_in_message":[],"_revision_number":83},{"id":"52047b321143c9f52814d49bc3941b795d490b6f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-02 00:51:33.000000000","message":"Uploaded patch set 84.","accounts_in_message":[],"_revision_number":84},{"id":"bc748f38744e6359229b68bf63f5850c137313cf","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-02 02:12:19.000000000","message":"Uploaded patch set 85.","accounts_in_message":[],"_revision_number":85},{"id":"fb4a6224892df4009e7edbd4fe859cd679292288","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-02 04:14:05.000000000","message":"Patch Set 85: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/85/check/requirements-check/6b51eb2/ : SUCCESS in 2m 41s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/85/check/openstack-tox-pep8/f2d0827/ : SUCCESS in 6m 16s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/85/check/openstack-tox-py27/e30cb79/ : SUCCESS in 5m 24s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/85/check/openstack-tox-py35/8e1f8ae/ : SUCCESS in 4m 07s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/85/check/build-openstack-sphinx-docs/7b1312a/html/ : SUCCESS in 7m 12s\n- openstack-tox-cover http://logs.openstack.org/12/435612/85/check/openstack-tox-cover/d992f19/cover/ : SUCCESS in 8m 08s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/85/check/neutron-lbaasv2-dsvm-api/615fdeb/ : FAILURE in 2h 00m 48s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/85/check/neutron-lbaasv2-dsvm-scenario/bf9ca74/ : SUCCESS in 1h 10m 35s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/85/check/octavia-tox-functional/40c8f4b/ : SUCCESS in 8m 19s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/85/check/octavia-v1-dsvm-scenario/6352080/ : SUCCESS in 59m 39s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/85/check/octavia-v1-dsvm-scenario-multinode/3c57036/ : SUCCESS in 47m 59s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/85/check/octavia-v1-dsvm-scenario-kvm-centos.7/fa005ee/ : SUCCESS in 48m 45s (non-voting)","accounts_in_message":[],"_revision_number":85},{"id":"a84d220c1285334f04f1d0cd35f2012b2c479599","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-04 19:59:43.000000000","message":"Uploaded patch set 86.","accounts_in_message":[],"_revision_number":86},{"id":"df911f9ac03d1803e31ac9811d85d208dcd8fe7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-04 22:08:48.000000000","message":"Patch Set 86: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/86/check/requirements-check/531271f/ : SUCCESS in 2m 38s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/86/check/openstack-tox-pep8/3b1a397/ : SUCCESS in 12m 49s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/86/check/openstack-tox-py27/4f5f46d/ : SUCCESS in 5m 17s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/86/check/openstack-tox-py35/54f612c/ : SUCCESS in 4m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/86/check/build-openstack-sphinx-docs/b36b257/html/ : SUCCESS in 12m 29s\n- openstack-tox-cover http://logs.openstack.org/12/435612/86/check/openstack-tox-cover/122ded6/cover/ : SUCCESS in 11m 26s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/86/check/neutron-lbaasv2-dsvm-api/ac9e99d/ : FAILURE in 2h 03m 49s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/86/check/neutron-lbaasv2-dsvm-py3x-api/73ed65a/ : FAILURE in 2h 00m 45s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/86/check/neutron-lbaasv2-dsvm-scenario/7343271/ : SUCCESS in 1h 23m 46s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/86/check/neutron-lbaasv2-dsvm-py3x-scenario/b0d3c4d/ : SUCCESS in 1h 12m 23s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/86/check/octavia-tox-functional/348b944/ : SUCCESS in 8m 00s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/86/check/octavia-tox-functional-py35/1d00d1b/ : SUCCESS in 6m 57s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/86/check/octavia-v1-dsvm-scenario/8aa99ce/ : SUCCESS in 41m 39s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/86/check/octavia-v1-dsvm-py3x-scenario/f574027/ : SUCCESS in 40m 53s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/86/check/octavia-v1-dsvm-scenario-multinode/c32dc1a/ : SUCCESS in 57m 51s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/86/check/octavia-v1-dsvm-py3x-scenario-multinode/cbfa94f/ : SUCCESS in 1h 26m 07s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/86/check/octavia-v1-dsvm-scenario-kvm-centos.7/4ccd390/ : SUCCESS in 44m 54s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/86/check/octavia-v1-dsvm-scenario-lxd/ee293bd/ : FAILURE in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":86},{"id":"ef4e327be5861e35a107b5fc2e04f8e919fad3f1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-05 01:55:56.000000000","message":"Uploaded patch set 87.","accounts_in_message":[],"_revision_number":87},{"id":"fc578f8eb5440f09761f68e90a6ba1ae4857ab72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 03:33:52.000000000","message":"Patch Set 87: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/87/check/requirements-check/6066cd0/ : SUCCESS in 3m 13s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/87/check/openstack-tox-pep8/a5d65d7/ : SUCCESS in 8m 27s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/87/check/openstack-tox-py27/c567413/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/87/check/openstack-tox-py35/86645b1/ : SUCCESS in 6m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/87/check/build-openstack-sphinx-docs/212a698/html/ : SUCCESS in 7m 12s\n- openstack-tox-cover http://logs.openstack.org/12/435612/87/check/openstack-tox-cover/fa5fee9/cover/ : SUCCESS in 12m 10s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/87/check/neutron-lbaasv2-dsvm-api/1154401/ : SUCCESS in 55m 47s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/87/check/neutron-lbaasv2-dsvm-py3x-api/476480f/ : SUCCESS in 1h 00m 25s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/87/check/neutron-lbaasv2-dsvm-scenario/330a56f/ : SUCCESS in 1h 13m 35s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/87/check/neutron-lbaasv2-dsvm-py3x-scenario/5e4219a/ : SUCCESS in 1h 29m 06s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/87/check/octavia-tox-functional/ec245af/ : SUCCESS in 8m 22s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/87/check/octavia-tox-functional-py35/cb021c6/ : SUCCESS in 10m 03s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/87/check/octavia-v1-dsvm-scenario/456595a/ : SUCCESS in 58m 26s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/87/check/octavia-v1-dsvm-py3x-scenario/9c56c57/ : SUCCESS in 44m 24s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/87/check/octavia-v1-dsvm-scenario-multinode/af17406/ : SUCCESS in 1h 06m 18s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/87/check/octavia-v1-dsvm-py3x-scenario-multinode/5ed1afd/ : SUCCESS in 52m 51s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/87/check/octavia-v1-dsvm-scenario-kvm-centos.7/6a4d349/ : SUCCESS in 53m 38s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/87/check/octavia-v1-dsvm-scenario-lxd/c786ee5/ : FAILURE in 37m 16s (non-voting)","accounts_in_message":[],"_revision_number":87},{"id":"2eb41f9593bb530e8ff5c34d04db6f500a635418","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-05 08:57:01.000000000","message":"Patch Set 87: Workflow-1","accounts_in_message":[],"_revision_number":87},{"id":"5ae628aebf5ea3624b62dd3c9b15611a8805b228","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-08 21:49:56.000000000","message":"Uploaded patch set 88.","accounts_in_message":[],"_revision_number":88},{"id":"d520c765a8885beea07f88b9f6389ca154abc12f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-08 23:30:39.000000000","message":"Patch Set 88: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/88/check/requirements-check/d987d23/ : SUCCESS in 3m 16s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/88/check/openstack-tox-pep8/57fe0f2/ : SUCCESS in 11m 27s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/88/check/openstack-tox-py27/b3fe836/ : SUCCESS in 6m 48s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/88/check/openstack-tox-py35/7260bd7/ : SUCCESS in 7m 34s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/88/check/build-openstack-sphinx-docs/c52cbdd/html/ : SUCCESS in 8m 19s\n- openstack-tox-cover http://logs.openstack.org/12/435612/88/check/openstack-tox-cover/2d0d547/cover/ : SUCCESS in 11m 24s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/88/check/neutron-lbaasv2-dsvm-api/e83b039/ : SUCCESS in 59m 00s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/88/check/neutron-lbaasv2-dsvm-py3x-api/8e7878e/ : SUCCESS in 1h 03m 23s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/88/check/neutron-lbaasv2-dsvm-scenario/fefb476/ : SUCCESS in 1h 34m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/88/check/neutron-lbaasv2-dsvm-py3x-scenario/054c9bb/ : SUCCESS in 55m 21s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/88/check/octavia-tox-functional/08a2bba/ : SUCCESS in 13m 00s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/88/check/octavia-tox-functional-py35/4acd29a/ : SUCCESS in 9m 22s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/88/check/octavia-v1-dsvm-scenario/0597e10/ : SUCCESS in 48m 17s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/88/check/octavia-v1-dsvm-py3x-scenario/3fc1ada/ : SUCCESS in 58m 47s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/88/check/octavia-v1-dsvm-scenario-multinode/c0d4ba0/ : SUCCESS in 53m 04s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/88/check/octavia-v1-dsvm-py3x-scenario-multinode/8062b77/ : SUCCESS in 1h 04m 53s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/88/check/octavia-v1-dsvm-scenario-kvm-centos.7/fac7840/ : SUCCESS in 45m 32s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/88/check/octavia-v1-dsvm-scenario-lxd/4c496f8/ : FAILURE in 38m 55s (non-voting)","accounts_in_message":[],"_revision_number":88},{"id":"805cb676a3540ee7a9f700b573dd69b730fdf2af","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-09 02:14:33.000000000","message":"Uploaded patch set 89.","accounts_in_message":[],"_revision_number":89},{"id":"a1583a34c90714a50d8174e129ba31aa5c70c140","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-09 03:59:02.000000000","message":"Patch Set 89: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/89/check/requirements-check/d954708/ : SUCCESS in 2m 19s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/89/check/openstack-tox-pep8/08bc82c/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/89/check/openstack-tox-py27/0954b12/ : SUCCESS in 8m 48s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/89/check/openstack-tox-py35/a9e443b/ : SUCCESS in 5m 14s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/89/check/build-openstack-sphinx-docs/a2921fd/html/ : SUCCESS in 5m 29s\n- openstack-tox-cover http://logs.openstack.org/12/435612/89/check/openstack-tox-cover/ee11caf/cover/ : SUCCESS in 9m 37s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/89/check/neutron-lbaasv2-dsvm-api/aecd230/ : SUCCESS in 59m 15s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/89/check/neutron-lbaasv2-dsvm-py3x-api/8d5ee91/ : SUCCESS in 53m 19s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/89/check/neutron-lbaasv2-dsvm-scenario/abb85fa/ : SUCCESS in 1h 08m 17s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/89/check/neutron-lbaasv2-dsvm-py3x-scenario/67550bf/ : FAILURE in 1h 41m 19s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/89/check/octavia-tox-functional/7950f56/ : SUCCESS in 6m 57s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/89/check/octavia-tox-functional-py35/57c4d2b/ : SUCCESS in 7m 23s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/89/check/octavia-v1-dsvm-scenario/6105180/ : SUCCESS in 44m 22s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/89/check/octavia-v1-dsvm-py3x-scenario/0765afa/ : SUCCESS in 40m 56s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/89/check/octavia-v1-dsvm-scenario-multinode/9d64427/ : SUCCESS in 52m 23s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/89/check/octavia-v1-dsvm-py3x-scenario-multinode/4cf6059/ : SUCCESS in 1h 08m 04s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/89/check/octavia-v1-dsvm-scenario-kvm-centos.7/1881f75/ : SUCCESS in 46m 59s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/89/check/octavia-v1-dsvm-scenario-lxd/1801f48/ : FAILURE in 34m 15s (non-voting)","accounts_in_message":[],"_revision_number":89},{"id":"c5d794a5e350a4c0c8a38aaae5f5d802ea5b3a5f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-12 23:15:14.000000000","message":"Uploaded patch set 90.","accounts_in_message":[],"_revision_number":90},{"id":"790c4af19a7a7e5edf1eb97d0205d44c3fc60f4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-13 01:28:08.000000000","message":"Patch Set 90: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/90/check/requirements-check/00b441f/ : SUCCESS in 3m 25s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/90/check/openstack-tox-pep8/cb1c21e/ : SUCCESS in 9m 25s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/90/check/openstack-tox-py27/3f6bd06/ : FAILURE in 5m 17s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/90/check/openstack-tox-py35/83483d1/ : SUCCESS in 5m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/90/check/build-openstack-sphinx-docs/55244ea/html/ : SUCCESS in 6m 05s\n- openstack-tox-cover http://logs.openstack.org/12/435612/90/check/openstack-tox-cover/bfc1467/cover/ : SUCCESS in 9m 07s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/90/check/neutron-lbaasv2-dsvm-api/de2bd71/ : SUCCESS in 53m 27s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/90/check/neutron-lbaasv2-dsvm-py3x-api/133f202/ : SUCCESS in 58m 17s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/90/check/neutron-lbaasv2-dsvm-scenario/efe52bc/ : SUCCESS in 1h 12m 13s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/90/check/neutron-lbaasv2-dsvm-py3x-scenario/a324a63/ : SUCCESS in 1h 15m 29s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/90/check/octavia-tox-functional/0f1c1c5/ : SUCCESS in 10m 54s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/90/check/octavia-tox-functional-py35/a316202/ : SUCCESS in 7m 17s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/90/check/octavia-v1-dsvm-scenario/92b62ad/ : SUCCESS in 41m 45s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/90/check/octavia-v1-dsvm-py3x-scenario/698d691/ : SUCCESS in 47m 49s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/90/check/octavia-v1-dsvm-scenario-multinode/a931931/ : SUCCESS in 56m 10s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/90/check/octavia-v1-dsvm-py3x-scenario-multinode/8fa4b43/ : SUCCESS in 51m 14s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/90/check/octavia-v1-dsvm-scenario-kvm-centos.7/e75c560/ : SUCCESS in 46m 14s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/90/check/octavia-v1-dsvm-scenario-lxd/dfe59d6/ : FAILURE in 31m 06s (non-voting)","accounts_in_message":[],"_revision_number":90},{"id":"daf78648ecf0bce8407330d01527873b1ff6e967","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-13 22:46:23.000000000","message":"Uploaded patch set 91.","accounts_in_message":[],"_revision_number":91},{"id":"9c89ff960b91a63f73d1ed5ce5b8e752a1628e75","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-13 23:36:09.000000000","message":"Uploaded patch set 92.","accounts_in_message":[],"_revision_number":92},{"id":"79aed7e439e565356aceac20ac30fe067c4e67b4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-14 00:32:00.000000000","message":"Uploaded patch set 93.","accounts_in_message":[],"_revision_number":93},{"id":"5ff3e00d493ad7b9ca7da01f4dfbac076e68d034","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-14 02:23:47.000000000","message":"Uploaded patch set 94.","accounts_in_message":[],"_revision_number":94},{"id":"1a7756337e881887f1cfefef11a3e0303b327998","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-14 03:44:24.000000000","message":"Patch Set 94: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/94/check/requirements-check/1742e2e/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/94/check/openstack-tox-pep8/9eddda6/ : SUCCESS in 6m 19s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/94/check/openstack-tox-py27/3f2c1f3/ : FAILURE in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/94/check/openstack-tox-py35/971884e/ : FAILURE in 8m 08s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/94/check/build-openstack-sphinx-docs/87f7d4e/html/ : SUCCESS in 6m 15s\n- openstack-tox-cover http://logs.openstack.org/12/435612/94/check/openstack-tox-cover/22e236c/ : POST_FAILURE in 12m 25s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/94/check/neutron-lbaasv2-dsvm-api/9795c5f/ : SUCCESS in 1h 12m 53s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/94/check/neutron-lbaasv2-dsvm-py3x-api/c2a4634/ : SUCCESS in 55m 23s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/94/check/neutron-lbaasv2-dsvm-scenario/c33339d/ : SUCCESS in 1h 18m 13s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/94/check/neutron-lbaasv2-dsvm-py3x-scenario/aaf0f03/ : SUCCESS in 58m 42s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/94/check/octavia-tox-functional/8583709/ : SUCCESS in 8m 34s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/94/check/octavia-tox-functional-py35/ee6cfd6/ : SUCCESS in 9m 27s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/94/check/octavia-v1-dsvm-scenario/a829d7d/ : SUCCESS in 49m 36s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/94/check/octavia-v1-dsvm-py3x-scenario/1ae3828/ : SUCCESS in 42m 20s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/94/check/octavia-v1-dsvm-scenario-multinode/10c682d/ : SUCCESS in 1h 03m 46s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/94/check/octavia-v1-dsvm-py3x-scenario-multinode/1711b3c/ : FAILURE in 15m 11s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/94/check/octavia-v1-dsvm-scenario-kvm-centos.7/976b1f4/ : SUCCESS in 55m 08s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/94/check/octavia-v1-dsvm-scenario-lxd/a7406fc/ : FAILURE in 34m 52s (non-voting)","accounts_in_message":[],"_revision_number":94},{"id":"abbd546e1529cef52f4dc74133a8101ef5630849","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-14 04:07:02.000000000","message":"Uploaded patch set 95.","accounts_in_message":[],"_revision_number":95},{"id":"881f745c38b0650c906aac00eff566dc7d91707e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-14 05:40:37.000000000","message":"Patch Set 95: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/95/check/requirements-check/1ca630f/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/95/check/openstack-tox-pep8/de66e8c/ : SUCCESS in 7m 10s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/95/check/openstack-tox-py27/551f4d4/ : SUCCESS in 6m 51s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/95/check/openstack-tox-py35/bd919f2/ : SUCCESS in 7m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/95/check/build-openstack-sphinx-docs/15f41a0/html/ : SUCCESS in 7m 56s\n- openstack-tox-cover http://logs.openstack.org/12/435612/95/check/openstack-tox-cover/9ead572/cover/ : SUCCESS in 9m 51s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/95/check/neutron-lbaasv2-dsvm-api/ebe217c/ : SUCCESS in 54m 12s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/95/check/neutron-lbaasv2-dsvm-py3x-api/4868dcf/ : SUCCESS in 57m 17s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/95/check/neutron-lbaasv2-dsvm-scenario/b365ccf/ : SUCCESS in 1h 19m 53s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/95/check/neutron-lbaasv2-dsvm-py3x-scenario/e0b1fb0/ : SUCCESS in 1h 27m 10s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/95/check/octavia-tox-functional/0a57c1c/ : SUCCESS in 9m 38s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/95/check/octavia-tox-functional-py35/d7932b8/ : SUCCESS in 11m 06s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/95/check/octavia-v1-dsvm-scenario/ceee0fa/ : SUCCESS in 50m 15s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/95/check/octavia-v1-dsvm-py3x-scenario/6e00c91/ : SUCCESS in 49m 53s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/95/check/octavia-v1-dsvm-scenario-multinode/a093f9a/ : SUCCESS in 1h 01m 42s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/95/check/octavia-v1-dsvm-py3x-scenario-multinode/54e2c79/ : SUCCESS in 56m 06s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/95/check/octavia-v1-dsvm-scenario-kvm-centos.7/5d45fff/ : SUCCESS in 47m 18s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/95/check/octavia-v1-dsvm-scenario-lxd/ac4aa87/ : FAILURE in 36m 45s (non-voting)","accounts_in_message":[],"_revision_number":95},{"id":"5ba3faa5cfb8545ef0b60f59a5d49a30b71fdc37","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-14 20:21:59.000000000","message":"Uploaded patch set 96.","accounts_in_message":[],"_revision_number":96},{"id":"a60aedb84e1c39392c0fe63fd62d0bd9c0c4cdc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-14 21:47:10.000000000","message":"Patch Set 96: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/96/check/requirements-check/495d99f/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/96/check/openstack-tox-pep8/a53b80f/ : SUCCESS in 6m 49s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/96/check/openstack-tox-py27/5b5884f/ : FAILURE in 5m 08s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/96/check/openstack-tox-py35/4c707ee/ : FAILURE in 5m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/96/check/build-openstack-sphinx-docs/aa13239/html/ : SUCCESS in 10m 02s\n- openstack-tox-cover http://logs.openstack.org/12/435612/96/check/openstack-tox-cover/3a37056/ : POST_FAILURE in 6m 57s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/96/check/neutron-lbaasv2-dsvm-api/82380e7/ : SUCCESS in 55m 21s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/96/check/neutron-lbaasv2-dsvm-py3x-api/b9287ba/ : SUCCESS in 58m 50s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/96/check/neutron-lbaasv2-dsvm-scenario/518b123/ : SUCCESS in 1h 10m 23s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/96/check/neutron-lbaasv2-dsvm-py3x-scenario/93e9fc6/ : SUCCESS in 1h 16m 14s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/96/check/octavia-tox-functional/917a3b0/ : SUCCESS in 8m 11s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/96/check/octavia-tox-functional-py35/9fa9ff7/ : SUCCESS in 9m 12s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/96/check/octavia-v1-dsvm-scenario/bfdb5b3/ : SUCCESS in 42m 23s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/96/check/octavia-v1-dsvm-py3x-scenario/f292d08/ : SUCCESS in 42m 52s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/96/check/octavia-v1-dsvm-scenario-multinode/196d8b5/ : SUCCESS in 56m 26s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/96/check/octavia-v1-dsvm-py3x-scenario-multinode/c4e7bea/ : SUCCESS in 52m 16s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/96/check/octavia-v1-dsvm-scenario-kvm-centos.7/691e010/ : SUCCESS in 45m 23s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/96/check/octavia-v1-dsvm-scenario-lxd/c018ef7/ : FAILURE in 33m 53s (non-voting)","accounts_in_message":[],"_revision_number":96},{"id":"79d9e2705c8906041393865dbc35d5fd0a078696","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-14 22:17:12.000000000","message":"Uploaded patch set 97.","accounts_in_message":[],"_revision_number":97},{"id":"9d8012ddee3ae5b2d014267ff1ddf0211f67b953","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-14 23:45:30.000000000","message":"Patch Set 97: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/97/check/requirements-check/eee4efe/ : SUCCESS in 3m 03s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/97/check/openstack-tox-pep8/90d709a/ : FAILURE in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/97/check/openstack-tox-py27/2e39edc/ : FAILURE in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/97/check/openstack-tox-py35/25aec10/ : FAILURE in 5m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/97/check/build-openstack-sphinx-docs/58976d5/html/ : SUCCESS in 6m 53s\n- openstack-tox-cover http://logs.openstack.org/12/435612/97/check/openstack-tox-cover/5319c45/ : POST_FAILURE in 4m 44s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/97/check/neutron-lbaasv2-dsvm-api/630f5fd/ : SUCCESS in 54m 20s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/97/check/neutron-lbaasv2-dsvm-py3x-api/505bfd4/ : SUCCESS in 54m 27s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/97/check/neutron-lbaasv2-dsvm-scenario/a20b836/ : SUCCESS in 1h 21m 23s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/97/check/neutron-lbaasv2-dsvm-py3x-scenario/2608a3e/ : SUCCESS in 1h 15m 26s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/97/check/octavia-tox-functional/3f65924/ : SUCCESS in 14m 14s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/97/check/octavia-tox-functional-py35/5e39a16/ : SUCCESS in 7m 49s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/97/check/octavia-v1-dsvm-scenario/39a869c/ : SUCCESS in 47m 25s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/97/check/octavia-v1-dsvm-py3x-scenario/bc283dd/ : SUCCESS in 41m 28s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/97/check/octavia-v1-dsvm-scenario-multinode/84e7414/ : SUCCESS in 54m 51s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/97/check/octavia-v1-dsvm-py3x-scenario-multinode/b3acfe0/ : SUCCESS in 53m 07s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/97/check/octavia-v1-dsvm-scenario-kvm-centos.7/46bcf5c/ : SUCCESS in 45m 39s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/97/check/octavia-v1-dsvm-scenario-lxd/e547320/ : FAILURE in 35m 27s (non-voting)","accounts_in_message":[],"_revision_number":97},{"id":"0c651e6ed634591a2033394c6dc533eeab8d8ed7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-15 00:02:09.000000000","message":"Uploaded patch set 98.","accounts_in_message":[],"_revision_number":98},{"id":"5872d16f42c88dc7b62b0c2afb0b070238a75b69","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-15 00:12:20.000000000","message":"Uploaded patch set 99.","accounts_in_message":[],"_revision_number":99},{"id":"8f857349127c7ba034c6e2f2852f7584db0f5877","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-15 00:49:59.000000000","message":"Uploaded patch set 100.","accounts_in_message":[],"_revision_number":100},{"id":"2fbfd65d524c8b45b0731405bd9dc9cbe7806306","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-15 02:04:40.000000000","message":"Patch Set 100: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/100/check/requirements-check/e676bbc/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/100/check/openstack-tox-pep8/782dd4c/ : SUCCESS in 8m 52s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/100/check/openstack-tox-py27/dc060ac/ : SUCCESS in 5m 59s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/100/check/openstack-tox-py35/2f15c93/ : SUCCESS in 5m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/100/check/build-openstack-sphinx-docs/2a5a522/html/ : SUCCESS in 7m 07s\n- openstack-tox-cover http://logs.openstack.org/12/435612/100/check/openstack-tox-cover/41d37ce/cover/ : SUCCESS in 9m 23s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/100/check/neutron-lbaasv2-dsvm-api/f8d8b41/ : SUCCESS in 58m 18s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/100/check/neutron-lbaasv2-dsvm-py3x-api/1746601/ : SUCCESS in 59m 56s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/100/check/neutron-lbaasv2-dsvm-scenario/fed1d83/ : SUCCESS in 1h 00m 40s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/100/check/neutron-lbaasv2-dsvm-py3x-scenario/180408e/ : SUCCESS in 57m 09s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/100/check/octavia-tox-functional/34e6cbf/ : SUCCESS in 7m 48s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/100/check/octavia-tox-functional-py35/6c36b70/ : SUCCESS in 7m 55s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/100/check/octavia-v1-dsvm-scenario/002a520/ : SUCCESS in 45m 32s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/100/check/octavia-v1-dsvm-py3x-scenario/b8932c1/ : SUCCESS in 49m 00s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/100/check/octavia-v1-dsvm-scenario-multinode/ea423ab/ : SUCCESS in 58m 29s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/100/check/octavia-v1-dsvm-py3x-scenario-multinode/9232866/ : SUCCESS in 51m 34s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/100/check/octavia-v1-dsvm-scenario-kvm-centos.7/5d1d3a7/ : SUCCESS in 46m 21s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/100/check/octavia-v1-dsvm-scenario-lxd/8e4e991/ : FAILURE in 41m 59s (non-voting)","accounts_in_message":[],"_revision_number":100},{"id":"ed8112502509544d8cf40d8279efad51bed21502","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-15 02:24:52.000000000","message":"Uploaded patch set 101.","accounts_in_message":[],"_revision_number":101},{"id":"87fe7679d5f99b29bae4e618482fd31b530e2e43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-15 03:47:35.000000000","message":"Patch Set 101: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/101/check/requirements-check/910946e/ : SUCCESS in 2m 39s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/101/check/openstack-tox-pep8/d8a6b7b/ : SUCCESS in 7m 02s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/101/check/openstack-tox-py27/dd93e4b/ : SUCCESS in 9m 18s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/101/check/openstack-tox-py35/fc18c93/ : SUCCESS in 8m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/101/check/build-openstack-sphinx-docs/0570b4b/html/ : SUCCESS in 6m 00s\n- openstack-tox-cover http://logs.openstack.org/12/435612/101/check/openstack-tox-cover/7c28615/cover/ : SUCCESS in 8m 43s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/101/check/neutron-lbaasv2-dsvm-api/31b48f7/ : SUCCESS in 1h 11m 47s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/101/check/neutron-lbaasv2-dsvm-py3x-api/9c9ed59/ : SUCCESS in 56m 16s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/101/check/neutron-lbaasv2-dsvm-scenario/b3ad09a/ : SUCCESS in 1h 13m 00s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/101/check/neutron-lbaasv2-dsvm-py3x-scenario/b3cfc08/ : SUCCESS in 1h 16m 57s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/101/check/octavia-tox-functional/b6e8c4a/ : SUCCESS in 5m 58s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/101/check/octavia-tox-functional-py35/c63ac25/ : SUCCESS in 6m 54s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/101/check/octavia-v1-dsvm-scenario/3104b6e/ : SUCCESS in 44m 07s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/101/check/octavia-v1-dsvm-py3x-scenario/2d2dfd5/ : SUCCESS in 39m 41s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/101/check/octavia-v1-dsvm-scenario-multinode/e615b29/ : SUCCESS in 55m 33s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/101/check/octavia-v1-dsvm-py3x-scenario-multinode/c0c8bf1/ : SUCCESS in 1h 04m 56s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/101/check/octavia-v1-dsvm-scenario-kvm-centos.7/e97ce28/ : SUCCESS in 50m 05s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/101/check/octavia-v1-dsvm-scenario-lxd/def7350/ : FAILURE in 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":101},{"id":"fcf458ca8deda54fe3381af54890ca35f89e7d5e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-15 19:03:27.000000000","message":"Uploaded patch set 102.","accounts_in_message":[],"_revision_number":102},{"id":"78c932b92595a28b2c2933f372200f36c818535b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-15 20:21:10.000000000","message":"Patch Set 102: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/102/check/requirements-check/f7e85f9/ : SUCCESS in 3m 16s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/102/check/openstack-tox-pep8/bdd49be/ : SUCCESS in 6m 19s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/102/check/openstack-tox-py27/1337d69/ : SUCCESS in 4m 39s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/102/check/openstack-tox-py35/ed4d99b/ : SUCCESS in 5m 23s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/102/check/build-openstack-sphinx-docs/3af427e/html/ : SUCCESS in 6m 16s\n- openstack-tox-cover http://logs.openstack.org/12/435612/102/check/openstack-tox-cover/4968fed/cover/ : SUCCESS in 8m 26s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/102/check/neutron-lbaasv2-dsvm-api/9320f40/ : SUCCESS in 1h 00m 21s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/102/check/neutron-lbaasv2-dsvm-py3x-api/7b9a976/ : SUCCESS in 52m 25s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/102/check/neutron-lbaasv2-dsvm-scenario/96b9b7a/ : SUCCESS in 1h 05m 41s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/102/check/neutron-lbaasv2-dsvm-py3x-scenario/5b6235b/ : SUCCESS in 1h 11m 56s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/102/check/octavia-tox-functional/1750847/ : SUCCESS in 7m 14s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/102/check/octavia-tox-functional-py35/c48c4d5/ : SUCCESS in 9m 33s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/102/check/octavia-v1-dsvm-scenario/59faf2a/ : SUCCESS in 49m 17s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/102/check/octavia-v1-dsvm-py3x-scenario/1185142/ : SUCCESS in 49m 40s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/102/check/octavia-v1-dsvm-scenario-multinode/9e8fe81/ : SUCCESS in 54m 31s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/102/check/octavia-v1-dsvm-py3x-scenario-multinode/9ed6b12/ : SUCCESS in 55m 25s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/102/check/octavia-v1-dsvm-scenario-kvm-centos.7/072535d/ : SUCCESS in 48m 01s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/102/check/octavia-v1-dsvm-scenario-lxd/ddcccbf/ : FAILURE in 42m 03s (non-voting)","accounts_in_message":[],"_revision_number":102},{"id":"0ded2e2b09e3117062b3013ed5017cfdf8d80f8c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-15 20:50:57.000000000","message":"Uploaded patch set 103.","accounts_in_message":[],"_revision_number":103},{"id":"09eac0d5c3b1565e6843dac75c663490e259d81f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-15 22:20:14.000000000","message":"Patch Set 103: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/103/check/requirements-check/62e5c5d/ : SUCCESS in 2m 51s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/103/check/openstack-tox-pep8/dd58fa7/ : SUCCESS in 7m 14s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/103/check/openstack-tox-py27/fcb92b6/ : SUCCESS in 4m 51s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/103/check/openstack-tox-py35/1b6008f/ : SUCCESS in 4m 00s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/103/check/build-openstack-sphinx-docs/073b6f5/html/ : SUCCESS in 5m 36s\n- openstack-tox-cover http://logs.openstack.org/12/435612/103/check/openstack-tox-cover/af2bb4d/cover/ : SUCCESS in 10m 58s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/103/check/neutron-lbaasv2-dsvm-api/262b538/ : SUCCESS in 54m 18s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/103/check/neutron-lbaasv2-dsvm-py3x-api/304efd7/ : SUCCESS in 1h 04m 57s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/103/check/neutron-lbaasv2-dsvm-scenario/3b73bb1/ : SUCCESS in 1h 19m 14s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/103/check/neutron-lbaasv2-dsvm-py3x-scenario/51d916f/ : SUCCESS in 1h 23m 55s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/103/check/octavia-tox-functional/53fd406/ : SUCCESS in 7m 07s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/103/check/octavia-tox-functional-py35/862308c/ : SUCCESS in 9m 20s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/103/check/octavia-v1-dsvm-scenario/9052463/ : SUCCESS in 43m 15s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/103/check/octavia-v1-dsvm-py3x-scenario/b4076cf/ : SUCCESS in 40m 22s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/103/check/octavia-v1-dsvm-scenario-multinode/f7c645f/ : SUCCESS in 57m 22s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/103/check/octavia-v1-dsvm-py3x-scenario-multinode/5d53e6e/ : SUCCESS in 1h 03m 07s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/103/check/octavia-v1-dsvm-scenario-kvm-centos.7/c239d69/ : SUCCESS in 48m 03s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/103/check/octavia-v1-dsvm-scenario-lxd/50406da/ : FAILURE in 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":103},{"id":"bc85b6defd97ec24de8f64b97a2529c12b222131","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-15 23:56:35.000000000","message":"Uploaded patch set 104.","accounts_in_message":[],"_revision_number":104},{"id":"0dcd6f852708a65cca805d9c31d2d4790f426710","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-16 01:20:33.000000000","message":"Patch Set 104: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/104/check/requirements-check/e6d76c5/ : SUCCESS in 2m 24s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/104/check/openstack-tox-pep8/417990f/ : SUCCESS in 8m 16s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/104/check/openstack-tox-py27/b2a78e6/ : SUCCESS in 5m 34s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/104/check/openstack-tox-py35/7e306c8/ : SUCCESS in 4m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/104/check/build-openstack-sphinx-docs/fed97f3/html/ : SUCCESS in 7m 55s\n- openstack-tox-cover http://logs.openstack.org/12/435612/104/check/openstack-tox-cover/eacb7cd/cover/ : SUCCESS in 9m 14s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/104/check/neutron-lbaasv2-dsvm-api/b59e58e/ : SUCCESS in 55m 00s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/104/check/neutron-lbaasv2-dsvm-py3x-api/a3dbf35/ : SUCCESS in 58m 35s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/104/check/neutron-lbaasv2-dsvm-scenario/5bab563/ : SUCCESS in 1h 08m 18s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/104/check/neutron-lbaasv2-dsvm-py3x-scenario/4393fed/ : SUCCESS in 1h 22m 14s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/104/check/octavia-tox-functional/3e2dd46/ : SUCCESS in 8m 17s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/104/check/octavia-tox-functional-py35/99682ec/ : SUCCESS in 7m 15s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/104/check/octavia-v1-dsvm-scenario/4577787/ : SUCCESS in 49m 36s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/104/check/octavia-v1-dsvm-py3x-scenario/3df9d07/ : SUCCESS in 48m 30s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/104/check/octavia-v1-dsvm-scenario-multinode/6aed91d/ : SUCCESS in 55m 20s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/104/check/octavia-v1-dsvm-py3x-scenario-multinode/97be1ff/ : SUCCESS in 52m 00s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/104/check/octavia-v1-dsvm-scenario-kvm-centos.7/9ced70a/ : SUCCESS in 48m 26s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/104/check/octavia-v1-dsvm-scenario-lxd/17e268f/ : FAILURE in 42m 37s (non-voting)","accounts_in_message":[],"_revision_number":104},{"id":"a07273fb342e0c951e77a998a32b4c2990341b25","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-12-21 20:18:16.000000000","message":"Uploaded patch set 105.","accounts_in_message":[],"_revision_number":105},{"id":"120609725275079e3631441310f1eba5f8428232","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-21 21:52:05.000000000","message":"Patch Set 105: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/105/check/requirements-check/2fb4f43/ : SUCCESS in 2m 28s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/105/check/openstack-tox-pep8/5fff883/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/105/check/openstack-tox-py27/a62f74f/ : SUCCESS in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/105/check/openstack-tox-py35/295ea8d/ : SUCCESS in 7m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/105/check/build-openstack-sphinx-docs/ca0fba3/html/ : SUCCESS in 5m 31s\n- openstack-tox-cover http://logs.openstack.org/12/435612/105/check/openstack-tox-cover/9e534fc/cover/ : SUCCESS in 9m 09s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/105/check/neutron-lbaasv2-dsvm-api/3c8059c/ : SUCCESS in 54m 30s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/105/check/neutron-lbaasv2-dsvm-py3x-api/640c40a/ : SUCCESS in 54m 34s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/105/check/neutron-lbaasv2-dsvm-scenario/4651faa/ : SUCCESS in 1h 11m 26s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/105/check/neutron-lbaasv2-dsvm-py3x-scenario/6ae4283/ : SUCCESS in 1h 25m 17s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/12/435612/105/check/octavia-tox-functional/675e88e/ : SUCCESS in 8m 43s\n- octavia-tox-functional-py35 http://logs.openstack.org/12/435612/105/check/octavia-tox-functional-py35/cc73246/ : SUCCESS in 6m 57s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/105/check/octavia-v1-dsvm-scenario/6d895ea/ : SUCCESS in 44m 02s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/105/check/octavia-v1-dsvm-py3x-scenario/007bb4a/ : SUCCESS in 46m 16s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/105/check/octavia-v1-dsvm-scenario-multinode/02d6152/ : SUCCESS in 51m 40s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/105/check/octavia-v1-dsvm-py3x-scenario-multinode/ea861a5/ : SUCCESS in 52m 03s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/105/check/octavia-v1-dsvm-scenario-kvm-centos.7/a55c5f1/ : SUCCESS in 49m 57s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/105/check/octavia-v1-dsvm-scenario-lxd/cfe9afa/ : FAILURE in 40m 29s (non-voting)","accounts_in_message":[],"_revision_number":105},{"id":"110933b146b92a93bb2a7e3bcf35ddadbb8c111f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-01-13 02:22:35.000000000","message":"Uploaded patch set 106.","accounts_in_message":[],"_revision_number":106},{"id":"0c95d475ff3b708f906f45198c5090b1dd62ac95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-13 03:49:16.000000000","message":"Patch Set 106: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/106/check/requirements-check/5f91ff0/ : SUCCESS in 2m 49s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/106/check/openstack-tox-pep8/27eb305/ : SUCCESS in 6m 37s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/106/check/openstack-tox-py27/0427761/ : SUCCESS in 6m 32s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/106/check/openstack-tox-py35/5285f34/ : SUCCESS in 5m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/106/check/build-openstack-sphinx-docs/14ee271/html/ : SUCCESS in 5m 54s\n- openstack-tox-cover http://logs.openstack.org/12/435612/106/check/openstack-tox-cover/b76bdbf/cover/ : SUCCESS in 9m 17s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/106/check/neutron-lbaasv2-dsvm-api/7f8b89d/ : SUCCESS in 56m 31s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/106/check/neutron-lbaasv2-dsvm-py3x-api/2dca3ca/ : SUCCESS in 1h 00m 30s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/106/check/neutron-lbaasv2-dsvm-scenario/89c90bf/ : SUCCESS in 1h 11m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/106/check/neutron-lbaasv2-dsvm-py3x-scenario/90e6bb4/ : SUCCESS in 1h 22m 57s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/12/435612/106/check/openstack-tox-functional/7c65b90/ : SUCCESS in 7m 27s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/106/check/openstack-tox-functional-py35/fcf09ec/ : SUCCESS in 7m 20s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/106/check/octavia-v1-dsvm-scenario/5bb852e/ : SUCCESS in 46m 28s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/106/check/octavia-v1-dsvm-py3x-scenario/20d5079/ : SUCCESS in 42m 18s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/106/check/octavia-v1-dsvm-scenario-multinode/2ef71a4/ : SUCCESS in 56m 38s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/106/check/octavia-v1-dsvm-py3x-scenario-multinode/4d6426d/ : SUCCESS in 51m 01s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/106/check/octavia-v1-dsvm-scenario-kvm-centos.7/0c21f17/ : SUCCESS in 44m 56s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/106/check/octavia-v1-dsvm-scenario-lxd/3d00362/ : FAILURE in 41m 05s (non-voting)","accounts_in_message":[],"_revision_number":106},{"id":"28537df0c1f1b4dda1d80733fc1081c34b1fd4f7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-02-02 20:45:16.000000000","message":"Uploaded patch set 107.","accounts_in_message":[],"_revision_number":107},{"id":"1a772b0481578fc03219b3a8c292d521d965cba3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-02-02 20:45:45.000000000","message":"Patch Set 107: Workflow-1","accounts_in_message":[],"_revision_number":107},{"id":"fa660ff115f10fc8e1b16acde8b6b2c26e162494","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-02 22:11:41.000000000","message":"Patch Set 107: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/107/check/requirements-check/8c6a4b3/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/107/check/openstack-tox-pep8/a76aefd/ : SUCCESS in 6m 50s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/107/check/openstack-tox-py27/8317feb/ : SUCCESS in 5m 50s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/107/check/openstack-tox-py35/f348863/ : SUCCESS in 5m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/107/check/build-openstack-sphinx-docs/070dee7/html/ : SUCCESS in 8m 31s\n- openstack-tox-cover http://logs.openstack.org/12/435612/107/check/openstack-tox-cover/b03d5b4/cover/ : SUCCESS in 8m 51s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/107/check/neutron-lbaasv2-dsvm-api/545b4a6/ : SUCCESS in 59m 28s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/107/check/neutron-lbaasv2-dsvm-py3x-api/69d15fe/ : SUCCESS in 56m 13s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/107/check/neutron-lbaasv2-dsvm-scenario/c6c9ef3/ : SUCCESS in 1h 16m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/107/check/neutron-lbaasv2-dsvm-py3x-scenario/f346f39/ : SUCCESS in 1h 14m 34s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/12/435612/107/check/openstack-tox-functional/fb25b08/ : SUCCESS in 11m 24s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/107/check/openstack-tox-functional-py35/02f22a3/ : SUCCESS in 7m 50s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/107/check/octavia-v1-dsvm-scenario/d19ddbc/ : SUCCESS in 41m 41s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/107/check/octavia-v1-dsvm-py3x-scenario/74322ae/ : SUCCESS in 43m 41s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/107/check/octavia-v1-dsvm-scenario-multinode/b54966c/ : SUCCESS in 57m 16s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/107/check/octavia-v1-dsvm-py3x-scenario-multinode/a21b728/ : SUCCESS in 1h 21m 03s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/107/check/octavia-v1-dsvm-scenario-kvm-centos.7/56fee45/ : SUCCESS in 1h 11m 54s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/107/check/octavia-v1-dsvm-scenario-lxd/a821231/ : FAILURE in 36m 23s (non-voting)","accounts_in_message":[],"_revision_number":107},{"id":"a1a03e46e3c8866d38adfdebf8a0e18c540e01bc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-02-06 00:22:24.000000000","message":"Uploaded patch set 108.","accounts_in_message":[],"_revision_number":108},{"id":"002010c6410999d71498f72c4be39bd632a96810","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-02-06 00:26:44.000000000","message":"Patch Set 108: Workflow-1","accounts_in_message":[],"_revision_number":108},{"id":"6ded3d158476766c8054ec6d61a6b40473d59f17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-06 02:14:38.000000000","message":"Patch Set 108: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/108/check/requirements-check/b3b9071/ : SUCCESS in 3m 06s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/108/check/openstack-tox-pep8/3d0d486/ : SUCCESS in 7m 24s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/108/check/openstack-tox-py27/d7e6333/ : SUCCESS in 5m 03s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/108/check/openstack-tox-py35/432147b/ : SUCCESS in 6m 37s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/108/check/build-openstack-sphinx-docs/d4c1c6b/html/ : SUCCESS in 5m 36s\n- openstack-tox-cover http://logs.openstack.org/12/435612/108/check/openstack-tox-cover/314526c/cover/ : SUCCESS in 9m 51s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/108/check/neutron-lbaasv2-dsvm-api/02bb020/ : FAILURE in 51m 00s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/108/check/neutron-lbaasv2-dsvm-py3x-api/4a281a8/ : SUCCESS in 54m 15s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/108/check/neutron-lbaasv2-dsvm-scenario/3245d66/ : SUCCESS in 1h 12m 06s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/108/check/neutron-lbaasv2-dsvm-py3x-scenario/4c99834/ : SUCCESS in 1h 50m 17s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/108/check/octavia-v1-dsvm-scenario/ce371b2/ : SUCCESS in 40m 10s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/108/check/octavia-v1-dsvm-py3x-scenario/34403d8/ : SUCCESS in 45m 13s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/108/check/octavia-v1-dsvm-scenario-multinode/76c7ca1/ : SUCCESS in 51m 55s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/108/check/octavia-v1-dsvm-py3x-scenario-multinode/b7ec7ba/ : SUCCESS in 1h 06m 42s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/108/check/octavia-v1-dsvm-scenario-kvm-centos.7/17ee7bc/ : FAILURE in 9m 00s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/108/check/octavia-v1-dsvm-scenario-lxd/2cc6ccd/ : FAILURE in 7m 46s (non-voting)","accounts_in_message":[],"_revision_number":108},{"id":"d91389b4aad5db17d752df5e0a49511e765f48bd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-02-15 03:19:41.000000000","message":"Uploaded patch set 109.","accounts_in_message":[],"_revision_number":109},{"id":"906b5978c3e2919e736fad85f55a1ec47168a8ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-15 04:45:28.000000000","message":"Patch Set 109: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/109/check/requirements-check/dda39fe/ : SUCCESS in 2m 39s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/109/check/openstack-tox-pep8/83f0c6f/ : SUCCESS in 6m 23s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/109/check/openstack-tox-py27/4f20691/ : SUCCESS in 4m 48s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/109/check/openstack-tox-py35/8c6901e/ : SUCCESS in 4m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/109/check/build-openstack-sphinx-docs/1e77a25/html/ : SUCCESS in 5m 44s\n- openstack-tox-cover http://logs.openstack.org/12/435612/109/check/openstack-tox-cover/4c1d31d/cover/ : SUCCESS in 9m 20s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/109/check/neutron-lbaasv2-dsvm-api/9d7a7eb/ : SUCCESS in 56m 27s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/109/check/neutron-lbaasv2-dsvm-py3x-api/99e1411/ : SUCCESS in 57m 56s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/109/check/neutron-lbaasv2-dsvm-scenario/ecb610e/ : SUCCESS in 1h 15m 05s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/109/check/neutron-lbaasv2-dsvm-py3x-scenario/1a0c9bb/ : SUCCESS in 1h 24m 13s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/109/check/octavia-v1-dsvm-scenario/3ee2254/ : SUCCESS in 47m 51s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/109/check/octavia-v1-dsvm-py3x-scenario/5f5056e/ : SUCCESS in 39m 35s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/109/check/octavia-v1-dsvm-scenario-multinode/1f11f6f/ : SUCCESS in 52m 22s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/109/check/octavia-v1-dsvm-py3x-scenario-multinode/3c63977/ : SUCCESS in 56m 11s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/109/check/octavia-v1-dsvm-scenario-kvm-centos.7/b5c0ef6/ : FAILURE in 7m 55s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/109/check/octavia-v1-dsvm-scenario-lxd/232685b/ : FAILURE in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":109},{"id":"6f19be709e09cf2057e979beba777e51a01f14c4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-02-15 10:00:15.000000000","message":"Uploaded patch set 110.","accounts_in_message":[],"_revision_number":110},{"id":"c2f883b3290605287974c9f65d3d2ff68580bbde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-15 11:28:41.000000000","message":"Patch Set 110: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/110/check/requirements-check/cf3e421/ : SUCCESS in 2m 58s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/110/check/openstack-tox-pep8/e6b1562/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/110/check/openstack-tox-py27/040364d/ : SUCCESS in 4m 28s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/110/check/openstack-tox-py35/cd8cbd4/ : SUCCESS in 5m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/110/check/build-openstack-sphinx-docs/050c4a8/html/ : SUCCESS in 5m 27s\n- openstack-tox-cover http://logs.openstack.org/12/435612/110/check/openstack-tox-cover/7ea11d4/cover/ : SUCCESS in 8m 38s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/110/check/neutron-lbaasv2-dsvm-api/32f863a/ : SUCCESS in 50m 47s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/110/check/neutron-lbaasv2-dsvm-py3x-api/ba9051e/ : SUCCESS in 50m 37s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/110/check/neutron-lbaasv2-dsvm-scenario/3e3c6a3/ : SUCCESS in 1h 19m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/110/check/neutron-lbaasv2-dsvm-py3x-scenario/263a1c0/ : SUCCESS in 1h 22m 56s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/110/check/octavia-v1-dsvm-scenario/daa5748/ : SUCCESS in 45m 19s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/110/check/octavia-v1-dsvm-py3x-scenario/975c876/ : SUCCESS in 39m 50s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/110/check/octavia-v1-dsvm-scenario-multinode/034804d/ : SUCCESS in 1h 02m 16s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/110/check/octavia-v1-dsvm-py3x-scenario-multinode/8672925/ : SUCCESS in 55m 32s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/110/check/octavia-v1-dsvm-scenario-kvm-centos.7/b691e28/ : FAILURE in 7m 17s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/110/check/octavia-v1-dsvm-scenario-lxd/66a7214/ : FAILURE in 7m 53s (non-voting)","accounts_in_message":[],"_revision_number":110},{"id":"4a7f2166a83d89004211315c3b52a4e54b7e0cd9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-02-15 23:25:46.000000000","message":"Patch Set 110: Workflow-1","accounts_in_message":[],"_revision_number":110},{"id":"03209bef5262c2491e9d419ec512dc40d91f1e65","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-02-21 01:04:36.000000000","message":"Uploaded patch set 111.","accounts_in_message":[],"_revision_number":111},{"id":"291698f2946ae2bacb9bd29441149376a89d8fa1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-02-21 01:57:29.000000000","message":"Patch Set 112: Patch Set 111 was rebased","accounts_in_message":[],"_revision_number":112},{"id":"740449f2c11e9ba9815503d526d54d58826efd95","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-02-21 01:57:49.000000000","message":"Patch Set 112: Workflow-1","accounts_in_message":[],"_revision_number":112},{"id":"8d2a83c6c65e7815e66277333b0f5dae12b72a91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-21 03:39:29.000000000","message":"Patch Set 112: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/112/check/requirements-check/984d24c/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/112/check/openstack-tox-pep8/b79d2f8/ : SUCCESS in 6m 32s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/112/check/openstack-tox-py27/71fdb0e/ : SUCCESS in 5m 03s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/112/check/openstack-tox-py35/827a866/ : SUCCESS in 5m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/112/check/build-openstack-sphinx-docs/b333064/html/ : SUCCESS in 5m 47s\n- openstack-tox-cover http://logs.openstack.org/12/435612/112/check/openstack-tox-cover/5517a97/cover/ : SUCCESS in 7m 47s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/112/check/neutron-lbaasv2-dsvm-api/51541d2/ : SUCCESS in 55m 19s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/112/check/neutron-lbaasv2-dsvm-py3x-api/f2119e1/ : SUCCESS in 53m 50s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/112/check/neutron-lbaasv2-dsvm-scenario/ad298a6/ : SUCCESS in 1h 08m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/112/check/neutron-lbaasv2-dsvm-py3x-scenario/29a993c/ : SUCCESS in 1h 34m 56s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/112/check/octavia-v1-dsvm-scenario/cb2fdef/ : SUCCESS in 39m 25s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/112/check/octavia-v1-dsvm-py3x-scenario/d74d175/ : SUCCESS in 50m 20s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/112/check/octavia-v1-dsvm-scenario-multinode/176d843/ : SUCCESS in 54m 02s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/112/check/octavia-v1-dsvm-py3x-scenario-multinode/25bb9be/ : SUCCESS in 55m 49s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/112/check/octavia-v1-dsvm-scenario-kvm-centos.7/5a6b31b/ : FAILURE in 7m 52s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/112/check/octavia-v1-dsvm-scenario-lxd/dfc5ceb/ : FAILURE in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":112},{"id":"7925ab8da5a75bfd457b634ba6dbf3c77652a662","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-02-27 23:22:29.000000000","message":"Uploaded patch set 113.","accounts_in_message":[],"_revision_number":113},{"id":"d2739d9c2e81ab054b3c7afaf91e2ff2937e88c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-28 00:46:56.000000000","message":"Patch Set 113: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/113/check/requirements-check/734e138/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/113/check/openstack-tox-pep8/0f287f5/ : SUCCESS in 6m 50s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/113/check/openstack-tox-py27/5f5a32b/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/113/check/openstack-tox-py35/b45ab18/ : SUCCESS in 4m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/113/check/build-openstack-sphinx-docs/d1724c0/html/ : SUCCESS in 5m 09s\n- openstack-tox-cover http://logs.openstack.org/12/435612/113/check/openstack-tox-cover/09c2855/cover/ : SUCCESS in 8m 35s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/113/check/neutron-lbaasv2-dsvm-api/d9d3417/ : SUCCESS in 58m 40s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/113/check/neutron-lbaasv2-dsvm-py3x-api/c02466a/ : SUCCESS in 51m 00s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/113/check/neutron-lbaasv2-dsvm-scenario/508c3a6/ : SUCCESS in 1h 19m 12s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/113/check/neutron-lbaasv2-dsvm-py3x-scenario/7d0bf65/ : SUCCESS in 1h 13m 17s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/113/check/octavia-v1-dsvm-scenario/c2c027b/ : SUCCESS in 58m 19s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/113/check/octavia-v1-dsvm-py3x-scenario/4f03253/ : SUCCESS in 53m 11s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/113/check/octavia-v1-dsvm-scenario-multinode/6595c51/ : SUCCESS in 1h 06m 26s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/113/check/octavia-v1-dsvm-py3x-scenario-multinode/b15a9ab/ : SUCCESS in 48m 13s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/113/check/octavia-v1-dsvm-scenario-kvm-centos.7/bc3ee43/ : FAILURE in 5m 27s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/113/check/octavia-v1-dsvm-scenario-lxd/2231381/ : FAILURE in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":113},{"id":"9b10e2b7fa7f2b53fd7948f00de862239387008d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-03-08 19:51:17.000000000","message":"Uploaded patch set 114.","accounts_in_message":[],"_revision_number":114},{"id":"fd1a29570e8f600b62a08d5a1b53753898d074c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-08 21:31:53.000000000","message":"Patch Set 114: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/114/check/requirements-check/e556cba/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/114/check/openstack-tox-pep8/1bf3026/ : SUCCESS in 9m 10s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/114/check/openstack-tox-py27/0c2e9b7/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/114/check/openstack-tox-py35/3b9fc1d/ : SUCCESS in 5m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/114/check/build-openstack-sphinx-docs/c145fc8/html/ : SUCCESS in 5m 11s\n- openstack-tox-cover http://logs.openstack.org/12/435612/114/check/openstack-tox-cover/3427a41/cover/ : SUCCESS in 8m 54s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/114/check/neutron-lbaasv2-dsvm-api/1bbaa50/ : SUCCESS in 51m 22s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/114/check/neutron-lbaasv2-dsvm-py3x-api/c1a0a1e/ : SUCCESS in 49m 20s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/114/check/neutron-lbaasv2-dsvm-scenario/5ebc7cc/ : FAILURE in 1h 00m 23s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/114/check/neutron-lbaasv2-dsvm-py3x-scenario/7a720f1/ : SUCCESS in 45m 48s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/114/check/octavia-v1-dsvm-scenario/2ab5300/ : SUCCESS in 46m 10s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/114/check/octavia-v1-dsvm-py3x-scenario/32cdfb8/ : SUCCESS in 48m 17s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/114/check/octavia-v1-dsvm-scenario-multinode/82193aa/ : SUCCESS in 1h 38m 08s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/114/check/octavia-v1-dsvm-py3x-scenario-multinode/48f1695/ : SUCCESS in 1h 04m 58s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/114/check/octavia-v1-dsvm-scenario-kvm-centos.7/c9c23b5/ : FAILURE in 6m 12s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/12/435612/114/check/octavia-v1-dsvm-scenario-lxd/ecfe60f/ : FAILURE in 7m 47s (non-voting)","accounts_in_message":[],"_revision_number":114},{"id":"4a25109b9ab3c02de0c9c8ac2b5976bf2feccf0e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-03-14 15:38:43.000000000","message":"Uploaded patch set 115.","accounts_in_message":[],"_revision_number":115},{"id":"5d71540546bd5e76ea802a115dde97a9de43670b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-14 16:57:25.000000000","message":"Patch Set 115: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/115/check/requirements-check/15c5e86/ : SUCCESS in 3m 22s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/115/check/openstack-tox-pep8/28aa501/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/115/check/openstack-tox-py27/41f8da0/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/115/check/openstack-tox-py35/5a7ad41/ : SUCCESS in 5m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/115/check/build-openstack-sphinx-docs/af96def/html/ : SUCCESS in 6m 11s\n- openstack-tox-cover http://logs.openstack.org/12/435612/115/check/openstack-tox-cover/06ef15f/cover/ : SUCCESS in 8m 10s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/115/check/neutron-lbaasv2-dsvm-api/f57ac84/ : FAILURE in 30m 58s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/115/check/neutron-lbaasv2-dsvm-py3x-api/baa62da/ : FAILURE in 24m 56s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/115/check/neutron-lbaasv2-dsvm-scenario/808ee81/ : FAILURE in 28m 50s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/115/check/neutron-lbaasv2-dsvm-py3x-scenario/0437d35/ : FAILURE in 31m 36s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/115/check/octavia-v1-dsvm-scenario/0add961/ : SUCCESS in 48m 37s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/115/check/octavia-v1-dsvm-py3x-scenario/c32eb68/ : SUCCESS in 49m 45s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/115/check/octavia-v1-dsvm-scenario-multinode/7c6c6d6/ : SUCCESS in 1h 10m 37s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/115/check/octavia-v1-dsvm-py3x-scenario-multinode/e76453f/ : SUCCESS in 1h 07m 37s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/115/check/octavia-v1-dsvm-scenario-kvm-centos.7/5357f4f/ : SUCCESS in 1h 00m 48s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/12/435612/115/check/openstack-tox-functional/15bb560/ : SUCCESS in 7m 05s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/115/check/openstack-tox-functional-py35/712b5d5/ : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":115},{"id":"530284d303b4187fc5228389e7ffa909e4144f04","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-03-15 06:25:00.000000000","message":"Patch Set 115: Workflow-1\n\nrecheck","accounts_in_message":[],"_revision_number":115},{"id":"e09820ee85ac41e60feede59f858ce22e6bb6b00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 07:47:03.000000000","message":"Patch Set 115: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/115/check/requirements-check/ff6cf08/ : SUCCESS in 2m 34s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/115/check/openstack-tox-pep8/747d725/ : SUCCESS in 6m 21s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/115/check/openstack-tox-py27/f33ac4d/ : SUCCESS in 4m 58s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/115/check/openstack-tox-py35/35a9e68/ : SUCCESS in 4m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/115/check/build-openstack-sphinx-docs/82523ff/html/ : SUCCESS in 5m 47s\n- openstack-tox-cover http://logs.openstack.org/12/435612/115/check/openstack-tox-cover/cd9bb48/cover/ : SUCCESS in 9m 34s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/115/check/neutron-lbaasv2-dsvm-api/5a1380a/ : SUCCESS in 1h 15m 52s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/115/check/neutron-lbaasv2-dsvm-py3x-api/c6d5015/ : SUCCESS in 1h 11m 10s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/115/check/neutron-lbaasv2-dsvm-scenario/fb7c8a5/ : SUCCESS in 1h 12m 09s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/115/check/neutron-lbaasv2-dsvm-py3x-scenario/1ba4ae2/ : SUCCESS in 1h 16m 42s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/115/check/octavia-v1-dsvm-scenario/35ed2d6/ : SUCCESS in 1h 09m 50s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/115/check/octavia-v1-dsvm-py3x-scenario/d597166/ : SUCCESS in 51m 14s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/115/check/octavia-v1-dsvm-scenario-multinode/fdf9d22/ : SUCCESS in 1h 13m 15s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/115/check/octavia-v1-dsvm-py3x-scenario-multinode/6997b74/ : SUCCESS in 1h 03m 21s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/115/check/octavia-v1-dsvm-scenario-kvm-centos.7/201bda2/ : SUCCESS in 55m 35s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/12/435612/115/check/openstack-tox-functional/a03e405/ : SUCCESS in 7m 17s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/115/check/openstack-tox-functional-py35/1f2aa9b/ : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":115},{"id":"2bb4cd6b4b5736eb915283d75dd389459db98879","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-03-19 18:12:56.000000000","message":"Patch Set 116: Patch Set 115 was rebased","accounts_in_message":[],"_revision_number":116},{"id":"95638b754236b322698adc877d5f845a2b9a2c77","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-03-19 18:13:03.000000000","message":"Patch Set 116: Workflow-1","accounts_in_message":[],"_revision_number":116},{"id":"0739702a6c222ef6426ebd09dd473c26917e9bf8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-19 20:22:02.000000000","message":"Patch Set 116: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/116/check/requirements-check/d96ad89/ : SUCCESS in 2m 53s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/116/check/openstack-tox-pep8/771790f/ : SUCCESS in 6m 37s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/116/check/openstack-tox-py27/64b4f2a/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/116/check/openstack-tox-py35/86bd3a8/ : SUCCESS in 4m 51s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/116/check/build-openstack-sphinx-docs/b1cb3ea/html/ : SUCCESS in 5m 18s\n- openstack-tox-cover http://logs.openstack.org/12/435612/116/check/openstack-tox-cover/ff9b28f/cover/ : SUCCESS in 7m 20s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/116/check/neutron-lbaasv2-dsvm-api/c273056/ : SUCCESS in 1h 14m 00s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/116/check/neutron-lbaasv2-dsvm-py3x-api/9321209/ : SUCCESS in 1h 17m 29s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/116/check/neutron-lbaasv2-dsvm-scenario/1a15905/ : SUCCESS in 1h 27m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/116/check/neutron-lbaasv2-dsvm-py3x-scenario/6461137/ : SUCCESS in 1h 42m 26s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/12/435612/116/check/openstack-tox-functional/d003eb9/ : SUCCESS in 6m 24s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/116/check/openstack-tox-functional-py35/4fdd98f/ : SUCCESS in 7m 42s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/116/check/octavia-v1-dsvm-scenario/0d73ee0/ : SUCCESS in 1h 04m 33s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/116/check/octavia-v1-dsvm-py3x-scenario/0f76753/ : SUCCESS in 1h 05m 07s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/116/check/octavia-v1-dsvm-scenario-multinode/192ef23/ : FAILURE in 1h 06m 45s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/116/check/octavia-v1-dsvm-py3x-scenario-multinode/5def564/ : FAILURE in 1h 14m 30s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/116/check/octavia-v1-dsvm-scenario-kvm-centos.7/322da05/ : SUCCESS in 1h 03m 17s (non-voting)","accounts_in_message":[],"_revision_number":116},{"id":"d39fdda17cfa0ae17cb3a61b7e0a85aa51060cc3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-03-27 05:50:34.000000000","message":"Patch Set 117: Patch Set 116 was rebased","accounts_in_message":[],"_revision_number":117},{"id":"ed5d5c5a72c9e9b4b3e01c264dbe4741475a0f92","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-03-27 05:51:01.000000000","message":"Patch Set 117: Workflow-1","accounts_in_message":[],"_revision_number":117},{"id":"e4caffe25f28711d90b37213507bb8061f82cde1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-27 07:26:32.000000000","message":"Patch Set 117: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/117/check/requirements-check/5eee7c7/ : SUCCESS in 3m 03s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/117/check/openstack-tox-pep8/9900a8a/ : SUCCESS in 6m 24s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/117/check/openstack-tox-py27/7afe2b3/ : SUCCESS in 5m 01s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/117/check/openstack-tox-py35/63e0e43/ : SUCCESS in 5m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/117/check/build-openstack-sphinx-docs/5f00208/html/ : SUCCESS in 5m 29s\n- openstack-tox-cover http://logs.openstack.org/12/435612/117/check/openstack-tox-cover/f5f135e/cover/ : SUCCESS in 7m 12s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/117/check/neutron-lbaasv2-dsvm-api/9134bb2/ : SUCCESS in 1h 11m 25s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/117/check/neutron-lbaasv2-dsvm-py3x-api/2bdce2c/ : SUCCESS in 1h 11m 27s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/117/check/neutron-lbaasv2-dsvm-scenario/6073381/ : SUCCESS in 1h 08m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/117/check/neutron-lbaasv2-dsvm-py3x-scenario/3535663/ : SUCCESS in 1h 27m 51s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/12/435612/117/check/openstack-tox-functional/e4958f4/ : SUCCESS in 6m 13s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/117/check/openstack-tox-functional-py35/25dfea7/ : SUCCESS in 6m 09s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/117/check/octavia-v1-dsvm-scenario/e096030/ : SUCCESS in 53m 53s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/117/check/octavia-v1-dsvm-py3x-scenario/bdb7462/ : SUCCESS in 48m 47s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/117/check/octavia-v1-dsvm-scenario-multinode/4f97bac/ : SUCCESS in 1h 10m 38s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/117/check/octavia-v1-dsvm-py3x-scenario-multinode/5fe3251/ : SUCCESS in 1h 12m 24s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/117/check/octavia-v1-dsvm-scenario-kvm-centos.7/ed9a7d6/ : SUCCESS in 54m 27s (non-voting)","accounts_in_message":[],"_revision_number":117},{"id":"7cb1ad28b4b32f1da49ef4edb890e29c5ec1d6a5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-04 22:42:26.000000000","message":"Patch Set 118: Patch Set 117 was rebased","accounts_in_message":[],"_revision_number":118},{"id":"48f3d9c4db07464fe70c212c80a8f1fd7078027b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-05 00:29:04.000000000","message":"Patch Set 118: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/118/check/requirements-check/9109475/ : FAILURE in 3m 02s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/118/check/openstack-tox-pep8/4049211/ : SUCCESS in 6m 09s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/118/check/openstack-tox-py27/ed513bc/ : SUCCESS in 4m 38s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/118/check/openstack-tox-py35/9a7f0d7/ : SUCCESS in 5m 14s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/118/check/build-openstack-sphinx-docs/d380160/html/ : SUCCESS in 5m 39s\n- openstack-tox-cover http://logs.openstack.org/12/435612/118/check/openstack-tox-cover/f140807/cover/ : SUCCESS in 6m 56s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/118/check/neutron-lbaasv2-dsvm-api/57b6676/ : SUCCESS in 1h 06m 58s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/118/check/neutron-lbaasv2-dsvm-py3x-api/c034973/ : SUCCESS in 1h 08m 56s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/118/check/neutron-lbaasv2-dsvm-scenario/b7b8b17/ : SUCCESS in 1h 40m 04s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/118/check/neutron-lbaasv2-dsvm-py3x-scenario/5b086ab/ : SUCCESS in 1h 19m 31s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/118/check/octavia-v1-dsvm-scenario/ae98605/ : SUCCESS in 50m 42s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/118/check/octavia-v1-dsvm-py3x-scenario/f12b946/ : SUCCESS in 47m 21s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/118/check/octavia-v1-dsvm-scenario-multinode/77ad869/ : FAILURE in 1h 02m 26s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/118/check/octavia-v1-dsvm-py3x-scenario-multinode/c2bc20b/ : FAILURE in 1h 02m 48s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/118/check/octavia-v1-dsvm-scenario-kvm-centos.7/efc7193/ : SUCCESS in 53m 16s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/118/check/openstack-tox-lower-constraints/5f7fde7/ : SUCCESS in 6m 35s\n- openstack-tox-functional http://logs.openstack.org/12/435612/118/check/openstack-tox-functional/2b9aed3/ : SUCCESS in 6m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/118/check/openstack-tox-functional-py35/2c81c46/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":118},{"id":"ea6ccbe8cc4f99659f967394662980e362b09ecc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-06 00:37:00.000000000","message":"Uploaded patch set 119.","accounts_in_message":[],"_revision_number":119},{"id":"a7c8487f2d2907d45fd9f58327d51ff86cca8039","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-06 02:09:51.000000000","message":"Patch Set 119: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/119/check/requirements-check/2213975/ : FAILURE in 2m 47s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/119/check/openstack-tox-pep8/85202b9/ : SUCCESS in 6m 08s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/119/check/openstack-tox-py27/4fdbce5/ : FAILURE in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/119/check/openstack-tox-py35/d0b607d/ : FAILURE in 4m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/119/check/build-openstack-sphinx-docs/6ac9508/html/ : SUCCESS in 5m 58s\n- openstack-tox-cover http://logs.openstack.org/12/435612/119/check/openstack-tox-cover/713abd9/ : FAILURE in 5m 19s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/119/check/neutron-lbaasv2-dsvm-api/3837d7a/ : SUCCESS in 1h 11m 10s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/119/check/neutron-lbaasv2-dsvm-py3x-api/371421d/ : SUCCESS in 1h 13m 24s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/119/check/neutron-lbaasv2-dsvm-scenario/05eba31/ : SUCCESS in 1h 26m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/119/check/neutron-lbaasv2-dsvm-py3x-scenario/a904503/ : SUCCESS in 1h 10m 09s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/119/check/octavia-v1-dsvm-scenario/e4b26eb/ : SUCCESS in 49m 21s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/119/check/octavia-v1-dsvm-py3x-scenario/aad0db8/ : SUCCESS in 49m 59s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/119/check/octavia-v1-dsvm-scenario-multinode/97bb2a5/ : SUCCESS in 1h 16m 59s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/119/check/octavia-v1-dsvm-py3x-scenario-multinode/9ae6aff/ : FAILURE in 58m 27s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/119/check/octavia-v1-dsvm-scenario-kvm-centos.7/21b8b21/ : SUCCESS in 49m 31s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/119/check/openstack-tox-lower-constraints/6256a7f/ : FAILURE in 5m 07s\n- openstack-tox-functional http://logs.openstack.org/12/435612/119/check/openstack-tox-functional/dee961e/ : FAILURE in 5m 52s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/119/check/openstack-tox-functional-py35/4c8cb97/ : FAILURE in 6m 15s","accounts_in_message":[],"_revision_number":119},{"id":"4457727524dcd5e698dff87718c508dbf69246a7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-06 23:24:26.000000000","message":"Uploaded patch set 120.","accounts_in_message":[],"_revision_number":120},{"id":"a124a496e4b7e06045674e3130694cb42089639b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-07 00:55:48.000000000","message":"Patch Set 120: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/120/check/requirements-check/d792ef8/ : FAILURE in 3m 04s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/120/check/openstack-tox-pep8/81edd3c/ : SUCCESS in 6m 55s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/120/check/openstack-tox-py27/afba29f/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/120/check/openstack-tox-py35/8210ffc/ : SUCCESS in 4m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/120/check/build-openstack-sphinx-docs/e22030d/html/ : SUCCESS in 5m 31s\n- openstack-tox-cover http://logs.openstack.org/12/435612/120/check/openstack-tox-cover/6b7c8fd/cover/ : SUCCESS in 7m 32s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/120/check/neutron-lbaasv2-dsvm-api/1741dea/ : SUCCESS in 1h 20m 42s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/120/check/neutron-lbaasv2-dsvm-py3x-api/9200137/ : SUCCESS in 1h 25m 52s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/120/check/neutron-lbaasv2-dsvm-scenario/cb5978b/ : SUCCESS in 1h 18m 58s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/120/check/neutron-lbaasv2-dsvm-py3x-scenario/cda3081/ : SUCCESS in 1h 12m 15s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/120/check/octavia-v1-dsvm-scenario/ca105ab/ : FAILURE in 27m 24s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/120/check/octavia-v1-dsvm-py3x-scenario/6abfaeb/ : FAILURE in 27m 38s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/120/check/octavia-v1-dsvm-scenario-multinode/c69500f/ : FAILURE in 28m 38s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/120/check/octavia-v1-dsvm-py3x-scenario-multinode/a8a7a4b/ : FAILURE in 28m 46s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/120/check/octavia-v1-dsvm-scenario-kvm-centos.7/38429fa/ : SUCCESS in 57m 46s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/120/check/openstack-tox-lower-constraints/d9c451a/ : SUCCESS in 6m 49s\n- openstack-tox-functional http://logs.openstack.org/12/435612/120/check/openstack-tox-functional/7897aad/ : SUCCESS in 7m 01s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/120/check/openstack-tox-functional-py35/c82967f/ : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":120},{"id":"859154d9861b9a63ee9e27378e5c2100cfdd9711","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-09 21:30:49.000000000","message":"Uploaded patch set 121.","accounts_in_message":[],"_revision_number":121},{"id":"cec20709fdfdb4fff6c952c3aceaae2b116314e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-09 22:51:14.000000000","message":"Patch Set 121: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/121/check/requirements-check/3d6a411/ : FAILURE in 2m 56s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/121/check/openstack-tox-pep8/2b1e38a/ : SUCCESS in 6m 26s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/121/check/openstack-tox-py27/9cd13a4/ : SUCCESS in 4m 41s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/121/check/openstack-tox-py35/cde4077/ : SUCCESS in 5m 09s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/121/check/build-openstack-sphinx-docs/464edea/html/ : SUCCESS in 5m 48s\n- openstack-tox-cover http://logs.openstack.org/12/435612/121/check/openstack-tox-cover/6f9ac82/cover/ : SUCCESS in 7m 06s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/121/check/neutron-lbaasv2-dsvm-api/bfe4839/ : SUCCESS in 1h 14m 15s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/121/check/neutron-lbaasv2-dsvm-scenario/e59807e/ : SUCCESS in 50m 01s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/121/check/octavia-v1-dsvm-scenario/55d2531/ : SUCCESS in 1h 03m 06s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/121/check/octavia-v1-dsvm-scenario-multinode/642fd7d/ : FAILURE in 58m 01s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/121/check/octavia-v1-dsvm-scenario-kvm-centos.7/b62fe20/ : SUCCESS in 1h 08m 07s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/121/check/neutron-lbaasv2-dsvm-py3x-api/810e681/ : SUCCESS in 1h 13m 00s\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/121/check/neutron-lbaasv2-dsvm-py3x-scenario/442b7a7/ : SUCCESS in 1h 09m 41s (non-voting)\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/121/check/octavia-v1-dsvm-py3x-scenario/461bbcd/ : SUCCESS in 42m 47s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/121/check/octavia-v1-dsvm-py3x-scenario-multinode/0fa7981/ : SUCCESS in 1h 06m 27s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/121/check/openstack-tox-lower-constraints/c7ab917/ : SUCCESS in 6m 54s\n- openstack-tox-functional http://logs.openstack.org/12/435612/121/check/openstack-tox-functional/b69eac2/ : SUCCESS in 6m 02s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/121/check/openstack-tox-functional-py35/8d63c88/ : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":121},{"id":"5af7e3f06662218923918db910695635e58a818d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-10 01:02:36.000000000","message":"Patch Set 121:\n\n(2 comments)","accounts_in_message":[],"_revision_number":121},{"id":"3ec480c9750e9ba657747ddaf619fd6bba5adb71","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-10 01:10:36.000000000","message":"Uploaded patch set 122.","accounts_in_message":[],"_revision_number":122},{"id":"8bb7482429e16b3b49b84154af0fa2586df20df4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-10 02:42:49.000000000","message":"Patch Set 122: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/122/check/requirements-check/7544329/ : FAILURE in 3m 15s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/122/check/openstack-tox-pep8/b52eeda/ : SUCCESS in 6m 26s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/122/check/openstack-tox-py27/fd5a266/ : SUCCESS in 5m 06s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/122/check/openstack-tox-py35/4514dc4/ : SUCCESS in 4m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/122/check/build-openstack-sphinx-docs/423ba7c/html/ : SUCCESS in 6m 25s\n- openstack-tox-cover http://logs.openstack.org/12/435612/122/check/openstack-tox-cover/73dd1cf/cover/ : SUCCESS in 6m 44s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/122/check/neutron-lbaasv2-dsvm-api/3940c93/ : SUCCESS in 1h 10m 49s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/122/check/neutron-lbaasv2-dsvm-scenario/4cf2d83/ : SUCCESS in 1h 01m 04s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/122/check/octavia-v1-dsvm-scenario/eb6334d/ : SUCCESS in 55m 54s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/122/check/octavia-v1-dsvm-scenario-multinode/15e9d65/ : FAILURE in 1h 11m 03s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/122/check/octavia-v1-dsvm-scenario-kvm-centos.7/e76820a/ : SUCCESS in 48m 44s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/122/check/neutron-lbaasv2-dsvm-py3x-api/85cbc80/ : SUCCESS in 1h 19m 23s\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/122/check/neutron-lbaasv2-dsvm-py3x-scenario/5d43035/ : SUCCESS in 1h 24m 04s (non-voting)\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/122/check/octavia-v1-dsvm-py3x-scenario/e340f7b/ : SUCCESS in 55m 34s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/122/check/octavia-v1-dsvm-py3x-scenario-multinode/efe2efa/ : SUCCESS in 57m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/122/check/openstack-tox-lower-constraints/87bdcd4/ : SUCCESS in 6m 16s\n- openstack-tox-functional http://logs.openstack.org/12/435612/122/check/openstack-tox-functional/a68aafd/ : SUCCESS in 5m 56s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/122/check/openstack-tox-functional-py35/35ddc04/ : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":122},{"id":"c7ebe070147d73b7c3eca0e5d7f84fc6ffb3c274","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-24 01:15:59.000000000","message":"Uploaded patch set 123.","accounts_in_message":[],"_revision_number":123},{"id":"145e4ea811a198389d54047952c235aaee5b9d24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-24 02:39:13.000000000","message":"Patch Set 123: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/123/check/requirements-check/f6e8bc0/ : FAILURE in 3m 00s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/123/check/openstack-tox-pep8/2251281/ : SUCCESS in 8m 27s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/123/check/openstack-tox-py27/1bb1f7f/ : SUCCESS in 5m 32s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/123/check/openstack-tox-py35/82e39d7/ : SUCCESS in 5m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/123/check/build-openstack-sphinx-docs/a686eb0/html/ : SUCCESS in 5m 17s\n- openstack-tox-cover http://logs.openstack.org/12/435612/123/check/openstack-tox-cover/774da61/cover/ : SUCCESS in 6m 42s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/123/check/neutron-lbaasv2-dsvm-api/ef6117f/ : SUCCESS in 1h 09m 17s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/123/check/neutron-lbaasv2-dsvm-py3x-api/76f2b37/ : SUCCESS in 1h 17m 48s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/123/check/neutron-lbaasv2-dsvm-scenario/10b41ca/ : SUCCESS in 1h 08m 56s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/123/check/neutron-lbaasv2-dsvm-py3x-scenario/faac2f7/ : SUCCESS in 1h 11m 07s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/123/check/openstack-tox-lower-constraints/8de1e23/ : SUCCESS in 7m 03s\n- openstack-tox-functional http://logs.openstack.org/12/435612/123/check/openstack-tox-functional/74cb789/ : SUCCESS in 6m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/123/check/openstack-tox-functional-py35/c4321ec/ : SUCCESS in 6m 20s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/123/check/octavia-v1-dsvm-scenario/47d9420/ : SUCCESS in 50m 29s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/123/check/octavia-v1-dsvm-py3x-scenario/9a1dc58/ : SUCCESS in 49m 43s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/123/check/octavia-v1-dsvm-scenario-multinode/a6ce5a9/ : SUCCESS in 1h 14m 41s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/123/check/octavia-v1-dsvm-py3x-scenario-multinode/2478c88/ : FAILURE in 1h 00m 58s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/123/check/octavia-v1-dsvm-scenario-kvm-centos.7/2ff4eff/ : SUCCESS in 54m 06s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/123/check/octavia-v1-dsvm-scenario-ubuntu.bionic/d889f19/ : SUCCESS in 57m 40s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/123/check/octavia-v2-dsvm-noop-api/7fad712/ : SUCCESS in 32m 12s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/123/check/octavia-v2-dsvm-noop-py35-api/06d592b/ : SUCCESS in 27m 39s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/123/check/octavia-v2-dsvm-scenario/76ad70a/ : SUCCESS in 47m 36s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/123/check/octavia-v2-dsvm-py35-scenario/3ddb45f/ : SUCCESS in 53m 08s (non-voting)","accounts_in_message":[],"_revision_number":123},{"id":"68e951b01723a77e34a3ab106dc726f630c36690","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-26 04:57:15.000000000","message":"Uploaded patch set 124.","accounts_in_message":[],"_revision_number":124},{"id":"86c5a7b22b859bbdefaaef51ac013a10c45460d8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-26 05:12:18.000000000","message":"Uploaded patch set 125.","accounts_in_message":[],"_revision_number":125},{"id":"12c11c0844fb8ba19021c91c2357d81fc787aaa5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-26 06:54:02.000000000","message":"Patch Set 125: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/125/check/requirements-check/a681b92/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/125/check/openstack-tox-pep8/1779414/ : SUCCESS in 6m 29s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/125/check/openstack-tox-py27/d188a01/ : SUCCESS in 5m 39s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/125/check/openstack-tox-py35/ec58930/ : SUCCESS in 5m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/125/check/build-openstack-sphinx-docs/d637492/html/ : SUCCESS in 6m 31s\n- openstack-tox-cover http://logs.openstack.org/12/435612/125/check/openstack-tox-cover/cee81a5/cover/ : SUCCESS in 7m 49s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/125/check/neutron-lbaasv2-dsvm-api/8ba1cd6/ : SUCCESS in 1h 07m 41s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/125/check/neutron-lbaasv2-dsvm-py3x-api/0088335/ : SUCCESS in 1h 11m 37s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/125/check/neutron-lbaasv2-dsvm-scenario/6dee4fd/ : SUCCESS in 1h 08m 47s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/125/check/neutron-lbaasv2-dsvm-py3x-scenario/1b33de6/ : SUCCESS in 1h 21m 17s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/125/check/octavia-v1-dsvm-scenario/1c422cf/ : SUCCESS in 56m 52s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/125/check/octavia-v1-dsvm-py3x-scenario/2968112/ : SUCCESS in 50m 47s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/125/check/octavia-v1-dsvm-scenario-multinode/0d5ffbf/ : SUCCESS in 1h 05m 30s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/125/check/octavia-v1-dsvm-py3x-scenario-multinode/f79862c/ : SUCCESS in 1h 34m 03s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/125/check/octavia-v1-dsvm-scenario-kvm-centos.7/15056a0/ : SUCCESS in 1h 06m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/125/check/openstack-tox-lower-constraints/4089c45/ : SUCCESS in 6m 16s\n- openstack-tox-functional http://logs.openstack.org/12/435612/125/check/openstack-tox-functional/e2d5dfe/ : SUCCESS in 6m 12s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/125/check/openstack-tox-functional-py35/a44f247/ : SUCCESS in 5m 57s\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/125/check/octavia-v1-dsvm-scenario-ubuntu.bionic/9148393/ : SUCCESS in 1h 00m 18s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/125/check/octavia-v2-dsvm-noop-api/2f8b5d1/ : FAILURE in 26m 24s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/125/check/octavia-v2-dsvm-noop-py35-api/33a2097/ : FAILURE in 31m 37s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/125/check/octavia-v2-dsvm-scenario/8f4685d/ : FAILURE in 49m 09s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/125/check/octavia-v2-dsvm-py35-scenario/8a8aab6/ : FAILURE in 41m 54s (non-voting)","accounts_in_message":[],"_revision_number":125},{"id":"410be1c976840c39f6ba23b4633743bb825e4119","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-26 20:37:25.000000000","message":"Uploaded patch set 126.","accounts_in_message":[],"_revision_number":126},{"id":"8ffd32d0189ab22b808a39799730ad3ed6041883","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-26 22:40:05.000000000","message":"Uploaded patch set 127.","accounts_in_message":[],"_revision_number":127},{"id":"1a9474592b9b0a9b00b26aab7ae51c27d10610db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 00:07:26.000000000","message":"Patch Set 127: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/127/check/requirements-check/db69f6a/ : SUCCESS in 3m 10s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/127/check/openstack-tox-pep8/10d2a42/ : SUCCESS in 7m 49s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/127/check/openstack-tox-py27/b5b0590/ : FAILURE in 5m 26s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/127/check/openstack-tox-py35/6e31630/ : FAILURE in 5m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/127/check/build-openstack-sphinx-docs/3f7c3b4/html/ : SUCCESS in 5m 44s\n- openstack-tox-cover http://logs.openstack.org/12/435612/127/check/openstack-tox-cover/3d8ede0/ : FAILURE in 5m 20s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/127/check/neutron-lbaasv2-dsvm-api/85fc2a3/ : SUCCESS in 1h 07m 44s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/127/check/neutron-lbaasv2-dsvm-py3x-api/75a2ce2/ : SUCCESS in 1h 16m 07s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/127/check/neutron-lbaasv2-dsvm-scenario/e81c562/ : SUCCESS in 1h 15m 40s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/127/check/neutron-lbaasv2-dsvm-py3x-scenario/6be12ba/ : SUCCESS in 1h 20m 05s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/127/check/octavia-v1-dsvm-scenario/475bb13/ : SUCCESS in 1h 11m 34s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/127/check/octavia-v1-dsvm-py3x-scenario/afb9d67/ : SUCCESS in 47m 09s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/127/check/octavia-v1-dsvm-scenario-multinode/330e485/ : SUCCESS in 57m 53s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/127/check/octavia-v1-dsvm-py3x-scenario-multinode/9736977/ : SUCCESS in 1h 16m 02s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/127/check/octavia-v1-dsvm-scenario-kvm-centos.7/ab0770a/ : SUCCESS in 1h 03m 28s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/127/check/openstack-tox-lower-constraints/12d9084/ : FAILURE in 5m 09s\n- openstack-tox-functional http://logs.openstack.org/12/435612/127/check/openstack-tox-functional/7db99ab/ : SUCCESS in 6m 09s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/127/check/openstack-tox-functional-py35/36ffe8c/ : SUCCESS in 6m 02s\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/127/check/octavia-v1-dsvm-scenario-ubuntu.bionic/6063962/ : SUCCESS in 57m 53s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/127/check/octavia-v2-dsvm-noop-api/e196906/ : FAILURE in 34m 16s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/127/check/octavia-v2-dsvm-noop-py35-api/5614b4b/ : FAILURE in 26m 51s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/127/check/octavia-v2-dsvm-scenario/bad442b/ : FAILURE in 47m 38s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/127/check/octavia-v2-dsvm-py35-scenario/e874af2/ : FAILURE in 42m 13s (non-voting)","accounts_in_message":[],"_revision_number":127},{"id":"7773387bc6b33272b83a7b7e220e61c84268657c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-27 20:29:43.000000000","message":"Patch Set 128: Patch Set 127 was rebased","accounts_in_message":[],"_revision_number":128},{"id":"9513fdeaab071de09b48e528908477065f86ac85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 22:16:50.000000000","message":"Patch Set 128: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/128/check/requirements-check/fa0b8ad/ : SUCCESS in 2m 59s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/128/check/openstack-tox-pep8/bb4d94d/ : SUCCESS in 10m 43s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/128/check/openstack-tox-py27/2335755/ : SUCCESS in 5m 31s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/128/check/openstack-tox-py35/b102a6f/ : SUCCESS in 6m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/128/check/build-openstack-sphinx-docs/12f9ab3/html/ : SUCCESS in 6m 11s\n- openstack-tox-cover http://logs.openstack.org/12/435612/128/check/openstack-tox-cover/f252d3f/cover/ : SUCCESS in 7m 24s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/128/check/neutron-lbaasv2-dsvm-api/ff71041/ : SUCCESS in 1h 14m 02s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/128/check/neutron-lbaasv2-dsvm-py3x-api/9236426/ : SUCCESS in 1h 12m 11s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/128/check/neutron-lbaasv2-dsvm-scenario/8318c0a/ : SUCCESS in 1h 11m 10s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/128/check/neutron-lbaasv2-dsvm-py3x-scenario/9541354/ : SUCCESS in 1h 23m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/128/check/openstack-tox-lower-constraints/5f13b77/ : SUCCESS in 7m 19s\n- openstack-tox-functional http://logs.openstack.org/12/435612/128/check/openstack-tox-functional/1a5455d/ : SUCCESS in 6m 24s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/128/check/openstack-tox-functional-py35/8b2d844/ : SUCCESS in 7m 38s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/128/check/octavia-v1-dsvm-scenario/14895a0/ : SUCCESS in 53m 40s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/128/check/octavia-v1-dsvm-py3x-scenario/322599a/ : SUCCESS in 58m 39s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/128/check/octavia-v1-dsvm-scenario-multinode/bd413b3/ : SUCCESS in 1h 05m 58s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/128/check/octavia-v1-dsvm-py3x-scenario-multinode/661c031/ : SUCCESS in 1h 20m 13s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/128/check/octavia-v1-dsvm-scenario-kvm-centos.7/9dfc4ac/ : SUCCESS in 59m 45s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/128/check/octavia-v1-dsvm-scenario-ubuntu.bionic/ee6b362/ : SUCCESS in 50m 25s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/128/check/octavia-v2-dsvm-noop-api/e5c73dd/ : FAILURE in 27m 19s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/128/check/octavia-v2-dsvm-noop-py35-api/9b5502f/ : FAILURE in 32m 27s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/128/check/octavia-v2-dsvm-scenario/eb3d14c/ : FAILURE in 46m 02s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/128/check/octavia-v2-dsvm-py35-scenario/027dfdc/ : FAILURE in 44m 04s (non-voting)","accounts_in_message":[],"_revision_number":128},{"id":"b8a6db12a308c54e06b323d84ceca340983757f9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-05-07 18:58:53.000000000","message":"Patch Set 129: Patch Set 128 was rebased","accounts_in_message":[],"_revision_number":129},{"id":"62e1292fe43e834840076ff8f4ce768da393d291","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-07 20:16:47.000000000","message":"Patch Set 129: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/129/check/requirements-check/e52177e/ : SUCCESS in 3m 08s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/129/check/openstack-tox-pep8/5a9148e/ : SUCCESS in 6m 24s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/129/check/openstack-tox-py27/cdb7ec4/ : SUCCESS in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/129/check/openstack-tox-py35/63763e8/ : SUCCESS in 4m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/129/check/build-openstack-sphinx-docs/99f5e34/html/ : SUCCESS in 5m 58s\n- openstack-tox-cover http://logs.openstack.org/12/435612/129/check/openstack-tox-cover/bad1742/cover/ : SUCCESS in 7m 25s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/129/check/neutron-lbaasv2-dsvm-api/0a1ccb5/ : FAILURE in 9m 10s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/129/check/neutron-lbaasv2-dsvm-py3x-api/793a843/ : FAILURE in 7m 38s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/129/check/neutron-lbaasv2-dsvm-scenario/d23d598/ : FAILURE in 10m 08s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/129/check/neutron-lbaasv2-dsvm-py3x-scenario/2e88ce7/ : FAILURE in 7m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/129/check/openstack-tox-lower-constraints/eb4f32b/ : SUCCESS in 8m 38s\n- openstack-tox-functional http://logs.openstack.org/12/435612/129/check/openstack-tox-functional/a79b283/ : SUCCESS in 5m 40s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/129/check/openstack-tox-functional-py35/16fc62c/ : SUCCESS in 7m 04s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/129/check/octavia-v1-dsvm-scenario/9dc7818/ : FAILURE in 7m 32s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/129/check/octavia-v1-dsvm-py3x-scenario/fb989e8/ : FAILURE in 7m 38s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/129/check/octavia-v1-dsvm-scenario-multinode/6f59b65/ : FAILURE in 11m 39s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/129/check/octavia-v1-dsvm-py3x-scenario-multinode/755fe38/ : FAILURE in 8m 28s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/129/check/octavia-v1-dsvm-scenario-kvm-centos.7/c0d79d6/ : FAILURE in 8m 52s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/129/check/octavia-v1-dsvm-scenario-ubuntu.bionic/938a4b6/ : SUCCESS in 50m 09s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/129/check/octavia-v2-dsvm-noop-api/008a56d/ : SUCCESS in 28m 10s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/129/check/octavia-v2-dsvm-noop-py35-api/9a44ca4/ : SUCCESS in 27m 51s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/129/check/octavia-v2-dsvm-scenario/d703830/ : SUCCESS in 51m 48s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/129/check/octavia-v2-dsvm-py35-scenario/2e16131/ : SUCCESS in 52m 08s (non-voting)","accounts_in_message":[],"_revision_number":129},{"id":"e911313c06c88b3535500a8d881c8918b2fcee1c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-05-07 22:59:58.000000000","message":"Patch Set 129:\n\nrecheck","accounts_in_message":[],"_revision_number":129},{"id":"f3bfd3040ce405a13f89ce1776d39ed4ce08ed89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-08 00:18:54.000000000","message":"Patch Set 129:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/129/check/requirements-check/c267807/ : SUCCESS in 3m 34s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/129/check/openstack-tox-pep8/d07991c/ : SUCCESS in 7m 43s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/129/check/openstack-tox-py27/433bf6b/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/129/check/openstack-tox-py35/dbe3595/ : SUCCESS in 5m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/129/check/build-openstack-sphinx-docs/bd85219/html/ : SUCCESS in 5m 03s\n- openstack-tox-cover http://logs.openstack.org/12/435612/129/check/openstack-tox-cover/a93ccdd/cover/ : SUCCESS in 7m 33s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/129/check/neutron-lbaasv2-dsvm-api/04ce711/ : SUCCESS in 1h 08m 49s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/129/check/neutron-lbaasv2-dsvm-py3x-api/f0a9d7c/ : SUCCESS in 1h 15m 01s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/129/check/neutron-lbaasv2-dsvm-scenario/511b23d/ : SUCCESS in 1h 12m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/129/check/neutron-lbaasv2-dsvm-py3x-scenario/9c9d194/ : SUCCESS in 1h 05m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/129/check/openstack-tox-lower-constraints/4e8f0cf/ : SUCCESS in 7m 37s\n- openstack-tox-functional http://logs.openstack.org/12/435612/129/check/openstack-tox-functional/f277b85/ : SUCCESS in 5m 46s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/129/check/openstack-tox-functional-py35/b41c295/ : SUCCESS in 6m 51s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/129/check/octavia-v1-dsvm-scenario/8c8a668/ : SUCCESS in 49m 56s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/129/check/octavia-v1-dsvm-py3x-scenario/9b36c4b/ : FAILURE in 27m 21s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/129/check/octavia-v1-dsvm-scenario-multinode/698adaf/ : FAILURE in 56m 46s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/129/check/octavia-v1-dsvm-py3x-scenario-multinode/fd20683/ : FAILURE in 58m 41s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/129/check/octavia-v1-dsvm-scenario-kvm-centos.7/8450275/ : SUCCESS in 1h 05m 55s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/129/check/octavia-v1-dsvm-scenario-ubuntu.bionic/575230d/ : SUCCESS in 1h 00m 11s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/129/check/octavia-v2-dsvm-noop-api/6f3146d/ : SUCCESS in 30m 31s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/129/check/octavia-v2-dsvm-noop-py35-api/1f36ec0/ : SUCCESS in 35m 38s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/129/check/octavia-v2-dsvm-scenario/95497e6/ : SUCCESS in 41m 11s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/129/check/octavia-v2-dsvm-py35-scenario/03c1a1c/ : SUCCESS in 47m 49s (non-voting)","accounts_in_message":[],"_revision_number":129},{"id":"08ac12b09235d3f8b946c1a0a78ae006b70cf8ed","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-05-09 01:54:44.000000000","message":"Uploaded patch set 130.","accounts_in_message":[],"_revision_number":130},{"id":"0117058ec57de3688f6679827c0e2c86727af6c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-09 03:32:14.000000000","message":"Patch Set 130: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/130/check/requirements-check/185aca1/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/130/check/openstack-tox-pep8/4a8798f/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/130/check/openstack-tox-py27/4e9ea91/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/130/check/openstack-tox-py35/fb8fa11/ : SUCCESS in 5m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/130/check/build-openstack-sphinx-docs/67c28ba/html/ : SUCCESS in 5m 49s\n- openstack-tox-cover http://logs.openstack.org/12/435612/130/check/openstack-tox-cover/e8820dc/cover/ : SUCCESS in 7m 02s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/130/check/neutron-lbaasv2-dsvm-api/3feeefd/ : SUCCESS in 1h 15m 53s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/130/check/neutron-lbaasv2-dsvm-py3x-api/2181b4b/ : SUCCESS in 1h 10m 32s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/130/check/neutron-lbaasv2-dsvm-scenario/9ef2569/ : SUCCESS in 1h 31m 33s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/130/check/neutron-lbaasv2-dsvm-py3x-scenario/80f61a1/ : SUCCESS in 1h 10m 16s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/130/check/openstack-tox-lower-constraints/2720ab9/ : SUCCESS in 6m 27s\n- openstack-tox-functional http://logs.openstack.org/12/435612/130/check/openstack-tox-functional/14bbe47/ : SUCCESS in 6m 01s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/130/check/openstack-tox-functional-py35/8065ed5/ : SUCCESS in 6m 51s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/130/check/octavia-v1-dsvm-scenario/395ea00/ : SUCCESS in 1h 01m 02s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/130/check/octavia-v1-dsvm-py3x-scenario/32bfb18/ : SUCCESS in 51m 37s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/130/check/octavia-v1-dsvm-scenario-multinode/4bb1896/ : SUCCESS in 1h 08m 01s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/130/check/octavia-v1-dsvm-py3x-scenario-multinode/cd68da3/ : FAILURE in 1h 04m 04s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/130/check/octavia-v1-dsvm-scenario-kvm-centos.7/87e2755/ : SUCCESS in 1h 06m 18s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/130/check/octavia-v1-dsvm-scenario-ubuntu.bionic/468ba40/ : SUCCESS in 51m 07s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/130/check/octavia-v2-dsvm-noop-api/6d721e8/ : SUCCESS in 29m 03s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/130/check/octavia-v2-dsvm-noop-py35-api/a94a728/ : SUCCESS in 40m 14s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/130/check/octavia-v2-dsvm-scenario/cd600d1/ : SUCCESS in 43m 45s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/130/check/octavia-v2-dsvm-py35-scenario/1cfeafc/ : SUCCESS in 45m 35s (non-voting)","accounts_in_message":[],"_revision_number":130},{"id":"f3b892ae1f8bb72f1c82937fd55f8e11312a81d8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-05-09 18:15:49.000000000","message":"Uploaded patch set 131.","accounts_in_message":[],"_revision_number":131},{"id":"6dbb6da1f00a013d8a0d5123084658f579f19ad8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-09 20:18:31.000000000","message":"Patch Set 131: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/131/check/requirements-check/2c78da0/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/131/check/openstack-tox-pep8/ffb175d/ : SUCCESS in 7m 19s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/131/check/openstack-tox-py27/5cf8006/ : SUCCESS in 6m 40s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/131/check/openstack-tox-py35/5afd46e/ : SUCCESS in 6m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/131/check/build-openstack-sphinx-docs/162d36a/html/ : SUCCESS in 7m 46s\n- openstack-tox-cover http://logs.openstack.org/12/435612/131/check/openstack-tox-cover/a1070fc/cover/ : SUCCESS in 7m 27s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/131/check/neutron-lbaasv2-dsvm-api/8296290/ : SUCCESS in 1h 21m 35s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/131/check/neutron-lbaasv2-dsvm-py3x-api/bb47002/ : SUCCESS in 1h 13m 44s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/131/check/neutron-lbaasv2-dsvm-scenario/ef3c297/ : SUCCESS in 1h 36m 29s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/131/check/neutron-lbaasv2-dsvm-py3x-scenario/36c1501/ : SUCCESS in 1h 28m 11s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/131/check/openstack-tox-lower-constraints/78aa94b/ : SUCCESS in 8m 40s\n- openstack-tox-functional http://logs.openstack.org/12/435612/131/check/openstack-tox-functional/aa59b0d/ : SUCCESS in 6m 57s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/131/check/openstack-tox-functional-py35/aec2aad/ : SUCCESS in 7m 18s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/131/check/octavia-v1-dsvm-scenario/e9c3473/ : SUCCESS in 1h 17m 26s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/131/check/octavia-v1-dsvm-py3x-scenario/ddc860d/ : SUCCESS in 1h 21m 04s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/131/check/octavia-v1-dsvm-scenario-multinode/9778334/ : SUCCESS in 1h 19m 28s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/131/check/octavia-v1-dsvm-py3x-scenario-multinode/5b55b75/ : FAILURE in 1h 16m 42s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/131/check/octavia-v1-dsvm-scenario-kvm-centos.7/2cb3c8e/ : SUCCESS in 1h 06m 57s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/131/check/octavia-v1-dsvm-scenario-ubuntu.bionic/7058a86/ : SUCCESS in 1h 02m 06s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/131/check/octavia-v2-dsvm-noop-api/21e2748/ : SUCCESS in 33m 40s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/131/check/octavia-v2-dsvm-noop-py35-api/f0ac625/ : SUCCESS in 34m 39s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/131/check/octavia-v2-dsvm-scenario/edd84bd/ : SUCCESS in 43m 38s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/131/check/octavia-v2-dsvm-py35-scenario/f501856/ : SUCCESS in 56m 33s (non-voting)","accounts_in_message":[],"_revision_number":131},{"id":"424fb6a98bd94e0a452b001e5fd2b7d25971f9a7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-05-11 04:13:48.000000000","message":"Uploaded patch set 132.","accounts_in_message":[],"_revision_number":132},{"id":"ebb77d8e220dfb6ddb5f02364406441b57d682f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-11 05:55:41.000000000","message":"Patch Set 132: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/132/check/requirements-check/76bfa53/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/132/check/openstack-tox-pep8/5117b6c/ : SUCCESS in 6m 40s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/132/check/openstack-tox-py27/38547cc/ : SUCCESS in 4m 08s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/132/check/openstack-tox-py35/0d3cddd/ : SUCCESS in 5m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/132/check/build-openstack-sphinx-docs/4c86f22/html/ : SUCCESS in 5m 44s\n- openstack-tox-cover http://logs.openstack.org/12/435612/132/check/openstack-tox-cover/cb2c3d7/cover/ : SUCCESS in 8m 09s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/132/check/neutron-lbaasv2-dsvm-api/57ffa99/ : SUCCESS in 1h 09m 12s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/132/check/neutron-lbaasv2-dsvm-py3x-api/02364a8/ : SUCCESS in 1h 13m 33s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/132/check/neutron-lbaasv2-dsvm-scenario/55cb537/ : SUCCESS in 1h 23m 29s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/132/check/neutron-lbaasv2-dsvm-py3x-scenario/7596767/ : SUCCESS in 55m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/132/check/openstack-tox-lower-constraints/4cfc873/ : SUCCESS in 7m 16s\n- openstack-tox-functional http://logs.openstack.org/12/435612/132/check/openstack-tox-functional/6066061/ : SUCCESS in 5m 59s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/132/check/openstack-tox-functional-py35/92ce61e/ : SUCCESS in 6m 39s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/132/check/octavia-v1-dsvm-scenario/edf151d/ : SUCCESS in 54m 12s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/132/check/octavia-v1-dsvm-py3x-scenario/090ef31/ : SUCCESS in 1h 02m 55s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/132/check/octavia-v1-dsvm-scenario-multinode/bd338c3/ : SUCCESS in 1h 36m 17s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/132/check/octavia-v1-dsvm-py3x-scenario-multinode/aac8836/ : FAILURE in 1h 10m 10s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/132/check/octavia-v1-dsvm-scenario-kvm-centos.7/014b6d9/ : FAILURE in 25m 29s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/132/check/octavia-v1-dsvm-scenario-ubuntu.bionic/143df54/ : SUCCESS in 50m 54s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/132/check/octavia-v2-dsvm-noop-api/06eb939/ : SUCCESS in 29m 43s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/132/check/octavia-v2-dsvm-noop-py35-api/e391247/ : SUCCESS in 33m 37s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/132/check/octavia-v2-dsvm-scenario/110d3b4/ : SUCCESS in 58m 53s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/132/check/octavia-v2-dsvm-py35-scenario/8d708eb/ : SUCCESS in 1h 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":132},{"id":"c67f9eb0a3836166c6b1ad45370ff60fd40beb8b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-05-31 23:54:46.000000000","message":"Uploaded patch set 133: Patch Set 132 was rebased.","accounts_in_message":[],"_revision_number":133},{"id":"71d4c53f41cbfbda39a44b6f522259d5e3bb1a17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-01 00:31:18.000000000","message":"Patch Set 133: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/133/check/requirements-check/9497554/ : SUCCESS in 2m 53s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/133/check/openstack-tox-pep8/5f651c6/ : FAILURE in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/133/check/openstack-tox-py27/1155c2d/ : FAILURE in 4m 44s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/133/check/openstack-tox-py35/d3ae876/ : FAILURE in 5m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/133/check/build-openstack-sphinx-docs/1d2bb9f/ : FAILURE in 4m 31s\n- openstack-tox-cover http://logs.openstack.org/12/435612/133/check/openstack-tox-cover/841cbb4/ : FAILURE in 5m 08s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/133/check/neutron-lbaasv2-dsvm-api/f1df289/ : FAILURE in 21m 48s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/133/check/neutron-lbaasv2-dsvm-py3x-api/7ec919a/ : FAILURE in 26m 37s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/133/check/neutron-lbaasv2-dsvm-scenario/0bc8d31/ : FAILURE in 28m 29s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/133/check/neutron-lbaasv2-dsvm-py3x-scenario/00be0fd/ : FAILURE in 19m 19s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/133/check/openstack-tox-lower-constraints/718f303/ : FAILURE in 6m 27s\n- openstack-tox-functional http://logs.openstack.org/12/435612/133/check/openstack-tox-functional/3d06644/ : FAILURE in 4m 35s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/133/check/openstack-tox-functional-py35/952832c/ : FAILURE in 4m 37s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/133/check/octavia-v1-dsvm-scenario/6d785b2/ : FAILURE in 23m 18s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/133/check/octavia-v1-dsvm-py3x-scenario/4f88f90/ : FAILURE in 24m 09s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/133/check/octavia-v1-dsvm-scenario-multinode/728c96b/ : FAILURE in 24m 10s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/133/check/octavia-v1-dsvm-py3x-scenario-multinode/08f29ea/ : FAILURE in 29m 10s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/133/check/octavia-v1-dsvm-scenario-kvm-centos.7/8c1560b/ : FAILURE in 23m 18s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/133/check/octavia-v1-dsvm-scenario-ubuntu.bionic/81d372e/ : FAILURE in 23m 34s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/133/check/octavia-v2-dsvm-noop-api/65e5dc6/ : FAILURE in 17m 37s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/133/check/octavia-v2-dsvm-noop-py35-api/c5b9ab9/ : FAILURE in 18m 48s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/133/check/octavia-v2-dsvm-scenario/0f36db0/ : FAILURE in 20m 36s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/133/check/octavia-v2-dsvm-py35-scenario/42da338/ : FAILURE in 19m 26s (non-voting)","accounts_in_message":[],"_revision_number":133},{"id":"a27baf5abf036d11f3b0836c0d416680752cfc6c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-02 02:01:02.000000000","message":"Uploaded patch set 134.","accounts_in_message":[],"_revision_number":134},{"id":"51786d4faaca052f1a623fdfd566e175d2989d52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-02 04:05:02.000000000","message":"Patch Set 134: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/134/check/requirements-check/92180c5/ : SUCCESS in 2m 52s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/134/check/openstack-tox-pep8/3910ae2/ : SUCCESS in 6m 39s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/134/check/openstack-tox-py27/f324294/ : SUCCESS in 4m 56s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/134/check/openstack-tox-py35/dd411ad/ : SUCCESS in 4m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/134/check/build-openstack-sphinx-docs/435bbca/html/ : SUCCESS in 5m 57s\n- openstack-tox-cover http://logs.openstack.org/12/435612/134/check/openstack-tox-cover/2055674/ : FAILURE in 7m 01s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/134/check/neutron-lbaasv2-dsvm-api/8ae2cc3/ : SUCCESS in 1h 12m 16s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/134/check/neutron-lbaasv2-dsvm-py3x-api/95c2f45/ : SUCCESS in 1h 11m 49s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/134/check/neutron-lbaasv2-dsvm-scenario/fe48ec4/ : SUCCESS in 1h 27m 55s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/134/check/neutron-lbaasv2-dsvm-py3x-scenario/5c6481b/ : SUCCESS in 1h 21m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/134/check/openstack-tox-lower-constraints/52ea52d/ : FAILURE in 6m 51s\n- openstack-tox-functional http://logs.openstack.org/12/435612/134/check/openstack-tox-functional/c61abf7/ : FAILURE in 6m 20s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/134/check/openstack-tox-functional-py35/0beb058/ : FAILURE in 6m 19s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/134/check/octavia-v1-dsvm-scenario/0de0839/ : SUCCESS in 54m 35s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/134/check/octavia-v1-dsvm-py3x-scenario/434d2cd/ : SUCCESS in 1h 03m 55s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/134/check/octavia-v1-dsvm-scenario-multinode/447fea1/ : FAILURE in 1h 15m 26s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/134/check/octavia-v1-dsvm-py3x-scenario-multinode/4144350/ : FAILURE in 1h 29m 32s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/134/check/octavia-v1-dsvm-scenario-kvm-centos.7/907b6c2/ : SUCCESS in 1h 17m 19s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/134/check/octavia-v1-dsvm-scenario-ubuntu.bionic/984d627/ : SUCCESS in 51m 59s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/134/check/octavia-v2-dsvm-noop-api/df5891d/ : SUCCESS in 29m 22s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/134/check/octavia-v2-dsvm-noop-py35-api/82cc233/ : SUCCESS in 35m 20s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/134/check/octavia-v2-dsvm-scenario/03a6e49/ : SUCCESS in 1h 09m 04s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/134/check/octavia-v2-dsvm-py35-scenario/0548527/ : SUCCESS in 1h 55m 16s (non-voting)","accounts_in_message":[],"_revision_number":134},{"id":"522fb5d3c2c18bf83d89ee5646f04be32dff7132","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-05 20:55:31.000000000","message":"Patch Set 135: Patch Set 134 was rebased","accounts_in_message":[],"_revision_number":135},{"id":"75bda8545ee94943dd35263436f9100df09701a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-05 22:59:27.000000000","message":"Patch Set 135: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/135/check/requirements-check/7cdd53d/ : SUCCESS in 2m 26s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/135/check/openstack-tox-pep8/e39c061/ : SUCCESS in 6m 23s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/135/check/openstack-tox-py27/d96ffca/ : SUCCESS in 5m 48s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/135/check/openstack-tox-py35/e904086/ : SUCCESS in 5m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/135/check/build-openstack-sphinx-docs/7cef038/html/ : SUCCESS in 6m 01s\n- openstack-tox-cover http://logs.openstack.org/12/435612/135/check/openstack-tox-cover/697cf9c/cover/ : SUCCESS in 7m 34s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/135/check/neutron-lbaasv2-dsvm-api/778d23d/ : SUCCESS in 1h 16m 26s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/135/check/neutron-lbaasv2-dsvm-py3x-api/3646a0e/ : SUCCESS in 1h 11m 35s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/135/check/neutron-lbaasv2-dsvm-scenario/7f86c5e/ : SUCCESS in 1h 26m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/135/check/neutron-lbaasv2-dsvm-py3x-scenario/1e54460/ : SUCCESS in 1h 28m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/135/check/openstack-tox-lower-constraints/53da0e0/ : SUCCESS in 6m 59s\n- openstack-tox-functional http://logs.openstack.org/12/435612/135/check/openstack-tox-functional/0ab57ad/ : SUCCESS in 6m 14s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/135/check/openstack-tox-functional-py35/0c13b09/ : SUCCESS in 7m 38s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/135/check/octavia-v1-dsvm-scenario/fbdc60a/ : SUCCESS in 58m 36s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/135/check/octavia-v1-dsvm-py3x-scenario/65efbdd/ : SUCCESS in 58m 15s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/135/check/octavia-v1-dsvm-scenario-multinode/b834d47/ : FAILURE in 1h 23m 42s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/135/check/octavia-v1-dsvm-py3x-scenario-multinode/70d679f/ : FAILURE in 1h 17m 36s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/135/check/octavia-v1-dsvm-scenario-kvm-centos.7/fa88300/ : SUCCESS in 1h 06m 23s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/135/check/octavia-v1-dsvm-scenario-ubuntu.bionic/a009f31/ : SUCCESS in 55m 38s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/135/check/octavia-v2-dsvm-noop-api/4e30e69/ : SUCCESS in 38m 24s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/135/check/octavia-v2-dsvm-noop-py35-api/de9bead/ : SUCCESS in 37m 24s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/135/check/octavia-v2-dsvm-scenario/12a5405/ : SUCCESS in 1h 12m 47s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/135/check/octavia-v2-dsvm-py35-scenario/9dce8b0/ : SUCCESS in 1h 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":135},{"id":"2c9063e60514ca97a8f447db61288583fae63fcb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-07 01:38:31.000000000","message":"Uploaded patch set 136: Patch Set 135 was rebased.","accounts_in_message":[],"_revision_number":136},{"id":"c211564db0ca1a36de23f343692b3ed88786ab3c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-07 02:00:45.000000000","message":"Uploaded patch set 137.","accounts_in_message":[],"_revision_number":137},{"id":"5ca7bc185538b2042663d704d5c632e36b8831dc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-07 02:44:00.000000000","message":"Uploaded patch set 138.","accounts_in_message":[],"_revision_number":138},{"id":"535b7122e1d639a7b215e9992b71f17d8c4a9607","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-07 04:15:41.000000000","message":"Patch Set 138: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/138/check/requirements-check/80996f9/ : SUCCESS in 2m 51s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/138/check/openstack-tox-pep8/3ca0558/ : SUCCESS in 6m 42s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/138/check/openstack-tox-py27/4ccb052/ : SUCCESS in 5m 44s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/138/check/openstack-tox-py35/c47955b/ : SUCCESS in 5m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/138/check/build-openstack-sphinx-docs/d2f3bb7/html/ : SUCCESS in 5m 57s\n- openstack-tox-cover http://logs.openstack.org/12/435612/138/check/openstack-tox-cover/4093887/cover/ : SUCCESS in 7m 08s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/138/check/neutron-lbaasv2-dsvm-api/4a699e4/ : SUCCESS in 1h 11m 45s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/138/check/neutron-lbaasv2-dsvm-py3x-api/dd727b9/ : SUCCESS in 1h 15m 41s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/138/check/neutron-lbaasv2-dsvm-scenario/c055bdb/ : SUCCESS in 1h 20m 58s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/138/check/neutron-lbaasv2-dsvm-py3x-scenario/2c6ff9e/ : SUCCESS in 1h 17m 11s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/138/check/openstack-tox-lower-constraints/fbb0712/ : SUCCESS in 6m 55s\n- openstack-tox-functional http://logs.openstack.org/12/435612/138/check/openstack-tox-functional/f89c273/ : SUCCESS in 7m 12s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/138/check/openstack-tox-functional-py35/971786f/ : SUCCESS in 6m 32s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/138/check/octavia-v1-dsvm-scenario/13a7c9c/ : SUCCESS in 54m 24s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/138/check/octavia-v1-dsvm-py3x-scenario/ec32688/ : SUCCESS in 52m 02s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/138/check/octavia-v1-dsvm-scenario-multinode/fce9433/ : FAILURE in 1h 16m 01s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/138/check/octavia-v1-dsvm-py3x-scenario-multinode/d71920d/ : FAILURE in 1h 20m 18s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/138/check/octavia-v1-dsvm-scenario-kvm-centos.7/e3af771/ : SUCCESS in 56m 57s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/138/check/octavia-v1-dsvm-scenario-ubuntu.bionic/edebf94/ : SUCCESS in 51m 16s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/138/check/octavia-v2-dsvm-noop-api/8762419/ : SUCCESS in 35m 05s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/138/check/octavia-v2-dsvm-noop-py35-api/b0a14d9/ : SUCCESS in 36m 57s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/138/check/octavia-v2-dsvm-scenario/16f030e/ : SUCCESS in 1h 25m 54s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/138/check/octavia-v2-dsvm-py35-scenario/201b6c9/ : SUCCESS in 1h 13m 45s (non-voting)","accounts_in_message":[],"_revision_number":138},{"id":"a0ec4675cd595f775d474d55bf3a930977433dff","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-07 05:01:51.000000000","message":"Uploaded patch set 139.","accounts_in_message":[],"_revision_number":139},{"id":"5c739f2782f381e6f9de4c55e0c375ec269ab603","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-07 06:31:58.000000000","message":"Patch Set 139: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/139/check/requirements-check/c4332dd/ : SUCCESS in 2m 33s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/139/check/openstack-tox-pep8/776e1dd/ : SUCCESS in 6m 24s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/139/check/openstack-tox-py27/2dc9f6e/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/139/check/openstack-tox-py35/6e630bf/ : SUCCESS in 4m 50s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/139/check/build-openstack-sphinx-docs/5ab95af/html/ : SUCCESS in 6m 44s\n- openstack-tox-cover http://logs.openstack.org/12/435612/139/check/openstack-tox-cover/f430539/cover/ : SUCCESS in 7m 53s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/139/check/neutron-lbaasv2-dsvm-api/2a6b064/ : SUCCESS in 1h 12m 46s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/139/check/neutron-lbaasv2-dsvm-py3x-api/2f9c91a/ : SUCCESS in 1h 15m 40s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/139/check/neutron-lbaasv2-dsvm-scenario/f81dffc/ : SUCCESS in 1h 18m 37s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/139/check/neutron-lbaasv2-dsvm-py3x-scenario/88a2ad9/ : SUCCESS in 1h 14m 48s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/139/check/openstack-tox-lower-constraints/fbcc378/ : SUCCESS in 6m 42s\n- openstack-tox-functional http://logs.openstack.org/12/435612/139/check/openstack-tox-functional/9294022/ : SUCCESS in 7m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/139/check/openstack-tox-functional-py35/57b2436/ : SUCCESS in 7m 25s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/139/check/octavia-v1-dsvm-scenario/a365646/ : SUCCESS in 58m 17s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/139/check/octavia-v1-dsvm-py3x-scenario/59fac26/ : SUCCESS in 52m 48s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/139/check/octavia-v1-dsvm-scenario-multinode/3252b29/ : FAILURE in 1h 22m 35s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/139/check/octavia-v1-dsvm-py3x-scenario-multinode/4d8dcec/ : FAILURE in 1h 18m 31s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/139/check/octavia-v1-dsvm-scenario-kvm-centos.7/84dde7e/ : SUCCESS in 1h 00m 20s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/139/check/octavia-v1-dsvm-scenario-ubuntu.bionic/7ec2c24/ : SUCCESS in 52m 54s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/139/check/octavia-v2-dsvm-noop-api/ad1bab8/ : SUCCESS in 36m 10s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/139/check/octavia-v2-dsvm-noop-py35-api/a3038b1/ : SUCCESS in 40m 32s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/139/check/octavia-v2-dsvm-scenario/01c10cb/ : SUCCESS in 1h 09m 56s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/139/check/octavia-v2-dsvm-py35-scenario/024fa5e/ : FAILURE in 1h 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":139},{"id":"dbea8c0a70b1416f963090b2c41431114816b0d1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-07 08:40:36.000000000","message":"Uploaded patch set 140.","accounts_in_message":[],"_revision_number":140},{"id":"8d4d8b8c6faeeef68f0f5c688cc3158e60c239d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-07 10:49:49.000000000","message":"Patch Set 140: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/140/check/requirements-check/31ced18/ : SUCCESS in 3m 15s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/140/check/openstack-tox-pep8/d2fcc36/ : SUCCESS in 6m 45s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/140/check/openstack-tox-py27/b12389c/ : SUCCESS in 5m 36s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/140/check/openstack-tox-py35/43444f6/ : SUCCESS in 5m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/140/check/build-openstack-sphinx-docs/665bb00/html/ : SUCCESS in 6m 12s\n- openstack-tox-cover http://logs.openstack.org/12/435612/140/check/openstack-tox-cover/2bd5a33/cover/ : SUCCESS in 8m 50s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/140/check/neutron-lbaasv2-dsvm-api/c1ea5d4/ : SUCCESS in 1h 15m 43s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/140/check/neutron-lbaasv2-dsvm-py3x-api/e997b94/ : SUCCESS in 1h 10m 48s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/140/check/neutron-lbaasv2-dsvm-scenario/700464e/ : SUCCESS in 1h 13m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/140/check/neutron-lbaasv2-dsvm-py3x-scenario/a00c89a/ : SUCCESS in 1h 15m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/140/check/openstack-tox-lower-constraints/16b22d5/ : SUCCESS in 6m 58s\n- openstack-tox-functional http://logs.openstack.org/12/435612/140/check/openstack-tox-functional/91f13d3/ : SUCCESS in 6m 13s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/140/check/openstack-tox-functional-py35/3e180b8/ : SUCCESS in 6m 31s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/140/check/octavia-v1-dsvm-scenario/a8473f3/ : SUCCESS in 51m 01s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/140/check/octavia-v1-dsvm-py3x-scenario/b9fdc9c/ : SUCCESS in 50m 36s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/140/check/octavia-v1-dsvm-scenario-multinode/f9c692d/ : FAILURE in 1h 12m 17s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/140/check/octavia-v1-dsvm-py3x-scenario-multinode/c0a89fb/ : FAILURE in 1h 15m 44s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/140/check/octavia-v1-dsvm-scenario-kvm-centos.7/2f1bdaf/ : SUCCESS in 1h 03m 09s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/140/check/octavia-v1-dsvm-scenario-ubuntu.bionic/a77d246/ : SUCCESS in 56m 52s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/140/check/octavia-v2-dsvm-noop-api/7e629aa/ : SUCCESS in 35m 10s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/140/check/octavia-v2-dsvm-noop-py35-api/c52c3d1/ : SUCCESS in 45m 27s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/140/check/octavia-v2-dsvm-scenario/021d7ba/ : SUCCESS in 1h 06m 09s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/140/check/octavia-v2-dsvm-py35-scenario/2340a5d/ : SUCCESS in 1h 32m 50s (non-voting)","accounts_in_message":[],"_revision_number":140},{"id":"1f09e68e06b250537aab2036ac7f118359e5c262","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-14 18:09:41.000000000","message":"Uploaded patch set 141.","accounts_in_message":[],"_revision_number":141},{"id":"69d56cce3754db09de2ecf1a5b1d882ec31e392b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-14 22:18:32.000000000","message":"Patch Set 141: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/141/check/requirements-check/8edd3d5/ : SUCCESS in 5m 27s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/141/check/openstack-tox-pep8/9c70403/ : SUCCESS in 8m 17s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/141/check/openstack-tox-py27/4a9899b/ : SUCCESS in 7m 51s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/141/check/openstack-tox-py35/d1c5ed2/ : SUCCESS in 7m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/141/check/build-openstack-sphinx-docs/4f3144b/html/ : SUCCESS in 8m 05s\n- openstack-tox-cover http://logs.openstack.org/12/435612/141/check/openstack-tox-cover/401e0be/cover/ : SUCCESS in 9m 24s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/141/check/neutron-lbaasv2-dsvm-api/acc3ace/ : SUCCESS in 1h 15m 49s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/141/check/neutron-lbaasv2-dsvm-py3x-api/fbea7ae/ : SUCCESS in 1h 15m 13s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/141/check/neutron-lbaasv2-dsvm-scenario/80bbf06/ : SUCCESS in 1h 16m 04s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/141/check/neutron-lbaasv2-dsvm-py3x-scenario/58c0c0f/ : SUCCESS in 1h 21m 27s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/141/check/openstack-tox-lower-constraints/c4e5c87/ : SUCCESS in 9m 18s\n- openstack-tox-functional http://logs.openstack.org/12/435612/141/check/openstack-tox-functional/53427a6/ : SUCCESS in 7m 12s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/141/check/openstack-tox-functional-py35/39caa13/ : SUCCESS in 7m 26s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/141/check/octavia-v1-dsvm-scenario/8de5431/ : SUCCESS in 53m 55s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/141/check/octavia-v1-dsvm-py3x-scenario/17f8f89/ : SUCCESS in 59m 04s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/141/check/octavia-v1-dsvm-scenario-multinode/ac657b0/ : FAILURE in 1h 23m 33s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/141/check/octavia-v1-dsvm-py3x-scenario-multinode/7129782/ : FAILURE in 1h 15m 18s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/141/check/octavia-v1-dsvm-scenario-kvm-centos.7/509b5be/ : SUCCESS in 57m 37s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/141/check/octavia-v1-dsvm-scenario-ubuntu.bionic/a17a867/ : SUCCESS in 1h 01m 16s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/141/check/octavia-v2-dsvm-noop-api/1b0321c/ : SUCCESS in 39m 51s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/141/check/octavia-v2-dsvm-noop-py35-api/e0ec40a/ : SUCCESS in 45m 08s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/141/check/octavia-v2-dsvm-scenario/2b0c219/ : SUCCESS in 1h 12m 02s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/141/check/octavia-v2-dsvm-py35-scenario/0473462/ : SUCCESS in 1h 25m 22s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/141/check/octavia-grenade/8875830/ : SUCCESS in 1h 03m 19s (non-voting)","accounts_in_message":[],"_revision_number":141},{"id":"a792cd109e81af14da6b5fade00a017f682a56a4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-29 20:25:45.000000000","message":"Uploaded patch set 142.","accounts_in_message":[],"_revision_number":142},{"id":"89535248a07a1b4626438d29526eff255aebe824","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-29 22:26:21.000000000","message":"Patch Set 142: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/142/check/requirements-check/65d7032/ : FAILURE in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/142/check/openstack-tox-pep8/e938182/ : SUCCESS in 8m 00s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/142/check/openstack-tox-py27/94a961b/ : SUCCESS in 5m 55s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/142/check/openstack-tox-py35/118eb7f/ : SUCCESS in 6m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/142/check/build-openstack-sphinx-docs/bbf8d8a/html/ : SUCCESS in 7m 46s\n- openstack-tox-cover http://logs.openstack.org/12/435612/142/check/openstack-tox-cover/f8b3714/cover/ : SUCCESS in 11m 57s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/142/check/neutron-lbaasv2-dsvm-api/dbd4751/ : SUCCESS in 1h 11m 14s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/142/check/neutron-lbaasv2-dsvm-py3x-api/b04eb53/ : SUCCESS in 1h 10m 23s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/142/check/neutron-lbaasv2-dsvm-scenario/c5c6003/ : SUCCESS in 1h 14m 12s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/142/check/neutron-lbaasv2-dsvm-py3x-scenario/149e8c4/ : SUCCESS in 1h 12m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/142/check/openstack-tox-lower-constraints/5f068a3/ : SUCCESS in 7m 39s\n- openstack-tox-functional http://logs.openstack.org/12/435612/142/check/openstack-tox-functional/473128d/ : SUCCESS in 7m 10s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/142/check/openstack-tox-functional-py35/8c1c615/ : SUCCESS in 7m 21s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/142/check/octavia-v1-dsvm-scenario/376e15f/ : SUCCESS in 1h 00m 44s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/142/check/octavia-v1-dsvm-py3x-scenario/4b57d13/ : SUCCESS in 52m 33s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/142/check/octavia-v1-dsvm-scenario-multinode/06e91e1/ : FAILURE in 1h 13m 54s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/142/check/octavia-v1-dsvm-py3x-scenario-multinode/23e1481/ : FAILURE in 1h 15m 20s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/142/check/octavia-v1-dsvm-scenario-kvm-centos.7/c064e49/ : SUCCESS in 1h 09m 40s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/142/check/octavia-v1-dsvm-scenario-ubuntu.bionic/5f287b9/ : SUCCESS in 51m 32s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/142/check/octavia-v2-dsvm-noop-api/34f64a3/ : SUCCESS in 36m 14s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/142/check/octavia-v2-dsvm-noop-py35-api/faaeaa4/ : SUCCESS in 37m 31s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/142/check/octavia-v2-dsvm-scenario/e051901/ : SUCCESS in 1h 20m 25s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/142/check/octavia-v2-dsvm-py35-scenario/c572ccb/ : SUCCESS in 1h 26m 05s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/142/check/octavia-grenade/6971a27/ : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":142},{"id":"ac7060004771471d1782a65e1288dd3c9b726e85","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-07-13 01:07:03.000000000","message":"Uploaded patch set 143: Patch Set 142 was rebased.","accounts_in_message":[],"_revision_number":143},{"id":"34a7f1c16dd720f6f0d09ef8e56e888efcd74bb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-13 03:21:32.000000000","message":"Patch Set 143: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/143/check/requirements-check/59394c4/ : SUCCESS in 4m 18s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/143/check/openstack-tox-pep8/91675a3/ : FAILURE in 8m 40s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/143/check/openstack-tox-py27/cfe1bbb/ : SUCCESS in 7m 11s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/143/check/openstack-tox-py35/e81ccdc/ : SUCCESS in 8m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/143/check/build-openstack-sphinx-docs/7d390bf/html/ : SUCCESS in 7m 25s\n- openstack-tox-cover http://logs.openstack.org/12/435612/143/check/openstack-tox-cover/1fe28bd/cover/ : SUCCESS in 9m 52s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/143/check/neutron-lbaasv2-dsvm-api/d023131/ : SUCCESS in 1h 14m 39s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/143/check/neutron-lbaasv2-dsvm-py3x-api/54e2a64/ : SUCCESS in 1h 11m 30s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/143/check/neutron-lbaasv2-dsvm-scenario/137a7ea/ : SUCCESS in 1h 22m 40s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/143/check/neutron-lbaasv2-dsvm-py3x-scenario/abfc674/ : SUCCESS in 58m 05s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/143/check/openstack-tox-lower-constraints/21e7f00/ : SUCCESS in 9m 29s\n- openstack-tox-functional http://logs.openstack.org/12/435612/143/check/openstack-tox-functional/54a38b7/ : SUCCESS in 7m 05s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/143/check/openstack-tox-functional-py35/f81a344/ : SUCCESS in 7m 45s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/143/check/octavia-v1-dsvm-scenario/6c3b991/ : SUCCESS in 59m 55s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/143/check/octavia-v1-dsvm-py3x-scenario/5177a19/ : SUCCESS in 53m 34s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/143/check/octavia-v1-dsvm-scenario-multinode/0feb075/ : FAILURE in 1h 17m 41s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/143/check/octavia-v1-dsvm-py3x-scenario-multinode/aef2215/ : FAILURE in 1h 13m 42s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/143/check/octavia-v1-dsvm-scenario-kvm-centos.7/8b7eaf2/ : SUCCESS in 1h 09m 50s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/143/check/octavia-v1-dsvm-scenario-ubuntu.bionic/5b9f9ad/ : SUCCESS in 1h 00m 36s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/143/check/octavia-v2-dsvm-noop-api/518b18d/ : SUCCESS in 48m 31s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/143/check/octavia-v2-dsvm-noop-py35-api/882d959/ : SUCCESS in 38m 06s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/143/check/octavia-v2-dsvm-scenario/d04db53/ : SUCCESS in 1h 10m 15s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/143/check/octavia-v2-dsvm-py35-scenario/328865f/ : SUCCESS in 1h 12m 48s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/143/check/octavia-grenade/92a4929/ : SUCCESS in 1h 10m 59s","accounts_in_message":[],"_revision_number":143},{"id":"73a0378b3d7e77c3016ea4dcf80e6c40d6fa2238","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-07-16 21:12:40.000000000","message":"Patch Set 144: Patch Set 143 was rebased","accounts_in_message":[],"_revision_number":144},{"id":"24bf5a869042798bac71d42bec3158a54f0f2d14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-17 00:39:27.000000000","message":"Patch Set 144: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/144/check/requirements-check/c6de12b/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/144/check/openstack-tox-pep8/19d1984/ : FAILURE in 8m 50s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/144/check/openstack-tox-py27/3f7d71e/ : SUCCESS in 6m 35s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/144/check/openstack-tox-py35/23a7909/ : SUCCESS in 6m 43s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/435612/144/check/build-openstack-sphinx-docs/533d9ba/html/ : SUCCESS in 7m 54s\n- openstack-tox-cover http://logs.openstack.org/12/435612/144/check/openstack-tox-cover/ae4f2f1/cover/ : SUCCESS in 9m 24s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/144/check/neutron-lbaasv2-dsvm-api/526e76e/ : SUCCESS in 1h 13m 34s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/144/check/neutron-lbaasv2-dsvm-py3x-api/34da11e/ : SUCCESS in 1h 14m 36s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/144/check/neutron-lbaasv2-dsvm-scenario/7c0939e/ : SUCCESS in 55m 18s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/144/check/neutron-lbaasv2-dsvm-py3x-scenario/17b0d46/ : SUCCESS in 1h 38m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/144/check/openstack-tox-lower-constraints/0c0e7c1/ : SUCCESS in 9m 26s\n- openstack-tox-functional http://logs.openstack.org/12/435612/144/check/openstack-tox-functional/404c516/ : SUCCESS in 8m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/144/check/openstack-tox-functional-py35/0894d9f/ : SUCCESS in 8m 22s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/144/check/octavia-v1-dsvm-scenario/a32f8fb/ : SUCCESS in 49m 50s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/144/check/octavia-v1-dsvm-py3x-scenario/7274819/ : FAILURE in 30m 04s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/144/check/octavia-v1-dsvm-scenario-multinode/7bca179/ : FAILURE in 1h 22m 33s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/144/check/octavia-v1-dsvm-py3x-scenario-multinode/3d6c689/ : FAILURE in 32m 10s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/144/check/octavia-v1-dsvm-scenario-kvm-centos.7/ab5b43f/ : SUCCESS in 1h 18m 30s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/144/check/octavia-v1-dsvm-scenario-ubuntu.bionic/319cae9/ : SUCCESS in 1h 04m 52s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/144/check/octavia-v2-dsvm-noop-api/cf3fc33/ : SUCCESS in 38m 43s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/144/check/octavia-v2-dsvm-noop-py35-api/6b12502/ : SUCCESS in 42m 48s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/144/check/octavia-v2-dsvm-scenario/40858ae/ : SUCCESS in 1h 13m 14s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/144/check/octavia-v2-dsvm-py35-scenario/774b864/ : SUCCESS in 1h 13m 20s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/144/check/octavia-grenade/96dd1b2/ : SUCCESS in 1h 18m 31s","accounts_in_message":[],"_revision_number":144},{"id":"ff79140191bbe87e3b8c29e1fcb7a43656d183ca","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-07-30 18:30:04.000000000","message":"Uploaded patch set 145.","accounts_in_message":[],"_revision_number":145},{"id":"4c6337d33ccf258409f8536c0656a25bc18ce308","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-30 20:13:40.000000000","message":"Patch Set 145: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/12/435612/145/check/requirements-check/f7e62d8/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/145/check/openstack-tox-pep8/27de0e7/ : SUCCESS in 10m 26s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/145/check/openstack-tox-py27/f938cbd/ : SUCCESS in 5m 32s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/145/check/openstack-tox-py35/f2432c8/ : SUCCESS in 6m 56s\n- openstack-tox-docs http://logs.openstack.org/12/435612/145/check/openstack-tox-docs/67646c3/ : SUCCESS in 6m 53s\n- openstack-tox-cover http://logs.openstack.org/12/435612/145/check/openstack-tox-cover/6f526dd/cover/ : SUCCESS in 9m 00s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/145/check/neutron-lbaasv2-dsvm-api/d76b021/ : SUCCESS in 1h 11m 10s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/145/check/neutron-lbaasv2-dsvm-py3x-api/b599116/ : SUCCESS in 1h 19m 20s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/145/check/neutron-lbaasv2-dsvm-scenario/e3f3681/ : SUCCESS in 1h 12m 06s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/145/check/neutron-lbaasv2-dsvm-py3x-scenario/3056e41/ : SUCCESS in 57m 22s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/145/check/openstack-tox-lower-constraints/dda5274/ : SUCCESS in 7m 41s\n- openstack-tox-functional http://logs.openstack.org/12/435612/145/check/openstack-tox-functional/744da37/ : SUCCESS in 9m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/145/check/openstack-tox-functional-py35/e49cf06/ : SUCCESS in 7m 10s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/145/check/octavia-v1-dsvm-scenario/b300417/ : SUCCESS in 1h 18m 40s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/145/check/octavia-v1-dsvm-py3x-scenario/f9adf94/ : SUCCESS in 52m 17s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/145/check/octavia-v1-dsvm-scenario-multinode/12333f1/ : FAILURE in 1h 10m 59s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/145/check/octavia-v1-dsvm-py3x-scenario-multinode/67f3236/ : FAILURE in 1h 25m 23s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/145/check/octavia-v1-dsvm-scenario-kvm-centos.7/8f8a922/ : SUCCESS in 49m 51s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/145/check/octavia-v1-dsvm-scenario-ubuntu.bionic/8505f61/ : SUCCESS in 1h 01m 07s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/145/check/octavia-v2-dsvm-noop-api/41456cf/ : SUCCESS in 43m 12s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/145/check/octavia-v2-dsvm-noop-py35-api/4d68367/ : SUCCESS in 36m 47s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/145/check/octavia-v2-dsvm-scenario/a83e7ea/ : SUCCESS in 1h 09m 54s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/145/check/octavia-v2-dsvm-py35-scenario/a636d82/ : SUCCESS in 1h 12m 19s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/145/check/octavia-grenade/ca2ee46/ : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":145},{"id":"dfc806b956137f893c15fb566652a633c52dbf0e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-08-21 01:55:37.000000000","message":"Uploaded patch set 146.","accounts_in_message":[],"_revision_number":146},{"id":"6799794fad4a197e1d58be14d6085471d83e6739","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-21 04:13:00.000000000","message":"Patch Set 146: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/146/check/requirements-check/2fb523c/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/146/check/openstack-tox-pep8/242d1df/ : FAILURE in 5m 44s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/146/check/openstack-tox-py27/1829f22/ : FAILURE in 6m 26s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/146/check/openstack-tox-py35/01095b4/ : FAILURE in 5m 19s\n- openstack-tox-docs http://logs.openstack.org/12/435612/146/check/openstack-tox-docs/75501dd/html/ : SUCCESS in 6m 08s\n- openstack-tox-cover http://logs.openstack.org/12/435612/146/check/openstack-tox-cover/ffe63ab/ : FAILURE in 5m 18s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/146/check/neutron-lbaasv2-dsvm-api/801c8c0/ : FAILURE in 2h 03m 27s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/146/check/neutron-lbaasv2-dsvm-py3x-api/27561cc/ : FAILURE in 2h 01m 28s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/146/check/neutron-lbaasv2-dsvm-scenario/bb71ef6/ : FAILURE in 1h 52m 34s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/146/check/neutron-lbaasv2-dsvm-py3x-scenario/f7e5368/ : FAILURE in 1h 59m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/146/check/openstack-tox-lower-constraints/385bb7b/ : FAILURE in 5m 25s\n- openstack-tox-functional http://logs.openstack.org/12/435612/146/check/openstack-tox-functional/9f2c27d/ : SUCCESS in 8m 00s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/146/check/openstack-tox-functional-py35/1754275/ : SUCCESS in 7m 18s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/146/check/octavia-v1-dsvm-scenario/7bc93cf/ : FAILURE in 1h 02m 28s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/146/check/octavia-v1-dsvm-py3x-scenario/b400cff/ : FAILURE in 45m 23s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/146/check/octavia-v1-dsvm-scenario-multinode/e0b780e/ : FAILURE in 1h 17m 54s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/146/check/octavia-v1-dsvm-py3x-scenario-multinode/15cde98/ : FAILURE in 1h 13m 14s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/146/check/octavia-v1-dsvm-scenario-kvm-centos.7/fbfe518/ : FAILURE in 1h 01m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/12/435612/146/check/octavia-v2-dsvm-scenario-ubuntu.bionic/f3f9fd3/ : FAILURE in 1h 16m 24s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/146/check/octavia-v2-dsvm-noop-api/6c3db75/ : TIMED_OUT in 2h 12m 15s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/146/check/octavia-v2-dsvm-noop-py35-api/8f00c50/ : TIMED_OUT in 2h 13m 44s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/146/check/octavia-v2-dsvm-scenario/d7b9d9a/ : FAILURE in 1h 11m 19s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/146/check/octavia-v2-dsvm-py35-scenario/ec8acfd/ : FAILURE in 1h 09m 14s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/146/check/octavia-grenade/bc77d18/ : SUCCESS in 1h 13m 06s","accounts_in_message":[],"_revision_number":146},{"id":"042aee80be78b0a751e235cebcf2f424efedbf16","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-08-21 04:13:22.000000000","message":"Patch Set 146:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- neutron-lbaasv2-dsvm-api: https://bugs.launchpad.net/bugs/1686542\n- neutron-lbaasv2-dsvm-py3x-api: https://bugs.launchpad.net/bugs/1686542\n- octavia-v1-dsvm-py3x-scenario: unrecognized error\n- octavia-v1-dsvm-scenario: unrecognized error\n- openstack-tox-cover: unrecognized error\n- openstack-tox-lower-constraints: unrecognized error\n- openstack-tox-pep8: unrecognized error\n- openstack-tox-py27: unrecognized error\n- openstack-tox-py35: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":146},{"id":"2a01a4b3f519c4bea092a00782175151c546e252","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-08-24 20:07:05.000000000","message":"Patch Set 146:\n\nrecheck","accounts_in_message":[],"_revision_number":146},{"id":"b87ce87c03fa607f903a3fb6307bdb9685e7f0fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-24 22:26:27.000000000","message":"Patch Set 146:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/146/check/requirements-check/0d8e97d/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/146/check/openstack-tox-pep8/519bc5d/ : FAILURE in 6m 26s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/146/check/openstack-tox-py27/c879427/ : FAILURE in 5m 36s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/146/check/openstack-tox-py35/7cfc57a/ : FAILURE in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/12/435612/146/check/openstack-tox-docs/f243a21/html/ : SUCCESS in 6m 38s\n- openstack-tox-cover http://logs.openstack.org/12/435612/146/check/openstack-tox-cover/421b9c1/ : FAILURE in 5m 59s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/146/check/neutron-lbaasv2-dsvm-api/5269c71/ : FAILURE in 2h 00m 41s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/146/check/neutron-lbaasv2-dsvm-py3x-api/0b6fcfe/ : FAILURE in 2h 01m 40s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/146/check/neutron-lbaasv2-dsvm-scenario/b70f44d/ : FAILURE in 2h 00m 42s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/146/check/neutron-lbaasv2-dsvm-py3x-scenario/4b9b3e8/ : FAILURE in 2h 02m 13s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/146/check/openstack-tox-lower-constraints/b5be7c6/ : FAILURE in 6m 20s\n- openstack-tox-functional http://logs.openstack.org/12/435612/146/check/openstack-tox-functional/c08ca1a/ : SUCCESS in 7m 30s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/146/check/openstack-tox-functional-py35/941ce39/ : SUCCESS in 8m 01s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/146/check/octavia-v1-dsvm-scenario/130e03f/ : FAILURE in 1h 02m 29s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/146/check/octavia-v1-dsvm-py3x-scenario/7cc6df6/ : FAILURE in 53m 48s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/146/check/octavia-v1-dsvm-scenario-multinode/a47e9d9/ : FAILURE in 1h 16m 32s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/146/check/octavia-v1-dsvm-py3x-scenario-multinode/540ff4a/ : FAILURE in 1h 35m 02s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/12/435612/146/check/octavia-v1-dsvm-scenario-kvm-centos.7/0d719e0/ : FAILURE in 26m 20s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/146/check/octavia-v2-dsvm-noop-api/7cc1bcd/ : TIMED_OUT in 2h 12m 42s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/146/check/octavia-v2-dsvm-noop-py35-api/b752f46/ : TIMED_OUT in 2h 13m 20s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/146/check/octavia-v2-dsvm-scenario/ba81e63/ : FAILURE in 1h 21m 30s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/146/check/octavia-v2-dsvm-py35-scenario/4098920/ : FAILURE in 1h 11m 51s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/146/check/octavia-grenade/01914d5/ : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":146},{"id":"351a02031859183b662501c081e74e5ce2c391d7","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-08-24 22:26:55.000000000","message":"Patch Set 146:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- neutron-lbaasv2-dsvm-api: https://bugs.launchpad.net/bugs/1686542\n- neutron-lbaasv2-dsvm-py3x-api: https://bugs.launchpad.net/bugs/1686542\n- octavia-v1-dsvm-py3x-scenario: unrecognized error\n- octavia-v1-dsvm-scenario: unrecognized error\n- openstack-tox-cover: unrecognized error\n- openstack-tox-lower-constraints: unrecognized error\n- openstack-tox-pep8: unrecognized error\n- openstack-tox-py27: unrecognized error\n- openstack-tox-py35: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":146},{"id":"825a712f012a5675f910800bf7a30c409935ef55","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-09-17 22:10:43.000000000","message":"Patch Set 147: Patch Set 146 was rebased","accounts_in_message":[],"_revision_number":147},{"id":"e254d29c5429dd9765cdc8ddb768a89307b09ca2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-09-17 23:03:03.000000000","message":"Uploaded patch set 148.","accounts_in_message":[],"_revision_number":148},{"id":"fbd4ab5209d0e4b544d9a174e9f439f928ee2868","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-18 00:50:06.000000000","message":"Patch Set 148: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/12/435612/148/check/requirements-check/3218f8c/ : SUCCESS in 4m 03s\n- openstack-tox-cover http://logs.openstack.org/12/435612/148/check/openstack-tox-cover/522726d/ : FAILURE in 10m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/148/check/openstack-tox-lower-constraints/d02cb3f/ : SUCCESS in 8m 24s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/148/check/openstack-tox-pep8/09dd9d9/ : SUCCESS in 8m 58s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/148/check/openstack-tox-py27/d901819/ : SUCCESS in 6m 18s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/148/check/openstack-tox-py35/9efbbcc/ : SUCCESS in 6m 30s\n- openstack-tox-py36 http://logs.openstack.org/12/435612/148/check/openstack-tox-py36/8a86fb3/ : FAILURE in 7m 22s\n- openstack-tox-docs http://logs.openstack.org/12/435612/148/check/openstack-tox-docs/ceda045/html/ : SUCCESS in 6m 29s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/148/check/neutron-lbaasv2-dsvm-api/5f19217/ : SUCCESS in 1h 13m 40s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/148/check/neutron-lbaasv2-dsvm-py3x-api/780b8fc/ : SUCCESS in 1h 08m 56s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/148/check/neutron-lbaasv2-dsvm-scenario/7227aae/ : SUCCESS in 1h 14m 42s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/148/check/neutron-lbaasv2-dsvm-py3x-scenario/60e54c8/ : SUCCESS in 1h 34m 01s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/12/435612/148/check/openstack-tox-functional/28786ea/ : SUCCESS in 7m 10s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/148/check/openstack-tox-functional-py35/4f730fd/ : SUCCESS in 6m 50s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/148/check/octavia-v1-dsvm-scenario/2a0d5f6/ : SUCCESS in 57m 20s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/148/check/octavia-v1-dsvm-py3x-scenario/f44e503/ : SUCCESS in 1h 06m 56s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/148/check/octavia-v1-dsvm-scenario-multinode/a36e1dd/ : FAILURE in 1h 04m 27s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/148/check/octavia-v1-dsvm-py3x-scenario-multinode/448df6b/ : FAILURE in 1h 25m 35s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/148/check/octavia-v2-dsvm-noop-api/cba1e32/ : SUCCESS in 38m 09s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/148/check/octavia-v2-dsvm-noop-py35-api/6921c84/ : SUCCESS in 43m 08s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/148/check/octavia-v2-dsvm-scenario/89ebc58/ : SUCCESS in 1h 28m 17s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/148/check/octavia-v2-dsvm-py35-scenario/61731e0/ : SUCCESS in 1h 10m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/12/435612/148/check/octavia-v2-dsvm-scenario-centos-7/0b2aa5e/ : SUCCESS in 1h 44m 38s\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/12/435612/148/check/octavia-v2-dsvm-scenario-ubuntu-bionic/96955a9/ : SUCCESS in 1h 19m 38s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/148/check/octavia-grenade/de2286a/ : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":148},{"id":"01fbed95f4764a1b70a37db668f5556ddebba123","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2018-09-24 07:45:34.000000000","message":"Patch Set 148:\n\n(1 comment)","accounts_in_message":[],"_revision_number":148},{"id":"602e9e4c0adf366730da398d9975c3bcaf5f12a2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-09-24 18:17:46.000000000","message":"Patch Set 148:\n\n(1 comment)","accounts_in_message":[],"_revision_number":148},{"id":"b8fc4b12d46ed5e5ed2335df9590332fb776a325","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2018-09-25 08:10:50.000000000","message":"Patch Set 148:\n\n(1 comment)","accounts_in_message":[],"_revision_number":148},{"id":"52ee493dc59f76e8a813eeabe7a0b80c11ff7285","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-10-01 22:14:56.000000000","message":"Uploaded patch set 149.","accounts_in_message":[],"_revision_number":149},{"id":"f6e99f204d6f593d09cc2c6ae5383d5ce3315fcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-02 01:14:36.000000000","message":"Patch Set 149: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack-python3 http://logs.openstack.org/12/435612/149/check/test-release-openstack-python3/7f7d783/ : SUCCESS in 4m 44s\n- requirements-check http://logs.openstack.org/12/435612/149/check/requirements-check/4de8fc8/ : SUCCESS in 5m 12s\n- openstack-tox-cover http://logs.openstack.org/12/435612/149/check/openstack-tox-cover/76d26c3/ : FAILURE in 8m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/149/check/openstack-tox-lower-constraints/73fcb63/ : SUCCESS in 8m 34s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/149/check/openstack-tox-pep8/1266adc/ : SUCCESS in 9m 00s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/149/check/openstack-tox-py27/527bacc/ : SUCCESS in 5m 37s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/149/check/openstack-tox-py35/17c460c/ : FAILURE in 6m 14s\n- openstack-tox-py36 http://logs.openstack.org/12/435612/149/check/openstack-tox-py36/bd88240/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/12/435612/149/check/openstack-tox-docs/d17afcb/html/ : SUCCESS in 6m 24s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/149/check/neutron-lbaasv2-dsvm-api/b34fece/ : SUCCESS in 1h 14m 50s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/149/check/neutron-lbaasv2-dsvm-py3x-api/528a893/ : SUCCESS in 1h 21m 29s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/149/check/neutron-lbaasv2-dsvm-scenario/9ed8971/ : SUCCESS in 53m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/149/check/neutron-lbaasv2-dsvm-py3x-scenario/1d1b02d/ : SUCCESS in 1h 43m 56s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/12/435612/149/check/openstack-tox-functional/a5d2979/ : SUCCESS in 7m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/149/check/openstack-tox-functional-py35/50e2eec/ : SUCCESS in 7m 08s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/149/check/octavia-v1-dsvm-scenario/e83cbc7/ : SUCCESS in 55m 52s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/149/check/octavia-v1-dsvm-py3x-scenario/48040a8/ : SUCCESS in 45m 06s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/149/check/octavia-v1-dsvm-scenario-multinode/4180f30/ : FAILURE in 1h 31m 21s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/149/check/octavia-v1-dsvm-py3x-scenario-multinode/1f62eb6/ : FAILURE in 1h 18m 57s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/149/check/octavia-v2-dsvm-noop-api/2b96259/ : SUCCESS in 38m 25s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/149/check/octavia-v2-dsvm-noop-py35-api/5b9857f/ : SUCCESS in 38m 47s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/149/check/octavia-v2-dsvm-scenario/d00d6f6/ : SUCCESS in 1h 23m 49s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/149/check/octavia-v2-dsvm-py35-scenario/b277c11/ : SUCCESS in 1h 18m 06s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/12/435612/149/check/octavia-v2-dsvm-scenario-centos-7/80f042a/ : SUCCESS in 2h 08m 03s\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/12/435612/149/check/octavia-v2-dsvm-scenario-ubuntu-bionic/4981fc8/ : FAILURE in 1h 58m 37s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/149/check/octavia-grenade/b309c5b/ : SUCCESS in 1h 02m 38s","accounts_in_message":[],"_revision_number":149},{"id":"5e619a1cc6dc20203def24c8339134217a0722fc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-11-05 22:02:14.000000000","message":"Uploaded patch set 150.","accounts_in_message":[],"_revision_number":150},{"id":"8b53f186782a5de88a61a9c240c02958b27cb50a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-11-05 22:44:07.000000000","message":"Uploaded patch set 151.","accounts_in_message":[],"_revision_number":151},{"id":"da0c79ae9696c071eb208d80fb466a95381b8fab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-06 00:27:54.000000000","message":"Patch Set 151: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/12/435612/151/check/test-release-openstack/ceb52eb/ : SUCCESS in 4m 02s\n- requirements-check http://logs.openstack.org/12/435612/151/check/requirements-check/8fa5329/ : SUCCESS in 3m 53s\n- openstack-tox-cover http://logs.openstack.org/12/435612/151/check/openstack-tox-cover/f572e26/ : FAILURE in 8m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/151/check/openstack-tox-lower-constraints/3192669/ : SUCCESS in 7m 51s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/151/check/openstack-tox-pep8/253651b/ : SUCCESS in 9m 08s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/151/check/openstack-tox-py27/fe58d72/ : SUCCESS in 6m 44s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/151/check/openstack-tox-py35/21bede2/ : SUCCESS in 7m 49s\n- openstack-tox-py36 http://logs.openstack.org/12/435612/151/check/openstack-tox-py36/5173d0c/ : SUCCESS in 7m 59s\n- openstack-tox-docs http://logs.openstack.org/12/435612/151/check/openstack-tox-docs/fedc1bf/html/ : SUCCESS in 9m 37s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/151/check/neutron-lbaasv2-dsvm-api/8c17313/ : SUCCESS in 1h 13m 54s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/151/check/neutron-lbaasv2-dsvm-py3x-api/9285d0a/ : SUCCESS in 1h 17m 15s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/151/check/neutron-lbaasv2-dsvm-scenario/1e527d3/ : SUCCESS in 1h 30m 30s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/151/check/neutron-lbaasv2-dsvm-py3x-scenario/b240920/ : SUCCESS in 1h 05m 02s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/12/435612/151/check/openstack-tox-functional/ed492a6/ : SUCCESS in 7m 44s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/151/check/openstack-tox-functional-py35/af1048b/ : SUCCESS in 7m 26s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/151/check/octavia-v1-dsvm-scenario/a5e8350/ : SUCCESS in 1h 00m 28s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/151/check/octavia-v1-dsvm-py3x-scenario/5f44af2/ : SUCCESS in 1h 15m 06s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/151/check/octavia-v1-dsvm-scenario-multinode/883838e/ : FAILURE in 1h 18m 35s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/151/check/octavia-v1-dsvm-py3x-scenario-multinode/a525a50/ : FAILURE in 1h 31m 21s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/151/check/octavia-v2-dsvm-noop-api/5f62a2b/ : SUCCESS in 38m 59s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/151/check/octavia-v2-dsvm-noop-py35-api/878c70d/ : SUCCESS in 42m 59s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/151/check/octavia-v2-dsvm-scenario/0105933/ : SUCCESS in 1h 35m 43s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/151/check/octavia-v2-dsvm-py35-scenario/0575967/ : SUCCESS in 1h 16m 48s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/12/435612/151/check/octavia-v2-dsvm-scenario-centos-7/a14c84f/ : SUCCESS in 1h 35m 21s\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/12/435612/151/check/octavia-v2-dsvm-scenario-ubuntu-bionic/170ddbf/ : SUCCESS in 1h 25m 02s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/151/check/octavia-grenade/25474f5/ : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":151},{"id":"c4576830459f0dbc514dbd1928aa070316b013ac","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-11-07 01:25:24.000000000","message":"Uploaded patch set 152.","accounts_in_message":[],"_revision_number":152},{"id":"b37436958b6993b2a932170c563edc3a815b8d65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 03:14:22.000000000","message":"Patch Set 152: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/12/435612/152/check/test-release-openstack/2a46c1d/ : SUCCESS in 3m 06s\n- requirements-check http://logs.openstack.org/12/435612/152/check/requirements-check/542bdca/ : SUCCESS in 3m 54s\n- openstack-tox-cover http://logs.openstack.org/12/435612/152/check/openstack-tox-cover/7ac4c0e/ : FAILURE in 5m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/152/check/openstack-tox-lower-constraints/c0480e5/ : SUCCESS in 6m 46s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/152/check/openstack-tox-pep8/b6646c9/ : SUCCESS in 9m 01s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/152/check/openstack-tox-py27/9af58ab/ : SUCCESS in 7m 23s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/152/check/openstack-tox-py35/4be9598/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/12/435612/152/check/openstack-tox-py36/a7878f1/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/12/435612/152/check/openstack-tox-docs/55afec1/html/ : SUCCESS in 6m 43s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/152/check/neutron-lbaasv2-dsvm-api/e30d9b5/ : SUCCESS in 1h 24m 10s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/152/check/neutron-lbaasv2-dsvm-py3x-api/6d15ce9/ : SUCCESS in 1h 08m 58s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/152/check/neutron-lbaasv2-dsvm-scenario/9b20e40/ : SUCCESS in 1h 27m 35s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/152/check/neutron-lbaasv2-dsvm-py3x-scenario/d9f4222/ : SUCCESS in 1h 01m 18s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/12/435612/152/check/openstack-tox-functional/b2138e9/ : SUCCESS in 5m 50s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/152/check/openstack-tox-functional-py35/b115519/ : SUCCESS in 8m 58s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/152/check/octavia-v1-dsvm-scenario/f328496/ : SUCCESS in 1h 04m 13s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/152/check/octavia-v1-dsvm-py3x-scenario/95514f9/ : SUCCESS in 1h 13m 48s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/12/435612/152/check/octavia-v1-dsvm-scenario-multinode/45aeaa5/ : FAILURE in 1h 18m 16s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/12/435612/152/check/octavia-v1-dsvm-py3x-scenario-multinode/ee2c150/ : FAILURE in 1h 11m 49s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/152/check/octavia-v2-dsvm-noop-api/2d59c64/ : SUCCESS in 37m 17s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/12/435612/152/check/octavia-v2-dsvm-noop-py35-api/0d92bc9/ : SUCCESS in 42m 49s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/152/check/octavia-v2-dsvm-scenario/1abf82e/ : SUCCESS in 1h 20m 30s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/12/435612/152/check/octavia-v2-dsvm-py35-scenario/25ce93c/ : SUCCESS in 1h 22m 13s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/12/435612/152/check/octavia-v2-dsvm-scenario-centos-7/947cc4f/ : SUCCESS in 1h 47m 02s\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/12/435612/152/check/octavia-v2-dsvm-scenario-ubuntu-bionic/7750c40/ : FAILURE in 1h 27m 51s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/152/check/octavia-grenade/36b5001/ : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":152},{"id":"041c9c0f4d73f7e67c700cd42ff818b55554f6cf","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-01-28 21:11:21.000000000","message":"Uploaded patch set 153.","accounts_in_message":[],"_revision_number":153},{"id":"d2626542c5bd60cb95ca3644beb72d87bca82f98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-28 23:59:57.000000000","message":"Patch Set 153: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/12/435612/153/check/test-release-openstack/a3be5fa/ : SUCCESS in 4m 33s\n- requirements-check http://logs.openstack.org/12/435612/153/check/requirements-check/69b71f5/ : SUCCESS in 4m 27s\n- openstack-tox-cover http://logs.openstack.org/12/435612/153/check/openstack-tox-cover/13bdcb6/ : FAILURE in 10m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/153/check/openstack-tox-lower-constraints/a45941c/ : SUCCESS in 8m 16s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/153/check/openstack-tox-pep8/4290f36/ : SUCCESS in 9m 26s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/153/check/openstack-tox-py27/5de992b/ : SUCCESS in 7m 24s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/153/check/openstack-tox-py35/166098c/ : SUCCESS in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/12/435612/153/check/openstack-tox-py36/d572303/ : SUCCESS in 7m 49s\n- openstack-tox-docs http://logs.openstack.org/12/435612/153/check/openstack-tox-docs/e3e8b40/html/ : SUCCESS in 7m 56s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/153/check/neutron-lbaasv2-dsvm-api/f3bcaa1/ : SUCCESS in 1h 12m 52s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/153/check/neutron-lbaasv2-dsvm-py3x-api/287650c/ : SUCCESS in 1h 20m 38s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/153/check/neutron-lbaasv2-dsvm-scenario/dc54ad1/ : SUCCESS in 1h 31m 05s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/153/check/neutron-lbaasv2-dsvm-py3x-scenario/6de8a1f/ : SUCCESS in 1h 38m 13s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/12/435612/153/check/openstack-tox-functional/fcd88a3/ : SUCCESS in 6m 24s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/153/check/openstack-tox-functional-py35/cd315b3/ : SUCCESS in 7m 05s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/153/check/octavia-v1-dsvm-scenario/92cde5d/ : SUCCESS in 1h 07m 14s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/153/check/octavia-v1-dsvm-py3x-scenario/c5e9c29/ : SUCCESS in 1h 02m 32s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/153/check/octavia-v2-dsvm-noop-api/4d6507f/ : SUCCESS in 43m 48s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/12/435612/153/check/octavia-v2-dsvm-noop-py2-api/06187dc/ : SUCCESS in 41m 26s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/153/check/octavia-v2-dsvm-scenario/d6bf239/ : SUCCESS in 1h 27m 51s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/12/435612/153/check/octavia-v2-dsvm-py2-scenario/81b2f8d/ : SUCCESS in 1h 32m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/12/435612/153/check/octavia-v2-dsvm-py2-scenario-centos-7/d262fba/ : SUCCESS in 2h 10m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/12/435612/153/check/octavia-v2-dsvm-scenario-ubuntu-bionic/25ff5fb/ : SUCCESS in 1h 47m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/12/435612/153/check/octavia-v2-dsvm-scenario-two-node/6225375/ : SUCCESS in 2h 01m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/12/435612/153/check/octavia-v2-dsvm-py2-scenario-two-node/e0e8e57/ : TIMED_OUT in 2h 35m 24s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/153/check/octavia-grenade/1a1c2d7/ : SUCCESS in 1h 10m 23s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/12/435612/153/check/octavia-v2-dsvm-tls-barbican/ed0e5df/ : SUCCESS in 44m 44s (non-voting)","accounts_in_message":[],"_revision_number":153},{"id":"dba85f46ca8b44a51d9ced08d5cb7056a8290189","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2019-01-29 05:43:38.000000000","message":"Patch Set 153:\n\n@Adam Harwell,\n\nThank you for rebasing the patch.","accounts_in_message":[],"_revision_number":153},{"id":"1c0dcdcecf33d3f674484584906dbb5484ec9cef","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-01-29 18:58:37.000000000","message":"Patch Set 153:\n\nno problem Dinesh!\nnote the coverage test is failing, but it\u0027s just because (as you might guess) the coverage is below our acceptable threshold, as I didn\u0027t finish the testing for this patch ever (time constraints, yada yada). It shouldn\u0027t be a problem. let me know in IRC if you run into any issues!","accounts_in_message":[],"_revision_number":153},{"id":"26999ac73621be05729df863c60d5cd80fe07740","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-01-30 20:47:31.000000000","message":"Uploaded patch set 154.","accounts_in_message":[],"_revision_number":154},{"id":"58101b9385b611f4ce00d100332a4d7a09494aa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-30 23:20:42.000000000","message":"Patch Set 154: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/12/435612/154/check/test-release-openstack/6980f76/ : SUCCESS in 3m 16s\n- requirements-check http://logs.openstack.org/12/435612/154/check/requirements-check/01e0729/ : SUCCESS in 3m 44s\n- openstack-tox-cover http://logs.openstack.org/12/435612/154/check/openstack-tox-cover/7f3cfba/cover/ : SUCCESS in 9m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/435612/154/check/openstack-tox-lower-constraints/b6a9a8f/ : SUCCESS in 8m 44s\n- openstack-tox-pep8 http://logs.openstack.org/12/435612/154/check/openstack-tox-pep8/0119de3/ : SUCCESS in 8m 41s\n- openstack-tox-py27 http://logs.openstack.org/12/435612/154/check/openstack-tox-py27/90c3c8a/ : SUCCESS in 7m 17s\n- openstack-tox-py35 http://logs.openstack.org/12/435612/154/check/openstack-tox-py35/664a0a3/ : SUCCESS in 5m 21s\n- openstack-tox-py36 http://logs.openstack.org/12/435612/154/check/openstack-tox-py36/9d27e7f/ : SUCCESS in 6m 26s\n- openstack-tox-docs http://logs.openstack.org/12/435612/154/check/openstack-tox-docs/3a8c5d8/html/ : SUCCESS in 6m 49s\n- build-openstack-releasenotes http://logs.openstack.org/12/435612/154/check/build-openstack-releasenotes/5dd9179/html/ : SUCCESS in 6m 01s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/12/435612/154/check/neutron-lbaasv2-dsvm-api/5d3dce3/ : SUCCESS in 1h 17m 21s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/12/435612/154/check/neutron-lbaasv2-dsvm-py3x-api/4cea3fc/ : SUCCESS in 1h 21m 56s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/12/435612/154/check/neutron-lbaasv2-dsvm-scenario/3e8e64f/ : SUCCESS in 1h 28m 05s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/12/435612/154/check/neutron-lbaasv2-dsvm-py3x-scenario/374d6ec/ : SUCCESS in 1h 21m 46s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/12/435612/154/check/openstack-tox-functional/f6afcba/ : SUCCESS in 8m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/12/435612/154/check/openstack-tox-functional-py35/97ebe23/ : SUCCESS in 8m 45s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/12/435612/154/check/octavia-v1-dsvm-scenario/d9196b5/ : SUCCESS in 1h 08m 29s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/12/435612/154/check/octavia-v1-dsvm-py3x-scenario/4c46bae/ : SUCCESS in 1h 03m 00s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/435612/154/check/octavia-v2-dsvm-noop-api/5fc11ff/ : SUCCESS in 48m 15s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/12/435612/154/check/octavia-v2-dsvm-noop-py2-api/3277450/ : SUCCESS in 40m 14s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/435612/154/check/octavia-v2-dsvm-scenario/23678b7/ : SUCCESS in 1h 35m 10s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/12/435612/154/check/octavia-v2-dsvm-py2-scenario/84e1181/ : SUCCESS in 1h 35m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/12/435612/154/check/octavia-v2-dsvm-py2-scenario-centos-7/b357958/ : SUCCESS in 2h 15m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/12/435612/154/check/octavia-v2-dsvm-scenario-ubuntu-bionic/f71a60e/ : SUCCESS in 1h 46m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/12/435612/154/check/octavia-v2-dsvm-scenario-two-node/5cf5ffe/ : SUCCESS in 1h 49m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/12/435612/154/check/octavia-v2-dsvm-py2-scenario-two-node/1c31ae3/ : SUCCESS in 1h 16m 22s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/435612/154/check/octavia-grenade/6ae26d2/ : SUCCESS in 1h 01m 18s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/12/435612/154/check/octavia-v2-dsvm-tls-barbican/475c37d/ : SUCCESS in 1h 07m 22s (non-voting)","accounts_in_message":[],"_revision_number":154},{"id":"0ea3a810d9d3aaa131a2be20e5715aff059f0c31","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-18 19:51:12.000000000","message":"Uploaded patch set 155.","accounts_in_message":[],"_revision_number":155},{"id":"7b418eb7549af7461bec80037c8167148d954bdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 22:25:40.000000000","message":"Patch Set 155: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b6c943da02b24a5f8d8f75a0440f9353 : SUCCESS in 5m 33s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8d25ffb04b3545d1b53e39f7bc0281b6 : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de20097eebeb4ad1a5c8510f0dde2b2a : SUCCESS in 6m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66f97418a280453a8e12df21fae1fcf3 : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97bdcc27f4eb413393c6d2b596c4301e : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/450e062220ec40a3b0dcbf8946f3c91d : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20e934c6e2024df4a27823712b1540f1 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9d1db3031c7415db911a464fd3df13b : FAILURE in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/732de070d0734aaea0c3864d84a6c1ba : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cc5bba5d9704c4d93c5810c2f62aeac : SUCCESS in 6m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1a3b698390d34b81bd787318fbd72c1e : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/036b50bc37704759b9da73d03f82e050 : SUCCESS in 7m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/795d5f4ed0074b3daa7de480693f63cf : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ad2eeb8f16454b6ca71aaa9c230c52ef : SUCCESS in 7m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f61123f698b04100867e0b0ac3d9c8ce : SUCCESS in 48m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ba5dbf10e8364c2ea40a6dee207cbcc7 : SUCCESS in 45m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c55c07d2799e4d7b9b668d5bcd50adf0 : SUCCESS in 1h 54m 08s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4f0611710a124205ab5b1d0a9cf676ec : SUCCESS in 1h 45m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1829a705a16c4eb3a5560209e938a1e4 : SUCCESS in 2h 27m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bfe4ae6389f544ba8c07a6e644266f13 : FAILURE in 36m 36s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c6ffac9c088b4e46aab08344dd047ac0 : SUCCESS in 2h 05m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e57cabb64ad942438f94c894355da9bb : FAILURE in 1h 04m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b5147ceb6f64c1482d69476fd4dddc7 : SUCCESS in 57m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b2dbedaf1b804807ac3db9ac32c2af99 : SUCCESS in 1h 53m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c546d074ba04d9da7568caa127d1d31 : SUCCESS in 1h 04m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7a4037075f7945ae933c007f5d69fc4d : SUCCESS in 1h 08m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/09334a6d700e4dbc9ef6f58788adadda : FAILURE in 1h 16m 44s (non-voting)","accounts_in_message":[],"_revision_number":155}],"current_revision_number":155,"current_revision":"0f69e8bf333fe79995c9f7cb960324273970691c","revisions":{"8aafc4bbbc909953afc56dab491bcca1f93d9069":{"kind":"REWORK","_number":1,"created":"2017-02-17 21:56:57.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/1"}}},"commit":{"parents":[{"commit":"88000edd5b04969774f5108dc2df0890aad563e1","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88000edd5b04969774f5108dc2df0890aad563e1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8aafc4bbbc909953afc56dab491bcca1f93d9069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8aafc4bbbc909953afc56dab491bcca1f93d9069"}]},"branch":"refs/heads/master"},"a1cc049f3b2867136eb18560db78e47ad30d4514":{"kind":"REWORK","_number":2,"created":"2017-02-18 00:32:24.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/2"}}},"commit":{"parents":[{"commit":"d34b93484aea5e66df29a64667da196d8c2116b9","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d34b93484aea5e66df29a64667da196d8c2116b9"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-18 00:31:45.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1cc049f3b2867136eb18560db78e47ad30d4514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1cc049f3b2867136eb18560db78e47ad30d4514"}]},"branch":"refs/heads/master"},"f086a7aa13c30217cbff9c314c18137a698a627a":{"kind":"REWORK","_number":3,"created":"2017-02-18 01:15:47.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/3"}}},"commit":{"parents":[{"commit":"d34b93484aea5e66df29a64667da196d8c2116b9","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d34b93484aea5e66df29a64667da196d8c2116b9"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-18 01:13:39.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f086a7aa13c30217cbff9c314c18137a698a627a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f086a7aa13c30217cbff9c314c18137a698a627a"}]},"branch":"refs/heads/master"},"f08986aa141073e71a04851fa88fabc2720bec59":{"kind":"REWORK","_number":4,"created":"2017-02-20 18:56:19.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/4"}}},"commit":{"parents":[{"commit":"716403097464f1f6998c921ba2b5827c09d8c4e4","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/716403097464f1f6998c921ba2b5827c09d8c4e4"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-20 18:56:02.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f08986aa141073e71a04851fa88fabc2720bec59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f08986aa141073e71a04851fa88fabc2720bec59"}]},"branch":"refs/heads/master"},"3deacf80f1f1ecd26ecd78be150c8dc1b8437075":{"kind":"REWORK","_number":5,"created":"2017-02-20 19:51:25.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/5"}}},"commit":{"parents":[{"commit":"716403097464f1f6998c921ba2b5827c09d8c4e4","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/716403097464f1f6998c921ba2b5827c09d8c4e4"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-20 19:51:09.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3deacf80f1f1ecd26ecd78be150c8dc1b8437075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3deacf80f1f1ecd26ecd78be150c8dc1b8437075"}]},"branch":"refs/heads/master"},"876e3189a2b00b0ede88ae551c255a5a0c811337":{"kind":"REWORK","_number":6,"created":"2017-02-20 21:39:17.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/6"}}},"commit":{"parents":[{"commit":"716403097464f1f6998c921ba2b5827c09d8c4e4","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/716403097464f1f6998c921ba2b5827c09d8c4e4"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-20 21:39:00.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/876e3189a2b00b0ede88ae551c255a5a0c811337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/876e3189a2b00b0ede88ae551c255a5a0c811337"}]},"branch":"refs/heads/master"},"ffa22b59317a3f42ea7b47ca56dc805646b6b4fc":{"kind":"REWORK","_number":7,"created":"2017-02-21 00:54:07.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/7"}}},"commit":{"parents":[{"commit":"716403097464f1f6998c921ba2b5827c09d8c4e4","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/716403097464f1f6998c921ba2b5827c09d8c4e4"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-20 21:52:34.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffa22b59317a3f42ea7b47ca56dc805646b6b4fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffa22b59317a3f42ea7b47ca56dc805646b6b4fc"}]},"branch":"refs/heads/master"},"04a755a5976b079b4e95449a1fb5f472c538af9a":{"kind":"REWORK","_number":8,"created":"2017-02-23 00:25:42.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/8"}}},"commit":{"parents":[{"commit":"716403097464f1f6998c921ba2b5827c09d8c4e4","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/716403097464f1f6998c921ba2b5827c09d8c4e4"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-23 00:25:18.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04a755a5976b079b4e95449a1fb5f472c538af9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04a755a5976b079b4e95449a1fb5f472c538af9a"}]},"branch":"refs/heads/master"},"a8b374a970197d21ae54e7166ad0a8c7f9afed89":{"kind":"REWORK","_number":9,"created":"2017-02-23 00:28:26.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/9"}}},"commit":{"parents":[{"commit":"716403097464f1f6998c921ba2b5827c09d8c4e4","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/716403097464f1f6998c921ba2b5827c09d8c4e4"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-23 00:28:16.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8b374a970197d21ae54e7166ad0a8c7f9afed89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8b374a970197d21ae54e7166ad0a8c7f9afed89"}]},"branch":"refs/heads/master"},"517abae8855df1aa54dbe5401e035be72fdb38db":{"kind":"REWORK","_number":10,"created":"2017-02-23 10:50:45.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/10"}}},"commit":{"parents":[{"commit":"716403097464f1f6998c921ba2b5827c09d8c4e4","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/716403097464f1f6998c921ba2b5827c09d8c4e4"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-23 10:50:26.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/517abae8855df1aa54dbe5401e035be72fdb38db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/517abae8855df1aa54dbe5401e035be72fdb38db"}]},"branch":"refs/heads/master"},"9aa499c253108115575c0bdffd047b7e0b7da095":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2017-02-28 01:06:38.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/11"}}},"commit":{"parents":[{"commit":"881db2fad554d82290d2495dae44ea75354cd177","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/881db2fad554d82290d2495dae44ea75354cd177"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-28 01:06:38.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9aa499c253108115575c0bdffd047b7e0b7da095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9aa499c253108115575c0bdffd047b7e0b7da095"}]},"branch":"refs/heads/master"},"0934d250d65439febe2fa57a25af04a6096123f3":{"kind":"REWORK","_number":12,"created":"2017-03-01 01:53:23.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/12"}}},"commit":{"parents":[{"commit":"881db2fad554d82290d2495dae44ea75354cd177","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/881db2fad554d82290d2495dae44ea75354cd177"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-28 22:39:26.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0934d250d65439febe2fa57a25af04a6096123f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0934d250d65439febe2fa57a25af04a6096123f3"}]},"branch":"refs/heads/master"},"83ac0365dea04109eff2677effa5f3802dd0fbc4":{"kind":"REWORK","_number":13,"created":"2017-03-07 13:28:11.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/13"}}},"commit":{"parents":[{"commit":"a9a9e5d6d9d4956754821e8559dd918af822f5a0","subject":"Allow to create vip in lb-mgmt-net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9a9e5d6d9d4956754821e8559dd918af822f5a0"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-03-07 13:27:20.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83ac0365dea04109eff2677effa5f3802dd0fbc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83ac0365dea04109eff2677effa5f3802dd0fbc4"}]},"branch":"refs/heads/master"},"fc67b5d778e5b04e194bb4fea404e2883f3cb461":{"kind":"REWORK","_number":14,"created":"2017-03-17 02:09:18.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/14"}}},"commit":{"parents":[{"commit":"114c6428f1eebbb67f5311c054411feba7186124","subject":"Merge \"Adds a new feature to limit the amphora build rate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/114c6428f1eebbb67f5311c054411feba7186124"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-03-17 02:08:57.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc67b5d778e5b04e194bb4fea404e2883f3cb461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc67b5d778e5b04e194bb4fea404e2883f3cb461"}]},"branch":"refs/heads/master"},"cb27feea1bf65c5b865083c2cb24d71badae0d1c":{"kind":"REWORK","_number":15,"created":"2017-03-18 06:17:41.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/15"}}},"commit":{"parents":[{"commit":"114c6428f1eebbb67f5311c054411feba7186124","subject":"Merge \"Adds a new feature to limit the amphora build rate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/114c6428f1eebbb67f5311c054411feba7186124"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-03-18 06:17:35.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb27feea1bf65c5b865083c2cb24d71badae0d1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb27feea1bf65c5b865083c2cb24d71badae0d1c"}]},"branch":"refs/heads/master"},"35eb352d0e0228a4b405001fcf5b739fb1d3fc7d":{"kind":"REWORK","_number":16,"created":"2017-04-11 19:30:33.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/16"}}},"commit":{"parents":[{"commit":"87bab9bf153237e16420f7fcfbf4b3880e47e40b","subject":"Merge \"Failover logic doesn\u0027t handle anti-affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87bab9bf153237e16420f7fcfbf4b3880e47e40b"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-04-11 19:17:17.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35eb352d0e0228a4b405001fcf5b739fb1d3fc7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35eb352d0e0228a4b405001fcf5b739fb1d3fc7d"}]},"branch":"refs/heads/master"},"6c8c48702bc36dc4991125d51a724fa44f1729fb":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2017-05-02 09:16:13.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/17"}}},"commit":{"parents":[{"commit":"a59ec9b9bd5d442dd3140974dea6c93e10b0ba9a","subject":"Merge \"Fix a bug where ports may not be deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a59ec9b9bd5d442dd3140974dea6c93e10b0ba9a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-05-02 09:16:13.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c8c48702bc36dc4991125d51a724fa44f1729fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c8c48702bc36dc4991125d51a724fa44f1729fb"}]},"branch":"refs/heads/master"},"d957234b7061387a663212ccc10afe5932ed35f7":{"kind":"REWORK","_number":18,"created":"2017-05-05 07:25:29.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/18"}}},"commit":{"parents":[{"commit":"7467aa28d4070867d47fcc35b8ce8dac1cd12d88","subject":"Merge \"Add some hacking checks borrowed from neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7467aa28d4070867d47fcc35b8ce8dac1cd12d88"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-05-05 07:25:20.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d957234b7061387a663212ccc10afe5932ed35f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d957234b7061387a663212ccc10afe5932ed35f7"}]},"branch":"refs/heads/master"},"cd77a0258b2b5b7d1f4674bb71d5913ed5a34f38":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2017-05-05 21:35:06.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/19"}}},"commit":{"parents":[{"commit":"8fdcb88deeebf96aea088ef7811be305667fd77d","subject":"Merge \"Don\u0027t show deleted objects in v2 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fdcb88deeebf96aea088ef7811be305667fd77d"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-05-05 21:35:06.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd77a0258b2b5b7d1f4674bb71d5913ed5a34f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd77a0258b2b5b7d1f4674bb71d5913ed5a34f38"}]},"branch":"refs/heads/master"},"4f6bc0f406c7771209a4da36aa0b599b3a92e770":{"kind":"REWORK","_number":20,"created":"2017-05-10 05:10:41.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/20"}}},"commit":{"parents":[{"commit":"da2f3d4a7ccfe45afe6f7337d177a1f31c9b5a3d","subject":"Merge \"Implement sorting and pagination for octavia\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da2f3d4a7ccfe45afe6f7337d177a1f31c9b5a3d"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-05-10 05:09:45.000000000","tz":-240},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nAlso tweak nova driver to allow selecting a management network from the\nnetworks provisioned for an amp if there are none configured (some cloud\nproviders require using only one default network).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f6bc0f406c7771209a4da36aa0b599b3a92e770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4f6bc0f406c7771209a4da36aa0b599b3a92e770"}]},"branch":"refs/heads/master"},"e0e34610142443a6edf6637b816641f2c8d52f15":{"kind":"REWORK","_number":21,"created":"2017-05-10 06:08:40.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/21"}}},"commit":{"parents":[{"commit":"793eb48d695c5e852fcc6f1850b69f56962bd96a","subject":"Allow specifying empty amp_boot_network_list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/793eb48d695c5e852fcc6f1850b69f56962bd96a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-05-10 06:08:26.000000000","tz":-240},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0e34610142443a6edf6637b816641f2c8d52f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0e34610142443a6edf6637b816641f2c8d52f15"}]},"branch":"refs/heads/master"},"ac8ae3d69edff7fe589c29d10eeac789decd85b8":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2017-05-10 13:55:03.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/22"}}},"commit":{"parents":[{"commit":"7feffccd1e566cc1fe2453b817acd66c91a45a28","subject":"CalculateAmphoraDelta networks based on vrrp_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7feffccd1e566cc1fe2453b817acd66c91a45a28"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-05-10 13:55:03.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac8ae3d69edff7fe589c29d10eeac789decd85b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac8ae3d69edff7fe589c29d10eeac789decd85b8"}]},"branch":"refs/heads/master"},"5ed9b0c5216e9cef9cfd2a5b62f2894d7eb1e14b":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2017-05-11 17:54:58.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/23"}}},"commit":{"parents":[{"commit":"4eb13a2d121a8f04b8011d098f417088b55c306c","subject":"Merge \"Trivial fix typos in documents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4eb13a2d121a8f04b8011d098f417088b55c306c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-05-11 17:54:58.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed9b0c5216e9cef9cfd2a5b62f2894d7eb1e14b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed9b0c5216e9cef9cfd2a5b62f2894d7eb1e14b"}]},"branch":"refs/heads/master"},"48e96c806df51d41ec334d0442c2aa32ed525a74":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2017-06-02 19:14:30.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/24"}}},"commit":{"parents":[{"commit":"29219e43453f0eecacebd7d3bf892519da7cc35c","subject":"Merge \"Don\u0027t leave LBs in PENDING_DELETE after refusing to cascade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29219e43453f0eecacebd7d3bf892519da7cc35c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-06-02 19:14:30.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48e96c806df51d41ec334d0442c2aa32ed525a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48e96c806df51d41ec334d0442c2aa32ed525a74"}]},"branch":"refs/heads/master"},"945f7259798f5e774475f2e7f8a1d4b6d6adf06a":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2017-06-06 00:53:48.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/25"}}},"commit":{"parents":[{"commit":"fcf6d1e59b2634b74edb79dc34dfa30a099da305","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcf6d1e59b2634b74edb79dc34dfa30a099da305"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-06-06 00:53:48.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAlso add a debug message in the heartbeat daemon to show heartbeats as\nthey are received.\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/945f7259798f5e774475f2e7f8a1d4b6d6adf06a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/945f7259798f5e774475f2e7f8a1d4b6d6adf06a"}]},"branch":"refs/heads/master"},"1e1f5dcae1854f31c8385acb6ff44e48260654fc":{"kind":"REWORK","_number":26,"created":"2017-06-12 23:53:23.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/26"}}},"commit":{"parents":[{"commit":"c8615b7ec447ea9e801bc244f642f0b29b75df59","subject":"Merge \"Fix an issue with failover on VIP net without DHCP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8615b7ec447ea9e801bc244f642f0b29b75df59"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-06-12 23:52:52.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e1f5dcae1854f31c8385acb6ff44e48260654fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e1f5dcae1854f31c8385acb6ff44e48260654fc"}]},"branch":"refs/heads/master"},"d7bec6f3cfd74cd342fc36bd9000dd5f78351111":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2017-06-29 01:49:47.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/27"}}},"commit":{"parents":[{"commit":"dac945260d0f04c4a99a0c0a003845cae8fe3f02","subject":"Merge \"Change default vagrant box to xenial64(16.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dac945260d0f04c4a99a0c0a003845cae8fe3f02"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-06-29 01:49:47.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7bec6f3cfd74cd342fc36bd9000dd5f78351111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7bec6f3cfd74cd342fc36bd9000dd5f78351111"}]},"branch":"refs/heads/master"},"84c517d5c6ab97fb6f1e05873578e08c7e3a2327":{"kind":"REWORK","_number":28,"created":"2017-06-29 21:31:10.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/28"}}},"commit":{"parents":[{"commit":"a7bdf0351f52b37f1eab250444e65ed8190a2495","subject":"Update api-ref for openstackdocstheme 1.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a7bdf0351f52b37f1eab250444e65ed8190a2495"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-06-29 21:31:01.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84c517d5c6ab97fb6f1e05873578e08c7e3a2327"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84c517d5c6ab97fb6f1e05873578e08c7e3a2327"}]},"branch":"refs/heads/master"},"30baa73a623da2c398ff58020a79cd106049c079":{"kind":"REWORK","_number":29,"created":"2017-06-29 21:31:52.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/29"}}},"commit":{"parents":[{"commit":"a7bdf0351f52b37f1eab250444e65ed8190a2495","subject":"Update api-ref for openstackdocstheme 1.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a7bdf0351f52b37f1eab250444e65ed8190a2495"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-06-29 21:31:43.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30baa73a623da2c398ff58020a79cd106049c079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30baa73a623da2c398ff58020a79cd106049c079"}]},"branch":"refs/heads/master"},"8b89c6a39334e1004c8c8a186883a0d0433b1a9d":{"kind":"REWORK","_number":30,"created":"2017-06-30 01:43:34.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/30"}}},"commit":{"parents":[{"commit":"a7bdf0351f52b37f1eab250444e65ed8190a2495","subject":"Update api-ref for openstackdocstheme 1.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a7bdf0351f52b37f1eab250444e65ed8190a2495"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-06-30 01:42:51.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b89c6a39334e1004c8c8a186883a0d0433b1a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b89c6a39334e1004c8c8a186883a0d0433b1a9d"}]},"branch":"refs/heads/master"},"0b2c81f0e714653ab301260f38bbdfb46048493c":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2017-07-07 22:05:04.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/31"}}},"commit":{"parents":[{"commit":"c0f81bdb120b7a18eb22a6059869acdb4b56c989","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0f81bdb120b7a18eb22a6059869acdb4b56c989"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-07-07 22:02:30.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b2c81f0e714653ab301260f38bbdfb46048493c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b2c81f0e714653ab301260f38bbdfb46048493c"}]},"branch":"refs/heads/master"},"ba7a1afd7476b9a35acf5ebd97f1977da8241161":{"kind":"REWORK","_number":32,"created":"2017-08-02 09:30:05.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/32"}}},"commit":{"parents":[{"commit":"64feb374a3bbb82c8ec3c7a43bdae0acaa48dc70","subject":"Merge \"Properly handle more states from HAProxy health messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64feb374a3bbb82c8ec3c7a43bdae0acaa48dc70"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-02 09:29:23.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba7a1afd7476b9a35acf5ebd97f1977da8241161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba7a1afd7476b9a35acf5ebd97f1977da8241161"}]},"branch":"refs/heads/master"},"d40c23a8fc6627214e203ea0b6027d0d6e363f1a":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2017-08-02 21:03:50.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/33"}}},"commit":{"parents":[{"commit":"5c74844ffed2f1a094a91e5fdb02aceba6b97ea7","subject":"Merge \"Barbicanclient is refactoring, this fixes one of our bad tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c74844ffed2f1a094a91e5fdb02aceba6b97ea7"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-02 21:03:49.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d40c23a8fc6627214e203ea0b6027d0d6e363f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d40c23a8fc6627214e203ea0b6027d0d6e363f1a"}]},"branch":"refs/heads/master"},"8181d1f5b70186293a2c2494e6ac0626be085079":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2017-08-08 00:56:53.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/34"}}},"commit":{"parents":[{"commit":"c87ec394c1f04b55f3d24099fd14a4ea16eff237","subject":"Merge \"Add allocate vip port when create loadbalancer in server side\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c87ec394c1f04b55f3d24099fd14a4ea16eff237"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-08 00:56:53.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8181d1f5b70186293a2c2494e6ac0626be085079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8181d1f5b70186293a2c2494e6ac0626be085079"}]},"branch":"refs/heads/master"},"6a57e2a3d4627ea2c1d2f8acc28f760e2463bf3f":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2017-08-16 20:17:58.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/35"}}},"commit":{"parents":[{"commit":"1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6","subject":"Merge \"Fix health monitor DB locking.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-16 20:17:58.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a57e2a3d4627ea2c1d2f8acc28f760e2463bf3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a57e2a3d4627ea2c1d2f8acc28f760e2463bf3f"}]},"branch":"refs/heads/master"},"47ee0f004e32d1d65fa021f2c32e3cf3ac495a78":{"kind":"REWORK","_number":36,"created":"2017-08-17 23:39:45.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/36"}}},"commit":{"parents":[{"commit":"1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6","subject":"Merge \"Fix health monitor DB locking.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-17 23:39:12.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47ee0f004e32d1d65fa021f2c32e3cf3ac495a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47ee0f004e32d1d65fa021f2c32e3cf3ac495a78"}]},"branch":"refs/heads/master"},"c4725ea7d45d97d6f08718fc1f619c7608ebcbd2":{"kind":"REWORK","_number":37,"created":"2017-08-18 01:38:59.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/37"}}},"commit":{"parents":[{"commit":"1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6","subject":"Merge \"Fix health monitor DB locking.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-18 01:36:30.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4725ea7d45d97d6f08718fc1f619c7608ebcbd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4725ea7d45d97d6f08718fc1f619c7608ebcbd2"}]},"branch":"refs/heads/master"},"fe860d2272f7d81b0275aff48fa74c8d5240919e":{"kind":"REWORK","_number":38,"created":"2017-08-19 01:52:14.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/38"}}},"commit":{"parents":[{"commit":"1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6","subject":"Merge \"Fix health monitor DB locking.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-19 01:51:52.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fe860d2272f7d81b0275aff48fa74c8d5240919e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fe860d2272f7d81b0275aff48fa74c8d5240919e"}]},"branch":"refs/heads/master"},"6c0e06e374f76d19a73acfbebadca736c6f2c5fc":{"kind":"REWORK","_number":39,"created":"2017-08-19 03:34:37.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/39"}}},"commit":{"parents":[{"commit":"1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6","subject":"Merge \"Fix health monitor DB locking.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-19 03:34:28.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c0e06e374f76d19a73acfbebadca736c6f2c5fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c0e06e374f76d19a73acfbebadca736c6f2c5fc"}]},"branch":"refs/heads/master"},"25fa4d01174c06380f9e1f3882b215b67339742b":{"kind":"REWORK","_number":40,"created":"2017-08-22 01:34:47.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/40"}}},"commit":{"parents":[{"commit":"1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6","subject":"Merge \"Fix health monitor DB locking.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-22 01:34:37.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25fa4d01174c06380f9e1f3882b215b67339742b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25fa4d01174c06380f9e1f3882b215b67339742b"}]},"branch":"refs/heads/master"},"e5e60952978626e2f8a7b23ef5c005c7d6dd6967":{"kind":"REWORK","_number":41,"created":"2017-08-22 06:03:07.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/41"}}},"commit":{"parents":[{"commit":"1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6","subject":"Merge \"Fix health monitor DB locking.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e97e2f13adcbbf4aa9098dc5a3cca4b4e44afa6"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-22 06:02:16.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5e60952978626e2f8a7b23ef5c005c7d6dd6967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5e60952978626e2f8a7b23ef5c005c7d6dd6967"}]},"branch":"refs/heads/master"},"30ff00ee6ca1991f1fcc8ca16619b2b3e3b3646c":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2017-08-22 06:08:08.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/42"}}},"commit":{"parents":[{"commit":"bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc","subject":"Merge \"Correct diskimage-create sshd removal bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-22 06:07:46.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30ff00ee6ca1991f1fcc8ca16619b2b3e3b3646c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30ff00ee6ca1991f1fcc8ca16619b2b3e3b3646c"}]},"branch":"refs/heads/master"},"b19eb815c9b69c8f97c63a075344b74fd427ee8e":{"kind":"REWORK","_number":43,"created":"2017-08-22 22:21:30.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/43"}}},"commit":{"parents":[{"commit":"bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc","subject":"Merge \"Correct diskimage-create sshd removal bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-22 22:21:03.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b19eb815c9b69c8f97c63a075344b74fd427ee8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b19eb815c9b69c8f97c63a075344b74fd427ee8e"}]},"branch":"refs/heads/master"},"be62349374b1830f4a660f4b0328b135f2d37b2b":{"kind":"REWORK","_number":44,"created":"2017-08-23 00:55:25.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/44"}}},"commit":{"parents":[{"commit":"bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc","subject":"Merge \"Correct diskimage-create sshd removal bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-23 00:55:08.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be62349374b1830f4a660f4b0328b135f2d37b2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be62349374b1830f4a660f4b0328b135f2d37b2b"}]},"branch":"refs/heads/master"},"8e3767203504d7f39101c5be19ccf8647f129f2b":{"kind":"REWORK","_number":45,"created":"2017-08-24 23:47:03.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/45"}}},"commit":{"parents":[{"commit":"bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc","subject":"Merge \"Correct diskimage-create sshd removal bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-24 23:46:52.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e3767203504d7f39101c5be19ccf8647f129f2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e3767203504d7f39101c5be19ccf8647f129f2b"}]},"branch":"refs/heads/master"},"df2f74781d08a24f4831e56ee539c8a176a92edb":{"kind":"REWORK","_number":46,"created":"2017-08-25 00:53:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/435612/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/46"}}},"commit":{"parents":[{"commit":"bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc","subject":"Merge \"Correct diskimage-create sshd removal bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-08-25 00:53:11.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df2f74781d08a24f4831e56ee539c8a176a92edb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df2f74781d08a24f4831e56ee539c8a176a92edb"}]},"branch":"refs/heads/master"},"c194ae16d452200e44e3ce91551bb487ea02244e":{"kind":"REWORK","_number":47,"created":"2017-08-29 23:03:45.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/47"}}},"commit":{"parents":[{"commit":"bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc","subject":"Merge \"Correct diskimage-create sshd removal bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-29 23:03:35.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c194ae16d452200e44e3ce91551bb487ea02244e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c194ae16d452200e44e3ce91551bb487ea02244e"}]},"branch":"refs/heads/master"},"260dcd9aa13f4f742d712f8e7c157bfa64a64f6e":{"kind":"REWORK","_number":48,"created":"2017-08-29 23:18:03.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/48"}}},"commit":{"parents":[{"commit":"bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc","subject":"Merge \"Correct diskimage-create sshd removal bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-08-29 23:17:54.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/260dcd9aa13f4f742d712f8e7c157bfa64a64f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/260dcd9aa13f4f742d712f8e7c157bfa64a64f6e"}]},"branch":"refs/heads/master"},"12a322de68fcbaaba8bf3b1fd9d4aeaf697ae2e2":{"kind":"REWORK","_number":49,"created":"2017-08-30 00:02:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/435612/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/49"}}},"commit":{"parents":[{"commit":"bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc","subject":"Merge \"Correct diskimage-create sshd removal bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-08-30 00:02:11.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12a322de68fcbaaba8bf3b1fd9d4aeaf697ae2e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12a322de68fcbaaba8bf3b1fd9d4aeaf697ae2e2"}]},"branch":"refs/heads/master"},"48d2dbb967d1553f6b7f274359b0c410296f50bb":{"kind":"REWORK","_number":50,"created":"2017-09-08 22:35:40.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/50"}}},"commit":{"parents":[{"commit":"bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc","subject":"Merge \"Correct diskimage-create sshd removal bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bacf5ff193b69ae4d29c42a9ff24a2e1a459d9dc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-08 22:34:42.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48d2dbb967d1553f6b7f274359b0c410296f50bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48d2dbb967d1553f6b7f274359b0c410296f50bb"}]},"branch":"refs/heads/master"},"86dea23acf1441d8520e00116a9125445ad8f6ed":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2017-09-08 22:36:22.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/51"}}},"commit":{"parents":[{"commit":"2975e7a159247e8372b0ea502e24a6d4bd990fef","subject":"Add \u0027futures\u0027 to requirements, as we use it multiple places","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2975e7a159247e8372b0ea502e24a6d4bd990fef"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-08 22:36:22.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/86dea23acf1441d8520e00116a9125445ad8f6ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/86dea23acf1441d8520e00116a9125445ad8f6ed"}]},"branch":"refs/heads/master"},"a12fd8bf04af1954f4b56741ad668bb1de0a8dc3":{"kind":"REWORK","_number":52,"created":"2017-09-14 19:58:40.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/52"}}},"commit":{"parents":[{"commit":"ec5b04ea8b0c9be69c1701c53bc8d7ee7de16a6e","subject":"Merge \"Whitespace fixes and simplification of keepalived template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec5b04ea8b0c9be69c1701c53bc8d7ee7de16a6e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-14 19:58:31.000000000","tz":-360},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a12fd8bf04af1954f4b56741ad668bb1de0a8dc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a12fd8bf04af1954f4b56741ad668bb1de0a8dc3"}]},"branch":"refs/heads/master"},"b8fa43e671d03da696587dec35cd2664edc31dff":{"kind":"REWORK","_number":53,"created":"2017-09-18 23:07:54.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/53"}}},"commit":{"parents":[{"commit":"b8e0f5e763da79154490a2726d4ece8134511d21","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8e0f5e763da79154490a2726d4ece8134511d21"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-18 23:07:42.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8fa43e671d03da696587dec35cd2664edc31dff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8fa43e671d03da696587dec35cd2664edc31dff"}]},"branch":"refs/heads/master"},"b2b33e21e2487ef7d18d5c386805d1bbb666ab89":{"kind":"REWORK","_number":54,"created":"2017-09-21 09:39:37.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/54"}}},"commit":{"parents":[{"commit":"b8e0f5e763da79154490a2726d4ece8134511d21","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8e0f5e763da79154490a2726d4ece8134511d21"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-21 09:19:25.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b2b33e21e2487ef7d18d5c386805d1bbb666ab89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b2b33e21e2487ef7d18d5c386805d1bbb666ab89"}]},"branch":"refs/heads/master"},"95f33203708038cc637d0c3213f3fe5ba92f948d":{"kind":"REWORK","_number":55,"created":"2017-09-22 22:09:07.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/55"}}},"commit":{"parents":[{"commit":"b8e0f5e763da79154490a2726d4ece8134511d21","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8e0f5e763da79154490a2726d4ece8134511d21"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-22 22:08:54.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95f33203708038cc637d0c3213f3fe5ba92f948d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95f33203708038cc637d0c3213f3fe5ba92f948d"}]},"branch":"refs/heads/master"},"619e505c090d5d398b2f6298ccdc68ef1026cdee":{"kind":"REWORK","_number":56,"created":"2017-09-25 19:00:49.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/56"}}},"commit":{"parents":[{"commit":"b8e0f5e763da79154490a2726d4ece8134511d21","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8e0f5e763da79154490a2726d4ece8134511d21"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-25 19:00:38.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/619e505c090d5d398b2f6298ccdc68ef1026cdee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/619e505c090d5d398b2f6298ccdc68ef1026cdee"}]},"branch":"refs/heads/master"},"b961339f8da3bbc7ef761269ebe0bea1bd130fb0":{"kind":"REWORK","_number":57,"created":"2017-09-27 20:03:38.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/57"}}},"commit":{"parents":[{"commit":"b8e0f5e763da79154490a2726d4ece8134511d21","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8e0f5e763da79154490a2726d4ece8134511d21"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-27 20:03:27.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b961339f8da3bbc7ef761269ebe0bea1bd130fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b961339f8da3bbc7ef761269ebe0bea1bd130fb0"}]},"branch":"refs/heads/master"},"ba1b666b621499eae4411807405350ab6637199c":{"kind":"REWORK","_number":58,"created":"2017-09-27 20:45:37.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/58"}}},"commit":{"parents":[{"commit":"b8e0f5e763da79154490a2726d4ece8134511d21","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8e0f5e763da79154490a2726d4ece8134511d21"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-27 20:45:27.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba1b666b621499eae4411807405350ab6637199c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba1b666b621499eae4411807405350ab6637199c"}]},"branch":"refs/heads/master"},"872bbfebe48d76c7d55fc83bb100ebe262feed6c":{"kind":"REWORK","_number":59,"created":"2017-09-27 23:11:04.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/59"}}},"commit":{"parents":[{"commit":"b8e0f5e763da79154490a2726d4ece8134511d21","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8e0f5e763da79154490a2726d4ece8134511d21"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-27 23:10:51.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/872bbfebe48d76c7d55fc83bb100ebe262feed6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/872bbfebe48d76c7d55fc83bb100ebe262feed6c"}]},"branch":"refs/heads/master"},"0e61845d2b618b727cce07f5ec7cf0b54376dcf0":{"kind":"REWORK","_number":60,"created":"2017-09-27 23:57:02.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/60"}}},"commit":{"parents":[{"commit":"b8e0f5e763da79154490a2726d4ece8134511d21","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8e0f5e763da79154490a2726d4ece8134511d21"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-27 23:56:49.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e61845d2b618b727cce07f5ec7cf0b54376dcf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e61845d2b618b727cce07f5ec7cf0b54376dcf0"}]},"branch":"refs/heads/master"},"954c25a334c759d62b9b857e6e4f55104ad7a412":{"kind":"REWORK","_number":61,"created":"2017-09-28 00:10:28.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/61"}}},"commit":{"parents":[{"commit":"b8e0f5e763da79154490a2726d4ece8134511d21","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8e0f5e763da79154490a2726d4ece8134511d21"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-28 00:10:18.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/954c25a334c759d62b9b857e6e4f55104ad7a412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/954c25a334c759d62b9b857e6e4f55104ad7a412"}]},"branch":"refs/heads/master"},"492866edacb928ff9d07cc7665048d8822854ef7":{"kind":"REWORK","_number":62,"created":"2017-09-28 00:30:32.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/62"}}},"commit":{"parents":[{"commit":"b8e0f5e763da79154490a2726d4ece8134511d21","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8e0f5e763da79154490a2726d4ece8134511d21"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-09-28 00:30:21.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/492866edacb928ff9d07cc7665048d8822854ef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/492866edacb928ff9d07cc7665048d8822854ef7"}]},"branch":"refs/heads/master"},"c27fe98788917542612e4f43791c63e6bcf40d9e":{"kind":"REWORK","_number":63,"created":"2017-10-13 03:03:50.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/63"}}},"commit":{"parents":[{"commit":"40484cc5c67cfad767bcda224ae4e08bbd2abdba","subject":"L7 policy API-REF update for result codes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40484cc5c67cfad767bcda224ae4e08bbd2abdba"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-10-13 03:03:31.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c27fe98788917542612e4f43791c63e6bcf40d9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c27fe98788917542612e4f43791c63e6bcf40d9e"}]},"branch":"refs/heads/master"},"3b069302ce9259928062b7a61ed3396314b4e799":{"kind":"REWORK","_number":64,"created":"2017-10-13 18:35:10.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/64"}}},"commit":{"parents":[{"commit":"40484cc5c67cfad767bcda224ae4e08bbd2abdba","subject":"L7 policy API-REF update for result codes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40484cc5c67cfad767bcda224ae4e08bbd2abdba"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-10-13 18:35:06.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b069302ce9259928062b7a61ed3396314b4e799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b069302ce9259928062b7a61ed3396314b4e799"}]},"branch":"refs/heads/master"},"5e90f21ddf3f0e8cb5eed0a0e41c553651ca6e8f":{"kind":"TRIVIAL_REBASE","_number":65,"created":"2017-10-16 17:52:05.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/65"}}},"commit":{"parents":[{"commit":"859bb3bae35fb26e3305d4ee691b4d5eba99d1ff","subject":"Merge \"Allow PUT to /pools/\u003cid\u003e/members to batch update members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/859bb3bae35fb26e3305d4ee691b4d5eba99d1ff"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-10-16 17:51:56.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e90f21ddf3f0e8cb5eed0a0e41c553651ca6e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e90f21ddf3f0e8cb5eed0a0e41c553651ca6e8f"}]},"branch":"refs/heads/master"},"6e6a19526186247b93cfd26ae8dd6aac90eb1e70":{"kind":"TRIVIAL_REBASE","_number":66,"created":"2017-10-31 23:26:16.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/66"}}},"commit":{"parents":[{"commit":"1c0bf468d5ab3de82c9ce9f6c1ca2436ecc9b424","subject":"Add cached_zone to the amphora record","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c0bf468d5ab3de82c9ce9f6c1ca2436ecc9b424"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-10-31 23:26:16.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e6a19526186247b93cfd26ae8dd6aac90eb1e70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e6a19526186247b93cfd26ae8dd6aac90eb1e70"}]},"branch":"refs/heads/master"},"ec8b1e9e27def32f3ab9facaa026896f4d9cd246":{"kind":"TRIVIAL_REBASE","_number":67,"created":"2017-11-02 01:21:24.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/67"}}},"commit":{"parents":[{"commit":"504cb6c682e4779b5889c0eb68705d0ab12e2c81","subject":"Merge \"Add cached_zone to the amphora record\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/504cb6c682e4779b5889c0eb68705d0ab12e2c81"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-11-02 01:21:24.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec8b1e9e27def32f3ab9facaa026896f4d9cd246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec8b1e9e27def32f3ab9facaa026896f4d9cd246"}]},"branch":"refs/heads/master"},"ddb549289c417c767637cdd8e21eaa9e78a4f5e6":{"kind":"REWORK","_number":68,"created":"2017-11-11 13:10:32.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/68"}}},"commit":{"parents":[{"commit":"2b7193764eb9e7fc7e53515abfc788042c267ffb","subject":"Merge \"Merge haproxy-octavia elements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2b7193764eb9e7fc7e53515abfc788042c267ffb"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-11-11 13:09:38.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ddb549289c417c767637cdd8e21eaa9e78a4f5e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ddb549289c417c767637cdd8e21eaa9e78a4f5e6"}]},"branch":"refs/heads/master"},"b125d040886435bf2b18349d578652c9a98a11ad":{"kind":"REWORK","_number":69,"created":"2017-11-11 13:31:36.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/69"}}},"commit":{"parents":[{"commit":"2b7193764eb9e7fc7e53515abfc788042c267ffb","subject":"Merge \"Merge haproxy-octavia elements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2b7193764eb9e7fc7e53515abfc788042c267ffb"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-11-11 13:30:26.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b125d040886435bf2b18349d578652c9a98a11ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b125d040886435bf2b18349d578652c9a98a11ad"}]},"branch":"refs/heads/master"},"c66bcdaec1ac04f6bbf63626205539a2263f2668":{"kind":"REWORK","_number":70,"created":"2017-11-16 09:17:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/70"}}},"commit":{"parents":[{"commit":"3bf9d1240e0304a2e027f24d5c0d8a26c98d7843","subject":"Merge \"Fix member operating status when add/remove HM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3bf9d1240e0304a2e027f24d5c0d8a26c98d7843"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-11-16 09:17:18.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c66bcdaec1ac04f6bbf63626205539a2263f2668"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c66bcdaec1ac04f6bbf63626205539a2263f2668"}]},"branch":"refs/heads/master"},"672bffe428fd5dc5db9f9850bf27adf36ff6fd9d":{"kind":"REWORK","_number":71,"created":"2017-11-16 09:52:28.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/71"}}},"commit":{"parents":[{"commit":"48bcd70af43182fa63da93734e08fe81d5c9a823","subject":"Optimize update_health process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48bcd70af43182fa63da93734e08fe81d5c9a823"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-11-16 09:52:02.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/672bffe428fd5dc5db9f9850bf27adf36ff6fd9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/672bffe428fd5dc5db9f9850bf27adf36ff6fd9d"}]},"branch":"refs/heads/master"},"1f98d92923091c6b419292bb51d69af1a5f9acae":{"kind":"REWORK","_number":72,"created":"2017-11-27 20:50:20.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/72"}}},"commit":{"parents":[{"commit":"309c4794c941dc7ae24e1ca00b26949437c699a5","subject":"Optimize update_health process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/309c4794c941dc7ae24e1ca00b26949437c699a5"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-11-27 20:50:06.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f98d92923091c6b419292bb51d69af1a5f9acae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f98d92923091c6b419292bb51d69af1a5f9acae"}]},"branch":"refs/heads/master"},"65b91630d34d96366a40d0aafec76b95e48ad4d5":{"kind":"REWORK","_number":73,"created":"2017-11-30 00:39:15.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/73"}}},"commit":{"parents":[{"commit":"fdeadd4a419aa09b7c170b553c442a5f75b88855","subject":"Clean up test_update_db.py a little bit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fdeadd4a419aa09b7c170b553c442a5f75b88855"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-11-30 00:07:53.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65b91630d34d96366a40d0aafec76b95e48ad4d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65b91630d34d96366a40d0aafec76b95e48ad4d5"}]},"branch":"refs/heads/master"},"d0ea14bfd8f0d83aeed1fd56d4ce57371c9247b5":{"kind":"REWORK","_number":74,"created":"2017-11-30 01:24:14.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/74"}}},"commit":{"parents":[{"commit":"fdeadd4a419aa09b7c170b553c442a5f75b88855","subject":"Clean up test_update_db.py a little bit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fdeadd4a419aa09b7c170b553c442a5f75b88855"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-11-30 01:23:34.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0ea14bfd8f0d83aeed1fd56d4ce57371c9247b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0ea14bfd8f0d83aeed1fd56d4ce57371c9247b5"}]},"branch":"refs/heads/master"},"164db6473cdac5499c4883bc82e5917dd01de31d":{"kind":"REWORK","_number":75,"created":"2017-11-30 02:13:57.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/75"}}},"commit":{"parents":[{"commit":"fdeadd4a419aa09b7c170b553c442a5f75b88855","subject":"Clean up test_update_db.py a little bit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fdeadd4a419aa09b7c170b553c442a5f75b88855"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-11-30 02:13:45.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/164db6473cdac5499c4883bc82e5917dd01de31d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/164db6473cdac5499c4883bc82e5917dd01de31d"}]},"branch":"refs/heads/master"},"c1f1e50a8e73487ac3e01dc29793276bbb43f777":{"kind":"REWORK","_number":76,"created":"2017-11-30 03:04:13.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/76"}}},"commit":{"parents":[{"commit":"fdeadd4a419aa09b7c170b553c442a5f75b88855","subject":"Clean up test_update_db.py a little bit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fdeadd4a419aa09b7c170b553c442a5f75b88855"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-11-30 03:03:57.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1f1e50a8e73487ac3e01dc29793276bbb43f777"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1f1e50a8e73487ac3e01dc29793276bbb43f777"}]},"branch":"refs/heads/master"},"643c754acc3215a7f06f8603c6ba434303dbe1d2":{"kind":"REWORK","_number":77,"created":"2017-11-30 05:34:02.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/77"}}},"commit":{"parents":[{"commit":"fdeadd4a419aa09b7c170b553c442a5f75b88855","subject":"Clean up test_update_db.py a little bit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fdeadd4a419aa09b7c170b553c442a5f75b88855"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-11-30 05:33:45.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/643c754acc3215a7f06f8603c6ba434303dbe1d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/643c754acc3215a7f06f8603c6ba434303dbe1d2"}]},"branch":"refs/heads/master"},"a3fb2f85f31bf6fe13a78a2beb919b2d7bf14bb9":{"kind":"REWORK","_number":78,"created":"2017-12-01 00:13:13.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/78"}}},"commit":{"parents":[{"commit":"3a53e54ff51fdeb94db0367839fe360d3772ea08","subject":"Clean up test_update_db.py a little bit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a53e54ff51fdeb94db0367839fe360d3772ea08"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-11-30 23:59:40.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a3fb2f85f31bf6fe13a78a2beb919b2d7bf14bb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a3fb2f85f31bf6fe13a78a2beb919b2d7bf14bb9"}]},"branch":"refs/heads/master"},"9fd44ddfe0df85cfc8df441058ba79f065f4a8f4":{"kind":"REWORK","_number":79,"created":"2017-12-01 19:27:04.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/79"}}},"commit":{"parents":[{"commit":"3a53e54ff51fdeb94db0367839fe360d3772ea08","subject":"Clean up test_update_db.py a little bit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a53e54ff51fdeb94db0367839fe360d3772ea08"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-01 19:24:34.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fd44ddfe0df85cfc8df441058ba79f065f4a8f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fd44ddfe0df85cfc8df441058ba79f065f4a8f4"}]},"branch":"refs/heads/master"},"9db312d4c6a025fa35c0fcc6388d64207a6c6c6a":{"kind":"REWORK","_number":80,"created":"2017-12-01 21:28:31.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/80"}}},"commit":{"parents":[{"commit":"bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f","subject":"Merge \"Clean up test_update_db.py a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-01 21:28:13.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9db312d4c6a025fa35c0fcc6388d64207a6c6c6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9db312d4c6a025fa35c0fcc6388d64207a6c6c6a"}]},"branch":"refs/heads/master"},"0c2fc30c7aace713666462fb5aff27229640284c":{"kind":"REWORK","_number":81,"created":"2017-12-01 22:18:48.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/81"}}},"commit":{"parents":[{"commit":"bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f","subject":"Merge \"Clean up test_update_db.py a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-01 22:18:35.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c2fc30c7aace713666462fb5aff27229640284c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c2fc30c7aace713666462fb5aff27229640284c"}]},"branch":"refs/heads/master"},"63152b03f3ac5377fe97d1081c5eafdfd202ed43":{"kind":"REWORK","_number":82,"created":"2017-12-01 23:44:26.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/82"}}},"commit":{"parents":[{"commit":"bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f","subject":"Merge \"Clean up test_update_db.py a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-01 23:44:13.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63152b03f3ac5377fe97d1081c5eafdfd202ed43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63152b03f3ac5377fe97d1081c5eafdfd202ed43"}]},"branch":"refs/heads/master"},"970ae443c31c8c0c815f55338df96719a0ed0ea6":{"kind":"REWORK","_number":83,"created":"2017-12-02 00:04:50.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/83"}}},"commit":{"parents":[{"commit":"bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f","subject":"Merge \"Clean up test_update_db.py a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-02 00:04:38.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/970ae443c31c8c0c815f55338df96719a0ed0ea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/970ae443c31c8c0c815f55338df96719a0ed0ea6"}]},"branch":"refs/heads/master"},"6a0adcdf1ea003b29c37465d18153aabc2785cec":{"kind":"REWORK","_number":84,"created":"2017-12-02 00:51:33.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/84"}}},"commit":{"parents":[{"commit":"bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f","subject":"Merge \"Clean up test_update_db.py a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-02 00:51:13.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a0adcdf1ea003b29c37465d18153aabc2785cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a0adcdf1ea003b29c37465d18153aabc2785cec"}]},"branch":"refs/heads/master"},"477a8402678fa0c49f004a4a2bf89a2987a54786":{"kind":"REWORK","_number":85,"created":"2017-12-02 02:12:19.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/85","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/85","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/85 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/85 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/85 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/85"}}},"commit":{"parents":[{"commit":"bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f","subject":"Merge \"Clean up test_update_db.py a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-02 02:11:47.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/477a8402678fa0c49f004a4a2bf89a2987a54786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/477a8402678fa0c49f004a4a2bf89a2987a54786"}]},"branch":"refs/heads/master"},"dd2138b5b671f4c365cf83683ef19802e748fcd8":{"kind":"REWORK","_number":86,"created":"2017-12-04 19:59:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/86","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/86","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/86 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/86 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/86 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/86"}}},"commit":{"parents":[{"commit":"bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f","subject":"Merge \"Clean up test_update_db.py a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc5a1e9ef5bb6cc3245c350eb0b57974f5f96b7f"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-04 19:59:31.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd2138b5b671f4c365cf83683ef19802e748fcd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd2138b5b671f4c365cf83683ef19802e748fcd8"}]},"branch":"refs/heads/master"},"8f8724ce8981bc61fe0b4d4e4491c7649f3fb663":{"kind":"REWORK","_number":87,"created":"2017-12-05 01:55:56.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/87","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/87","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/87 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/87 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/87 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/87"}}},"commit":{"parents":[{"commit":"5d2f1af4c1d5b4f6312aa20027cece3fc5709522","subject":"Update compute noop driver/tests (many were disabled/useless)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d2f1af4c1d5b4f6312aa20027cece3fc5709522"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-05 01:55:34.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f8724ce8981bc61fe0b4d4e4491c7649f3fb663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f8724ce8981bc61fe0b4d4e4491c7649f3fb663"}]},"branch":"refs/heads/master"},"69ef25bbfa9045b8d360fcc4661e1d8d59b838a2":{"kind":"REWORK","_number":88,"created":"2017-12-08 21:49:56.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/88","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/88","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/88 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/88 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/88 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/88"}}},"commit":{"parents":[{"commit":"52cbda00bd659ef408ee812b388b7e3db1c3bb84","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52cbda00bd659ef408ee812b388b7e3db1c3bb84"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-08 21:49:45.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69ef25bbfa9045b8d360fcc4661e1d8d59b838a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69ef25bbfa9045b8d360fcc4661e1d8d59b838a2"}]},"branch":"refs/heads/master"},"8f2160fcfa0a3c6fa0d1c000906dd44c46fb93a0":{"kind":"REWORK","_number":89,"created":"2017-12-09 02:14:33.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/89","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/89","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/89 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/89 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/89 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/89"}}},"commit":{"parents":[{"commit":"23ebdc77a8b66e2bd6683dc27f71a9e00f8802f0","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23ebdc77a8b66e2bd6683dc27f71a9e00f8802f0"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-09 02:11:32.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f2160fcfa0a3c6fa0d1c000906dd44c46fb93a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f2160fcfa0a3c6fa0d1c000906dd44c46fb93a0"}]},"branch":"refs/heads/master"},"6e7d6e228c86e9359185b4f0c5529ec5bef93d60":{"kind":"REWORK","_number":90,"created":"2017-12-12 23:15:14.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/90","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/90","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/90 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/90 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/90 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/90"}}},"commit":{"parents":[{"commit":"4ee8fd5a60ec3afd5b707b7792f268a9e755d931","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ee8fd5a60ec3afd5b707b7792f268a9e755d931"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-12 23:13:18.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e7d6e228c86e9359185b4f0c5529ec5bef93d60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e7d6e228c86e9359185b4f0c5529ec5bef93d60"}]},"branch":"refs/heads/master"},"6bf55ab8a1a5dae40bf5ad8df4d5199476a437fe":{"kind":"REWORK","_number":91,"created":"2017-12-13 22:46:23.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/91","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/91","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/91 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/91 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/91 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/91"}}},"commit":{"parents":[{"commit":"4ee8fd5a60ec3afd5b707b7792f268a9e755d931","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ee8fd5a60ec3afd5b707b7792f268a9e755d931"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-13 22:41:24.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bf55ab8a1a5dae40bf5ad8df4d5199476a437fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bf55ab8a1a5dae40bf5ad8df4d5199476a437fe"}]},"branch":"refs/heads/master"},"f306e8e0e8cb1d6526ee168676fef9ced8c05ed4":{"kind":"REWORK","_number":92,"created":"2017-12-13 23:36:09.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/92","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/92","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/92 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/92 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/92 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/92"}}},"commit":{"parents":[{"commit":"90df915bb612e7f8f1c3fdc4561675a8d1081739","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/90df915bb612e7f8f1c3fdc4561675a8d1081739"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-13 23:33:29.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f306e8e0e8cb1d6526ee168676fef9ced8c05ed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f306e8e0e8cb1d6526ee168676fef9ced8c05ed4"}]},"branch":"refs/heads/master"},"221bac6b67334e18410928007a9b714f53e8d8b6":{"kind":"REWORK","_number":93,"created":"2017-12-14 00:32:00.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/93","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/93","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/93 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/93 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/93 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/93"}}},"commit":{"parents":[{"commit":"b37c3e4c710c06f47149c5bb2342dbde94fb0bc3","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b37c3e4c710c06f47149c5bb2342dbde94fb0bc3"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-14 00:30:50.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/221bac6b67334e18410928007a9b714f53e8d8b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/221bac6b67334e18410928007a9b714f53e8d8b6"}]},"branch":"refs/heads/master"},"4e0a92e5d69bbe268dffa6cfdc56d4af088122f5":{"kind":"REWORK","_number":94,"created":"2017-12-14 02:23:47.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/94","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/94","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/94 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/94 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/94 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/94"}}},"commit":{"parents":[{"commit":"b37c3e4c710c06f47149c5bb2342dbde94fb0bc3","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b37c3e4c710c06f47149c5bb2342dbde94fb0bc3"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-14 02:16:37.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e0a92e5d69bbe268dffa6cfdc56d4af088122f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e0a92e5d69bbe268dffa6cfdc56d4af088122f5"}]},"branch":"refs/heads/master"},"eec17d05e960c3548bac2c092f075e99b8e966c2":{"kind":"REWORK","_number":95,"created":"2017-12-14 04:07:02.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/95","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/95","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/95 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/95 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/95 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/95"}}},"commit":{"parents":[{"commit":"b37c3e4c710c06f47149c5bb2342dbde94fb0bc3","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b37c3e4c710c06f47149c5bb2342dbde94fb0bc3"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-14 04:06:44.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eec17d05e960c3548bac2c092f075e99b8e966c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eec17d05e960c3548bac2c092f075e99b8e966c2"}]},"branch":"refs/heads/master"},"b351d82992d66c1269780bb1fafc2aed5b4cb383":{"kind":"REWORK","_number":96,"created":"2017-12-14 20:21:59.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/96","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/96","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/96 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/96 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/96 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/96"}}},"commit":{"parents":[{"commit":"b37c3e4c710c06f47149c5bb2342dbde94fb0bc3","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b37c3e4c710c06f47149c5bb2342dbde94fb0bc3"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-14 20:01:19.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b351d82992d66c1269780bb1fafc2aed5b4cb383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b351d82992d66c1269780bb1fafc2aed5b4cb383"}]},"branch":"refs/heads/master"},"9940aa4cd9d120986abd88949d23c3c530f3305d":{"kind":"REWORK","_number":97,"created":"2017-12-14 22:17:12.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/97","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/97","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/97 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/97 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/97 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/97"}}},"commit":{"parents":[{"commit":"b37c3e4c710c06f47149c5bb2342dbde94fb0bc3","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b37c3e4c710c06f47149c5bb2342dbde94fb0bc3"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-14 22:15:51.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9940aa4cd9d120986abd88949d23c3c530f3305d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9940aa4cd9d120986abd88949d23c3c530f3305d"}]},"branch":"refs/heads/master"},"7eca4ecb692842f0d6de0e6c75ea4da97daeccaa":{"kind":"REWORK","_number":98,"created":"2017-12-15 00:02:09.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/98","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/98","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/98 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/98 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/98 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/98"}}},"commit":{"parents":[{"commit":"b37c3e4c710c06f47149c5bb2342dbde94fb0bc3","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b37c3e4c710c06f47149c5bb2342dbde94fb0bc3"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-15 00:01:54.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7eca4ecb692842f0d6de0e6c75ea4da97daeccaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7eca4ecb692842f0d6de0e6c75ea4da97daeccaa"}]},"branch":"refs/heads/master"},"5757920c7eee8cedb54183478abd57dd8a19a613":{"kind":"REWORK","_number":99,"created":"2017-12-15 00:12:20.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/99","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/99","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/99 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/99 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/99 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/99"}}},"commit":{"parents":[{"commit":"f05e285874f692d41143ac8b115fb7e4406b7712","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f05e285874f692d41143ac8b115fb7e4406b7712"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-15 00:12:07.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5757920c7eee8cedb54183478abd57dd8a19a613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5757920c7eee8cedb54183478abd57dd8a19a613"}]},"branch":"refs/heads/master"},"3af6375e4bd65c5f3ff77a8f83e9c08b77fc9577":{"kind":"REWORK","_number":100,"created":"2017-12-15 00:49:59.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/100","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/100","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/100 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/100 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/100 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/100"}}},"commit":{"parents":[{"commit":"f05e285874f692d41143ac8b115fb7e4406b7712","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f05e285874f692d41143ac8b115fb7e4406b7712"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-15 00:49:40.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3af6375e4bd65c5f3ff77a8f83e9c08b77fc9577"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3af6375e4bd65c5f3ff77a8f83e9c08b77fc9577"}]},"branch":"refs/heads/master"},"c3801a512c06d760a717e7a72523b7e47a9f37ea":{"kind":"REWORK","_number":101,"created":"2017-12-15 02:24:52.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/101","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/101","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/101 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/101 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/101 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/101"}}},"commit":{"parents":[{"commit":"f05e285874f692d41143ac8b115fb7e4406b7712","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f05e285874f692d41143ac8b115fb7e4406b7712"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-15 02:24:37.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3801a512c06d760a717e7a72523b7e47a9f37ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3801a512c06d760a717e7a72523b7e47a9f37ea"}]},"branch":"refs/heads/master"},"1244d3693c191191c7a713876c44f08d34d91a5f":{"kind":"REWORK","_number":102,"created":"2017-12-15 19:03:27.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/102","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/102","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/102 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/102 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/102 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/102"}}},"commit":{"parents":[{"commit":"f05e285874f692d41143ac8b115fb7e4406b7712","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f05e285874f692d41143ac8b115fb7e4406b7712"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-15 19:03:15.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1244d3693c191191c7a713876c44f08d34d91a5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1244d3693c191191c7a713876c44f08d34d91a5f"}]},"branch":"refs/heads/master"},"72ae52c2f1f600f010fd3fdee53b2eff5a02dc08":{"kind":"REWORK","_number":103,"created":"2017-12-15 20:50:57.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/103","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/103","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/103 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/103 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/103 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/103"}}},"commit":{"parents":[{"commit":"f05e285874f692d41143ac8b115fb7e4406b7712","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f05e285874f692d41143ac8b115fb7e4406b7712"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-15 20:50:30.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72ae52c2f1f600f010fd3fdee53b2eff5a02dc08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72ae52c2f1f600f010fd3fdee53b2eff5a02dc08"}]},"branch":"refs/heads/master"},"bfb7caf6044d90469553b2f4b709ca0e52a0319f":{"kind":"REWORK","_number":104,"created":"2017-12-15 23:56:35.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/104","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/104","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/104 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/104 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/104 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/104"}}},"commit":{"parents":[{"commit":"f05e285874f692d41143ac8b115fb7e4406b7712","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f05e285874f692d41143ac8b115fb7e4406b7712"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-15 23:56:18.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfb7caf6044d90469553b2f4b709ca0e52a0319f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfb7caf6044d90469553b2f4b709ca0e52a0319f"}]},"branch":"refs/heads/master"},"e7a471bef15c195ecdc617fcecd792750f1b0752":{"kind":"REWORK","_number":105,"created":"2017-12-21 20:18:16.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/105","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/105","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/105 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/105 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/105 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/105"}}},"commit":{"parents":[{"commit":"1c44d1a9b5cbf0534ee72eca331541ce36bc7895","subject":"Add unit tests for neutron utils, add model/util for floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c44d1a9b5cbf0534ee72eca331541ce36bc7895"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-12-21 20:14:31.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7a471bef15c195ecdc617fcecd792750f1b0752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7a471bef15c195ecdc617fcecd792750f1b0752"}]},"branch":"refs/heads/master"},"510f28861b38516496e5111c8b8a14f918e24fac":{"kind":"REWORK","_number":106,"created":"2018-01-13 02:22:35.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/106","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/106","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/106 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/106 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/106 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/106"}}},"commit":{"parents":[{"commit":"ab45b15c27031d37cf7b7eb8051308d42514d08a","subject":"Merge \"Add unit tests for neutron utils, add model/util for floating_ip\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab45b15c27031d37cf7b7eb8051308d42514d08a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-01-13 02:14:41.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510f28861b38516496e5111c8b8a14f918e24fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510f28861b38516496e5111c8b8a14f918e24fac"}]},"branch":"refs/heads/master"},"63e333f90d21767d38f19d791693fcda5ed1a7b2":{"kind":"REWORK","_number":107,"created":"2018-02-02 20:45:16.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/107","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/107","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/107 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/107 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/107 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/107"}}},"commit":{"parents":[{"commit":"d1693c523b40e401ea34c9593fd40dd3f929f434","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1693c523b40e401ea34c9593fd40dd3f929f434"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-02-02 20:44:48.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63e333f90d21767d38f19d791693fcda5ed1a7b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63e333f90d21767d38f19d791693fcda5ed1a7b2"}]},"branch":"refs/heads/master"},"95b25a6649e9cf43e4a18388ec7694edec094a6b":{"kind":"REWORK","_number":108,"created":"2018-02-06 00:22:24.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/108","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/108","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/108 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/108 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/108 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/108"}}},"commit":{"parents":[{"commit":"7ca4508f456d53766377350fd344f29545a78cfd","subject":"Re-enable \"Page Contents\" and update user docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ca4508f456d53766377350fd344f29545a78cfd"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-02-06 00:21:55.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95b25a6649e9cf43e4a18388ec7694edec094a6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95b25a6649e9cf43e4a18388ec7694edec094a6b"}]},"branch":"refs/heads/master"},"f40d690ee9a0a7a987c796afc5266bab871af30c":{"kind":"REWORK","_number":109,"created":"2018-02-15 03:19:41.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/109","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/109","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/109 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/109 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/109 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/109"}}},"commit":{"parents":[{"commit":"26b0802ace77b9a9b89d3b2d265950fa1b407696","subject":"Merge \"Fix missing test-requirement and cleanup docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26b0802ace77b9a9b89d3b2d265950fa1b407696"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-02-15 03:14:11.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f40d690ee9a0a7a987c796afc5266bab871af30c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f40d690ee9a0a7a987c796afc5266bab871af30c"}]},"branch":"refs/heads/master"},"4810d23e84867e7ce351f4f956fca42b29cf3dfa":{"kind":"REWORK","_number":110,"created":"2018-02-15 10:00:15.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/110","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/110","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/110 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/110 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/110 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/110"}}},"commit":{"parents":[{"commit":"26b0802ace77b9a9b89d3b2d265950fa1b407696","subject":"Merge \"Fix missing test-requirement and cleanup docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26b0802ace77b9a9b89d3b2d265950fa1b407696"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-02-15 09:59:55.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4810d23e84867e7ce351f4f956fca42b29cf3dfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4810d23e84867e7ce351f4f956fca42b29cf3dfa"}]},"branch":"refs/heads/master"},"c260dbee4905acb79ff26829363544c7a5235a31":{"kind":"REWORK","_number":111,"created":"2018-02-21 01:04:36.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/111","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/111","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/111 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/111 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/111 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/111"}}},"commit":{"parents":[{"commit":"21117d1053580723daa1b1ba9e18901814f7e3ab","subject":"Overhaul HealthManager update threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21117d1053580723daa1b1ba9e18901814f7e3ab"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-02-21 01:02:46.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c260dbee4905acb79ff26829363544c7a5235a31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c260dbee4905acb79ff26829363544c7a5235a31"}]},"branch":"refs/heads/master"},"6c81a14c3d039d8700b3bd8ef38c3ac9cf120854":{"kind":"TRIVIAL_REBASE","_number":112,"created":"2018-02-21 01:57:29.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/112","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/112","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/112 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/112 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/112 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/112"}}},"commit":{"parents":[{"commit":"82b9d78ebb7a4ba2b4db251a4e185214e0090b13","subject":"Overhaul HealthManager update threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/82b9d78ebb7a4ba2b4db251a4e185214e0090b13"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-02-21 01:57:29.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c81a14c3d039d8700b3bd8ef38c3ac9cf120854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c81a14c3d039d8700b3bd8ef38c3ac9cf120854"}]},"branch":"refs/heads/master"},"372e5e1ade184793e4aeccf218f5f5886c32d960":{"kind":"REWORK","_number":113,"created":"2018-02-27 23:22:29.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/113","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/113","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/113 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/113 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/113 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/113"}}},"commit":{"parents":[{"commit":"a94b62ed6599027ae44cfd440a984061995a8dc9","subject":"Merge \"Update configuration samples (QoS)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a94b62ed6599027ae44cfd440a984061995a8dc9"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-02-27 23:21:17.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/372e5e1ade184793e4aeccf218f5f5886c32d960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/372e5e1ade184793e4aeccf218f5f5886c32d960"}]},"branch":"refs/heads/master"},"7c68fab7f67a88fc1c015d57c99db8d64204988a":{"kind":"REWORK","_number":114,"created":"2018-03-08 19:51:17.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/114","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/114","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/114 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/114 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/114 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/114"}}},"commit":{"parents":[{"commit":"7f3506b2956c4f6edc24703ea8099ce783715897","subject":"Migrate to stestr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f3506b2956c4f6edc24703ea8099ce783715897"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-03-08 19:50:53.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c68fab7f67a88fc1c015d57c99db8d64204988a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c68fab7f67a88fc1c015d57c99db8d64204988a"}]},"branch":"refs/heads/master"},"2ff7eeb7892ae335a9bf0d45c0f57fb747bf987c":{"kind":"REWORK","_number":115,"created":"2018-03-14 15:38:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/115","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/115","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/115 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/115 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/115 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/115"}}},"commit":{"parents":[{"commit":"03e5ae07992734340f105d495c9169e8d4882c16","subject":"Merge \"Reduce stestr concurrency to 1 for tempest jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/03e5ae07992734340f105d495c9169e8d4882c16"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-03-14 15:37:57.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ff7eeb7892ae335a9bf0d45c0f57fb747bf987c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ff7eeb7892ae335a9bf0d45c0f57fb747bf987c"}]},"branch":"refs/heads/master"},"c573968c46228c4d6ab2ca2fdaeb07f13dbbc9d6":{"kind":"TRIVIAL_REBASE","_number":116,"created":"2018-03-19 18:12:56.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/116","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/116","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/116 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/116 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/116 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/116"}}},"commit":{"parents":[{"commit":"54a4cf00cf304b108013ca4487c138c13f30983d","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54a4cf00cf304b108013ca4487c138c13f30983d"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-03-19 18:12:56.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c573968c46228c4d6ab2ca2fdaeb07f13dbbc9d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c573968c46228c4d6ab2ca2fdaeb07f13dbbc9d6"}]},"branch":"refs/heads/master"},"11224a311aa8cdbb1d043e13830a419b2d357a1e":{"kind":"TRIVIAL_REBASE","_number":117,"created":"2018-03-27 05:50:34.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/117","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/117","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/117 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/117 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/117 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/117"}}},"commit":{"parents":[{"commit":"df9770a77377e1618e532b2553ea30cea3bf2eca","subject":"Merge \"Switch multinode tests to non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df9770a77377e1618e532b2553ea30cea3bf2eca"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-03-27 05:50:34.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11224a311aa8cdbb1d043e13830a419b2d357a1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11224a311aa8cdbb1d043e13830a419b2d357a1e"}]},"branch":"refs/heads/master"},"c073cb9b616642388dd01f46f3b1c61c544316d6":{"kind":"TRIVIAL_REBASE","_number":118,"created":"2018-04-04 22:42:26.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/118","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/118","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/118 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/118 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/118 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/118"}}},"commit":{"parents":[{"commit":"e819e4521ad6c4a02086309abb922206125ef596","subject":"Allow members to be set as \"backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e819e4521ad6c4a02086309abb922206125ef596"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 22:42:26.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c073cb9b616642388dd01f46f3b1c61c544316d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c073cb9b616642388dd01f46f3b1c61c544316d6"}]},"branch":"refs/heads/master"},"df296a3de510111c4fe4c3449d852161a1e6508b":{"kind":"REWORK","_number":119,"created":"2018-04-06 00:37:00.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/119","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/119","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/119 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/119 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/119 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/119"}}},"commit":{"parents":[{"commit":"b8b57f9622cccd075422ce565a160c0adee882ec","subject":"WIP: Experimental multi-az support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8b57f9622cccd075422ce565a160c0adee882ec"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-06 00:34:14.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nNow with more ACTIVE_STANDBY!\nNow with more AZ Anti-Affinity!\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df296a3de510111c4fe4c3449d852161a1e6508b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df296a3de510111c4fe4c3449d852161a1e6508b"}]},"branch":"refs/heads/master"},"b83791a148d3895e5089141c11f196b36bfbd591":{"kind":"REWORK","_number":120,"created":"2018-04-06 23:24:26.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/120","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/120","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/120 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/120 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/120 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/120"}}},"commit":{"parents":[{"commit":"0b4bf6e856b03e57b5ff4f83f959acf655037a41","subject":"Experimental multi-az support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b4bf6e856b03e57b5ff4f83f959acf655037a41"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-06 23:13:14.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b83791a148d3895e5089141c11f196b36bfbd591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b83791a148d3895e5089141c11f196b36bfbd591"}]},"branch":"refs/heads/master"},"43a118e7d47523eaa0c70f84ba4745b26f2e33f1":{"kind":"REWORK","_number":121,"created":"2018-04-09 21:30:49.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/121","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/121","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/121 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/121 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/121 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/121"}}},"commit":{"parents":[{"commit":"55390d218a5352eb46f3354050a7ae2d4a52a11a","subject":"Experimental multi-az support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55390d218a5352eb46f3354050a7ae2d4a52a11a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-09 21:30:18.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43a118e7d47523eaa0c70f84ba4745b26f2e33f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43a118e7d47523eaa0c70f84ba4745b26f2e33f1"}]},"branch":"refs/heads/master"},"4e530108b487d684f6d7411352e512c3405ab24e":{"kind":"REWORK","_number":122,"created":"2018-04-10 01:10:36.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/122","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/122","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/122 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/122 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/122 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/122"}}},"commit":{"parents":[{"commit":"0332356d6c92f2e21f0180b028b3c718c4beb338","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0332356d6c92f2e21f0180b028b3c718c4beb338"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-10 01:09:25.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e530108b487d684f6d7411352e512c3405ab24e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e530108b487d684f6d7411352e512c3405ab24e"}]},"branch":"refs/heads/master"},"dc6ce786d10213af6b47cb0f6c3f3b88740d0ab1":{"kind":"REWORK","_number":123,"created":"2018-04-24 01:15:59.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/123","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/123","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/123 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/123 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/123 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/123"}}},"commit":{"parents":[{"commit":"869d9f861992587082f9b89bdf69b93fcea7b1af","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/869d9f861992587082f9b89bdf69b93fcea7b1af"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-24 01:15:45.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc6ce786d10213af6b47cb0f6c3f3b88740d0ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc6ce786d10213af6b47cb0f6c3f3b88740d0ab1"}]},"branch":"refs/heads/master"},"fcfb3b2c07f099407625a94dbf20cda3062a7870":{"kind":"REWORK","_number":124,"created":"2018-04-26 04:57:15.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/124","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/124","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/124 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/124 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/124 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/124"}}},"commit":{"parents":[{"commit":"fc814fb39d356702530ec97cbb1b197135339aa4","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc814fb39d356702530ec97cbb1b197135339aa4"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-26 04:56:49.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcfb3b2c07f099407625a94dbf20cda3062a7870"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcfb3b2c07f099407625a94dbf20cda3062a7870"}]},"branch":"refs/heads/master"},"a41591c32b92ddad1f6fa6870d368ab5c78482f4":{"kind":"REWORK","_number":125,"created":"2018-04-26 05:12:18.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/125","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/125","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/125 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/125 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/125 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/125"}}},"commit":{"parents":[{"commit":"fc814fb39d356702530ec97cbb1b197135339aa4","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc814fb39d356702530ec97cbb1b197135339aa4"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-26 05:12:06.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a41591c32b92ddad1f6fa6870d368ab5c78482f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a41591c32b92ddad1f6fa6870d368ab5c78482f4"}]},"branch":"refs/heads/master"},"55d3d2cca1e336146fc6e2c9df38117cd6cc9fb2":{"kind":"REWORK","_number":126,"created":"2018-04-26 20:37:25.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/126","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/126","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/126 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/126 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/126 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/126"}}},"commit":{"parents":[{"commit":"efc2d2f9f84399ec9e5e80374f5a69cda5991a58","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/efc2d2f9f84399ec9e5e80374f5a69cda5991a58"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-26 20:37:12.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55d3d2cca1e336146fc6e2c9df38117cd6cc9fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55d3d2cca1e336146fc6e2c9df38117cd6cc9fb2"}]},"branch":"refs/heads/master"},"a0b06d84cd4735770da22034758f2500bd73b7a9":{"kind":"REWORK","_number":127,"created":"2018-04-26 22:40:05.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/127","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/127","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/127 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/127 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/127 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/127"}}},"commit":{"parents":[{"commit":"0c05f78b362418fb68845160d588c1bdd66e55a3","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c05f78b362418fb68845160d588c1bdd66e55a3"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-26 22:39:51.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0b06d84cd4735770da22034758f2500bd73b7a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0b06d84cd4735770da22034758f2500bd73b7a9"}]},"branch":"refs/heads/master"},"8002abb40f12b3c8e8bf8a3adde893783a89e116":{"kind":"TRIVIAL_REBASE","_number":128,"created":"2018-04-27 20:29:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/128","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/128","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/128 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/128 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/128 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/128"}}},"commit":{"parents":[{"commit":"aace25cbe7a8b7dfd2052c05225426f9dbc812b1","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aace25cbe7a8b7dfd2052c05225426f9dbc812b1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-27 20:29:43.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8002abb40f12b3c8e8bf8a3adde893783a89e116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8002abb40f12b3c8e8bf8a3adde893783a89e116"}]},"branch":"refs/heads/master"},"9fb7dc39ecad251ff0d72331eb845d6cbd43f778":{"kind":"TRIVIAL_REBASE","_number":129,"created":"2018-05-07 18:58:53.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/129","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/129","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/129 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/129 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/129 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/129"}}},"commit":{"parents":[{"commit":"7d1b30cb3191ae74ae2618fc862840542da8d9dd","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d1b30cb3191ae74ae2618fc862840542da8d9dd"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-05-07 18:58:53.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fb7dc39ecad251ff0d72331eb845d6cbd43f778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fb7dc39ecad251ff0d72331eb845d6cbd43f778"}]},"branch":"refs/heads/master"},"ba63285ada6523a6479f97189f8dee97a6fc14c5":{"kind":"REWORK","_number":130,"created":"2018-05-09 01:54:44.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/130","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/130","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/130 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/130 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/130 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/130"}}},"commit":{"parents":[{"commit":"7d1b30cb3191ae74ae2618fc862840542da8d9dd","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d1b30cb3191ae74ae2618fc862840542da8d9dd"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-05-09 01:54:05.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba63285ada6523a6479f97189f8dee97a6fc14c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba63285ada6523a6479f97189f8dee97a6fc14c5"}]},"branch":"refs/heads/master"},"e8cecf7d33dae8a506736c54c82b30d6848a144d":{"kind":"REWORK","_number":131,"created":"2018-05-09 18:15:49.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/131","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/131","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/131 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/131 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/131 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/131"}}},"commit":{"parents":[{"commit":"7d1b30cb3191ae74ae2618fc862840542da8d9dd","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d1b30cb3191ae74ae2618fc862840542da8d9dd"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-05-09 18:15:37.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e8cecf7d33dae8a506736c54c82b30d6848a144d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e8cecf7d33dae8a506736c54c82b30d6848a144d"}]},"branch":"refs/heads/master"},"c7822b1eb698b61a53831c9e13453a794aafe010":{"kind":"REWORK","_number":132,"created":"2018-05-11 04:13:48.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/132","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/132","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/132 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/132 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/132 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/132"}}},"commit":{"parents":[{"commit":"7d1b30cb3191ae74ae2618fc862840542da8d9dd","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d1b30cb3191ae74ae2618fc862840542da8d9dd"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-05-11 04:13:25.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7822b1eb698b61a53831c9e13453a794aafe010"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7822b1eb698b61a53831c9e13453a794aafe010"}]},"branch":"refs/heads/master"},"ce5c304729e2dffa24717b7070ea8fbb7e72d93e":{"kind":"TRIVIAL_REBASE","_number":133,"created":"2018-05-31 23:54:46.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/133","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/133","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/133 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/133 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/133 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/133"}}},"commit":{"parents":[{"commit":"2a4368a9427bce7da6e1a42721b1d24f01262299","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a4368a9427bce7da6e1a42721b1d24f01262299"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-05-31 23:54:28.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce5c304729e2dffa24717b7070ea8fbb7e72d93e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce5c304729e2dffa24717b7070ea8fbb7e72d93e"}]},"branch":"refs/heads/master"},"f558df47e99790fea3b9839d73f6ed82eaa2b8f1":{"kind":"REWORK","_number":134,"created":"2018-06-02 02:01:02.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/134","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/134","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/134 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/134 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/134 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/134"}}},"commit":{"parents":[{"commit":"72774f016559fe232275edb52da68962b8beb6b7","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72774f016559fe232275edb52da68962b8beb6b7"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-02 02:00:57.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f558df47e99790fea3b9839d73f6ed82eaa2b8f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f558df47e99790fea3b9839d73f6ed82eaa2b8f1"}]},"branch":"refs/heads/master"},"05542ddfb66a18f1a3291af29f39b51b9c585535":{"kind":"TRIVIAL_REBASE","_number":135,"created":"2018-06-05 20:55:31.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/135","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/135","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/135 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/135 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/135 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/135"}}},"commit":{"parents":[{"commit":"a9ae12a9ebfa77f0aeadfd8bad540b77d0a44661","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9ae12a9ebfa77f0aeadfd8bad540b77d0a44661"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-05 20:55:31.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05542ddfb66a18f1a3291af29f39b51b9c585535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05542ddfb66a18f1a3291af29f39b51b9c585535"}]},"branch":"refs/heads/master"},"c062bc8af4e50a379941d47e8ef89eaa0918558d":{"kind":"TRIVIAL_REBASE","_number":136,"created":"2018-06-07 01:38:31.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/136","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/136","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/136 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/136 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/136 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/136"}}},"commit":{"parents":[{"commit":"336d1dbbe469cf669448b002a0f33d99fdbde0da","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/336d1dbbe469cf669448b002a0f33d99fdbde0da"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-07 01:38:21.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c062bc8af4e50a379941d47e8ef89eaa0918558d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c062bc8af4e50a379941d47e8ef89eaa0918558d"}]},"branch":"refs/heads/master"},"b0570e0d89bbf1802d76797d6992c365478425a0":{"kind":"REWORK","_number":137,"created":"2018-06-07 02:00:45.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/137","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/137","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/137 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/137 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/137 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/137"}}},"commit":{"parents":[{"commit":"336d1dbbe469cf669448b002a0f33d99fdbde0da","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/336d1dbbe469cf669448b002a0f33d99fdbde0da"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-07 02:00:33.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0570e0d89bbf1802d76797d6992c365478425a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0570e0d89bbf1802d76797d6992c365478425a0"}]},"branch":"refs/heads/master"},"4c35f8e6057e4c0545b3a138de66a4dec98b0ebc":{"kind":"REWORK","_number":138,"created":"2018-06-07 02:44:00.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/138","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/138","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/138 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/138 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/138 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/138"}}},"commit":{"parents":[{"commit":"336d1dbbe469cf669448b002a0f33d99fdbde0da","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/336d1dbbe469cf669448b002a0f33d99fdbde0da"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-07 02:43:51.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c35f8e6057e4c0545b3a138de66a4dec98b0ebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c35f8e6057e4c0545b3a138de66a4dec98b0ebc"}]},"branch":"refs/heads/master"},"a9ace1fd1dfde95950a279022f7ed479f7c84216":{"kind":"REWORK","_number":139,"created":"2018-06-07 05:01:51.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/139","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/139","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/139 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/139 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/139 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/139"}}},"commit":{"parents":[{"commit":"336d1dbbe469cf669448b002a0f33d99fdbde0da","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/336d1dbbe469cf669448b002a0f33d99fdbde0da"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-07 05:01:41.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9ace1fd1dfde95950a279022f7ed479f7c84216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9ace1fd1dfde95950a279022f7ed479f7c84216"}]},"branch":"refs/heads/master"},"526fc77aaadd5975050147aecc0e7366eb3b507a":{"kind":"REWORK","_number":140,"created":"2018-06-07 08:40:36.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/140","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/140","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/140 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/140 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/140 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/140"}}},"commit":{"parents":[{"commit":"336d1dbbe469cf669448b002a0f33d99fdbde0da","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/336d1dbbe469cf669448b002a0f33d99fdbde0da"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-07 08:40:27.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/526fc77aaadd5975050147aecc0e7366eb3b507a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/526fc77aaadd5975050147aecc0e7366eb3b507a"}]},"branch":"refs/heads/master"},"78aec48402a9639fbea4ada63bfc1f92302f81ef":{"kind":"REWORK","_number":141,"created":"2018-06-14 18:09:41.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/141","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/141","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/141 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/141 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/141 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/141"}}},"commit":{"parents":[{"commit":"72be69bbd8ebabeabecd80ce25ac2c1477177608","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72be69bbd8ebabeabecd80ce25ac2c1477177608"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-14 18:08:52.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78aec48402a9639fbea4ada63bfc1f92302f81ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78aec48402a9639fbea4ada63bfc1f92302f81ef"}]},"branch":"refs/heads/master"},"27b0f11194edc1264f33ab426feba3972d7fa902":{"kind":"REWORK","_number":142,"created":"2018-06-29 20:25:45.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/142","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/142","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/142 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/142 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/142 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/142"}}},"commit":{"parents":[{"commit":"0ca751417001aa94c022c8a6568982484a51cbd9","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ca751417001aa94c022c8a6568982484a51cbd9"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-29 20:22:11.000000000","tz":-300},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27b0f11194edc1264f33ab426feba3972d7fa902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27b0f11194edc1264f33ab426feba3972d7fa902"}]},"branch":"refs/heads/master"},"52acb7c5ac556d5900f263635ae0c40e825eb77b":{"kind":"TRIVIAL_REBASE","_number":143,"created":"2018-07-13 01:07:03.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/143","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/143","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/143 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/143 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/143 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/143"}}},"commit":{"parents":[{"commit":"369b9e6c1f8464da14e9b7594b09c9ec1ecc7f4e","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/369b9e6c1f8464da14e9b7594b09c9ec1ecc7f4e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-07-13 01:06:49.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52acb7c5ac556d5900f263635ae0c40e825eb77b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52acb7c5ac556d5900f263635ae0c40e825eb77b"}]},"branch":"refs/heads/master"},"89229e1e6539647f5c90d517b81c03d2cea99463":{"kind":"TRIVIAL_REBASE","_number":144,"created":"2018-07-16 21:12:40.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/144","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/144","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/144 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/144 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/144 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/144"}}},"commit":{"parents":[{"commit":"594ecf92d574107a6143da663fc516e6d8fb4471","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/594ecf92d574107a6143da663fc516e6d8fb4471"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-07-16 21:12:40.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89229e1e6539647f5c90d517b81c03d2cea99463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89229e1e6539647f5c90d517b81c03d2cea99463"}]},"branch":"refs/heads/master"},"f4e55082ec298e9d6b7b8f8db3ac1d87d8de3b27":{"kind":"REWORK","_number":145,"created":"2018-07-30 18:30:04.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/145","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/145","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/145 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/145 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/145 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/145"}}},"commit":{"parents":[{"commit":"04432b1b70c4169cfe7d982b637b19131ebffec3","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04432b1b70c4169cfe7d982b637b19131ebffec3"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-07-30 18:29:52.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4e55082ec298e9d6b7b8f8db3ac1d87d8de3b27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4e55082ec298e9d6b7b8f8db3ac1d87d8de3b27"}]},"branch":"refs/heads/master"},"a0684362038de5ba8c0808fed7aea04bb6e90b8f":{"kind":"REWORK","_number":146,"created":"2018-08-21 01:55:37.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/146","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/146","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/146 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/146 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/146 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/146"}}},"commit":{"parents":[{"commit":"35a0666d4bf85707cb5724a5008dd94884c47640","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35a0666d4bf85707cb5724a5008dd94884c47640"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-08-21 01:55:16.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0684362038de5ba8c0808fed7aea04bb6e90b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0684362038de5ba8c0808fed7aea04bb6e90b8f"}]},"branch":"refs/heads/master"},"cb27e070e135487e01f3d609a24b986fc219234c":{"kind":"TRIVIAL_REBASE","_number":147,"created":"2018-09-17 22:10:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/147","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/147","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/147 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/147 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/147 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/147"}}},"commit":{"parents":[{"commit":"e4506e9b4beed996873e983019b8304556fac1d3","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4506e9b4beed996873e983019b8304556fac1d3"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-09-17 22:10:43.000000000","tz":0},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb27e070e135487e01f3d609a24b986fc219234c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb27e070e135487e01f3d609a24b986fc219234c"}]},"branch":"refs/heads/master"},"a8d28daf1a3bf7eaeab9ab8bf2d495ddf072d1c6":{"kind":"REWORK","_number":148,"created":"2018-09-17 23:03:03.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/148","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/148","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/148 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/148 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/148 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/148"}}},"commit":{"parents":[{"commit":"e4506e9b4beed996873e983019b8304556fac1d3","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4506e9b4beed996873e983019b8304556fac1d3"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-09-17 23:02:51.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8d28daf1a3bf7eaeab9ab8bf2d495ddf072d1c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8d28daf1a3bf7eaeab9ab8bf2d495ddf072d1c6"}]},"branch":"refs/heads/master"},"96cdf0d1a016fd2cd7a786513034594f8e35c6f3":{"kind":"REWORK","_number":149,"created":"2018-10-01 22:14:56.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/149","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/149","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/149 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/149 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/149 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/149"}}},"commit":{"parents":[{"commit":"f0875cefb8d3fda9db73ab9f4f098118d2191e1f","subject":"WIP: AZ Evacuation resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0875cefb8d3fda9db73ab9f4f098118d2191e1f"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-10-01 22:12:48.000000000","tz":-420},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96cdf0d1a016fd2cd7a786513034594f8e35c6f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96cdf0d1a016fd2cd7a786513034594f8e35c6f3"}]},"branch":"refs/heads/master"},"59b6e26aa245957ea10b6dbf74e02df799a78690":{"kind":"REWORK","_number":150,"created":"2018-11-05 22:02:14.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/150","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/150","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/150 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/150 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/150 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/150"}}},"commit":{"parents":[{"commit":"f9d1508c42654899b8ca4c78b66af2c6386df7aa","subject":"Experimental multi-az support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9d1508c42654899b8ca4c78b66af2c6386df7aa"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-11-05 22:02:11.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/59b6e26aa245957ea10b6dbf74e02df799a78690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/59b6e26aa245957ea10b6dbf74e02df799a78690"}]},"branch":"refs/heads/master"},"abfdfe56179a29d324645ab2d9642a45771b1e90":{"kind":"REWORK","_number":151,"created":"2018-11-05 22:44:07.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/151","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/151","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/151 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/151 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/151 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/151"}}},"commit":{"parents":[{"commit":"6cdc71cd3e4daea4d6a7bee4f4e3da5254f2ace7","subject":"Experimental multi-az support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cdc71cd3e4daea4d6a7bee4f4e3da5254f2ace7"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-11-05 22:33:53.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/abfdfe56179a29d324645ab2d9642a45771b1e90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/abfdfe56179a29d324645ab2d9642a45771b1e90"}]},"branch":"refs/heads/master"},"a9335354e528bdf3a6a3b88b5619642acd4ec0ed":{"kind":"REWORK","_number":152,"created":"2018-11-07 01:25:24.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/152","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/152","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/152 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/152 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/152 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/152"}}},"commit":{"parents":[{"commit":"6cdc71cd3e4daea4d6a7bee4f4e3da5254f2ace7","subject":"Experimental multi-az support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cdc71cd3e4daea4d6a7bee4f4e3da5254f2ace7"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-11-07 01:25:06.000000000","tz":540},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9335354e528bdf3a6a3b88b5619642acd4ec0ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9335354e528bdf3a6a3b88b5619642acd4ec0ed"}]},"branch":"refs/heads/master"},"23fadd97e475602b72e5d5994a13bdf1fc06b62e":{"kind":"REWORK","_number":153,"created":"2019-01-28 21:11:21.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/153","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/153","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/153 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/153 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/153 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/153"}}},"commit":{"parents":[{"commit":"ca61ce5dd268e443a45fea3e334150c349ae2974","subject":"Experimental multi-az support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ca61ce5dd268e443a45fea3e334150c349ae2974"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-01-28 21:09:06.000000000","tz":-360},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23fadd97e475602b72e5d5994a13bdf1fc06b62e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23fadd97e475602b72e5d5994a13bdf1fc06b62e"}]},"branch":"refs/heads/master"},"d606b8f52b84cbfa7347ea5805c1b756cdbf26b1":{"kind":"REWORK","_number":154,"created":"2019-01-30 20:47:31.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/154","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/154","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/154 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/154 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/154 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/154"}}},"commit":{"parents":[{"commit":"e43d8ff6ed5e66d323a93f9ca7b095f3e2ec54f6","subject":"Experimental multi-az support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e43d8ff6ed5e66d323a93f9ca7b095f3e2ec54f6"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-01-30 20:46:57.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d606b8f52b84cbfa7347ea5805c1b756cdbf26b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d606b8f52b84cbfa7347ea5805c1b756cdbf26b1"}]},"branch":"refs/heads/master"},"0f69e8bf333fe79995c9f7cb960324273970691c":{"kind":"REWORK","_number":155,"created":"2019-11-18 19:51:12.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/435612/155","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/435612/155","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/155 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/155 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/435612/155 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/435612/155"}}},"commit":{"parents":[{"commit":"07e3870d422b7688d15733afa52f873f7e8833cb","subject":"Experimental multi-az support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/07e3870d422b7688d15733afa52f873f7e8833cb"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2017-02-17 21:56:02.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-18 19:47:06.000000000","tz":-480},"subject":"WIP: Floating IP Network Driver (spans L3s)","message":"WIP: Floating IP Network Driver (spans L3s)\n\nAt GoDaddy, our hypervisors are scattered across many different L2\nnetworks, so we can\u0027t count on the ability to move ports back and\nforth between amphorae to route traffic. We need something that operates\nat L3 instead.\n\nThis driver should allow operators to use Floating IPs as the VIP for\nloadbalancers, solving the problem (since FLIPs are L3).\n\nChange-Id: I5d3f8a63d795ec14f6cdadfb7d7b263cbe7328ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f69e8bf333fe79995c9f7cb960324273970691c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f69e8bf333fe79995c9f7cb960324273970691c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
