)]}'
{"id":"openstack%2Foctavia~478637","triplet_id":"openstack%2Foctavia~master~I4f5bafe8d194b23a55cb4de0ba42ad7b6eb7a6cc","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I4f5bafe8d194b23a55cb4de0ba42ad7b6eb7a6cc","subject":"Enable uwsgi support for octavia devstack","status":"MERGED","created":"2017-06-28 21:09:37.000000000","updated":"2017-06-29 03:19:33.000000000","submitted":"2017-06-29 03:19:33.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ac081ce70b8c4248530ac122449470a64f53ac80","_number":478637,"virtual_id_number":478637,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2017-06-29 03:19:33.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2017-06-29 01:20:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2017-06-29 02:25:38.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2017-06-29 02:25:38.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-06-29 01:20:22.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2017-06-29 02:25:38.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2017-06-29 03:19:33.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"4cf2014ebc86df9ce93c311e3eb72d75dc37fc1d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2017-06-28 21:09:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cfe61a2c08968b561427882de3482341570ce30","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2017-06-28 21:09:48.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"9f81f0235c14a34683342f0ec98527c0011a01d2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-28 22:02:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/37/478637/1/check/gate-octavia-docs-ubuntu-xenial/ecef7a1//doc/build/html/ : SUCCESS in 4m 41s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/37/478637/1/check/gate-octavia-pep8-ubuntu-xenial/c15239a/ : SUCCESS in 4m 13s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/37/478637/1/check/gate-octavia-python27-ubuntu-xenial/97143fc/ : SUCCESS in 2m 56s\n- gate-octavia-python35 http://logs.openstack.org/37/478637/1/check/gate-octavia-python35/5223d1b/ : SUCCESS in 2m 52s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/37/478637/1/check/octavia-coverage-ubuntu-xenial/43818da/ : SUCCESS in 8m 26s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/37/478637/1/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/782a469/ : FAILURE in 27m 30s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/37/478637/1/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/27ceef1/ : FAILURE in 39m 20s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/37/478637/1/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/6a3daf7/ : FAILURE in 44m 17s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/37/478637/1/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/50418a4/ : FAILURE in 42m 38s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/37/478637/1/check/gate-octavia-tox-functional-ubuntu-xenial/532c110/ : SUCCESS in 6m 35s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/37/478637/1/check/gate-octavia-tox-functional-py35-ubuntu-xenial/c67a6f4/ : SUCCESS in 4m 43s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/37/478637/1/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/8d7bf2a/ : SUCCESS in 44m 24s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/37/478637/1/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/9930797/ : SUCCESS in 47m 40s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/37/478637/1/check/gate-octavia-v1-dsvm-scenario-multinode/8620358/ : SUCCESS in 43m 08s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/37/478637/1/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/6522c82/ : SUCCESS in 34m 47s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/37/478637/1/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/e0e488b/ : SUCCESS in 50m 48s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/37/478637/1/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/f159457/ : FAILURE in 42m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3208431a5f4b4756575c27a967a37931af3deb6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2017-06-28 22:07:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7cdc2d58b3a6ac45b7adc44d9d03c89f2b5dddae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2017-06-28 22:59:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"18890732fb82eeb0f320ee1edf958703b9edaa1a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-29 00:16:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/37/478637/3/check/gate-octavia-docs-ubuntu-xenial/25d38c4//doc/build/html/ : SUCCESS in 4m 03s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/37/478637/3/check/gate-octavia-pep8-ubuntu-xenial/4a1ed5e/ : SUCCESS in 4m 03s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/37/478637/3/check/gate-octavia-python27-ubuntu-xenial/965adbd/ : SUCCESS in 5m 01s\n- gate-octavia-python35 http://logs.openstack.org/37/478637/3/check/gate-octavia-python35/79048bf/ : SUCCESS in 4m 09s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/37/478637/3/check/octavia-coverage-ubuntu-xenial/1b71d8a/ : SUCCESS in 6m 06s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/37/478637/3/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/5b4db72/ : SUCCESS in 51m 32s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/37/478637/3/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/198a2ee/ : SUCCESS in 58m 47s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/37/478637/3/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/0a7cc30/ : SUCCESS in 35m 44s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/37/478637/3/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/08e0d03/ : SUCCESS in 1h 16m 49s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/37/478637/3/check/gate-octavia-tox-functional-ubuntu-xenial/dc9094a/ : SUCCESS in 4m 52s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/37/478637/3/check/gate-octavia-tox-functional-py35-ubuntu-xenial/17a5e6a/ : SUCCESS in 5m 34s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/37/478637/3/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/f6620d9/ : SUCCESS in 29m 10s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/37/478637/3/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/d432e09/ : SUCCESS in 26m 26s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/37/478637/3/check/gate-octavia-v1-dsvm-scenario-multinode/8b1c3b2/ : SUCCESS in 34m 38s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/37/478637/3/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/1df9a63/ : SUCCESS in 31m 49s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/37/478637/3/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/d1d0dc0/ : SUCCESS in 47m 40s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/37/478637/3/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/f051449/ : FAILURE in 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f94665b6f3a5232d5deda36fd201d69baefaba1e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2017-06-29 01:20:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9928a67c4eef373ffc09976dc9a2aec73fcf5c23","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2017-06-29 02:25:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2eb878825e9f27436874f26057f88b6166744994","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-29 02:25:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"9d82f02889e088c58fcc3e216d9332b15bfad986","date":"2017-06-29 03:19:33.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"ac081ce70b8c4248530ac122449470a64f53ac80","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-29 03:19:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/37/478637/3/gate/gate-octavia-docs-ubuntu-xenial/61f35ef//doc/build/html/ : SUCCESS in 3m 14s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/37/478637/3/gate/gate-octavia-pep8-ubuntu-xenial/8762133/ : SUCCESS in 3m 44s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/37/478637/3/gate/gate-octavia-python27-ubuntu-xenial/d8bcc2b/ : SUCCESS in 4m 05s\n- gate-octavia-python35 http://logs.openstack.org/37/478637/3/gate/gate-octavia-python35/0503a40/ : SUCCESS in 4m 23s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/37/478637/3/gate/octavia-coverage-ubuntu-xenial/dc47df6/ : SUCCESS in 6m 46s\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/37/478637/3/gate/gate-octavia-tox-functional-ubuntu-xenial/491cbe2/ : SUCCESS in 5m 54s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/37/478637/3/gate/gate-octavia-tox-functional-py35-ubuntu-xenial/ed2af43/ : SUCCESS in 6m 41s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/37/478637/3/gate/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/0b6a5b8/ : SUCCESS in 40m 43s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/37/478637/3/gate/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/2fbf49e/ : SUCCESS in 39m 10s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/37/478637/3/gate/gate-octavia-v1-dsvm-scenario-multinode/dd681bf/ : SUCCESS in 33m 42s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/37/478637/3/gate/gate-octavia-v1-dsvm-py3x-scenario-multinode/343dba0/ : SUCCESS in 34m 03s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/37/478637/3/gate/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/9252905/ : SUCCESS in 52m 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"55305c5f6ce6b0e97ccbad5b980c21ce9f686603","revisions":{"ea43726fa47bb2467b9472f51036540f45de8689":{"kind":"REWORK","_number":1,"created":"2017-06-28 21:09:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/478637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/478637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/478637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/478637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/478637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/478637/1"}}},"commit":{"parents":[{"commit":"3ccd8a31624e2839d1e19f5cd52ac3341303675c","subject":"Merge \"Allow operators to disallow creation of TLS Termination listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ccd8a31624e2839d1e19f5cd52ac3341303675c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-06-28 21:08:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-06-28 21:09:36.000000000","tz":-420},"subject":"Enable uwsgi support for octavia devstack","message":"Enable uwsgi support for octavia devstack\n\nThe Pike goal [1] has changed to requiring uwsgi for projects.\nThis patch enables octavia to run under uwsgi.\n\n[1] https://governance.openstack.org/tc/goals/pike/deploy-api-in-wsgi.html\n\nChange-Id: I4f5bafe8d194b23a55cb4de0ba42ad7b6eb7a6cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea43726fa47bb2467b9472f51036540f45de8689"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea43726fa47bb2467b9472f51036540f45de8689"}]},"branch":"refs/heads/master"},"1d6becde9131b186f18a73d46d34a065f5d7dafa":{"kind":"REWORK","_number":2,"created":"2017-06-28 22:07:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/478637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/478637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/478637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/478637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/478637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/478637/2"}}},"commit":{"parents":[{"commit":"3ccd8a31624e2839d1e19f5cd52ac3341303675c","subject":"Merge \"Allow operators to disallow creation of TLS Termination listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ccd8a31624e2839d1e19f5cd52ac3341303675c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-06-28 21:08:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-06-28 22:06:45.000000000","tz":-420},"subject":"Enable uwsgi support for octavia devstack","message":"Enable uwsgi support for octavia devstack\n\nThe Pike goal [1] has changed to requiring uwsgi for projects.\nThis patch enables octavia to run under uwsgi.\n\n[1] https://governance.openstack.org/tc/goals/pike/deploy-api-in-wsgi.html\n\nChange-Id: I4f5bafe8d194b23a55cb4de0ba42ad7b6eb7a6cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d6becde9131b186f18a73d46d34a065f5d7dafa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d6becde9131b186f18a73d46d34a065f5d7dafa"}]},"branch":"refs/heads/master"},"55305c5f6ce6b0e97ccbad5b980c21ce9f686603":{"kind":"REWORK","_number":3,"created":"2017-06-28 22:59:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/478637/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/478637/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/478637/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/478637/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/478637/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/478637/3"}}},"commit":{"parents":[{"commit":"3ccd8a31624e2839d1e19f5cd52ac3341303675c","subject":"Merge \"Allow operators to disallow creation of TLS Termination listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ccd8a31624e2839d1e19f5cd52ac3341303675c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-06-28 21:08:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-06-28 22:59:00.000000000","tz":-420},"subject":"Enable uwsgi support for octavia devstack","message":"Enable uwsgi support for octavia devstack\n\nThe Pike goal [1] has changed to requiring uwsgi for projects.\nThis patch enables octavia to run under uwsgi.\n\n[1] https://governance.openstack.org/tc/goals/pike/deploy-api-in-wsgi.html\n\nChange-Id: I4f5bafe8d194b23a55cb4de0ba42ad7b6eb7a6cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55305c5f6ce6b0e97ccbad5b980c21ce9f686603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55305c5f6ce6b0e97ccbad5b980c21ce9f686603"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
