)]}'
{"id":"openstack%2Foctavia~498909","triplet_id":"openstack%2Foctavia~master~I82f7e39e40663892438c0dd2a1cedcf1459cc337","project":"openstack/octavia","branch":"master","topic":"modify-diskimage-elements-for-suse-support","hashtags":[],"change_id":"I82f7e39e40663892438c0dd2a1cedcf1459cc337","subject":"Enable devstack octavia plugin to support SUSE distros","status":"ABANDONED","created":"2017-08-29 18:44:25.000000000","updated":"2024-05-07 16:00:58.000000000","total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"577de86da5073a7eed06c39bf7e8199f8895da60","_number":498909,"virtual_id_number":498909,"owner":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"_account_id":13770,"name":"Adolfo Duarte","email":"adduarte@redhat.com","username":"adolfoduarte"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-18 16:56:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13770,"name":"Adolfo Duarte","email":"adduarte@redhat.com","username":"adolfoduarte"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"_account_id":13770,"name":"Adolfo Duarte","email":"adduarte@redhat.com","username":"adolfoduarte"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"date":"2020-05-18 14:41:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13770,"name":"Adolfo Duarte","email":"adduarte@redhat.com","username":"adolfoduarte"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"_account_id":13770,"name":"Adolfo Duarte","email":"adduarte@redhat.com","username":"adolfoduarte"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-29 19:04:03.000000000","updated_by":{"_account_id":13770,"name":"Adolfo Duarte","email":"adduarte@redhat.com","username":"adolfoduarte"},"reviewer":{"_account_id":13770,"name":"Adolfo Duarte","email":"adduarte@redhat.com","username":"adolfoduarte"},"state":"REVIEWER"},{"updated":"2017-08-29 20:03:54.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-10-20 06:38:36.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2017-10-27 16:36:51.000000000","updated_by":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"reviewer":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"state":"REVIEWER"},{"updated":"2018-10-15 21:10:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-05-15 21:27:22.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-05-18 16:56:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"897e657c9ea97ad5c181c558ede624ee1f574952","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-08-29 18:44:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c3f2886f90ff7332afefffe212a1db5dc6e97db","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-08-29 18:47:54.000000000","message":"Patch Set 1:\n\nThis is a work in progress patch.","accounts_in_message":[],"_revision_number":1},{"id":"136c962ba5140f437c6030224f66969807b35dfb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-29 20:03:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-octavia-docs-ubuntu-xenial http://docs-draft.openstack.org/09/498909/1/check/gate-octavia-docs-ubuntu-xenial/7e14e9d//doc/build/html/ : SUCCESS in 6m 24s\n- gate-octavia-pep8-ubuntu-xenial http://logs.openstack.org/09/498909/1/check/gate-octavia-pep8-ubuntu-xenial/157bac8/ : SUCCESS in 6m 56s\n- gate-octavia-python27-ubuntu-xenial http://logs.openstack.org/09/498909/1/check/gate-octavia-python27-ubuntu-xenial/7b2f08b/ : SUCCESS in 4m 22s\n- gate-octavia-python35 http://logs.openstack.org/09/498909/1/check/gate-octavia-python35/ebeb4fb/ : SUCCESS in 4m 28s\n- octavia-coverage-ubuntu-xenial http://logs.openstack.org/09/498909/1/check/octavia-coverage-ubuntu-xenial/f276abf/ : SUCCESS in 8m 04s\n- gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial http://logs.openstack.org/09/498909/1/check/gate-neutron-lbaasv2-dsvm-api-ubuntu-xenial/b4d5db9/ : SUCCESS in 59m 54s\n- gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv http://logs.openstack.org/09/498909/1/check/gate-neutron-lbaasv2-dsvm-py3x-api-ubuntu-xenial-nv/a3dd034/ : SUCCESS in 57m 09s (non-voting)\n- gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv http://logs.openstack.org/09/498909/1/check/gate-neutron-lbaasv2-dsvm-scenario-ubuntu-xenial-nv/d17d6d3/ : SUCCESS in 1h 16m 46s (non-voting)\n- gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv http://logs.openstack.org/09/498909/1/check/gate-neutron-lbaasv2-dsvm-py3x-scenario-ubuntu-xenial-nv/71e119f/ : SUCCESS in 1h 15m 04s (non-voting)\n- gate-octavia-tox-functional-ubuntu-xenial http://logs.openstack.org/09/498909/1/check/gate-octavia-tox-functional-ubuntu-xenial/cebc118/ : SUCCESS in 7m 14s\n- gate-octavia-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/09/498909/1/check/gate-octavia-tox-functional-py35-ubuntu-xenial/61393a9/ : SUCCESS in 7m 41s\n- gate-octavia-v1-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/09/498909/1/check/gate-octavia-v1-dsvm-scenario-ubuntu-xenial/551512f/ : SUCCESS in 53m 12s\n- gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial http://logs.openstack.org/09/498909/1/check/gate-octavia-v1-dsvm-py3x-scenario-ubuntu-xenial/9a84fc6/ : SUCCESS in 41m 09s\n- gate-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/09/498909/1/check/gate-octavia-v1-dsvm-scenario-multinode/18e6769/ : SUCCESS in 59m 36s\n- gate-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/09/498909/1/check/gate-octavia-v1-dsvm-py3x-scenario-multinode/f2d8b3d/ : SUCCESS in 1h 00m 09s\n- gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv http://logs.openstack.org/09/498909/1/check/gate-octavia-v1-dsvm-scenario-kvm-centos.7-ubuntu-xenial-nv/3896b66/ : SUCCESS in 44m 28s (non-voting)\n- gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv http://logs.openstack.org/09/498909/1/check/gate-octavia-v1-dsvm-scenario-lxd-ubuntu-xenial-nv/0e00964/ : FAILURE in 36m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e5e5e42ec4ecb9fdfe59f18ee27088e93effefe","author":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"date":"2017-09-12 16:43:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"490d3c3138a834fe39961676e23add6ac97643d1","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-12 20:51:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cce4a3729c395bd2ea75d5fc23166cd932dba317","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2017-10-20 06:38:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2d5b94a9fa16b8915d4a81bac8928daf4ff33995","author":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"date":"2017-10-20 15:57:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90b1fed24dac6c3ed76b0b060257dc7455cc4794","author":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"date":"2017-10-20 16:39:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0203ce6a9d8c433faac729f9185dcc67dc4e6081","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-20 22:16:46.000000000","message":"Patch Set 1:\n\n(11 comments)\n\nWill address review comments in the follow up patch.","accounts_in_message":[],"_revision_number":1},{"id":"8098b1507b43cdba375ad27c4e3fc6ce58d08fa9","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-20 22:16:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d9f7c692aec2015425026c9b3395f6a3feca73b0","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-20 22:17:16.000000000","message":"Patch Set 2:\n\nAddressed review comments.","accounts_in_message":[],"_revision_number":2},{"id":"574063f0783afaec42b0c65f0cabfe88dc01774d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-20 22:52:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/09/498909/2/check/openstack-tox-pep8/5219e42/ : SUCCESS in 7m 34s\n- openstack-tox-py27 http://logs.openstack.org/09/498909/2/check/openstack-tox-py27/f52c983/ : SUCCESS in 6m 07s\n- openstack-tox-py35 http://logs.openstack.org/09/498909/2/check/openstack-tox-py35/08e6c81/ : SUCCESS in 7m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/498909/2/check/build-openstack-sphinx-docs/713fbea/html/ : SUCCESS in 6m 51s\n- openstack-tox-cover http://logs.openstack.org/09/498909/2/check/openstack-tox-cover/9890b7e/ : SUCCESS in 14m 45s\n- legacy-neutron-lbaasv2-dsvm-api http://logs.openstack.org/09/498909/2/check/legacy-neutron-lbaasv2-dsvm-api/4371b9f/ : FAILURE in 9m 44s\n- legacy-neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/09/498909/2/check/legacy-neutron-lbaasv2-dsvm-py3x-api/0f07f30/ : FAILURE in 11m 59s\n- legacy-neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/09/498909/2/check/legacy-neutron-lbaasv2-dsvm-scenario/2baa993/ : FAILURE in 13m 45s (non-voting)\n- legacy-neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/09/498909/2/check/legacy-neutron-lbaasv2-dsvm-py3x-scenario/2ca077e/ : FAILURE in 10m 32s (non-voting)\n- legacy-octavia-tox-functional http://logs.openstack.org/09/498909/2/check/legacy-octavia-tox-functional/a5885ae/ : SUCCESS in 8m 18s\n- legacy-octavia-tox-functional-py35 http://logs.openstack.org/09/498909/2/check/legacy-octavia-tox-functional-py35/106865c/ : SUCCESS in 12m 58s\n- legacy-octavia-v1-dsvm-scenario http://logs.openstack.org/09/498909/2/check/legacy-octavia-v1-dsvm-scenario/25d1d32/ : FAILURE in 11m 32s\n- legacy-octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/09/498909/2/check/legacy-octavia-v1-dsvm-py3x-scenario/3b498d3/ : FAILURE in 8m 44s\n- legacy-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/09/498909/2/check/legacy-octavia-v1-dsvm-scenario-multinode/594b79e/ : FAILURE in 16m 33s\n- legacy-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/09/498909/2/check/legacy-octavia-v1-dsvm-py3x-scenario-multinode/bdfb377/ : FAILURE in 11m 25s\n- legacy-octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/09/498909/2/check/legacy-octavia-v1-dsvm-scenario-kvm-centos.7/7546883/ : FAILURE in 13m 50s (non-voting)\n- legacy-octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/09/498909/2/check/legacy-octavia-v1-dsvm-scenario-lxd/fe94ba8/ : FAILURE in 14m 03s (non-voting)\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/09/498909/2/check/neutron-lbaasv2-dsvm-api/fa7fb0d/ : FAILURE in 22m 44s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/09/498909/2/check/neutron-lbaasv2-dsvm-py3x-api/21be424/ : FAILURE in 14m 41s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/09/498909/2/check/neutron-lbaasv2-dsvm-scenario/fcb6b33/ : FAILURE in 12m 53s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/09/498909/2/check/neutron-lbaasv2-dsvm-py3x-scenario/4642d43/ : FAILURE in 10m 23s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/09/498909/2/check/octavia-tox-functional/83c4562/ : SUCCESS in 8m 42s\n- octavia-tox-functional-py35 http://logs.openstack.org/09/498909/2/check/octavia-tox-functional-py35/2a3ffbf/ : SUCCESS in 8m 44s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/09/498909/2/check/octavia-v1-dsvm-scenario/901575f/ : FAILURE in 13m 00s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/09/498909/2/check/octavia-v1-dsvm-py3x-scenario/1e06669/ : FAILURE in 8m 59s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/09/498909/2/check/octavia-v1-dsvm-scenario-multinode/3f05d25/ : FAILURE in 16m 30s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/09/498909/2/check/octavia-v1-dsvm-py3x-scenario-multinode/33940c6/ : FAILURE in 12m 37s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/09/498909/2/check/octavia-v1-dsvm-scenario-kvm-centos.7/b96bf60/ : FAILURE in 13m 49s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/09/498909/2/check/octavia-v1-dsvm-scenario-lxd/1cbf9e7/ : FAILURE in 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"522195f6a1d183e37bcfaf29da759a0c20177747","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-24 17:42:20.000000000","message":"Patch Set 2: Code-Review-1\n\nStill work in progress","accounts_in_message":[],"_revision_number":2},{"id":"c66e90494ebebe9299c792bb03eeb311f87e765a","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-24 22:56:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e683024b5f581a65534a0753bd9dc9f51cff9dc8","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-24 22:57:11.000000000","message":"Patch Set 3: Code-Review-1\n\nWork in progress","accounts_in_message":[],"_revision_number":3},{"id":"3e8b42a72c660fc8957900c0522b26901d21ed98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-24 23:36:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/09/498909/3/check/openstack-tox-pep8/8373acc/ : SUCCESS in 7m 43s\n- openstack-tox-py27 http://logs.openstack.org/09/498909/3/check/openstack-tox-py27/a4969d6/ : SUCCESS in 5m 34s\n- openstack-tox-py35 http://logs.openstack.org/09/498909/3/check/openstack-tox-py35/f402aff/ : SUCCESS in 6m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/498909/3/check/build-openstack-sphinx-docs/e410969/html/ : SUCCESS in 7m 28s\n- openstack-tox-cover http://logs.openstack.org/09/498909/3/check/openstack-tox-cover/7ada962/cover/ : SUCCESS in 10m 23s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/09/498909/3/check/neutron-lbaasv2-dsvm-api/705a408/ : FAILURE in 9m 34s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/09/498909/3/check/neutron-lbaasv2-dsvm-py3x-api/eb496f1/ : FAILURE in 9m 17s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/09/498909/3/check/neutron-lbaasv2-dsvm-scenario/1cc0769/ : FAILURE in 13m 32s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/09/498909/3/check/neutron-lbaasv2-dsvm-py3x-scenario/4c2b454/ : FAILURE in 8m 09s (non-voting)\n- legacy-octavia-tox-functional http://logs.openstack.org/09/498909/3/check/legacy-octavia-tox-functional/b211284/ : SUCCESS in 8m 24s\n- legacy-octavia-tox-functional-py35 http://logs.openstack.org/09/498909/3/check/legacy-octavia-tox-functional-py35/45a6bd8/ : SUCCESS in 7m 39s\n- legacy-octavia-v1-dsvm-scenario http://logs.openstack.org/09/498909/3/check/legacy-octavia-v1-dsvm-scenario/16dc2d6/ : FAILURE in 8m 15s\n- legacy-octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/09/498909/3/check/legacy-octavia-v1-dsvm-py3x-scenario/f8ace8a/ : FAILURE in 10m 55s\n- legacy-octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/09/498909/3/check/legacy-octavia-v1-dsvm-scenario-multinode/01977b1/ : FAILURE in 13m 43s\n- legacy-octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/09/498909/3/check/legacy-octavia-v1-dsvm-py3x-scenario-multinode/c00e494/ : FAILURE in 12m 12s\n- legacy-octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/09/498909/3/check/legacy-octavia-v1-dsvm-scenario-kvm-centos.7/97c9b03/ : FAILURE in 13m 18s (non-voting)\n- legacy-octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/09/498909/3/check/legacy-octavia-v1-dsvm-scenario-lxd/4e4e43d/ : FAILURE in 18m 18s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/09/498909/3/check/octavia-tox-functional/c763a74/ : SUCCESS in 8m 20s\n- octavia-tox-functional-py35 http://logs.openstack.org/09/498909/3/check/octavia-tox-functional-py35/6f054ed/ : SUCCESS in 8m 21s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/09/498909/3/check/octavia-v1-dsvm-scenario/56e07d3/ : FAILURE in 8m 48s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/09/498909/3/check/octavia-v1-dsvm-py3x-scenario/28666bb/ : FAILURE in 11m 58s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/09/498909/3/check/octavia-v1-dsvm-scenario-multinode/7def672/ : FAILURE in 13m 47s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/09/498909/3/check/octavia-v1-dsvm-py3x-scenario-multinode/6831118/ : FAILURE in 13m 29s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/09/498909/3/check/octavia-v1-dsvm-scenario-kvm-centos.7/5860de4/ : FAILURE in 12m 07s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/09/498909/3/check/octavia-v1-dsvm-scenario-lxd/e37173f/ : FAILURE in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"46cec113aecbf01edff7f641208c6ea5df0f62fe","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-25 19:10:16.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"0a45f8d506e69ce7ce83acd89141da3d4268ff38","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-25 19:10:44.000000000","message":"Patch Set 4: Code-Review-1\n\nWork in progress","accounts_in_message":[],"_revision_number":4},{"id":"d19066b8d7e0d6fd077ff52e4f409f1fea70539a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-25 20:45:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/09/498909/4/check/openstack-tox-pep8/7c883d4/ : SUCCESS in 11m 45s\n- openstack-tox-py27 http://logs.openstack.org/09/498909/4/check/openstack-tox-py27/a103bba/ : SUCCESS in 10m 26s\n- openstack-tox-py35 http://logs.openstack.org/09/498909/4/check/openstack-tox-py35/c5f0f24/ : SUCCESS in 5m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/498909/4/check/build-openstack-sphinx-docs/1ecdb3c/html/ : SUCCESS in 11m 54s\n- openstack-tox-cover http://logs.openstack.org/09/498909/4/check/openstack-tox-cover/ce8c01b/cover/ : SUCCESS in 12m 44s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/09/498909/4/check/neutron-lbaasv2-dsvm-api/2886ae5/ : FAILURE in 1h 17m 29s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/09/498909/4/check/neutron-lbaasv2-dsvm-py3x-api/f4e3af6/ : SUCCESS in 1h 05m 01s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/09/498909/4/check/neutron-lbaasv2-dsvm-scenario/32141ba/ : SUCCESS in 1h 24m 59s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/09/498909/4/check/neutron-lbaasv2-dsvm-py3x-scenario/2d5b966/ : FAILURE in 1h 27m 11s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/09/498909/4/check/octavia-tox-functional/8c57cf9/ : SUCCESS in 12m 54s\n- octavia-tox-functional-py35 http://logs.openstack.org/09/498909/4/check/octavia-tox-functional-py35/fb9efd9/ : SUCCESS in 8m 48s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/09/498909/4/check/octavia-v1-dsvm-scenario/697411a/ : SUCCESS in 52m 37s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/09/498909/4/check/octavia-v1-dsvm-py3x-scenario/e8a4d5e/ : SUCCESS in 51m 44s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/09/498909/4/check/octavia-v1-dsvm-scenario-multinode/74e171d/ : SUCCESS in 55m 44s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/09/498909/4/check/octavia-v1-dsvm-py3x-scenario-multinode/d2b134c/ : SUCCESS in 1h 04m 27s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/09/498909/4/check/octavia-v1-dsvm-scenario-kvm-centos.7/764e7b5/ : SUCCESS in 54m 11s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/09/498909/4/check/octavia-v1-dsvm-scenario-lxd/77d52f9/ : FAILURE in 39m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b4ea438c0028b35be1816ef243800ca2cace61bf","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-26 17:13:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9ff59ff567b669bdf3fbde7dc3540bf0bfbcf154","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-26 20:55:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/498909/4/check/openstack-tox-pep8/9be7297/ : SUCCESS in 10m 03s\n- openstack-tox-py27 http://logs.openstack.org/09/498909/4/check/openstack-tox-py27/40088fc/ : SUCCESS in 5m 39s\n- openstack-tox-py35 http://logs.openstack.org/09/498909/4/check/openstack-tox-py35/c6605d5/ : SUCCESS in 6m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/498909/4/check/build-openstack-sphinx-docs/e76a369/html/ : SUCCESS in 9m 26s\n- openstack-tox-cover http://logs.openstack.org/09/498909/4/check/openstack-tox-cover/acfee1f/cover/ : SUCCESS in 10m 56s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/09/498909/4/check/neutron-lbaasv2-dsvm-api/fbf983c/ : SUCCESS in 58m 46s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/09/498909/4/check/neutron-lbaasv2-dsvm-py3x-api/8124d24/ : SUCCESS in 1h 06m 49s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/09/498909/4/check/neutron-lbaasv2-dsvm-scenario/8d04906/ : SUCCESS in 55m 51s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/09/498909/4/check/neutron-lbaasv2-dsvm-py3x-scenario/8db7d8a/ : SUCCESS in 1h 29m 53s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/09/498909/4/check/octavia-tox-functional/c052682/ : SUCCESS in 7m 54s\n- octavia-tox-functional-py35 http://logs.openstack.org/09/498909/4/check/octavia-tox-functional-py35/9223d8a/ : SUCCESS in 9m 02s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/09/498909/4/check/octavia-v1-dsvm-scenario/3a54d72/ : SUCCESS in 50m 12s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/09/498909/4/check/octavia-v1-dsvm-py3x-scenario/ff2162f/ : SUCCESS in 1h 00m 57s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/09/498909/4/check/octavia-v1-dsvm-scenario-multinode/b35daf5/ : SUCCESS in 1h 04m 47s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/09/498909/4/check/octavia-v1-dsvm-py3x-scenario-multinode/7ae1563/ : SUCCESS in 1h 10m 35s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/09/498909/4/check/octavia-v1-dsvm-scenario-kvm-centos.7/5584643/ : SUCCESS in 51m 55s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/09/498909/4/check/octavia-v1-dsvm-scenario-lxd/12f2a5f/ : FAILURE in 39m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"17e166914c15f97b0a74da991a32713c7e91d17a","author":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"date":"2017-10-27 16:36:51.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0edb5b5faa82a9f6134eadb8b51788b35711dd26","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-27 23:09:12.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nreplied to review comments.","accounts_in_message":[],"_revision_number":4},{"id":"38121b38b0be1f2432528716c2f3ab3440bea1e2","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-27 23:09:42.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"c873e21aaea1efb86b1fea26a51ba6d3290b4501","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2017-10-27 23:10:23.000000000","message":"Patch Set 5: Code-Review-1\n\nWork in progress.\nStill appreciate if there are any review comments.","accounts_in_message":[],"_revision_number":5},{"id":"250a250354fdfaffa021963f7f74de3658d6d389","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-28 02:30:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/09/498909/5/check/openstack-tox-pep8/bd227bb/ : SUCCESS in 7m 22s\n- openstack-tox-py27 http://logs.openstack.org/09/498909/5/check/openstack-tox-py27/1984d44/ : SUCCESS in 7m 20s\n- openstack-tox-py35 http://logs.openstack.org/09/498909/5/check/openstack-tox-py35/23f36f8/ : SUCCESS in 5m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/498909/5/check/build-openstack-sphinx-docs/cbc547a/html/ : SUCCESS in 8m 24s\n- openstack-tox-cover http://logs.openstack.org/09/498909/5/check/openstack-tox-cover/eea25ae/cover/ : SUCCESS in 10m 22s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/09/498909/5/check/neutron-lbaasv2-dsvm-api/d8354fe/ : SUCCESS in 1h 01m 33s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/09/498909/5/check/neutron-lbaasv2-dsvm-py3x-api/75df72c/ : SUCCESS in 1h 02m 17s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/09/498909/5/check/neutron-lbaasv2-dsvm-scenario/057dfd5/ : SUCCESS in 1h 12m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/09/498909/5/check/neutron-lbaasv2-dsvm-py3x-scenario/78332c1/ : FAILURE in 2h 01m 16s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/09/498909/5/check/octavia-tox-functional/d6e7ebb/ : SUCCESS in 8m 15s\n- octavia-tox-functional-py35 http://logs.openstack.org/09/498909/5/check/octavia-tox-functional-py35/f52501c/ : SUCCESS in 9m 03s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/09/498909/5/check/octavia-v1-dsvm-scenario/b2dce2d/ : SUCCESS in 54m 39s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/09/498909/5/check/octavia-v1-dsvm-py3x-scenario/8730b14/ : FAILURE in 45m 21s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/09/498909/5/check/octavia-v1-dsvm-scenario-multinode/22e8eeb/ : SUCCESS in 1h 12m 39s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/09/498909/5/check/octavia-v1-dsvm-py3x-scenario-multinode/4e895a8/ : SUCCESS in 1h 11m 59s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/09/498909/5/check/octavia-v1-dsvm-scenario-kvm-centos.7/6271004/ : SUCCESS in 56m 16s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/09/498909/5/check/octavia-v1-dsvm-scenario-lxd/35f609b/ : FAILURE in 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cedf359f143afd02638ab8b0fc4573f23c45ffc4","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2017-11-01 08:39:00.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"2285dd85c1e9e8c9ffe60e58c9ac273a85256f4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-01 10:01:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/09/498909/6/check/openstack-tox-pep8/e942a8c/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/09/498909/6/check/openstack-tox-py27/3f0352c/ : SUCCESS in 5m 27s\n- openstack-tox-py35 http://logs.openstack.org/09/498909/6/check/openstack-tox-py35/f8676f7/ : SUCCESS in 5m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/498909/6/check/build-openstack-sphinx-docs/5d4191f/html/ : SUCCESS in 6m 18s\n- openstack-tox-cover http://logs.openstack.org/09/498909/6/check/openstack-tox-cover/6e5aa6f/cover/ : SUCCESS in 13m 07s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/09/498909/6/check/neutron-lbaasv2-dsvm-api/ceeec05/ : FAILURE in 59m 55s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/09/498909/6/check/neutron-lbaasv2-dsvm-py3x-api/4fc0d13/ : SUCCESS in 1h 00m 34s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/09/498909/6/check/neutron-lbaasv2-dsvm-scenario/8431b18/ : SUCCESS in 1h 18m 27s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/09/498909/6/check/neutron-lbaasv2-dsvm-py3x-scenario/2065604/ : SUCCESS in 1h 21m 52s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/09/498909/6/check/octavia-tox-functional/764a4c3/ : SUCCESS in 7m 34s\n- octavia-tox-functional-py35 http://logs.openstack.org/09/498909/6/check/octavia-tox-functional-py35/c5c971b/ : SUCCESS in 7m 34s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/09/498909/6/check/octavia-v1-dsvm-scenario/23a9cd6/ : SUCCESS in 46m 36s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/09/498909/6/check/octavia-v1-dsvm-py3x-scenario/6dd1bb1/ : SUCCESS in 49m 48s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/09/498909/6/check/octavia-v1-dsvm-scenario-multinode/f4d1d1c/ : SUCCESS in 1h 06m 28s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/09/498909/6/check/octavia-v1-dsvm-py3x-scenario-multinode/f6f34d5/ : SUCCESS in 58m 54s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/09/498909/6/check/octavia-v1-dsvm-scenario-kvm-centos.7/6b30dfc/ : SUCCESS in 47m 30s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/09/498909/6/check/octavia-v1-dsvm-scenario-lxd/a453eec/ : FAILURE in 38m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ca58cec88499933c0c74e445d23e761c63cbf926","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-10-15 18:26:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2cce75ae5bc095ac8a2890569ccb757ec60a6228","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2018-10-15 21:10:22.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nI know this is WIP, but a couple of comments:\n1. There are three different things going on in this patch; changing the devstack plugin, adding the ability to build an amp image, and updating the amphora agent code. Can we split these out?\n2. We have moved away from having distro specific elements to having elements that work with all of the distributions. Can we please merge the opensuse changes into the existing amphora-agent and haproxy elements? Ping me if you have questions on how to do that.\n3. Do we really need to put in a permanent ifcfg-o-hm0 file on the devstack hosts? It seems like that would be annoying on reboots after devstack has been shutdown.","accounts_in_message":[],"_revision_number":7},{"id":"ded668a7f818f85ff4969d8230884b749766ef92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-15 23:15:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/09/498909/7/check/openstack-tox-cover/20bf57e/ : FAILURE in 7m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/498909/7/check/openstack-tox-lower-constraints/de6eceb/ : SUCCESS in 7m 47s\n- openstack-tox-pep8 http://logs.openstack.org/09/498909/7/check/openstack-tox-pep8/86521a4/ : FAILURE in 7m 04s\n- openstack-tox-py27 http://logs.openstack.org/09/498909/7/check/openstack-tox-py27/316dea1/ : SUCCESS in 6m 42s\n- openstack-tox-py35 http://logs.openstack.org/09/498909/7/check/openstack-tox-py35/7723205/ : SUCCESS in 5m 52s\n- openstack-tox-py36 http://logs.openstack.org/09/498909/7/check/openstack-tox-py36/626314b/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/09/498909/7/check/openstack-tox-docs/d1220dd/html/ : SUCCESS in 6m 54s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/09/498909/7/check/neutron-lbaasv2-dsvm-api/4416f24/ : SUCCESS in 1h 15m 09s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/09/498909/7/check/neutron-lbaasv2-dsvm-py3x-api/50e42d3/ : SUCCESS in 1h 14m 14s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/09/498909/7/check/neutron-lbaasv2-dsvm-scenario/72d258c/ : SUCCESS in 1h 15m 23s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/09/498909/7/check/neutron-lbaasv2-dsvm-py3x-scenario/82a0c78/ : SUCCESS in 52m 56s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/09/498909/7/check/openstack-tox-functional/354021f/ : SUCCESS in 6m 31s\n- openstack-tox-functional-py35 http://logs.openstack.org/09/498909/7/check/openstack-tox-functional-py35/fcbee1c/ : SUCCESS in 7m 29s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/09/498909/7/check/octavia-v1-dsvm-scenario/a912ad2/ : SUCCESS in 57m 17s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/09/498909/7/check/octavia-v1-dsvm-py3x-scenario/6eef818/ : SUCCESS in 1h 02m 06s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/09/498909/7/check/octavia-v1-dsvm-scenario-multinode/ec1de26/ : FAILURE in 1h 29m 13s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/09/498909/7/check/octavia-v1-dsvm-py3x-scenario-multinode/5ec36a9/ : FAILURE in 1h 06m 37s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/09/498909/7/check/octavia-v2-dsvm-noop-api/dfc823c/ : SUCCESS in 39m 04s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/09/498909/7/check/octavia-v2-dsvm-noop-py35-api/8b00237/ : SUCCESS in 46m 05s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/09/498909/7/check/octavia-v2-dsvm-scenario/34bbfc7/ : SUCCESS in 1h 23m 05s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/09/498909/7/check/octavia-v2-dsvm-py35-scenario/6f7170b/ : SUCCESS in 1h 18m 16s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/09/498909/7/check/octavia-v2-dsvm-scenario-centos-7/eb149e1/ : SUCCESS in 2h 07m 44s\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/09/498909/7/check/octavia-v2-dsvm-scenario-ubuntu-bionic/43d79f7/ : FAILURE in 1h 52m 12s (non-voting)\n- octavia-grenade http://logs.openstack.org/09/498909/7/check/octavia-grenade/5eae456/ : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":7},{"id":"4640ddaa625af07d1c6dbc22ad71653303e681e9","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2018-10-17 21:41:50.000000000","message":"Patch Set 7:\n\nThanks for the comments, I will split this patch into respective modules to make it simpler for merging.","accounts_in_message":[],"_revision_number":7},{"id":"46cd573ad7d0992f1c035212188c79def7fc72bd","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-08-21 22:12:45.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4ee251027c0a15ac505f488c1d83c023ca9f565c","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-08-21 22:15:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"10887630bdd9c3777e6b8a49cd4fcfa245755365","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-08-21 22:22:18.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"678e470e467d84318d2667d0d7caab023b46c88f","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-08-21 22:23:22.000000000","message":"Patch Set 10:\n\nBased on the comments from patch set 7 I am splitting this up into three different patches, one for devstack plugin related changes, other for the Amphor image dib elements and the other for the osutils.","accounts_in_message":[],"_revision_number":10},{"id":"df80bec9008528acdc55e55ae72633c2039e846a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 00:40:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80c14736b6ac422b9b799bfb4cb2ac7e : SUCCESS in 11m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c48a66f590748a6994844d9c8704b9a : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05adfd43a55b47d0acf540451a62f18f : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32338a63e49246dba4238a5f7647b150 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/458ef6a6506f4bd99a27df128f22e1a4 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7a76c0eea704888a9c6be9bb53ec74e : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/307ea2cbee0a42b282b83c3d5e36f587 : SUCCESS in 8m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a6f06357201942a4bb1d6c55acc07fd9 : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b3d922d841aa4e78b6f298bfaef35eb5 : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/66e70c4cbe3d42cca633f3a7debb1b50 : SUCCESS in 47m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b66f863c8cca4fa9b9c9065501d5dd65 : SUCCESS in 44m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7f9a8b73f51423da3be4b71d7bf0347 : SUCCESS in 1h 35m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8eae0aee61fc474ea0707fdeb7bfc31c : SUCCESS in 1h 42m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1b0b2a00d6c94535bfa770d04084795e : SUCCESS in 2h 06m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dffedc92e7f741479433b09f368647eb : SUCCESS in 2h 17m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/587e4b0c938b473092ff246a4abe96d2 : SUCCESS in 1h 45m 21s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bc2ae1a1f7c94526a9a921539724edbf : SUCCESS in 1h 16m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/45b37cbed7b640e2bac068363acafdbb : SUCCESS in 54m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d122554f63c84db9b0c1a70e2e904bc8 : SUCCESS in 1h 38m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e684aa0bc9f64382902ed2ca582bb852 : SUCCESS in 1h 23m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f62c6a2d214844708075127673704a19 : SUCCESS in 1h 17m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/14ccdeb6b77e467ab1ed3d930e2348b6 : FAILURE in 1h 29m 19s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d798fbfed8ef79f1ce2a04ec0b5409992ecc10f3","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-08-26 04:59:24.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"aafb8c7117a6c3a91ff2fb9872fbf91c2cbd849b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 07:24:17.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80f8f62193d741edbb76f02351afc7b5 : SUCCESS in 8m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb6413db8dcf453cbf5236ad6ce759f2 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba759bb2c919416ca6e599eb44ae8199 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c7afb7a1d7b4095afda999cb68d1bc1 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f330584a892a4167a631eaf93365ca6c : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e6d4b6061654083b7915865d919e7cb : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea234c71201f47efba32b3d20519d572 : SUCCESS in 9m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/762ea19d14c14d12832b252c3d7b5454 : SUCCESS in 4m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/11d86a60a6ee4b83a83f1022cdd860c4 : SUCCESS in 6m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9e084f46549c46aeb571bb9d95a1c7ff : SUCCESS in 7m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d02445cb52934404835d28203e6679be : SUCCESS in 5m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/842f5e5af09248629720a86de1891cd6 : SUCCESS in 44m 44s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/657c56a8c37d40d6933e84ae486584f1 : SUCCESS in 49m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ee3170e6bb44f12937a7a49d4d7d46b : SUCCESS in 1h 31m 26s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3f482026559c431dbd081367fc91075f : SUCCESS in 1h 43m 32s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9eb6dee517f24d11b8a4cfab9916bab6 : SUCCESS in 2h 21m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eb97323c1bc44841bf298715cb485f71 : SUCCESS in 1h 20m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4e419a7791534050a0700948e54f7cbb : SUCCESS in 1h 08m 54s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5bfaa6961508476bb32a94199a4bdd44 : SUCCESS in 1h 06m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dc63ae43743e4d0bae4262696e179017 : SUCCESS in 45m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/590742cc8811415b8ed44a85e62dbdc7 : SUCCESS in 1h 35m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ffee7697bb18482c80829193885a05f8 : SUCCESS in 1h 02m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/26527e0455cb45ad8212777f2940bbb7 : SUCCESS in 1h 13m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f4a47e131da549a890c6da0e30090a90 : FAILURE in 1h 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6815e843e46ff6b14e59b537887bfa5a5684acc5","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-05-15 21:19:33.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"70693b3ad7b13cf4756e26130dcd5b12d11c540f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:15:21.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28be61bc4d0d49ea8da4e90e53f2012c : SUCCESS in 11m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53bc9e10c0b446af8c706222a7934655 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f755da20024db49d06eadf05d08548 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c3e11d2edd2401fa370778cc5bb2a02 : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d535fb827b8049588833e3f721106731 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1def76d339f408c98d1ed9c260d4281 : SUCCESS in 15m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d3eba599f7ea4a0cb0bb2e7e79f1cf7b : SUCCESS in 8m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6d7b1607d92f430d8da6a0c1ca49dbda : SUCCESS in 8m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c54bcee6cd14d138e01894225affded : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3adbfb6574574ca3b33c9a4d13d8da2e : SUCCESS in 51m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2e2a41302154c5cb4d084457caad701 : SUCCESS in 1h 35m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9f8970d363f348a0913bc6b1ace21375 : SUCCESS in 59m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/803888070d4f47ac9fe96142bab57b60 : SUCCESS in 1h 02m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/859f4a5fdfb14b1b9217121ee37503ec : SUCCESS in 1h 08m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b7ae672a974d4b388534d49d3ce9388e : SUCCESS in 1h 27m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/929abc78304c4bd7ad1615a99ccc5c1f : SUCCESS in 1h 03m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7b8c56dd466a4ebcb09ec99adf51c9b3 : FAILURE in 1h 52m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/38b51bfa8df848108ff3abac0673ea66 : SUCCESS in 1h 37m 08s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6afc4b20805eac6da0ea06f18fdd63cec04ab8cf","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-05-18 14:41:10.000000000","message":"Patch Set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"be392683b9214d4c5f907b63893d395c941500af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 16:56:50.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c49b6a48b88d49a5b9b1e9cb214dd1b5 : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/711acbd678664ad9866c16d1a8245f64 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ab5dce137fe41a9b1c43f6dc90fa110 : SUCCESS in 9m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a740a7285dab4a2389f7ce689ea96f40 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0d5ff9bce8a4d9b96f37036953e8cd5 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c8eb97329a4438e97901cad97748c44 : SUCCESS in 12m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6a75931d95f94042b4a12ea571e78944 : SUCCESS in 5m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/51bc58091de04f2b9ff8761baea78b4b : SUCCESS in 9m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bcde5b360f5240a7b0b5407acc1b82aa : SUCCESS in 8m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a1d46ea66c394f9091ec826eac07e78d : SUCCESS in 49m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51bc989d9a9a4a14bcf2d4fd5ee02c56 : SUCCESS in 1h 04m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8a088aea74f74a1bb409d71e05513c40 : SUCCESS in 1h 12m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/75a022e6e5874d4dbf62c98f9959a558 : SUCCESS in 59m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4015554e9d746d380c76850e1bf404e : SUCCESS in 49m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/063948036aa94e57940716b24faaf5eb : SUCCESS in 1h 31m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3466874a763b47959f3f9add5f0da876 : SUCCESS in 1h 04m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c9380cab19194c9eb0d2630ec80aca01 : SUCCESS in 2h 09m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4a613243cabe475b8c01e91d0c310794 : SUCCESS in 1h 55m 50s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"577de86da5073a7eed06c39bf7e8199f8895da60","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-07 16:00:58.000000000","message":"Abandoned\n\nThis patch has not been touched for four years. Abandoning","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"c20f9511e9e6bd21a7d78cdfd23bec0e62a2f2a3","revisions":{"70510829ca8b2d52706e5cef90f003b1b2c98172":{"kind":"REWORK","_number":1,"created":"2017-08-29 18:44:25.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/09/498909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/1"}}},"commit":{"parents":[{"commit":"404e8ad63a3841c5a0cf930c5984971ab26e6c41","subject":"Add ability to build amp image on SUSE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/404e8ad63a3841c5a0cf930c5984971ab26e6c41"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:44:10.000000000","tz":-420},"subject":"(WIP):Support opensuse based Amphora Image for Octavia","message":"(WIP):Support opensuse based Amphora Image for Octavia\n\nProvide support for creating an OpenSUSE based Amphora\nImage for Octavia.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70510829ca8b2d52706e5cef90f003b1b2c98172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70510829ca8b2d52706e5cef90f003b1b2c98172"}]},"branch":"refs/heads/master"},"cfce614e864f2b5052ae22c155b1e425d44b48c3":{"kind":"REWORK","_number":2,"created":"2017-10-20 22:16:58.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/09/498909/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/2"}}},"commit":{"parents":[{"commit":"ec923445e155cfbf62957842bd999f0298d32192","subject":"Merge \"Move Octavia zuulv3 configuration in repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec923445e155cfbf62957842bd999f0298d32192"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-10-20 22:14:36.000000000","tz":-420},"subject":"(WIP):Support SUSE distro based Amphora Image for Octavia","message":"(WIP):Support SUSE distro based Amphora Image for Octavia\n\nProvide support for creating a SUSE distro based Amphora\nImage for Octavia.\nThe elements and changes required to accomodate SUSE distro\nhas been added or edited based on the requirement.\nAlso added a Subclass to configure the SUSE distro specific\nnetworks in osutils.py.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfce614e864f2b5052ae22c155b1e425d44b48c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfce614e864f2b5052ae22c155b1e425d44b48c3"}]},"branch":"refs/heads/master"},"e56bd5bd0e12e2f6816a3c36e4b8177f1dab24dd":{"kind":"REWORK","_number":3,"created":"2017-10-24 22:56:51.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/09/498909/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/3"}}},"commit":{"parents":[{"commit":"ba0f87f84abbb1661a170b66be2deb9d0ee03cfb","subject":"Merge \"Update tests to do not use deprecated test.services()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba0f87f84abbb1661a170b66be2deb9d0ee03cfb"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-10-24 22:56:39.000000000","tz":-420},"subject":"(WIP):Support SUSE distro based Amphora Image for Octavia","message":"(WIP):Support SUSE distro based Amphora Image for Octavia\n\nProvide support for creating a SUSE distro based Amphora\nImage for Octavia.\nThe elements and changes required to accomodate SUSE distro\nhas been added or edited based on the requirement.\nAlso added a Subclass to configure the SUSE distro specific\nnetworks in osutils.py.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e56bd5bd0e12e2f6816a3c36e4b8177f1dab24dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e56bd5bd0e12e2f6816a3c36e4b8177f1dab24dd"}]},"branch":"refs/heads/master"},"4bd9091f8495b672368ca375e287bc683a40e975":{"kind":"REWORK","_number":4,"created":"2017-10-25 19:10:16.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/09/498909/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/4"}}},"commit":{"parents":[{"commit":"ba0f87f84abbb1661a170b66be2deb9d0ee03cfb","subject":"Merge \"Update tests to do not use deprecated test.services()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba0f87f84abbb1661a170b66be2deb9d0ee03cfb"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-10-25 19:08:57.000000000","tz":0},"subject":"(WIP):Support SUSE distro based Amphora Image for Octavia","message":"(WIP):Support SUSE distro based Amphora Image for Octavia\n\nProvide support for creating a SUSE distro based Amphora\nImage for Octavia.\nThe elements and changes required to accomodate SUSE distro\nhas been added or edited based on the requirement.\nAlso added a Subclass to configure the SUSE distro specific\nnetworks in osutils.py.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9091f8495b672368ca375e287bc683a40e975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9091f8495b672368ca375e287bc683a40e975"}]},"branch":"refs/heads/master"},"88aa2151056f19db87450dd7570c5bac04ba6c4a":{"kind":"REWORK","_number":5,"created":"2017-10-27 23:09:42.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/09/498909/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/5"}}},"commit":{"parents":[{"commit":"ba0f87f84abbb1661a170b66be2deb9d0ee03cfb","subject":"Merge \"Update tests to do not use deprecated test.services()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba0f87f84abbb1661a170b66be2deb9d0ee03cfb"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-10-27 23:09:40.000000000","tz":0},"subject":"(WIP):Support SUSE distro based Amphora Image for Octavia","message":"(WIP):Support SUSE distro based Amphora Image for Octavia\n\nProvide support for creating a SUSE distro based Amphora\nImage for Octavia.\nThe elements and changes required to accomodate SUSE distro\nhas been added or edited based on the requirement.\nAlso added a Subclass to configure the SUSE distro specific\nnetworks in osutils.py.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88aa2151056f19db87450dd7570c5bac04ba6c4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88aa2151056f19db87450dd7570c5bac04ba6c4a"}]},"branch":"refs/heads/master"},"c12c51ec12c4b967cf73e13981009a2da96ad374":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2017-11-01 08:39:00.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/09/498909/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/6"}}},"commit":{"parents":[{"commit":"926fb27fbeb41cec405e35f0814e14e152aff927","subject":"Disable kvm on OVH infra instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926fb27fbeb41cec405e35f0814e14e152aff927"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2017-11-01 08:39:00.000000000","tz":0},"subject":"(WIP):Support SUSE distro based Amphora Image for Octavia","message":"(WIP):Support SUSE distro based Amphora Image for Octavia\n\nProvide support for creating a SUSE distro based Amphora\nImage for Octavia.\nThe elements and changes required to accomodate SUSE distro\nhas been added or edited based on the requirement.\nAlso added a Subclass to configure the SUSE distro specific\nnetworks in osutils.py.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c12c51ec12c4b967cf73e13981009a2da96ad374"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c12c51ec12c4b967cf73e13981009a2da96ad374"}]},"branch":"refs/heads/master"},"5dd2f85227cb38a7bf071966696adca303b457ff":{"kind":"REWORK","_number":7,"created":"2018-10-15 18:26:57.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/09/498909/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/7"}}},"commit":{"parents":[{"commit":"ed301d7fa2ca362a75e6a1f1c93b9446c81fda1c","subject":"Merge \"Add API versioning to the api-ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed301d7fa2ca362a75e6a1f1c93b9446c81fda1c"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2018-10-15 18:26:35.000000000","tz":-420},"subject":"(WIP):Support SUSE distro based Amphora Image for Octavia","message":"(WIP):Support SUSE distro based Amphora Image for Octavia\n\nProvide support for creating a SUSE distro based Amphora\nImage for Octavia.\nThe elements and changes required to accomodate SUSE distro\nhas been added or edited based on the requirement.\nAlso added a Subclass to configure the SUSE distro specific\nnetworks in osutils.py.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd2f85227cb38a7bf071966696adca303b457ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd2f85227cb38a7bf071966696adca303b457ff"}]},"branch":"refs/heads/master"},"22b252e0ea5e8ae170310252834952b2236dfa3c":{"kind":"REWORK","_number":8,"created":"2019-08-21 22:12:45.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/09/498909/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/8"}}},"commit":{"parents":[{"commit":"bc8fbd7a7ba0ec270fda671b8bdf7a0a0487e190","subject":"Merge \"Remove amphora-agent build deps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc8fbd7a7ba0ec270fda671b8bdf7a0a0487e190"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-21 22:11:40.000000000","tz":0},"subject":"(WIP):Enable devstack octavia plugin to support SUSE distros","message":"(WIP):Enable devstack octavia plugin to support SUSE distros\n\nEnable devstack octavia plugin support for creating a SUSE\ndistro based Amphora Image for Octavia.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22b252e0ea5e8ae170310252834952b2236dfa3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22b252e0ea5e8ae170310252834952b2236dfa3c"}]},"branch":"refs/heads/master"},"f36e0e105de700d88b12a787af2f08b0fe500604":{"kind":"REWORK","_number":9,"created":"2019-08-21 22:15:21.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/09/498909/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/9"}}},"commit":{"parents":[{"commit":"bc8fbd7a7ba0ec270fda671b8bdf7a0a0487e190","subject":"Merge \"Remove amphora-agent build deps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc8fbd7a7ba0ec270fda671b8bdf7a0a0487e190"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-21 22:15:04.000000000","tz":0},"subject":"(WIP):Enable devstack octavia plugin to support SUSE distros","message":"(WIP):Enable devstack octavia plugin to support SUSE distros\n\nEnable devstack octavia plugin support for creating a SUSE\ndistro based Amphora Image for Octavia.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f36e0e105de700d88b12a787af2f08b0fe500604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f36e0e105de700d88b12a787af2f08b0fe500604"}]},"branch":"refs/heads/master"},"98b1a2aa7b50979573c554e926d51d36b5e98468":{"kind":"REWORK","_number":10,"created":"2019-08-21 22:22:18.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/09/498909/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/10"}}},"commit":{"parents":[{"commit":"7aa928cebdb287b5ef63241ed25d6f47ac214c77","subject":"Update osutil support for SUSE distro","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7aa928cebdb287b5ef63241ed25d6f47ac214c77"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-21 22:22:01.000000000","tz":0},"subject":"(WIP):Enable devstack octavia plugin to support SUSE distros","message":"(WIP):Enable devstack octavia plugin to support SUSE distros\n\nEnable devstack octavia plugin support for creating a SUSE\ndistro based Amphora Image for Octavia.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98b1a2aa7b50979573c554e926d51d36b5e98468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98b1a2aa7b50979573c554e926d51d36b5e98468"}]},"branch":"refs/heads/master"},"6e89fe783c3bd93e114f90447505f3c32b6c2a0a":{"kind":"REWORK","_number":11,"created":"2019-08-26 04:59:24.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/09/498909/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/11"}}},"commit":{"parents":[{"commit":"24309d11547caba580c6082836798b0f28ed0859","subject":"Update osutil support for SUSE distro","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24309d11547caba580c6082836798b0f28ed0859"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-26 04:37:44.000000000","tz":0},"subject":"(WIP):Enable devstack octavia plugin to support SUSE distros","message":"(WIP):Enable devstack octavia plugin to support SUSE distros\n\nEnable devstack octavia plugin support for creating a SUSE\ndistro based Amphora Image for Octavia.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e89fe783c3bd93e114f90447505f3c32b6c2a0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e89fe783c3bd93e114f90447505f3c32b6c2a0a"}]},"branch":"refs/heads/master"},"c1f491e45454fa99e125dbd60b7ddc79868c715c":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-05-15 21:19:33.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/09/498909/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/12"}}},"commit":{"parents":[{"commit":"37898b49c0bfdcf326bdbfe76ea2d7e3e0f07dc3","subject":"Update osutil support for SUSE distro","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37898b49c0bfdcf326bdbfe76ea2d7e3e0f07dc3"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Keith Berger","email":"keith.berger@suse.com","date":"2020-05-15 21:19:33.000000000","tz":0},"subject":"(WIP):Enable devstack octavia plugin to support SUSE distros","message":"(WIP):Enable devstack octavia plugin to support SUSE distros\n\nEnable devstack octavia plugin support for creating a SUSE\ndistro based Amphora Image for Octavia.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1f491e45454fa99e125dbd60b7ddc79868c715c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1f491e45454fa99e125dbd60b7ddc79868c715c"}]},"branch":"refs/heads/master"},"c20f9511e9e6bd21a7d78cdfd23bec0e62a2f2a3":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2020-05-18 14:41:10.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/09/498909/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/498909/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/498909/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/498909/13"}}},"commit":{"parents":[{"commit":"37898b49c0bfdcf326bdbfe76ea2d7e3e0f07dc3","subject":"Update osutil support for SUSE distro","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37898b49c0bfdcf326bdbfe76ea2d7e3e0f07dc3"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-08-29 18:23:50.000000000","tz":-420},"committer":{"name":"Keith Berger","email":"keith.berger@suse.com","date":"2020-05-18 14:40:14.000000000","tz":0},"subject":"Enable devstack octavia plugin to support SUSE distros","message":"Enable devstack octavia plugin to support SUSE distros\n\nEnable devstack octavia plugin support for creating a SUSE\ndistro based Amphora Image for Octavia.\n\nChange-Id: I82f7e39e40663892438c0dd2a1cedcf1459cc337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c20f9511e9e6bd21a7d78cdfd23bec0e62a2f2a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c20f9511e9e6bd21a7d78cdfd23bec0e62a2f2a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
