)]}'
{"id":"openstack%2Foctavia~529191","triplet_id":"openstack%2Foctavia~master~I3b98c693d334a1025f633bd319c88dde724be30e","project":"openstack/octavia","branch":"master","topic":"active-active-with-distributor","hashtags":[],"change_id":"I3b98c693d334a1025f633bd319c88dde724be30e","subject":"ACTIVE-ACTIVE: Amphora driver updates","status":"NEW","created":"2017-12-20 00:34:37.000000000","updated":"2019-06-19 03:39:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"da37097c9d9448f441347fd1114c3d1412e3157e","_number":529191,"virtual_id_number":529191,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":20226,"name":"Ganpat Agarwal","email":"gans.developer@gmail.com","username":"gans"},{"_account_id":20279,"name":"Jason Niesz","email":"Jason.Niesz@walmart.com","username":"jniesz"},{"value":1,"date":"2019-06-19 03:39:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20226,"name":"Ganpat Agarwal","email":"gans.developer@gmail.com","username":"gans"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20279,"name":"Jason Niesz","email":"Jason.Niesz@walmart.com","username":"jniesz"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":20226,"name":"Ganpat Agarwal","email":"gans.developer@gmail.com","username":"gans"},{"_account_id":20279,"name":"Jason Niesz","email":"Jason.Niesz@walmart.com","username":"jniesz"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2019-06-19 01:25:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20226,"name":"Ganpat Agarwal","email":"gans.developer@gmail.com","username":"gans"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20279,"name":"Jason Niesz","email":"Jason.Niesz@walmart.com","username":"jniesz"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":20226,"name":"Ganpat Agarwal","email":"gans.developer@gmail.com","username":"gans"},{"_account_id":20279,"name":"Jason Niesz","email":"Jason.Niesz@walmart.com","username":"jniesz"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-16 18:56:26.000000000","updated_by":{"_account_id":20279,"name":"Jason Niesz","email":"Jason.Niesz@walmart.com","username":"jniesz"},"reviewer":{"_account_id":20279,"name":"Jason Niesz","email":"Jason.Niesz@walmart.com","username":"jniesz"},"state":"REVIEWER"},{"updated":"2018-01-17 06:10:27.000000000","updated_by":{"_account_id":20226,"name":"Ganpat Agarwal","email":"gans.developer@gmail.com","username":"gans"},"reviewer":{"_account_id":20226,"name":"Ganpat Agarwal","email":"gans.developer@gmail.com","username":"gans"},"state":"REVIEWER"},{"updated":"2019-06-19 01:25:09.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-06-19 03:39:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44d6b106c9b5925ebe99686b85da31a9f52fccb6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2017-12-20 00:34:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b93fb0d444599ea6c67685ba994dc000427015f9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2017-12-20 00:34:50.000000000","message":"Patch Set 1: Code-Review-1\n\nWIP","accounts_in_message":[],"_revision_number":1},{"id":"7b2bc79a8f4e173d60ad576a0c27e74aea92214b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2017-12-20 00:34:55.000000000","message":"Patch Set 1: Workflow-1 -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"42286a416db9d88e06e9dd51129a4059c6b9c034","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-20 02:42:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/91/529191/1/check/openstack-tox-pep8/06e1ad8/ : SUCCESS in 6m 52s\n- openstack-tox-py27 http://logs.openstack.org/91/529191/1/check/openstack-tox-py27/d70eedf/ : SUCCESS in 5m 20s\n- openstack-tox-py35 http://logs.openstack.org/91/529191/1/check/openstack-tox-py35/8ffd432/ : SUCCESS in 5m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/91/529191/1/check/build-openstack-sphinx-docs/167106a/html/ : SUCCESS in 5m 57s\n- openstack-tox-cover http://logs.openstack.org/91/529191/1/check/openstack-tox-cover/5a74438/cover/ : SUCCESS in 9m 54s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/91/529191/1/check/neutron-lbaasv2-dsvm-api/8ef6b5f/ : SUCCESS in 1h 01m 55s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/91/529191/1/check/neutron-lbaasv2-dsvm-py3x-api/57c9159/ : SUCCESS in 1h 03m 06s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/91/529191/1/check/neutron-lbaasv2-dsvm-scenario/d4fffb8/ : SUCCESS in 1h 27m 27s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/91/529191/1/check/neutron-lbaasv2-dsvm-py3x-scenario/9e8997b/ : SUCCESS in 1h 59m 03s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/91/529191/1/check/octavia-tox-functional/773cd36/ : SUCCESS in 7m 46s\n- octavia-tox-functional-py35 http://logs.openstack.org/91/529191/1/check/octavia-tox-functional-py35/a92203e/ : SUCCESS in 7m 55s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/91/529191/1/check/octavia-v1-dsvm-scenario/298a0f7/ : SUCCESS in 48m 01s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/91/529191/1/check/octavia-v1-dsvm-py3x-scenario/50334c0/ : SUCCESS in 1h 29m 22s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/91/529191/1/check/octavia-v1-dsvm-scenario-multinode/d6f9805/ : SUCCESS in 53m 08s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/91/529191/1/check/octavia-v1-dsvm-py3x-scenario-multinode/6dca56b/ : SUCCESS in 57m 09s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/91/529191/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/57742be/ : SUCCESS in 48m 40s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/91/529191/1/check/octavia-v1-dsvm-scenario-lxd/585eddb/ : FAILURE in 1h 38m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"302476dd3f86943cc5444214105d8e8bc311b153","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2018-01-12 22:21:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6ac12628c80b29cefd24313fb154b903ac9ea34","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2018-01-12 22:24:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5ad9bc4c5c2f791f925d0393f5b2636a0d3a965e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-12 23:54:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/91/529191/3/check/openstack-tox-pep8/8bee31f/ : SUCCESS in 7m 30s\n- openstack-tox-py27 http://logs.openstack.org/91/529191/3/check/openstack-tox-py27/030187c/ : SUCCESS in 5m 48s\n- openstack-tox-py35 http://logs.openstack.org/91/529191/3/check/openstack-tox-py35/61c1569/ : SUCCESS in 5m 23s\n- build-openstack-sphinx-docs http://logs.openstack.org/91/529191/3/check/build-openstack-sphinx-docs/961efdf/html/ : SUCCESS in 7m 28s\n- openstack-tox-cover http://logs.openstack.org/91/529191/3/check/openstack-tox-cover/118f975/cover/ : SUCCESS in 9m 09s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/91/529191/3/check/neutron-lbaasv2-dsvm-api/93f5e05/ : SUCCESS in 55m 49s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/91/529191/3/check/neutron-lbaasv2-dsvm-py3x-api/e133671/ : SUCCESS in 1h 05m 31s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/91/529191/3/check/neutron-lbaasv2-dsvm-scenario/f8f4ef4/ : SUCCESS in 1h 11m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/91/529191/3/check/neutron-lbaasv2-dsvm-py3x-scenario/27df4d5/ : SUCCESS in 1h 13m 45s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/91/529191/3/check/openstack-tox-functional/6500cbc/ : SUCCESS in 8m 43s\n- openstack-tox-functional-py35 http://logs.openstack.org/91/529191/3/check/openstack-tox-functional-py35/34ea62e/ : SUCCESS in 9m 58s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/91/529191/3/check/octavia-v1-dsvm-scenario/b3e5245/ : SUCCESS in 43m 03s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/91/529191/3/check/octavia-v1-dsvm-py3x-scenario/a9c14ed/ : SUCCESS in 44m 00s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/91/529191/3/check/octavia-v1-dsvm-scenario-multinode/5f2a9a3/ : SUCCESS in 56m 48s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/91/529191/3/check/octavia-v1-dsvm-py3x-scenario-multinode/c09a81e/ : SUCCESS in 51m 30s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/91/529191/3/check/octavia-v1-dsvm-scenario-kvm-centos.7/d20c6ca/ : SUCCESS in 46m 53s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/91/529191/3/check/octavia-v1-dsvm-scenario-lxd/52db2f5/ : FAILURE in 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"88623bb50d1ebf37f8094db4a32afd7dfce11801","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2018-01-14 20:31:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0f83415252bea06b23c5abefc086ca8db2906fc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-14 21:53:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 finger://ze01.openstack.org/0cbcefa404064c87b09927f89d3b57c8 : POST_FAILURE in 7m 12s\n- openstack-tox-py27 finger://ze08.openstack.org/263f1fe748184276aeb5ff87e42b4229 : POST_FAILURE in 4m 06s\n- openstack-tox-py35 finger://ze04.openstack.org/601af9f6b2de40d5827b6d973a20c177 : POST_FAILURE in 9m 19s\n- build-openstack-sphinx-docs finger://ze08.openstack.org/ad4db9dc08894d1d84f2134676143979 : POST_FAILURE in 6m 08s\n- openstack-tox-cover finger://ze10.openstack.org/2b71b6a9f83a457fa7934d075ea80560 : POST_FAILURE in 8m 54s\n- neutron-lbaasv2-dsvm-api finger://ze09/5b682bb1a1114da48e344cd7c38db2a3 : POST_FAILURE in 54m 57s\n- neutron-lbaasv2-dsvm-py3x-api finger://ze02.openstack.org/b3320c90148c4c5b8278b02866b01f2b : POST_FAILURE in 1h 00m 51s\n- neutron-lbaasv2-dsvm-scenario finger://ze01.openstack.org/d6c8ddc1e85f421c846c4b6072d6aa3b : POST_FAILURE in 1h 20m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario finger://ze07.openstack.org/0ef0862aba054e2ebe675613e2b255a5 : POST_FAILURE in 1h 13m 27s (non-voting)\n- openstack-tox-functional finger://ze01.openstack.org/64b4ceb0b8ee46a899e77245d8da3ef1 : POST_FAILURE in 9m 33s\n- openstack-tox-functional-py35 finger://ze10.openstack.org/3f30cfcf522a4e3eab03b4a21b441a62 : POST_FAILURE in 8m 37s\n- octavia-v1-dsvm-scenario finger://ze02.openstack.org/f855212b21d747dfb803e292277848f2 : POST_FAILURE in 36m 43s\n- octavia-v1-dsvm-py3x-scenario finger://ze05.openstack.org/c7e3562544a94889aea677e8b93d5757 : POST_FAILURE in 46m 45s\n- octavia-v1-dsvm-scenario-multinode finger://ze06.openstack.org/2f00bcf331294f83adbf258a89035349 : POST_FAILURE in 52m 07s\n- octavia-v1-dsvm-py3x-scenario-multinode finger://ze04.openstack.org/240599131b984f2083b6f3aac4421177 : POST_FAILURE in 52m 46s\n- octavia-v1-dsvm-scenario-kvm-centos.7 finger://ze08.openstack.org/610a4049043e488db3a36f01915dde2b : POST_FAILURE in 48m 39s (non-voting)\n- octavia-v1-dsvm-scenario-lxd finger://ze08.openstack.org/497e616fd3304a0f9e1eda75031d4758 : POST_FAILURE in 36m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9626fa32a8fbcbf904074a5e08f5cd5b647fd985","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2018-01-15 16:29:04.000000000","message":"Patch Set 4:\n\nrecheck\nzuul","accounts_in_message":[],"_revision_number":4},{"id":"2ed583428dd09bb4d5c511800fc1b29167e92cf1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2018-01-15 18:52:38.000000000","message":"Patch Set 4:\n\nrecheck\nzuul again","accounts_in_message":[],"_revision_number":4},{"id":"de04dfa5405f2afc1d68dd601b5947cbe66fe2bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-15 20:32:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/91/529191/4/check/openstack-tox-pep8/e88e270/ : SUCCESS in 10m 59s\n- openstack-tox-py27 http://logs.openstack.org/91/529191/4/check/openstack-tox-py27/e1bb69b/ : SUCCESS in 6m 20s\n- openstack-tox-py35 http://logs.openstack.org/91/529191/4/check/openstack-tox-py35/9fbf54a/ : SUCCESS in 6m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/91/529191/4/check/build-openstack-sphinx-docs/c95b536/html/ : SUCCESS in 7m 01s\n- openstack-tox-cover http://logs.openstack.org/91/529191/4/check/openstack-tox-cover/6c39809/cover/ : SUCCESS in 10m 47s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/91/529191/4/check/neutron-lbaasv2-dsvm-api/ceb41ae/ : SUCCESS in 1h 07m 50s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/91/529191/4/check/neutron-lbaasv2-dsvm-py3x-api/5a39965/ : SUCCESS in 55m 09s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/91/529191/4/check/neutron-lbaasv2-dsvm-scenario/0631324/ : SUCCESS in 1h 20m 45s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/91/529191/4/check/neutron-lbaasv2-dsvm-py3x-scenario/bb90853/ : SUCCESS in 1h 06m 24s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/91/529191/4/check/openstack-tox-functional/071257c/ : SUCCESS in 8m 31s\n- openstack-tox-functional-py35 http://logs.openstack.org/91/529191/4/check/openstack-tox-functional-py35/8d885b5/ : SUCCESS in 9m 29s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/91/529191/4/check/octavia-v1-dsvm-scenario/62b025c/ : SUCCESS in 1h 03m 41s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/91/529191/4/check/octavia-v1-dsvm-py3x-scenario/c284ae6/ : SUCCESS in 48m 42s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/91/529191/4/check/octavia-v1-dsvm-scenario-multinode/9bc4a4c/ : SUCCESS in 1h 00m 05s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/91/529191/4/check/octavia-v1-dsvm-py3x-scenario-multinode/38212c7/ : SUCCESS in 1h 04m 49s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/91/529191/4/check/octavia-v1-dsvm-scenario-kvm-centos.7/d460e49/ : SUCCESS in 49m 56s (non-voting)\n- octavia-v1-dsvm-scenario-lxd http://logs.openstack.org/91/529191/4/check/octavia-v1-dsvm-scenario-lxd/f9b3b35/ : FAILURE in 37m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bac88c93af9e1ffd95360ccd00c92a8d856fd44a","author":{"_account_id":20279,"name":"Jason Niesz","email":"Jason.Niesz@walmart.com","username":"jniesz"},"date":"2018-01-16 18:56:26.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM, besides making the rest path for plug_vip consistent, as you stated.","accounts_in_message":[],"_revision_number":4},{"id":"0e4e493c6a09fea8d71a66415c25a3c3efb75798","author":{"_account_id":20226,"name":"Ganpat Agarwal","email":"gans.developer@gmail.com","username":"gans"},"date":"2018-01-17 06:10:27.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"045026387a23a5fd0b1d7dc03053b4b39ff507dd","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2018-04-10 14:39:19.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"5366ec3b8ecfd6e2a871610076790a4130b654f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-10 16:15:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/91/529191/5/check/openstack-tox-pep8/5e84d98/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/91/529191/5/check/openstack-tox-py27/de1285b/ : SUCCESS in 5m 02s\n- openstack-tox-py35 http://logs.openstack.org/91/529191/5/check/openstack-tox-py35/b2a24d0/ : SUCCESS in 5m 04s\n- build-openstack-sphinx-docs http://logs.openstack.org/91/529191/5/check/build-openstack-sphinx-docs/b9c08b4/html/ : SUCCESS in 5m 55s\n- openstack-tox-cover http://logs.openstack.org/91/529191/5/check/openstack-tox-cover/ddeeb0f/cover/ : SUCCESS in 7m 53s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/91/529191/5/check/neutron-lbaasv2-dsvm-api/96fad50/ : SUCCESS in 1h 15m 52s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/91/529191/5/check/neutron-lbaasv2-dsvm-scenario/c54c362/ : SUCCESS in 1h 09m 23s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/91/529191/5/check/octavia-v1-dsvm-scenario/96a4fed/ : SUCCESS in 55m 48s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/91/529191/5/check/octavia-v1-dsvm-scenario-multinode/5740b29/ : SUCCESS in 1h 03m 36s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/91/529191/5/check/octavia-v1-dsvm-scenario-kvm-centos.7/b0d5397/ : SUCCESS in 54m 35s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/91/529191/5/check/neutron-lbaasv2-dsvm-py3x-api/4238cac/ : SUCCESS in 1h 15m 08s\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/91/529191/5/check/neutron-lbaasv2-dsvm-py3x-scenario/969246d/ : SUCCESS in 1h 29m 42s (non-voting)\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/91/529191/5/check/octavia-v1-dsvm-py3x-scenario/702dc82/ : SUCCESS in 1h 07m 27s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/91/529191/5/check/octavia-v1-dsvm-py3x-scenario-multinode/006f49e/ : SUCCESS in 1h 16m 06s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/91/529191/5/check/openstack-tox-lower-constraints/4a433f6/ : SUCCESS in 7m 00s\n- openstack-tox-functional http://logs.openstack.org/91/529191/5/check/openstack-tox-functional/7b687b6/ : SUCCESS in 6m 42s\n- openstack-tox-functional-py35 http://logs.openstack.org/91/529191/5/check/openstack-tox-functional-py35/28fd6e5/ : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":5},{"id":"e07489525f91d4eeed3a684e26c14ccd2b8fabc6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-19 01:25:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"087b5897cb284df4427a46fc19c0d71da3e85cb6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-19 01:41:00.000000000","message":"Topic changed from active-active to active-active-with-distributor","accounts_in_message":[],"_revision_number":6},{"id":"da37097c9d9448f441347fd1114c3d1412e3157e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 03:39:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/91/529191/6/check/openstack-tox-cover/77cb94b/cover/ : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/529191/6/check/openstack-tox-lower-constraints/8178b0a/ : SUCCESS in 7m 49s\n- openstack-tox-pep8 http://logs.openstack.org/91/529191/6/check/openstack-tox-pep8/f99aa0c/ : SUCCESS in 6m 52s\n- openstack-tox-py27 http://logs.openstack.org/91/529191/6/check/openstack-tox-py27/c72f414/ : SUCCESS in 5m 12s\n- openstack-tox-py36 http://logs.openstack.org/91/529191/6/check/openstack-tox-py36/5c3d52b/ : SUCCESS in 5m 40s\n- openstack-tox-py37 http://logs.openstack.org/91/529191/6/check/openstack-tox-py37/564b07c/ : SUCCESS in 5m 53s\n- openstack-tox-docs http://logs.openstack.org/91/529191/6/check/openstack-tox-docs/332fd60/html/ : SUCCESS in 6m 54s\n- openstack-tox-functional http://logs.openstack.org/91/529191/6/check/openstack-tox-functional/544a706/ : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/529191/6/check/openstack-tox-functional-py36/e8f602b/ : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/91/529191/6/check/octavia-v2-dsvm-noop-api/f08a6f2/ : SUCCESS in 41m 52s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/91/529191/6/check/octavia-v2-dsvm-noop-py2-api/f7ca0ca/ : SUCCESS in 41m 44s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/91/529191/6/check/octavia-v2-dsvm-scenario/a7f0126/ : SUCCESS in 1h 41m 16s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/91/529191/6/check/octavia-v2-dsvm-py2-scenario/c241bb7/ : SUCCESS in 1h 37m 14s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/91/529191/6/check/octavia-v2-dsvm-py2-scenario-centos-7/08f6705/ : FAILURE in 2h 06m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/91/529191/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/e03e25c/ : SUCCESS in 1h 40m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/91/529191/6/check/octavia-v2-dsvm-scenario-two-node/b0702d6/ : SUCCESS in 1h 27m 12s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/91/529191/6/check/octavia-v2-dsvm-py2-scenario-two-node/b1c02f5/ : SUCCESS in 1h 47m 08s (non-voting)\n- octavia-grenade http://logs.openstack.org/91/529191/6/check/octavia-grenade/877eddc/ : SUCCESS in 1h 10m 54s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/91/529191/6/check/octavia-v2-dsvm-tls-barbican/84f8b54/ : SUCCESS in 52m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/91/529191/6/check/octavia-v2-dsvm-scenario-amphora-v2/f86f527/ : SUCCESS in 1h 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"80d6d3a5b80f0bd1a0e7547e7f8934fe7f78c87c","revisions":{"41567604fca91a3a3fbb18ef95d922e579fe95c0":{"kind":"REWORK","_number":1,"created":"2017-12-20 00:34:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/91/529191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/529191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/529191/1"}}},"commit":{"parents":[{"commit":"2850beb39b01053334d2d91285c0fda126871bc5","subject":"ACTIVE-ACTIVE: Initial distributor data model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2850beb39b01053334d2d91285c0fda126871bc5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-12-19 22:08:21.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-12-20 00:34:29.000000000","tz":-480},"subject":"ACTIVE-ACTIVE: Amphora driver updates","message":"ACTIVE-ACTIVE: Amphora driver updates\n\nThis patch extends the amphora driver to support methods required for\nActive/Active topologies and amphora in the distributor role.\n\nThis patch is a decompostion of the following patch:\nhttps://review.openstack.org/#/c/313006\n\nThis patch also moves a class comment from a previous patch into a class\ndocstring.\n\nStory: 2001288\nTask: 6114\n\nChange-Id: I3b98c693d334a1025f633bd319c88dde724be30e\nCo-Authored-By: Valeria Perelman \u003cperelman@il.ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41567604fca91a3a3fbb18ef95d922e579fe95c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41567604fca91a3a3fbb18ef95d922e579fe95c0"}]},"branch":"refs/heads/master"},"60a95f00681528c0765d1137d6be62a3e8d8ee3b":{"kind":"REWORK","_number":2,"created":"2018-01-12 22:21:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/91/529191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/529191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/529191/2"}}},"commit":{"parents":[{"commit":"9ab04a6997a472bac8f0d0a667c236c693403dfb","subject":"ACTIVE-ACTIVE: Initial distributor data model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ab04a6997a472bac8f0d0a667c236c693403dfb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-12-19 22:08:21.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2018-01-12 22:20:49.000000000","tz":-480},"subject":"ACTIVE-ACTIVE: Amphora driver updates","message":"ACTIVE-ACTIVE: Amphora driver updates\n\nThis patch extends the amphora driver to support methods required for\nActive/Active topologies and amphora in the distributor role.\n\nThis patch is a decompostion of the following patch:\nhttps://review.openstack.org/#/c/313006\n\nThis patch also moves a class comment from a previous patch into a class\ndocstring.\n\nStory: 2001288\nTask: 6114\n\nChange-Id: I3b98c693d334a1025f633bd319c88dde724be30e\nCo-Authored-By: Valeria Perelman \u003cperelman@il.ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60a95f00681528c0765d1137d6be62a3e8d8ee3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60a95f00681528c0765d1137d6be62a3e8d8ee3b"}]},"branch":"refs/heads/master"},"f0d253adcdf7b3afcda613daaf1d2c7fc8a99b08":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-01-12 22:24:57.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/91/529191/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/529191/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/529191/3"}}},"commit":{"parents":[{"commit":"158f29a7977c5e3c0cbebcf50b4ed0c4d919aa16","subject":"ACTIVE-ACTIVE: Initial distributor data model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/158f29a7977c5e3c0cbebcf50b4ed0c4d919aa16"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-12-19 22:08:21.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2018-01-12 22:24:57.000000000","tz":0},"subject":"ACTIVE-ACTIVE: Amphora driver updates","message":"ACTIVE-ACTIVE: Amphora driver updates\n\nThis patch extends the amphora driver to support methods required for\nActive/Active topologies and amphora in the distributor role.\n\nThis patch is a decompostion of the following patch:\nhttps://review.openstack.org/#/c/313006\n\nThis patch also moves a class comment from a previous patch into a class\ndocstring.\n\nStory: 2001288\nTask: 6114\n\nChange-Id: I3b98c693d334a1025f633bd319c88dde724be30e\nCo-Authored-By: Valeria Perelman \u003cperelman@il.ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0d253adcdf7b3afcda613daaf1d2c7fc8a99b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0d253adcdf7b3afcda613daaf1d2c7fc8a99b08"}]},"branch":"refs/heads/master"},"7cbf1e3932f326c4b812e334ff2c7d3696fdcb0f":{"kind":"REWORK","_number":4,"created":"2018-01-14 20:31:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/91/529191/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/529191/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/529191/4"}}},"commit":{"parents":[{"commit":"158f29a7977c5e3c0cbebcf50b4ed0c4d919aa16","subject":"ACTIVE-ACTIVE: Initial distributor data model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/158f29a7977c5e3c0cbebcf50b4ed0c4d919aa16"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-12-19 22:08:21.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2018-01-14 20:31:09.000000000","tz":-480},"subject":"ACTIVE-ACTIVE: Amphora driver updates","message":"ACTIVE-ACTIVE: Amphora driver updates\n\nThis patch extends the amphora driver to support methods required for\nActive/Active topologies and amphora in the distributor role.\n\nThis patch is a decompostion of the following patch:\nhttps://review.openstack.org/#/c/313006\n\nThis patch also moves a class comment from a previous patch into a class\ndocstring.\n\nStory: 2001288\nTask: 6114\n\nChange-Id: I3b98c693d334a1025f633bd319c88dde724be30e\nCo-Authored-By: Valeria Perelman \u003cperelman@il.ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7cbf1e3932f326c4b812e334ff2c7d3696fdcb0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7cbf1e3932f326c4b812e334ff2c7d3696fdcb0f"}]},"branch":"refs/heads/master"},"7468c9ab8c036b523ecfacb1f4a37625c9258fce":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2018-04-10 14:39:19.000000000","uploader":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"ref":"refs/changes/91/529191/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/529191/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/529191/5"}}},"commit":{"parents":[{"commit":"414eb3adf9f980e1e277708d726ee1e686aef417","subject":"ACTIVE-ACTIVE: Initial distributor data model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/414eb3adf9f980e1e277708d726ee1e686aef417"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-12-19 22:08:21.000000000","tz":-480},"committer":{"name":"Jacky Hu","email":"hudayou@hotmail.com","date":"2018-04-10 14:39:19.000000000","tz":0},"subject":"ACTIVE-ACTIVE: Amphora driver updates","message":"ACTIVE-ACTIVE: Amphora driver updates\n\nThis patch extends the amphora driver to support methods required for\nActive/Active topologies and amphora in the distributor role.\n\nThis patch is a decompostion of the following patch:\nhttps://review.openstack.org/#/c/313006\n\nThis patch also moves a class comment from a previous patch into a class\ndocstring.\n\nStory: 2001288\nTask: 6114\n\nChange-Id: I3b98c693d334a1025f633bd319c88dde724be30e\nCo-Authored-By: Valeria Perelman \u003cperelman@il.ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7468c9ab8c036b523ecfacb1f4a37625c9258fce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7468c9ab8c036b523ecfacb1f4a37625c9258fce"}]},"branch":"refs/heads/master"},"80d6d3a5b80f0bd1a0e7547e7f8934fe7f78c87c":{"kind":"REWORK","_number":6,"created":"2019-06-19 01:25:09.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/91/529191/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/529191/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/529191/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/529191/6"}}},"commit":{"parents":[{"commit":"aec27ebf00c64625512978899e40a9b25d3455b9","subject":"ACTIVE-ACTIVE: Initial distributor data model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aec27ebf00c64625512978899e40a9b25d3455b9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2017-12-19 22:08:21.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-19 01:24:12.000000000","tz":-420},"subject":"ACTIVE-ACTIVE: Amphora driver updates","message":"ACTIVE-ACTIVE: Amphora driver updates\n\nThis patch extends the amphora driver to support methods required for\nActive/Active topologies and amphora in the distributor role.\n\nThis patch is a decompostion of the following patch:\nhttps://review.openstack.org/#/c/313006\n\nThis patch also moves a class comment from a previous patch into a class\ndocstring.\n\nStory: 2001288\nTask: 6114\n\nChange-Id: I3b98c693d334a1025f633bd319c88dde724be30e\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Valeria Perelman \u003cperelman@il.ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80d6d3a5b80f0bd1a0e7547e7f8934fe7f78c87c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80d6d3a5b80f0bd1a0e7547e7f8934fe7f78c87c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
