)]}'
{"id":"openstack%2Foctavia~558962","triplet_id":"openstack%2Foctavia~master~I1ebd42810587a75d82d54312ffc7569255aea145","project":"openstack/octavia","branch":"master","topic":"octavia-availability-zones","attention_set":{"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2022-01-27 20:00:07.000000000","reason":"A robot voted negatively on a label"},"29268":{"account":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"last_update":"2024-03-12 06:36:58.000000000","reason":"A robot voted negatively on a label"},"27256":{"account":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"last_update":"2025-12-10 17:13:55.000000000","reason":"\u003cGERRIT_ACCOUNT_36250\u003e replied on the change","reason_account":{"_account_id":36250,"name":"Przemysław Kuczyński","display_name":"przemekk","email":"przemyslaw.kuczynski@ochk.pl","username":"przemekk"}},"22072":{"account":{"_account_id":22072,"name":"Viktor Křivák","email":"viktor.krivak@ultimum.io","username":"viktor-krivak"},"last_update":"2024-03-12 04:08:12.000000000","reason":"\u003cGERRIT_ACCOUNT_29268\u003e replied on the change","reason_account":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"}},"36250":{"account":{"_account_id":36250,"name":"Przemysław Kuczyński","display_name":"przemekk","email":"przemyslaw.kuczynski@ochk.pl","username":"przemekk"},"last_update":"2025-12-10 17:52:11.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-03-28 14:47:50.000000000","reason":"\u003cGERRIT_ACCOUNT_32053\u003e replied on the change","reason_account":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-03-12 03:48:33.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"32053":{"account":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"last_update":"2024-03-13 10:31:28.000000000","reason":"\u003cGERRIT_ACCOUNT_32053\u003e replied on the change","reason_account":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-12-10 18:01:46.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-12-11 08:06:23.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I1ebd42810587a75d82d54312ffc7569255aea145","subject":"Implement amphora multi-az support","status":"NEW","created":"2018-04-04 23:44:59.000000000","updated":"2025-12-11 08:06:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":34,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"cb64b4a13f6abfd9af1cd592192ce9ed34a67829","_number":558962,"virtual_id_number":558962,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-12 18:52:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-1,"date":"2024-07-12 13:03:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"}],"CC":[{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22072,"name":"Viktor Křivák","email":"viktor.krivak@ultimum.io","username":"viktor-krivak"},{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},{"_account_id":36250,"name":"Przemysław Kuczyński","display_name":"przemekk","email":"przemyslaw.kuczynski@ochk.pl","username":"przemekk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-04-08 22:59:32.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2018-05-23 17:35:48.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2018-09-20 07:25:21.000000000","updated_by":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"reviewer":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"state":"REVIEWER"},{"updated":"2018-11-13 12:25:42.000000000","updated_by":{"_account_id":28067,"name":"Patrik Martinsson","email":"martinsson.patrik@gmail.com","username":"patchon"},"reviewer":{"_account_id":28067,"name":"Patrik Martinsson","email":"martinsson.patrik@gmail.com","username":"patchon"},"state":"REVIEWER"},{"updated":"2019-05-27 08:30:00.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2019-11-18 20:44:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 01:09:56.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2021-03-02 17:42:58.000000000","updated_by":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"reviewer":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"state":"REVIEWER"},{"updated":"2021-03-24 08:24:18.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-04-07 16:23:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-07-29 22:19:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"},{"updated":"2022-01-27 08:53:58.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-12-16 08:01:44.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REMOVED"},{"updated":"2022-12-16 16:30:40.000000000","updated_by":{"_account_id":28067,"name":"Patrik Martinsson","email":"martinsson.patrik@gmail.com","username":"patchon"},"reviewer":{"_account_id":28067,"name":"Patrik Martinsson","email":"martinsson.patrik@gmail.com","username":"patchon"},"state":"REMOVED"},{"updated":"2022-12-19 07:30:10.000000000","updated_by":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"reviewer":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"state":"CC"},{"updated":"2023-03-27 17:13:06.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-03-28 13:33:02.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2023-03-28 14:38:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-28 15:40:41.000000000","updated_by":{"_account_id":22072,"name":"Viktor Křivák","email":"viktor.krivak@ultimum.io","username":"viktor-krivak"},"reviewer":{"_account_id":22072,"name":"Viktor Křivák","email":"viktor.krivak@ultimum.io","username":"viktor-krivak"},"state":"CC"},{"updated":"2024-03-12 04:08:12.000000000","updated_by":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"reviewer":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"state":"CC"},{"updated":"2024-07-12 13:04:09.000000000","updated_by":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"reviewer":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"state":"CC"},{"updated":"2024-07-17 08:57:41.000000000","updated_by":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"CC"},{"updated":"2025-12-10 17:13:55.000000000","updated_by":{"_account_id":36250,"name":"Przemysław Kuczyński","display_name":"przemekk","email":"przemyslaw.kuczynski@ochk.pl","username":"przemekk"},"reviewer":{"_account_id":36250,"name":"Przemysław Kuczyński","display_name":"przemekk","email":"przemyslaw.kuczynski@ochk.pl","username":"przemekk"},"state":"CC"}],"messages":[{"id":"abdceb4d7b4cb1afc29bedf301db003cad8c2ae6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-04 23:44:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08fc170bdf6d8b650b4b824e7227d6a9f00f71c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-05 02:04:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/1/check/openstack-tox-pep8/3513765/ : SUCCESS in 13m 34s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/1/check/openstack-tox-py27/acbfe0b/ : SUCCESS in 4m 29s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/1/check/openstack-tox-py35/fcf69f1/ : SUCCESS in 11m 56s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/1/check/build-openstack-sphinx-docs/a05db7f/html/ : SUCCESS in 5m 30s\n- openstack-tox-cover http://logs.openstack.org/62/558962/1/check/openstack-tox-cover/134ea1d/cover/ : SUCCESS in 7m 43s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/1/check/neutron-lbaasv2-dsvm-api/813bf75/ : FAILURE in 2h 04m 58s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/1/check/neutron-lbaasv2-dsvm-py3x-api/462d5f1/ : SUCCESS in 1h 14m 00s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/1/check/neutron-lbaasv2-dsvm-scenario/584f113/ : SUCCESS in 1h 09m 29s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/1/check/neutron-lbaasv2-dsvm-py3x-scenario/12e7e4a/ : SUCCESS in 1h 05m 10s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/1/check/octavia-v1-dsvm-scenario/577844c/ : SUCCESS in 1h 01m 46s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/1/check/octavia-v1-dsvm-py3x-scenario/9f8cbf6/ : SUCCESS in 59m 04s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/1/check/octavia-v1-dsvm-scenario-multinode/0c56e90/ : FAILURE in 1h 06m 25s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/1/check/octavia-v1-dsvm-py3x-scenario-multinode/55ac9ee/ : SUCCESS in 1h 07m 22s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/f7d1926/ : SUCCESS in 1h 05m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/1/check/openstack-tox-lower-constraints/362a8f3/ : SUCCESS in 7m 22s\n- openstack-tox-functional http://logs.openstack.org/62/558962/1/check/openstack-tox-functional/2fba1d4/ : SUCCESS in 5m 55s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/1/check/openstack-tox-functional-py35/488fd12/ : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"0aa29188d1ef368f4ba4a3a550a598a97afd87ca","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-05 23:54:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e0bf5a8c7d194152d56e4ca27d4abfcf53fa5d0f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-06 00:37:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"836409f743d2355f787d7c823afc268782d250c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-06 02:27:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/2/check/openstack-tox-pep8/1ee425a/ : SUCCESS in 6m 10s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/2/check/openstack-tox-py27/f255322/ : SUCCESS in 4m 29s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/2/check/openstack-tox-py35/24bdf86/ : SUCCESS in 4m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/2/check/build-openstack-sphinx-docs/7733b44/html/ : SUCCESS in 5m 43s\n- openstack-tox-cover http://logs.openstack.org/62/558962/2/check/openstack-tox-cover/fd6946a/cover/ : SUCCESS in 6m 39s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/2/check/neutron-lbaasv2-dsvm-api/d67d921/ : SUCCESS in 1h 13m 49s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/2/check/neutron-lbaasv2-dsvm-py3x-api/d6d1d77/ : SUCCESS in 1h 13m 45s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/2/check/neutron-lbaasv2-dsvm-scenario/864e2fd/ : SUCCESS in 1h 34m 31s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/2/check/neutron-lbaasv2-dsvm-py3x-scenario/c0b7112/ : SUCCESS in 1h 43m 58s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/2/check/octavia-v1-dsvm-scenario/6a05dd0/ : SUCCESS in 50m 38s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/2/check/octavia-v1-dsvm-py3x-scenario/6fe0ffa/ : SUCCESS in 59m 22s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/2/check/octavia-v1-dsvm-scenario-multinode/644008b/ : FAILURE in 1h 05m 27s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/2/check/octavia-v1-dsvm-py3x-scenario-multinode/5900e5b/ : FAILURE in 1h 06m 15s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/2/check/octavia-v1-dsvm-scenario-kvm-centos.7/8ae5483/ : SUCCESS in 52m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/2/check/openstack-tox-lower-constraints/7926a3f/ : SUCCESS in 6m 29s\n- openstack-tox-functional http://logs.openstack.org/62/558962/2/check/openstack-tox-functional/8023518/ : SUCCESS in 6m 48s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/2/check/openstack-tox-functional-py35/d88c506/ : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":2},{"id":"ee7e4c1b461f58d3ecf49a16d01681d187e01853","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-06 20:57:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3b7eea8d2887149bc9103d80b9b1c66c0f2493c7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-06 22:14:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4af19003b34e217c72ab3056c924d9660a788799","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-06 23:49:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/4/check/openstack-tox-pep8/d024fba/ : SUCCESS in 6m 15s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/4/check/openstack-tox-py27/7fd7cc5/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/4/check/openstack-tox-py35/e898994/ : SUCCESS in 5m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/4/check/build-openstack-sphinx-docs/3775888/html/ : SUCCESS in 5m 37s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/4/check/build-openstack-releasenotes/cabb867/html/ : SUCCESS in 4m 21s\n- openstack-tox-cover http://logs.openstack.org/62/558962/4/check/openstack-tox-cover/e7c383c/cover/ : SUCCESS in 6m 59s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/4/check/neutron-lbaasv2-dsvm-api/00f5875/ : SUCCESS in 1h 08m 22s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/4/check/neutron-lbaasv2-dsvm-py3x-api/0fac793/ : SUCCESS in 1h 18m 37s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/4/check/neutron-lbaasv2-dsvm-scenario/fb2159c/ : SUCCESS in 1h 08m 14s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/4/check/neutron-lbaasv2-dsvm-py3x-scenario/972b170/ : SUCCESS in 1h 29m 13s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/4/check/octavia-v1-dsvm-scenario/959b8f9/ : FAILURE in 21m 57s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/4/check/octavia-v1-dsvm-py3x-scenario/54348b1/ : FAILURE in 30m 37s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/4/check/octavia-v1-dsvm-scenario-multinode/b1e8be9/ : FAILURE in 24m 02s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/4/check/octavia-v1-dsvm-py3x-scenario-multinode/7011a49/ : FAILURE in 29m 14s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/4/check/octavia-v1-dsvm-scenario-kvm-centos.7/15fbf69/ : SUCCESS in 51m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/4/check/openstack-tox-lower-constraints/678f730/ : SUCCESS in 6m 31s\n- openstack-tox-functional http://logs.openstack.org/62/558962/4/check/openstack-tox-functional/fecafc0/ : SUCCESS in 6m 01s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/4/check/openstack-tox-functional-py35/270a098/ : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":4},{"id":"4b6f811427138bd64da5781cfd2aa9595b1e2c2d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-04-08 22:59:32.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"33bb7edf3bd6004f0ee143ca6340a179aa18793c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-09 00:13:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/4/check/openstack-tox-pep8/4f75d7d/ : SUCCESS in 5m 54s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/4/check/openstack-tox-py27/a5dbb8b/ : SUCCESS in 4m 44s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/4/check/openstack-tox-py35/242f0da/ : SUCCESS in 4m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/4/check/build-openstack-sphinx-docs/6682f09/html/ : SUCCESS in 5m 38s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/4/check/build-openstack-releasenotes/cc40982/html/ : SUCCESS in 4m 10s\n- openstack-tox-cover http://logs.openstack.org/62/558962/4/check/openstack-tox-cover/800911f/cover/ : SUCCESS in 6m 55s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/4/check/neutron-lbaasv2-dsvm-api/c2aa4f9/ : SUCCESS in 1h 07m 54s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/4/check/neutron-lbaasv2-dsvm-py3x-api/571353f/ : SUCCESS in 1h 08m 31s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/4/check/neutron-lbaasv2-dsvm-scenario/2bb2d4a/ : SUCCESS in 1h 09m 16s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/4/check/neutron-lbaasv2-dsvm-py3x-scenario/403a868/ : SUCCESS in 1h 11m 19s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/4/check/octavia-v1-dsvm-scenario/b5aba81/ : SUCCESS in 49m 01s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/4/check/octavia-v1-dsvm-py3x-scenario/98c172e/ : SUCCESS in 48m 38s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/4/check/octavia-v1-dsvm-scenario-multinode/f446663/ : FAILURE in 57m 20s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/4/check/octavia-v1-dsvm-py3x-scenario-multinode/2781f11/ : FAILURE in 57m 20s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/4/check/octavia-v1-dsvm-scenario-kvm-centos.7/2d7010e/ : SUCCESS in 50m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/4/check/openstack-tox-lower-constraints/0bb058e/ : SUCCESS in 6m 26s\n- openstack-tox-functional http://logs.openstack.org/62/558962/4/check/openstack-tox-functional/a06175a/ : SUCCESS in 5m 46s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/4/check/openstack-tox-functional-py35/5dd928b/ : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":4},{"id":"bd6382edfca49e5eee1e448805eb94ceb5ba189e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-09 21:20:08.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"be03d86538f5de728cc3dbd9a4d03a7c52e8d5ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-09 22:49:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/5/check/openstack-tox-pep8/94ec1fa/ : SUCCESS in 6m 15s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/5/check/openstack-tox-py27/e1fbeb4/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/5/check/openstack-tox-py35/583ec8f/ : SUCCESS in 4m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/5/check/build-openstack-sphinx-docs/b9ad1fe/html/ : SUCCESS in 5m 45s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/5/check/build-openstack-releasenotes/1fcdf9d/html/ : SUCCESS in 4m 49s\n- openstack-tox-cover http://logs.openstack.org/62/558962/5/check/openstack-tox-cover/a8a1063/cover/ : SUCCESS in 6m 30s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/5/check/neutron-lbaasv2-dsvm-api/09f6f3d/ : SUCCESS in 1h 13m 04s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/5/check/neutron-lbaasv2-dsvm-scenario/f535e7f/ : SUCCESS in 1h 07m 31s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/5/check/octavia-v1-dsvm-scenario/02eab78/ : SUCCESS in 54m 59s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/5/check/octavia-v1-dsvm-scenario-multinode/f783e0d/ : SUCCESS in 1h 05m 28s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/5/check/octavia-v1-dsvm-scenario-kvm-centos.7/f32c95c/ : SUCCESS in 1h 00m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/5/check/neutron-lbaasv2-dsvm-py3x-api/61575c8/ : SUCCESS in 1h 13m 58s\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/5/check/neutron-lbaasv2-dsvm-py3x-scenario/3392bdf/ : SUCCESS in 1h 23m 13s (non-voting)\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/5/check/octavia-v1-dsvm-py3x-scenario/ac20332/ : SUCCESS in 56m 29s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/5/check/octavia-v1-dsvm-py3x-scenario-multinode/b1e8841/ : SUCCESS in 1h 02m 45s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/5/check/openstack-tox-lower-constraints/e3f00ec/ : SUCCESS in 7m 37s\n- openstack-tox-functional http://logs.openstack.org/62/558962/5/check/openstack-tox-functional/ccf5d68/ : SUCCESS in 7m 23s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/5/check/openstack-tox-functional-py35/be99479/ : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":5},{"id":"05e82bd465138c1486d95080a67af427a5574fcc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-24 01:14:48.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"7810096fd37a91acecc625900c0b95d118e4b19b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-24 02:50:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/6/check/openstack-tox-pep8/a59abca/ : SUCCESS in 6m 40s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/6/check/openstack-tox-py27/67710e5/ : SUCCESS in 5m 01s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/6/check/openstack-tox-py35/2aa0588/ : SUCCESS in 5m 56s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/6/check/build-openstack-sphinx-docs/9f2a151/html/ : SUCCESS in 5m 52s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/6/check/build-openstack-releasenotes/10ff210/html/ : SUCCESS in 4m 21s\n- openstack-tox-cover http://logs.openstack.org/62/558962/6/check/openstack-tox-cover/c1e6b26/cover/ : SUCCESS in 8m 01s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/6/check/neutron-lbaasv2-dsvm-api/1a8fb93/ : SUCCESS in 1h 20m 42s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/6/check/neutron-lbaasv2-dsvm-py3x-api/780370e/ : SUCCESS in 1h 08m 52s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/6/check/neutron-lbaasv2-dsvm-scenario/da906f5/ : SUCCESS in 1h 06m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/6/check/neutron-lbaasv2-dsvm-py3x-scenario/ebe2331/ : SUCCESS in 1h 22m 30s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/6/check/openstack-tox-lower-constraints/d7b05bb/ : SUCCESS in 6m 32s\n- openstack-tox-functional http://logs.openstack.org/62/558962/6/check/openstack-tox-functional/2fcdb6a/ : SUCCESS in 5m 34s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/6/check/openstack-tox-functional-py35/d117440/ : SUCCESS in 6m 03s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/6/check/octavia-v1-dsvm-scenario/62bbaa4/ : SUCCESS in 52m 00s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/6/check/octavia-v1-dsvm-py3x-scenario/860492d/ : SUCCESS in 59m 11s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/6/check/octavia-v1-dsvm-scenario-multinode/a6db9d0/ : SUCCESS in 1h 05m 05s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/6/check/octavia-v1-dsvm-py3x-scenario-multinode/a805e0b/ : SUCCESS in 1h 28m 57s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/6/check/octavia-v1-dsvm-scenario-kvm-centos.7/8b50942/ : SUCCESS in 57m 53s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/6/check/octavia-v1-dsvm-scenario-ubuntu.bionic/fc8ef54/ : SUCCESS in 51m 28s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/6/check/octavia-v2-dsvm-noop-api/b098084/ : SUCCESS in 30m 42s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/6/check/octavia-v2-dsvm-noop-py35-api/63731c9/ : SUCCESS in 33m 44s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/6/check/octavia-v2-dsvm-scenario/f461233/ : SUCCESS in 41m 33s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/6/check/octavia-v2-dsvm-py35-scenario/737b2c4/ : SUCCESS in 44m 46s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3d6b057ad4f26867b21e8cff3d63e6dc01a69471","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-25 01:51:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5c320b7c23a05c77bca8847a79b3b4317cd437f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-25 03:24:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/7/check/openstack-tox-pep8/6041ab8/ : SUCCESS in 7m 03s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/7/check/openstack-tox-py27/5dc7964/ : SUCCESS in 5m 45s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/7/check/openstack-tox-py35/a2941b6/ : SUCCESS in 5m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/7/check/build-openstack-sphinx-docs/629c9cd/html/ : SUCCESS in 6m 28s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/7/check/build-openstack-releasenotes/59ff3f0/html/ : SUCCESS in 4m 59s\n- openstack-tox-cover http://logs.openstack.org/62/558962/7/check/openstack-tox-cover/710acc4/cover/ : SUCCESS in 9m 48s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/7/check/neutron-lbaasv2-dsvm-api/fea130d/ : SUCCESS in 1h 09m 01s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/7/check/neutron-lbaasv2-dsvm-py3x-api/a47871b/ : SUCCESS in 1h 14m 05s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/7/check/neutron-lbaasv2-dsvm-scenario/d9c5b6d/ : SUCCESS in 1h 08m 43s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/7/check/neutron-lbaasv2-dsvm-py3x-scenario/91ff6be/ : SUCCESS in 52m 51s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/7/check/octavia-v1-dsvm-scenario/f92cf45/ : SUCCESS in 1h 05m 13s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/7/check/octavia-v1-dsvm-py3x-scenario/5e018ae/ : SUCCESS in 1h 24m 24s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/7/check/octavia-v1-dsvm-scenario-multinode/81dcfc4/ : SUCCESS in 1h 06m 12s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/7/check/octavia-v1-dsvm-py3x-scenario-multinode/b833198/ : FAILURE in 1h 10m 12s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/7/check/octavia-v1-dsvm-scenario-kvm-centos.7/e66b211/ : SUCCESS in 1h 01m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/7/check/openstack-tox-lower-constraints/e969da3/ : SUCCESS in 6m 45s\n- openstack-tox-functional http://logs.openstack.org/62/558962/7/check/openstack-tox-functional/b314927/ : SUCCESS in 7m 43s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/7/check/openstack-tox-functional-py35/e9908e9/ : SUCCESS in 6m 34s\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/7/check/octavia-v1-dsvm-scenario-ubuntu.bionic/932ddbb/ : SUCCESS in 48m 56s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/7/check/octavia-v2-dsvm-noop-api/90630b4/ : SUCCESS in 27m 09s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/7/check/octavia-v2-dsvm-noop-py35-api/4aea76f/ : SUCCESS in 37m 23s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/7/check/octavia-v2-dsvm-scenario/37ff975/ : SUCCESS in 43m 20s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/7/check/octavia-v2-dsvm-py35-scenario/e3521de/ : SUCCESS in 43m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"39681a7d352bb9f994b129cfdd66346b0b5a7aa2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-26 04:54:29.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"a2dd0559749ffa2d215df16b27061bab51621f91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-26 06:20:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/8/check/openstack-tox-pep8/3963fb8/ : SUCCESS in 5m 47s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/8/check/openstack-tox-py27/98cc683/ : SUCCESS in 4m 44s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/8/check/openstack-tox-py35/061b7e4/ : SUCCESS in 4m 56s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/8/check/build-openstack-sphinx-docs/3af45db/html/ : SUCCESS in 5m 37s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/8/check/build-openstack-releasenotes/1fd02b5/html/ : SUCCESS in 4m 13s\n- openstack-tox-cover http://logs.openstack.org/62/558962/8/check/openstack-tox-cover/0615078/cover/ : SUCCESS in 6m 50s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/8/check/neutron-lbaasv2-dsvm-api/d314318/ : SUCCESS in 1h 08m 08s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/8/check/neutron-lbaasv2-dsvm-py3x-api/e945a53/ : SUCCESS in 1h 14m 08s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/8/check/neutron-lbaasv2-dsvm-scenario/2a387ae/ : SUCCESS in 52m 50s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/8/check/neutron-lbaasv2-dsvm-py3x-scenario/7397608/ : SUCCESS in 1h 19m 10s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/8/check/octavia-v1-dsvm-scenario/361fa32/ : SUCCESS in 48m 20s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/8/check/octavia-v1-dsvm-py3x-scenario/35d823a/ : SUCCESS in 55m 57s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/8/check/octavia-v1-dsvm-scenario-multinode/9ea0cd8/ : SUCCESS in 1h 06m 33s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/8/check/octavia-v1-dsvm-py3x-scenario-multinode/fbddb54/ : FAILURE in 1h 07m 42s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/8/check/octavia-v1-dsvm-scenario-kvm-centos.7/19abf9d/ : SUCCESS in 1h 01m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/8/check/openstack-tox-lower-constraints/8e82360/ : SUCCESS in 6m 17s\n- openstack-tox-functional http://logs.openstack.org/62/558962/8/check/openstack-tox-functional/d4e34c7/ : SUCCESS in 5m 40s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/8/check/openstack-tox-functional-py35/bf615fd/ : SUCCESS in 6m 10s\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/8/check/octavia-v1-dsvm-scenario-ubuntu.bionic/e5ea4e7/ : SUCCESS in 42m 35s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/8/check/octavia-v2-dsvm-noop-api/4b5bb74/ : FAILURE in 25m 14s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/8/check/octavia-v2-dsvm-noop-py35-api/c4e6cca/ : FAILURE in 30m 35s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/8/check/octavia-v2-dsvm-scenario/9b1c9c3/ : FAILURE in 47m 11s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/8/check/octavia-v2-dsvm-py35-scenario/a4562ce/ : FAILURE in 43m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b835285804cdaa05d1b277223ec0df1c9df27cb4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-26 20:37:25.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"53854395b637ddead0a319432e76c6c0ce2fbf68","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-26 22:40:05.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"fc901c36b4002922ef6027a8d7c53bf8ff755242","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 00:33:59.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/10/check/openstack-tox-pep8/5e2c30a/ : SUCCESS in 6m 04s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/10/check/openstack-tox-py27/e10f693/ : FAILURE in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/10/check/openstack-tox-py35/0a72f42/ : FAILURE in 5m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/10/check/build-openstack-sphinx-docs/710c4ef/html/ : SUCCESS in 7m 10s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/10/check/build-openstack-releasenotes/c117d4c/html/ : SUCCESS in 4m 02s\n- openstack-tox-cover http://logs.openstack.org/62/558962/10/check/openstack-tox-cover/06b8361/ : FAILURE in 4m 59s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/10/check/neutron-lbaasv2-dsvm-api/e1b80fd/ : SUCCESS in 1h 10m 36s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/10/check/neutron-lbaasv2-dsvm-py3x-api/60575b2/ : SUCCESS in 1h 19m 53s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/10/check/neutron-lbaasv2-dsvm-scenario/77b9b0a/ : SUCCESS in 1h 47m 32s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/10/check/neutron-lbaasv2-dsvm-py3x-scenario/169bec3/ : SUCCESS in 1h 27m 01s (non-voting)\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/10/check/octavia-v1-dsvm-scenario/91492c3/ : SUCCESS in 52m 37s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/10/check/octavia-v1-dsvm-py3x-scenario/49de307/ : SUCCESS in 49m 16s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/10/check/octavia-v1-dsvm-scenario-multinode/3c7d8cf/ : SUCCESS in 1h 23m 28s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/10/check/octavia-v1-dsvm-py3x-scenario-multinode/f653865/ : SUCCESS in 1h 02m 31s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/10/check/octavia-v1-dsvm-scenario-kvm-centos.7/5a708a9/ : SUCCESS in 1h 07m 54s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/10/check/openstack-tox-lower-constraints/2a2a2a2/ : FAILURE in 5m 47s\n- openstack-tox-functional http://logs.openstack.org/62/558962/10/check/openstack-tox-functional/d821081/ : SUCCESS in 6m 17s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/10/check/openstack-tox-functional-py35/beff25f/ : SUCCESS in 7m 57s\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/10/check/octavia-v1-dsvm-scenario-ubuntu.bionic/87236c4/ : SUCCESS in 49m 55s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/10/check/octavia-v2-dsvm-noop-api/71e252e/ : FAILURE in 29m 17s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/10/check/octavia-v2-dsvm-noop-py35-api/4447ad4/ : FAILURE in 28m 08s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/10/check/octavia-v2-dsvm-scenario/66aa791/ : FAILURE in 1h 00m 36s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/10/check/octavia-v2-dsvm-py35-scenario/b89b211/ : FAILURE in 49m 58s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"33670915ffec1ce02c61a05064b158e7fdd52b63","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-04-27 20:29:31.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"925086c071eab0af1ff23d154a064b151016a2a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 22:19:34.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/11/check/openstack-tox-pep8/fa77243/ : SUCCESS in 6m 53s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/11/check/openstack-tox-py27/923cc8a/ : SUCCESS in 5m 34s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/11/check/openstack-tox-py35/a0a9158/ : SUCCESS in 5m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/11/check/build-openstack-sphinx-docs/588943e/html/ : SUCCESS in 5m 47s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/11/check/build-openstack-releasenotes/d497c56/html/ : SUCCESS in 4m 32s\n- openstack-tox-cover http://logs.openstack.org/62/558962/11/check/openstack-tox-cover/91c5f15/cover/ : SUCCESS in 8m 14s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/11/check/neutron-lbaasv2-dsvm-api/fceb91a/ : SUCCESS in 1h 14m 16s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/11/check/neutron-lbaasv2-dsvm-py3x-api/32274ef/ : SUCCESS in 1h 21m 07s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/11/check/neutron-lbaasv2-dsvm-scenario/322d17e/ : SUCCESS in 1h 12m 46s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/11/check/neutron-lbaasv2-dsvm-py3x-scenario/870da3b/ : SUCCESS in 1h 28m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/11/check/openstack-tox-lower-constraints/fbfdac6/ : SUCCESS in 6m 53s\n- openstack-tox-functional http://logs.openstack.org/62/558962/11/check/openstack-tox-functional/0a1bc19/ : SUCCESS in 7m 06s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/11/check/openstack-tox-functional-py35/c91b748/ : SUCCESS in 7m 36s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/11/check/octavia-v1-dsvm-scenario/cd96c42/ : SUCCESS in 57m 21s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/11/check/octavia-v1-dsvm-py3x-scenario/7f183a2/ : SUCCESS in 50m 30s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/11/check/octavia-v1-dsvm-scenario-multinode/78249e5/ : FAILURE in 1h 08m 57s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/11/check/octavia-v1-dsvm-py3x-scenario-multinode/debfec2/ : SUCCESS in 1h 04m 39s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/11/check/octavia-v1-dsvm-scenario-kvm-centos.7/3a81252/ : SUCCESS in 1h 01m 06s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/11/check/octavia-v1-dsvm-scenario-ubuntu.bionic/6cc89ad/ : SUCCESS in 58m 25s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/11/check/octavia-v2-dsvm-noop-api/70a8550/ : FAILURE in 31m 46s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/11/check/octavia-v2-dsvm-noop-py35-api/aec73f1/ : FAILURE in 34m 30s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/11/check/octavia-v2-dsvm-scenario/f6c6652/ : FAILURE in 51m 35s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/11/check/octavia-v2-dsvm-py35-scenario/65b4113/ : FAILURE in 46m 56s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"de4942f569e8b6fb098f96c9ae0477a49d3c5544","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-05-07 18:58:40.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"0a1a79bc5a23877dc7ef5c9dcc1bc4c3855d175d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-07 20:30:54.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/12/check/openstack-tox-pep8/457121c/ : SUCCESS in 5m 54s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/12/check/openstack-tox-py27/867d2e5/ : SUCCESS in 5m 06s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/12/check/openstack-tox-py35/b385513/ : SUCCESS in 4m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/12/check/build-openstack-sphinx-docs/9f41a27/html/ : SUCCESS in 6m 50s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/12/check/build-openstack-releasenotes/0a8dfb9/html/ : SUCCESS in 4m 23s\n- openstack-tox-cover http://logs.openstack.org/62/558962/12/check/openstack-tox-cover/6299b11/cover/ : SUCCESS in 8m 07s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/12/check/neutron-lbaasv2-dsvm-api/7adeb10/ : SUCCESS in 1h 10m 36s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/12/check/neutron-lbaasv2-dsvm-py3x-api/eec7904/ : SUCCESS in 1h 09m 19s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/12/check/neutron-lbaasv2-dsvm-scenario/630a54b/ : SUCCESS in 1h 14m 31s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/12/check/neutron-lbaasv2-dsvm-py3x-scenario/e8fa667/ : SUCCESS in 1h 11m 00s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/12/check/openstack-tox-lower-constraints/75ad142/ : SUCCESS in 6m 25s\n- openstack-tox-functional http://logs.openstack.org/62/558962/12/check/openstack-tox-functional/b7d1c28/ : SUCCESS in 6m 19s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/12/check/openstack-tox-functional-py35/2d0f769/ : SUCCESS in 6m 25s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/12/check/octavia-v1-dsvm-scenario/80049ea/ : SUCCESS in 50m 48s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/12/check/octavia-v1-dsvm-py3x-scenario/656a1a4/ : SUCCESS in 59m 41s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/12/check/octavia-v1-dsvm-scenario-multinode/6cf5ea7/ : FAILURE in 59m 20s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/12/check/octavia-v1-dsvm-py3x-scenario-multinode/a8fdf85/ : SUCCESS in 1h 15m 15s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/12/check/octavia-v1-dsvm-scenario-kvm-centos.7/a9910ce/ : SUCCESS in 55m 36s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/12/check/octavia-v1-dsvm-scenario-ubuntu.bionic/ee8ffdc/ : SUCCESS in 59m 21s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/12/check/octavia-v2-dsvm-noop-api/2b37737/ : SUCCESS in 31m 00s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/12/check/octavia-v2-dsvm-noop-py35-api/1c4d6f6/ : SUCCESS in 32m 38s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/12/check/octavia-v2-dsvm-scenario/cca2618/ : SUCCESS in 50m 01s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/12/check/octavia-v2-dsvm-py35-scenario/8b3886e/ : SUCCESS in 44m 27s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"38a1deab68bdfddb9f2430fdc795b21ba2b54d46","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2018-05-23 17:35:48.000000000","message":"Patch Set 12:\n\n(1 comment)\n\nI also don\u0027t think we can address AZ in isolation with only nova. We also have neutron AZs and, admittedly haven\u0027t put much thought into them, but they might go fhand in hand with nova (why would somebody segregate nova into AZs but leave neutron as a single point of failure?)\n\nBecause it\u0027s experimental I am not giving a -1","accounts_in_message":[],"_revision_number":12},{"id":"8a6864459a70974e8feb8bce9e62b7fd5c385f6a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-05-31 23:52:44.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"c055615a93fc852e42e149d06e54dc4ce1bf7581","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-01 01:30:16.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/13/check/openstack-tox-pep8/ecf7eaa/ : SUCCESS in 6m 41s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/13/check/openstack-tox-py27/d11bc6b/ : SUCCESS in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/13/check/openstack-tox-py35/b3d3bf9/ : SUCCESS in 4m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/13/check/build-openstack-sphinx-docs/74170e1/html/ : SUCCESS in 8m 04s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/13/check/build-openstack-releasenotes/dc9d0b1/html/ : SUCCESS in 5m 24s\n- openstack-tox-cover http://logs.openstack.org/62/558962/13/check/openstack-tox-cover/e48b7c0/cover/ : SUCCESS in 7m 18s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/13/check/neutron-lbaasv2-dsvm-api/5a300e7/ : SUCCESS in 1h 08m 04s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/13/check/neutron-lbaasv2-dsvm-py3x-api/b103a35/ : SUCCESS in 1h 12m 11s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/13/check/neutron-lbaasv2-dsvm-scenario/721768d/ : SUCCESS in 1h 19m 11s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/13/check/neutron-lbaasv2-dsvm-py3x-scenario/820908d/ : SUCCESS in 1h 28m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/13/check/openstack-tox-lower-constraints/b9156e9/ : SUCCESS in 7m 00s\n- openstack-tox-functional http://logs.openstack.org/62/558962/13/check/openstack-tox-functional/e704d34/ : SUCCESS in 7m 23s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/13/check/openstack-tox-functional-py35/9c6b2d0/ : SUCCESS in 6m 39s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/13/check/octavia-v1-dsvm-scenario/fbb1298/ : SUCCESS in 44m 12s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/13/check/octavia-v1-dsvm-py3x-scenario/a445964/ : SUCCESS in 57m 29s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/13/check/octavia-v1-dsvm-scenario-multinode/d629303/ : FAILURE in 1h 13m 03s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/13/check/octavia-v1-dsvm-py3x-scenario-multinode/c700e2a/ : FAILURE in 1h 31m 07s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/13/check/octavia-v1-dsvm-scenario-kvm-centos.7/8b2423f/ : SUCCESS in 57m 14s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/13/check/octavia-v1-dsvm-scenario-ubuntu.bionic/0453efe/ : SUCCESS in 50m 26s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/13/check/octavia-v2-dsvm-noop-api/3f97558/ : SUCCESS in 32m 51s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/13/check/octavia-v2-dsvm-noop-py35-api/5ad05da/ : SUCCESS in 37m 47s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/13/check/octavia-v2-dsvm-scenario/d2e38bc/ : FAILURE in 1h 03m 16s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/13/check/octavia-v2-dsvm-py35-scenario/f5de661/ : SUCCESS in 53m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"49f06f07cc3dac97811a66c8bd269a451b4491f3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-02 01:53:55.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"ca9075218f50a4170bf47d3a865fd4d1ba8d6a2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-02 03:18:01.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/14/check/openstack-tox-pep8/a9f6332/ : SUCCESS in 6m 44s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/14/check/openstack-tox-py27/636eef3/ : SUCCESS in 5m 50s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/14/check/openstack-tox-py35/6898141/ : SUCCESS in 5m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/14/check/build-openstack-sphinx-docs/05f572f/html/ : SUCCESS in 5m 29s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/14/check/build-openstack-releasenotes/df6cbd6/html/ : SUCCESS in 3m 46s\n- openstack-tox-cover http://logs.openstack.org/62/558962/14/check/openstack-tox-cover/37149d5/cover/ : SUCCESS in 6m 59s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/14/check/neutron-lbaasv2-dsvm-api/3ebd477/ : SUCCESS in 1h 13m 33s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/14/check/neutron-lbaasv2-dsvm-py3x-api/fe18c4c/ : SUCCESS in 1h 10m 08s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/14/check/neutron-lbaasv2-dsvm-scenario/471d4db/ : SUCCESS in 1h 16m 32s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/14/check/neutron-lbaasv2-dsvm-py3x-scenario/d3af1e1/ : SUCCESS in 1h 13m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/14/check/openstack-tox-lower-constraints/6637979/ : SUCCESS in 7m 19s\n- openstack-tox-functional http://logs.openstack.org/62/558962/14/check/openstack-tox-functional/49743b2/ : SUCCESS in 6m 09s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/14/check/openstack-tox-functional-py35/887df92/ : SUCCESS in 6m 30s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/14/check/octavia-v1-dsvm-scenario/1e6e568/ : SUCCESS in 1h 05m 16s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/14/check/octavia-v1-dsvm-py3x-scenario/ef13e1b/ : SUCCESS in 54m 59s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/14/check/octavia-v1-dsvm-scenario-multinode/c13d3e7/ : FAILURE in 1h 18m 10s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/14/check/octavia-v1-dsvm-py3x-scenario-multinode/4569745/ : FAILURE in 1h 16m 26s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/14/check/octavia-v1-dsvm-scenario-kvm-centos.7/3bdfaaa/ : SUCCESS in 55m 29s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/14/check/octavia-v1-dsvm-scenario-ubuntu.bionic/a161962/ : SUCCESS in 51m 08s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/14/check/octavia-v2-dsvm-noop-api/c709e86/ : SUCCESS in 31m 53s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/14/check/octavia-v2-dsvm-noop-py35-api/49a9a37/ : SUCCESS in 34m 55s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/14/check/octavia-v2-dsvm-scenario/45bcb83/ : SUCCESS in 59m 00s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/14/check/octavia-v2-dsvm-py35-scenario/1833c1d/ : SUCCESS in 54m 55s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"8616ab7dca3f38f35adcdd60a84a760f0f79f17c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-05 20:34:39.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"f708f9de0fcc216c2e71607e6ffe7966cee105a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-05 23:00:03.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/15/check/openstack-tox-pep8/0374635/ : SUCCESS in 7m 09s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/15/check/openstack-tox-py27/cb44d87/ : SUCCESS in 4m 59s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/15/check/openstack-tox-py35/d9c9c07/ : SUCCESS in 5m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/15/check/build-openstack-sphinx-docs/fdffc19/html/ : SUCCESS in 7m 16s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/15/check/build-openstack-releasenotes/7a3b0bd/html/ : SUCCESS in 4m 12s\n- openstack-tox-cover http://logs.openstack.org/62/558962/15/check/openstack-tox-cover/08fb07a/cover/ : SUCCESS in 7m 27s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/15/check/neutron-lbaasv2-dsvm-api/20974e4/ : SUCCESS in 1h 16m 09s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/15/check/neutron-lbaasv2-dsvm-py3x-api/653c01e/ : SUCCESS in 1h 16m 52s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/15/check/neutron-lbaasv2-dsvm-scenario/30a0e92/ : SUCCESS in 1h 22m 55s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/15/check/neutron-lbaasv2-dsvm-py3x-scenario/1c409ed/ : SUCCESS in 1h 14m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/15/check/openstack-tox-lower-constraints/24c6007/ : SUCCESS in 8m 52s\n- openstack-tox-functional http://logs.openstack.org/62/558962/15/check/openstack-tox-functional/cf81127/ : SUCCESS in 6m 30s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/15/check/openstack-tox-functional-py35/7bf69f5/ : SUCCESS in 6m 57s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/15/check/octavia-v1-dsvm-scenario/2a59e43/ : SUCCESS in 52m 51s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/15/check/octavia-v1-dsvm-py3x-scenario/3db6f6d/ : SUCCESS in 53m 57s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/15/check/octavia-v1-dsvm-scenario-multinode/ed2eb65/ : FAILURE in 1h 20m 57s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/15/check/octavia-v1-dsvm-py3x-scenario-multinode/53db02c/ : FAILURE in 1h 17m 02s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/15/check/octavia-v1-dsvm-scenario-kvm-centos.7/4fba90f/ : SUCCESS in 1h 07m 11s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/15/check/octavia-v1-dsvm-scenario-ubuntu.bionic/cb9f574/ : SUCCESS in 1h 30m 33s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/15/check/octavia-v2-dsvm-noop-api/08f945d/ : SUCCESS in 36m 56s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/15/check/octavia-v2-dsvm-noop-py35-api/f3d9f59/ : SUCCESS in 42m 27s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/15/check/octavia-v2-dsvm-scenario/6235baf/ : SUCCESS in 1h 19m 41s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/15/check/octavia-v2-dsvm-py35-scenario/1864066/ : SUCCESS in 1h 09m 04s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"9c7f49e936c60f5a5f83ce643fd2f7cddaff0104","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-07 01:38:31.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"ec518635413bfa9ca2865a9d0044d7e8faaebb05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-07 03:09:15.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/16/check/openstack-tox-pep8/a7191fa/ : SUCCESS in 6m 17s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/16/check/openstack-tox-py27/31984bc/ : SUCCESS in 6m 42s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/16/check/openstack-tox-py35/1383860/ : SUCCESS in 5m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/16/check/build-openstack-sphinx-docs/782fc5f/html/ : SUCCESS in 6m 25s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/16/check/build-openstack-releasenotes/5ad0178/html/ : SUCCESS in 4m 15s\n- openstack-tox-cover http://logs.openstack.org/62/558962/16/check/openstack-tox-cover/c90cd7b/cover/ : SUCCESS in 7m 55s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/16/check/neutron-lbaasv2-dsvm-api/886b6d9/ : SUCCESS in 1h 18m 52s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/16/check/neutron-lbaasv2-dsvm-py3x-api/0273448/ : SUCCESS in 1h 11m 18s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/16/check/neutron-lbaasv2-dsvm-scenario/4e808fd/ : SUCCESS in 1h 12m 05s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/16/check/neutron-lbaasv2-dsvm-py3x-scenario/a8fdd14/ : SUCCESS in 1h 22m 54s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/16/check/openstack-tox-lower-constraints/67ac8b2/ : SUCCESS in 7m 44s\n- openstack-tox-functional http://logs.openstack.org/62/558962/16/check/openstack-tox-functional/6c4b3e9/ : SUCCESS in 6m 37s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/16/check/openstack-tox-functional-py35/7fa7e4c/ : SUCCESS in 7m 26s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/16/check/octavia-v1-dsvm-scenario/2396cca/ : SUCCESS in 51m 11s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/16/check/octavia-v1-dsvm-py3x-scenario/b9d97db/ : SUCCESS in 55m 21s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/16/check/octavia-v1-dsvm-scenario-multinode/b0128b2/ : FAILURE in 1h 24m 22s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/16/check/octavia-v1-dsvm-py3x-scenario-multinode/085b61d/ : FAILURE in 1h 03m 40s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/16/check/octavia-v1-dsvm-scenario-kvm-centos.7/dac59d9/ : SUCCESS in 57m 28s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/16/check/octavia-v1-dsvm-scenario-ubuntu.bionic/ae96e75/ : SUCCESS in 52m 51s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/16/check/octavia-v2-dsvm-noop-api/abdca61/ : SUCCESS in 39m 21s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/16/check/octavia-v2-dsvm-noop-py35-api/96d9eed/ : SUCCESS in 41m 35s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/16/check/octavia-v2-dsvm-scenario/70acf71/ : SUCCESS in 1h 13m 32s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/16/check/octavia-v2-dsvm-py35-scenario/4fc27d5/ : SUCCESS in 1h 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"869284fa8799ff843e2167499e1d4ec781377f4c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-14 18:00:20.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"0d2d1f9171e37e096938f7b52b014a3a14184471","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-14 22:04:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/17/check/openstack-tox-pep8/8678b83/ : SUCCESS in 7m 12s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/17/check/openstack-tox-py27/3ac08ce/ : SUCCESS in 6m 27s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/17/check/openstack-tox-py35/ac25909/ : SUCCESS in 5m 51s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/17/check/build-openstack-sphinx-docs/fc80827/html/ : SUCCESS in 8m 06s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/17/check/build-openstack-releasenotes/502556b/html/ : SUCCESS in 5m 25s\n- openstack-tox-cover http://logs.openstack.org/62/558962/17/check/openstack-tox-cover/78b3411/cover/ : SUCCESS in 9m 04s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/17/check/neutron-lbaasv2-dsvm-api/8e3a686/ : SUCCESS in 1h 12m 38s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/17/check/neutron-lbaasv2-dsvm-py3x-api/7a57452/ : SUCCESS in 1h 13m 02s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/17/check/neutron-lbaasv2-dsvm-scenario/197a060/ : SUCCESS in 1h 26m 12s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/17/check/neutron-lbaasv2-dsvm-py3x-scenario/929dd9e/ : SUCCESS in 1h 19m 16s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/17/check/openstack-tox-lower-constraints/d6a4b4f/ : SUCCESS in 8m 41s\n- openstack-tox-functional http://logs.openstack.org/62/558962/17/check/openstack-tox-functional/dff43e0/ : SUCCESS in 7m 57s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/17/check/openstack-tox-functional-py35/aae1654/ : SUCCESS in 7m 25s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/17/check/octavia-v1-dsvm-scenario/401d6f0/ : SUCCESS in 53m 26s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/17/check/octavia-v1-dsvm-py3x-scenario/56365ca/ : SUCCESS in 58m 29s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/17/check/octavia-v1-dsvm-scenario-multinode/437b5f4/ : FAILURE in 1h 23m 22s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/17/check/octavia-v1-dsvm-py3x-scenario-multinode/6c4d5ab/ : FAILURE in 57m 36s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/17/check/octavia-v1-dsvm-scenario-kvm-centos.7/6c86269/ : SUCCESS in 55m 06s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/17/check/octavia-v1-dsvm-scenario-ubuntu.bionic/9f7de8a/ : SUCCESS in 1h 00m 43s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/17/check/octavia-v2-dsvm-noop-api/6dc23cd/ : SUCCESS in 37m 01s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/17/check/octavia-v2-dsvm-noop-py35-api/37d3590/ : SUCCESS in 42m 19s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/17/check/octavia-v2-dsvm-scenario/56e4481/ : SUCCESS in 1h 13m 16s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/17/check/octavia-v2-dsvm-py35-scenario/8a9fad0/ : SUCCESS in 1h 21m 28s (non-voting)\n- octavia-grenade octavia-grenade : ERROR Unable to find playbook /var/lib/zuul/builds/fe8d0a4aa9cd449b9abad02611d6dc64/untrusted/project_3/git.openstack.org/openstack/octavia/playbooks/legacy/grenade-devstack-octavia/run.yaml (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"c1103d8a2b42e5914f622197dcb568e2ffe7c6b1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-06-29 20:25:45.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9f935d63b3e86bb8233e8fc58e1e01592483de65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-29 22:21:17.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/18/check/openstack-tox-pep8/87440cb/ : SUCCESS in 7m 58s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/18/check/openstack-tox-py27/3976a9d/ : SUCCESS in 6m 29s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/18/check/openstack-tox-py35/adeb11c/ : SUCCESS in 11m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/18/check/build-openstack-sphinx-docs/f9d3447/html/ : SUCCESS in 7m 31s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/18/check/build-openstack-releasenotes/8d3f997/html/ : SUCCESS in 9m 45s\n- openstack-tox-cover http://logs.openstack.org/62/558962/18/check/openstack-tox-cover/30bb7a1/cover/ : SUCCESS in 7m 51s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/18/check/neutron-lbaasv2-dsvm-api/311bf14/ : SUCCESS in 1h 11m 50s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/18/check/neutron-lbaasv2-dsvm-py3x-api/e69b87c/ : SUCCESS in 1h 15m 55s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/18/check/neutron-lbaasv2-dsvm-scenario/084acdc/ : SUCCESS in 1h 04m 22s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/18/check/neutron-lbaasv2-dsvm-py3x-scenario/07d259b/ : FAILURE in 57m 08s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/18/check/openstack-tox-lower-constraints/1e29dae/ : SUCCESS in 9m 00s\n- openstack-tox-functional http://logs.openstack.org/62/558962/18/check/openstack-tox-functional/1343d4b/ : SUCCESS in 7m 29s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/18/check/openstack-tox-functional-py35/fe2506f/ : SUCCESS in 7m 37s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/18/check/octavia-v1-dsvm-scenario/f61f897/ : SUCCESS in 57m 24s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/18/check/octavia-v1-dsvm-py3x-scenario/2845e72/ : SUCCESS in 47m 46s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/18/check/octavia-v1-dsvm-scenario-multinode/275fead/ : FAILURE in 1h 17m 58s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/18/check/octavia-v1-dsvm-py3x-scenario-multinode/38a6182/ : FAILURE in 1h 29m 00s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/18/check/octavia-v1-dsvm-scenario-kvm-centos.7/064ffa1/ : SUCCESS in 1h 05m 11s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/18/check/octavia-v1-dsvm-scenario-ubuntu.bionic/ddf4fe4/ : SUCCESS in 1h 00m 45s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/18/check/octavia-v2-dsvm-noop-api/9d51cf3/ : SUCCESS in 39m 58s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/18/check/octavia-v2-dsvm-noop-py35-api/a4bcd24/ : SUCCESS in 36m 33s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/18/check/octavia-v2-dsvm-scenario/f9b9042/ : SUCCESS in 1h 21m 31s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/18/check/octavia-v2-dsvm-py35-scenario/968226c/ : SUCCESS in 1h 17m 24s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/18/check/octavia-grenade/ad48936/ : SUCCESS in 1h 18m 41s","accounts_in_message":[],"_revision_number":18},{"id":"91f91ef0710d2a993f90965807222d58c2cf8294","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-07-13 01:07:03.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"63132dc9212db5b022c6842ab2358bb93fdd7c0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-13 03:32:22.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/19/check/openstack-tox-pep8/d2483f4/ : SUCCESS in 9m 07s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/19/check/openstack-tox-py27/3d0da6b/ : SUCCESS in 6m 45s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/19/check/openstack-tox-py35/f20ab01/ : SUCCESS in 6m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/19/check/build-openstack-sphinx-docs/c59e8f3/html/ : SUCCESS in 8m 26s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/19/check/build-openstack-releasenotes/751395d/html/ : SUCCESS in 4m 53s\n- openstack-tox-cover http://logs.openstack.org/62/558962/19/check/openstack-tox-cover/7081602/cover/ : SUCCESS in 9m 51s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/19/check/neutron-lbaasv2-dsvm-api/5665059/ : SUCCESS in 1h 13m 27s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/19/check/neutron-lbaasv2-dsvm-py3x-api/9de7670/ : SUCCESS in 1h 22m 00s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/19/check/neutron-lbaasv2-dsvm-scenario/52a7aab/ : SUCCESS in 1h 12m 40s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/19/check/neutron-lbaasv2-dsvm-py3x-scenario/7af2ff3/ : SUCCESS in 1h 21m 45s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/19/check/openstack-tox-lower-constraints/7ce3e62/ : SUCCESS in 9m 10s\n- openstack-tox-functional http://logs.openstack.org/62/558962/19/check/openstack-tox-functional/4c2933f/ : SUCCESS in 7m 42s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/19/check/openstack-tox-functional-py35/8c2a4b3/ : SUCCESS in 11m 24s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/19/check/octavia-v1-dsvm-scenario/1646b07/ : SUCCESS in 47m 57s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/19/check/octavia-v1-dsvm-py3x-scenario/34fff1e/ : SUCCESS in 47m 30s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/19/check/octavia-v1-dsvm-scenario-multinode/4f26d6a/ : FAILURE in 1h 15m 11s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/19/check/octavia-v1-dsvm-py3x-scenario-multinode/01c4f0a/ : FAILURE in 1h 36m 20s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/19/check/octavia-v1-dsvm-scenario-kvm-centos.7/9768217/ : SUCCESS in 50m 47s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/19/check/octavia-v1-dsvm-scenario-ubuntu.bionic/b50c811/ : SUCCESS in 49m 02s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/19/check/octavia-v2-dsvm-noop-api/44c0dee/ : SUCCESS in 37m 45s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/19/check/octavia-v2-dsvm-noop-py35-api/0418b6a/ : SUCCESS in 43m 27s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/19/check/octavia-v2-dsvm-scenario/5723546/ : SUCCESS in 1h 20m 42s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/19/check/octavia-v2-dsvm-py35-scenario/e98c576/ : SUCCESS in 1h 29m 23s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/19/check/octavia-grenade/4e7c60d/ : SUCCESS in 1h 09m 25s","accounts_in_message":[],"_revision_number":19},{"id":"2d80b2d41a7b34c5e6b85d30b37a312503805dd7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-07-16 21:12:09.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"c4e8c18082eeb54b42384ef1673dc8e64d6b460e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-17 00:17:14.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/20/check/openstack-tox-pep8/8b7ae74/ : SUCCESS in 8m 49s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/20/check/openstack-tox-py27/73585bf/ : SUCCESS in 6m 06s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/20/check/openstack-tox-py35/25439ce/ : SUCCESS in 6m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/20/check/build-openstack-sphinx-docs/07e5dfd/html/ : SUCCESS in 7m 18s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/20/check/build-openstack-releasenotes/f37d025/html/ : SUCCESS in 5m 22s\n- openstack-tox-cover http://logs.openstack.org/62/558962/20/check/openstack-tox-cover/fc5d9ff/cover/ : SUCCESS in 9m 04s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/20/check/neutron-lbaasv2-dsvm-api/323d8bf/ : SUCCESS in 1h 10m 38s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/20/check/neutron-lbaasv2-dsvm-py3x-api/a96d7d5/ : SUCCESS in 1h 13m 16s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/20/check/neutron-lbaasv2-dsvm-scenario/451f284/ : SUCCESS in 1h 21m 22s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/20/check/neutron-lbaasv2-dsvm-py3x-scenario/282bdd2/ : SUCCESS in 1h 27m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/20/check/openstack-tox-lower-constraints/22c0362/ : SUCCESS in 9m 55s\n- openstack-tox-functional http://logs.openstack.org/62/558962/20/check/openstack-tox-functional/df9fabb/ : SUCCESS in 9m 04s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/20/check/openstack-tox-functional-py35/ae35568/ : SUCCESS in 8m 23s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/20/check/octavia-v1-dsvm-scenario/5fc3702/ : SUCCESS in 59m 33s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/20/check/octavia-v1-dsvm-py3x-scenario/870e75d/ : FAILURE in 27m 28s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/20/check/octavia-v1-dsvm-scenario-multinode/1db600d/ : FAILURE in 1h 20m 02s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/20/check/octavia-v1-dsvm-py3x-scenario-multinode/7c494b3/ : FAILURE in 40m 06s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/20/check/octavia-v1-dsvm-scenario-kvm-centos.7/73d39b3/ : FAILURE in 33m 39s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/20/check/octavia-v1-dsvm-scenario-ubuntu.bionic/b99eb92/ : SUCCESS in 1h 01m 23s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/20/check/octavia-v2-dsvm-noop-api/49ab9e1/ : SUCCESS in 41m 52s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/20/check/octavia-v2-dsvm-noop-py35-api/91d4a6d/ : SUCCESS in 45m 06s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/20/check/octavia-v2-dsvm-scenario/6fe6735/ : SUCCESS in 1h 07m 01s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/20/check/octavia-v2-dsvm-py35-scenario/a78480d/ : SUCCESS in 1h 08m 37s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/20/check/octavia-grenade/6fd3325/ : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":20},{"id":"caf617105cfc5c23515333e39e79536052c4b99a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-07-23 09:57:40.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"a2da1c6cf4f812c9bc17ec42b5df2e1f6a0f4ce7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-23 11:43:52.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/20/check/openstack-tox-pep8/fc2f925/ : SUCCESS in 8m 50s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/20/check/openstack-tox-py27/228fe5a/ : SUCCESS in 7m 18s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/20/check/openstack-tox-py35/bde59cb/ : SUCCESS in 6m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/558962/20/check/build-openstack-sphinx-docs/417ecee/ : POST_FAILURE in 7m 41s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/20/check/build-openstack-releasenotes/6a6cb89/html/ : SUCCESS in 5m 49s\n- openstack-tox-cover http://logs.openstack.org/62/558962/20/check/openstack-tox-cover/d496f3b/cover/ : SUCCESS in 10m 27s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/20/check/neutron-lbaasv2-dsvm-api/e2deea5/ : SUCCESS in 1h 21m 49s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/20/check/neutron-lbaasv2-dsvm-py3x-api/1b37738/ : SUCCESS in 1h 12m 07s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/20/check/neutron-lbaasv2-dsvm-scenario/a8bdd70/ : SUCCESS in 54m 02s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/20/check/neutron-lbaasv2-dsvm-py3x-scenario/9890ad8/ : SUCCESS in 1h 01m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/20/check/openstack-tox-lower-constraints/9171174/ : SUCCESS in 8m 06s\n- openstack-tox-functional http://logs.openstack.org/62/558962/20/check/openstack-tox-functional/dc26060/ : SUCCESS in 9m 06s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/20/check/openstack-tox-functional-py35/00395aa/ : SUCCESS in 7m 53s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/20/check/octavia-v1-dsvm-scenario/94c07e8/ : SUCCESS in 45m 02s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/20/check/octavia-v1-dsvm-py3x-scenario/f96dea7/ : SUCCESS in 50m 53s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/20/check/octavia-v1-dsvm-scenario-multinode/28e8789/ : FAILURE in 1h 31m 09s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/20/check/octavia-v1-dsvm-py3x-scenario-multinode/e59c00a/ : FAILURE in 1h 33m 00s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/20/check/octavia-v1-dsvm-scenario-kvm-centos.7/d059228/ : FAILURE in 26m 21s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/20/check/octavia-v1-dsvm-scenario-ubuntu.bionic/4f25520/ : SUCCESS in 52m 10s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/20/check/octavia-v2-dsvm-noop-api/f3e94f6/ : SUCCESS in 42m 27s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/20/check/octavia-v2-dsvm-noop-py35-api/8e9d66b/ : SUCCESS in 39m 38s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/20/check/octavia-v2-dsvm-scenario/824df3d/ : SUCCESS in 1h 09m 56s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/20/check/octavia-v2-dsvm-py35-scenario/4c4d3a0/ : SUCCESS in 1h 25m 58s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/20/check/octavia-grenade/dab58a6/ : SUCCESS in 1h 11m 00s","accounts_in_message":[],"_revision_number":20},{"id":"9f717c3bbd64c1441c6c020b6a9050c04f9397bb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-07-27 20:33:09.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"84224d7a0bf1ad9b64d4d78186179e200d3fcb9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-28 00:05:57.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/21/check/openstack-tox-pep8/608a857/ : SUCCESS in 8m 18s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/21/check/openstack-tox-py27/acaa761/ : SUCCESS in 7m 07s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/21/check/openstack-tox-py35/77a17cb/ : SUCCESS in 6m 57s\n- openstack-tox-docs http://logs.openstack.org/62/558962/21/check/openstack-tox-docs/9e4a79f/ : SUCCESS in 6m 13s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/21/check/build-openstack-releasenotes/abc8bb0/html/ : SUCCESS in 5m 33s\n- openstack-tox-cover http://logs.openstack.org/62/558962/21/check/openstack-tox-cover/b97e76d/cover/ : SUCCESS in 9m 39s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/21/check/neutron-lbaasv2-dsvm-api/02cba04/ : SUCCESS in 1h 18m 33s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/21/check/neutron-lbaasv2-dsvm-py3x-api/90d3a2b/ : SUCCESS in 1h 18m 48s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/21/check/neutron-lbaasv2-dsvm-scenario/d72516b/ : SUCCESS in 1h 27m 47s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/21/check/neutron-lbaasv2-dsvm-py3x-scenario/38aa628/ : SUCCESS in 1h 20m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/21/check/openstack-tox-lower-constraints/91aaa94/ : SUCCESS in 10m 36s\n- openstack-tox-functional http://logs.openstack.org/62/558962/21/check/openstack-tox-functional/395b49e/ : SUCCESS in 10m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/21/check/openstack-tox-functional-py35/4bedec9/ : SUCCESS in 8m 30s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/21/check/octavia-v1-dsvm-scenario/5666bb9/ : SUCCESS in 1h 04m 25s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/21/check/octavia-v1-dsvm-py3x-scenario/6f0cd2c/ : SUCCESS in 1h 08m 19s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/21/check/octavia-v1-dsvm-scenario-multinode/083d21c/ : FAILURE in 1h 22m 57s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/21/check/octavia-v1-dsvm-py3x-scenario-multinode/733fae0/ : FAILURE in 1h 19m 39s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/21/check/octavia-v1-dsvm-scenario-kvm-centos.7/507c052/ : SUCCESS in 1h 10m 54s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/21/check/octavia-v1-dsvm-scenario-ubuntu.bionic/30f0773/ : SUCCESS in 55m 34s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/21/check/octavia-v2-dsvm-noop-api/37c2f53/ : SUCCESS in 39m 33s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/21/check/octavia-v2-dsvm-noop-py35-api/f2c141a/ : SUCCESS in 41m 38s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/21/check/octavia-v2-dsvm-scenario/0cf14af/ : SUCCESS in 1h 18m 03s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/21/check/octavia-v2-dsvm-py35-scenario/1d5a110/ : SUCCESS in 1h 21m 08s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/21/check/octavia-grenade/24a5fd5/ : SUCCESS in 1h 15m 29s","accounts_in_message":[],"_revision_number":21},{"id":"6b77aead24d9eccc665e81de1c1d895a93d38915","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-07-30 18:08:51.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"71cdf76c7ef48459637718503dff060fc4fb126b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-30 20:05:59.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/22/check/openstack-tox-pep8/4d092c6/ : SUCCESS in 9m 20s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/22/check/openstack-tox-py27/268a38b/ : SUCCESS in 5m 27s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/22/check/openstack-tox-py35/098e341/ : SUCCESS in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/62/558962/22/check/openstack-tox-docs/77ba7dc/ : SUCCESS in 7m 39s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/22/check/build-openstack-releasenotes/98521a6/html/ : SUCCESS in 5m 47s\n- openstack-tox-cover http://logs.openstack.org/62/558962/22/check/openstack-tox-cover/ed9b278/cover/ : SUCCESS in 8m 31s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/22/check/neutron-lbaasv2-dsvm-api/2099a0d/ : SUCCESS in 1h 18m 57s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/22/check/neutron-lbaasv2-dsvm-py3x-api/340a0e5/ : SUCCESS in 1h 12m 53s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/22/check/neutron-lbaasv2-dsvm-scenario/cc548b3/ : SUCCESS in 56m 12s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/22/check/neutron-lbaasv2-dsvm-py3x-scenario/eb0dbc9/ : SUCCESS in 55m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/22/check/openstack-tox-lower-constraints/5456a2a/ : SUCCESS in 7m 59s\n- openstack-tox-functional http://logs.openstack.org/62/558962/22/check/openstack-tox-functional/6293f38/ : SUCCESS in 7m 50s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/22/check/openstack-tox-functional-py35/6708a72/ : SUCCESS in 8m 48s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/22/check/octavia-v1-dsvm-scenario/dc2c624/ : SUCCESS in 1h 35m 51s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/22/check/octavia-v1-dsvm-py3x-scenario/cbdde1f/ : SUCCESS in 48m 04s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/22/check/octavia-v1-dsvm-scenario-multinode/5e1db54/ : FAILURE in 1h 22m 41s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/22/check/octavia-v1-dsvm-py3x-scenario-multinode/0909c16/ : FAILURE in 1h 17m 08s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/22/check/octavia-v1-dsvm-scenario-kvm-centos.7/3a2433e/ : SUCCESS in 48m 47s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/22/check/octavia-v1-dsvm-scenario-ubuntu.bionic/b0c6901/ : SUCCESS in 46m 23s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/22/check/octavia-v2-dsvm-noop-api/481ddbf/ : SUCCESS in 37m 57s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/22/check/octavia-v2-dsvm-noop-py35-api/63ba2ee/ : SUCCESS in 43m 31s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/22/check/octavia-v2-dsvm-scenario/db49c07/ : SUCCESS in 1h 32m 51s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/22/check/octavia-v2-dsvm-py35-scenario/d9ad419/ : SUCCESS in 1h 28m 10s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/22/check/octavia-grenade/cc9dea0/ : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":22},{"id":"be8cbf7a4a2998de5642ada5cd232822acb5e181","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-08-21 01:51:26.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"9832e90e88939621266918c2bf70b7ce4b47e05d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-21 03:32:53.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/23/check/openstack-tox-pep8/3547194/ : SUCCESS in 8m 14s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/23/check/openstack-tox-py27/14501af/ : SUCCESS in 5m 25s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/23/check/openstack-tox-py35/deb385f/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/62/558962/23/check/openstack-tox-docs/fdf35c0/html/ : SUCCESS in 6m 51s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/23/check/build-openstack-releasenotes/c93f98f/html/ : SUCCESS in 4m 08s\n- openstack-tox-cover http://logs.openstack.org/62/558962/23/check/openstack-tox-cover/1061c03/cover/ : SUCCESS in 8m 08s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/23/check/neutron-lbaasv2-dsvm-api/ce78421/ : SUCCESS in 1h 13m 56s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/23/check/neutron-lbaasv2-dsvm-py3x-api/a8ad1e4/ : SUCCESS in 1h 11m 06s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/23/check/neutron-lbaasv2-dsvm-scenario/1448824/ : SUCCESS in 1h 28m 29s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/23/check/neutron-lbaasv2-dsvm-py3x-scenario/de8aec9/ : SUCCESS in 53m 13s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/23/check/openstack-tox-lower-constraints/54f49b0/ : FAILURE in 6m 22s\n- openstack-tox-functional http://logs.openstack.org/62/558962/23/check/openstack-tox-functional/b8e2796/ : SUCCESS in 6m 34s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/23/check/openstack-tox-functional-py35/b6ff34a/ : SUCCESS in 7m 06s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/23/check/octavia-v1-dsvm-scenario/8650707/ : SUCCESS in 1h 03m 10s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/23/check/octavia-v1-dsvm-py3x-scenario/e918586/ : SUCCESS in 59m 10s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/23/check/octavia-v1-dsvm-scenario-multinode/69e5434/ : FAILURE in 1h 32m 46s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/23/check/octavia-v1-dsvm-py3x-scenario-multinode/50198d6/ : FAILURE in 1h 11m 23s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/23/check/octavia-v1-dsvm-scenario-kvm-centos.7/55babc3/ : FAILURE in 27m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/62/558962/23/check/octavia-v2-dsvm-scenario-ubuntu.bionic/df21e27/ : SUCCESS in 1h 19m 56s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/23/check/octavia-v2-dsvm-noop-api/a89f727/ : SUCCESS in 32m 46s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/23/check/octavia-v2-dsvm-noop-py35-api/de54ec3/ : SUCCESS in 36m 55s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/23/check/octavia-v2-dsvm-scenario/b86ef01/ : SUCCESS in 1h 39m 34s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/23/check/octavia-v2-dsvm-py35-scenario/b001d98/ : SUCCESS in 1h 14m 37s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/23/check/octavia-grenade/faa31d0/ : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":23},{"id":"9a64059da26303212599bedafd6ac7f4b3792b2c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-08-24 06:54:09.000000000","message":"Patch Set 23:\n\nrecheck\n\nWeird...","accounts_in_message":[],"_revision_number":23},{"id":"78a847c2af6153b3fcaaab39e531c26d37674702","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-24 09:29:05.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/23/check/openstack-tox-pep8/abecbef/ : SUCCESS in 8m 07s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/23/check/openstack-tox-py27/d7c8350/ : SUCCESS in 5m 10s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/23/check/openstack-tox-py35/f03b68d/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/62/558962/23/check/openstack-tox-docs/0b92df5/html/ : SUCCESS in 6m 11s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/23/check/build-openstack-releasenotes/3b1f82f/html/ : SUCCESS in 4m 29s\n- openstack-tox-cover http://logs.openstack.org/62/558962/23/check/openstack-tox-cover/ccb96be/cover/ : SUCCESS in 9m 50s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/23/check/neutron-lbaasv2-dsvm-api/12d01c0/ : SUCCESS in 1h 13m 50s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/23/check/neutron-lbaasv2-dsvm-py3x-api/9d8cb4c/ : SUCCESS in 1h 19m 26s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/23/check/neutron-lbaasv2-dsvm-scenario/05b3d26/ : SUCCESS in 1h 35m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/23/check/neutron-lbaasv2-dsvm-py3x-scenario/33c0c8c/ : SUCCESS in 1h 21m 42s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/23/check/openstack-tox-lower-constraints/fbb6711/ : SUCCESS in 7m 45s\n- openstack-tox-functional http://logs.openstack.org/62/558962/23/check/openstack-tox-functional/138cff4/ : SUCCESS in 6m 58s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/23/check/openstack-tox-functional-py35/8ca61c2/ : SUCCESS in 7m 07s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/23/check/octavia-v1-dsvm-scenario/436c49a/ : SUCCESS in 1h 02m 12s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/23/check/octavia-v1-dsvm-py3x-scenario/ebc0f7f/ : SUCCESS in 1h 05m 22s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/23/check/octavia-v1-dsvm-scenario-multinode/6aa7031/ : FAILURE in 1h 16m 15s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/23/check/octavia-v1-dsvm-py3x-scenario-multinode/5466179/ : FAILURE in 1h 22m 00s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/62/558962/23/check/octavia-v1-dsvm-scenario-kvm-centos.7/6c9996f/ : SUCCESS in 57m 14s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/23/check/octavia-v2-dsvm-noop-api/e312a00/ : SUCCESS in 34m 54s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/23/check/octavia-v2-dsvm-noop-py35-api/1ff7008/ : SUCCESS in 36m 23s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/23/check/octavia-v2-dsvm-scenario/27c2a4b/ : SUCCESS in 1h 11m 15s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/23/check/octavia-v2-dsvm-py35-scenario/87b85b5/ : SUCCESS in 1h 24m 54s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/23/check/octavia-grenade/a5bb0b0/ : SUCCESS in 1h 25m 52s","accounts_in_message":[],"_revision_number":23},{"id":"d829e78a5654e5c1dcbaa42cb249b0b67a0ed998","author":{"_account_id":28067,"name":"Patrik Martinsson","email":"martinsson.patrik@gmail.com","username":"patchon"},"date":"2018-08-24 09:29:48.000000000","message":"Patch Set 23:\n\nHi, \n\nThanks for this. \n\nEven though this is experimental, we are trying it out in our OSP 13 installation. \nWe have manually bumped octavia to the stable/queens release 2.0.2 and merged this patch-set. \n\nThis works as I would expect and we are quite happy with this, given our \"pretty straightforward setup\". \n\nI have a few questions, and may also have found a bug (but not sure), anyways here goes, \n\nWe simply ha two az\u0027s, called A and B (actually we have a C also, but we leave that one out this), these are defined as a comma-separated list in octavia.conf as \"availability_zone \u003d A,B\" on all octavia-containers (which runs on all controllers). \n\nThe AZ placement doesn\u0027t seem to work when a amphora gets recreated from the health-manager (that is, I manually delete MASTER / BACKUP amp and wait for it to be recreated). The filtering seems to work according to the logs, *but* the actual request to nova-api seems to include the wrong AZ. Please see my findings at http://paste.openstack.org/show/728729/ (this seems like a bug if I\u0027m not mistaken).\n\nSecond question is, if we have this configuration with two AZ\u0027s (A B again), and the AZ B goes \"offline\", the health-manager will notice that the amp in that AZ got \u0027stale\u0027 and thus try to recreate the amp, *however* since we now already have an amp in AZ A, it will filter that AZ out and only leave us with AZ B. Will it then send that request off to nova-api and nova will try to create the instance according to instructions but fail (and if nova fails, will octavia try to recreate the amp until it succeeds) ? Or what will happen. I\u0027m just curios here. \n\nThanks for a great job on this patch, it really solves *our specific* needs. \nI know there is some discussion about this and that it tends to lean at \"that nova should handle the placement *not* octavia\" - but until that is in place in nova (I don\u0027t think it is, right? (I can\u0027t find any information about this), this will absolutely do the trick for us. \n\nThanks again!\n\nBest regards, \nPatrik Martinsson \nSweden","accounts_in_message":[],"_revision_number":23},{"id":"28988f68ea07926129401d4e41c84276bcb2a617","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-09-06 15:30:47.000000000","message":"Patch Set 23:\n\n\u003e Hi,\n \u003e \n \u003e Thanks for this.\n \u003e \n \u003e Even though this is experimental, we are trying it out in our OSP\n \u003e 13 installation.\n \u003e We have manually bumped octavia to the stable/queens release 2.0.2\n \u003e and merged this patch-set.\n \u003e \n \u003e This works as I would expect and we are quite happy with this,\n \u003e given our \"pretty straightforward setup\".\n\nThanks for testing this, and glad it works for you! I really do hope we can merge it officially in Stein, as it is an \"optional\" thing that shouldn\u0027t hurt anyone not configuring it.\n\n \u003e \n \u003e I have a few questions, and may also have found a bug (but not\n \u003e sure), anyways here goes,\n \u003e \n \u003e We simply ha two az\u0027s, called A and B (actually we have a C also,\n \u003e but we leave that one out this), these are defined as a\n \u003e comma-separated list in octavia.conf as \"availability_zone \u003d A,B\"\n \u003e on all octavia-containers (which runs on all controllers).\n \u003e \n\nThis will really only work sanely with Three AZs, for reasons you point out later.\n\n \u003e The AZ placement doesn\u0027t seem to work when a amphora gets recreated\n \u003e from the health-manager (that is, I manually delete MASTER / BACKUP\n \u003e amp and wait for it to be recreated). The filtering seems to work\n \u003e according to the logs, *but* the actual request to nova-api seems\n \u003e to include the wrong AZ. Please see my findings at\n \u003e http://paste.openstack.org/show/728729/ (this seems like a bug if\n \u003e I\u0027m not mistaken).\n\nYeah, that\u0027s a weird log. I\u0027ve been running/testing this in my environments for about 10 months now, and I have not recently seen anything like this (there were some bugs originally that caused issues like this, but I resolved all of the ones I was able to find). The code path should be the same for the failover scheduling as the initial scheduling as far as the AZ-picker goes, but I\u0027ll take a deeper look in a bit.\n\n \u003e \n \u003e Second question is, if we have this configuration with two AZ\u0027s (A\n \u003e B again), and the AZ B goes \"offline\", the health-manager will\n \u003e notice that the amp in that AZ got \u0027stale\u0027 and thus try to recreate\n \u003e the amp, *however* since we now already have an amp in AZ A, it\n \u003e will filter that AZ out and only leave us with AZ B. Will it then\n \u003e send that request off to nova-api and nova will try to create the\n \u003e instance according to instructions but fail (and if nova fails,\n \u003e will octavia try to recreate the amp until it succeeds) ? Or what\n \u003e will happen. I\u0027m just curios here.\n\nYeah, this is not a good thing. You really need to have 3 AZs for this to work very reliably, or at least count on nova being able to reschedule to the same AZ as the dead amp.\n \u003e \n \u003e Thanks for a great job on this patch, it really solves *our\n \u003e specific* needs.\n \u003e I know there is some discussion about this and that it tends to\n \u003e lean at \"that nova should handle the placement *not* octavia\" - but\n \u003e until that is in place in nova (I don\u0027t think it is, right? (I\n \u003e can\u0027t find any information about this), this will absolutely do the\n \u003e trick for us.\n \u003e \n \u003e Thanks again!\n \u003e \n \u003e Best regards,\n \u003e Patrik Martinsson\n \u003e Sweden\n\nAgain, glad you\u0027re finding it useful. Please feel free to poke other cores about reviewing this, as I know for a fact that there are others besides you and I that also need this functionality.","accounts_in_message":[],"_revision_number":23},{"id":"9309d33aa3a1e3099c2abe58e5e9d800efae20f5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-09-17 22:10:29.000000000","message":"Patch Set 24: Patch Set 23 was rebased","accounts_in_message":[],"_revision_number":24},{"id":"50960d036cfc62f5130ba893a4ff350b73a050c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-18 00:10:45.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/62/558962/24/check/openstack-tox-cover/6e8e0a6/cover/ : SUCCESS in 10m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/24/check/openstack-tox-lower-constraints/2e6539f/ : SUCCESS in 10m 19s\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/24/check/openstack-tox-pep8/67d0516/ : SUCCESS in 9m 25s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/24/check/openstack-tox-py27/969037f/ : SUCCESS in 6m 49s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/24/check/openstack-tox-py35/5d3b4c6/ : SUCCESS in 6m 38s\n- openstack-tox-py36 http://logs.openstack.org/62/558962/24/check/openstack-tox-py36/5bc1ed3/ : SUCCESS in 8m 17s\n- openstack-tox-docs http://logs.openstack.org/62/558962/24/check/openstack-tox-docs/8baa9ad/html/ : SUCCESS in 7m 34s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/24/check/build-openstack-releasenotes/1ed3659/html/ : SUCCESS in 6m 05s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/24/check/neutron-lbaasv2-dsvm-api/c4e9fe0/ : SUCCESS in 1h 06m 43s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/24/check/neutron-lbaasv2-dsvm-py3x-api/f206f85/ : SUCCESS in 1h 18m 21s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/24/check/neutron-lbaasv2-dsvm-scenario/4b3b532/ : SUCCESS in 1h 16m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/24/check/neutron-lbaasv2-dsvm-py3x-scenario/33c9d94/ : SUCCESS in 1h 27m 13s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/62/558962/24/check/openstack-tox-functional/5b6bf4e/ : SUCCESS in 7m 32s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/24/check/openstack-tox-functional-py35/d89144d/ : SUCCESS in 8m 48s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/24/check/octavia-v1-dsvm-scenario/d3176ba/ : SUCCESS in 58m 26s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/24/check/octavia-v1-dsvm-py3x-scenario/a53fcec/ : SUCCESS in 59m 43s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/24/check/octavia-v1-dsvm-scenario-multinode/e9e700b/ : FAILURE in 1h 05m 08s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/24/check/octavia-v1-dsvm-py3x-scenario-multinode/a4d8f34/ : FAILURE in 1h 15m 11s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/24/check/octavia-v2-dsvm-noop-api/b8eaf7c/ : SUCCESS in 37m 50s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/24/check/octavia-v2-dsvm-noop-py35-api/4ad3155/ : SUCCESS in 43m 59s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/24/check/octavia-v2-dsvm-scenario/52e74b9/ : SUCCESS in 1h 17m 37s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/24/check/octavia-v2-dsvm-py35-scenario/95afeb0/ : SUCCESS in 1h 15m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/62/558962/24/check/octavia-v2-dsvm-scenario-centos-7/aa58230/ : SUCCESS in 1h 54m 48s\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/62/558962/24/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6a9d9dc/ : SUCCESS in 1h 40m 36s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/24/check/octavia-grenade/a8997f0/ : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":24},{"id":"769370a75bbf06459a7db864ab96836466ac2cd5","author":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"date":"2018-09-20 07:25:21.000000000","message":"Patch Set 24:\n\n(7 comments)\n\nThis is a interesting patch, also in experimental status. No -1. ;-)","accounts_in_message":[],"_revision_number":24},{"id":"436a04e72d9f6453f4541fec4d7629728591cb58","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-10-01 22:10:31.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"98425fdc0da10caacf83fbb109ab15ebda035110","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-02 01:01:35.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/62/558962/25/check/openstack-tox-cover/6ae3e94/cover/ : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/25/check/openstack-tox-lower-constraints/2a2f4e2/ : SUCCESS in 8m 13s\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/25/check/openstack-tox-pep8/44006ca/ : SUCCESS in 8m 07s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/25/check/openstack-tox-py27/c3e576c/ : SUCCESS in 6m 29s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/25/check/openstack-tox-py35/84ea65b/ : SUCCESS in 7m 24s\n- openstack-tox-py36 http://logs.openstack.org/62/558962/25/check/openstack-tox-py36/bb31d1c/ : SUCCESS in 6m 19s\n- openstack-tox-docs http://logs.openstack.org/62/558962/25/check/openstack-tox-docs/62c8823/html/ : SUCCESS in 8m 02s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/25/check/build-openstack-releasenotes/bbcd23b/html/ : SUCCESS in 5m 36s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/25/check/neutron-lbaasv2-dsvm-api/529830c/ : SUCCESS in 1h 10m 26s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/25/check/neutron-lbaasv2-dsvm-py3x-api/2b48364/ : SUCCESS in 1h 14m 29s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/25/check/neutron-lbaasv2-dsvm-scenario/96a16d2/ : SUCCESS in 1h 16m 17s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/25/check/neutron-lbaasv2-dsvm-py3x-scenario/7c9ab66/ : SUCCESS in 1h 16m 48s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/62/558962/25/check/openstack-tox-functional/e2177d6/ : SUCCESS in 7m 03s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/25/check/openstack-tox-functional-py35/1c6672f/ : SUCCESS in 6m 55s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/25/check/octavia-v1-dsvm-scenario/223a450/ : SUCCESS in 55m 48s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/25/check/octavia-v1-dsvm-py3x-scenario/2ca2c26/ : SUCCESS in 56m 16s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/25/check/octavia-v1-dsvm-scenario-multinode/4e66624/ : FAILURE in 1h 16m 18s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/25/check/octavia-v1-dsvm-py3x-scenario-multinode/82245e0/ : FAILURE in 1h 21m 44s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/25/check/octavia-v2-dsvm-noop-api/2d03fb4/ : SUCCESS in 41m 12s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/25/check/octavia-v2-dsvm-noop-py35-api/16ec7c1/ : SUCCESS in 41m 29s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/25/check/octavia-v2-dsvm-scenario/f9b6548/ : SUCCESS in 1h 50m 17s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/25/check/octavia-v2-dsvm-py35-scenario/f3d1ed3/ : SUCCESS in 1h 35m 30s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/62/558962/25/check/octavia-v2-dsvm-scenario-centos-7/31fbb85/ : FAILURE in 33m 21s\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/62/558962/25/check/octavia-v2-dsvm-scenario-ubuntu-bionic/98760a1/ : SUCCESS in 1h 19m 31s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/25/check/octavia-grenade/5520942/ : SUCCESS in 1h 04m 39s","accounts_in_message":[],"_revision_number":25},{"id":"a2de71d3937674124eb0ccacf638a3cfb50402b3","author":{"_account_id":28067,"name":"Patrik Martinsson","email":"martinsson.patrik@gmail.com","username":"patchon"},"date":"2018-11-04 13:29:53.000000000","message":"Patch Set 25: Code-Review-1\n\n(1 comment)\n\nHi Adam,\n\nI heard you were on vacation, hope it\u0027s treating you well!\n\nI found the problem with the issue I had when recreating amp\u0027s (the one where they didn\u0027t respect the az-placement).\n\nI\u0027ve commented on the problematic part of the code, hope I\u0027m making sense.\n\nBest regards, \nPatrik Martinsson\nSweden","accounts_in_message":[],"_revision_number":25},{"id":"2b6b7b51e8d7006b639ea00e6b2a64b5290a5cb5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-11-05 21:50:37.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"4d09d3970cfa37cd1564f23c5bc4588efcd47d92","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-11-05 21:55:43.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"8ba817922dfb8e252832c5e7b0cbe0ba6f1dfbeb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-11-05 21:58:40.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"0fa0232a5b6eb1ca4575a6a24550312708a1b4b7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2018-11-05 22:44:07.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"5804aa1621892a0dfd6b99078741806dc38396e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-06 00:58:50.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/62/558962/28/check/openstack-tox-cover/c53956d/cover/ : SUCCESS in 8m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/28/check/openstack-tox-lower-constraints/5bcd898/ : SUCCESS in 7m 44s\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/28/check/openstack-tox-pep8/5c0d51e/ : SUCCESS in 10m 29s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/28/check/openstack-tox-py27/6ab74e6/ : SUCCESS in 5m 49s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/28/check/openstack-tox-py35/91e514f/ : SUCCESS in 6m 56s\n- openstack-tox-py36 http://logs.openstack.org/62/558962/28/check/openstack-tox-py36/ab87cca/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/62/558962/28/check/openstack-tox-docs/6b867e8/html/ : SUCCESS in 9m 16s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/28/check/build-openstack-releasenotes/65c1360/html/ : SUCCESS in 6m 57s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/28/check/neutron-lbaasv2-dsvm-api/6af44d6/ : SUCCESS in 1h 09m 33s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/28/check/neutron-lbaasv2-dsvm-py3x-api/5c271ef/ : SUCCESS in 1h 20m 43s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/28/check/neutron-lbaasv2-dsvm-scenario/5edc8c8/ : SUCCESS in 1h 30m 05s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/28/check/neutron-lbaasv2-dsvm-py3x-scenario/06281bf/ : SUCCESS in 1h 23m 44s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/62/558962/28/check/openstack-tox-functional/67305e4/ : SUCCESS in 7m 59s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/28/check/openstack-tox-functional-py35/768cf7a/ : SUCCESS in 7m 44s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/28/check/octavia-v1-dsvm-scenario/34df3f9/ : SUCCESS in 1h 00m 37s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/28/check/octavia-v1-dsvm-py3x-scenario/294aa5f/ : SUCCESS in 1h 35m 13s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/62/558962/28/check/octavia-v1-dsvm-scenario-multinode/c49550b/ : FAILURE in 1h 23m 01s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/62/558962/28/check/octavia-v1-dsvm-py3x-scenario-multinode/c80e7ba/ : FAILURE in 1h 31m 25s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/28/check/octavia-v2-dsvm-noop-api/1d41d97/ : SUCCESS in 33m 28s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/62/558962/28/check/octavia-v2-dsvm-noop-py35-api/8f1ed03/ : SUCCESS in 42m 35s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/28/check/octavia-v2-dsvm-scenario/8d3e1fa/ : SUCCESS in 1h 57m 50s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/62/558962/28/check/octavia-v2-dsvm-py35-scenario/35266a8/ : SUCCESS in 1h 31m 11s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/62/558962/28/check/octavia-v2-dsvm-scenario-centos-7/c993f2f/ : SUCCESS in 1h 55m 44s\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/62/558962/28/check/octavia-v2-dsvm-scenario-ubuntu-bionic/5a7beab/ : SUCCESS in 2h 12m 07s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/28/check/octavia-grenade/d8f1ae8/ : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":28},{"id":"9fd2d1360228fe1976e22cfe6d7de9e9610d975b","author":{"_account_id":28067,"name":"Patrik Martinsson","email":"martinsson.patrik@gmail.com","username":"patchon"},"date":"2018-11-13 12:25:42.000000000","message":"Patch Set 28:\n\nCool, that\u0027s exactly what I did internally.","accounts_in_message":[],"_revision_number":28},{"id":"b49c94cc9911df9fd97e06ec0dd27b8e317fd123","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-01-28 21:00:58.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"49d83bb2e7671c2e087c6d14a8a8235916ec6ae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-28 23:42:10.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/62/558962/29/check/openstack-tox-cover/01a7ff9/cover/ : SUCCESS in 10m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/29/check/openstack-tox-lower-constraints/06e6827/ : SUCCESS in 10m 33s\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/29/check/openstack-tox-pep8/52ebf53/ : SUCCESS in 8m 01s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/29/check/openstack-tox-py27/8df6764/ : SUCCESS in 8m 07s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/29/check/openstack-tox-py35/9f4d3ea/ : SUCCESS in 8m 43s\n- openstack-tox-py36 http://logs.openstack.org/62/558962/29/check/openstack-tox-py36/067633a/ : FAILURE in 6m 32s\n- openstack-tox-docs http://logs.openstack.org/62/558962/29/check/openstack-tox-docs/f1f1573/html/ : SUCCESS in 7m 25s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/29/check/build-openstack-releasenotes/2d40179/html/ : SUCCESS in 6m 05s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/29/check/neutron-lbaasv2-dsvm-api/51eccc0/ : SUCCESS in 1h 18m 56s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/29/check/neutron-lbaasv2-dsvm-py3x-api/ad4724f/ : SUCCESS in 1h 16m 08s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/29/check/neutron-lbaasv2-dsvm-scenario/9637847/ : SUCCESS in 1h 03m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/29/check/neutron-lbaasv2-dsvm-py3x-scenario/7e35929/ : SUCCESS in 1h 25m 09s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/62/558962/29/check/openstack-tox-functional/10f0e30/ : SUCCESS in 8m 20s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/29/check/openstack-tox-functional-py35/863198c/ : SUCCESS in 9m 23s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/29/check/octavia-v1-dsvm-scenario/7522666/ : SUCCESS in 1h 04m 55s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/29/check/octavia-v1-dsvm-py3x-scenario/1a04816/ : SUCCESS in 1h 02m 05s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/29/check/octavia-v2-dsvm-noop-api/1b9b462/ : SUCCESS in 38m 45s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/62/558962/29/check/octavia-v2-dsvm-noop-py2-api/467fa57/ : SUCCESS in 37m 13s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/29/check/octavia-v2-dsvm-scenario/2664dba/ : SUCCESS in 1h 30m 48s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/62/558962/29/check/octavia-v2-dsvm-py2-scenario/947fa08/ : SUCCESS in 1h 27m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/62/558962/29/check/octavia-v2-dsvm-py2-scenario-centos-7/924390a/ : TIMED_OUT in 2h 33m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/62/558962/29/check/octavia-v2-dsvm-scenario-ubuntu-bionic/e40d5fb/ : FAILURE in 1h 30m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/62/558962/29/check/octavia-v2-dsvm-scenario-two-node/ec42534/ : SUCCESS in 1h 34m 15s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/62/558962/29/check/octavia-v2-dsvm-py2-scenario-two-node/c54e36b/ : SUCCESS in 1h 35m 05s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/29/check/octavia-grenade/1b64d07/ : SUCCESS in 1h 28m 51s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/62/558962/29/check/octavia-v2-dsvm-tls-barbican/15d1e02/ : SUCCESS in 53m 05s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"e54b260d6a775a581ae72de86dc3636ddbf30c97","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-01-30 19:42:33.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"daa18ff2dfb70943b822a1c75ff362ed62519979","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-30 22:24:58.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/62/558962/30/check/openstack-tox-cover/37c9a1c/cover/ : SUCCESS in 9m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/30/check/openstack-tox-lower-constraints/8868cba/ : SUCCESS in 10m 56s\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/30/check/openstack-tox-pep8/b195aae/ : SUCCESS in 8m 48s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/30/check/openstack-tox-py27/6ff2103/ : SUCCESS in 7m 00s\n- openstack-tox-py35 http://logs.openstack.org/62/558962/30/check/openstack-tox-py35/43fadb0/ : SUCCESS in 5m 51s\n- openstack-tox-py36 http://logs.openstack.org/62/558962/30/check/openstack-tox-py36/515c825/ : SUCCESS in 7m 21s\n- openstack-tox-docs http://logs.openstack.org/62/558962/30/check/openstack-tox-docs/7913747/html/ : SUCCESS in 6m 58s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/30/check/build-openstack-releasenotes/046d381/html/ : SUCCESS in 5m 00s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/62/558962/30/check/neutron-lbaasv2-dsvm-api/700cd61/ : SUCCESS in 1h 30m 07s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/62/558962/30/check/neutron-lbaasv2-dsvm-py3x-api/2159843/ : SUCCESS in 1h 29m 40s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/62/558962/30/check/neutron-lbaasv2-dsvm-scenario/4cb172f/ : SUCCESS in 1h 34m 31s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/62/558962/30/check/neutron-lbaasv2-dsvm-py3x-scenario/c8c1afc/ : SUCCESS in 1h 33m 10s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/62/558962/30/check/openstack-tox-functional/0802f09/ : SUCCESS in 7m 18s\n- openstack-tox-functional-py35 http://logs.openstack.org/62/558962/30/check/openstack-tox-functional-py35/8343055/ : SUCCESS in 7m 57s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/62/558962/30/check/octavia-v1-dsvm-scenario/4473aa0/ : SUCCESS in 1h 05m 55s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/62/558962/30/check/octavia-v1-dsvm-py3x-scenario/3fcd305/ : SUCCESS in 1h 06m 17s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/30/check/octavia-v2-dsvm-noop-api/b79dba0/ : SUCCESS in 46m 45s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/62/558962/30/check/octavia-v2-dsvm-noop-py2-api/b240962/ : SUCCESS in 35m 21s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/30/check/octavia-v2-dsvm-scenario/1a25f7a/ : FAILURE in 1h 34m 49s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/62/558962/30/check/octavia-v2-dsvm-py2-scenario/be3e4d2/ : SUCCESS in 1h 26m 36s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/62/558962/30/check/octavia-v2-dsvm-py2-scenario-centos-7/90389fd/ : TIMED_OUT in 2h 33m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/62/558962/30/check/octavia-v2-dsvm-scenario-ubuntu-bionic/2d0370d/ : SUCCESS in 1h 46m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/62/558962/30/check/octavia-v2-dsvm-scenario-two-node/df20660/ : FAILURE in 1h 50m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/62/558962/30/check/octavia-v2-dsvm-py2-scenario-two-node/9f1f77f/ : SUCCESS in 1h 49m 36s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/30/check/octavia-grenade/11a4793/ : SUCCESS in 1h 07m 47s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/62/558962/30/check/octavia-v2-dsvm-tls-barbican/fe8dd21/ : SUCCESS in 50m 29s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"2bb875cc46100727176b06072a3c60e0d0850aa1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-11 21:42:38.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"c218367e002b3c6aa7bf39e97657d4881b6f71f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 23:50:24.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/62/558962/31/check/openstack-tox-cover/a2fdb45/cover/ : SUCCESS in 11m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/558962/31/check/openstack-tox-lower-constraints/b37a46d/ : SUCCESS in 7m 57s\n- openstack-tox-pep8 http://logs.openstack.org/62/558962/31/check/openstack-tox-pep8/10299eb/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/62/558962/31/check/openstack-tox-py27/58ccfcb/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/62/558962/31/check/openstack-tox-py36/78b490f/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/62/558962/31/check/openstack-tox-docs/3d2c61b/html/ : SUCCESS in 6m 28s\n- build-openstack-releasenotes http://logs.openstack.org/62/558962/31/check/build-openstack-releasenotes/dd60836/html/ : SUCCESS in 4m 50s\n- openstack-tox-functional http://logs.openstack.org/62/558962/31/check/openstack-tox-functional/ca34d5a/ : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 http://logs.openstack.org/62/558962/31/check/openstack-tox-functional-py36/25b809b/ : SUCCESS in 12m 12s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/62/558962/31/check/octavia-v2-dsvm-noop-api/2641835/ : SUCCESS in 48m 27s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/62/558962/31/check/octavia-v2-dsvm-noop-py2-api/8cf426f/ : SUCCESS in 45m 14s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/62/558962/31/check/octavia-v2-dsvm-scenario/bbc0e25/ : SUCCESS in 1h 28m 09s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/62/558962/31/check/octavia-v2-dsvm-py2-scenario/f64c3cf/ : SUCCESS in 1h 35m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/62/558962/31/check/octavia-v2-dsvm-py2-scenario-centos-7/8cc4244/ : FAILURE in 2h 01m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/62/558962/31/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6d6308f/ : SUCCESS in 1h 27m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/62/558962/31/check/octavia-v2-dsvm-scenario-two-node/c8f5bef/ : FAILURE in 1h 48m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/62/558962/31/check/octavia-v2-dsvm-py2-scenario-two-node/608e85a/ : SUCCESS in 1h 09m 20s (non-voting)\n- octavia-grenade http://logs.openstack.org/62/558962/31/check/octavia-grenade/2f7784b/ : SUCCESS in 1h 20m 15s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/62/558962/31/check/octavia-v2-dsvm-tls-barbican/4ff78ea/ : SUCCESS in 46m 12s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"16201595767f89989e5ab20e7737bfe6e4cdc87e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-18 18:15:06.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"e30365534394378fe55a93ca4147fb0e09679e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 20:44:39.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f60b6e2c4cfb41fa955f7fff925cefaf : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95ff583b105c4f96a54eb07c201d3c35 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e07a54b8a26491e966c4b2525b486c0 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4110d952ea1f4e1da3c92feb9ba1161c : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5347b0667a7467aacb0090220310065 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fcf58b539b8149378f872f1e40395cb5 : FAILURE in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f615db6beeb4de3856b1fdcf867efd0 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fe05eda5b254502b417d94694276d29 : SUCCESS in 7m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d15ee125d7a145119d904f72bb507fae : SUCCESS in 5m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/def7ef886fb94f31a841d1781bf5a196 : SUCCESS in 6m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c2c66c17533446859a30477b6163bc95 : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a2853e88d4d748488b7eae90abf7a3e6 : SUCCESS in 6m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ef08168fe8fd49379a1b5b743994b5d6 : SUCCESS in 42m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0261f84c6380433fad378ae2d7633042 : SUCCESS in 54m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73b7a6c6c24641fbb198dbf6d905fa97 : SUCCESS in 1h 44m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c65040f6d9354dadb0b6f6c1ac2048d7 : SUCCESS in 2h 01m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fe97fcdb68ea43c99909895ec70c7c7e : SUCCESS in 2h 18m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/193e06c2145341649393cafbd72b3b48 : FAILURE in 36m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fc7be9d7b72546df82f009ee001687c8 : SUCCESS in 1h 58m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8b9829be291443d3a71da2fc78f8bd50 : FAILURE in 1h 22m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b97e327d235249bd800bb18073cb61a5 : SUCCESS in 53m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0a545593932e44688df6e4561ba15cf3 : SUCCESS in 1h 40m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2cb15b3e4f64d2989277fcd8a34d88f : FAILURE in 42m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/420c71d284214fd09f20891a94d22135 : SUCCESS in 1h 24m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a31588f1d81446c786c588b7412efbae : FAILURE in 1h 17m 34s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"428a696d8df5cb21619bdfc98d827054affd3603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-24 05:33:18.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"54b176ce7bd45b377c86fb662b460abf74467f15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 08:10:05.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9692c500a0f49f49db092e63703cd63 : FAILURE in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf230b9f88424c6c829e4d9880e96545 : FAILURE in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d4d9a00f8b4b1ea239ed60ed4a8551 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e612b3c60b246dd8217184d6b8b5390 : FAILURE in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5897316dada44d1f86a734f6a39fb590 : FAILURE in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5eb8580e5b1491e94a44c49a308b6fa : FAILURE in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecf0f04126644dbb9de35c83727b18b0 : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b46a1056da264b27952e368b0a4643da : SUCCESS in 5m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/be83bf1b38364fa491f639f783c7291d : FAILURE in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b728098ae46b469dadb49e96d0f3d14d : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/27e1def1f69344a19812c804b87b8d8d : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/124bcca3046d4f2f942d1a8b4b1dd994 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/61d67fad0f37419e9f0b758dd9f32871 : TIMED_OUT in 2h 12m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/309002757d95455aa21036c889600d96 : FAILURE in 2h 28m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3d28192755e1412d81e24152f2e35973 : FAILURE in 1h 03m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/74d479d12e5645d290baaeddb4974a52 : FAILURE in 54m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f99ffc2c2b714557b049ff7ac4ab2267 : FAILURE in 1h 00m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6447ea4949540c5a3dfb4b7bbfc59ab : FAILURE in 1h 18m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d06e2e66423449578f7b1af855d25e47 : TIMED_OUT in 2h 32m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/38cfc9c17a674ab4bf694b9afb1cb4f7 : FAILURE in 48m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/daf3f1f76aa44d46b402af9d1ff4aceb : FAILURE in 2h 31m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/82c7b9159e514c5d94347a17e4e61c19 : FAILURE in 2h 12m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0f0b8eb227f94d2aa697012eada8cffe : FAILURE in 2h 29m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/53e3e291e5ce43fab878121f66b0772f : FAILURE in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"6b1e2b3f4857eadc66229380ab645864baea9608","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:19:34.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":33},{"id":"989e8ccdf67e7f8e996e75c1a5047021bd9ec7ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-27 10:49:14.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"fa6efd6ce2d17f56f931b5b90bb065215d6deddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-27 11:08:13.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"ee156f4cba1339ed8851d69df0f7da76252a5fd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-27 11:09:34.000000000","message":"Patch Set 36: Published edit on patch set 35.","accounts_in_message":[],"_revision_number":36},{"id":"5b1a99e6ba8f037c10d87995a6f16b4270c44317","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-27 14:16:19.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"f7b9236c108ca68ba4c86d9aaec2775afdfe2e35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-01-27 14:46:42.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"741204ec414a43602556930db763b1d6bd8e694a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-27 15:44:44.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"0b8d76af6c369809a5c4461994af19adc3f7610b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-27 16:43:55.000000000","message":"Patch Set 40: Published edit on patch set 39.","accounts_in_message":[],"_revision_number":40},{"id":"fd76efe62aadff7331a9d0450f1a8a1e903f768b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-27 16:47:06.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"a6439063762f228289f887eb53f31b44120c83db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 20:00:07.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e2b9e4976984027b031feaaadb2f282 : SUCCESS in 11m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eae00b7dcbe54ef9b22ba55d09dec231 : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a3b6de8e3cb4053b4085f79a29d7cd2 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d7b8a57458e4e3ca7424725c14c6439 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6c8ba68de8f477098cb59c131947d79 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb6f9fcd3bd24657927d18eee90a569a : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ad6bb1a756a4d2d9de956db73040331 : SUCCESS in 14m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4067355a32704994bbca58c17bf87806 : SUCCESS in 6m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b0392daca9784d1b98ffd35734a5cf4c : SUCCESS in 5m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ede3eaa5388e4a44bfce1578d58802ae : SUCCESS in 11m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5b7cffa55c264263bf5bd053f89f0e77 : SUCCESS in 5m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f7563735217d4020b58470d115c34af6 : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/32b68a0412f4485099e97a0f5defa4b3 : TIMED_OUT in 3h 04m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/305b92801f5844c18c98fcd80d3fc3a4 : FAILURE in 2h 31m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/02d69911b9044b95bae0643c3d347ab5 : FAILURE in 50m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f426b97209044883b678b36280a6f3ae : FAILURE in 1h 04m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7dcf3582bea649faa09438a160322096 : FAILURE in 52m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f02b9b82f7d74b39ac066149df27e498 : FAILURE in 2h 14m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5cfc62a6f4ad46fda25553ad5d7b338a : FAILURE in 2h 21m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e745298d7b16477c90f7acbe4c808b7e : FAILURE in 2h 10m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/711668af50434628bd7a8b92a079accf : FAILURE in 2h 12m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/02a8fa4ad1f84b3ea9f22d5a70812cca : FAILURE in 2h 10m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/2739729bb31c4e368b0b85e20bd73b69 : TIMED_OUT in 2h 33m 55s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"f2be553898398703d48734bc829f8994a97f0b70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"date":"2022-12-13 08:42:13.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"9ebf45694e2e3641da4381fb66d96b24e476dce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 09:51:50.000000000","message":"Patch Set 42: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3fba3bfdfa3492e9371bcb3883c76fe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28aaff5a335142dfb384008385da7d9e : FAILURE in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5afd48ca28f44caaa088f4f998d2d40e : FAILURE in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2da5a26af154440dbeac3898155e0742 : FAILURE in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e262b64dda244fe3b5f4164cd87b1b96 : FAILURE in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792fa55d8bb64bfabe25b60163270351 : FAILURE in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63e7cda3a5d243a1ac7b58a80233e148 : SUCCESS in 5m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7132c2f995514b8588778cb07575f1ec : FAILURE in 5m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/da14282da6fd4b3dbafc44d315b8c3e4 : SUCCESS in 10m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1ae06272fe1c4c62a1015150b085b4c6 : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3d66593007f6449d899a94c7d74fb5c7 : SUCCESS in 10m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6a1cd5bea1114127a549acdd64532e15 : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ccf7539ca4c472bbfb2f13a85b58d7b : SUCCESS in 9m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7b97df56518043968d780abbec4b245f : FAILURE in 32m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b36ef42730a4468dbabc831268da9f94 : FAILURE in 35m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e2dd689a8ac34b9990f8d35a6d900352 : FAILURE in 32m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/33aa103340a94efcaf78004cfe974c78 : FAILURE in 50m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a33de0449a7a4a01afa89354ea62de92 : FAILURE in 59m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0cf85c2ee1cb4d0e9084ecb1a4c1aa3b : FAILURE in 38m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/33c71e7be1174c5a8505e96bf78e6f93 : FAILURE in 35m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6f3a6c4809414e148f231eef9072a1ec : FAILURE in 47m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1cd34ffe243a4ff08996c310e02b4105 : FAILURE in 37m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/aceed8b430ed4a4a8752c073f9456fa3 : FAILURE in 32m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2d12de2d00934ab287dd671fd6444819 : FAILURE in 38m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a0acd1f1a194f77bd2caf286394695e : FAILURE in 40m 32s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"4b5124967618620bd9ecc23384d685b85c30ffb6","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2022-12-16 08:01:44.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_6732\u003e.","accounts_in_message":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"_revision_number":42},{"id":"d3d3304cfedf6bd6dce8cb8a69bbf90d051eba8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"date":"2022-12-16 12:31:34.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"2f2f88784dc69634ddca22334b0e94f345475df4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"date":"2022-12-16 13:44:09.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"1f9152d04073b11e50b07cc2e667ed6adfabf55d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 16:25:52.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec9d492e012b4a788856a797d6e1c9de\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bcd8bdd744341eea4c8120be57222e1 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1715b4a2aa14ebb94a1acfcb2d9c94f : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2cb30711aee4159a00647bdef6a6096 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50d5edb731374786a80f63b3119edd88 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b64c6d31384bf785b63c26c6652f7f : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05b5820412d7411493ef5ffa7bb04ad1 : SUCCESS in 5m 03s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/828d0b95accf4ed0aed4d3e2ee179ffc : SUCCESS in 4m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7262d5bf41a74d9ea894569e629ae025 : SUCCESS in 6m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ea83adf478544bf7bff81ee546f688a7 : SUCCESS in 4m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/34550e914da34f8ea570e036c4cf3532 : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f4e1b32a49774354b38b52626eefc2a1 : SUCCESS in 6m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a5c97ff18dfe4054a7ecda8299472af6 : SUCCESS in 6m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8293ef2ec19b4d03adb219f3e1d795b7 : FAILURE in 17m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8a46ac988734604984dd27a371a3c73 : SUCCESS in 1h 29m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/423d64eb514341468e344f25cc28abe3 : SUCCESS in 34m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/606373da8daf43e3b1596dbc32c6f2e7 : SUCCESS in 49m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b463db0c7e97467b9a597a656f086231 : FAILURE in 1h 11m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c354875349fe4e6985a2573e2d522bc7 : SUCCESS in 33m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/444f6813a13d4619b7622b8d35809798 : SUCCESS in 1h 40m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ddc158b54a714390a16053671e14e5ff : SUCCESS in 2h 00m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ffd8d1e6253a48a68879d3a36efe17df : SUCCESS in 1h 36m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/510380bae92a4b588659848f89abb4e9 : SUCCESS in 1h 57m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0480f344ad72487fb70dddfc2062a642 : SUCCESS in 1h 23m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d7bedbe2fea4e649d1d207626834c1d : TIMED_OUT in 2h 34m 43s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"80fe613d80ad607d282e4650fc4b134e61a6c85c","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":28067,"name":"Patrik Martinsson","email":"martinsson.patrik@gmail.com","username":"patchon"},"date":"2022-12-16 16:30:40.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_28067\u003e.","accounts_in_message":[{"_account_id":28067,"name":"Patrik Martinsson","email":"martinsson.patrik@gmail.com","username":"patchon"}],"_revision_number":44},{"id":"e4fa59d12327e1c047ef10ee49cc35cb7432ae0f","author":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"date":"2022-12-19 07:30:10.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"076cc7aa47d4eba65a201e12ea596ee999445818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 10:11:50.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bede00e6f206407a9cb7c5ace4921a13\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea6786bd9fb84dd5814fb8d746c196f6 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ef57860c1f74e0ea374a58434e2760b : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f00f7256ee564ba585a802e1fc05b569 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/574047ba829d44488278b59077b83515 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e1911bac0a8450da6c023bf0c2d572a : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3336c0c5ddd2430e9cef2221af5ff553 : SUCCESS in 5m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4b494def5af04d4283f7ba3744ea6597 : SUCCESS in 6m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8f43ed6a8e344111be640ccc166e3142 : SUCCESS in 7m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c1d427a05d5842c39a8a5d187f7d7ee6 : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b2a7b41d75fa42f1824295b04291d9ee : SUCCESS in 7m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/47f7410d49a24c219153fa6d0e79a527 : SUCCESS in 9m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ea22e85ddc84099b8c862169a54c2a3 : SUCCESS in 7m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c61c8e41d956455f8dc0ea5bc7d1ec90 : SUCCESS in 1h 34m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a3c39476e76419890e20d3a929b57b6 : SUCCESS in 2h 20m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/358a1b85c9ca41068c957d367c88988b : SUCCESS in 44m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8dc00a2529164fc2a1061a6e2405cda2 : SUCCESS in 1h 09m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3990e54f82214531a865642235c128e3 : FAILURE in 1h 24m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/57dfd50d07254f5789082a97a771bc08 : SUCCESS in 39m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/534c3ab971c34f899a8ed7c0a1a867cb : SUCCESS in 2h 22m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a76b5b12c7064d199d02a21a0b8152df : FAILURE in 25m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f6c67b8bd6ae44a0a215b99494b91f4e : SUCCESS in 2h 31m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c1ba1269302a407c80965cdb9303df12 : SUCCESS in 2h 12m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b95d7e9a836d4899a527497c5f9c74b1 : SUCCESS in 1h 52m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/080134055f154a99b5993ce3a1eec72c : SUCCESS in 2h 32m 19s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"4e792586775c357769bd9c6bb5130b9b758bfbdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-12 09:13:38.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"022467fa49c8aebae74896b1e37971b2ef0c9b27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:19:25.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28a4621748bb4a39bf5c533acebdb091\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27f8653d3910488a87126d0af5b561f0 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d2864f4e35f48f6aac6e7a327674a69 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/433b54ff50ee408dbfedd21145376111 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c13c71750940498fa7daa5c5ccf86d42 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f556b69066cf48d4a1b7b5ae0b17ef00 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/819ef83e8c5447b69113cce4ec0ff1c7 : SUCCESS in 4m 13s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/77b10b76476c471e94f3616001be2cb4 : SUCCESS in 4m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/934ca7b875c04f038da1fd203d8d885c : SUCCESS in 9m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/50a285ade5c14f798a692544a04f5be4 : SUCCESS in 4m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5427c119cc954acb99a0336521d47da8 : SUCCESS in 7m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c33b4d261f914e0cab79318417d9c4df : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bc1f606a2b14448885d8be7079668833 : SUCCESS in 6m 37s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d2959b129b494742b45151d6727821a9 : SUCCESS in 1h 37m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/87cfad0d8014407b974965580bbc6fc4 : SUCCESS in 2h 14m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e8618576297c47c1b23a6a59df0b5e65 : SUCCESS in 38m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/af7f93a961ca4daaab9316d652d1e4e5 : SUCCESS in 1h 05m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bcebed88a7c94e2bbba531eb17d22970 : SUCCESS in 1h 20m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b9e7d2c297b43c087e8f4e808ca4377 : SUCCESS in 31m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c41508fc7c0c4e838ad6612df544056a : SUCCESS in 2h 17m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/31d0e468c06a450aa6669ef8fe3b01c1 : SUCCESS in 3h 01m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/32697ecb2aa646318878cebc6c511903 : TIMED_OUT in 2h 33m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4187b1a479e44d1e8a1b0c1ec3338f0b : FAILURE in 2h 19m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a91b5fbb79bd49f9bd144d47a64ee6f9 : SUCCESS in 2h 03m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/08f76da6f3f94655ba7b9c643dcc902f : SUCCESS in 2h 28m 20s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"1886339106c7027d2ddb2c4c9f79cf5b5d497727","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"date":"2023-02-18 13:54:06.000000000","message":"Patch Set 46: Patch Set 45 was rebased","accounts_in_message":[],"_revision_number":46},{"id":"a19d566eab405c8360a8b0262d725567ddeef6f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 16:33:05.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb8849b23ad43538c5c8ef6170d5c39\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/161ed5b758c4481ca4a81776dcb092ac : SUCCESS in 11m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e02429d607646f6833f147adf4eec36 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e876950e14a54c65bbf861175be86d1e : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11496d377cb941079417b70ed9833b8e : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54770d6cbe5b468f96a77b8a95574b27 : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60694204e2244f7d963b9b7ba3af32ed : SUCCESS in 4m 34s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7624c49de8c74c6bb3da45c93ea1aacb : SUCCESS in 5m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e319c5c3bd5f4bb2b6ac81b924a856ef : SUCCESS in 7m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/be5315d9cb95484091177668c96ee60d : SUCCESS in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/87acf14048ff4b428c9cab5428f58e12 : SUCCESS in 6m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/df4be457a02843a2ba9c0ec70b7d3dbd : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5c014b0a0e4a4bf4ae587be4fd300616 : SUCCESS in 5m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/067605c971a641e29f17572d50187873 : SUCCESS in 1h 09m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/abe4d21d235741408876f8d07c8fa355 : SUCCESS in 1h 48m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e2eefb770b64c3280ed8100576f6e66 : SUCCESS in 38m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3948421053a444e79af1d7c2236a0466 : SUCCESS in 44m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b986a42dbd2d42b7aecd5379d94e25de : SUCCESS in 1h 01m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d36bea66a664cecb70a8ed0900906e7 : SUCCESS in 35m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/13b2f0c98341413abf8aacc06eff3201 : SUCCESS in 1h 56m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/36628e93e2034ed69214266f4ef1e30d : SUCCESS in 1h 58m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/886507c32112445a9ce74c37dde5cba7 : SUCCESS in 1h 50m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1f3ec062000a46d1bd821b15b99f0f00 : SUCCESS in 1h 46m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1516218ba5984892bf39a728afba6f29 : SUCCESS in 1h 43m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e61a508d19444dbfba7806aab2881b53 : SUCCESS in 2h 05m 13s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"f2fbc4188ec2f2716b79f50919461da367afd7f5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-27 17:13:06.000000000","message":"Patch Set 46: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"101b57d0c4f96268945aed4abd7f2f21d1345fbf","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-28 13:33:02.000000000","message":"Patch Set 46:\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"ff55c07b570a798a8ee3de243867e66955bf0109","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-28 14:08:28.000000000","message":"Patch Set 46:\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"323df3b09bf1c696b52f2a98fb0da6b3db522ce1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-28 14:23:11.000000000","message":"Patch Set 46:\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"d09c12088f940b2836b5976332007b8aedc3038d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-28 14:26:18.000000000","message":"Topic 548989 removed","accounts_in_message":[],"_revision_number":46},{"id":"41ed80a5b4af8006b168be13ddb05db869f3f5f5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-28 14:26:31.000000000","message":"Topic set to octavia-availability-zones","accounts_in_message":[],"_revision_number":46},{"id":"a27248a40f0a380a474e27144bac66f883abe0c8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-28 14:38:39.000000000","message":"Patch Set 46: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":46},{"id":"f332993731d6cf98eae01e4b2f3a4f5dd30d9447","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-28 14:44:50.000000000","message":"Patch Set 46: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"3de6d8062a904caf2c26417474bdd89f6bab9b54","author":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"date":"2023-03-28 14:47:50.000000000","message":"Patch Set 46:\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"f1a4b2d0e390b9934e3240ec4a106e331fbc72e0","author":{"_account_id":22072,"name":"Viktor Křivák","email":"viktor.krivak@ultimum.io","username":"viktor-krivak"},"date":"2023-03-28 15:40:41.000000000","message":"Patch Set 46:\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"59aab38c380a3d7bb70360dfd30af14c0dbe8abe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"date":"2024-03-12 03:48:33.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":47},{"id":"bb401df7c59e5bb5758ccde86129c9986c8f08f8","author":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"date":"2024-03-12 04:08:12.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"3f53a0ddb638513520975d5edb1f232c584ae5d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 06:36:58.000000000","message":"Patch Set 47: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/60516aa7add143bdb16d4b9e949c18aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40ec1c176ef04ff09a2881ae1e671b0d : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e494facaf3c04b98a7b5f4cec6e4710f : FAILURE in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73d266caaabd4ab78432376e9b288f21 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4dcfa9dcbba84fcea942dad0f2c6e6b2 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/683c2a82e6454f9285ce20eb0f675185 : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef92df5445bc4d48bc2b698663ad3276 : SUCCESS in 4m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/27ccc2c95abc405d9f202a5eff8b3242 : SUCCESS in 6m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bbf1a5483fba4bc6ae2051a1e84b18da : SUCCESS in 8m 37s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3224f29983b741f69b8f603d53117602 : SUCCESS in 10m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/941c8da0aadf420daa07e6001d063526 : SUCCESS in 5m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a33a2fd4193b44ddb82608c3df0126af : SUCCESS in 7m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/baf2f9b8bf7d49989076423acdc6939f : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4cc5fe049bc3470d90e66ad21a8f39b2 : SUCCESS in 7m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/503f257992c342b7a001a43d926d5bd9 : SUCCESS in 1h 30m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/de7813323ba940f7855b55afa3d3a9ab : SUCCESS in 2h 12m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0cec74269e6543ff9f458462575a7fac : SUCCESS in 1h 21m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f175746478564cdfa0403155926d35e1 : SUCCESS in 2h 20m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3107cdd2dc7542f7922224a9cdcc35f9 : SUCCESS in 1h 16m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c609b50334444b82b2c2e72a5df02d0c : SUCCESS in 41m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7833887e0a304d8d8aa84cdf45c4d7bd : SUCCESS in 38m 21s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/97e161379bd14f509639cca69eb94bd5 : SUCCESS in 43m 32s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/399ed39c72534cf6877ac9b29e858452 : SUCCESS in 1h 03m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb2cde6e3c394a75985aa74399be9aa7 : SUCCESS in 37m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f0b6bf865d3841c0ae3b447c25b14b86 : SUCCESS in 2h 13m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d4a255cc26db4a519124c87af597628b : SUCCESS in 2h 46m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3654a332c68042a1ad90b0778104ae43 : FAILURE in 27m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d952a68e09145918c4a2dbc971f9996 : TIMED_OUT in 2h 33m 27s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"91df11836e6d8e91eafe4597e0336c285d975e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"date":"2024-03-12 13:51:24.000000000","message":"Uploaded patch set 48.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":48},{"id":"ef05e411c91d3e551f2115b08c926b64805b1f86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 16:43:13.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1a589249cb34ca9968b89098c31d76b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/071fc8ab556c4dadb033ce54f79b5646 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaea000894be42db856f018913c2bf63 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88c4b8e5c39742f9aa6a50966d2cdcaf : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a3f44763b66409f811b1ed90276e478 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ad3a36794a4253bb828303a6922940 : SUCCESS in 13m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd5f751f65c24fba934e7ad06a5b5eb5 : SUCCESS in 6m 17s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1b42e33e3f324c2781359288c1022eb0 : SUCCESS in 6m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ec12a93bc471452b9f635580836165bf : SUCCESS in 7m 31s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f43c893ecef34724b6182e54f612f980 : SUCCESS in 10m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2a38c68b1483487cbb6508ab20a34544 : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/60b91f2ea5c34fc792b3361c9803f3e8 : SUCCESS in 7m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e00a32ca14e949baab80fd311b20106b : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/45de040ab8d641eabb55b0505749b126 : SUCCESS in 7m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f46aa4123ba74555b7af23636d01c305 : SUCCESS in 1h 38m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/01388b9714c54fbb8a2fe2760c114b5a : SUCCESS in 2h 12m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5c333b0fdd2b47cf8736695c2655f913 : FAILURE in 1h 19m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4b0121da3b12451a89662c1fc88a3314 : SUCCESS in 2h 04m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9f184b80a2a04fa4b819ad35ec29a9f5 : SUCCESS in 1h 04m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a85176de564940a1b566a9bbe0f7526a : SUCCESS in 37m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/244896dde3b0475095a56a334b7f850a : SUCCESS in 37m 13s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c9fdffaee31d452aa8128b0b3ee57a57 : SUCCESS in 38m 33s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c276ee5dd0564d80b53a7ba198c3981e : SUCCESS in 1h 13m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e7e73d8fb094f4b837651f88b7cc421 : SUCCESS in 32m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a638d3631c75446d855f0da6d331ef93 : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ea5c386e75f4686a12b3acf3fc8995f : SUCCESS in 2h 44m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e1bf4de84ed946e8a5ee0c0e124953f5 : FAILURE in 22m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ddc9631e57047078968113c4c882552 : SUCCESS in 2h 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"0de795223438e7efb452a4bc86b66c801d771980","author":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"date":"2024-03-13 10:31:28.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"d51fd9e495c630f6f4d1a4b5297801b224cab7a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2024-07-12 12:57:20.000000000","message":"Uploaded patch set 49.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":49},{"id":"6c5b8deb86b4d2c982aff7b417684e00112145af","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-12 13:03:29.000000000","message":"Patch Set 49: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"fb07902356da6dfe526b6157558d0e878f3f2ab2","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2024-07-12 13:04:09.000000000","message":"Patch Set 49:\n\n(3 comments)","accounts_in_message":[],"_revision_number":49},{"id":"37ab919c8a03b55f94849447c58c98ddaff17e21","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-12 13:07:31.000000000","message":"Patch Set 49:\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"5688adc3d91daf5c6ae45a16f14b8151b8ed892b","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2024-07-12 13:09:54.000000000","message":"Patch Set 49:\n\n(2 comments)","accounts_in_message":[],"_revision_number":49},{"id":"e6464621db644f827af41da6cf8af6ca0748f2e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 18:52:53.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6ff35f3b4ab4df7bcf6e35d60452ebc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4f9980baf6a42bc8e72f86d5221ecdc : SUCCESS in 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/867cfd49558d40128a52266097b0fe6d : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dabdbc2cadae44ab99549e849aa9b663 : SUCCESS in 7m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/92573f7e5e8746c6a5ae5918105c9791 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/539bf4203329424eacbd53ea70aa04b5 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e00d53a43965418380808e8fbe317ad1 : SUCCESS in 13m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1447a22c119406291d0aeb593911336 : SUCCESS in 4m 18s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a1d725d853994682ac151654c1c9d48d : SUCCESS in 5m 20s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/6d734980632f44bdb38d24e636d19ddd : SUCCESS in 7m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fc6fea4921db41c8a826392d4f7dae5e : SUCCESS in 4m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1662be60e0a74c4088b15d629aea63ab : SUCCESS in 7m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e9e5e3c08f154375b4989036f9e91fa7 : SUCCESS in 7m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f1fb11e7a16846c5ad02d9dfa6f6be4e : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59eee081022e4a599722be10fa017a44 : SUCCESS in 1h 37m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/75fe5beff89f4377b929bf3bd8526661 : SUCCESS in 2h 14m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/65f5c41f8f764693adf4c916ab93b5cd : SUCCESS in 1h 28m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ae871cde1d0b4f49837d62a476735675 : SUCCESS in 2h 22m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/12364df0e5f847ebb1c99d3a86b05327 : SUCCESS in 1h 04m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/366a204782f840bfba426872745a2be1 : SUCCESS in 31m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7f825430a5864852b69b8d7920b34d81 : SUCCESS in 50m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/94611396f75f40418ceeb117e2ea3a58 : SUCCESS in 1h 07m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/115ec03f505a40f597a4c3560738b868 : SUCCESS in 33m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0a560f7113494eb18f42a0ae1c3360f3 : TIMED_OUT in 2h 32m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/117a5b5206ca4a6eb4e6388413de15c7 : SUCCESS in 2h 46m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8a24238f468e466cbff5dc71689e31e7 : FAILURE in 21m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8bc5d59cecd48d5bd1cf45b88a5b847 : SUCCESS in 2h 29m 53s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"c0e551561cae0392a2fc35ccc53a77ff59136c5a","author":{"_account_id":36250,"name":"Przemysław Kuczyński","display_name":"przemekk","email":"przemyslaw.kuczynski@ochk.pl","username":"przemekk"},"date":"2025-12-10 17:13:55.000000000","message":"Patch Set 49:\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"5a39a187a62654a0de919c3df769f34f16ae4013","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-10 17:52:11.000000000","message":"Patch Set 49:\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"32e38e8c8d552b871dc00e496ff2327579f22e9d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-10 18:01:46.000000000","message":"Patch Set 49:\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"cb64b4a13f6abfd9af1cd592192ce9ed34a67829","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-11 08:06:23.000000000","message":"Patch Set 49:\n\n(1 comment)","accounts_in_message":[],"_revision_number":49}],"current_revision_number":49,"current_revision":"7df248aa4f2a9e41b2d1c7c9697faa6b512c7c9e","revisions":{"eba89c9f17eabf14f5f2b2eed587d1a255e0b5ce":{"kind":"REWORK","_number":1,"created":"2018-04-04 23:44:59.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/1"}}},"commit":{"parents":[{"commit":"e819e4521ad6c4a02086309abb922206125ef596","subject":"Allow members to be set as \"backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e819e4521ad6c4a02086309abb922206125ef596"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"subject":"WIP: Experimental multi-az support","message":"WIP: Experimental multi-az support\n\nThis requires nova to be patched to schedule to multiple AZs.\n... I might be able to fix that without too much effort though.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eba89c9f17eabf14f5f2b2eed587d1a255e0b5ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eba89c9f17eabf14f5f2b2eed587d1a255e0b5ce"}]},"branch":"refs/heads/master"},"b8b57f9622cccd075422ce565a160c0adee882ec":{"kind":"REWORK","_number":2,"created":"2018-04-06 00:37:00.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/2"}}},"commit":{"parents":[{"commit":"e819e4521ad6c4a02086309abb922206125ef596","subject":"Allow members to be set as \"backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e819e4521ad6c4a02086309abb922206125ef596"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-06 00:15:01.000000000","tz":540},"subject":"WIP: Experimental multi-az support","message":"WIP: Experimental multi-az support\n\nThis requires nova to be patched to schedule to multiple AZs.\n... I might be able to fix that without too much effort though.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8b57f9622cccd075422ce565a160c0adee882ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8b57f9622cccd075422ce565a160c0adee882ec"}]},"branch":"refs/heads/master"},"56f0fd7b3eb22d6fc6fec1c503fdc58fe70d89a6":{"kind":"REWORK","_number":3,"created":"2018-04-06 20:57:48.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/3"}}},"commit":{"parents":[{"commit":"e819e4521ad6c4a02086309abb922206125ef596","subject":"Allow members to be set as \"backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e819e4521ad6c4a02086309abb922206125ef596"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-06 20:56:26.000000000","tz":540},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56f0fd7b3eb22d6fc6fec1c503fdc58fe70d89a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56f0fd7b3eb22d6fc6fec1c503fdc58fe70d89a6"}]},"branch":"refs/heads/master"},"0b4bf6e856b03e57b5ff4f83f959acf655037a41":{"kind":"REWORK","_number":4,"created":"2018-04-06 22:14:07.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/4"}}},"commit":{"parents":[{"commit":"e819e4521ad6c4a02086309abb922206125ef596","subject":"Allow members to be set as \"backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e819e4521ad6c4a02086309abb922206125ef596"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-06 22:07:40.000000000","tz":540},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b4bf6e856b03e57b5ff4f83f959acf655037a41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b4bf6e856b03e57b5ff4f83f959acf655037a41"}]},"branch":"refs/heads/master"},"55390d218a5352eb46f3354050a7ae2d4a52a11a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2018-04-09 21:20:08.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/5"}}},"commit":{"parents":[{"commit":"1f278e7ab336bb203dd4c8a01f33e8ddf5fb0220","subject":"Merge \"Expose timeout options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f278e7ab336bb203dd4c8a01f33e8ddf5fb0220"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-09 21:20:08.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55390d218a5352eb46f3354050a7ae2d4a52a11a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55390d218a5352eb46f3354050a7ae2d4a52a11a"}]},"branch":"refs/heads/master"},"da70dd8c234dcf09a7dde5ddcdda3ede083b62fd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2018-04-24 01:14:48.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/6"}}},"commit":{"parents":[{"commit":"afb47816ac7a1299e0ea8a3ce65d24fa6c8f9db6","subject":"Merge \"Trivial: fix syntax error in command examples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/afb47816ac7a1299e0ea8a3ce65d24fa6c8f9db6"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-24 01:14:48.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da70dd8c234dcf09a7dde5ddcdda3ede083b62fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da70dd8c234dcf09a7dde5ddcdda3ede083b62fd"}]},"branch":"refs/heads/master"},"18c787190d17bd7db7fa5c7132ce32bc3832350d":{"kind":"REWORK","_number":7,"created":"2018-04-25 01:51:42.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/7"}}},"commit":{"parents":[{"commit":"afb47816ac7a1299e0ea8a3ce65d24fa6c8f9db6","subject":"Merge \"Trivial: fix syntax error in command examples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/afb47816ac7a1299e0ea8a3ce65d24fa6c8f9db6"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-25 01:51:34.000000000","tz":-420},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18c787190d17bd7db7fa5c7132ce32bc3832350d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18c787190d17bd7db7fa5c7132ce32bc3832350d"}]},"branch":"refs/heads/master"},"50ccb1ecfda494e8e852a34a15162850e97c4921":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2018-04-26 04:54:29.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/8"}}},"commit":{"parents":[{"commit":"0c6f281ccd95e4f42f81a1289c92ec7b4f26efd2","subject":"Merge \"Fix lower-constraints to actually match our reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c6f281ccd95e4f42f81a1289c92ec7b4f26efd2"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-26 04:54:29.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50ccb1ecfda494e8e852a34a15162850e97c4921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50ccb1ecfda494e8e852a34a15162850e97c4921"}]},"branch":"refs/heads/master"},"c796b794e4aa3d7a227214a742e282a1181869b6":{"kind":"REWORK","_number":9,"created":"2018-04-26 20:37:25.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/9"}}},"commit":{"parents":[{"commit":"0c6f281ccd95e4f42f81a1289c92ec7b4f26efd2","subject":"Merge \"Fix lower-constraints to actually match our reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c6f281ccd95e4f42f81a1289c92ec7b4f26efd2"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-26 20:36:51.000000000","tz":-420},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c796b794e4aa3d7a227214a742e282a1181869b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c796b794e4aa3d7a227214a742e282a1181869b6"}]},"branch":"refs/heads/master"},"4b688ef2a319d41e724402410dcdbe0661837315":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2018-04-26 22:40:05.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/10"}}},"commit":{"parents":[{"commit":"838ba5dad434433ab72696f1d94ba9e7b394ac63","subject":"Fix amp failover where failover already failed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/838ba5dad434433ab72696f1d94ba9e7b394ac63"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-26 22:39:43.000000000","tz":-420},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b688ef2a319d41e724402410dcdbe0661837315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b688ef2a319d41e724402410dcdbe0661837315"}]},"branch":"refs/heads/master"},"da4d4da887fc9c13f1a267ca9a2f17e23763a652":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2018-04-27 20:29:31.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/11"}}},"commit":{"parents":[{"commit":"1decf3627113a1a0a67bfbded5ac73a030657cc2","subject":"Fix amp failover where failover already failed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1decf3627113a1a0a67bfbded5ac73a030657cc2"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-27 20:29:31.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da4d4da887fc9c13f1a267ca9a2f17e23763a652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da4d4da887fc9c13f1a267ca9a2f17e23763a652"}]},"branch":"refs/heads/master"},"28b8046ecdb171f25145b6d9df8af066b1994277":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2018-05-07 18:58:40.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/12"}}},"commit":{"parents":[{"commit":"b7536a323f1e5a9cfa1463c0083ae7b9c52f35d9","subject":"Fix amp failover where failover already failed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7536a323f1e5a9cfa1463c0083ae7b9c52f35d9"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-05-07 18:58:40.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28b8046ecdb171f25145b6d9df8af066b1994277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28b8046ecdb171f25145b6d9df8af066b1994277"}]},"branch":"refs/heads/master"},"e81dcd9ac723b3e0c64aea3950168664aa58eb8e":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2018-05-31 23:52:44.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/13"}}},"commit":{"parents":[{"commit":"a684d061c5a9ccc8c9b047fd62fa6d53b99b416a","subject":"Fix amp failover where failover already failed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a684d061c5a9ccc8c9b047fd62fa6d53b99b416a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-05-31 23:52:44.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e81dcd9ac723b3e0c64aea3950168664aa58eb8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e81dcd9ac723b3e0c64aea3950168664aa58eb8e"}]},"branch":"refs/heads/master"},"004a3cc90398b7eb01b0e50cd11d1f1bc85a8dc4":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2018-06-02 01:53:55.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/14"}}},"commit":{"parents":[{"commit":"928a716acdacbe48bdc2b08372530f064d4e3b4d","subject":"Fix amp failover where failover already failed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/928a716acdacbe48bdc2b08372530f064d4e3b4d"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-02 01:53:55.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/004a3cc90398b7eb01b0e50cd11d1f1bc85a8dc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/004a3cc90398b7eb01b0e50cd11d1f1bc85a8dc4"}]},"branch":"refs/heads/master"},"d45ed557f8f22c589eec53d907dc4f5a98d0144a":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2018-06-05 20:34:39.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/15"}}},"commit":{"parents":[{"commit":"06ddbd16f208fcab3216023ef4146c633c2d38a1","subject":"Fix amp failover where failover already failed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06ddbd16f208fcab3216023ef4146c633c2d38a1"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-05 20:34:39.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d45ed557f8f22c589eec53d907dc4f5a98d0144a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d45ed557f8f22c589eec53d907dc4f5a98d0144a"}]},"branch":"refs/heads/master"},"6f51e5e78a59738d9c36b7f6e69ddf59d0d6f040":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2018-06-07 01:38:31.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/16"}}},"commit":{"parents":[{"commit":"96cce3ed74c63e5756e216a8e48c60de09b027c2","subject":"Fix amp failover where failover already failed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96cce3ed74c63e5756e216a8e48c60de09b027c2"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-07 01:37:55.000000000","tz":-420},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f51e5e78a59738d9c36b7f6e69ddf59d0d6f040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f51e5e78a59738d9c36b7f6e69ddf59d0d6f040"}]},"branch":"refs/heads/master"},"aafad06eafe67d57e85ba4106b628a227d065d33":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2018-06-14 18:00:20.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/17"}}},"commit":{"parents":[{"commit":"b9a8309d502e57cee918e169d1429a2ab0aac5e2","subject":"Merge \"Improve the error logging for zombie amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b9a8309d502e57cee918e169d1429a2ab0aac5e2"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-14 18:00:20.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aafad06eafe67d57e85ba4106b628a227d065d33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aafad06eafe67d57e85ba4106b628a227d065d33"}]},"branch":"refs/heads/master"},"2f93d0e5d0c56e54718b7e253793769a4d9089a0":{"kind":"REWORK","_number":18,"created":"2018-06-29 20:25:45.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/18"}}},"commit":{"parents":[{"commit":"3d545ab1b6e133f4696fdd389472b869e86d177b","subject":"Merge \"Fix configured haproxy restarts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d545ab1b6e133f4696fdd389472b869e86d177b"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-06-29 20:13:42.000000000","tz":-300},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f93d0e5d0c56e54718b7e253793769a4d9089a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f93d0e5d0c56e54718b7e253793769a4d9089a0"}]},"branch":"refs/heads/master"},"202164ad4497867187efcf38105102cf04150ad4":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2018-07-13 01:07:03.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/19"}}},"commit":{"parents":[{"commit":"26852b00de0127c5823ec060a24b4c0f5f62e4fb","subject":"Merge \"Improve resource quota response message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26852b00de0127c5823ec060a24b4c0f5f62e4fb"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-07-13 01:06:37.000000000","tz":-420},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/202164ad4497867187efcf38105102cf04150ad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/202164ad4497867187efcf38105102cf04150ad4"}]},"branch":"refs/heads/master"},"0b7404cb325ab7bbc6bccf80e29a576e3c3cfcb5":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2018-07-16 21:12:09.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/20"}}},"commit":{"parents":[{"commit":"7edad9390983bcd2abd05e82dae9b41a6092c403","subject":"Add release note for HMAC python3 fix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7edad9390983bcd2abd05e82dae9b41a6092c403"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-07-16 21:12:09.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b7404cb325ab7bbc6bccf80e29a576e3c3cfcb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b7404cb325ab7bbc6bccf80e29a576e3c3cfcb5"}]},"branch":"refs/heads/master"},"d279203e2dc594d282ef596396d035f5f04ee70e":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2018-07-27 20:33:09.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/21"}}},"commit":{"parents":[{"commit":"d3092c0facdbd4e3d0c0190710e286914ee14006","subject":"Merge \"Implement provider drivers - Driver Library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3092c0facdbd4e3d0c0190710e286914ee14006"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-07-27 20:33:09.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d279203e2dc594d282ef596396d035f5f04ee70e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d279203e2dc594d282ef596396d035f5f04ee70e"}]},"branch":"refs/heads/master"},"0cc6f21364e499ba61694c3c3a828d1e2688949f":{"kind":"REWORK","_number":22,"created":"2018-07-30 18:08:51.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/22"}}},"commit":{"parents":[{"commit":"2fe50a4146be1263266fb1b5dcd3e2bd3e4d3ff7","subject":"Merge \"Use openstack client command to replace nova client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fe50a4146be1263266fb1b5dcd3e2bd3e4d3ff7"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-07-30 18:08:40.000000000","tz":-420},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0cc6f21364e499ba61694c3c3a828d1e2688949f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0cc6f21364e499ba61694c3c3a828d1e2688949f"}]},"branch":"refs/heads/master"},"27e49db51a6d380fc3d68de2b70ae17237b8f042":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2018-08-21 01:51:26.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/23"}}},"commit":{"parents":[{"commit":"798ce95760e3df09c328f01db9a53866a3d07045","subject":"Disable KVM at OVH","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/798ce95760e3df09c328f01db9a53866a3d07045"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-08-21 01:51:26.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e49db51a6d380fc3d68de2b70ae17237b8f042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e49db51a6d380fc3d68de2b70ae17237b8f042"}]},"branch":"refs/heads/master"},"3bf9211414549306dec24e7f6001ab12542cf238":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2018-09-17 22:10:29.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/24"}}},"commit":{"parents":[{"commit":"aee5c6e10e51bbb4e0999c43b00ea2f8dee6991a","subject":"Merge \"Gate on octavia-dsvm-base based jobs and housekeeping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aee5c6e10e51bbb4e0999c43b00ea2f8dee6991a"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-09-17 22:10:29.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3bf9211414549306dec24e7f6001ab12542cf238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3bf9211414549306dec24e7f6001ab12542cf238"}]},"branch":"refs/heads/master"},"c1b1a5426732621d8c4ac8535c768b98ac9dc3f7":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2018-10-01 22:10:31.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/25"}}},"commit":{"parents":[{"commit":"687a0e84727f433911fb771b69154368295fbc67","subject":"Merge \"Separate the thread pool for health and stats update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/687a0e84727f433911fb771b69154368295fbc67"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-10-01 22:10:31.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1b1a5426732621d8c4ac8535c768b98ac9dc3f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1b1a5426732621d8c4ac8535c768b98ac9dc3f7"}]},"branch":"refs/heads/master"},"a62bd4b700be9fbe148c4494c6f448d273f0eb37":{"kind":"REWORK","_number":26,"created":"2018-11-05 21:50:37.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/26"}}},"commit":{"parents":[{"commit":"5bd7cc7cb41aee81e81c7f472d98558420e986d9","subject":"Merge \"Update README by adding Mailing List and Wiki URL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bd7cc7cb41aee81e81c7f472d98558420e986d9"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-11-05 21:50:37.000000000","tz":540},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a62bd4b700be9fbe148c4494c6f448d273f0eb37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a62bd4b700be9fbe148c4494c6f448d273f0eb37"}]},"branch":"refs/heads/master"},"f9d1508c42654899b8ca4c78b66af2c6386df7aa":{"kind":"REWORK","_number":27,"created":"2018-11-05 21:58:40.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/27"}}},"commit":{"parents":[{"commit":"5bd7cc7cb41aee81e81c7f472d98558420e986d9","subject":"Merge \"Update README by adding Mailing List and Wiki URL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bd7cc7cb41aee81e81c7f472d98558420e986d9"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-11-05 21:58:42.000000000","tz":540},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9d1508c42654899b8ca4c78b66af2c6386df7aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9d1508c42654899b8ca4c78b66af2c6386df7aa"}]},"branch":"refs/heads/master"},"6cdc71cd3e4daea4d6a7bee4f4e3da5254f2ace7":{"kind":"REWORK","_number":28,"created":"2018-11-05 22:44:07.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/28"}}},"commit":{"parents":[{"commit":"5bd7cc7cb41aee81e81c7f472d98558420e986d9","subject":"Merge \"Update README by adding Mailing List and Wiki URL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bd7cc7cb41aee81e81c7f472d98558420e986d9"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-11-05 22:33:49.000000000","tz":540},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cdc71cd3e4daea4d6a7bee4f4e3da5254f2ace7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cdc71cd3e4daea4d6a7bee4f4e3da5254f2ace7"}]},"branch":"refs/heads/master"},"ca61ce5dd268e443a45fea3e334150c349ae2974":{"kind":"REWORK","_number":29,"created":"2019-01-28 21:00:58.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/29"}}},"commit":{"parents":[{"commit":"6f96401e9c9c0fbac879dc174f129cfcaafb5f09","subject":"Merge \"Clarify the amp_ssh_key_name configuration setting.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f96401e9c9c0fbac879dc174f129cfcaafb5f09"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-01-28 21:00:47.000000000","tz":-360},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ca61ce5dd268e443a45fea3e334150c349ae2974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ca61ce5dd268e443a45fea3e334150c349ae2974"}]},"branch":"refs/heads/master"},"e43d8ff6ed5e66d323a93f9ca7b095f3e2ec54f6":{"kind":"REWORK","_number":30,"created":"2019-01-30 19:42:33.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/30"}}},"commit":{"parents":[{"commit":"979144f2fdf7c391b6c154c01a22f107f45da833","subject":"Merge \"Fix a topology bug in flavors support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/979144f2fdf7c391b6c154c01a22f107f45da833"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-01-30 19:40:48.000000000","tz":-480},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e43d8ff6ed5e66d323a93f9ca7b095f3e2ec54f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e43d8ff6ed5e66d323a93f9ca7b095f3e2ec54f6"}]},"branch":"refs/heads/master"},"80aa79db60aed85dd55c83eb39f9238aa74b99de":{"kind":"REWORK","_number":31,"created":"2019-05-11 21:42:38.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/31"}}},"commit":{"parents":[{"commit":"29d4340e9fd0c93595183ce16110198fef58c86b","subject":"Remove v1 API and associated code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29d4340e9fd0c93595183ce16110198fef58c86b"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-11 21:39:18.000000000","tz":-420},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80aa79db60aed85dd55c83eb39f9238aa74b99de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80aa79db60aed85dd55c83eb39f9238aa74b99de"}]},"branch":"refs/heads/master"},"07e3870d422b7688d15733afa52f873f7e8833cb":{"kind":"REWORK","_number":32,"created":"2019-11-18 18:15:06.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/32"}}},"commit":{"parents":[{"commit":"42bb73d36402184d9d21b7d59edfbe5795d363e8","subject":"Merge \"Fix filtering for provider capabilities list API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42bb73d36402184d9d21b7d59edfbe5795d363e8"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-18 18:14:28.000000000","tz":-480},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/07e3870d422b7688d15733afa52f873f7e8833cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/07e3870d422b7688d15733afa52f873f7e8833cb"}]},"branch":"refs/heads/master"},"e41944a7e1b70ef902e9999e68fe3222f5f41ada":{"kind":"REWORK","_number":33,"created":"2021-03-24 05:33:18.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/33"}}},"commit":{"parents":[{"commit":"6c54eab5b5d8fd5300cd5d62e23e6cadc1641636","subject":"Make /healthcheck cache results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c54eab5b5d8fd5300cd5d62e23e6cadc1641636"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2021-03-24 05:31:09.000000000","tz":-420},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e41944a7e1b70ef902e9999e68fe3222f5f41ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e41944a7e1b70ef902e9999e68fe3222f5f41ada"}]},"branch":"refs/heads/master"},"8f75599740bb36cf96619245a45a3ce52a68e6f0":{"kind":"REWORK","_number":34,"created":"2022-01-27 10:49:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/558962/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/34"}}},"commit":{"parents":[{"commit":"c766d891e512d28ab6056126a977d467ce15e317","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c766d891e512d28ab6056126a977d467ce15e317"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-01-27 10:49:03.000000000","tz":120},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f75599740bb36cf96619245a45a3ce52a68e6f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f75599740bb36cf96619245a45a3ce52a68e6f0"}]},"branch":"refs/heads/master"},"5248b42a54e9724f42202772b8a7eef2e058a27d":{"kind":"REWORK","_number":35,"created":"2022-01-27 11:08:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/558962/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/35"}}},"commit":{"parents":[{"commit":"c766d891e512d28ab6056126a977d467ce15e317","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c766d891e512d28ab6056126a977d467ce15e317"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-27 11:08:13.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5248b42a54e9724f42202772b8a7eef2e058a27d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5248b42a54e9724f42202772b8a7eef2e058a27d"}]},"branch":"refs/heads/master"},"3dd11bc64a1dc89aab39885b467a5d26cd477c1d":{"kind":"REWORK","_number":36,"created":"2022-01-27 11:09:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/558962/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/36"}}},"commit":{"parents":[{"commit":"c766d891e512d28ab6056126a977d467ce15e317","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c766d891e512d28ab6056126a977d467ce15e317"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-27 11:09:34.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3dd11bc64a1dc89aab39885b467a5d26cd477c1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3dd11bc64a1dc89aab39885b467a5d26cd477c1d"}]},"branch":"refs/heads/master"},"6a2ab25c7ff5ad0f66fec3d61b36d1cd07196f19":{"kind":"REWORK","_number":37,"created":"2022-01-27 14:16:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/558962/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/37"}}},"commit":{"parents":[{"commit":"c766d891e512d28ab6056126a977d467ce15e317","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c766d891e512d28ab6056126a977d467ce15e317"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-01-27 14:16:05.000000000","tz":120},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a2ab25c7ff5ad0f66fec3d61b36d1cd07196f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a2ab25c7ff5ad0f66fec3d61b36d1cd07196f19"}]},"branch":"refs/heads/master"},"b29cedb6e722a06cc67fc7f511f994a7d6dcb8d3":{"kind":"REWORK","_number":38,"created":"2022-01-27 14:46:42.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/558962/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/38"}}},"commit":{"parents":[{"commit":"c766d891e512d28ab6056126a977d467ce15e317","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c766d891e512d28ab6056126a977d467ce15e317"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2022-01-27 14:46:30.000000000","tz":-480},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b29cedb6e722a06cc67fc7f511f994a7d6dcb8d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b29cedb6e722a06cc67fc7f511f994a7d6dcb8d3"}]},"branch":"refs/heads/master"},"7c029944985eab78723a2eef95a8d307f83cb55e":{"kind":"REWORK","_number":39,"created":"2022-01-27 15:44:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/558962/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/39"}}},"commit":{"parents":[{"commit":"c766d891e512d28ab6056126a977d467ce15e317","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c766d891e512d28ab6056126a977d467ce15e317"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-01-27 15:44:33.000000000","tz":120},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c029944985eab78723a2eef95a8d307f83cb55e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c029944985eab78723a2eef95a8d307f83cb55e"}]},"branch":"refs/heads/master"},"1deb334b2ac264c517783f917d9409c08995b613":{"kind":"REWORK","_number":40,"created":"2022-01-27 16:43:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/558962/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/40"}}},"commit":{"parents":[{"commit":"c766d891e512d28ab6056126a977d467ce15e317","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c766d891e512d28ab6056126a977d467ce15e317"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-27 16:43:36.000000000","tz":0},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1deb334b2ac264c517783f917d9409c08995b613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1deb334b2ac264c517783f917d9409c08995b613"}]},"branch":"refs/heads/master"},"a50426c502c2d1adad0e4cded43811316ced8c99":{"kind":"REWORK","_number":41,"created":"2022-01-27 16:47:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/558962/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/41"}}},"commit":{"parents":[{"commit":"c766d891e512d28ab6056126a977d467ce15e317","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c766d891e512d28ab6056126a977d467ce15e317"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-01-27 16:46:47.000000000","tz":120},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a50426c502c2d1adad0e4cded43811316ced8c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a50426c502c2d1adad0e4cded43811316ced8c99"}]},"branch":"refs/heads/master"},"2d88db56d7729f70d3fd9807657e7c49016a11e8":{"kind":"REWORK","_number":42,"created":"2022-12-13 08:42:13.000000000","uploader":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"ref":"refs/changes/62/558962/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/42"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Denys Mischenko","email":"denis@mischenko.org.ua","date":"2022-12-13 08:42:01.000000000","tz":60},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nIf AZs provided as a list either in flavor or in nova section of the\nconfig, Octavia will try to create amphoras in different availability\nzones. Master amphora will be calculated based on the hash from\nloadbalancer id.\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d88db56d7729f70d3fd9807657e7c49016a11e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d88db56d7729f70d3fd9807657e7c49016a11e8"}]},"branch":"refs/heads/master"},"632d9d2959ddd2ac507045e9eb9bf820cfaac16f":{"kind":"REWORK","_number":43,"created":"2022-12-16 12:31:34.000000000","uploader":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"ref":"refs/changes/62/558962/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/43"}}},"commit":{"parents":[{"commit":"8d27bdb5462474dafc934d164f3a299bfca8dd89","subject":"Merge \"Filter ComputeWaitTimeoutException when jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d27bdb5462474dafc934d164f3a299bfca8dd89"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Denys Mischenko","email":"denis@mischenko.org.ua","date":"2022-12-16 11:44:51.000000000","tz":60},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nIf AZs provided as a list either in flavor or in nova section of the\nconfig, Octavia will try to create amphoras in different availability\nzones. Master amphora will be calculated based on the hash from\nloadbalancer id.\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/632d9d2959ddd2ac507045e9eb9bf820cfaac16f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/632d9d2959ddd2ac507045e9eb9bf820cfaac16f"}]},"branch":"refs/heads/master"},"79b90722b277dac051a9a8d96be84d70d0cdfeaa":{"kind":"REWORK","_number":44,"created":"2022-12-16 13:44:09.000000000","uploader":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"ref":"refs/changes/62/558962/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/44"}}},"commit":{"parents":[{"commit":"8d27bdb5462474dafc934d164f3a299bfca8dd89","subject":"Merge \"Filter ComputeWaitTimeoutException when jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d27bdb5462474dafc934d164f3a299bfca8dd89"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Denys Mischenko","email":"denis@mischenko.org.ua","date":"2022-12-16 13:42:04.000000000","tz":60},"subject":"Experimental multi-az support","message":"Experimental multi-az support\n\nIf AZs provided as a list either in flavor or in nova section of the\nconfig, Octavia will try to create amphoras in different availability\nzones. Master amphora will be calculated based on the hash from\nloadbalancer id.\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79b90722b277dac051a9a8d96be84d70d0cdfeaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79b90722b277dac051a9a8d96be84d70d0cdfeaa"}]},"branch":"refs/heads/master"},"2aba8af6e25b98c77db5b5cc16db133f9fe3c986":{"kind":"REWORK","_number":45,"created":"2023-01-12 09:13:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/558962/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/45"}}},"commit":{"parents":[{"commit":"c455a261883ddd2becb0282d801f9c23eca31e92","subject":"Merge \"Fix listener creation with fully-populated API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c455a261883ddd2becb0282d801f9c23eca31e92"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2023-01-12 09:12:50.000000000","tz":60},"subject":"Implement amphora multi-az support","message":"Implement amphora multi-az support\n\nIf AZs provided as a list either in flavor or in nova section of the\nconfig, Octavia will try to create amphoras in different availability\nzones. Master amphora will be calculated based on the hash from\nloadbalancer id.\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aba8af6e25b98c77db5b5cc16db133f9fe3c986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aba8af6e25b98c77db5b5cc16db133f9fe3c986"}]},"branch":"refs/heads/master"},"a2f34bb311b8950c784c35ed61d407b475630140":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2023-02-18 13:54:06.000000000","uploader":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"ref":"refs/changes/62/558962/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/46"}}},"commit":{"parents":[{"commit":"b4b6e07fe07f2b1103ebe733db3f76f05d7acf0c","subject":"Merge \"Fix typo in lvs-masquerade.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4b6e07fe07f2b1103ebe733db3f76f05d7acf0c"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Matthew Heler","email":"matthew.heler@hotmail.com","date":"2023-02-18 13:54:06.000000000","tz":0},"subject":"Implement amphora multi-az support","message":"Implement amphora multi-az support\n\nIf AZs provided as a list either in flavor or in nova section of the\nconfig, Octavia will try to create amphoras in different availability\nzones. Master amphora will be calculated based on the hash from\nloadbalancer id.\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2f34bb311b8950c784c35ed61d407b475630140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2f34bb311b8950c784c35ed61d407b475630140"}]},"branch":"refs/heads/master","description":"Rebase"},"f5dccefea72df7c016a7b389d4d31c5ba63d10cb":{"kind":"REWORK","_number":47,"created":"2024-03-12 03:48:33.000000000","uploader":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"ref":"refs/changes/62/558962/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/47"}}},"commit":{"parents":[{"commit":"ffc6f83f07b088fbd8979a5610fa371ff5c6467e","subject":"Merge \"Add --wait to Octavia cookbook\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffc6f83f07b088fbd8979a5610fa371ff5c6467e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Matthew N Heler","email":"matthew.heler@hotmail.com","date":"2024-03-12 03:47:36.000000000","tz":-300},"subject":"Implement amphora multi-az support","message":"Implement amphora multi-az support\n\nIf AZs provided as a list either in flavor or in nova section of the\nconfig, Octavia will try to create amphoras in different availability\nzones. Master amphora will be calculated based on the hash from\nloadbalancer id.\n\nOptionally, nova can to be patched to schedule to multiple AZs, which\nimproves the reliability of scheduling. This must be enabled with the\nconfig option `accepts_multiple_az`.\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5dccefea72df7c016a7b389d4d31c5ba63d10cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5dccefea72df7c016a7b389d4d31c5ba63d10cb"}]},"branch":"refs/heads/master"},"7e33eb5b576254217a1e025adbf5b38926cb9bdd":{"kind":"REWORK","_number":48,"created":"2024-03-12 13:51:24.000000000","uploader":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"ref":"refs/changes/62/558962/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/48"}}},"commit":{"parents":[{"commit":"ffc6f83f07b088fbd8979a5610fa371ff5c6467e","subject":"Merge \"Add --wait to Octavia cookbook\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffc6f83f07b088fbd8979a5610fa371ff5c6467e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-04-04 23:42:07.000000000","tz":540},"committer":{"name":"Matthew N Heler","email":"matthew.heler@hotmail.com","date":"2024-03-12 13:50:45.000000000","tz":-300},"subject":"Implement amphora multi-az support","message":"Implement amphora multi-az support\n\nIf AZs provided as a list either in flavor or in nova section of the\nconfig, Octavia will try to create amphoras in different availability\nzones. Master amphora will be calculated based on the hash from\nloadbalancer id.\n\nIf using the default network driver (allowed_address_pairs), all AZs\nmust use the same L2 network or failovers will not work.\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e33eb5b576254217a1e025adbf5b38926cb9bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e33eb5b576254217a1e025adbf5b38926cb9bdd"}]},"branch":"refs/heads/master"},"7df248aa4f2a9e41b2d1c7c9697faa6b512c7c9e":{"kind":"REWORK","_number":49,"created":"2024-07-12 12:57:20.000000000","uploader":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"ref":"refs/changes/62/558962/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/558962/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/558962/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/558962/49"}}},"commit":{"parents":[{"commit":"7974e755e18bed376ea5b1bce2a927eeb34494b2","subject":"Merge \"jobboard: Support authentication and SSL for Redis Sentinel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7974e755e18bed376ea5b1bce2a927eeb34494b2"}]}],"author":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2024-07-12 12:44:51.000000000","tz":0},"committer":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2024-07-12 12:44:51.000000000","tz":0},"subject":"Implement amphora multi-az support","message":"Implement amphora multi-az support\n\nThis change adds support to define multiple compute availability zones for the\namphora creation. Multiple ways to define a list of compute zones available:\n* Define a list of Octavia availability zones in flavor.\n* Define list of compute zones in octavia flavor\n* Define list of compute zones in nova section of the config file\n\nOnce we create a load balancer with multiple compute availability zones\n(using either of the above ways), we calculate a zone for the initial master\namphora based on the loadbalancer_id hash. Then we pick another zone for the\nfailover amphora. Once we execute a failover, in the compute task we find\namphora which is going to be deleted using a status \"pending delete\" then put\na new amphora in exactly the same availability zone. This lets us preserve the\nlocation of the amphoras.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nCo-Authored-By: Denys Mishchenko \u003cdenis@mischenko.org.ua\u003e\n\nChange-Id: I1ebd42810587a75d82d54312ffc7569255aea145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7df248aa4f2a9e41b2d1c7c9697faa6b512c7c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7df248aa4f2a9e41b2d1c7c9697faa6b512c7c9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
