)]}'
{"id":"openstack%2Foctavia~580778","triplet_id":"openstack%2Foctavia~master~If5fbfc34ff0e91608f402811aa2b78c9cbcb00ec","project":"openstack/octavia","branch":"master","topic":"rm-extra-ints","hashtags":[],"change_id":"If5fbfc34ff0e91608f402811aa2b78c9cbcb00ec","subject":"Remove unnecessary interface configurations","status":"MERGED","created":"2018-07-06 23:02:57.000000000","updated":"2019-11-27 20:34:30.000000000","submitted":"2019-11-27 20:32:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"580778-1574886738213-a496186d","meta_rev_id":"8a7013016734fa75c5e57a3460de8f012b1b3857","_number":580778,"virtual_id_number":580778,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"date":"2019-11-27 15:52:25.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 20:32:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-11-27 12:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-11-22 08:32:23.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-11-21 15:14:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-11-26 12:55:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-11-27 12:18:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 05:59:01.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-11-21 15:14:52.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"},{"updated":"2019-11-22 08:32:23.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-26 12:55:52.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-11-27 12:18:24.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-27 20:32:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e278883a59d989f7cc65bffe278f77a492236d86","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2018-07-06 23:02:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edec1b95ea9a2cdfed239bb9aff05fc5ec2949e8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2018-07-06 23:03:21.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c6b70b88961672a5b6a3e75039376bd7d5350fcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-07 00:47:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/78/580778/1/check/openstack-tox-pep8/05eee30/ : SUCCESS in 7m 32s\n- openstack-tox-py27 http://logs.openstack.org/78/580778/1/check/openstack-tox-py27/62b4515/ : SUCCESS in 5m 05s\n- openstack-tox-py35 http://logs.openstack.org/78/580778/1/check/openstack-tox-py35/8e59511/ : SUCCESS in 5m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/78/580778/1/check/build-openstack-sphinx-docs/5cba9df/html/ : SUCCESS in 6m 01s\n- openstack-tox-cover http://logs.openstack.org/78/580778/1/check/openstack-tox-cover/b0048ea/cover/ : SUCCESS in 11m 27s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/78/580778/1/check/neutron-lbaasv2-dsvm-api/3694afd/ : SUCCESS in 1h 07m 30s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/78/580778/1/check/neutron-lbaasv2-dsvm-py3x-api/450b03e/ : SUCCESS in 1h 09m 44s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/78/580778/1/check/neutron-lbaasv2-dsvm-scenario/6a5fc1f/ : SUCCESS in 53m 47s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/78/580778/1/check/neutron-lbaasv2-dsvm-py3x-scenario/f7609ec/ : SUCCESS in 1h 21m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/78/580778/1/check/openstack-tox-lower-constraints/4b60a39/ : SUCCESS in 8m 03s\n- openstack-tox-functional http://logs.openstack.org/78/580778/1/check/openstack-tox-functional/88fe888/ : SUCCESS in 6m 57s\n- openstack-tox-functional-py35 http://logs.openstack.org/78/580778/1/check/openstack-tox-functional-py35/7b0386f/ : SUCCESS in 7m 16s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/78/580778/1/check/octavia-v1-dsvm-scenario/a0ceb4c/ : SUCCESS in 44m 51s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/78/580778/1/check/octavia-v1-dsvm-py3x-scenario/9beb970/ : SUCCESS in 1h 01m 28s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/78/580778/1/check/octavia-v1-dsvm-scenario-multinode/6e0a112/ : FAILURE in 1h 16m 13s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/78/580778/1/check/octavia-v1-dsvm-py3x-scenario-multinode/e901fc8/ : FAILURE in 1h 28m 10s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/78/580778/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/da14fc6/ : SUCCESS in 50m 27s (non-voting)\n- octavia-v1-dsvm-scenario-ubuntu.bionic http://logs.openstack.org/78/580778/1/check/octavia-v1-dsvm-scenario-ubuntu.bionic/ee174c5/ : SUCCESS in 44m 04s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/78/580778/1/check/octavia-v2-dsvm-noop-api/b4416f9/ : SUCCESS in 39m 58s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/78/580778/1/check/octavia-v2-dsvm-noop-py35-api/8f48e9a/ : SUCCESS in 43m 55s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/78/580778/1/check/octavia-v2-dsvm-scenario/2338078/ : SUCCESS in 1h 38m 33s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/78/580778/1/check/octavia-v2-dsvm-py35-scenario/24532ff/ : SUCCESS in 1h 30m 23s (non-voting)\n- octavia-grenade http://logs.openstack.org/78/580778/1/check/octavia-grenade/faea52c/ : SUCCESS in 1h 01m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9b5774c1138252f1017f532955300cc52c28fb74","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2018-07-08 12:35:38.000000000","message":"Patch Set 1:\n\nMichael, is there a need for a CentOS equivalent for this?","accounts_in_message":[],"_revision_number":1},{"id":"61a2a1d5e7e4c4775a44cde7d6e4e74fd10b451b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2018-07-08 21:36:00.000000000","message":"Patch Set 1:\n\nMaybe, I would need to look at the centos images.\nThis is partly why I set WIP on this patch.","accounts_in_message":[],"_revision_number":1},{"id":"e3e01924d2d94784890fa8d5e60d3fea33172f62","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-05 14:54:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a57bdfc14f47936881cb53c9932c5396d777b333","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 21:38:48.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"306a99cbdc354c1030ddc6e2bf2ffde1a1d2d75b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 21:39:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"99ce1f4d77516303fc5668854936b1cd68dd7e61","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 21:40:42.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5e0e447bb521d04bd1dc91178bea9982790a9b52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 00:23:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/78/580778/4/check/openstack-tox-cover/3300b1b/cover/ : SUCCESS in 8m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/78/580778/4/check/openstack-tox-lower-constraints/de9214f/ : SUCCESS in 10m 07s\n- openstack-tox-pep8 http://logs.openstack.org/78/580778/4/check/openstack-tox-pep8/311b6b0/ : SUCCESS in 18m 08s\n- openstack-tox-py27 http://logs.openstack.org/78/580778/4/check/openstack-tox-py27/f938df3/ : SUCCESS in 5m 28s\n- openstack-tox-py36 http://logs.openstack.org/78/580778/4/check/openstack-tox-py36/aee8a10/ : SUCCESS in 5m 37s\n- openstack-tox-py37 http://logs.openstack.org/78/580778/4/check/openstack-tox-py37/f8bac40/ : SUCCESS in 6m 09s\n- openstack-tox-docs http://logs.openstack.org/78/580778/4/check/openstack-tox-docs/a84985a/html/ : SUCCESS in 7m 03s\n- openstack-tox-functional http://logs.openstack.org/78/580778/4/check/openstack-tox-functional/333ba2e/ : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 http://logs.openstack.org/78/580778/4/check/openstack-tox-functional-py36/5ec1cc4/ : SUCCESS in 6m 57s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/78/580778/4/check/octavia-v2-dsvm-noop-api/b019e5e/ : SUCCESS in 46m 19s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/78/580778/4/check/octavia-v2-dsvm-noop-py2-api/7a7d324/ : SUCCESS in 48m 13s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/78/580778/4/check/octavia-v2-dsvm-scenario/c757322/ : SUCCESS in 1h 31m 08s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/78/580778/4/check/octavia-v2-dsvm-py2-scenario/1b5aa22/ : SUCCESS in 1h 39m 37s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/78/580778/4/check/octavia-v2-dsvm-py2-scenario-centos-7/ba2a575/ : FAILURE in 2h 08m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/78/580778/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/1cced07/ : SUCCESS in 1h 52m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/78/580778/4/check/octavia-v2-dsvm-scenario-two-node/98f7795/ : SUCCESS in 1h 56m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/78/580778/4/check/octavia-v2-dsvm-py2-scenario-two-node/e388193/ : SUCCESS in 1h 46m 41s (non-voting)\n- octavia-grenade http://logs.openstack.org/78/580778/4/check/octavia-grenade/94ab437/ : SUCCESS in 1h 21m 22s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/78/580778/4/check/octavia-v2-dsvm-tls-barbican/f6d32a3/ : SUCCESS in 54m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/78/580778/4/check/octavia-v2-dsvm-scenario-amphora-v2/c7f050d/ : TIMED_OUT in 2h 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"47d90550ee0100a2f72637021f47f24fd3fc717d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-21 01:55:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"572d8e6a70da9837b3eea769e5a7eb746dedf898","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-06-27 05:59:01.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5c879d42f314e61e4c5e627a2c9084e0448f770f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-22 20:33:02.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"33b0468fdfa4efec4278c9e0fd22c346c41e508c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-20 23:59:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b5e5f1353e2b55cc8581b090acd4113a547d0afe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-21 00:02:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b094b1fa08c3f229dbf8a31c919117a5192195cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 02:09:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/394b235a5eb54ff9b48f211d61cf1fbf : SUCCESS in 8m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c55361530dfa4febb02dc1d5553b0a8b : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07e202b1d7f64de6aea0d9415fc1828a : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6075627d7c5045ef917ee75f8ad2d8e7 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7af4bef1fedc4067b7a43bc6f95365d5 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a9d4204b0ef4a52a352df763bd01357 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eff8aff1b2fa44f9a9ed36884c37d9ff : SUCCESS in 12m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/26dab986b36944c88bfbd13db0d0f77d : SUCCESS in 4m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/894646a1817c4690aee360410d9de887 : SUCCESS in 6m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5f5946fadbab4b31b1bee4c33359328b : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/314235d7e3654a9f8f6a153963472d3c : SUCCESS in 5m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/540cef7df5eb404981a82a90cadb26f2 : SUCCESS in 46m 26s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ebb65e86d19c4ce2b65880e4c36ad225 : SUCCESS in 46m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9370f5d04c3d468e9bab7a6c9443c66a : SUCCESS in 1h 32m 15s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/09579276c2c14f438e482a5be4112d8a : SUCCESS in 1h 41m 34s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b39484c0d007443bb2e8ae1df0febf60 : SUCCESS in 2h 03m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/580173439a74429ea01ebda7df645e51 : FAILURE in 37m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a672eed449f540b08e2f665fd24707f6 : SUCCESS in 1h 54m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/599c0ec2a927424c87f89a03934f44d2 : SUCCESS in 1h 20m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3701d7db8b834b7cbd60b96d72a3375d : SUCCESS in 52m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e1c95cbcef674a9eacfd4576ae0337e5 : SUCCESS in 1h 34m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e88831c19a2d42bdb5e4ad1a7a7fc1bb : SUCCESS in 1h 06m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d95c5918fa574fae8ef959a75fafd49e : SUCCESS in 1h 04m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/24013374e42446a0bb2c441f3592aa9c : FAILURE in 1h 26m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"020de38794376fc4c318d74c84bcc59beeda3461","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-11-21 15:14:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a7e57d715c190bc60df8f1deb84a2c8b24b01804","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-22 08:32:23.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"900162278270e0a3ad3845c3ffff595634bee31b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-26 12:55:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"4523d2465050f8b7217ffc7ed6f8fe460cee315b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 12:18:24.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b61630fc49447e2e023641f0f26e35ff27898a3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 12:18:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"765354848ff741fe06a929ff86f13497eae4ffc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:09:07.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7472dcf844a4cd9b24a6a0e52ff00ff : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff1b830e65447b28457167403e4d8a9 : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b50ea842a159407794d572bc85a7c35d : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6b51b00793247f09f17113076deb407 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be024d54d81f47b5a266a59c541bf6e5 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe83b7377594ff98bffa2ac14956882 : SUCCESS in 11m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fe94eb5c0733452fa0faa9bec0d5b90a : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0f9edd220eb646a3a680428d64b9ea03 : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d0e1e5109be42d58647e25f4813528f : CANCELED in 38m 44s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0788cd94520d48079424cc48ea67ac09 : CANCELED in 42m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/235056b91d4e4eb3b225d2b50e813a58 : FAILURE in 38m 34s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6149bfcbaacd437f8dc402743ff48a32 : CANCELED in 39m 03s","accounts_in_message":[],"_revision_number":5},{"id":"280bc7152a889e214bedd50c6d1343db049aa82d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 15:48:42.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"02bd04acd344a2df622a358fa6375a36b884a3bd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-27 15:52:25.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"6f2e8ed2ec4c230d59a749cd437f9fd044a3552f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:25:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a2a015c25184f1888e8a43109d30158 : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82096fa9ad0a435893c41c38bbdbf554 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c28cba7b973a4cb1b79cd4138c7500cb : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cdfe01a826b84ce99dc97dbf35d49f95 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/364d15907ede4007bcb468d5e6adc7ec : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0ea8785d3b7468b8a3633d163614bfb : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/640b10f4063d498a885e48d58c3de5f9 : SUCCESS in 10m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3420c6a03ef24b62bc2dad5535e5b1c1 : SUCCESS in 5m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ad6a075854b642e1bdd1253ddaea7edf : SUCCESS in 6m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/69cb098c8aa24ca9bfc20332a25848fe : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/004f713345ec41c78145fc89c6370518 : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a01b4675b9ae437895ce8c7b2b8222c9 : SUCCESS in 43m 31s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5ececb4a12e940589ba2726b391aba52 : SUCCESS in 47m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce261a0bbe1b4a5380a70849fa6c931c : SUCCESS in 1h 30m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e684bbf5831e4ae3a272f3c91dc4c90e : SUCCESS in 1h 33m 59s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4929272ce3394675ad04a5a56cbe12ba : SUCCESS in 1h 50m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9e4a31aee64146dbb241c43382567ffb : FAILURE in 36m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4425d4e40ea64fae8dd24182ef703a15 : SUCCESS in 2h 26m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bfc26f1f9ae24e2b99f9b3df682ddeb0 : FAILURE in 1h 01m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7eb0aa82339a4ea48a3804cd3343ec30 : SUCCESS in 48m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9ee99751579c4a22bd08356e959d1b86 : SUCCESS in 1h 35m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d0174e7e35848efa89d05f89d87f3f1 : FAILURE in 1h 53m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/119ba0aab8b649fc85e6f368ea7bc35d : SUCCESS in 1h 27m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f20b3d841a7f453b85be193d269c6912 : FAILURE in 1h 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d79826c00674ed5ec105a5836f2e0aacaddac39a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:25:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f6daacea4c9b86c7f6c00ef8f286fb295caa333a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 20:32:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"e95297ae9e23f1bd0d67ee3b65322301a5aa6f48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 20:32:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efc40cfd30d64c62adcdae54fdf1527e : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55116b5bb5a44f18b7bc18f0810e5b57 : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c202cfb550764ec1839806ba35431d70 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecfe6bc609f44356aeae9613a0f6903e : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8144149d84394992b75af246aa3467c0 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a8842a2a88e4c2496d15ee84c6fac8b : SUCCESS in 10m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/577bb79596284be1b733572ffae3564f : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d52bb81a847403a8bb9c72a7c222b17 : SUCCESS in 6m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da483555ccb544baa3b31495899e358b : SUCCESS in 42m 17s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d99af4020f844e73a5efffd47f98b968 : SUCCESS in 43m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d53447a68384bfba4eeef65064ab961 : SUCCESS in 1h 37m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/17d8624e375448d5a99804fb77e6cc91 : SUCCESS in 1h 59m 11s","accounts_in_message":[],"_revision_number":5},{"id":"0f3e6653b4372ca63a2cce0d09e72cef7afea633","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 20:34:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/653900520c194988b828d5fa57062fc2 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7ad61bb451ecdc738d6057cf1b2f736a33630774","revisions":{"3f0cb61b3d9cf43615033ea889bd2e66dd41ef3f":{"kind":"REWORK","_number":1,"created":"2018-07-06 23:02:57.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/78/580778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/580778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/580778/1"}}},"commit":{"parents":[{"commit":"464d8f9371f62e33bb871167519d6ba0333a318b","subject":"Introduce ipvsadm to the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/464d8f9371f62e33bb871167519d6ba0333a318b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2018-07-06 22:59:26.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2018-07-06 23:02:56.000000000","tz":-420},"subject":"Remove unnecessary interface configurations","message":"Remove unnecessary interface configurations\n\nThe image building process adds some unnecessary network interface\nconfiguration files to /etc/network/interfaces.d. This element cleans those\nout as they are not needed because cloud-init will create the required\nconfiguration file.\n\nChange-Id: If5fbfc34ff0e91608f402811aa2b78c9cbcb00ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f0cb61b3d9cf43615033ea889bd2e66dd41ef3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f0cb61b3d9cf43615033ea889bd2e66dd41ef3f"}]},"branch":"refs/heads/master"},"7eae357c12053e00a0a7ea84a1523b9c4fe46aa2":{"kind":"REWORK","_number":2,"created":"2019-06-20 21:38:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/78/580778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/580778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/580778/2"}}},"commit":{"parents":[{"commit":"464d8f9371f62e33bb871167519d6ba0333a318b","subject":"Introduce ipvsadm to the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/464d8f9371f62e33bb871167519d6ba0333a318b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2018-07-06 22:59:26.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-20 21:38:27.000000000","tz":0},"subject":"Remove unnecessary interface configurations","message":"Remove unnecessary interface configurations\n\nThe image building process adds some unnecessary network interface\nconfiguration files to /etc/network/interfaces.d. This element cleans those\nout as they are not needed because cloud-init will create the required\nconfiguration file.\n\nChange-Id: If5fbfc34ff0e91608f402811aa2b78c9cbcb00ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7eae357c12053e00a0a7ea84a1523b9c4fe46aa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7eae357c12053e00a0a7ea84a1523b9c4fe46aa2"}]},"branch":"refs/heads/master"},"098c7f94bef639f1d586ad03581efc1bdce5a05d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-20 21:39:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/78/580778/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/580778/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/580778/3"}}},"commit":{"parents":[{"commit":"a485934de3e5226a1604c649a6210efb79987617","subject":"Merge \"Clarify that the certificate guide should be used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a485934de3e5226a1604c649a6210efb79987617"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2018-07-06 22:59:26.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-20 21:39:10.000000000","tz":0},"subject":"Remove unnecessary interface configurations","message":"Remove unnecessary interface configurations\n\nThe image building process adds some unnecessary network interface\nconfiguration files to /etc/network/interfaces.d. This element cleans those\nout as they are not needed because cloud-init will create the required\nconfiguration file.\n\nChange-Id: If5fbfc34ff0e91608f402811aa2b78c9cbcb00ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/098c7f94bef639f1d586ad03581efc1bdce5a05d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/098c7f94bef639f1d586ad03581efc1bdce5a05d"}]},"branch":"refs/heads/master"},"3ae796149f7fdbaa2aea4db39d7e3839c7d17858":{"kind":"REWORK","_number":4,"created":"2019-06-20 21:40:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/78/580778/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/580778/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/580778/4"}}},"commit":{"parents":[{"commit":"a485934de3e5226a1604c649a6210efb79987617","subject":"Merge \"Clarify that the certificate guide should be used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a485934de3e5226a1604c649a6210efb79987617"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2018-07-06 22:59:26.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-20 21:40:39.000000000","tz":0},"subject":"Remove unnecessary interface configurations","message":"Remove unnecessary interface configurations\n\nThe image building process adds some unnecessary network interface\nconfiguration files to /etc/network/interfaces.d. This element cleans those\nout as they are not needed because cloud-init will create the required\nconfiguration file.\n\nChange-Id: If5fbfc34ff0e91608f402811aa2b78c9cbcb00ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ae796149f7fdbaa2aea4db39d7e3839c7d17858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ae796149f7fdbaa2aea4db39d7e3839c7d17858"}]},"branch":"refs/heads/master"},"7ad61bb451ecdc738d6057cf1b2f736a33630774":{"kind":"REWORK","_number":5,"created":"2019-11-21 00:02:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/78/580778/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/580778/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/580778/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/580778/5"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2018-07-06 22:59:26.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-21 00:02:42.000000000","tz":-480},"subject":"Remove unnecessary interface configurations","message":"Remove unnecessary interface configurations\n\nThe image building process adds some unnecessary network interface\nconfiguration files to /etc/network/interfaces.d. This element cleans those\nout as they are not needed because cloud-init will create the required\nconfiguration file.\n\nChange-Id: If5fbfc34ff0e91608f402811aa2b78c9cbcb00ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ad61bb451ecdc738d6057cf1b2f736a33630774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ad61bb451ecdc738d6057cf1b2f736a33630774"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
