)]}'
{"id":"openstack%2Foctavia~596665","triplet_id":"openstack%2Foctavia~master~Idd313f20c183ade94588844ea63d0e18a63a009c","project":"openstack/octavia","branch":"master","topic":"refactor_quota_code","hashtags":[],"change_id":"Idd313f20c183ade94588844ea63d0e18a63a009c","subject":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","status":"ABANDONED","created":"2018-08-27 08:38:32.000000000","updated":"2022-03-17 01:46:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee594a4dd541ac509a317a279239508928ac1e9b","_number":596665,"virtual_id_number":596665,"owner":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-11-18 20:12:34.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-14 21:38:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-11 06:01:11.000000000","updated_by":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"reviewer":{"_account_id":15309,"name":"zhaobo","email":"bzhaojyathousandy@gmail.com","username":"ZhaoBo"},"state":"REVIEWER"},{"updated":"2020-01-14 21:38:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 20:12:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"6da4e00169d452834bf5d0869914b13b7283a4a7","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2018-08-27 08:38:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ceb2989ad509cebbbacb5b2aa9f3251a0cccb788","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2018-08-27 08:44:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fdf0c243522f251adcf684d4ba67c35a7a5e9c81","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2018-08-27 09:35:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fe4a4c6ab5407846692c304339ef3e47d09eca50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-27 10:49:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/65/596665/3/check/openstack-tox-pep8/59752ef/ : FAILURE in 6m 12s\n- openstack-tox-py27 http://logs.openstack.org/65/596665/3/check/openstack-tox-py27/bfc129d/ : SUCCESS in 5m 13s\n- openstack-tox-py35 http://logs.openstack.org/65/596665/3/check/openstack-tox-py35/0edede9/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/65/596665/3/check/openstack-tox-docs/e491297/html/ : SUCCESS in 6m 31s\n- openstack-tox-cover http://logs.openstack.org/65/596665/3/check/openstack-tox-cover/6b19010/ : FAILURE in 8m 23s\n- openstack-tox-py36 http://logs.openstack.org/65/596665/3/check/openstack-tox-py36/33b3c77/ : SUCCESS in 6m 35s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/65/596665/3/check/neutron-lbaasv2-dsvm-api/37eb4f6/ : FAILURE in 47m 12s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/65/596665/3/check/neutron-lbaasv2-dsvm-py3x-api/8944d9b/ : FAILURE in 53m 23s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/65/596665/3/check/neutron-lbaasv2-dsvm-scenario/c0dd81c/ : FAILURE in 43m 17s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/65/596665/3/check/neutron-lbaasv2-dsvm-py3x-scenario/1ee281b/ : FAILURE in 42m 42s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/65/596665/3/check/openstack-tox-lower-constraints/ed5b941/ : FAILURE in 8m 39s\n- openstack-tox-functional http://logs.openstack.org/65/596665/3/check/openstack-tox-functional/7908316/ : FAILURE in 7m 35s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/596665/3/check/openstack-tox-functional-py35/7e42ba2/ : FAILURE in 7m 35s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/65/596665/3/check/octavia-v1-dsvm-scenario/91b28bb/ : FAILURE in 42m 13s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/65/596665/3/check/octavia-v1-dsvm-py3x-scenario/68fbcd8/ : FAILURE in 44m 29s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/65/596665/3/check/octavia-v1-dsvm-scenario-multinode/0e0f767/ : FAILURE in 1h 04m 15s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/65/596665/3/check/octavia-v1-dsvm-py3x-scenario-multinode/338cb05/ : FAILURE in 53m 24s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/65/596665/3/check/octavia-v1-dsvm-scenario-kvm-centos.7/db0e43d/ : FAILURE in 40m 32s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/65/596665/3/check/octavia-v2-dsvm-noop-api/5cae06b/ : FAILURE in 25m 30s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/65/596665/3/check/octavia-v2-dsvm-noop-py35-api/74c27b8/ : FAILURE in 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/65/596665/3/check/octavia-v2-dsvm-scenario/e6c27b6/ : FAILURE in 41m 59s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/65/596665/3/check/octavia-v2-dsvm-py35-scenario/cb38af1/ : FAILURE in 38m 41s (non-voting)\n- octavia-grenade http://logs.openstack.org/65/596665/3/check/octavia-grenade/f170adb/ : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":3},{"id":"96ac4a4e7ac566e4b738ce2038f1464007c1cc36","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2018-08-27 11:42:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a873f41eea09115ca5ac604001dd276f16a0fac6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-27 13:19:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/65/596665/4/check/openstack-tox-pep8/884738c/ : SUCCESS in 7m 40s\n- openstack-tox-py27 http://logs.openstack.org/65/596665/4/check/openstack-tox-py27/2cd97ec/ : SUCCESS in 8m 02s\n- openstack-tox-py35 http://logs.openstack.org/65/596665/4/check/openstack-tox-py35/ab81a49/ : SUCCESS in 4m 57s\n- openstack-tox-docs http://logs.openstack.org/65/596665/4/check/openstack-tox-docs/7c8195b/html/ : SUCCESS in 6m 32s\n- openstack-tox-cover http://logs.openstack.org/65/596665/4/check/openstack-tox-cover/8fddf6e/ : FAILURE in 8m 31s\n- openstack-tox-py36 http://logs.openstack.org/65/596665/4/check/openstack-tox-py36/3e1801e/ : SUCCESS in 6m 47s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/65/596665/4/check/neutron-lbaasv2-dsvm-api/d4f2e15/ : SUCCESS in 1h 11m 43s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/65/596665/4/check/neutron-lbaasv2-dsvm-py3x-api/e0c510d/ : SUCCESS in 1h 12m 04s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/65/596665/4/check/neutron-lbaasv2-dsvm-scenario/d973607/ : SUCCESS in 1h 27m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/65/596665/4/check/neutron-lbaasv2-dsvm-py3x-scenario/4eb475e/ : SUCCESS in 1h 20m 54s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/65/596665/4/check/openstack-tox-lower-constraints/4df7ec6/ : FAILURE in 8m 03s\n- openstack-tox-functional http://logs.openstack.org/65/596665/4/check/openstack-tox-functional/650b839/ : FAILURE in 7m 07s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/596665/4/check/openstack-tox-functional-py35/67b8725/ : FAILURE in 7m 34s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/65/596665/4/check/octavia-v1-dsvm-scenario/8149c5e/ : SUCCESS in 58m 05s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/65/596665/4/check/octavia-v1-dsvm-py3x-scenario/6404389/ : SUCCESS in 54m 32s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/65/596665/4/check/octavia-v1-dsvm-scenario-multinode/9f22960/ : FAILURE in 1h 18m 43s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/65/596665/4/check/octavia-v1-dsvm-py3x-scenario-multinode/31c193d/ : FAILURE in 1h 23m 10s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/65/596665/4/check/octavia-v1-dsvm-scenario-kvm-centos.7/929128b/ : SUCCESS in 1h 06m 09s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/65/596665/4/check/octavia-v2-dsvm-noop-api/2343cad/ : SUCCESS in 43m 04s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/65/596665/4/check/octavia-v2-dsvm-noop-py35-api/e76550d/ : SUCCESS in 41m 20s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/65/596665/4/check/octavia-v2-dsvm-scenario/7ae14da/ : SUCCESS in 1h 30m 09s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/65/596665/4/check/octavia-v2-dsvm-py35-scenario/3eb7d48/ : SUCCESS in 1h 10m 02s (non-voting)\n- octavia-grenade http://logs.openstack.org/65/596665/4/check/octavia-grenade/fefdf3c/ : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":4},{"id":"ffa5cb4c76b3cf11eba8dbe578709dcb1c4b7bef","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2018-09-11 02:39:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2e06ef81f50f0ae50ee64334add395fa68f733da","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2018-09-11 02:55:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cb9a8c08f10d1ba6e1c5ff9b235c314eb65dbe96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-11 04:21:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/65/596665/6/check/openstack-tox-cover/9ccd583/ : FAILURE in 8m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/596665/6/check/openstack-tox-lower-constraints/43f6cce/ : FAILURE in 6m 56s\n- openstack-tox-pep8 http://logs.openstack.org/65/596665/6/check/openstack-tox-pep8/28b9776/ : SUCCESS in 7m 22s\n- openstack-tox-py27 http://logs.openstack.org/65/596665/6/check/openstack-tox-py27/873a6ec/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/65/596665/6/check/openstack-tox-py35/76aa935/ : SUCCESS in 5m 09s\n- openstack-tox-py36 http://logs.openstack.org/65/596665/6/check/openstack-tox-py36/882b038/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/65/596665/6/check/openstack-tox-docs/f3c8d10/html/ : SUCCESS in 5m 45s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/65/596665/6/check/neutron-lbaasv2-dsvm-api/c95ecfe/ : SUCCESS in 1h 07m 07s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/65/596665/6/check/neutron-lbaasv2-dsvm-py3x-api/b567532/ : SUCCESS in 1h 07m 33s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/65/596665/6/check/neutron-lbaasv2-dsvm-scenario/39ef4d1/ : SUCCESS in 1h 10m 25s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/65/596665/6/check/neutron-lbaasv2-dsvm-py3x-scenario/f91cbb9/ : SUCCESS in 1h 08m 51s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/596665/6/check/openstack-tox-functional/487de82/ : FAILURE in 6m 40s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/596665/6/check/openstack-tox-functional-py35/48ee52c/ : FAILURE in 6m 21s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/65/596665/6/check/octavia-v1-dsvm-scenario/f8a1550/ : SUCCESS in 50m 24s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/65/596665/6/check/octavia-v1-dsvm-py3x-scenario/44d506b/ : SUCCESS in 49m 00s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/65/596665/6/check/octavia-v1-dsvm-scenario-multinode/8d8be9c/ : FAILURE in 1h 20m 28s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/65/596665/6/check/octavia-v1-dsvm-py3x-scenario-multinode/5117101/ : FAILURE in 1h 14m 01s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/65/596665/6/check/octavia-v1-dsvm-scenario-kvm-centos.7/43fa777/ : SUCCESS in 1h 06m 38s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/65/596665/6/check/octavia-v2-dsvm-noop-api/894e071/ : SUCCESS in 40m 56s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/65/596665/6/check/octavia-v2-dsvm-noop-py35-api/77487a3/ : SUCCESS in 36m 45s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/65/596665/6/check/octavia-v2-dsvm-scenario/e69d56c/ : SUCCESS in 1h 13m 36s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/65/596665/6/check/octavia-v2-dsvm-py35-scenario/17f798d/ : SUCCESS in 1h 11m 09s (non-voting)\n- octavia-grenade http://logs.openstack.org/65/596665/6/check/octavia-grenade/f229d21/ : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":6},{"id":"faadb4f29ada930bbd9c147a642fc9f36c118c82","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2018-09-11 05:37:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"674decef6c177fd37a9ea9a1a55e7357dbc35e6a","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2018-09-11 05:59:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c999bfc6fef347c6d4e998ff84d080f881880a07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-11 07:49:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/65/596665/8/check/openstack-tox-cover/e69ed6d/cover/ : SUCCESS in 8m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/596665/8/check/openstack-tox-lower-constraints/a510046/ : SUCCESS in 6m 45s\n- openstack-tox-pep8 http://logs.openstack.org/65/596665/8/check/openstack-tox-pep8/5dfa8f2/ : SUCCESS in 7m 53s\n- openstack-tox-py27 http://logs.openstack.org/65/596665/8/check/openstack-tox-py27/59a4026/ : SUCCESS in 5m 53s\n- openstack-tox-py35 http://logs.openstack.org/65/596665/8/check/openstack-tox-py35/2af5a2a/ : SUCCESS in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/65/596665/8/check/openstack-tox-py36/b089ae9/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/65/596665/8/check/openstack-tox-docs/60013a3/html/ : SUCCESS in 6m 34s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/65/596665/8/check/neutron-lbaasv2-dsvm-api/f2d98c9/ : SUCCESS in 1h 18m 59s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/65/596665/8/check/neutron-lbaasv2-dsvm-py3x-api/441511d/ : SUCCESS in 1h 31m 22s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/65/596665/8/check/neutron-lbaasv2-dsvm-scenario/2d15dad/ : SUCCESS in 1h 36m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/65/596665/8/check/neutron-lbaasv2-dsvm-py3x-scenario/8df35a3/ : SUCCESS in 1h 33m 32s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/596665/8/check/openstack-tox-functional/211610a/ : SUCCESS in 7m 16s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/596665/8/check/openstack-tox-functional-py35/52b4251/ : SUCCESS in 6m 32s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/65/596665/8/check/octavia-v1-dsvm-scenario/d69c43e/ : SUCCESS in 59m 23s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/65/596665/8/check/octavia-v1-dsvm-py3x-scenario/5c3e86c/ : SUCCESS in 1h 05m 43s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/65/596665/8/check/octavia-v1-dsvm-scenario-multinode/2e8ca21/ : FAILURE in 1h 20m 35s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/65/596665/8/check/octavia-v1-dsvm-py3x-scenario-multinode/ab75c81/ : FAILURE in 1h 13m 32s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/65/596665/8/check/octavia-v1-dsvm-scenario-kvm-centos.7/6de1c0f/ : FAILURE in 34m 45s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/65/596665/8/check/octavia-v2-dsvm-noop-api/57851b4/ : SUCCESS in 37m 07s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/65/596665/8/check/octavia-v2-dsvm-noop-py35-api/598b98e/ : SUCCESS in 40m 30s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/65/596665/8/check/octavia-v2-dsvm-scenario/e80ef22/ : SUCCESS in 1h 14m 52s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/65/596665/8/check/octavia-v2-dsvm-py35-scenario/fc3cba2/ : SUCCESS in 1h 31m 59s (non-voting)\n- octavia-grenade http://logs.openstack.org/65/596665/8/check/octavia-grenade/91137d9/ : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":8},{"id":"2d1cf668de17c805411e2d03bf5ef7ae7195bd67","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2018-09-20 09:58:12.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"5f610e6978685361f670236d88cfd134ca5c254e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-20 15:42:46.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/65/596665/9/check/openstack-tox-cover/d2fd85c/cover/ : SUCCESS in 10m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/596665/9/check/openstack-tox-lower-constraints/3208f52/ : SUCCESS in 9m 50s\n- openstack-tox-pep8 http://logs.openstack.org/65/596665/9/check/openstack-tox-pep8/cd5eb84/ : FAILURE in 9m 05s\n- openstack-tox-py27 http://logs.openstack.org/65/596665/9/check/openstack-tox-py27/5b95746/ : SUCCESS in 7m 02s\n- openstack-tox-py35 http://logs.openstack.org/65/596665/9/check/openstack-tox-py35/e4a02b4/ : SUCCESS in 6m 47s\n- openstack-tox-py36 http://logs.openstack.org/65/596665/9/check/openstack-tox-py36/3449574/ : SUCCESS in 7m 59s\n- openstack-tox-docs http://logs.openstack.org/65/596665/9/check/openstack-tox-docs/76d201f/html/ : SUCCESS in 7m 41s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/65/596665/9/check/neutron-lbaasv2-dsvm-api/884c211/ : SUCCESS in 1h 12m 05s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/65/596665/9/check/neutron-lbaasv2-dsvm-py3x-api/d1f2041/ : SUCCESS in 1h 11m 29s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/65/596665/9/check/neutron-lbaasv2-dsvm-scenario/4b43486/ : SUCCESS in 1h 25m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/65/596665/9/check/neutron-lbaasv2-dsvm-py3x-scenario/1f337be/ : SUCCESS in 54m 44s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/596665/9/check/openstack-tox-functional/a432516/ : SUCCESS in 6m 42s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/596665/9/check/openstack-tox-functional-py35/cae849d/ : SUCCESS in 7m 35s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/65/596665/9/check/octavia-v1-dsvm-scenario/3df262d/ : SUCCESS in 1h 00m 55s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/65/596665/9/check/octavia-v1-dsvm-py3x-scenario/7fbddfb/ : SUCCESS in 46m 41s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/65/596665/9/check/octavia-v1-dsvm-scenario-multinode/2a97582/ : FAILURE in 1h 05m 29s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/65/596665/9/check/octavia-v1-dsvm-py3x-scenario-multinode/9439234/ : FAILURE in 1h 22m 04s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/65/596665/9/check/octavia-v2-dsvm-noop-api/b1e6df4/ : SUCCESS in 40m 46s (non-voting)\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/65/596665/9/check/octavia-v2-dsvm-noop-py35-api/7335170/ : SUCCESS in 34m 59s (non-voting)\n- octavia-v2-dsvm-scenario http://logs.openstack.org/65/596665/9/check/octavia-v2-dsvm-scenario/58580ee/ : SUCCESS in 1h 42m 40s (non-voting)\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/65/596665/9/check/octavia-v2-dsvm-py35-scenario/d528d17/ : SUCCESS in 1h 09m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/65/596665/9/check/octavia-v2-dsvm-scenario-centos-7/e909bec/ : FAILURE in 39m 06s\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/65/596665/9/check/octavia-v2-dsvm-scenario-ubuntu-bionic/cb202ed/ : SUCCESS in 1h 42m 01s (non-voting)\n- octavia-grenade http://logs.openstack.org/65/596665/9/check/octavia-grenade/1e8b2d9/ : SUCCESS in 1h 09m 07s","accounts_in_message":[],"_revision_number":9},{"id":"127e62719f5c05a3436cb35f65b7d262bb979193","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2018-10-15 05:55:41.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"0dfc20ba9f0c8c695ae9ac8fba4680c95e374266","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2018-10-15 06:21:31.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"37bda0aa5e496ed1b87087afd4ed379ca011ee01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-15 07:27:25.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/65/596665/11/check/openstack-tox-cover/093f2df/ : FAILURE in 6m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/596665/11/check/openstack-tox-lower-constraints/9e9bbad/ : FAILURE in 6m 07s\n- openstack-tox-pep8 http://logs.openstack.org/65/596665/11/check/openstack-tox-pep8/1476fa8/ : SUCCESS in 8m 57s\n- openstack-tox-py27 http://logs.openstack.org/65/596665/11/check/openstack-tox-py27/8746b1a/ : FAILURE in 6m 13s\n- openstack-tox-py35 http://logs.openstack.org/65/596665/11/check/openstack-tox-py35/490b43c/ : FAILURE in 6m 28s\n- openstack-tox-py36 http://logs.openstack.org/65/596665/11/check/openstack-tox-py36/2fb78a9/ : FAILURE in 7m 00s\n- openstack-tox-docs http://logs.openstack.org/65/596665/11/check/openstack-tox-docs/ae35ea2/html/ : SUCCESS in 7m 00s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/65/596665/11/check/neutron-lbaasv2-dsvm-api/10cfa74/ : FAILURE in 25m 53s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/65/596665/11/check/neutron-lbaasv2-dsvm-py3x-api/0644d43/ : FAILURE in 24m 17s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/65/596665/11/check/neutron-lbaasv2-dsvm-scenario/08e7785/ : FAILURE in 23m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/65/596665/11/check/neutron-lbaasv2-dsvm-py3x-scenario/01e1251/ : FAILURE in 28m 12s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/65/596665/11/check/openstack-tox-functional/602cc5d/ : SUCCESS in 7m 17s\n- openstack-tox-functional-py35 http://logs.openstack.org/65/596665/11/check/openstack-tox-functional-py35/089d327/ : SUCCESS in 7m 25s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/65/596665/11/check/octavia-v1-dsvm-scenario/332b28d/ : FAILURE in 29m 58s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/65/596665/11/check/octavia-v1-dsvm-py3x-scenario/6ffa184/ : FAILURE in 22m 57s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/65/596665/11/check/octavia-v1-dsvm-scenario-multinode/1fed1ad/ : FAILURE in 25m 11s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/65/596665/11/check/octavia-v1-dsvm-py3x-scenario-multinode/67c01a7/ : FAILURE in 28m 35s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/65/596665/11/check/octavia-v2-dsvm-noop-api/d44619a/ : FAILURE in 22m 18s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/65/596665/11/check/octavia-v2-dsvm-noop-py35-api/ed50cd5/ : FAILURE in 21m 06s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/65/596665/11/check/octavia-v2-dsvm-scenario/fa1b904/ : FAILURE in 22m 40s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/65/596665/11/check/octavia-v2-dsvm-py35-scenario/d2c8812/ : FAILURE in 24m 28s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/65/596665/11/check/octavia-v2-dsvm-scenario-centos-7/03e8919/ : FAILURE in 27m 22s\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/65/596665/11/check/octavia-v2-dsvm-scenario-ubuntu-bionic/b209a22/ : FAILURE in 21m 53s (non-voting)\n- octavia-grenade http://logs.openstack.org/65/596665/11/check/octavia-grenade/1f5f14f/ : FAILURE in 55m 06s","accounts_in_message":[],"_revision_number":11},{"id":"578628867a703752dde0e53363bdfac1ac37cb3f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-14 19:48:02.000000000","message":"Patch Set 11:\n\nI\u0027ll re-base this one on-top of the one Carlos just did.","accounts_in_message":[],"_revision_number":11},{"id":"1a6003f0b753aa3dc9f53a2a2135d7ddc1d55140","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-14 19:48:11.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a7f6a7c94f035427aa6c817d265617418050035f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 21:38:41.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6971d97ffbed406e89bac94fb920b7cf : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9fbb8c652f948dc9a37bdbd7ba96345 : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8812416c3c40f4b7dd14eef7569620 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fa06e933e6e4e51a5e50c20b4a5f086 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed0d208a85464c20be37de955a703815 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0110d347b804473aab5b385950027cdc : SUCCESS in 10m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/45e3c67b897a4b6297258712b06d5757 : SUCCESS in 6m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9ad3ee21e5cc433083a20b124a9994ec : SUCCESS in 16m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ffedbcdf08674339abc04d750c232db8 : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/322f0ca6eb3f44e691626515aa3e5660 : SUCCESS in 55m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7bc18b873b84e698924a271cfe418bf : SUCCESS in 1h 35m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5dbf03dd194841fe9c5c7d92fc11f882 : SUCCESS in 1h 00m 37s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ba6063d3a4c64c0daff206ce662b7168 : FAILURE in 12m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce3a4151a0894595aa1dedee08648fe6 : FAILURE in 1h 02m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4f67439dc05145689ae7bce420350abd : SUCCESS in 1h 40m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d4b8bc13aa84586aeb448c536665e29 : SUCCESS in 1h 09m 21s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"53538e998e2177ca187fa303289294d079ca093a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-18 20:12:34.000000000","message":"Patch Set 12:\n\nThis will need a manual rebase.","accounts_in_message":[],"_revision_number":12},{"id":"ee594a4dd541ac509a317a279239508928ac1e9b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-03-17 01:46:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"0a6d2dcfa1fbd02d8fc383dce91eb25217ada847","revisions":{"3a1d2508795868ada121161f54f316f7ffbb2c44":{"kind":"REWORK","_number":1,"created":"2018-08-27 08:38:32.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/65/596665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/596665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/596665/1"}}},"commit":{"parents":[{"commit":"23ba1e28166b576e538efbbe0ba064528c2a313c","subject":"Add quota support to octavia\u0027s l7policy and l7rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23ba1e28166b576e538efbbe0ba064528c2a313c"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:36:50.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:36:50.000000000","tz":0},"subject":"[WIP] Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","message":"[WIP] Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027\n\nChange-Id: Idd313f20c183ade94588844ea63d0e18a63a009c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a1d2508795868ada121161f54f316f7ffbb2c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a1d2508795868ada121161f54f316f7ffbb2c44"}]},"branch":"refs/heads/master"},"624b3fe0a6a9919011427c2cf22f6777647c20f1":{"kind":"REWORK","_number":2,"created":"2018-08-27 08:44:38.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/65/596665/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/596665/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/596665/2"}}},"commit":{"parents":[{"commit":"23ba1e28166b576e538efbbe0ba064528c2a313c","subject":"Add quota support to octavia\u0027s l7policy and l7rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23ba1e28166b576e538efbbe0ba064528c2a313c"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:36:50.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:44:23.000000000","tz":0},"subject":"[WIP] Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","message":"[WIP] Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027\n\nChange-Id: Idd313f20c183ade94588844ea63d0e18a63a009c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/624b3fe0a6a9919011427c2cf22f6777647c20f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/624b3fe0a6a9919011427c2cf22f6777647c20f1"}]},"branch":"refs/heads/master"},"4ee195d32a1115bbf9978eb04841b4143a066ac1":{"kind":"REWORK","_number":3,"created":"2018-08-27 09:35:52.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/65/596665/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/596665/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/596665/3"}}},"commit":{"parents":[{"commit":"23ba1e28166b576e538efbbe0ba064528c2a313c","subject":"Add quota support to octavia\u0027s l7policy and l7rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23ba1e28166b576e538efbbe0ba064528c2a313c"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:36:50.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 09:35:33.000000000","tz":0},"subject":"[WIP] Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","message":"[WIP] Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027\n\nChange-Id: Idd313f20c183ade94588844ea63d0e18a63a009c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ee195d32a1115bbf9978eb04841b4143a066ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ee195d32a1115bbf9978eb04841b4143a066ac1"}]},"branch":"refs/heads/master"},"81dd0c91cb75d301f711e01592e24c369fb53624":{"kind":"REWORK","_number":4,"created":"2018-08-27 11:42:54.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/65/596665/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/596665/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/596665/4"}}},"commit":{"parents":[{"commit":"23ba1e28166b576e538efbbe0ba064528c2a313c","subject":"Add quota support to octavia\u0027s l7policy and l7rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23ba1e28166b576e538efbbe0ba064528c2a313c"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:36:50.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 11:42:37.000000000","tz":0},"subject":"[WIP] Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","message":"[WIP] Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027\n\nChange-Id: Idd313f20c183ade94588844ea63d0e18a63a009c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81dd0c91cb75d301f711e01592e24c369fb53624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81dd0c91cb75d301f711e01592e24c369fb53624"}]},"branch":"refs/heads/master"},"55894aada2fe62ef5f8e42fc77f5ef730df52c18":{"kind":"REWORK","_number":5,"created":"2018-09-11 02:39:19.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/65/596665/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/596665/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/596665/5"}}},"commit":{"parents":[{"commit":"e3f3461556162059a87000fe6afd48a4d43e2689","subject":"Add quota support to octavia\u0027s l7policy and l7rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3f3461556162059a87000fe6afd48a4d43e2689"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:36:50.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-09-11 02:38:49.000000000","tz":480},"subject":"[WIP] Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","message":"[WIP] Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027\n\nChange-Id: Idd313f20c183ade94588844ea63d0e18a63a009c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55894aada2fe62ef5f8e42fc77f5ef730df52c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55894aada2fe62ef5f8e42fc77f5ef730df52c18"}]},"branch":"refs/heads/master"},"528828a821aac3d9421188669e8911e8429bc471":{"kind":"REWORK","_number":6,"created":"2018-09-11 02:55:41.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/65/596665/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/596665/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/596665/6"}}},"commit":{"parents":[{"commit":"e3f3461556162059a87000fe6afd48a4d43e2689","subject":"Add quota support to octavia\u0027s l7policy and l7rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3f3461556162059a87000fe6afd48a4d43e2689"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:36:50.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-09-11 02:55:15.000000000","tz":480},"subject":"[WIP] Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","message":"[WIP] Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027\n\nChange-Id: Idd313f20c183ade94588844ea63d0e18a63a009c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/528828a821aac3d9421188669e8911e8429bc471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/528828a821aac3d9421188669e8911e8429bc471"}]},"branch":"refs/heads/master"},"ebfaa2335789f13b56e9f125d6babb512829c21a":{"kind":"REWORK","_number":7,"created":"2018-09-11 05:37:13.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/65/596665/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/596665/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/596665/7"}}},"commit":{"parents":[{"commit":"e3f3461556162059a87000fe6afd48a4d43e2689","subject":"Add quota support to octavia\u0027s l7policy and l7rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3f3461556162059a87000fe6afd48a4d43e2689"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:36:50.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-09-11 03:14:49.000000000","tz":480},"subject":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","message":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027\n\nThe functions \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027 contains many\nduplicate codes.The purpose of this patch is to simplify these codes.\n\nChange-Id: Idd313f20c183ade94588844ea63d0e18a63a009c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ebfaa2335789f13b56e9f125d6babb512829c21a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ebfaa2335789f13b56e9f125d6babb512829c21a"}]},"branch":"refs/heads/master"},"18438903ff70ccb612b52fc158c66043ab19bf57":{"kind":"REWORK","_number":8,"created":"2018-09-11 05:59:10.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/65/596665/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/596665/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/596665/8"}}},"commit":{"parents":[{"commit":"e3f3461556162059a87000fe6afd48a4d43e2689","subject":"Add quota support to octavia\u0027s l7policy and l7rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3f3461556162059a87000fe6afd48a4d43e2689"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:36:50.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-09-11 05:58:53.000000000","tz":480},"subject":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","message":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027\n\nThe functions \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027 contains many\nduplicate codes.The purpose of this patch is to simplify these codes.\n\nChange-Id: Idd313f20c183ade94588844ea63d0e18a63a009c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18438903ff70ccb612b52fc158c66043ab19bf57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18438903ff70ccb612b52fc158c66043ab19bf57"}]},"branch":"refs/heads/master"},"bf401ae8443bdd19e4b72071efa47ac572d6abd2":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2018-09-20 09:58:12.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/65/596665/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/596665/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/596665/9"}}},"commit":{"parents":[{"commit":"10ef961d9aef0387cd519a5b87059b93534f5dee","subject":"Add quota support to octavia\u0027s l7policy and l7rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/10ef961d9aef0387cd519a5b87059b93534f5dee"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:36:50.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-09-20 09:58:12.000000000","tz":0},"subject":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","message":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027\n\nThe functions \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027 contains many\nduplicate codes.The purpose of this patch is to simplify these codes.\n\nChange-Id: Idd313f20c183ade94588844ea63d0e18a63a009c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf401ae8443bdd19e4b72071efa47ac572d6abd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf401ae8443bdd19e4b72071efa47ac572d6abd2"}]},"branch":"refs/heads/master"},"5c8d8fd74f86d848e46435ca774f43a8552ac258":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2018-10-15 05:55:41.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/65/596665/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/596665/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/596665/10"}}},"commit":{"parents":[{"commit":"f84a5a0fbdb80e20019a232bc76aa8958a4d8874","subject":"Add quota support to octavia\u0027s l7policy and l7rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f84a5a0fbdb80e20019a232bc76aa8958a4d8874"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:36:50.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-10-15 05:55:41.000000000","tz":0},"subject":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","message":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027\n\nThe functions \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027 contains many\nduplicate codes.The purpose of this patch is to simplify these codes.\n\nChange-Id: Idd313f20c183ade94588844ea63d0e18a63a009c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c8d8fd74f86d848e46435ca774f43a8552ac258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c8d8fd74f86d848e46435ca774f43a8552ac258"}]},"branch":"refs/heads/master"},"46d7b0385e3a21d081e72dfbbd4e27d33cfa6d30":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2018-10-15 06:21:31.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/65/596665/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/596665/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/596665/11"}}},"commit":{"parents":[{"commit":"18fcf03c58ee38f3c66edd2be5b4ac96d75634b9","subject":"Add quota support to octavia\u0027s l7policy and l7rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18fcf03c58ee38f3c66edd2be5b4ac96d75634b9"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-27 08:36:50.000000000","tz":0},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-10-15 06:21:31.000000000","tz":0},"subject":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","message":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027\n\nThe functions \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027 contains many\nduplicate codes.The purpose of this patch is to simplify these codes.\n\nChange-Id: Idd313f20c183ade94588844ea63d0e18a63a009c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46d7b0385e3a21d081e72dfbbd4e27d33cfa6d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46d7b0385e3a21d081e72dfbbd4e27d33cfa6d30"}]},"branch":"refs/heads/master"},"0a6d2dcfa1fbd02d8fc383dce91eb25217ada847":{"kind":"REWORK","_number":12,"created":"2020-01-14 19:48:11.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/65/596665/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/596665/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/596665/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/596665/12"}}},"commit":{"parents":[{"commit":"327041a7353107ac5bbc9ba1643e9a07f18827b4","subject":"Add quota support to octavia\u0027s l7policy and l7rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/327041a7353107ac5bbc9ba1643e9a07f18827b4"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-14 19:45:43.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-14 19:45:43.000000000","tz":-300},"subject":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027","message":"Refactor \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027\n\nThe functions \u0027check_quota_met\u0027 and \u0027decrement_quota\u0027 contains many\nduplicate codes.The purpose of this patch is to simplify these codes.\n\nChange-Id: Idd313f20c183ade94588844ea63d0e18a63a009c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a6d2dcfa1fbd02d8fc383dce91eb25217ada847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a6d2dcfa1fbd02d8fc383dce91eb25217ada847"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
