)]}'
{"id":"openstack%2Foctavia~615382","triplet_id":"openstack%2Foctavia~master~I533000136437af12a5e507dcbf554f03a63ce447","project":"openstack/octavia","branch":"master","topic":"oslo-notifications","hashtags":[],"change_id":"I533000136437af12a5e507dcbf554f03a63ce447","subject":"Add notifications about changed status to api server","status":"NEW","created":"2018-11-03 15:35:55.000000000","updated":"2019-02-20 22:44:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"79f52cb3621984c88682e67e276e252ce4daf9df","_number":615382,"virtual_id_number":615382,"owner":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2018-11-03 17:38:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2019-02-20 22:44:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2018-12-05 19:16:57.000000000","_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-03 17:38:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-12-05 19:16:57.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2019-02-20 22:44:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"7283dc534e3bf61ebbb98133c8098ac68fd7a857","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2018-11-03 15:35:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c366b458ff89ebf50a85143fb13e84c105aa022e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-03 17:38:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/615382/1/check/openstack-tox-cover/2b5988f/cover/ : SUCCESS in 7m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/615382/1/check/openstack-tox-lower-constraints/847710d/ : SUCCESS in 5m 57s\n- openstack-tox-pep8 http://logs.openstack.org/82/615382/1/check/openstack-tox-pep8/db5b2c9/ : SUCCESS in 7m 58s\n- openstack-tox-py27 http://logs.openstack.org/82/615382/1/check/openstack-tox-py27/eb67333/ : SUCCESS in 5m 31s\n- openstack-tox-py35 http://logs.openstack.org/82/615382/1/check/openstack-tox-py35/15a9fb0/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/82/615382/1/check/openstack-tox-py36/62c3ce2/ : SUCCESS in 6m 46s\n- openstack-tox-docs http://logs.openstack.org/82/615382/1/check/openstack-tox-docs/050cce6/html/ : SUCCESS in 5m 36s\n- build-openstack-releasenotes http://logs.openstack.org/82/615382/1/check/build-openstack-releasenotes/2598f93/html/ : SUCCESS in 4m 15s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/82/615382/1/check/neutron-lbaasv2-dsvm-api/a91aa62/ : SUCCESS in 1h 06m 42s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/82/615382/1/check/neutron-lbaasv2-dsvm-py3x-api/0247950/ : SUCCESS in 1h 08m 03s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/82/615382/1/check/neutron-lbaasv2-dsvm-scenario/3152c03/ : SUCCESS in 1h 28m 45s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/82/615382/1/check/neutron-lbaasv2-dsvm-py3x-scenario/7b86e5c/ : SUCCESS in 1h 19m 11s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/82/615382/1/check/openstack-tox-functional/4b0008d/ : SUCCESS in 6m 54s\n- openstack-tox-functional-py35 http://logs.openstack.org/82/615382/1/check/openstack-tox-functional-py35/c2223d0/ : SUCCESS in 7m 05s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/82/615382/1/check/octavia-v1-dsvm-scenario/b4f3b2e/ : SUCCESS in 58m 23s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/82/615382/1/check/octavia-v1-dsvm-py3x-scenario/66ccf50/ : SUCCESS in 48m 49s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/82/615382/1/check/octavia-v1-dsvm-scenario-multinode/98cb192/ : FAILURE in 1h 12m 09s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/82/615382/1/check/octavia-v1-dsvm-py3x-scenario-multinode/41ae935/ : FAILURE in 1h 11m 48s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/82/615382/1/check/octavia-v2-dsvm-noop-api/980cd75/ : SUCCESS in 38m 12s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/82/615382/1/check/octavia-v2-dsvm-noop-py35-api/b154e2f/ : SUCCESS in 39m 06s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/82/615382/1/check/octavia-v2-dsvm-scenario/9991c56/ : SUCCESS in 1h 32m 12s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/82/615382/1/check/octavia-v2-dsvm-py35-scenario/3caddea/ : SUCCESS in 1h 19m 08s\n- octavia-v2-dsvm-scenario-centos-7 http://logs.openstack.org/82/615382/1/check/octavia-v2-dsvm-scenario-centos-7/afc9128/ : SUCCESS in 1h 57m 19s\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/82/615382/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/e1a10db/ : SUCCESS in 1h 19m 13s (non-voting)\n- octavia-grenade http://logs.openstack.org/82/615382/1/check/octavia-grenade/975bdf0/ : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a652361babdfe4ac9c75d7829453d2c38e951f0a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-02-20 22:44:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5b8eea23dcd058be0db4235436d218043025273","revisions":{"d5b8eea23dcd058be0db4235436d218043025273":{"kind":"REWORK","_number":1,"created":"2018-11-03 15:35:55.000000000","uploader":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"ref":"refs/changes/82/615382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/615382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/615382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/615382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/615382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/615382/1"}}},"commit":{"parents":[{"commit":"61617411285b9777171d7f5cf16b08d82454960b","subject":"Add notifications about changed status to worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61617411285b9777171d7f5cf16b08d82454960b"}]}],"author":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2018-11-03 15:33:09.000000000","tz":180},"committer":{"name":"Vadim Ponomarev","email":"velizarx@gmail.com","date":"2018-11-03 15:34:39.000000000","tz":180},"subject":"Add notifications about changed status to api server","message":"Add notifications about changed status to api server\n\nChange-Id: I533000136437af12a5e507dcbf554f03a63ce447\nStory: 2004122\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5b8eea23dcd058be0db4235436d218043025273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5b8eea23dcd058be0db4235436d218043025273"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
