)]}'
{"id":"openstack%2Foctavia~628509","triplet_id":"openstack%2Foctavia~master~Ie160436913e4d935bab118d31ba10193ac38bd8f","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ie160436913e4d935bab118d31ba10193ac38bd8f","subject":"Update hacking version to latest","status":"MERGED","created":"2019-01-04 15:59:28.000000000","updated":"2019-05-15 19:30:30.000000000","submitted":"2019-05-15 19:30:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"628509-1557948630732-cd9c5273","meta_rev_id":"8151e5076f41b9871eeebc50db2de733f0cfdef9","_number":628509,"virtual_id_number":628509,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-05-15 19:30:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":2,"date":"2019-05-15 12:43:00.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-05-15 08:48:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":1,"date":"2019-05-15 12:43:00.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-15 00:38:59.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 00:38:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-05-15 08:48:41.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-05-15 12:43:00.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-05-15 19:30:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c0d4bb223cadfebc232b1d297200118ed19104f","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-01-04 15:59:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14fb2041e03bcca9e98612780d61e44b0777921d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-04 18:45:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/09/628509/1/check/build-openstack-api-ref/f2c48b7/html/ : SUCCESS in 6m 02s\n- requirements-check http://logs.openstack.org/09/628509/1/check/requirements-check/360b05e/ : SUCCESS in 3m 46s\n- openstack-tox-cover http://logs.openstack.org/09/628509/1/check/openstack-tox-cover/36aa105/cover/ : SUCCESS in 8m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/628509/1/check/openstack-tox-lower-constraints/5bd3959/ : SUCCESS in 7m 24s\n- openstack-tox-pep8 http://logs.openstack.org/09/628509/1/check/openstack-tox-pep8/27e767f/ : FAILURE in 5m 41s\n- openstack-tox-py27 http://logs.openstack.org/09/628509/1/check/openstack-tox-py27/c2e1272/ : SUCCESS in 5m 56s\n- openstack-tox-py35 http://logs.openstack.org/09/628509/1/check/openstack-tox-py35/44c347e/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/09/628509/1/check/openstack-tox-py36/684b40a/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/09/628509/1/check/openstack-tox-docs/d783c1c/html/ : SUCCESS in 6m 22s\n- build-openstack-releasenotes http://logs.openstack.org/09/628509/1/check/build-openstack-releasenotes/8d8ffb4/html/ : SUCCESS in 5m 10s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/09/628509/1/check/neutron-lbaasv2-dsvm-api/4893c5e/ : SUCCESS in 1h 17m 33s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/09/628509/1/check/neutron-lbaasv2-dsvm-py3x-api/5688048/ : SUCCESS in 1h 16m 25s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/09/628509/1/check/neutron-lbaasv2-dsvm-scenario/d8d44a0/ : SUCCESS in 1h 21m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/09/628509/1/check/neutron-lbaasv2-dsvm-py3x-scenario/548a417/ : SUCCESS in 2h 00m 39s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/09/628509/1/check/openstack-tox-functional/1571286/ : SUCCESS in 6m 54s\n- openstack-tox-functional-py35 http://logs.openstack.org/09/628509/1/check/openstack-tox-functional-py35/492541e/ : SUCCESS in 6m 51s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/09/628509/1/check/octavia-v1-dsvm-scenario/d83adbf/ : SUCCESS in 54m 04s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/09/628509/1/check/octavia-v1-dsvm-py3x-scenario/4595c8a/ : SUCCESS in 1h 04m 44s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/09/628509/1/check/octavia-v2-dsvm-noop-api/9e3506f/ : SUCCESS in 39m 25s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/09/628509/1/check/octavia-v2-dsvm-noop-py2-api/320e63c/ : SUCCESS in 39m 40s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/09/628509/1/check/octavia-v2-dsvm-scenario/b5b752e/ : SUCCESS in 1h 26m 27s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/09/628509/1/check/octavia-v2-dsvm-py2-scenario/5656d77/ : SUCCESS in 1h 29m 52s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/09/628509/1/check/octavia-v2-dsvm-py2-scenario-centos-7/e6f5c3c/ : TIMED_OUT in 2h 33m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/09/628509/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/082b93a/ : FAILURE in 34m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/09/628509/1/check/octavia-v2-dsvm-scenario-two-node/b926edc/ : FAILURE in 2h 06m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/09/628509/1/check/octavia-v2-dsvm-py2-scenario-two-node/75f2e89/ : SUCCESS in 1h 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a4bef4d7953c278dce621a0a42b4b7174466000","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-01-06 09:00:23.000000000","message":"Patch Set 1:\n\nA. Why? you provided zero information about the reasoning for this change.\nB. Your patch will not update hacking to the latest since it will restrict it to \u003c1.2.0\nC. fails PEP8\nD. I see you proposed it to numerous projects, you might be better of following the instructions at https://wiki.openstack.org/wiki/Requirements","accounts_in_message":[],"_revision_number":1},{"id":"2fb82473ee437bbd563683b108643b8c328e4ade","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-01-06 09:01:17.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"457503621b30522abd356facb562c1b6d6796065","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-15 00:38:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7a438b8353d233d2097a114f60e51cf16273b79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 03:22:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/09/628509/2/check/build-openstack-api-ref/6cc676e/html/ : SUCCESS in 4m 25s\n- requirements-check http://logs.openstack.org/09/628509/2/check/requirements-check/30ac8dc/ : SUCCESS in 4m 08s\n- openstack-tox-cover http://logs.openstack.org/09/628509/2/check/openstack-tox-cover/8d7ee58/cover/ : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/628509/2/check/openstack-tox-lower-constraints/225a924/ : SUCCESS in 5m 56s\n- openstack-tox-pep8 http://logs.openstack.org/09/628509/2/check/openstack-tox-pep8/08b453b/ : SUCCESS in 5m 54s\n- openstack-tox-py27 http://logs.openstack.org/09/628509/2/check/openstack-tox-py27/d4c102e/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/09/628509/2/check/openstack-tox-py36/c057496/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/09/628509/2/check/openstack-tox-docs/990a885/html/ : SUCCESS in 6m 11s\n- build-openstack-releasenotes http://logs.openstack.org/09/628509/2/check/build-openstack-releasenotes/9418f76/html/ : SUCCESS in 6m 14s\n- openstack-tox-functional http://logs.openstack.org/09/628509/2/check/openstack-tox-functional/1f1fdaf/ : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 http://logs.openstack.org/09/628509/2/check/openstack-tox-functional-py36/74da112/ : SUCCESS in 7m 01s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-noop-api/0889965/ : SUCCESS in 42m 13s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-noop-py2-api/aeeda7f/ : SUCCESS in 45m 26s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-scenario/0c7aebe/ : FAILURE in 1h 35m 00s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-py2-scenario/ca14f53/ : SUCCESS in 1h 26m 56s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-py2-scenario-centos-7/e77cc0f/ : FAILURE in 1h 58m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/10bec4d/ : SUCCESS in 1h 44m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-scenario-two-node/455fa99/ : FAILURE in 2h 29m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-py2-scenario-two-node/5a33d13/ : SUCCESS in 2h 01m 50s (non-voting)\n- octavia-grenade http://logs.openstack.org/09/628509/2/check/octavia-grenade/4c6a4c2/ : SUCCESS in 1h 03m 12s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-tls-barbican/75d7eb6/ : SUCCESS in 53m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e3a0cf10263880654dc7d8029ee7297a67efeee","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-15 05:58:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b5e6fe60d95c7b477811cb58505488cfc60e5682","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 08:21:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/09/628509/2/check/build-openstack-api-ref/04a9567/html/ : SUCCESS in 4m 35s\n- requirements-check http://logs.openstack.org/09/628509/2/check/requirements-check/73e79ff/ : SUCCESS in 3m 57s\n- openstack-tox-cover http://logs.openstack.org/09/628509/2/check/openstack-tox-cover/85b1a58/cover/ : SUCCESS in 7m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/628509/2/check/openstack-tox-lower-constraints/b2494d8/ : SUCCESS in 8m 06s\n- openstack-tox-pep8 http://logs.openstack.org/09/628509/2/check/openstack-tox-pep8/5c9cf89/ : SUCCESS in 7m 34s\n- openstack-tox-py27 http://logs.openstack.org/09/628509/2/check/openstack-tox-py27/bb98a33/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/09/628509/2/check/openstack-tox-py36/ce7f5a8/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/09/628509/2/check/openstack-tox-docs/8064077/html/ : SUCCESS in 5m 34s\n- build-openstack-releasenotes http://logs.openstack.org/09/628509/2/check/build-openstack-releasenotes/9ca640c/html/ : SUCCESS in 6m 00s\n- openstack-tox-functional http://logs.openstack.org/09/628509/2/check/openstack-tox-functional/9479a01/ : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 http://logs.openstack.org/09/628509/2/check/openstack-tox-functional-py36/c6c40c3/ : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-noop-api/d07107f/ : SUCCESS in 41m 44s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-noop-py2-api/350a122/ : SUCCESS in 47m 30s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-scenario/1632f20/ : SUCCESS in 1h 31m 29s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-py2-scenario/9644de5/ : SUCCESS in 1h 36m 33s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-py2-scenario-centos-7/a7aa2d7/ : POST_FAILURE in 2h 17m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/1b1eafa/ : SUCCESS in 1h 31m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-scenario-two-node/e217bc4/ : SUCCESS in 1h 54m 48s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-py2-scenario-two-node/d353ac2/ : SUCCESS in 1h 44m 33s (non-voting)\n- octavia-grenade http://logs.openstack.org/09/628509/2/check/octavia-grenade/a219c9b/ : SUCCESS in 1h 32m 49s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-tls-barbican/8815999/ : SUCCESS in 57m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe46e4f9bf86e496bf07376117dc880ba3aeee41","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-15 08:48:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"93855865bc2cd6cfd05f91e3479e514723370036","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-05-15 12:43:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"f33b4041a59762548e0b5f197251ac97b8501976","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 12:43:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd44a493f7237a6bc099920cb8f14bcb04e8e700","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 14:39:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/09/628509/2/gate/build-openstack-api-ref/2bcd177/html/ : SUCCESS in 6m 11s\n- requirements-check http://logs.openstack.org/09/628509/2/gate/requirements-check/db4c75c/ : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/628509/2/gate/openstack-tox-lower-constraints/b6ed554/ : SUCCESS in 6m 58s\n- openstack-tox-pep8 http://logs.openstack.org/09/628509/2/gate/openstack-tox-pep8/0c9c732/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/09/628509/2/gate/openstack-tox-py27/c5ab3d2/ : SUCCESS in 5m 21s\n- openstack-tox-py36 http://logs.openstack.org/09/628509/2/gate/openstack-tox-py36/0af33a7/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/09/628509/2/gate/openstack-tox-docs/3e1c327/html/ : SUCCESS in 6m 09s\n- build-openstack-releasenotes http://logs.openstack.org/09/628509/2/gate/build-openstack-releasenotes/6a0a86f/html/ : SUCCESS in 5m 39s\n- openstack-tox-functional http://logs.openstack.org/09/628509/2/gate/openstack-tox-functional/e6de840/ : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 http://logs.openstack.org/09/628509/2/gate/openstack-tox-functional-py36/d771a96/ : SUCCESS in 6m 18s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/09/628509/2/gate/octavia-v2-dsvm-noop-api/1b3c9f5/ : SUCCESS in 41m 24s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/09/628509/2/gate/octavia-v2-dsvm-noop-py2-api/d100bfc/ : SUCCESS in 42m 18s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/09/628509/2/gate/octavia-v2-dsvm-scenario/7687a8a/ : FAILURE in 1h 43m 30s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/09/628509/2/gate/octavia-v2-dsvm-py2-scenario/21231de/ : FAILURE in 1h 34m 14s\n- octavia-grenade http://logs.openstack.org/09/628509/2/gate/octavia-grenade/cad4b50/ : SUCCESS in 1h 12m 47s","accounts_in_message":[],"_revision_number":2},{"id":"5d481da772709450e6a6b8a7ef131c1055ffbb2e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-15 15:04:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c82abd95483e39fffcab090c34ba5fe0bdb0cb6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 17:36:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/09/628509/2/check/build-openstack-api-ref/2fdc49c/html/ : SUCCESS in 5m 35s\n- requirements-check http://logs.openstack.org/09/628509/2/check/requirements-check/26eb0b8/ : SUCCESS in 3m 18s\n- openstack-tox-cover http://logs.openstack.org/09/628509/2/check/openstack-tox-cover/35ec31b/cover/ : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/628509/2/check/openstack-tox-lower-constraints/b24ac9e/ : SUCCESS in 7m 06s\n- openstack-tox-pep8 http://logs.openstack.org/09/628509/2/check/openstack-tox-pep8/c678130/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/09/628509/2/check/openstack-tox-py27/d8c07e7/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/09/628509/2/check/openstack-tox-py36/f9a31d2/ : SUCCESS in 4m 48s\n- openstack-tox-docs http://logs.openstack.org/09/628509/2/check/openstack-tox-docs/aec74d4/html/ : SUCCESS in 6m 19s\n- build-openstack-releasenotes http://logs.openstack.org/09/628509/2/check/build-openstack-releasenotes/5c7771b/html/ : SUCCESS in 4m 53s\n- openstack-tox-functional http://logs.openstack.org/09/628509/2/check/openstack-tox-functional/f257b6b/ : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 http://logs.openstack.org/09/628509/2/check/openstack-tox-functional-py36/4aa784d/ : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-noop-api/34ec097/ : SUCCESS in 45m 34s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-noop-py2-api/80100a6/ : SUCCESS in 1h 12m 55s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-scenario/2dbf73b/ : SUCCESS in 1h 36m 38s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-py2-scenario/83002a4/ : SUCCESS in 1h 41m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-py2-scenario-centos-7/87fdf9a/ : FAILURE in 2h 08m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/479e268/ : SUCCESS in 1h 40m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-scenario-two-node/2172b27/ : FAILURE in 1h 57m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-py2-scenario-two-node/28ab91c/ : FAILURE in 2h 14m 40s (non-voting)\n- octavia-grenade http://logs.openstack.org/09/628509/2/check/octavia-grenade/adc3889/ : SUCCESS in 1h 13m 05s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/09/628509/2/check/octavia-v2-dsvm-tls-barbican/f3de766/ : SUCCESS in 56m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a52cd04b4d5dcbbc087184bfc62daf598231fb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 17:36:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d7f1cf1785dd5111ef4d09e536fb467b1fabc3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 19:30:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8151e5076f41b9871eeebc50db2de733f0cfdef9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 19:30:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/09/628509/2/gate/build-openstack-api-ref/227b9d0/html/ : SUCCESS in 5m 20s\n- requirements-check http://logs.openstack.org/09/628509/2/gate/requirements-check/4d1572c/ : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/628509/2/gate/openstack-tox-lower-constraints/c59c1c4/ : SUCCESS in 6m 20s\n- openstack-tox-pep8 http://logs.openstack.org/09/628509/2/gate/openstack-tox-pep8/448c80e/ : SUCCESS in 5m 53s\n- openstack-tox-py27 http://logs.openstack.org/09/628509/2/gate/openstack-tox-py27/d16cdd0/ : SUCCESS in 5m 03s\n- openstack-tox-py36 http://logs.openstack.org/09/628509/2/gate/openstack-tox-py36/d245a96/ : SUCCESS in 4m 31s\n- openstack-tox-docs http://logs.openstack.org/09/628509/2/gate/openstack-tox-docs/3056ae9/html/ : SUCCESS in 5m 37s\n- build-openstack-releasenotes http://logs.openstack.org/09/628509/2/gate/build-openstack-releasenotes/dfdec37/html/ : SUCCESS in 4m 47s\n- openstack-tox-functional http://logs.openstack.org/09/628509/2/gate/openstack-tox-functional/967f390/ : SUCCESS in 6m 38s\n- openstack-tox-functional-py36 http://logs.openstack.org/09/628509/2/gate/openstack-tox-functional-py36/1a276aa/ : SUCCESS in 6m 24s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/09/628509/2/gate/octavia-v2-dsvm-noop-api/c9e6ddc/ : SUCCESS in 45m 41s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/09/628509/2/gate/octavia-v2-dsvm-noop-py2-api/7e2958b/ : SUCCESS in 40m 55s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/09/628509/2/gate/octavia-v2-dsvm-scenario/6b000de/ : SUCCESS in 1h 47m 55s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/09/628509/2/gate/octavia-v2-dsvm-py2-scenario/704b508/ : SUCCESS in 1h 29m 06s\n- octavia-grenade http://logs.openstack.org/09/628509/2/gate/octavia-grenade/5a7281b/ : SUCCESS in 1h 17m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff50886d796a25fa606c71536f89734fa5d20dfa","revisions":{"f434dc7699020e2437900f6e5f2a8c24d4bc74c1":{"kind":"REWORK","_number":1,"created":"2019-01-04 15:59:28.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/09/628509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/628509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/628509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/628509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/628509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/628509/1"}}},"commit":{"parents":[{"commit":"cb2c5fc1a56247f0c837534156ada758f67a8d7f","subject":"Fix grenade job to clone Octavia from base branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb2c5fc1a56247f0c837534156ada758f67a8d7f"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 15:59:14.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 15:59:24.000000000","tz":480},"subject":"Update hacking version to latest","message":"Update hacking version to latest\n\nChange-Id: Ie160436913e4d935bab118d31ba10193ac38bd8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f434dc7699020e2437900f6e5f2a8c24d4bc74c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f434dc7699020e2437900f6e5f2a8c24d4bc74c1"}]},"branch":"refs/heads/master"},"ff50886d796a25fa606c71536f89734fa5d20dfa":{"kind":"REWORK","_number":2,"created":"2019-05-15 00:38:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/09/628509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/628509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/628509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/628509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/628509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/628509/2"}}},"commit":{"parents":[{"commit":"2c99d0109c8ac3970ce2d5548ad5a2bc8c35dfb3","subject":"Cap to sphinx 2.0.0 on Python 2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c99d0109c8ac3970ce2d5548ad5a2bc8c35dfb3"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 15:59:14.000000000","tz":480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-15 00:38:58.000000000","tz":-420},"subject":"Update hacking version to latest","message":"Update hacking version to latest\n\nThis resolves extranous \"improper escape sequence\" warnings on\npython 3.6+[1].\n\nNote, this does not resolve those warnings from pylint. There\nis already another proposed patch to address pylint[2].\n\n[1] https://review.opendev.org/494322\n[2] https://review.opendev.org/635236\n\nChange-Id: Ie160436913e4d935bab118d31ba10193ac38bd8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff50886d796a25fa606c71536f89734fa5d20dfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff50886d796a25fa606c71536f89734fa5d20dfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
