)]}'
{"id":"openstack%2Foctavia~634302","triplet_id":"openstack%2Foctavia~master~Ib3ec668290819b07aebca741c3d3197b451ef30d","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ib3ec668290819b07aebca741c3d3197b451ef30d","subject":"Allow members to be accessed as a base resource","status":"NEW","created":"2019-01-31 20:07:04.000000000","updated":"2019-02-04 22:58:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b3073da0517b0ff01dc8cbe853a34bbffe1e4c38","_number":634302,"virtual_id_number":634302,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-01-31 22:48:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-1,"date":"2019-02-01 10:07:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":-1,"date":"2019-02-04 22:58:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-31 22:48:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-01 10:07:12.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"adf0caca894354e23909ef316b5adc161cd8b022","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-01-31 20:07:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9f6b1501b7fc8429b906ee0e941c46fabcf9f4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 22:48:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/02/634302/1/check/build-openstack-api-ref/81c92f9/html/ : SUCCESS in 6m 54s\n- openstack-tox-cover http://logs.openstack.org/02/634302/1/check/openstack-tox-cover/a1472bf/cover/ : SUCCESS in 9m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/634302/1/check/openstack-tox-lower-constraints/45f4615/ : SUCCESS in 8m 08s\n- openstack-tox-pep8 http://logs.openstack.org/02/634302/1/check/openstack-tox-pep8/de658ae/ : SUCCESS in 9m 40s\n- openstack-tox-py27 http://logs.openstack.org/02/634302/1/check/openstack-tox-py27/1147894/ : SUCCESS in 6m 21s\n- openstack-tox-py35 http://logs.openstack.org/02/634302/1/check/openstack-tox-py35/a29fc5f/ : SUCCESS in 7m 13s\n- openstack-tox-py36 http://logs.openstack.org/02/634302/1/check/openstack-tox-py36/a50e652/ : SUCCESS in 7m 26s\n- openstack-tox-docs http://logs.openstack.org/02/634302/1/check/openstack-tox-docs/cb29409/html/ : SUCCESS in 7m 05s\n- build-openstack-releasenotes http://logs.openstack.org/02/634302/1/check/build-openstack-releasenotes/85345c9/html/ : SUCCESS in 5m 43s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/02/634302/1/check/neutron-lbaasv2-dsvm-api/2aa6539/ : SUCCESS in 1h 15m 33s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/02/634302/1/check/neutron-lbaasv2-dsvm-py3x-api/82f38d3/ : SUCCESS in 1h 18m 18s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/02/634302/1/check/neutron-lbaasv2-dsvm-scenario/9541155/ : SUCCESS in 1h 28m 23s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/02/634302/1/check/neutron-lbaasv2-dsvm-py3x-scenario/852e9b8/ : SUCCESS in 1h 25m 46s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/02/634302/1/check/openstack-tox-functional/3950ce1/ : SUCCESS in 7m 36s\n- openstack-tox-functional-py35 http://logs.openstack.org/02/634302/1/check/openstack-tox-functional-py35/8361662/ : SUCCESS in 7m 19s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/02/634302/1/check/octavia-v1-dsvm-scenario/018edab/ : SUCCESS in 1h 04m 45s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/02/634302/1/check/octavia-v1-dsvm-py3x-scenario/5eb4fcb/ : SUCCESS in 59m 55s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/02/634302/1/check/octavia-v2-dsvm-noop-api/cd6c550/ : SUCCESS in 42m 35s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/02/634302/1/check/octavia-v2-dsvm-noop-py2-api/1463285/ : SUCCESS in 39m 13s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/02/634302/1/check/octavia-v2-dsvm-scenario/af7c40d/ : SUCCESS in 1h 31m 02s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/02/634302/1/check/octavia-v2-dsvm-py2-scenario/4728eec/ : SUCCESS in 1h 33m 45s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/02/634302/1/check/octavia-v2-dsvm-py2-scenario-centos-7/655e736/ : TIMED_OUT in 2h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/02/634302/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/afc89a0/ : SUCCESS in 1h 40m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/02/634302/1/check/octavia-v2-dsvm-scenario-two-node/5226493/ : FAILURE in 1h 51m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/02/634302/1/check/octavia-v2-dsvm-py2-scenario-two-node/1e85c61/ : FAILURE in 1h 34m 30s (non-voting)\n- octavia-grenade http://logs.openstack.org/02/634302/1/check/octavia-grenade/bceb6ad/ : SUCCESS in 1h 08m 02s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/02/634302/1/check/octavia-v2-dsvm-tls-barbican/c01943d/ : SUCCESS in 51m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c64312bcbf12757b20f3e51ab4931d19ffd6f31","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-01 10:07:12.000000000","message":"Patch Set 1: Code-Review-1\n\nI haven\u0027t reviewed the code per se yet but one thing caught my attention: API version needs to be bumped in octavia/api/root_controller.py","accounts_in_message":[],"_revision_number":1},{"id":"b3073da0517b0ff01dc8cbe853a34bbffe1e4c38","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-02-04 22:58:17.000000000","message":"Patch Set 1: Workflow-1\n\nTruth! And honestly this isn\u0027t QUITE done, should probably WIP it, because there\u0027s a couple more tests I need to write before I\u0027m comfortable with it.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66bb793b4fa2c8ab96812c9b2570b0713100f326","revisions":{"66bb793b4fa2c8ab96812c9b2570b0713100f326":{"kind":"REWORK","_number":1,"created":"2019-01-31 20:07:04.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/02/634302/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/634302/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/634302/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/634302/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/634302/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/634302/1"}}},"commit":{"parents":[{"commit":"979144f2fdf7c391b6c154c01a22f107f45da833","subject":"Merge \"Fix a topology bug in flavors support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/979144f2fdf7c391b6c154c01a22f107f45da833"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-01-31 20:06:37.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-01-31 20:06:37.000000000","tz":-480},"subject":"Allow members to be accessed as a base resource","message":"Allow members to be accessed as a base resource\n\nChange-Id: Ib3ec668290819b07aebca741c3d3197b451ef30d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66bb793b4fa2c8ab96812c9b2570b0713100f326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66bb793b4fa2c8ab96812c9b2570b0713100f326"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
