)]}'
{"id":"openstack%2Foctavia~635236","triplet_id":"openstack%2Foctavia~master~I52301d763797d619f195bd8a1c32bc47f1e68420","project":"openstack/octavia","branch":"master","topic":"652665","hashtags":[],"change_id":"I52301d763797d619f195bd8a1c32bc47f1e68420","subject":"Add Python 3.7 support","status":"MERGED","created":"2019-02-06 18:13:52.000000000","updated":"2019-05-16 06:30:28.000000000","submitted":"2019-05-16 06:30:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"635236-1557988228066-96febb6b","meta_rev_id":"b92f6b6daec4eabc90d9707c2a03d41be34f69ed","_number":635236,"virtual_id_number":635236,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"date":"2019-05-15 05:51:47.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-05-16 06:30:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-05-16 04:47:45.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-05-15 16:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-05-16 04:47:45.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-14 14:15:45.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-04-19 10:11:13.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-05-10 20:38:47.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-05-15 16:48:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-05-16 04:47:45.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-05-16 06:30:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3b4e7824a3b2bfa81dfe106b95456545872c95e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-06 18:13:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b21e6aa46efe028bda2fbabbafc568da8c5c178","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-06 18:38:58.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5fbf9434365a8f412aa6ebe067fb5b90146c4c5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-06 20:30:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/1/check/build-openstack-api-ref/3c17774/html/ : SUCCESS in 5m 41s\n- requirements-check http://logs.openstack.org/36/635236/1/check/requirements-check/7b164f2/ : SUCCESS in 4m 05s\n- openstack-tox-cover http://logs.openstack.org/36/635236/1/check/openstack-tox-cover/e07d4c5/cover/ : SUCCESS in 9m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/1/check/openstack-tox-lower-constraints/7eab2e3/ : SUCCESS in 8m 21s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/1/check/openstack-tox-pep8/8c79911/ : FAILURE in 8m 57s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/1/check/openstack-tox-py27/6f50dee/ : SUCCESS in 6m 11s\n- openstack-tox-py35 http://logs.openstack.org/36/635236/1/check/openstack-tox-py35/64057e1/ : SUCCESS in 6m 42s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/1/check/openstack-tox-py36/acdf295/ : SUCCESS in 6m 17s\n- openstack-tox-docs http://logs.openstack.org/36/635236/1/check/openstack-tox-docs/a7a07b1/html/ : SUCCESS in 8m 35s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/1/check/build-openstack-releasenotes/9a392d9/html/ : SUCCESS in 6m 30s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/1/check/neutron-lbaasv2-dsvm-api/3590673/ : SUCCESS in 1h 20m 56s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/1/check/neutron-lbaasv2-dsvm-py3x-api/9225188/ : SUCCESS in 1h 18m 29s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/1/check/neutron-lbaasv2-dsvm-scenario/99075c8/ : SUCCESS in 1h 32m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/1/check/neutron-lbaasv2-dsvm-py3x-scenario/84a8acd/ : SUCCESS in 1h 22m 58s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/1/check/openstack-tox-functional/dd10df0/ : SUCCESS in 7m 17s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/635236/1/check/openstack-tox-functional-py35/6e224b1/ : SUCCESS in 8m 11s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/1/check/octavia-v1-dsvm-scenario/50dbd2e/ : SUCCESS in 1h 15m 53s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/1/check/octavia-v1-dsvm-py3x-scenario/3d1e512/ : SUCCESS in 1h 01m 30s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/1/check/octavia-v2-dsvm-noop-api/d94d287/ : SUCCESS in 39m 29s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/1/check/octavia-v2-dsvm-noop-py2-api/fe153f6/ : SUCCESS in 40m 45s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/1/check/octavia-v2-dsvm-scenario/8346eb2/ : SUCCESS in 1h 36m 37s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/1/check/octavia-v2-dsvm-py2-scenario/495eb41/ : SUCCESS in 1h 35m 45s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/1/check/octavia-v2-dsvm-py2-scenario-centos-7/761c2f0/ : FAILURE in 51m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/04201a6/ : SUCCESS in 1h 32m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/1/check/octavia-v2-dsvm-scenario-two-node/629d405/ : FAILURE in 2h 03m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/1/check/octavia-v2-dsvm-py2-scenario-two-node/06de115/ : FAILURE in 1h 26m 09s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/1/check/octavia-v2-dsvm-tls-barbican/777262e/ : SUCCESS in 49m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99b438d0fb7cfe3457aadab3c2d89540dcdae4c2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-07 22:25:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3af24ca205569210d0c48b2f742f389e997bc651","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-07 22:25:16.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"e9caec65065c0b969154e1d4dd1a08aec74d2603","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-08 01:06:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/2/check/build-openstack-api-ref/7cad7fb/html/ : SUCCESS in 7m 56s\n- requirements-check http://logs.openstack.org/36/635236/2/check/requirements-check/a7335c2/ : SUCCESS in 6m 44s\n- openstack-tox-cover http://logs.openstack.org/36/635236/2/check/openstack-tox-cover/7a4fca4/ : FAILURE in 12m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/2/check/openstack-tox-lower-constraints/683d4e4/ : FAILURE in 11m 52s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/2/check/openstack-tox-pep8/2dc523b/ : FAILURE in 13m 24s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/2/check/openstack-tox-py27/0d918ca/ : SUCCESS in 8m 03s\n- openstack-tox-py35 http://logs.openstack.org/36/635236/2/check/openstack-tox-py35/02c588b/ : SUCCESS in 7m 26s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/2/check/openstack-tox-py36/81d122f/ : SUCCESS in 8m 19s\n- openstack-tox-docs http://logs.openstack.org/36/635236/2/check/openstack-tox-docs/ebb9aea/html/ : SUCCESS in 6m 55s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/2/check/build-openstack-releasenotes/439d282/html/ : SUCCESS in 5m 36s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/2/check/neutron-lbaasv2-dsvm-api/8247d5c/ : SUCCESS in 1h 15m 47s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/2/check/neutron-lbaasv2-dsvm-py3x-api/1f68e20/ : SUCCESS in 1h 18m 37s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/2/check/neutron-lbaasv2-dsvm-scenario/1079e66/ : FAILURE in 1h 42m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/2/check/neutron-lbaasv2-dsvm-py3x-scenario/a26ef50/ : FAILURE in 2h 00m 07s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/2/check/openstack-tox-functional/820852b/ : FAILURE in 8m 18s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/635236/2/check/openstack-tox-functional-py35/2af1043/ : FAILURE in 8m 35s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/2/check/octavia-v1-dsvm-scenario/3be6647/ : FAILURE in 1h 19m 49s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/2/check/octavia-v1-dsvm-py3x-scenario/50b2660/ : FAILURE in 1h 46m 02s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/2/check/octavia-v2-dsvm-noop-api/5a66d03/ : SUCCESS in 58m 00s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/2/check/octavia-v2-dsvm-noop-py2-api/b5a4b19/ : SUCCESS in 40m 16s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/2/check/octavia-v2-dsvm-scenario/a09fed8/ : TIMED_OUT in 2h 32m 17s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/2/check/octavia-v2-dsvm-py2-scenario/7177503/ : TIMED_OUT in 2h 32m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/2/check/octavia-v2-dsvm-py2-scenario-centos-7/1f6b506/ : TIMED_OUT in 2h 33m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/cc0aee4/ : TIMED_OUT in 2h 32m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/2/check/octavia-v2-dsvm-scenario-two-node/5bd4340/ : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/2/check/octavia-v2-dsvm-py2-scenario-two-node/0049a30/ : TIMED_OUT in 2h 33m 23s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/2/check/octavia-grenade/5060793/ : SUCCESS in 1h 04m 56s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/2/check/octavia-v2-dsvm-tls-barbican/4113ab3/ : FAILURE in 1h 05m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e6df8526c17c5491f5402b28b479882300e4eecc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-09 01:09:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"23ae119887aad0305777597f83af1e486fc95946","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-09 01:09:17.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  Job openstack-tox-functional-py37 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python35-jobs\n        - openstack-python36-jobs\n        - openstack-python37-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - neutron-lbaasv2-dsvm-api\n          - neutron-lbaasv2-dsvm-py3x-api\n          - neutron-lbaasv2-dsvm-scenario:\n              voting: false\n          - neutron-lbaasv2-dsvm-py3x-scenario:\n              voting: false\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py35:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py37:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v1-dsvm-scenario\n          - octavia-v1-dsvm-py3x-scenario\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-v2-dsvm-py2-scenario-centos-7:\n              voting: false\n          - octavia-v2-dsvm-scenario-ubuntu-bionic:\n              voting: false\n          - octavia-v2-dsvm-scenario-two-node:\n              voting: false\n          - octavia-v2-dsvm-py2-scenario-two-node:\n              voting: false\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n          - octavia-v2-dsvm-tls-barbican:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n              voting: false\n      gate:\n        queue: octavia\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py35:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py37:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v1-dsvm-scenario\n          - octavia-v1-dsvm-py3x-scenario\n          - neutron-lbaasv2-dsvm-api\n          - neutron-lbaasv2-dsvm-py3x-api\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-ubuntu:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-fedora:\n              branches: ^(?!stable/.*).*$\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":3},{"id":"3ff83d7bcdda9bcdb960191566ccd71c0b7c21d0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-09 01:12:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"71111012f18d3dbcbef2666bc11bc8950f32ce7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-09 03:53:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/4/check/build-openstack-api-ref/cb7d0bb/html/ : SUCCESS in 6m 30s\n- requirements-check http://logs.openstack.org/36/635236/4/check/requirements-check/4043481/ : SUCCESS in 3m 31s\n- openstack-tox-cover http://logs.openstack.org/36/635236/4/check/openstack-tox-cover/fb43f9a/cover/ : SUCCESS in 9m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/4/check/openstack-tox-lower-constraints/c0d8a62/ : SUCCESS in 8m 04s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/4/check/openstack-tox-pep8/c24e68d/ : FAILURE in 7m 51s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/4/check/openstack-tox-py27/52ed2db/ : SUCCESS in 6m 08s\n- openstack-tox-py35 http://logs.openstack.org/36/635236/4/check/openstack-tox-py35/63590d5/ : SUCCESS in 6m 34s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/4/check/openstack-tox-py36/6cdc21b/ : SUCCESS in 5m 27s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/4/check/openstack-tox-py37/923c89c/ : SUCCESS in 7m 42s\n- openstack-tox-docs http://logs.openstack.org/36/635236/4/check/openstack-tox-docs/d6e7fea/html/ : SUCCESS in 7m 28s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/4/check/build-openstack-releasenotes/df05761/html/ : SUCCESS in 5m 18s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/4/check/neutron-lbaasv2-dsvm-api/288ad6f/ : SUCCESS in 1h 19m 07s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/4/check/neutron-lbaasv2-dsvm-py3x-api/48c04ae/ : SUCCESS in 1h 23m 17s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/4/check/neutron-lbaasv2-dsvm-scenario/5eb7dea/ : FAILURE in 1h 43m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/4/check/neutron-lbaasv2-dsvm-py3x-scenario/17eb9d5/ : FAILURE in 1h 54m 48s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/4/check/openstack-tox-functional/edd99cb/ : SUCCESS in 8m 05s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/635236/4/check/openstack-tox-functional-py35/d43131c/ : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/4/check/openstack-tox-functional-py36/753c7b2/ : SUCCESS in 7m 56s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/4/check/octavia-v1-dsvm-scenario/bd59e4e/ : FAILURE in 1h 18m 25s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/4/check/octavia-v1-dsvm-py3x-scenario/d641281/ : FAILURE in 1h 20m 44s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/4/check/octavia-v2-dsvm-noop-api/532d727/ : SUCCESS in 41m 07s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/4/check/octavia-v2-dsvm-noop-py2-api/5d6dd43/ : SUCCESS in 40m 32s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/4/check/octavia-v2-dsvm-scenario/e8f8c32/ : TIMED_OUT in 2h 32m 40s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/4/check/octavia-v2-dsvm-py2-scenario/1754f49/ : TIMED_OUT in 2h 32m 41s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/4/check/octavia-v2-dsvm-py2-scenario-centos-7/783b578/ : TIMED_OUT in 2h 33m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/871237e/ : TIMED_OUT in 2h 32m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/4/check/octavia-v2-dsvm-scenario-two-node/a1c5127/ : TIMED_OUT in 2h 33m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/4/check/octavia-v2-dsvm-py2-scenario-two-node/3bcb16e/ : TIMED_OUT in 2h 33m 22s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/4/check/octavia-grenade/e3eeb63/ : SUCCESS in 1h 14m 18s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/4/check/octavia-v2-dsvm-tls-barbican/f49f20d/ : FAILURE in 1h 16m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a4b97ff4f60d0c9ae2ce71c4879a8cd5373c0f00","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-09 11:20:39.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"8fa63adb88bc3d597934724c0add5743138130db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-09 14:01:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/5/check/build-openstack-api-ref/4c2b6db/html/ : SUCCESS in 4m 27s\n- requirements-check http://logs.openstack.org/36/635236/5/check/requirements-check/9a938ec/ : SUCCESS in 2m 56s\n- openstack-tox-cover http://logs.openstack.org/36/635236/5/check/openstack-tox-cover/4f0c004/cover/ : SUCCESS in 7m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/5/check/openstack-tox-lower-constraints/587731d/ : SUCCESS in 6m 26s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/5/check/openstack-tox-pep8/81a2cb0/ : FAILURE in 8m 31s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/5/check/openstack-tox-py27/ddb6ca9/ : SUCCESS in 5m 30s\n- openstack-tox-py35 http://logs.openstack.org/36/635236/5/check/openstack-tox-py35/0856a1f/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/5/check/openstack-tox-py36/1b84c90/ : SUCCESS in 5m 10s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/5/check/openstack-tox-py37/6c3ead8/ : SUCCESS in 7m 19s\n- openstack-tox-docs http://logs.openstack.org/36/635236/5/check/openstack-tox-docs/576c214/html/ : SUCCESS in 5m 53s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/5/check/build-openstack-releasenotes/8297d6f/html/ : SUCCESS in 4m 10s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/5/check/neutron-lbaasv2-dsvm-api/fe9c5fc/ : SUCCESS in 1h 13m 12s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/5/check/neutron-lbaasv2-dsvm-py3x-api/c72f631/ : SUCCESS in 1h 20m 53s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/5/check/neutron-lbaasv2-dsvm-scenario/f34fcbc/ : FAILURE in 1h 44m 45s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/5/check/neutron-lbaasv2-dsvm-py3x-scenario/b654745/ : FAILURE in 1h 30m 51s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/5/check/openstack-tox-functional/88bfe09/ : SUCCESS in 5m 57s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/635236/5/check/openstack-tox-functional-py35/d790f18/ : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/5/check/openstack-tox-functional-py36/92f6881/ : SUCCESS in 6m 18s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/5/check/octavia-v1-dsvm-scenario/668a638/ : FAILURE in 1h 17m 39s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/5/check/octavia-v1-dsvm-py3x-scenario/8c25389/ : FAILURE in 1h 17m 45s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/5/check/octavia-v2-dsvm-noop-api/a1bb652/ : SUCCESS in 38m 11s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/5/check/octavia-v2-dsvm-noop-py2-api/7d169ab/ : SUCCESS in 39m 04s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/5/check/octavia-v2-dsvm-scenario/3a263c7/ : TIMED_OUT in 2h 33m 14s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/5/check/octavia-v2-dsvm-py2-scenario/232d3bc/ : TIMED_OUT in 2h 33m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/5/check/octavia-v2-dsvm-py2-scenario-centos-7/aad794d/ : TIMED_OUT in 2h 33m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/9ae5520/ : TIMED_OUT in 2h 32m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/5/check/octavia-v2-dsvm-scenario-two-node/12bc1a3/ : TIMED_OUT in 2h 32m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/5/check/octavia-v2-dsvm-py2-scenario-two-node/ee7bf66/ : TIMED_OUT in 2h 33m 29s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/5/check/octavia-grenade/25f4c5e/ : SUCCESS in 1h 13m 31s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/5/check/octavia-v2-dsvm-tls-barbican/90604d5/ : FAILURE in 1h 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"57e1e8b753198b73486b596483658656cadd8ce1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-09 16:32:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"61252961f89e82b69366706fe18c919e7fac013a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-09 18:46:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/6/check/build-openstack-api-ref/0945f9c/html/ : SUCCESS in 5m 07s\n- requirements-check http://logs.openstack.org/36/635236/6/check/requirements-check/26295f3/ : SUCCESS in 3m 04s\n- openstack-tox-cover http://logs.openstack.org/36/635236/6/check/openstack-tox-cover/ab1c257/cover/ : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/6/check/openstack-tox-lower-constraints/bb1ef19/ : SUCCESS in 7m 24s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/6/check/openstack-tox-pep8/2a666dc/ : SUCCESS in 7m 43s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/6/check/openstack-tox-py27/779c730/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/36/635236/6/check/openstack-tox-py35/f467062/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/6/check/openstack-tox-py36/5633bb2/ : SUCCESS in 5m 03s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/6/check/openstack-tox-py37/6fc145f/ : SUCCESS in 5m 17s\n- openstack-tox-docs http://logs.openstack.org/36/635236/6/check/openstack-tox-docs/d66c562/html/ : SUCCESS in 6m 52s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/6/check/build-openstack-releasenotes/18f4e47/html/ : SUCCESS in 4m 50s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/6/check/neutron-lbaasv2-dsvm-api/f351369/ : SUCCESS in 1h 19m 42s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/6/check/neutron-lbaasv2-dsvm-py3x-api/a9ffa77/ : SUCCESS in 1h 20m 32s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/6/check/neutron-lbaasv2-dsvm-scenario/502d539/ : SUCCESS in 1h 26m 54s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/6/check/neutron-lbaasv2-dsvm-py3x-scenario/74b059c/ : SUCCESS in 1h 36m 35s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/6/check/openstack-tox-functional/05109b7/ : SUCCESS in 6m 46s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/635236/6/check/openstack-tox-functional-py35/8f4196c/ : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/6/check/openstack-tox-functional-py36/fad2053/ : SUCCESS in 6m 49s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/6/check/octavia-v1-dsvm-scenario/8c49d62/ : SUCCESS in 1h 00m 01s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/6/check/octavia-v1-dsvm-py3x-scenario/3e9a53c/ : SUCCESS in 1h 10m 16s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/6/check/octavia-v2-dsvm-noop-api/879c172/ : SUCCESS in 36m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/6/check/octavia-v2-dsvm-noop-py2-api/ae814e9/ : SUCCESS in 37m 57s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/6/check/octavia-v2-dsvm-scenario/f4efc3f/ : SUCCESS in 1h 25m 33s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/6/check/octavia-v2-dsvm-py2-scenario/276a4c1/ : SUCCESS in 1h 20m 57s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/6/check/octavia-v2-dsvm-py2-scenario-centos-7/1dfa8da/ : SUCCESS in 2h 11m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/98debdf/ : SUCCESS in 1h 17m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/6/check/octavia-v2-dsvm-scenario-two-node/1f6c146/ : SUCCESS in 1h 30m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/6/check/octavia-v2-dsvm-py2-scenario-two-node/3a2b0ea/ : FAILURE in 1h 08m 39s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/6/check/octavia-grenade/2e32638/ : SUCCESS in 1h 07m 33s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/6/check/octavia-v2-dsvm-tls-barbican/9ea48f9/ : SUCCESS in 42m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"51ec22c643fa85300826d9acc53a00fa5f9f5ef0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-13 20:01:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c0d14518490dba969fdde51565391b0a97f5e418","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-13 22:26:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/7/check/build-openstack-api-ref/2c1699e/html/ : SUCCESS in 4m 55s\n- requirements-check http://logs.openstack.org/36/635236/7/check/requirements-check/a94f2b6/ : SUCCESS in 3m 39s\n- openstack-tox-cover http://logs.openstack.org/36/635236/7/check/openstack-tox-cover/ad966c6/cover/ : SUCCESS in 8m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/7/check/openstack-tox-lower-constraints/398f990/ : SUCCESS in 8m 04s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/7/check/openstack-tox-pep8/b24751c/ : SUCCESS in 9m 01s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/7/check/openstack-tox-py27/285a883/ : SUCCESS in 7m 25s\n- openstack-tox-py35 http://logs.openstack.org/36/635236/7/check/openstack-tox-py35/f2f1e03/ : SUCCESS in 5m 47s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/7/check/openstack-tox-py36/dd7e0e0/ : SUCCESS in 5m 47s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/7/check/openstack-tox-py37/3fdb7f2/ : SUCCESS in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/36/635236/7/check/openstack-tox-docs/32baea8/html/ : SUCCESS in 6m 35s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/7/check/build-openstack-releasenotes/9d566bf/html/ : SUCCESS in 5m 10s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/7/check/neutron-lbaasv2-dsvm-api/c656907/ : SUCCESS in 1h 17m 37s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/7/check/neutron-lbaasv2-dsvm-py3x-api/23e452c/ : SUCCESS in 1h 19m 48s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/7/check/neutron-lbaasv2-dsvm-scenario/d267063/ : FAILURE in 1h 41m 15s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/7/check/neutron-lbaasv2-dsvm-py3x-scenario/a330acc/ : SUCCESS in 1h 43m 01s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/7/check/openstack-tox-functional/495f751/ : SUCCESS in 7m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/635236/7/check/openstack-tox-functional-py35/b58743e/ : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/7/check/openstack-tox-functional-py36/aeb59cf/ : SUCCESS in 8m 56s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/7/check/octavia-v1-dsvm-scenario/19055b5/ : SUCCESS in 1h 08m 31s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/7/check/octavia-v1-dsvm-py3x-scenario/3d8c35f/ : SUCCESS in 59m 59s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/7/check/octavia-v2-dsvm-noop-api/d33749d/ : SUCCESS in 44m 08s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/7/check/octavia-v2-dsvm-noop-py2-api/89644c0/ : SUCCESS in 41m 50s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/7/check/octavia-v2-dsvm-scenario/e1bc4ce/ : SUCCESS in 1h 41m 48s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/7/check/octavia-v2-dsvm-py2-scenario/d25a72a/ : SUCCESS in 1h 38m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/7/check/octavia-v2-dsvm-py2-scenario-centos-7/9529cef/ : SUCCESS in 2h 20m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/7/check/octavia-v2-dsvm-scenario-ubuntu-bionic/25cf06c/ : SUCCESS in 1h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/7/check/octavia-v2-dsvm-scenario-two-node/99926ab/ : SUCCESS in 1h 49m 56s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/7/check/octavia-v2-dsvm-py2-scenario-two-node/c04379d/ : FAILURE in 1h 26m 57s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/7/check/octavia-grenade/ca931c5/ : SUCCESS in 1h 11m 02s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/7/check/octavia-v2-dsvm-tls-barbican/40f95ff/ : SUCCESS in 44m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"66c61125397204160fdfc5a0a4b3e502e7b0c604","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-02-15 22:08:43.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"b98ba75ed955fb3e2271a3e405171cb1fff0b97a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-16 00:54:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/8/check/build-openstack-api-ref/15127ba/html/ : SUCCESS in 5m 21s\n- requirements-check http://logs.openstack.org/36/635236/8/check/requirements-check/75315ae/ : SUCCESS in 5m 49s\n- openstack-tox-cover http://logs.openstack.org/36/635236/8/check/openstack-tox-cover/f2b5074/cover/ : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/8/check/openstack-tox-lower-constraints/4d361ac/ : SUCCESS in 7m 56s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/8/check/openstack-tox-pep8/9371c90/ : SUCCESS in 10m 35s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/8/check/openstack-tox-py27/7a89573/ : SUCCESS in 5m 42s\n- openstack-tox-py35 http://logs.openstack.org/36/635236/8/check/openstack-tox-py35/b538daf/ : SUCCESS in 7m 06s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/8/check/openstack-tox-py36/ccedf05/ : SUCCESS in 5m 43s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/8/check/openstack-tox-py37/83d26f2/ : SUCCESS in 7m 02s\n- openstack-tox-docs http://logs.openstack.org/36/635236/8/check/openstack-tox-docs/11b7a0b/html/ : SUCCESS in 7m 20s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/8/check/build-openstack-releasenotes/dc9db5f/html/ : SUCCESS in 6m 14s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/8/check/neutron-lbaasv2-dsvm-api/3c19515/ : SUCCESS in 1h 23m 04s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/8/check/neutron-lbaasv2-dsvm-py3x-api/1aeecf5/ : SUCCESS in 1h 22m 03s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/8/check/neutron-lbaasv2-dsvm-scenario/d9a00ad/ : SUCCESS in 1h 35m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/8/check/neutron-lbaasv2-dsvm-py3x-scenario/545c568/ : SUCCESS in 1h 40m 31s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/8/check/openstack-tox-functional/1d97f62/ : SUCCESS in 6m 44s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/635236/8/check/openstack-tox-functional-py35/e66a122/ : SUCCESS in 9m 20s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/8/check/openstack-tox-functional-py36/44572b3/ : SUCCESS in 6m 57s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/8/check/octavia-v1-dsvm-scenario/05a3a96/ : SUCCESS in 59m 03s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/8/check/octavia-v1-dsvm-py3x-scenario/695eb82/ : SUCCESS in 1h 09m 18s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/8/check/octavia-v2-dsvm-noop-api/0a793ee/ : SUCCESS in 39m 16s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/8/check/octavia-v2-dsvm-noop-py2-api/2a92a04/ : SUCCESS in 1h 19m 34s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/8/check/octavia-v2-dsvm-scenario/a4521cf/ : SUCCESS in 1h 37m 35s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/8/check/octavia-v2-dsvm-py2-scenario/bc91aee/ : SUCCESS in 1h 25m 48s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/8/check/octavia-v2-dsvm-py2-scenario-centos-7/b9f485c/ : FAILURE in 2h 30m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/8/check/octavia-v2-dsvm-scenario-ubuntu-bionic/17b199b/ : SUCCESS in 1h 35m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/8/check/octavia-v2-dsvm-scenario-two-node/e3bca85/ : FAILURE in 1h 50m 14s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/8/check/octavia-v2-dsvm-py2-scenario-two-node/0af6439/ : FAILURE in 1h 12m 46s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/8/check/octavia-grenade/d4483ce/ : SUCCESS in 1h 12m 13s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/8/check/octavia-v2-dsvm-tls-barbican/3f5a6dd/ : SUCCESS in 51m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"44d78affe5887e9d3e9cd67d9e6979fea13b15a9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-22 15:09:24.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6f9caaf212134806fcfa2c352511fa7db6211e44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-22 17:37:26.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/9/check/build-openstack-api-ref/08b0900/html/ : SUCCESS in 5m 28s\n- requirements-check http://logs.openstack.org/36/635236/9/check/requirements-check/76c1bf3/ : SUCCESS in 3m 52s\n- openstack-tox-cover http://logs.openstack.org/36/635236/9/check/openstack-tox-cover/e93152d/cover/ : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/9/check/openstack-tox-lower-constraints/30572e3/ : SUCCESS in 8m 25s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/9/check/openstack-tox-pep8/8858956/ : SUCCESS in 7m 12s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/9/check/openstack-tox-py27/2e69f0f/ : SUCCESS in 7m 13s\n- openstack-tox-py35 http://logs.openstack.org/36/635236/9/check/openstack-tox-py35/f500412/ : SUCCESS in 6m 31s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/9/check/openstack-tox-py36/ac16fc0/ : SUCCESS in 6m 02s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/9/check/openstack-tox-py37/c5c4eca/ : SUCCESS in 6m 23s\n- openstack-tox-docs http://logs.openstack.org/36/635236/9/check/openstack-tox-docs/f6cea21/html/ : SUCCESS in 6m 47s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/9/check/build-openstack-releasenotes/63e7e71/html/ : SUCCESS in 5m 15s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/9/check/neutron-lbaasv2-dsvm-api/8fa77b7/ : SUCCESS in 1h 20m 39s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/9/check/neutron-lbaasv2-dsvm-py3x-api/aff875e/ : SUCCESS in 1h 20m 06s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/9/check/neutron-lbaasv2-dsvm-scenario/fa45e88/ : SUCCESS in 1h 24m 35s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/9/check/neutron-lbaasv2-dsvm-py3x-scenario/e466088/ : SUCCESS in 1h 31m 43s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/9/check/openstack-tox-functional/a500f1b/ : SUCCESS in 7m 14s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/635236/9/check/openstack-tox-functional-py35/3f1aaae/ : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/9/check/openstack-tox-functional-py36/65d0b97/ : SUCCESS in 8m 14s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/9/check/octavia-v1-dsvm-scenario/39ffc4d/ : SUCCESS in 58m 08s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/9/check/octavia-v1-dsvm-py3x-scenario/77948dd/ : SUCCESS in 56m 40s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/9/check/octavia-v2-dsvm-noop-api/eb01e16/ : SUCCESS in 55m 22s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/9/check/octavia-v2-dsvm-noop-py2-api/de311c1/ : SUCCESS in 39m 17s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/9/check/octavia-v2-dsvm-scenario/4ed067e/ : SUCCESS in 1h 40m 09s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/9/check/octavia-v2-dsvm-py2-scenario/223271f/ : SUCCESS in 1h 37m 56s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/9/check/octavia-v2-dsvm-py2-scenario-centos-7/01cff90/ : FAILURE in 1h 54m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/9/check/octavia-v2-dsvm-scenario-ubuntu-bionic/8a09bcd/ : FAILURE in 2h 04m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/9/check/octavia-v2-dsvm-scenario-two-node/658bda5/ : SUCCESS in 1h 54m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/9/check/octavia-v2-dsvm-py2-scenario-two-node/d3a0c0b/ : SUCCESS in 1h 37m 15s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/9/check/octavia-grenade/075a028/ : SUCCESS in 1h 19m 56s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/9/check/octavia-v2-dsvm-tls-barbican/35056af/ : SUCCESS in 58m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"528835f64f04e814227196db371472297a9b4ed6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-08 17:16:19.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c762ba5f9d1ea17f149bd865b18a45a45e9bf415","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 19:35:16.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/10/check/build-openstack-api-ref/5424a7a/html/ : SUCCESS in 7m 01s\n- requirements-check http://logs.openstack.org/36/635236/10/check/requirements-check/5fe39c5/ : SUCCESS in 4m 22s\n- openstack-tox-cover http://logs.openstack.org/36/635236/10/check/openstack-tox-cover/df7e6da/cover/ : SUCCESS in 11m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/10/check/openstack-tox-lower-constraints/229d4de/ : SUCCESS in 9m 31s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/10/check/openstack-tox-pep8/c71808a/ : SUCCESS in 8m 24s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/10/check/openstack-tox-py27/4db0a1f/ : SUCCESS in 7m 52s\n- openstack-tox-py35 http://logs.openstack.org/36/635236/10/check/openstack-tox-py35/f0b0005/ : SUCCESS in 7m 43s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/10/check/openstack-tox-py36/ed08adf/ : SUCCESS in 7m 22s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/10/check/openstack-tox-py37/0b534f7/ : SUCCESS in 6m 37s\n- openstack-tox-docs http://logs.openstack.org/36/635236/10/check/openstack-tox-docs/cead791/html/ : SUCCESS in 7m 42s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/10/check/build-openstack-releasenotes/7f917fd/html/ : SUCCESS in 6m 23s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/10/check/neutron-lbaasv2-dsvm-api/bbc214d/ : SUCCESS in 1h 20m 02s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/10/check/neutron-lbaasv2-dsvm-py3x-api/46c6ad9/ : SUCCESS in 1h 18m 19s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/10/check/neutron-lbaasv2-dsvm-scenario/34eeedc/ : SUCCESS in 59m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/10/check/neutron-lbaasv2-dsvm-py3x-scenario/1e8d6bc/ : SUCCESS in 1h 32m 40s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/10/check/openstack-tox-functional/3f51f4c/ : SUCCESS in 9m 03s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/635236/10/check/openstack-tox-functional-py35/4b412cb/ : SUCCESS in 8m 49s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/10/check/openstack-tox-functional-py36/6820fdb/ : SUCCESS in 9m 36s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/10/check/octavia-v1-dsvm-scenario/57ae209/ : SUCCESS in 1h 11m 27s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/10/check/octavia-v1-dsvm-py3x-scenario/fc615a2/ : SUCCESS in 1h 18m 42s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/10/check/octavia-v2-dsvm-noop-api/96127cb/ : SUCCESS in 45m 56s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/10/check/octavia-v2-dsvm-noop-py2-api/cbf4f01/ : SUCCESS in 46m 34s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/10/check/octavia-v2-dsvm-scenario/b02e57a/ : SUCCESS in 1h 40m 20s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/10/check/octavia-v2-dsvm-py2-scenario/b8e8fc9/ : SUCCESS in 1h 31m 36s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/10/check/octavia-v2-dsvm-py2-scenario-centos-7/6514894/ : FAILURE in 2h 08m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/10/check/octavia-v2-dsvm-scenario-ubuntu-bionic/f5ad133/ : SUCCESS in 2h 05m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/10/check/octavia-v2-dsvm-scenario-two-node/c014276/ : SUCCESS in 1h 53m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/10/check/octavia-v2-dsvm-py2-scenario-two-node/b36b5cc/ : SUCCESS in 1h 33m 48s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/10/check/octavia-grenade/0f0058c/ : SUCCESS in 1h 10m 51s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/10/check/octavia-v2-dsvm-tls-barbican/9006209/ : SUCCESS in 1h 03m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ab5045581ec40a6d5351bb2c8e5a20242741abce","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-15 10:08:42.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"3f1a666fe162b85b06fb0671e9efaa4f6f55a165","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-15 10:13:40.000000000","message":"Patch Set 11:\n\nPython Runtime for Train: https://governance.openstack.org/tc/reference/runtimes/train.html#python-runtime-for-train\n\nShould we drop Python 3.5? I\u0027m in favor keeping it for as long as possible. Distros like Ubuntu 16.04 have 3.5 as default Python runtime.","accounts_in_message":[],"_revision_number":11},{"id":"a531adf7e70dfe60e9e4f8b446acfb4d897d41bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 12:21:53.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/11/check/build-openstack-api-ref/a2894d8/html/ : SUCCESS in 6m 51s\n- test-release-openstack http://logs.openstack.org/36/635236/11/check/test-release-openstack/c9cd536/ : SUCCESS in 5m 24s\n- requirements-check http://logs.openstack.org/36/635236/11/check/requirements-check/0dc18a6/ : FAILURE in 4m 42s\n- openstack-tox-cover http://logs.openstack.org/36/635236/11/check/openstack-tox-cover/98492f8/cover/ : SUCCESS in 8m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/11/check/openstack-tox-lower-constraints/d2a3a41/ : SUCCESS in 8m 48s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/11/check/openstack-tox-pep8/3786024/ : SUCCESS in 8m 01s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/11/check/openstack-tox-py27/5927146/ : SUCCESS in 5m 54s\n- openstack-tox-py35 http://logs.openstack.org/36/635236/11/check/openstack-tox-py35/4cc40b8/ : SUCCESS in 6m 07s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/11/check/openstack-tox-py36/6a2758a/ : SUCCESS in 8m 18s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/11/check/openstack-tox-py37/1734846/ : SUCCESS in 7m 28s\n- openstack-tox-docs http://logs.openstack.org/36/635236/11/check/openstack-tox-docs/1daa28d/html/ : SUCCESS in 7m 44s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/11/check/build-openstack-releasenotes/e310178/html/ : SUCCESS in 5m 45s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/11/check/neutron-lbaasv2-dsvm-api/913c2a3/ : SUCCESS in 1h 16m 12s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/11/check/neutron-lbaasv2-dsvm-py3x-api/05b0f09/ : SUCCESS in 1h 15m 48s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/11/check/neutron-lbaasv2-dsvm-scenario/3d9ca46/ : SUCCESS in 1h 20m 48s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/11/check/neutron-lbaasv2-dsvm-py3x-scenario/f341398/ : FAILURE in 1h 22m 14s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/11/check/openstack-tox-functional/2e1fe37/ : SUCCESS in 8m 16s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/635236/11/check/openstack-tox-functional-py35/c030240/ : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/11/check/openstack-tox-functional-py36/c7789e8/ : SUCCESS in 8m 04s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/11/check/octavia-v1-dsvm-scenario/88c2021/ : SUCCESS in 57m 17s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/11/check/octavia-v1-dsvm-py3x-scenario/18e7c9c/ : SUCCESS in 57m 58s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/11/check/octavia-v2-dsvm-noop-api/b1ace47/ : SUCCESS in 40m 25s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/11/check/octavia-v2-dsvm-noop-py2-api/4a90c70/ : SUCCESS in 44m 32s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/11/check/octavia-v2-dsvm-scenario/fbbdfff/ : SUCCESS in 1h 26m 35s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/11/check/octavia-v2-dsvm-py2-scenario/2456a74/ : SUCCESS in 1h 32m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/11/check/octavia-v2-dsvm-py2-scenario-centos-7/788179b/ : SUCCESS in 2h 08m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/11/check/octavia-v2-dsvm-scenario-ubuntu-bionic/3e1dd1e/ : SUCCESS in 1h 33m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/11/check/octavia-v2-dsvm-scenario-two-node/9a5136a/ : FAILURE in 1h 48m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/11/check/octavia-v2-dsvm-py2-scenario-two-node/54a1c47/ : SUCCESS in 1h 32m 38s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/11/check/octavia-grenade/45313dc/ : SUCCESS in 1h 08m 16s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/11/check/octavia-v2-dsvm-tls-barbican/ff30b1c/ : SUCCESS in 49m 33s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"66f5e5c19fcb9dea4bbd8987bad8c6e11a6c8938","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-17 10:29:11.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"bc8d21b208c88c17bebbcd39ee4b736275bbc2a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 13:12:24.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/12/check/build-openstack-api-ref/c361a35/html/ : SUCCESS in 5m 07s\n- test-release-openstack http://logs.openstack.org/36/635236/12/check/test-release-openstack/5beab8a/ : SUCCESS in 4m 26s\n- requirements-check http://logs.openstack.org/36/635236/12/check/requirements-check/27dbe49/ : SUCCESS in 3m 46s\n- openstack-tox-cover http://logs.openstack.org/36/635236/12/check/openstack-tox-cover/7cee3ae/cover/ : SUCCESS in 8m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/12/check/openstack-tox-lower-constraints/f09a38f/ : SUCCESS in 8m 57s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/12/check/openstack-tox-pep8/8d29710/ : SUCCESS in 7m 17s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/12/check/openstack-tox-py27/7c7720c/ : SUCCESS in 5m 40s\n- openstack-tox-py35 http://logs.openstack.org/36/635236/12/check/openstack-tox-py35/3adb605/ : SUCCESS in 5m 48s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/12/check/openstack-tox-py36/1fffa80/ : SUCCESS in 6m 02s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/12/check/openstack-tox-py37/12f05ce/ : SUCCESS in 6m 12s\n- openstack-tox-docs http://logs.openstack.org/36/635236/12/check/openstack-tox-docs/04336dc/html/ : SUCCESS in 7m 37s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/12/check/build-openstack-releasenotes/5762649/html/ : SUCCESS in 5m 28s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/12/check/neutron-lbaasv2-dsvm-api/5b24b6b/ : SUCCESS in 1h 19m 48s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/12/check/neutron-lbaasv2-dsvm-py3x-api/6324d19/ : SUCCESS in 1h 23m 02s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/12/check/neutron-lbaasv2-dsvm-scenario/487cc22/ : SUCCESS in 1h 28m 59s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/12/check/neutron-lbaasv2-dsvm-py3x-scenario/423e4b9/ : SUCCESS in 1h 29m 25s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/12/check/openstack-tox-functional/2982a0f/ : SUCCESS in 7m 42s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/635236/12/check/openstack-tox-functional-py35/7bcef54/ : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/12/check/openstack-tox-functional-py36/e39c3d1/ : SUCCESS in 8m 18s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/12/check/octavia-v1-dsvm-scenario/5604ce9/ : SUCCESS in 1h 22m 46s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/12/check/octavia-v1-dsvm-py3x-scenario/4d8b26e/ : FAILURE in 37m 50s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/12/check/octavia-v2-dsvm-noop-api/37fc895/ : SUCCESS in 49m 24s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/12/check/octavia-v2-dsvm-noop-py2-api/0209391/ : SUCCESS in 43m 39s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/12/check/octavia-v2-dsvm-scenario/17d2b50/ : SUCCESS in 1h 35m 03s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/12/check/octavia-v2-dsvm-py2-scenario/52ead23/ : SUCCESS in 1h 28m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/12/check/octavia-v2-dsvm-py2-scenario-centos-7/38a7a1f/ : FAILURE in 2h 07m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/12/check/octavia-v2-dsvm-scenario-ubuntu-bionic/9ded936/ : SUCCESS in 1h 45m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/12/check/octavia-v2-dsvm-scenario-two-node/62c098d/ : SUCCESS in 1h 55m 02s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/12/check/octavia-v2-dsvm-py2-scenario-two-node/39459da/ : SUCCESS in 1h 57m 34s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/12/check/octavia-grenade/25bdd74/ : SUCCESS in 1h 34m 22s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/12/check/octavia-v2-dsvm-tls-barbican/08a6384/ : SUCCESS in 1h 02m 25s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0334ab9275814b429663b7150e1701a45f332fc9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-18 00:16:49.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"806a5cec7a512e6f409552df2284507180f0ac54","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-18 00:22:17.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1967907d7c71e7cf86bbb7dcb85c43cfff367da4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 02:21:39.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/14/check/build-openstack-api-ref/0cffe6c/html/ : SUCCESS in 10m 14s\n- test-release-openstack http://logs.openstack.org/36/635236/14/check/test-release-openstack/9bd623a/ : SUCCESS in 4m 56s\n- requirements-check http://logs.openstack.org/36/635236/14/check/requirements-check/a28c606/ : SUCCESS in 4m 06s\n- openstack-tox-cover http://logs.openstack.org/36/635236/14/check/openstack-tox-cover/300a48a/cover/ : SUCCESS in 8m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/14/check/openstack-tox-lower-constraints/43fd92e/ : SUCCESS in 7m 55s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/14/check/openstack-tox-pep8/1d429d6/ : SUCCESS in 8m 24s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/14/check/openstack-tox-py27/3349e69/ : SUCCESS in 7m 22s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/14/check/openstack-tox-py36/270f4a1/ : SUCCESS in 6m 52s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/14/check/openstack-tox-py37/7a52fc7/ : SUCCESS in 6m 12s\n- openstack-tox-docs http://logs.openstack.org/36/635236/14/check/openstack-tox-docs/960d324/html/ : SUCCESS in 9m 14s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/14/check/build-openstack-releasenotes/e177162/html/ : SUCCESS in 7m 03s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/14/check/neutron-lbaasv2-dsvm-api/2273258/ : SUCCESS in 1h 22m 29s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/14/check/neutron-lbaasv2-dsvm-py3x-api/04e83a0/ : SUCCESS in 1h 21m 28s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/14/check/neutron-lbaasv2-dsvm-scenario/d406cb7/ : SUCCESS in 1h 26m 09s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/14/check/neutron-lbaasv2-dsvm-py3x-scenario/0c05ad3/ : SUCCESS in 1h 25m 59s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/14/check/openstack-tox-functional/00cee34/ : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/14/check/openstack-tox-functional-py36/e5c5052/ : SUCCESS in 9m 44s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/14/check/octavia-v1-dsvm-scenario/6eba254/ : SUCCESS in 1h 09m 09s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/14/check/octavia-v1-dsvm-py3x-scenario/55fda09/ : SUCCESS in 1h 06m 37s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/14/check/octavia-v2-dsvm-noop-api/9933d2c/ : SUCCESS in 43m 06s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/14/check/octavia-v2-dsvm-noop-py2-api/7de4acc/ : SUCCESS in 42m 52s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/14/check/octavia-v2-dsvm-scenario/d77bc9f/ : SUCCESS in 1h 32m 07s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/14/check/octavia-v2-dsvm-py2-scenario/acdf22c/ : SUCCESS in 1h 36m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/14/check/octavia-v2-dsvm-py2-scenario-centos-7/951d1c1/ : FAILURE in 30m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/14/check/octavia-v2-dsvm-scenario-ubuntu-bionic/8880bf4/ : SUCCESS in 1h 52m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/14/check/octavia-v2-dsvm-scenario-two-node/4cf05c5/ : SUCCESS in 1h 50m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/14/check/octavia-v2-dsvm-py2-scenario-two-node/c2e50d2/ : SUCCESS in 1h 46m 31s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/14/check/octavia-grenade/e26517e/ : SUCCESS in 1h 36m 32s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/14/check/octavia-v2-dsvm-tls-barbican/9d0b8f2/ : SUCCESS in 54m 55s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"7f84150a09daf3c6b5a97e6adfe4c89d73e2ebe1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-04-19 10:11:13.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"9225bcac33e5f86dc6a6c8b53ee7836dae74bbbd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-10 13:09:45.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"e2de2a2c9ee64539bca01d93126d73e210ebce7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:21:50.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/15/check/build-openstack-api-ref/97debfc/html/ : SUCCESS in 4m 43s\n- test-release-openstack http://logs.openstack.org/36/635236/15/check/test-release-openstack/615e13a/ : SUCCESS in 3m 13s\n- requirements-check http://logs.openstack.org/36/635236/15/check/requirements-check/2728876/ : SUCCESS in 3m 38s\n- openstack-tox-cover http://logs.openstack.org/36/635236/15/check/openstack-tox-cover/e16d5f0/cover/ : SUCCESS in 8m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/15/check/openstack-tox-lower-constraints/1e68d24/ : SUCCESS in 7m 54s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/15/check/openstack-tox-pep8/cb7e8b9/ : FAILURE in 7m 06s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/15/check/openstack-tox-py27/2632ee0/ : SUCCESS in 6m 05s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/15/check/openstack-tox-py36/5137033/ : SUCCESS in 6m 13s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/15/check/openstack-tox-py37/061d913/ : SUCCESS in 5m 52s\n- openstack-tox-docs http://logs.openstack.org/36/635236/15/check/openstack-tox-docs/9414dd4/html/ : SUCCESS in 6m 13s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/15/check/build-openstack-releasenotes/6c0140e/html/ : SUCCESS in 5m 17s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/635236/15/check/neutron-lbaasv2-dsvm-api/454321f/ : SUCCESS in 1h 22m 00s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/635236/15/check/neutron-lbaasv2-dsvm-py3x-api/93c63bc/ : SUCCESS in 1h 16m 43s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/635236/15/check/neutron-lbaasv2-dsvm-scenario/61f238e/ : SUCCESS in 1h 32m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/635236/15/check/neutron-lbaasv2-dsvm-py3x-scenario/dc8f57b/ : SUCCESS in 1h 21m 01s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/635236/15/check/openstack-tox-functional/310ef88/ : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/15/check/openstack-tox-functional-py36/8c0e20e/ : SUCCESS in 7m 38s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/635236/15/check/octavia-v1-dsvm-scenario/13ce553/ : SUCCESS in 1h 11m 28s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/635236/15/check/octavia-v1-dsvm-py3x-scenario/92bde8a/ : SUCCESS in 1h 06m 28s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/15/check/octavia-v2-dsvm-noop-api/9096897/ : SUCCESS in 44m 39s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/15/check/octavia-v2-dsvm-noop-py2-api/360ce40/ : SUCCESS in 47m 52s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/15/check/octavia-v2-dsvm-scenario/bb90c62/ : SUCCESS in 1h 29m 35s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/15/check/octavia-v2-dsvm-py2-scenario/6078908/ : SUCCESS in 1h 32m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/15/check/octavia-v2-dsvm-py2-scenario-centos-7/a75760d/ : FAILURE in 2h 07m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/15/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6fe2587/ : SUCCESS in 1h 44m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/15/check/octavia-v2-dsvm-scenario-two-node/2982a04/ : SUCCESS in 1h 59m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/15/check/octavia-v2-dsvm-py2-scenario-two-node/76dbfbc/ : SUCCESS in 1h 23m 30s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/15/check/octavia-grenade/fbbf6e1/ : SUCCESS in 1h 20m 35s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/15/check/octavia-v2-dsvm-tls-barbican/65ee2e1/ : SUCCESS in 1h 00m 21s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8cd160519e599f301450eb17530b4303d19db301","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-10 20:38:47.000000000","message":"Patch Set 15: Code-Review-1\n\nIMO we should rebase this on top of the v1 removal -- no sense in refactoring a bunch of code that\u0027s being deleted","accounts_in_message":[],"_revision_number":15},{"id":"db8ed28306624bee915ba12b143ef299202bebe8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-14 15:46:21.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"acd983b4e3f605db60040536a101853538db9c64","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-14 17:11:22.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"def658a0792a5c4ead108b50f6195f67256c920d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 20:09:48.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/17/check/build-openstack-api-ref/8ae51c5/html/ : SUCCESS in 5m 12s\n- test-release-openstack http://logs.openstack.org/36/635236/17/check/test-release-openstack/7bd1fe5/ : SUCCESS in 3m 19s\n- requirements-check http://logs.openstack.org/36/635236/17/check/requirements-check/7f1f332/ : SUCCESS in 4m 31s\n- openstack-tox-cover http://logs.openstack.org/36/635236/17/check/openstack-tox-cover/7ad7cc3/cover/ : SUCCESS in 9m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/17/check/openstack-tox-lower-constraints/9e86468/ : SUCCESS in 6m 25s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/17/check/openstack-tox-pep8/1eb269a/ : SUCCESS in 6m 50s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/17/check/openstack-tox-py27/ac3d586/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/17/check/openstack-tox-py36/d4080eb/ : SUCCESS in 5m 42s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/17/check/openstack-tox-py37/975efcc/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/36/635236/17/check/openstack-tox-docs/140b888/html/ : SUCCESS in 6m 02s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/17/check/build-openstack-releasenotes/85763d9/html/ : SUCCESS in 4m 57s\n- openstack-tox-functional http://logs.openstack.org/36/635236/17/check/openstack-tox-functional/84e3aae/ : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/17/check/openstack-tox-functional-py36/5f994ab/ : SUCCESS in 8m 50s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-noop-api/72ce71a/ : SUCCESS in 44m 34s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-noop-py2-api/3f678a6/ : SUCCESS in 43m 02s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-scenario/af1f041/ : SUCCESS in 1h 29m 08s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-py2-scenario/7161d40/ : SUCCESS in 1h 49m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-py2-scenario-centos-7/a322ce3/ : FAILURE in 1h 55m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-scenario-ubuntu-bionic/78c692c/ : SUCCESS in 1h 53m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-scenario-two-node/cddf0aa/ : TIMED_OUT in 2h 34m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-py2-scenario-two-node/832959b/ : FAILURE in 1h 26m 01s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/17/check/octavia-grenade/338de83/ : FAILURE in 33m 24s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-tls-barbican/12572dd/ : SUCCESS in 44m 18s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"dbc29079421e227de61dcdd74106c618061e08a2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-15 05:51:47.000000000","message":"Patch Set 17:\n\nrecheck\n\nFailed getting release file http://archive.ubuntu.com/ubuntu/dists/xenial/Release","accounts_in_message":[],"_revision_number":17},{"id":"0d86477be4a540ab3f531663743e676edfa194c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 08:04:32.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/17/check/build-openstack-api-ref/3a9a152/html/ : SUCCESS in 4m 59s\n- test-release-openstack http://logs.openstack.org/36/635236/17/check/test-release-openstack/e5c54cc/ : SUCCESS in 3m 04s\n- requirements-check http://logs.openstack.org/36/635236/17/check/requirements-check/4a17c4b/ : SUCCESS in 3m 07s\n- openstack-tox-cover http://logs.openstack.org/36/635236/17/check/openstack-tox-cover/29dbbaa/cover/ : SUCCESS in 7m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/17/check/openstack-tox-lower-constraints/71175d7/ : SUCCESS in 6m 41s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/17/check/openstack-tox-pep8/25e47f2/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/17/check/openstack-tox-py27/ae47b72/ : SUCCESS in 4m 52s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/17/check/openstack-tox-py36/bedf5eb/ : SUCCESS in 5m 33s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/17/check/openstack-tox-py37/25137fd/ : SUCCESS in 5m 50s\n- openstack-tox-docs http://logs.openstack.org/36/635236/17/check/openstack-tox-docs/a2f71f3/html/ : SUCCESS in 5m 33s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/17/check/build-openstack-releasenotes/0cecc7a/html/ : SUCCESS in 4m 41s\n- openstack-tox-functional http://logs.openstack.org/36/635236/17/check/openstack-tox-functional/e755606/ : SUCCESS in 7m 03s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/17/check/openstack-tox-functional-py36/b757163/ : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-noop-api/805dad7/ : SUCCESS in 43m 47s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-noop-py2-api/639a382/ : SUCCESS in 45m 22s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-scenario/0970ae2/ : SUCCESS in 1h 36m 42s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-py2-scenario/baaad10/ : SUCCESS in 1h 33m 34s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-py2-scenario-centos-7/b934ea0/ : FAILURE in 2h 07m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-scenario-ubuntu-bionic/69a9c2b/ : SUCCESS in 1h 47m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-scenario-two-node/1fc1aed/ : SUCCESS in 1h 51m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-py2-scenario-two-node/aa1106f/ : SUCCESS in 1h 48m 57s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/635236/17/check/octavia-grenade/0fec3b4/ : SUCCESS in 1h 11m 00s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/635236/17/check/octavia-v2-dsvm-tls-barbican/0be7717/ : SUCCESS in 44m 53s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"15176b86c82b32564b7b60cb6b1c4f63e06c15fb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-15 16:48:27.000000000","message":"Patch Set 17: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":17},{"id":"e0c2ebb46ff1306438d93be7147dad1b53eedc4b","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-05-16 04:47:45.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"e7fc3dcf8dfe9673dfce971a7346de00ff6587cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 04:48:05.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"890f3df60818a115bab86c15e55317e37be56549","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:30:27.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/36/635236/17/gate/build-openstack-api-ref/1317782/html/ : SUCCESS in 5m 17s\n- test-release-openstack http://logs.openstack.org/36/635236/17/gate/test-release-openstack/cf8c231/ : SUCCESS in 3m 10s\n- requirements-check http://logs.openstack.org/36/635236/17/gate/requirements-check/d8f513f/ : SUCCESS in 3m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/635236/17/gate/openstack-tox-lower-constraints/2a0fa8d/ : SUCCESS in 6m 17s\n- openstack-tox-pep8 http://logs.openstack.org/36/635236/17/gate/openstack-tox-pep8/7239a54/ : SUCCESS in 6m 57s\n- openstack-tox-py27 http://logs.openstack.org/36/635236/17/gate/openstack-tox-py27/9061b09/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/36/635236/17/gate/openstack-tox-py36/52f240e/ : SUCCESS in 6m 19s\n- openstack-tox-py37 http://logs.openstack.org/36/635236/17/gate/openstack-tox-py37/fcc8f6d/ : SUCCESS in 8m 27s\n- openstack-tox-docs http://logs.openstack.org/36/635236/17/gate/openstack-tox-docs/4a89331/html/ : SUCCESS in 6m 14s\n- build-openstack-releasenotes http://logs.openstack.org/36/635236/17/gate/build-openstack-releasenotes/362987e/html/ : SUCCESS in 4m 59s\n- openstack-tox-functional http://logs.openstack.org/36/635236/17/gate/openstack-tox-functional/0c4c28b/ : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/635236/17/gate/openstack-tox-functional-py36/65edf09/ : SUCCESS in 6m 03s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/635236/17/gate/octavia-v2-dsvm-noop-api/c5b287e/ : SUCCESS in 40m 16s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/635236/17/gate/octavia-v2-dsvm-noop-py2-api/452b8e6/ : SUCCESS in 44m 35s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/635236/17/gate/octavia-v2-dsvm-scenario/3b60381/ : SUCCESS in 1h 27m 40s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/635236/17/gate/octavia-v2-dsvm-py2-scenario/afd9eb8/ : SUCCESS in 1h 28m 15s\n- octavia-grenade http://logs.openstack.org/36/635236/17/gate/octavia-grenade/b806f3e/ : SUCCESS in 1h 13m 58s","accounts_in_message":[],"_revision_number":17},{"id":"0344072a80a23e7b5f3c14ced552103ec5f3dcb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:30:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"c4faac25de85ca3d8b4f12964589a36b2bcd3b57","revisions":{"9f7e46ac878bb41247e5f83f388a6634decc800c":{"kind":"REWORK","_number":1,"created":"2019-02-06 18:13:52.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/1"}}},"commit":{"parents":[{"commit":"ddcae3e2291fe4cdaca5b9ce53eed1c3311338e7","subject":"Merge \"Fix the amphora noop driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ddcae3e2291fe4cdaca5b9ce53eed1c3311338e7"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:13:18.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:13:18.000000000","tz":60},"subject":"Update pylint version","message":"Update pylint version\n\nPython 3.7 support was added in Pylint 2.0.0.\n\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f7e46ac878bb41247e5f83f388a6634decc800c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f7e46ac878bb41247e5f83f388a6634decc800c"}]},"branch":"refs/heads/master"},"b7b624b6ab92f7dc8aabcad4f9a4a33bb4ea5b6e":{"kind":"REWORK","_number":2,"created":"2019-02-07 22:25:04.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/2"}}},"commit":{"parents":[{"commit":"ddcae3e2291fe4cdaca5b9ce53eed1c3311338e7","subject":"Merge \"Fix the amphora noop driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ddcae3e2291fe4cdaca5b9ce53eed1c3311338e7"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-07 22:24:35.000000000","tz":60},"subject":"WIP: Update pylint version","message":"WIP: Update pylint version\n\nPylint 2.0.0 introduced support to Python 3.7 while dropping support to\nPython 2.\n\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7b624b6ab92f7dc8aabcad4f9a4a33bb4ea5b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7b624b6ab92f7dc8aabcad4f9a4a33bb4ea5b6e"}]},"branch":"refs/heads/master"},"30f7a039fe9d24d959fb9d3586ec9b0d75633189":{"kind":"REWORK","_number":3,"created":"2019-02-09 01:09:00.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/3"}}},"commit":{"parents":[{"commit":"b5dda3ce0570dbc888d77bfff9b13321eb33608c","subject":"Fix functional tests under Python \u003e\u003d 3.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5dda3ce0570dbc888d77bfff9b13321eb33608c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-09 01:06:18.000000000","tz":60},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30f7a039fe9d24d959fb9d3586ec9b0d75633189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30f7a039fe9d24d959fb9d3586ec9b0d75633189"}]},"branch":"refs/heads/master"},"8f81b7f4c3fb2c8e72dab35b982f081a7b30d1b9":{"kind":"REWORK","_number":4,"created":"2019-02-09 01:12:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/4"}}},"commit":{"parents":[{"commit":"b5dda3ce0570dbc888d77bfff9b13321eb33608c","subject":"Fix functional tests under Python \u003e\u003d 3.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5dda3ce0570dbc888d77bfff9b13321eb33608c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-09 01:12:30.000000000","tz":60},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f81b7f4c3fb2c8e72dab35b982f081a7b30d1b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f81b7f4c3fb2c8e72dab35b982f081a7b30d1b9"}]},"branch":"refs/heads/master"},"93060c8645649f6199f3e3b1aab806b7c386bf20":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-02-09 11:20:39.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/5"}}},"commit":{"parents":[{"commit":"fae5b059806d97781306ec81f8ee82f2b1289461","subject":"Merge \"Fix functional tests under Python \u003e\u003d 3.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fae5b059806d97781306ec81f8ee82f2b1289461"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-09 11:20:39.000000000","tz":0},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93060c8645649f6199f3e3b1aab806b7c386bf20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93060c8645649f6199f3e3b1aab806b7c386bf20"}]},"branch":"refs/heads/master"},"cc9d9269ecc6693029b9e69a027ef2375cd8e288":{"kind":"REWORK","_number":6,"created":"2019-02-09 16:32:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/6"}}},"commit":{"parents":[{"commit":"fae5b059806d97781306ec81f8ee82f2b1289461","subject":"Merge \"Fix functional tests under Python \u003e\u003d 3.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fae5b059806d97781306ec81f8ee82f2b1289461"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-09 16:30:26.000000000","tz":60},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc9d9269ecc6693029b9e69a027ef2375cd8e288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc9d9269ecc6693029b9e69a027ef2375cd8e288"}]},"branch":"refs/heads/master"},"96429822791bdd7d561b1bb5aea3658fb79ef212":{"kind":"REWORK","_number":7,"created":"2019-02-13 20:01:52.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/7"}}},"commit":{"parents":[{"commit":"f026372216ce45e3e5ddc95541e235cfd30a1a3e","subject":"Merge \"Refactor the pluggin of the VIP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f026372216ce45e3e5ddc95541e235cfd30a1a3e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-13 20:00:17.000000000","tz":60},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96429822791bdd7d561b1bb5aea3658fb79ef212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96429822791bdd7d561b1bb5aea3658fb79ef212"}]},"branch":"refs/heads/master"},"12d7b16136d12239a22a98c2ac6fa8bbb6b1cc40":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-02-15 22:08:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/36/635236/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/8"}}},"commit":{"parents":[{"commit":"f026372216ce45e3e5ddc95541e235cfd30a1a3e","subject":"Merge \"Refactor the pluggin of the VIP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f026372216ce45e3e5ddc95541e235cfd30a1a3e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-02-15 22:08:40.000000000","tz":0},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nStory: 2004073\nTask: 27434\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d7b16136d12239a22a98c2ac6fa8bbb6b1cc40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d7b16136d12239a22a98c2ac6fa8bbb6b1cc40"}]},"branch":"refs/heads/master"},"2c77b38bb2ae78c69a11dc963655160666fc589c":{"kind":"REWORK","_number":9,"created":"2019-02-22 15:09:24.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/9"}}},"commit":{"parents":[{"commit":"6e0bed1c540769e22af0dfff4af8c630dae2a5dd","subject":"Merge \"Set the default retries down to 120\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0bed1c540769e22af0dfff4af8c630dae2a5dd"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-22 15:09:04.000000000","tz":0},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nStory: 2004073\nTask: 27434\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c77b38bb2ae78c69a11dc963655160666fc589c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c77b38bb2ae78c69a11dc963655160666fc589c"}]},"branch":"refs/heads/master"},"1576bdcdc1392f524dc907f45dfffa78975db532":{"kind":"REWORK","_number":10,"created":"2019-04-08 17:16:19.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/10"}}},"commit":{"parents":[{"commit":"b207213204e4710d5ee7f2d90ac52a3e87c7f370","subject":"Merge \"Fix spare amphora check and creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b207213204e4710d5ee7f2d90ac52a3e87c7f370"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-04-08 17:16:04.000000000","tz":120},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nStory: 2004073\nTask: 27434\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1576bdcdc1392f524dc907f45dfffa78975db532"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1576bdcdc1392f524dc907f45dfffa78975db532"}]},"branch":"refs/heads/master"},"c3567cf65b23b5113ed4a1045ba38e709ee4718d":{"kind":"REWORK","_number":11,"created":"2019-04-15 10:08:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/11"}}},"commit":{"parents":[{"commit":"6e321eb2182b259eb55d0539f52d41b0ebad1f67","subject":"Merge \"Fix the amphora base port coming up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e321eb2182b259eb55d0539f52d41b0ebad1f67"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-04-15 10:08:39.000000000","tz":120},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nStory: 2004073\nTask: 27434\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3567cf65b23b5113ed4a1045ba38e709ee4718d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3567cf65b23b5113ed4a1045ba38e709ee4718d"}]},"branch":"refs/heads/master"},"9cd3bf4a4f339f57cee782a3b3894377d768ab7b":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-04-17 10:29:11.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/12"}}},"commit":{"parents":[{"commit":"66656bceb6657ac0a4d0d1ce86b6ac2925cf043e","subject":"Merge \"Update documentation about rotating amphorae\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66656bceb6657ac0a4d0d1ce86b6ac2925cf043e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-04-17 10:29:11.000000000","tz":0},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nStory: 2004073\nTask: 27434\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9cd3bf4a4f339f57cee782a3b3894377d768ab7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9cd3bf4a4f339f57cee782a3b3894377d768ab7b"}]},"branch":"refs/heads/master"},"24a94604b1e74fe94701d707b5941b294281ebdb":{"kind":"REWORK","_number":13,"created":"2019-04-18 00:16:49.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/13"}}},"commit":{"parents":[{"commit":"3924e443dfda4c4ec839361d4edac31ee0f25577","subject":"Remove python3.5 jobs for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3924e443dfda4c4ec839361d4edac31ee0f25577"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-04-18 00:16:22.000000000","tz":60},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nStory: 2004073\nTask: 27434\n\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24a94604b1e74fe94701d707b5941b294281ebdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24a94604b1e74fe94701d707b5941b294281ebdb"}]},"branch":"refs/heads/master"},"39248c743f1c5d747502fb4680a09ae10b4d4cdf":{"kind":"REWORK","_number":14,"created":"2019-04-18 00:22:17.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/14"}}},"commit":{"parents":[{"commit":"3924e443dfda4c4ec839361d4edac31ee0f25577","subject":"Remove python3.5 jobs for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3924e443dfda4c4ec839361d4edac31ee0f25577"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-04-18 00:19:13.000000000","tz":60},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nStory: 2004073\nTask: 27434\n\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39248c743f1c5d747502fb4680a09ae10b4d4cdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39248c743f1c5d747502fb4680a09ae10b4d4cdf"}]},"branch":"refs/heads/master"},"4e8b6054283fcde8571a4133053f990c2c360374":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-05-10 13:09:45.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/15"}}},"commit":{"parents":[{"commit":"28f4f9e44a07c224ea4da74169470c4e5cd2b17e","subject":"Fix missing INCLUDEPKGS var in haproxy element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28f4f9e44a07c224ea4da74169470c4e5cd2b17e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-10 13:08:52.000000000","tz":120},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nStory: 2004073\nTask: 27434\n\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e8b6054283fcde8571a4133053f990c2c360374"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e8b6054283fcde8571a4133053f990c2c360374"}]},"branch":"refs/heads/master"},"0261e94bf0327ec35fe7729a10c2f057c8689f48":{"kind":"REWORK","_number":16,"created":"2019-05-14 15:46:21.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/16"}}},"commit":{"parents":[{"commit":"987a6b3f1dc83fcaa86c7c02418b7f33afdd59b3","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/987a6b3f1dc83fcaa86c7c02418b7f33afdd59b3"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-14 15:40:02.000000000","tz":120},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nStory: 2004073\nTask: 27434\n\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0261e94bf0327ec35fe7729a10c2f057c8689f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0261e94bf0327ec35fe7729a10c2f057c8689f48"}]},"branch":"refs/heads/master"},"c4faac25de85ca3d8b4f12964589a36b2bcd3b57":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-05-14 17:11:22.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/635236/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/635236/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/635236/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/635236/17"}}},"commit":{"parents":[{"commit":"2c99d0109c8ac3970ce2d5548ad5a2bc8c35dfb3","subject":"Cap to sphinx 2.0.0 on Python 2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c99d0109c8ac3970ce2d5548ad5a2bc8c35dfb3"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-06 18:42:28.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-14 17:11:22.000000000","tz":0},"subject":"Add Python 3.7 support","message":"Add Python 3.7 support\n\nIn order to support Python 3.7, pylint has to be updated to 2.0.0\nminimum. Newer versions of Pylint enforce additional checkers which can\nbe addressed with some code refactoring rather than silently ignoring\nthem in pylintrc; except useless-object-inheritance which is required to\nbe silented so that we stay compatible with Python 2.x.\n\nStory: 2004073\nTask: 27434\n\nChange-Id: I52301d763797d619f195bd8a1c32bc47f1e68420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4faac25de85ca3d8b4f12964589a36b2bcd3b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4faac25de85ca3d8b4f12964589a36b2bcd3b57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
