)]}'
{"id":"openstack%2Foctavia~638581","triplet_id":"openstack%2Foctavia~master~Id11459ea70479aa0145059f88af847dddcd93553","project":"openstack/octavia","branch":"master","topic":"rhel8","hashtags":[],"change_id":"Id11459ea70479aa0145059f88af847dddcd93553","subject":"Add RHEL 8 amphora support","status":"MERGED","created":"2019-02-22 04:56:07.000000000","updated":"2019-06-09 13:56:08.000000000","submitted":"2019-06-09 13:56:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"638581-1560088568774-22ee9161","meta_rev_id":"0c17082f828c16931c3c68e407ebb8941a6b1967","_number":638581,"virtual_id_number":638581,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":23253,"name":"haokunsheng","email":"haoks@awcloud.com","username":"haoks"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-06-09 13:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2019-06-09 12:25:50.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":23253,"name":"haokunsheng","email":"haoks@awcloud.com","username":"haoks"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-07 16:38:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":1,"date":"2019-06-09 12:25:50.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":23253,"name":"haokunsheng","email":"haoks@awcloud.com","username":"haoks"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":23253,"name":"haokunsheng","email":"haoks@awcloud.com","username":"haoks"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23253,"name":"haokunsheng","email":"haoks@awcloud.com","username":"haoks"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-05 21:55:15.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2019-03-06 12:22:14.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-06-02 08:37:12.000000000","updated_by":{"_account_id":23253,"name":"haokunsheng","email":"haoks@awcloud.com","username":"haoks"},"reviewer":{"_account_id":23253,"name":"haokunsheng","email":"haoks@awcloud.com","username":"haoks"},"state":"REVIEWER"},{"updated":"2019-06-03 09:45:46.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-06-07 16:38:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-06-09 12:25:50.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-06-09 13:56:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e84b5a43531b57cb4186f066d4bd2809a40f17da","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-22 04:56:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d83b576f6d9252b4f24305d7913abd51f0119ea4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-22 04:56:22.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e6b197b354cf06f594171265a9db81915c486e1c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-22 05:00:17.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"db03a131beb386c93032843d856482e989b749bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-22 07:33:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/81/638581/1/check/openstack-tox-cover/35b3669/cover/ : SUCCESS in 8m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/638581/1/check/openstack-tox-lower-constraints/b9df65f/ : SUCCESS in 9m 45s\n- openstack-tox-pep8 http://logs.openstack.org/81/638581/1/check/openstack-tox-pep8/2cbdc58/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/81/638581/1/check/openstack-tox-py27/e58f424/ : SUCCESS in 6m 36s\n- openstack-tox-py35 http://logs.openstack.org/81/638581/1/check/openstack-tox-py35/efb6941/ : SUCCESS in 8m 22s\n- openstack-tox-py36 http://logs.openstack.org/81/638581/1/check/openstack-tox-py36/f6f65d1/ : SUCCESS in 6m 18s\n- openstack-tox-docs http://logs.openstack.org/81/638581/1/check/openstack-tox-docs/5cbdcc6/html/ : SUCCESS in 10m 17s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/81/638581/1/check/neutron-lbaasv2-dsvm-api/092b1c2/ : SUCCESS in 1h 16m 00s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/81/638581/1/check/neutron-lbaasv2-dsvm-py3x-api/c2f64d1/ : SUCCESS in 1h 21m 15s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/81/638581/1/check/neutron-lbaasv2-dsvm-scenario/98d0163/ : FAILURE in 27m 44s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/81/638581/1/check/neutron-lbaasv2-dsvm-py3x-scenario/97d6c59/ : FAILURE in 47m 47s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/81/638581/1/check/openstack-tox-functional/2f4176c/ : SUCCESS in 6m 57s\n- openstack-tox-functional-py35 http://logs.openstack.org/81/638581/1/check/openstack-tox-functional-py35/370d258/ : SUCCESS in 10m 22s\n- openstack-tox-functional-py36 http://logs.openstack.org/81/638581/1/check/openstack-tox-functional-py36/37c07a0/ : SUCCESS in 6m 58s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/81/638581/1/check/octavia-v1-dsvm-scenario/cd8e507/ : FAILURE in 47m 07s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/81/638581/1/check/octavia-v1-dsvm-py3x-scenario/db67317/ : FAILURE in 27m 59s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/81/638581/1/check/octavia-v2-dsvm-noop-api/226d661/ : SUCCESS in 59m 02s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/81/638581/1/check/octavia-v2-dsvm-noop-py2-api/ebdc4f6/ : SUCCESS in 36m 24s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/81/638581/1/check/octavia-v2-dsvm-scenario/20197af/ : FAILURE in 44m 50s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/81/638581/1/check/octavia-v2-dsvm-py2-scenario/9443715/ : FAILURE in 25m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/81/638581/1/check/octavia-v2-dsvm-py2-scenario-centos-7/9459663/ : FAILURE in 2h 13m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/81/638581/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/aab8da3/ : FAILURE in 26m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/81/638581/1/check/octavia-v2-dsvm-scenario-two-node/6778dda/ : FAILURE in 35m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/81/638581/1/check/octavia-v2-dsvm-py2-scenario-two-node/f1a5b76/ : FAILURE in 26m 23s (non-voting)\n- octavia-grenade http://logs.openstack.org/81/638581/1/check/octavia-grenade/e3a0506/ : SUCCESS in 1h 06m 09s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/81/638581/1/check/octavia-v2-dsvm-tls-barbican/709bd45/ : FAILURE in 27m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdaab86a0c33e333c917843958faded1249ac287","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-22 12:43:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85ef176bca8e41f4db71e4b04a490ae061c7e58c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-22 12:49:03.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"f8758be1795d9460190565b6f06f5d3cddfb957c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-22 14:56:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/81/638581/2/check/openstack-tox-cover/5fd957d/cover/ : SUCCESS in 8m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/638581/2/check/openstack-tox-lower-constraints/2e88edf/ : SUCCESS in 8m 27s\n- openstack-tox-pep8 http://logs.openstack.org/81/638581/2/check/openstack-tox-pep8/7a9cb6f/ : SUCCESS in 8m 11s\n- openstack-tox-py27 http://logs.openstack.org/81/638581/2/check/openstack-tox-py27/4b04424/ : SUCCESS in 5m 28s\n- openstack-tox-py35 http://logs.openstack.org/81/638581/2/check/openstack-tox-py35/dc4cdef/ : SUCCESS in 5m 49s\n- openstack-tox-py36 http://logs.openstack.org/81/638581/2/check/openstack-tox-py36/4f395bd/ : SUCCESS in 5m 11s\n- openstack-tox-docs http://logs.openstack.org/81/638581/2/check/openstack-tox-docs/c4e6a6f/html/ : SUCCESS in 7m 29s\n- build-openstack-releasenotes http://logs.openstack.org/81/638581/2/check/build-openstack-releasenotes/72a8c3a/html/ : SUCCESS in 6m 14s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/81/638581/2/check/neutron-lbaasv2-dsvm-api/83f2b89/ : SUCCESS in 1h 21m 06s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/81/638581/2/check/neutron-lbaasv2-dsvm-py3x-api/9a94234/ : SUCCESS in 1h 20m 32s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/81/638581/2/check/neutron-lbaasv2-dsvm-scenario/ed8caed/ : SUCCESS in 1h 25m 12s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/81/638581/2/check/neutron-lbaasv2-dsvm-py3x-scenario/7e9a156/ : SUCCESS in 1h 23m 31s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/81/638581/2/check/openstack-tox-functional/f1d089f/ : SUCCESS in 6m 42s\n- openstack-tox-functional-py35 http://logs.openstack.org/81/638581/2/check/openstack-tox-functional-py35/d8b386e/ : SUCCESS in 8m 48s\n- openstack-tox-functional-py36 http://logs.openstack.org/81/638581/2/check/openstack-tox-functional-py36/db90b9f/ : SUCCESS in 7m 43s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/81/638581/2/check/octavia-v1-dsvm-scenario/9386fa3/ : SUCCESS in 1h 00m 59s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/81/638581/2/check/octavia-v1-dsvm-py3x-scenario/2db1afa/ : SUCCESS in 1h 03m 29s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/81/638581/2/check/octavia-v2-dsvm-noop-api/6bf70c1/ : SUCCESS in 42m 58s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/81/638581/2/check/octavia-v2-dsvm-noop-py2-api/6b74857/ : SUCCESS in 43m 02s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/81/638581/2/check/octavia-v2-dsvm-scenario/993538f/ : SUCCESS in 1h 39m 34s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/81/638581/2/check/octavia-v2-dsvm-py2-scenario/d24c8d6/ : SUCCESS in 1h 31m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/81/638581/2/check/octavia-v2-dsvm-py2-scenario-centos-7/933b1da/ : FAILURE in 2h 00m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/81/638581/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/38f0da3/ : SUCCESS in 1h 37m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/81/638581/2/check/octavia-v2-dsvm-scenario-two-node/2b9506d/ : SUCCESS in 1h 48m 28s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/81/638581/2/check/octavia-v2-dsvm-py2-scenario-two-node/93f1321/ : FAILURE in 1h 36m 21s (non-voting)\n- octavia-grenade http://logs.openstack.org/81/638581/2/check/octavia-grenade/ba2a281/ : SUCCESS in 1h 19m 41s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/81/638581/2/check/octavia-v2-dsvm-tls-barbican/e999275/ : SUCCESS in 51m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74374334f61eb18992931c398efa912a7943f595","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-02-22 15:03:11.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"3d7b3cb1d806d3e6784a3c1939fe7b14dfd36393","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-03-05 17:04:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"82175fa553ffafdfe55027a2c4103d9cf1eea3d8","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-03-05 20:11:33.000000000","message":"Patch Set 2: Code-Review+2\n\nWorks in my tests.","accounts_in_message":[],"_revision_number":2},{"id":"949e67de2e06b914b65af582c45b80591e1e4f21","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-03-05 21:55:15.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fa1c26a7e12447f8d0ecb873c3545bdfe65bb681","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-03-19 14:46:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"00816fc743374904df22f4d71a111806bbb75483","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-03-19 14:46:46.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"6292a89175e4bcd30caf6c33e6610090614f23d9","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-03-19 15:28:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ab1fd4c43c14d920ace623727f4e32ce2a55efe2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-03-19 15:48:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f67180690e9094bf0fda207f3d134e9d80b5f233","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-03-19 15:55:27.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"3d62734d8daf03e2b5d00510ad3d88db7644998c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-19 16:18:27.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"d6d8c212092429387905d71840322c939b851d22","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-03-19 16:21:12.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"bf8d30c82252907e784572ebdb220d83c9bf52ba","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-03-24 13:08:15.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"1f9a20c53cae0d120c35913f19075193d76f1897","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-24 15:02:30.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/81/638581/6/check/openstack-tox-cover/cad9125/cover/ : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/638581/6/check/openstack-tox-lower-constraints/a2bc8c3/ : SUCCESS in 8m 13s\n- openstack-tox-pep8 http://logs.openstack.org/81/638581/6/check/openstack-tox-pep8/91f83b5/ : SUCCESS in 6m 33s\n- openstack-tox-py27 http://logs.openstack.org/81/638581/6/check/openstack-tox-py27/8352ed7/ : SUCCESS in 7m 30s\n- openstack-tox-py35 http://logs.openstack.org/81/638581/6/check/openstack-tox-py35/9cb6d2d/ : SUCCESS in 5m 51s\n- openstack-tox-py36 http://logs.openstack.org/81/638581/6/check/openstack-tox-py36/edb89e4/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/81/638581/6/check/openstack-tox-docs/6c07d08/html/ : SUCCESS in 6m 05s\n- build-openstack-releasenotes http://logs.openstack.org/81/638581/6/check/build-openstack-releasenotes/5594b38/html/ : SUCCESS in 4m 45s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/81/638581/6/check/neutron-lbaasv2-dsvm-api/2a2af99/ : SUCCESS in 1h 15m 22s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/81/638581/6/check/neutron-lbaasv2-dsvm-py3x-api/94f5906/ : SUCCESS in 1h 21m 04s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/81/638581/6/check/neutron-lbaasv2-dsvm-scenario/c778c3f/ : FAILURE in 1h 09m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/81/638581/6/check/neutron-lbaasv2-dsvm-py3x-scenario/1b93691/ : SUCCESS in 1h 18m 36s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/81/638581/6/check/openstack-tox-functional/c2fc9b2/ : SUCCESS in 8m 00s\n- openstack-tox-functional-py35 http://logs.openstack.org/81/638581/6/check/openstack-tox-functional-py35/78f91cf/ : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 http://logs.openstack.org/81/638581/6/check/openstack-tox-functional-py36/dd34c51/ : SUCCESS in 7m 01s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/81/638581/6/check/octavia-v1-dsvm-scenario/b987028/ : SUCCESS in 1h 04m 30s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/81/638581/6/check/octavia-v1-dsvm-py3x-scenario/31576da/ : FAILURE in 49m 34s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-noop-api/d95eabe/ : SUCCESS in 44m 31s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-noop-py2-api/4339a66/ : SUCCESS in 42m 02s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-scenario/0f5f459/ : SUCCESS in 1h 34m 46s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-py2-scenario/d3e600d/ : SUCCESS in 1h 27m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-py2-scenario-centos-7/b320924/ : FAILURE in 1h 47m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/a893e6d/ : SUCCESS in 1h 30m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-scenario-two-node/1755afc/ : SUCCESS in 1h 48m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-py2-scenario-two-node/ec841c8/ : SUCCESS in 1h 41m 54s (non-voting)\n- octavia-grenade http://logs.openstack.org/81/638581/6/check/octavia-grenade/d17c96f/ : SUCCESS in 1h 06m 11s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-tls-barbican/c94e6ee/ : SUCCESS in 49m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fc3ba5394f052015eef4bcc9b9fc18ec6bc6ee7e","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-03-24 15:04:04.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"d62b9baeb79fb36f5bec83fd299f696acc016954","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-24 17:24:59.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/81/638581/6/check/openstack-tox-cover/1cad94a/cover/ : SUCCESS in 9m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/638581/6/check/openstack-tox-lower-constraints/b85db29/ : SUCCESS in 8m 33s\n- openstack-tox-pep8 http://logs.openstack.org/81/638581/6/check/openstack-tox-pep8/666ddde/ : SUCCESS in 6m 25s\n- openstack-tox-py27 http://logs.openstack.org/81/638581/6/check/openstack-tox-py27/a8ee1e5/ : SUCCESS in 5m 03s\n- openstack-tox-py35 http://logs.openstack.org/81/638581/6/check/openstack-tox-py35/6719580/ : SUCCESS in 5m 26s\n- openstack-tox-py36 http://logs.openstack.org/81/638581/6/check/openstack-tox-py36/b54e34c/ : SUCCESS in 6m 47s\n- openstack-tox-docs http://logs.openstack.org/81/638581/6/check/openstack-tox-docs/1f5b88c/html/ : SUCCESS in 6m 57s\n- build-openstack-releasenotes http://logs.openstack.org/81/638581/6/check/build-openstack-releasenotes/1a7bf9b/html/ : SUCCESS in 6m 47s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/81/638581/6/check/neutron-lbaasv2-dsvm-api/3173e75/ : SUCCESS in 1h 22m 51s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/81/638581/6/check/neutron-lbaasv2-dsvm-py3x-api/c59ba81/ : SUCCESS in 1h 18m 16s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/81/638581/6/check/neutron-lbaasv2-dsvm-scenario/623cc2a/ : SUCCESS in 1h 22m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/81/638581/6/check/neutron-lbaasv2-dsvm-py3x-scenario/15bf8a8/ : SUCCESS in 1h 34m 29s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/81/638581/6/check/openstack-tox-functional/9d9789c/ : SUCCESS in 7m 03s\n- openstack-tox-functional-py35 http://logs.openstack.org/81/638581/6/check/openstack-tox-functional-py35/e560114/ : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/81/638581/6/check/openstack-tox-functional-py36/8b8b5dd/ : SUCCESS in 7m 08s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/81/638581/6/check/octavia-v1-dsvm-scenario/c5ded05/ : SUCCESS in 58m 55s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/81/638581/6/check/octavia-v1-dsvm-py3x-scenario/e638130/ : SUCCESS in 1h 11m 20s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-noop-api/6591ebc/ : SUCCESS in 41m 15s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-noop-py2-api/fc6544b/ : SUCCESS in 37m 27s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-scenario/e87a597/ : FAILURE in 1h 32m 50s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-py2-scenario/6a51f03/ : SUCCESS in 1h 39m 53s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-py2-scenario-centos-7/a966735/ : FAILURE in 2h 16m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6547217/ : SUCCESS in 1h 46m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-scenario-two-node/b9c2dc7/ : SUCCESS in 1h 40m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-py2-scenario-two-node/eb7e97e/ : FAILURE in 1h 53m 59s (non-voting)\n- octavia-grenade http://logs.openstack.org/81/638581/6/check/octavia-grenade/541e0b4/ : SUCCESS in 1h 07m 17s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-tls-barbican/2c99398/ : SUCCESS in 49m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2a6592b18ce8a47a3e71943771a16fb497227c8b","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-03-25 15:20:28.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"fda88bf34b8c1ef779b42a9ceb5441b123131e25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 17:34:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/81/638581/6/check/openstack-tox-cover/8c46140/cover/ : SUCCESS in 10m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/638581/6/check/openstack-tox-lower-constraints/dd03c97/ : SUCCESS in 9m 12s\n- openstack-tox-pep8 http://logs.openstack.org/81/638581/6/check/openstack-tox-pep8/9f8dbe1/ : SUCCESS in 7m 53s\n- openstack-tox-py27 http://logs.openstack.org/81/638581/6/check/openstack-tox-py27/5e02bd0/ : SUCCESS in 7m 09s\n- openstack-tox-py35 http://logs.openstack.org/81/638581/6/check/openstack-tox-py35/8923ecd/ : SUCCESS in 7m 08s\n- openstack-tox-py36 http://logs.openstack.org/81/638581/6/check/openstack-tox-py36/6d34f04/ : SUCCESS in 10m 18s\n- openstack-tox-docs http://logs.openstack.org/81/638581/6/check/openstack-tox-docs/95b9fab/html/ : SUCCESS in 8m 10s\n- build-openstack-releasenotes http://logs.openstack.org/81/638581/6/check/build-openstack-releasenotes/bbe3f8c/html/ : SUCCESS in 7m 31s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/81/638581/6/check/neutron-lbaasv2-dsvm-api/41aa141/ : SUCCESS in 1h 19m 26s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/81/638581/6/check/neutron-lbaasv2-dsvm-py3x-api/7e374d8/ : SUCCESS in 1h 30m 54s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/81/638581/6/check/neutron-lbaasv2-dsvm-scenario/4aecf6a/ : SUCCESS in 1h 30m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/81/638581/6/check/neutron-lbaasv2-dsvm-py3x-scenario/089a45f/ : SUCCESS in 1h 31m 47s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/81/638581/6/check/openstack-tox-functional/60a2263/ : SUCCESS in 11m 54s\n- openstack-tox-functional-py35 http://logs.openstack.org/81/638581/6/check/openstack-tox-functional-py35/83ae526/ : SUCCESS in 9m 07s\n- openstack-tox-functional-py36 http://logs.openstack.org/81/638581/6/check/openstack-tox-functional-py36/676fa4c/ : SUCCESS in 14m 55s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/81/638581/6/check/octavia-v1-dsvm-scenario/4943c6a/ : SUCCESS in 1h 03m 31s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/81/638581/6/check/octavia-v1-dsvm-py3x-scenario/932b0ed/ : SUCCESS in 1h 03m 19s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-noop-api/35d8356/ : SUCCESS in 44m 02s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-noop-py2-api/2a406ec/ : SUCCESS in 41m 16s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-scenario/8d006df/ : SUCCESS in 1h 51m 06s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-py2-scenario/5ce785d/ : SUCCESS in 1h 43m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-py2-scenario-centos-7/79cb0ca/ : FAILURE in 2h 03m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/187c045/ : SUCCESS in 2h 04m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-scenario-two-node/f3cecec/ : SUCCESS in 1h 50m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-py2-scenario-two-node/4b10bed/ : FAILURE in 1h 36m 27s (non-voting)\n- octavia-grenade http://logs.openstack.org/81/638581/6/check/octavia-grenade/671ce5a/ : SUCCESS in 1h 18m 10s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/81/638581/6/check/octavia-v2-dsvm-tls-barbican/4bd9b41/ : SUCCESS in 1h 03m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a35fcf098f3da352fe1b43148bdc57b9a6510fb0","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-05-28 09:15:06.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f0754be261e13b470e12ec54f9181e5d311cc9d7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-28 09:47:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"dd7427ae4ebabfeb18319f0a83749c3271dc5963","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-28 09:47:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"da31bd89f190eb8e63a3d9ea4306b745e71dd69b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 12:06:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/81/638581/7/check/openstack-tox-cover/7cbdbf3/cover/ : SUCCESS in 7m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/638581/7/check/openstack-tox-lower-constraints/117bfc0/ : SUCCESS in 6m 51s\n- openstack-tox-pep8 http://logs.openstack.org/81/638581/7/check/openstack-tox-pep8/a8d248d/ : SUCCESS in 6m 03s\n- openstack-tox-py27 http://logs.openstack.org/81/638581/7/check/openstack-tox-py27/49448c5/ : SUCCESS in 5m 16s\n- openstack-tox-py36 http://logs.openstack.org/81/638581/7/check/openstack-tox-py36/5fd8059/ : SUCCESS in 6m 04s\n- openstack-tox-py37 http://logs.openstack.org/81/638581/7/check/openstack-tox-py37/e808def/ : SUCCESS in 5m 33s\n- openstack-tox-docs http://logs.openstack.org/81/638581/7/check/openstack-tox-docs/a64402a/html/ : SUCCESS in 6m 04s\n- build-openstack-releasenotes http://logs.openstack.org/81/638581/7/check/build-openstack-releasenotes/5e3451f/html/ : SUCCESS in 5m 02s\n- openstack-tox-functional http://logs.openstack.org/81/638581/7/check/openstack-tox-functional/3695be5/ : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/81/638581/7/check/openstack-tox-functional-py36/2e6839b/ : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/81/638581/7/check/octavia-v2-dsvm-noop-api/0533eb4/ : SUCCESS in 41m 59s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/81/638581/7/check/octavia-v2-dsvm-noop-py2-api/767f040/ : SUCCESS in 45m 37s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/81/638581/7/check/octavia-v2-dsvm-scenario/615927b/ : SUCCESS in 1h 29m 53s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/81/638581/7/check/octavia-v2-dsvm-py2-scenario/8b51a66/ : SUCCESS in 1h 33m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/81/638581/7/check/octavia-v2-dsvm-py2-scenario-centos-7/795ceaa/ : FAILURE in 2h 01m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/81/638581/7/check/octavia-v2-dsvm-scenario-ubuntu-bionic/89f4931/ : SUCCESS in 1h 40m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/81/638581/7/check/octavia-v2-dsvm-scenario-two-node/a158d17/ : SUCCESS in 1h 59m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/81/638581/7/check/octavia-v2-dsvm-py2-scenario-two-node/68eac4a/ : SUCCESS in 1h 44m 17s (non-voting)\n- octavia-grenade http://logs.openstack.org/81/638581/7/check/octavia-grenade/065cf83/ : SUCCESS in 1h 12m 15s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/81/638581/7/check/octavia-v2-dsvm-tls-barbican/d47b0ef/ : SUCCESS in 56m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a2722093a03820df32c3296f6d518d31ea6621c8","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-05-28 13:15:18.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"15739a6696751ec0ce79980377f4375e2ef8502a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-28 13:17:29.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ad611b8731d9275976e31617fd1c823c26956af0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 15:42:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/81/638581/8/check/openstack-tox-cover/7286b5c/cover/ : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/638581/8/check/openstack-tox-lower-constraints/83d69e0/ : SUCCESS in 7m 18s\n- openstack-tox-pep8 http://logs.openstack.org/81/638581/8/check/openstack-tox-pep8/62f6a80/ : SUCCESS in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/81/638581/8/check/openstack-tox-py27/05f2462/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/81/638581/8/check/openstack-tox-py36/ade38ea/ : SUCCESS in 5m 03s\n- openstack-tox-py37 http://logs.openstack.org/81/638581/8/check/openstack-tox-py37/cee443f/ : SUCCESS in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/81/638581/8/check/openstack-tox-docs/ccd5e7d/html/ : SUCCESS in 5m 56s\n- build-openstack-releasenotes http://logs.openstack.org/81/638581/8/check/build-openstack-releasenotes/01e21bf/html/ : SUCCESS in 5m 29s\n- openstack-tox-functional http://logs.openstack.org/81/638581/8/check/openstack-tox-functional/7b8c78b/ : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 http://logs.openstack.org/81/638581/8/check/openstack-tox-functional-py36/5aa59dd/ : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/81/638581/8/check/octavia-v2-dsvm-noop-api/6e17326/ : SUCCESS in 44m 35s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/81/638581/8/check/octavia-v2-dsvm-noop-py2-api/b111c47/ : SUCCESS in 42m 57s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/81/638581/8/check/octavia-v2-dsvm-scenario/090163c/ : SUCCESS in 1h 31m 31s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/81/638581/8/check/octavia-v2-dsvm-py2-scenario/e33f2f8/ : SUCCESS in 1h 32m 26s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/81/638581/8/check/octavia-v2-dsvm-py2-scenario-centos-7/a86a158/ : FAILURE in 2h 06m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/81/638581/8/check/octavia-v2-dsvm-scenario-ubuntu-bionic/353995f/ : SUCCESS in 1h 29m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/81/638581/8/check/octavia-v2-dsvm-scenario-two-node/409550c/ : SUCCESS in 1h 35m 49s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/81/638581/8/check/octavia-v2-dsvm-py2-scenario-two-node/5f82e88/ : SUCCESS in 1h 40m 16s (non-voting)\n- octavia-grenade http://logs.openstack.org/81/638581/8/check/octavia-grenade/789d493/ : SUCCESS in 1h 21m 40s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/81/638581/8/check/octavia-v2-dsvm-tls-barbican/b1bd0db/ : SUCCESS in 54m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6810f97f852db401459eaa10d315017e4e48571a","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-05-29 11:46:22.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"de6e9342ec186af039be18d2481c6238fdbf177d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 13:57:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/81/638581/9/check/openstack-tox-cover/d41fab8/cover/ : SUCCESS in 8m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/638581/9/check/openstack-tox-lower-constraints/f77f7a4/ : SUCCESS in 6m 52s\n- openstack-tox-pep8 http://logs.openstack.org/81/638581/9/check/openstack-tox-pep8/f927c04/ : SUCCESS in 5m 56s\n- openstack-tox-py27 http://logs.openstack.org/81/638581/9/check/openstack-tox-py27/4950907/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/81/638581/9/check/openstack-tox-py36/7e5417f/ : SUCCESS in 5m 25s\n- openstack-tox-py37 http://logs.openstack.org/81/638581/9/check/openstack-tox-py37/39d337b/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/81/638581/9/check/openstack-tox-docs/999efe0/html/ : SUCCESS in 6m 19s\n- build-openstack-releasenotes http://logs.openstack.org/81/638581/9/check/build-openstack-releasenotes/2df7360/html/ : SUCCESS in 4m 49s\n- openstack-tox-functional http://logs.openstack.org/81/638581/9/check/openstack-tox-functional/752a7bd/ : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 http://logs.openstack.org/81/638581/9/check/openstack-tox-functional-py36/4a08c27/ : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/81/638581/9/check/octavia-v2-dsvm-noop-api/730dd70/ : SUCCESS in 40m 25s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/81/638581/9/check/octavia-v2-dsvm-noop-py2-api/10a25aa/ : SUCCESS in 40m 51s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/81/638581/9/check/octavia-v2-dsvm-scenario/f22bd70/ : SUCCESS in 1h 31m 52s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/81/638581/9/check/octavia-v2-dsvm-py2-scenario/0b2b89d/ : SUCCESS in 1h 34m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/81/638581/9/check/octavia-v2-dsvm-py2-scenario-centos-7/86fcb09/ : FAILURE in 2h 01m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/81/638581/9/check/octavia-v2-dsvm-scenario-ubuntu-bionic/611da92/ : FAILURE in 1h 49m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/81/638581/9/check/octavia-v2-dsvm-scenario-two-node/e668937/ : SUCCESS in 1h 52m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/81/638581/9/check/octavia-v2-dsvm-py2-scenario-two-node/f5e53ae/ : SUCCESS in 1h 45m 34s (non-voting)\n- octavia-grenade http://logs.openstack.org/81/638581/9/check/octavia-grenade/5c2ba3f/ : SUCCESS in 1h 14m 19s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/81/638581/9/check/octavia-v2-dsvm-tls-barbican/a88a9e7/ : SUCCESS in 52m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"97e6c9fc2dc0edd1cf34454242ff80a10754c5fb","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-06-02 07:29:55.000000000","message":"Patch Set 9: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":9},{"id":"fb0ded194887313d50274a21b92b4f60272401ec","author":{"_account_id":23253,"name":"haokunsheng","email":"haoks@awcloud.com","username":"haoks"},"date":"2019-06-02 08:37:12.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"13058202ae939a967d31cbab2a60b445beb70efb","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-06-03 06:36:17.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"77c71ce7b1efd2be942db5e1f20533d2521ec3c2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-03 08:34:37.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"07d25235b944097f165e714942e25f9df2edb6ac","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-06-03 08:37:28.000000000","message":"Patch Set 10: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"5b79cfef4f1183ad4a04a4a1630bac98dc06301a","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-06-03 09:45:46.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"dac26763d71889c44c0feb58abf15eb1c07801f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 11:02:25.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/81/638581/10/check/openstack-tox-cover/70b9689/cover/ : SUCCESS in 7m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/638581/10/check/openstack-tox-lower-constraints/3be4fc7/ : SUCCESS in 8m 00s\n- openstack-tox-pep8 http://logs.openstack.org/81/638581/10/check/openstack-tox-pep8/affbe38/ : SUCCESS in 6m 31s\n- openstack-tox-py27 http://logs.openstack.org/81/638581/10/check/openstack-tox-py27/9b9df59/ : SUCCESS in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/81/638581/10/check/openstack-tox-py36/d134dae/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/81/638581/10/check/openstack-tox-py37/56c0e6d/ : SUCCESS in 6m 43s\n- openstack-tox-docs http://logs.openstack.org/81/638581/10/check/openstack-tox-docs/dc04392/html/ : SUCCESS in 6m 00s\n- build-openstack-releasenotes http://logs.openstack.org/81/638581/10/check/build-openstack-releasenotes/b146382/html/ : SUCCESS in 5m 06s\n- openstack-tox-functional http://logs.openstack.org/81/638581/10/check/openstack-tox-functional/98037eb/ : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 http://logs.openstack.org/81/638581/10/check/openstack-tox-functional-py36/a796173/ : SUCCESS in 6m 37s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/81/638581/10/check/octavia-v2-dsvm-noop-api/a891ad8/ : SUCCESS in 38m 11s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/81/638581/10/check/octavia-v2-dsvm-noop-py2-api/d0a171e/ : SUCCESS in 45m 25s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/81/638581/10/check/octavia-v2-dsvm-scenario/c262f51/ : SUCCESS in 1h 28m 49s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/81/638581/10/check/octavia-v2-dsvm-py2-scenario/40700bb/ : SUCCESS in 1h 46m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/81/638581/10/check/octavia-v2-dsvm-py2-scenario-centos-7/4f04719/ : FAILURE in 1h 50m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/81/638581/10/check/octavia-v2-dsvm-scenario-ubuntu-bionic/2518d7e/ : SUCCESS in 1h 43m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/81/638581/10/check/octavia-v2-dsvm-scenario-two-node/6b627fa/ : SUCCESS in 1h 49m 02s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/81/638581/10/check/octavia-v2-dsvm-py2-scenario-two-node/83e3a88/ : SUCCESS in 2h 01m 23s (non-voting)\n- octavia-grenade http://logs.openstack.org/81/638581/10/check/octavia-grenade/7cb8fcc/ : SUCCESS in 1h 01m 15s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/81/638581/10/check/octavia-v2-dsvm-tls-barbican/c7830dd/ : SUCCESS in 51m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9b8410df790b33c4a74c9005f3c8828a151cedb6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-06 22:22:54.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"dd96d52f56af9fd3fcc3d8f28d720d0464c8d51c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-07 16:35:16.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c1d946d39a3195641148b339c613fdeffc8324a1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-07 16:35:45.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7a29fa99e31bed1e37b47501e2d08f2edfe7c71d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-07 16:38:04.000000000","message":"Patch Set 11: Code-Review+2\n\nWorked for me.","accounts_in_message":[],"_revision_number":11},{"id":"d1ff289ae45a3924847e8741be0026480116f8b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 18:48:47.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/81/638581/11/check/openstack-tox-cover/63b0229/cover/ : SUCCESS in 8m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/638581/11/check/openstack-tox-lower-constraints/796d8fc/ : SUCCESS in 6m 30s\n- openstack-tox-pep8 http://logs.openstack.org/81/638581/11/check/openstack-tox-pep8/1bd764c/ : SUCCESS in 6m 47s\n- openstack-tox-py27 http://logs.openstack.org/81/638581/11/check/openstack-tox-py27/6dac150/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/81/638581/11/check/openstack-tox-py36/c8174d0/ : SUCCESS in 5m 37s\n- openstack-tox-py37 http://logs.openstack.org/81/638581/11/check/openstack-tox-py37/e3c1a5e/ : SUCCESS in 6m 37s\n- openstack-tox-docs http://logs.openstack.org/81/638581/11/check/openstack-tox-docs/b8ebb42/html/ : SUCCESS in 6m 36s\n- build-openstack-releasenotes http://logs.openstack.org/81/638581/11/check/build-openstack-releasenotes/7775c5a/html/ : SUCCESS in 4m 53s\n- openstack-tox-functional http://logs.openstack.org/81/638581/11/check/openstack-tox-functional/a91c70f/ : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 http://logs.openstack.org/81/638581/11/check/openstack-tox-functional-py36/6c03786/ : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/81/638581/11/check/octavia-v2-dsvm-noop-api/219116a/ : SUCCESS in 43m 17s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/81/638581/11/check/octavia-v2-dsvm-noop-py2-api/600fa0c/ : SUCCESS in 43m 41s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/81/638581/11/check/octavia-v2-dsvm-scenario/6940657/ : SUCCESS in 1h 36m 21s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/81/638581/11/check/octavia-v2-dsvm-py2-scenario/561c6dc/ : SUCCESS in 1h 37m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/81/638581/11/check/octavia-v2-dsvm-py2-scenario-centos-7/f3a65f0/ : FAILURE in 2h 05m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/81/638581/11/check/octavia-v2-dsvm-scenario-ubuntu-bionic/dca778d/ : SUCCESS in 1h 36m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/81/638581/11/check/octavia-v2-dsvm-scenario-two-node/9f02406/ : SUCCESS in 1h 44m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/81/638581/11/check/octavia-v2-dsvm-py2-scenario-two-node/373b6a0/ : SUCCESS in 1h 56m 28s (non-voting)\n- octavia-grenade http://logs.openstack.org/81/638581/11/check/octavia-grenade/8d0fb4e/ : SUCCESS in 1h 11m 04s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/81/638581/11/check/octavia-v2-dsvm-tls-barbican/f437de7/ : SUCCESS in 50m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/81/638581/11/check/octavia-v2-dsvm-scenario-amphora-v2/cb02ef5/ : SUCCESS in 1h 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"56bf1384f2c736ddffeb94ac327eeee78e0e3d77","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-06-09 12:25:50.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"e45fc85dcb5e1f8588a21b3b68a022173cadfd93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 12:26:02.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"eb370f63fa7aa7770f9d8dcfd8ee57900dd92070","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 13:56:07.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/81/638581/11/gate/openstack-tox-lower-constraints/1ec66a9/ : SUCCESS in 6m 55s\n- openstack-tox-pep8 http://logs.openstack.org/81/638581/11/gate/openstack-tox-pep8/f8b68f7/ : SUCCESS in 6m 51s\n- openstack-tox-py27 http://logs.openstack.org/81/638581/11/gate/openstack-tox-py27/ba0af26/ : SUCCESS in 9m 51s\n- openstack-tox-py36 http://logs.openstack.org/81/638581/11/gate/openstack-tox-py36/75baca6/ : SUCCESS in 9m 43s\n- openstack-tox-py37 http://logs.openstack.org/81/638581/11/gate/openstack-tox-py37/c050b19/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/81/638581/11/gate/openstack-tox-docs/4289140/html/ : SUCCESS in 11m 22s\n- build-openstack-releasenotes http://logs.openstack.org/81/638581/11/gate/build-openstack-releasenotes/6be7ad9/html/ : SUCCESS in 5m 03s\n- openstack-tox-functional http://logs.openstack.org/81/638581/11/gate/openstack-tox-functional/3cc9f01/ : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 http://logs.openstack.org/81/638581/11/gate/openstack-tox-functional-py36/6f475e1/ : SUCCESS in 6m 07s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/81/638581/11/gate/octavia-v2-dsvm-noop-api/4d46742/ : SUCCESS in 35m 29s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/81/638581/11/gate/octavia-v2-dsvm-noop-py2-api/f85d247/ : SUCCESS in 36m 30s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/81/638581/11/gate/octavia-v2-dsvm-scenario/5baf1ad/ : SUCCESS in 1h 28m 58s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/81/638581/11/gate/octavia-v2-dsvm-py2-scenario/5ee46d5/ : SUCCESS in 1h 16m 33s\n- octavia-grenade http://logs.openstack.org/81/638581/11/gate/octavia-grenade/b9428d4/ : SUCCESS in 1h 05m 43s","accounts_in_message":[],"_revision_number":11},{"id":"e1d8b1b818f5e8811721c0e13718e7d87d76e71a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 13:56:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"8821737359549f824fd208a25d3258801e5251b5","revisions":{"60081a2763f6ea8b9d70a466a95624258d33f372":{"kind":"REWORK","_number":1,"created":"2019-02-22 04:56:07.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/81/638581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/638581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/638581/1"}}},"commit":{"parents":[{"commit":"6e0bed1c540769e22af0dfff4af8c630dae2a5dd","subject":"Merge \"Set the default retries down to 120\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0bed1c540769e22af0dfff4af8c630dae2a5dd"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-21 18:54:55.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-22 04:55:44.000000000","tz":60},"subject":"WIP: Add RHEL 8 amphora support","message":"WIP: Add RHEL 8 amphora support\n\nNetwork scripts are deprecated in RHEL 8. This patch makes sure the\npackage is installed and the service enabled. Sometime in the future\n(Train release or newer), support for Network Manager will be added as\nit became the default network configuration tool in RHEL/CentOS 8 and\nFedora 28+.\n\nDepends-On: https://review.openstack.org/#/c/623137/\n\nChange-Id: Id11459ea70479aa0145059f88af847dddcd93553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60081a2763f6ea8b9d70a466a95624258d33f372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60081a2763f6ea8b9d70a466a95624258d33f372"}]},"branch":"refs/heads/master"},"e5f4bacb9f18df166c2c70797f2c309486f06dde":{"kind":"REWORK","_number":2,"created":"2019-02-22 12:43:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/81/638581/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/638581/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/638581/2"}}},"commit":{"parents":[{"commit":"6e0bed1c540769e22af0dfff4af8c630dae2a5dd","subject":"Merge \"Set the default retries down to 120\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0bed1c540769e22af0dfff4af8c630dae2a5dd"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-21 18:54:55.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-22 12:29:51.000000000","tz":0},"subject":"Add RHEL 8 amphora support","message":"Add RHEL 8 amphora support\n\nNetwork scripts are deprecated in RHEL 8. This patch makes sure the\npackage is installed and the service enabled. Sometime in the future\n(Train release or newer), support for Network Manager will be added as\nit became the default network configuration tool in RHEL/CentOS 8 and\nFedora 28+.\n\nDepends-On: https://review.openstack.org/#/c/623137/\n\nChange-Id: Id11459ea70479aa0145059f88af847dddcd93553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5f4bacb9f18df166c2c70797f2c309486f06dde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5f4bacb9f18df166c2c70797f2c309486f06dde"}]},"branch":"refs/heads/master"},"2001ef4aa55f07c7dc337f173eec8cd08900cb74":{"kind":"REWORK","_number":3,"created":"2019-03-19 14:46:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/81/638581/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/638581/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/638581/3"}}},"commit":{"parents":[{"commit":"6e0bed1c540769e22af0dfff4af8c630dae2a5dd","subject":"Merge \"Set the default retries down to 120\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0bed1c540769e22af0dfff4af8c630dae2a5dd"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-21 18:54:55.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-03-19 14:46:31.000000000","tz":60},"subject":"Add RHEL 8 amphora support","message":"Add RHEL 8 amphora support\n\nNetwork scripts are deprecated in RHEL 8. This patch makes sure the\npackage is installed and the service enabled. Sometime in the future\n(Train release or newer), support for Network Manager will be added as\nit became the default network configuration tool in RHEL/CentOS 8 and\nFedora 28+.\n\nDepends-On: https://review.openstack.org/#/c/643731/\n\nChange-Id: Id11459ea70479aa0145059f88af847dddcd93553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2001ef4aa55f07c7dc337f173eec8cd08900cb74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2001ef4aa55f07c7dc337f173eec8cd08900cb74"}]},"branch":"refs/heads/master"},"a6d1a7f502bacdbc84dd1bebbed7c7d178b111d1":{"kind":"REWORK","_number":4,"created":"2019-03-19 15:28:32.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/81/638581/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/638581/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/638581/4"}}},"commit":{"parents":[{"commit":"6e0bed1c540769e22af0dfff4af8c630dae2a5dd","subject":"Merge \"Set the default retries down to 120\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0bed1c540769e22af0dfff4af8c630dae2a5dd"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-21 18:54:55.000000000","tz":60},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-03-19 15:18:35.000000000","tz":120},"subject":"Add RHEL 8 amphora support","message":"Add RHEL 8 amphora support\n\nNetwork scripts are deprecated in RHEL 8. This patch makes sure the\npackage is installed and the service enabled. Sometime in the future\n(Train release or newer), support for Network Manager will be added as\nit became the default network configuration tool in RHEL/CentOS 8 and\nFedora 28+.\n\nThis patch also reflects a change to diskimage-builder that will now\nhave a version-less \u0027rhel\u0027 element as opposed to a separate \u0027rhel7\u0027\nand \u0027rhel8\u0027 elements.\n\nDepends-On: https://review.openstack.org/#/c/643731/\n\nCo-Authored-By: Nir Magnezi \u003cnmagnezi@redhat.com\u003e\nChange-Id: Id11459ea70479aa0145059f88af847dddcd93553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6d1a7f502bacdbc84dd1bebbed7c7d178b111d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6d1a7f502bacdbc84dd1bebbed7c7d178b111d1"}]},"branch":"refs/heads/master"},"ff8a3ddcf4b2d5eb522bd8b3453728ce713cab54":{"kind":"REWORK","_number":5,"created":"2019-03-19 15:48:31.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/81/638581/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/638581/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/638581/5"}}},"commit":{"parents":[{"commit":"6e0bed1c540769e22af0dfff4af8c630dae2a5dd","subject":"Merge \"Set the default retries down to 120\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0bed1c540769e22af0dfff4af8c630dae2a5dd"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-21 18:54:55.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-03-19 15:48:29.000000000","tz":60},"subject":"Add RHEL 8 amphora support","message":"Add RHEL 8 amphora support\n\nNetwork scripts are deprecated in RHEL 8. This patch makes sure the\npackage is installed and the service enabled. Sometime in the future\n(Train release or newer), support for Network Manager will be added as\nit became the default network configuration tool in RHEL/CentOS 8 and\nFedora 28+.\n\nThis patch also reflects a change to diskimage-builder that will now\nhave a version-less \u0027rhel\u0027 element as opposed to a separate \u0027rhel7\u0027\nand \u0027rhel8\u0027 elements.\n\nDepends-On: https://review.openstack.org/#/c/643731/\n\nCo-Authored-By: Nir Magnezi \u003cnmagnezi@redhat.com\u003e\nChange-Id: Id11459ea70479aa0145059f88af847dddcd93553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff8a3ddcf4b2d5eb522bd8b3453728ce713cab54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff8a3ddcf4b2d5eb522bd8b3453728ce713cab54"}]},"branch":"refs/heads/master"},"498ed8b7ff597c28af0d374bf624056f41741bc9":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-03-19 15:55:27.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/81/638581/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/638581/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/638581/6"}}},"commit":{"parents":[{"commit":"6e0bed1c540769e22af0dfff4af8c630dae2a5dd","subject":"Merge \"Set the default retries down to 120\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0bed1c540769e22af0dfff4af8c630dae2a5dd"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-21 18:54:55.000000000","tz":60},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-03-19 15:55:22.000000000","tz":0},"subject":"Add RHEL 8 amphora support","message":"Add RHEL 8 amphora support\n\nNetwork scripts are deprecated in RHEL 8. This patch makes sure the\npackage is installed and the service enabled. Sometime in the future\n(Train release or newer), support for Network Manager will be added as\nit became the default network configuration tool in RHEL/CentOS 8 and\nFedora 28+.\n\nThis patch also reflects a change to diskimage-builder that will now\nhave a version-less \u0027rhel\u0027 element as opposed to a separate \u0027rhel7\u0027\nand \u0027rhel8\u0027 elements.\n\nDepends-On: https://review.openstack.org/#/c/643731/\n\nChange-Id: Id11459ea70479aa0145059f88af847dddcd93553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/498ed8b7ff597c28af0d374bf624056f41741bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/498ed8b7ff597c28af0d374bf624056f41741bc9"}]},"branch":"refs/heads/master"},"9e42634fdfb96f14c07a8992316b1ce534d58641":{"kind":"REWORK","_number":7,"created":"2019-05-28 09:47:45.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/81/638581/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/638581/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/638581/7"}}},"commit":{"parents":[{"commit":"6e0bed1c540769e22af0dfff4af8c630dae2a5dd","subject":"Merge \"Set the default retries down to 120\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0bed1c540769e22af0dfff4af8c630dae2a5dd"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-21 18:54:55.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-28 09:46:40.000000000","tz":120},"subject":"Add RHEL 8 amphora support","message":"Add RHEL 8 amphora support\n\nNetwork scripts are deprecated in RHEL 8. This patch makes sure the\npackage is installed and the service enabled. Sometime in the future\n(Train release or newer), support for Network Manager will be added as\nit became the default network configuration tool in RHEL/CentOS 8 and\nFedora 28+.\n\nThis patch also reflects a change to diskimage-builder that will now\nhave a version-less \u0027rhel\u0027 element as opposed to a separate \u0027rhel7\u0027\nand \u0027rhel8\u0027 elements.\n\nDepends-On: https://review.openstack.org/#/c/643731/\n\nChange-Id: Id11459ea70479aa0145059f88af847dddcd93553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e42634fdfb96f14c07a8992316b1ce534d58641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e42634fdfb96f14c07a8992316b1ce534d58641"}]},"branch":"refs/heads/master"},"4b3f2608e69225a43054567617399771e22ba5b3":{"kind":"REWORK","_number":8,"created":"2019-05-28 13:17:29.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/81/638581/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/638581/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/638581/8"}}},"commit":{"parents":[{"commit":"6e0bed1c540769e22af0dfff4af8c630dae2a5dd","subject":"Merge \"Set the default retries down to 120\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0bed1c540769e22af0dfff4af8c630dae2a5dd"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-21 18:54:55.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-28 13:17:19.000000000","tz":120},"subject":"Add RHEL 8 amphora support","message":"Add RHEL 8 amphora support\n\nNetwork scripts are deprecated in RHEL 8. This patch makes sure the\npackage is installed and the service enabled. Sometime in the future\n(Train release or newer), support for Network Manager will be added as\nit became the default network configuration tool in RHEL/CentOS 8 and\nFedora 28+.\n\nThis patch also reflects a change to diskimage-builder that will now\nhave a version-less \u0027rhel\u0027 element as opposed to a separate \u0027rhel7\u0027\nand \u0027rhel8\u0027 elements.\n\nDepends-On: https://review.openstack.org/#/c/643731/\n\nChange-Id: Id11459ea70479aa0145059f88af847dddcd93553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b3f2608e69225a43054567617399771e22ba5b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b3f2608e69225a43054567617399771e22ba5b3"}]},"branch":"refs/heads/master"},"953d386ed417c3720e23ee704e3a3de9a459c9f4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-05-29 11:46:22.000000000","uploader":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"ref":"refs/changes/81/638581/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/638581/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/638581/9"}}},"commit":{"parents":[{"commit":"21a54d6b78d002c76ca2e68b3aad34681d51686f","subject":"Merge \"Rename review.openstack.org to review.opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21a54d6b78d002c76ca2e68b3aad34681d51686f"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-21 18:54:55.000000000","tz":60},"committer":{"name":"Nir Magnezi","email":"nmagnezi@redhat.com","date":"2019-05-29 11:43:11.000000000","tz":180},"subject":"Add RHEL 8 amphora support","message":"Add RHEL 8 amphora support\n\nNetwork scripts are deprecated in RHEL 8. This patch makes sure the\npackage is installed and the service enabled. Sometime in the future\n(Train release or newer), support for Network Manager will be added as\nit became the default network configuration tool in RHEL/CentOS 8 and\nFedora 28+.\n\nThis patch also reflects a change to diskimage-builder that will now\nhave a version-less \u0027rhel\u0027 element as opposed to a separate \u0027rhel7\u0027\nand \u0027rhel8\u0027 elements.\n\nDepends-On: https://review.openstack.org/#/c/643731/\n\nChange-Id: Id11459ea70479aa0145059f88af847dddcd93553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/953d386ed417c3720e23ee704e3a3de9a459c9f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/953d386ed417c3720e23ee704e3a3de9a459c9f4"}]},"branch":"refs/heads/master"},"39a686acba4cc34bef2e67178704b92ebc2826f1":{"kind":"REWORK","_number":10,"created":"2019-06-03 08:34:37.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/81/638581/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/638581/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/638581/10"}}},"commit":{"parents":[{"commit":"21a54d6b78d002c76ca2e68b3aad34681d51686f","subject":"Merge \"Rename review.openstack.org to review.opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21a54d6b78d002c76ca2e68b3aad34681d51686f"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-21 18:54:55.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-03 08:34:27.000000000","tz":0},"subject":"Add RHEL 8 amphora support","message":"Add RHEL 8 amphora support\n\nNetwork scripts are deprecated in RHEL 8. This patch makes sure the\npackage is installed and the service enabled. Sometime in the future\n(Train release or newer), support for Network Manager will be added as\nit became the default network configuration tool in RHEL/CentOS 8 and\nFedora 28+.\n\nThis patch also reflects a change to diskimage-builder that will now\nhave a version-less \u0027rhel\u0027 element as opposed to a separate \u0027rhel7\u0027\nand \u0027rhel8\u0027 elements.\n\nDepends-On: https://review.openstack.org/#/c/643731/\n\nChange-Id: Id11459ea70479aa0145059f88af847dddcd93553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39a686acba4cc34bef2e67178704b92ebc2826f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39a686acba4cc34bef2e67178704b92ebc2826f1"}]},"branch":"refs/heads/master"},"8821737359549f824fd208a25d3258801e5251b5":{"kind":"REWORK","_number":11,"created":"2019-06-07 16:35:16.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/81/638581/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/638581/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/638581/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/638581/11"}}},"commit":{"parents":[{"commit":"21a54d6b78d002c76ca2e68b3aad34681d51686f","subject":"Merge \"Rename review.openstack.org to review.opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21a54d6b78d002c76ca2e68b3aad34681d51686f"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-02-21 18:54:55.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-06-07 16:34:29.000000000","tz":120},"subject":"Add RHEL 8 amphora support","message":"Add RHEL 8 amphora support\n\nNetwork scripts are deprecated in RHEL 8. This patch makes sure the\npackage is installed and the service enabled. Sometime in the future\n(Train release or newer), support for Network Manager will be added as\nit became the default network configuration tool in RHEL/CentOS 8 and\nFedora 28+.\n\nThis patch also reflects a change to diskimage-builder that will now\nhave a version-less \u0027rhel\u0027 element as opposed to a separate \u0027rhel7\u0027\nand \u0027rhel8\u0027 elements.\n\nChange-Id: Id11459ea70479aa0145059f88af847dddcd93553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8821737359549f824fd208a25d3258801e5251b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8821737359549f824fd208a25d3258801e5251b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
