)]}'
{"id":"openstack%2Foctavia~639736","triplet_id":"openstack%2Foctavia~master~I276188530a83598ed75560f02ed9d80ce9afca2f","project":"openstack/octavia","branch":"master","topic":"http_proxy_to_wsgi","hashtags":[],"change_id":"I276188530a83598ed75560f02ed9d80ce9afca2f","subject":"Add support for oslo_middleware http_proxy_to_wsgi","status":"MERGED","created":"2019-02-27 16:34:29.000000000","updated":"2019-08-01 04:21:52.000000000","submitted":"2019-08-01 04:21:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"639736-1564633312822-05bf691b","meta_rev_id":"1c820f3dd8e42814a99a82b96b703f3b13e7a49d","_number":639736,"virtual_id_number":639736,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20408,"name":"Pavel Abalikhin","email":"anpavl@gmail.com","username":"pabalikhin"},{"value":0,"_account_id":28058,"name":"Mikhail Kalinin","email":"objque@gmail.com","username":"objque"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":13323,"name":"Tatiana Kholkina","email":"tlashchova@gmail.com","username":"tlashchova"},{"value":2,"date":"2019-08-01 04:21:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-01 00:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20408,"name":"Pavel Abalikhin","email":"anpavl@gmail.com","username":"pabalikhin"},{"value":0,"_account_id":28058,"name":"Mikhail Kalinin","email":"objque@gmail.com","username":"objque"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":13323,"name":"Tatiana Kholkina","email":"tlashchova@gmail.com","username":"tlashchova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-01 00:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20408,"name":"Pavel Abalikhin","email":"anpavl@gmail.com","username":"pabalikhin"},{"value":0,"_account_id":28058,"name":"Mikhail Kalinin","email":"objque@gmail.com","username":"objque"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":13323,"name":"Tatiana Kholkina","email":"tlashchova@gmail.com","username":"tlashchova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-01 00:49:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20408,"name":"Pavel Abalikhin","email":"anpavl@gmail.com","username":"pabalikhin"},{"value":0,"_account_id":28058,"name":"Mikhail Kalinin","email":"objque@gmail.com","username":"objque"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":13323,"name":"Tatiana Kholkina","email":"tlashchova@gmail.com","username":"tlashchova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13323,"name":"Tatiana Kholkina","email":"tlashchova@gmail.com","username":"tlashchova"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":20408,"name":"Pavel Abalikhin","email":"anpavl@gmail.com","username":"pabalikhin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28058,"name":"Mikhail Kalinin","email":"objque@gmail.com","username":"objque"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-28 12:56:21.000000000","updated_by":{"_account_id":28058,"name":"Mikhail Kalinin","email":"objque@gmail.com","username":"objque"},"reviewer":{"_account_id":28058,"name":"Mikhail Kalinin","email":"objque@gmail.com","username":"objque"},"state":"REVIEWER"},{"updated":"2019-02-28 13:22:28.000000000","updated_by":{"_account_id":13323,"name":"Tatiana Kholkina","email":"tlashchova@gmail.com","username":"tlashchova"},"reviewer":{"_account_id":13323,"name":"Tatiana Kholkina","email":"tlashchova@gmail.com","username":"tlashchova"},"state":"REVIEWER"},{"updated":"2019-03-19 15:39:52.000000000","updated_by":{"_account_id":20408,"name":"Pavel Abalikhin","email":"anpavl@gmail.com","username":"pabalikhin"},"reviewer":{"_account_id":20408,"name":"Pavel Abalikhin","email":"anpavl@gmail.com","username":"pabalikhin"},"state":"REVIEWER"},{"updated":"2019-03-20 14:38:09.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-07-24 09:13:02.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-07-26 20:43:24.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-08-01 00:12:18.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-01 00:49:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-01 04:21:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f864ec8c49da26491cf1259933c12ad03b47314","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-02-27 16:34:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec5ee47a942698749eb35a0a6a0669ac573d11c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 18:53:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/36/639736/1/check/openstack-tox-cover/0f11d46/cover/ : SUCCESS in 10m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/639736/1/check/openstack-tox-lower-constraints/c79b13f/ : SUCCESS in 8m 09s\n- openstack-tox-pep8 http://logs.openstack.org/36/639736/1/check/openstack-tox-pep8/28b5a2a/ : SUCCESS in 9m 00s\n- openstack-tox-py27 http://logs.openstack.org/36/639736/1/check/openstack-tox-py27/da1d313/ : SUCCESS in 6m 19s\n- openstack-tox-py35 http://logs.openstack.org/36/639736/1/check/openstack-tox-py35/0eabdb4/ : SUCCESS in 8m 30s\n- openstack-tox-py36 http://logs.openstack.org/36/639736/1/check/openstack-tox-py36/fbfd6d6/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/36/639736/1/check/openstack-tox-docs/620379a/html/ : SUCCESS in 7m 01s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/639736/1/check/neutron-lbaasv2-dsvm-api/c351277/ : SUCCESS in 1h 15m 30s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/639736/1/check/neutron-lbaasv2-dsvm-py3x-api/3313695/ : SUCCESS in 1h 24m 19s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/639736/1/check/neutron-lbaasv2-dsvm-scenario/621a9df/ : SUCCESS in 1h 38m 13s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/639736/1/check/neutron-lbaasv2-dsvm-py3x-scenario/82b6a0f/ : SUCCESS in 1h 37m 38s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/639736/1/check/openstack-tox-functional/e783a56/ : SUCCESS in 7m 02s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/639736/1/check/openstack-tox-functional-py35/4ae2d82/ : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/639736/1/check/openstack-tox-functional-py36/caa7f2e/ : SUCCESS in 11m 05s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/639736/1/check/octavia-v1-dsvm-scenario/0e9bb3f/ : SUCCESS in 55m 59s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/639736/1/check/octavia-v1-dsvm-py3x-scenario/b355a33/ : SUCCESS in 55m 45s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/639736/1/check/octavia-v2-dsvm-noop-api/fdcaa3c/ : SUCCESS in 39m 10s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/639736/1/check/octavia-v2-dsvm-noop-py2-api/7fca141/ : SUCCESS in 38m 50s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/639736/1/check/octavia-v2-dsvm-scenario/029792a/ : SUCCESS in 1h 36m 26s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/639736/1/check/octavia-v2-dsvm-py2-scenario/ec5db7b/ : SUCCESS in 1h 17m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/639736/1/check/octavia-v2-dsvm-py2-scenario-centos-7/946e469/ : SUCCESS in 2h 12m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/639736/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/7d4c246/ : SUCCESS in 1h 33m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/639736/1/check/octavia-v2-dsvm-scenario-two-node/0535c47/ : SUCCESS in 1h 39m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/639736/1/check/octavia-v2-dsvm-py2-scenario-two-node/f0ea3f5/ : SUCCESS in 1h 16m 51s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/639736/1/check/octavia-grenade/2463aff/ : SUCCESS in 1h 06m 10s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/639736/1/check/octavia-v2-dsvm-tls-barbican/4453ea7/ : SUCCESS in 48m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32a72cfc4a46e337b66d64211cac4e90704bbe0b","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-02-27 20:22:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ae6a2beab62dcf56f8aec8cc322d01a2dec7f918","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-02-27 20:23:46.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4246ef0023ae6d15d721d423ca008ff94d0014a0","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-02-27 20:24:17.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"6de53cfc73b4bc811fff3bce70664de00d064a87","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-02-27 20:36:40.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"2588e7158642bef8296930649d41f9e77252410a","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-02-27 21:18:41.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"be8832c89023b8f8fea5425c437f07f0ba04e406","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 23:36:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/36/639736/6/check/openstack-tox-cover/7d3720a/cover/ : SUCCESS in 10m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/36/639736/6/check/openstack-tox-lower-constraints/41b55f5/ : SUCCESS in 7m 45s\n- openstack-tox-pep8 http://logs.openstack.org/36/639736/6/check/openstack-tox-pep8/b274a38/ : SUCCESS in 11m 11s\n- openstack-tox-py27 http://logs.openstack.org/36/639736/6/check/openstack-tox-py27/57ff6aa/ : SUCCESS in 8m 11s\n- openstack-tox-py35 http://logs.openstack.org/36/639736/6/check/openstack-tox-py35/4e24471/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/36/639736/6/check/openstack-tox-py36/5dc3f47/ : SUCCESS in 6m 10s\n- openstack-tox-docs http://logs.openstack.org/36/639736/6/check/openstack-tox-docs/d81a9f9/html/ : SUCCESS in 9m 02s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/36/639736/6/check/neutron-lbaasv2-dsvm-api/0475a9f/ : SUCCESS in 1h 14m 29s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/36/639736/6/check/neutron-lbaasv2-dsvm-py3x-api/22f1a87/ : SUCCESS in 1h 17m 52s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/36/639736/6/check/neutron-lbaasv2-dsvm-scenario/3f67669/ : SUCCESS in 1h 25m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/36/639736/6/check/neutron-lbaasv2-dsvm-py3x-scenario/9fd2a61/ : SUCCESS in 1h 25m 58s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/36/639736/6/check/openstack-tox-functional/72520f6/ : SUCCESS in 7m 09s\n- openstack-tox-functional-py35 http://logs.openstack.org/36/639736/6/check/openstack-tox-functional-py35/63bed42/ : SUCCESS in 8m 28s\n- openstack-tox-functional-py36 http://logs.openstack.org/36/639736/6/check/openstack-tox-functional-py36/e245481/ : SUCCESS in 12m 50s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/36/639736/6/check/octavia-v1-dsvm-scenario/80663cd/ : SUCCESS in 59m 29s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/36/639736/6/check/octavia-v1-dsvm-py3x-scenario/03f0090/ : SUCCESS in 1h 04m 09s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/36/639736/6/check/octavia-v2-dsvm-noop-api/cbb0985/ : SUCCESS in 44m 36s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/36/639736/6/check/octavia-v2-dsvm-noop-py2-api/60e01ad/ : SUCCESS in 45m 05s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/36/639736/6/check/octavia-v2-dsvm-scenario/77c0bae/ : SUCCESS in 1h 36m 11s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/36/639736/6/check/octavia-v2-dsvm-py2-scenario/0be055d/ : SUCCESS in 1h 24m 44s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/36/639736/6/check/octavia-v2-dsvm-py2-scenario-centos-7/deefea0/ : SUCCESS in 2h 12m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/36/639736/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/f548e6c/ : SUCCESS in 1h 33m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/36/639736/6/check/octavia-v2-dsvm-scenario-two-node/437e5c8/ : FAILURE in 1h 19m 12s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/36/639736/6/check/octavia-v2-dsvm-py2-scenario-two-node/909a9c7/ : FAILURE in 1h 20m 29s (non-voting)\n- octavia-grenade http://logs.openstack.org/36/639736/6/check/octavia-grenade/23348bb/ : SUCCESS in 1h 32m 41s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/36/639736/6/check/octavia-v2-dsvm-tls-barbican/731f093/ : SUCCESS in 58m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"acf5a26926256ff06f7785736c35209068a278a7","author":{"_account_id":28058,"name":"Mikhail Kalinin","email":"objque@gmail.com","username":"objque"},"date":"2019-02-28 12:56:21.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f926492c9ce22dae086eed3a69ee56a6da6560f2","author":{"_account_id":13323,"name":"Tatiana Kholkina","email":"tlashchova@gmail.com","username":"tlashchova"},"date":"2019-02-28 13:22:28.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"fe075cce7878e06ec8d622042a9a37d6a702e01a","author":{"_account_id":20408,"name":"Pavel Abalikhin","email":"anpavl@gmail.com","username":"pabalikhin"},"date":"2019-03-19 15:39:52.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d68e1262301ee6052175e46235563d67834ecb09","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-23 21:49:37.000000000","message":"Patch Set 6: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"8c74b15299486d822a292be57bfaa9e64f4027a2","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-24 09:13:02.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"2956897b340386d91e56432d8dd8ca4a9306a8b5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-26 20:43:24.000000000","message":"Patch Set 6: Code-Review-1\n\nThis looks like something we\u0027d also want to have backported to stable branches. Could you please add a release note? Also, maybe it would be good to add it to octavia/etc/octavia.conf for awareness.","accounts_in_message":[],"_revision_number":6},{"id":"6bc5b4daa2dfe838b36225f030d78dc2e3760d2d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-26 22:48:32.000000000","message":"Patch Set 6: Code-Review-1\n\nYes, definitely need a release-note for this! This is something people really need to see called out so they can configure it accordingly.\n\nAdding it to the example octavia.conf would also be nice.","accounts_in_message":[],"_revision_number":6},{"id":"23d83bc528bd3ad7f355a820a64800117b4edf4e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-01 00:03:10.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"40c3026b62af913c625cd05d11b2cca48c2b4377","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-01 00:11:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"94440d32f9cb0233875f3a06f6911561df50039c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-01 00:12:18.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"0f9561f69748ea00d1b7fe52a4dd2fc653165501","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-01 00:49:58.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":8},{"id":"6c9cd108a6ace4ddec3208c1256df5196682e4f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 02:36:10.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/36/639736/8/check/openstack-tox-cover/3822a8d/cover/ : SUCCESS in 7m 33s\n- openstack-tox-lower-constraints https://logs.opendev.org/36/639736/8/check/openstack-tox-lower-constraints/4cb05e7/ : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://logs.opendev.org/36/639736/8/check/openstack-tox-pep8/214370c/ : SUCCESS in 5m 37s\n- openstack-tox-py27 https://logs.opendev.org/36/639736/8/check/openstack-tox-py27/c2a482d/ : SUCCESS in 5m 21s\n- openstack-tox-py36 https://logs.opendev.org/36/639736/8/check/openstack-tox-py36/9d8895f/ : SUCCESS in 3m 37s\n- openstack-tox-py37 https://logs.opendev.org/36/639736/8/check/openstack-tox-py37/db556a1/ : SUCCESS in 5m 33s\n- openstack-tox-docs https://logs.opendev.org/36/639736/8/check/openstack-tox-docs/534d776/html/ : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://logs.opendev.org/36/639736/8/check/build-openstack-releasenotes/62106ac/html/ : SUCCESS in 4m 11s\n- openstack-tox-functional https://logs.opendev.org/36/639736/8/check/openstack-tox-functional/9dc8270/ : SUCCESS in 4m 58s\n- openstack-tox-functional-py36 https://logs.opendev.org/36/639736/8/check/openstack-tox-functional-py36/8a4ec40/ : SUCCESS in 4m 57s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/36/639736/8/check/octavia-v2-dsvm-noop-api/77c53a9/ : SUCCESS in 38m 39s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/36/639736/8/check/octavia-v2-dsvm-noop-py2-api/91bd068/ : SUCCESS in 50m 11s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/36/639736/8/check/octavia-v2-dsvm-scenario/5a20348/ : SUCCESS in 1h 38m 46s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/36/639736/8/check/octavia-v2-dsvm-py2-scenario/5025bde/ : SUCCESS in 2h 04m 01s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/36/639736/8/check/octavia-v2-dsvm-py2-scenario-centos-7/5ec2507/ : FAILURE in 1h 52m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/36/639736/8/check/octavia-v2-dsvm-scenario-two-node/c355416/ : SUCCESS in 1h 56m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/36/639736/8/check/octavia-v2-dsvm-py2-scenario-two-node/5e48a6f/ : SUCCESS in 2h 18m 18s (non-voting)\n- octavia-grenade https://logs.opendev.org/36/639736/8/check/octavia-grenade/da1c722/ : SUCCESS in 1h 17m 38s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/36/639736/8/check/octavia-v2-dsvm-tls-barbican/8459215/ : SUCCESS in 46m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/36/639736/8/check/octavia-v2-dsvm-scenario-amphora-v2/e6b7ad1/ : SUCCESS in 1h 27m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/36/639736/8/check/octavia-v2-act-stdby-iptables-dsvm-scenario/4128608/ : FAILURE in 1h 15m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/36/639736/8/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/1c4a283/ : FAILURE in 1h 19m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/36/639736/8/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/407ebcf/ : FAILURE in 1h 43m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b0142cf4fa481fbb6d3a0f89f88cac88e6167da0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 02:36:21.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"81d5f9137e96eac0a4a02074181776b089c30e70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 04:21:52.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/36/639736/8/gate/openstack-tox-lower-constraints/ccf5040/ : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://logs.opendev.org/36/639736/8/gate/openstack-tox-pep8/3fdfe7b/ : SUCCESS in 5m 25s\n- openstack-tox-py27 https://logs.opendev.org/36/639736/8/gate/openstack-tox-py27/c84b2f0/ : SUCCESS in 4m 01s\n- openstack-tox-py36 https://logs.opendev.org/36/639736/8/gate/openstack-tox-py36/d7e8d68/ : SUCCESS in 4m 35s\n- openstack-tox-py37 https://logs.opendev.org/36/639736/8/gate/openstack-tox-py37/fada174/ : SUCCESS in 4m 16s\n- openstack-tox-docs https://logs.opendev.org/36/639736/8/gate/openstack-tox-docs/aeb88a2/html/ : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://logs.opendev.org/36/639736/8/gate/build-openstack-releasenotes/5e8e018/html/ : SUCCESS in 4m 18s\n- openstack-tox-functional https://logs.opendev.org/36/639736/8/gate/openstack-tox-functional/2f3205c/ : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://logs.opendev.org/36/639736/8/gate/openstack-tox-functional-py36/6ce70e7/ : SUCCESS in 5m 02s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/36/639736/8/gate/octavia-v2-dsvm-noop-api/2b6e41b/ : SUCCESS in 43m 12s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/36/639736/8/gate/octavia-v2-dsvm-noop-py2-api/b9a34ef/ : SUCCESS in 44m 03s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/36/639736/8/gate/octavia-v2-dsvm-scenario/3e8c618/ : SUCCESS in 1h 39m 50s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/36/639736/8/gate/octavia-v2-dsvm-py2-scenario/6a00486/ : SUCCESS in 1h 35m 56s\n- octavia-grenade https://logs.opendev.org/36/639736/8/gate/octavia-grenade/0290401/ : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":8},{"id":"a04207fa81c9a9dffb8bbf0e1d1c0b0416501c03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 04:21:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"ec83c69372d647aaeec02fea6eea29c7ee11115a","revisions":{"6e7756c705c1b5127dd218d035fa142292fd2279":{"kind":"REWORK","_number":1,"created":"2019-02-27 16:34:29.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/36/639736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/639736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/639736/1"}}},"commit":{"parents":[{"commit":"612b042e64d9fcffb983f958ed0d6de6a9f765d3","subject":"Merge \"L7rule support client certificate cases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/612b042e64d9fcffb983f958ed0d6de6a9f765d3"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 16:34:02.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 16:34:26.000000000","tz":180},"subject":"WIP: Add support for the oslo_middleware.http_proxy_to_wsgi","message":"WIP: Add support for the oslo_middleware.http_proxy_to_wsgi\n\nTODO:\n1. release note\n2. task in the storyboard\n\nChange-Id: I276188530a83598ed75560f02ed9d80ce9afca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e7756c705c1b5127dd218d035fa142292fd2279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e7756c705c1b5127dd218d035fa142292fd2279"}]},"branch":"refs/heads/master"},"65375b6cbf891e971ad2578df7cc82685761cc94":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-02-27 20:22:59.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/36/639736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/639736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/639736/2"}}},"commit":{"parents":[{"commit":"612b042e64d9fcffb983f958ed0d6de6a9f765d3","subject":"Merge \"L7rule support client certificate cases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/612b042e64d9fcffb983f958ed0d6de6a9f765d3"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 16:34:02.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 20:22:05.000000000","tz":0},"subject":"Add support for the oslo_middleware.http_proxy_to_wsgi","message":"Add support for the oslo_middleware.http_proxy_to_wsgi\n\nAdd oslo_middleware.http_proxy_to_wsgi for application\n\nIt can be enabled with the option in octavia.conf:\n[oslo_middleware] enable_proxy_headers_parsing\u003dTrue\n\nTODO:\n1. release note\n\nStory: 2005105\nTask: 29732\n\nChange-Id: I276188530a83598ed75560f02ed9d80ce9afca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65375b6cbf891e971ad2578df7cc82685761cc94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/65375b6cbf891e971ad2578df7cc82685761cc94"}]},"branch":"refs/heads/master"},"3e45a47564b38452b829d08c8dd3163269fe308c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-02-27 20:23:46.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/36/639736/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/639736/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/639736/3"}}},"commit":{"parents":[{"commit":"612b042e64d9fcffb983f958ed0d6de6a9f765d3","subject":"Merge \"L7rule support client certificate cases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/612b042e64d9fcffb983f958ed0d6de6a9f765d3"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 16:34:02.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 20:23:43.000000000","tz":0},"subject":"Add support for the oslo_middleware.http_proxy_to_wsgi","message":"Add support for the oslo_middleware.http_proxy_to_wsgi\n\nAdd oslo_middleware http_proxy_to_wsgi for application\n\nIt can be enabled with the option in octavia.conf:\n[oslo_middleware] enable_proxy_headers_parsing\u003dTrue\n\nTODO:\n1. release note\n\nStory: 2005105\nTask: 29732\n\nChange-Id: I276188530a83598ed75560f02ed9d80ce9afca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e45a47564b38452b829d08c8dd3163269fe308c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e45a47564b38452b829d08c8dd3163269fe308c"}]},"branch":"refs/heads/master"},"6a79892838485e697ae56ece3e72478c6aea6bc7":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-02-27 20:24:17.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/36/639736/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/639736/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/639736/4"}}},"commit":{"parents":[{"commit":"612b042e64d9fcffb983f958ed0d6de6a9f765d3","subject":"Merge \"L7rule support client certificate cases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/612b042e64d9fcffb983f958ed0d6de6a9f765d3"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 16:34:02.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 20:24:15.000000000","tz":0},"subject":"Add support for the oslo_middleware http_proxy_to_wsgi","message":"Add support for the oslo_middleware http_proxy_to_wsgi\n\nAdd oslo_middleware.http_proxy_to_wsgi for application\n\nIt can be enabled with the option in octavia.conf:\n[oslo_middleware] enable_proxy_headers_parsing\u003dTrue\n\nTODO:\n1. release note\n\nStory: 2005105\nTask: 29732\n\nChange-Id: I276188530a83598ed75560f02ed9d80ce9afca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a79892838485e697ae56ece3e72478c6aea6bc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a79892838485e697ae56ece3e72478c6aea6bc7"}]},"branch":"refs/heads/master"},"7f88d499a022fbc77c56dfd349695d620cf8759b":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-02-27 20:36:40.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/36/639736/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/639736/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/639736/5"}}},"commit":{"parents":[{"commit":"612b042e64d9fcffb983f958ed0d6de6a9f765d3","subject":"Merge \"L7rule support client certificate cases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/612b042e64d9fcffb983f958ed0d6de6a9f765d3"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 16:34:02.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 20:36:12.000000000","tz":0},"subject":"Add support for the oslo_middleware http_proxy_to_wsgi","message":"Add support for the oslo_middleware http_proxy_to_wsgi\n\nThis sets up the HTTPProxyToWSGI middleware in front of Octavia API. The\npurpose of this middleware is to set up the request URL correctly in\nthe case there is a proxy (For instance, a loadbalancer such as HAProxy)\nin front of Octavia API.\n\nSo, when TLS connections are terminated at the proxy, and one tries to\nget the versions from the \u0027/\u0027 resource from Octavia API, one will notice\nthat the protocol is incorrect; It will show \u0027http\u0027 instead of \u0027https\u0027.\nSo this middleware handles such cases.\n\nThe HTTPProxyToWSGI is off by default and needs to be enabled via a\nconfiguration value.\n\nIt can be enabled with the option in octavia.conf:\n[oslo_middleware]\nenable_proxy_headers_parsing\u003dTrue\n\nStory: 2005105\nTask: 29732\nChange-Id: I276188530a83598ed75560f02ed9d80ce9afca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f88d499a022fbc77c56dfd349695d620cf8759b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f88d499a022fbc77c56dfd349695d620cf8759b"}]},"branch":"refs/heads/master"},"6d327ecce5b63dd09edfdf0405f75034f0f043e2":{"kind":"REWORK","_number":6,"created":"2019-02-27 21:18:41.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/36/639736/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/639736/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/639736/6"}}},"commit":{"parents":[{"commit":"612b042e64d9fcffb983f958ed0d6de6a9f765d3","subject":"Merge \"L7rule support client certificate cases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/612b042e64d9fcffb983f958ed0d6de6a9f765d3"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 16:34:02.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 21:18:34.000000000","tz":0},"subject":"Add support for the oslo_middleware http_proxy_to_wsgi","message":"Add support for the oslo_middleware http_proxy_to_wsgi\n\nThis sets up the HTTPProxyToWSGI middleware in front of Octavia API. The\npurpose of this middleware is to set up the request URL correctly in\nthe case there is a proxy (For instance, a loadbalancer such as HAProxy)\nin front of Octavia API.\n\nSo, when TLS connections are terminated at the proxy, and one tries to\nget the versions from the \u0027/\u0027 resource from Octavia API, one will notice\nthat the protocol is incorrect; It will show \u0027http\u0027 instead of \u0027https\u0027.\nSo this middleware handles such cases.\n\nThe HTTPProxyToWSGI is off by default and needs to be enabled via a\nconfiguration value.\n\nIt can be enabled with the option in octavia.conf:\n[oslo_middleware]\nenable_proxy_headers_parsing\u003dTrue\n\nStory: 2005105\nTask: 29732\nChange-Id: I276188530a83598ed75560f02ed9d80ce9afca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d327ecce5b63dd09edfdf0405f75034f0f043e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d327ecce5b63dd09edfdf0405f75034f0f043e2"}]},"branch":"refs/heads/master"},"2e23b6680febb67c5347f652310e43c549196fba":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-01 00:03:10.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/36/639736/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/639736/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/639736/7"}}},"commit":{"parents":[{"commit":"4e72fa9c2e86f863a0ea4af2a4d7950f873ba8cd","subject":"Merge \"Add missing test requirement \"fixtures\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e72fa9c2e86f863a0ea4af2a4d7950f873ba8cd"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 16:34:02.000000000","tz":180},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-01 00:03:10.000000000","tz":0},"subject":"Add support for the oslo_middleware http_proxy_to_wsgi","message":"Add support for the oslo_middleware http_proxy_to_wsgi\n\nThis sets up the HTTPProxyToWSGI middleware in front of Octavia API. The\npurpose of this middleware is to set up the request URL correctly in\nthe case there is a proxy (For instance, a loadbalancer such as HAProxy)\nin front of Octavia API.\n\nSo, when TLS connections are terminated at the proxy, and one tries to\nget the versions from the \u0027/\u0027 resource from Octavia API, one will notice\nthat the protocol is incorrect; It will show \u0027http\u0027 instead of \u0027https\u0027.\nSo this middleware handles such cases.\n\nThe HTTPProxyToWSGI is off by default and needs to be enabled via a\nconfiguration value.\n\nIt can be enabled with the option in octavia.conf:\n[oslo_middleware]\nenable_proxy_headers_parsing\u003dTrue\n\nStory: 2005105\nTask: 29732\nChange-Id: I276188530a83598ed75560f02ed9d80ce9afca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e23b6680febb67c5347f652310e43c549196fba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e23b6680febb67c5347f652310e43c549196fba"}]},"branch":"refs/heads/master"},"ec83c69372d647aaeec02fea6eea29c7ee11115a":{"kind":"REWORK","_number":8,"created":"2019-08-01 00:11:35.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/36/639736/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/639736/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/639736/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/639736/8"}}},"commit":{"parents":[{"commit":"4e72fa9c2e86f863a0ea4af2a4d7950f873ba8cd","subject":"Merge \"Add missing test requirement \"fixtures\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e72fa9c2e86f863a0ea4af2a4d7950f873ba8cd"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-27 16:34:02.000000000","tz":180},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-01 00:10:54.000000000","tz":-420},"subject":"Add support for oslo_middleware http_proxy_to_wsgi","message":"Add support for oslo_middleware http_proxy_to_wsgi\n\nThis sets up the HTTPProxyToWSGI middleware in front of Octavia API. The\npurpose of this middleware is to set up the request URL correctly in\nthe case there is a proxy (For instance, a loadbalancer such as HAProxy)\nin front of Octavia API.\n\nSo, when TLS connections are terminated at the proxy, and one tries to\nget the versions from the \u0027/\u0027 resource from Octavia API, one will notice\nthat the protocol is incorrect; It will show \u0027http\u0027 instead of \u0027https\u0027.\nSo this middleware handles such cases.\n\nThe HTTPProxyToWSGI is off by default and needs to be enabled via a\nconfiguration value.\n\nIt can be enabled with the option in octavia.conf:\n[oslo_middleware]\nenable_proxy_headers_parsing\u003dTrue\n\nStory: 2005105\nTask: 29732\nChange-Id: I276188530a83598ed75560f02ed9d80ce9afca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec83c69372d647aaeec02fea6eea29c7ee11115a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec83c69372d647aaeec02fea6eea29c7ee11115a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
