)]}'
{"id":"openstack%2Foctavia~648069","triplet_id":"openstack%2Foctavia~stable%2Fstein~I2346c8a3f2fa4b4eb8a004b4aaab93986228c371","project":"openstack/octavia","branch":"stable/stein","topic":"628016-stable/stein","hashtags":[],"change_id":"I2346c8a3f2fa4b4eb8a004b4aaab93986228c371","subject":"Enabling SG customization on loadbalancer listerners","status":"ABANDONED","created":"2019-03-27 10:16:47.000000000","updated":"2020-07-24 11:50:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"905030975c7242dd98672228c707cddc06a13e98","_number":648069,"virtual_id_number":648069,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-09-18 17:22:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2019-09-18 13:48:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-1,"date":"2019-09-18 13:49:04.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 13:49:04.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-18 17:22:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7013bdc6d4e21577b423a500bfc66d7b02d002ff","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-03-27 10:16:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"57b6cc3d2c4362fba19b8479650b07cb20d4d9b2","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-03-27 10:18:36.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0aea9129cb47d2e9db11e9db79f1665354f4a5f1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-03-27 10:19:20.000000000","message":"Patch Set 1: Code-Review-2 Workflow-1\n\nIn line with same patch in other branches, this patch is not expected to be merged upstream. It is only used to validate from time to time it does not break tests.","accounts_in_message":[],"_revision_number":1},{"id":"b5edcbc76fac51f0acc0ec8f70342f7b4fd4a8c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 12:22:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/69/648069/1/check/openstack-tox-cover/cfaca83/cover/ : SUCCESS in 8m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/648069/1/check/openstack-tox-lower-constraints/7b74cc4/ : SUCCESS in 11m 18s\n- openstack-tox-pep8 http://logs.openstack.org/69/648069/1/check/openstack-tox-pep8/4c9265a/ : SUCCESS in 6m 52s\n- openstack-tox-py27 http://logs.openstack.org/69/648069/1/check/openstack-tox-py27/f2cac8d/ : SUCCESS in 5m 51s\n- openstack-tox-py35 http://logs.openstack.org/69/648069/1/check/openstack-tox-py35/0fca30a/ : SUCCESS in 5m 48s\n- openstack-tox-py36 http://logs.openstack.org/69/648069/1/check/openstack-tox-py36/d26f9b7/ : SUCCESS in 6m 13s\n- openstack-tox-docs http://logs.openstack.org/69/648069/1/check/openstack-tox-docs/e5da337/html/ : SUCCESS in 6m 44s\n- build-openstack-releasenotes http://logs.openstack.org/69/648069/1/check/build-openstack-releasenotes/184704f/html/ : SUCCESS in 5m 33s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/69/648069/1/check/neutron-lbaasv2-dsvm-api/cea0174/ : SUCCESS in 1h 14m 24s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/69/648069/1/check/neutron-lbaasv2-dsvm-py3x-api/a699d1d/ : SUCCESS in 1h 17m 03s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/69/648069/1/check/neutron-lbaasv2-dsvm-scenario/fa4a422/ : SUCCESS in 1h 29m 40s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/69/648069/1/check/neutron-lbaasv2-dsvm-py3x-scenario/8d203b4/ : SUCCESS in 1h 40m 57s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/69/648069/1/check/openstack-tox-functional/7dfc7d6/ : SUCCESS in 9m 20s\n- openstack-tox-functional-py35 http://logs.openstack.org/69/648069/1/check/openstack-tox-functional-py35/9975241/ : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 http://logs.openstack.org/69/648069/1/check/openstack-tox-functional-py36/bb3db08/ : SUCCESS in 8m 10s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/69/648069/1/check/octavia-v1-dsvm-scenario/2dce297/ : SUCCESS in 1h 10m 53s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/69/648069/1/check/octavia-v1-dsvm-py3x-scenario/70e203a/ : SUCCESS in 56m 36s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/69/648069/1/check/octavia-v2-dsvm-noop-api/b0beb4b/ : SUCCESS in 42m 13s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/69/648069/1/check/octavia-v2-dsvm-noop-py2-api/0833e47/ : SUCCESS in 43m 17s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/69/648069/1/check/octavia-v2-dsvm-scenario/2b9e19e/ : SUCCESS in 1h 37m 49s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/69/648069/1/check/octavia-v2-dsvm-py2-scenario/2350acf/ : SUCCESS in 1h 31m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/69/648069/1/check/octavia-v2-dsvm-py2-scenario-centos-7/009fbb1/ : FAILURE in 1h 58m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/69/648069/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6a6b617/ : SUCCESS in 1h 30m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/69/648069/1/check/octavia-v2-dsvm-scenario-two-node/6c4ff35/ : FAILURE in 1h 52m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/69/648069/1/check/octavia-v2-dsvm-py2-scenario-two-node/b322135/ : SUCCESS in 1h 36m 16s (non-voting)\n- octavia-grenade http://logs.openstack.org/69/648069/1/check/octavia-grenade/4a40faf/ : SUCCESS in 1h 15m 23s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/69/648069/1/check/octavia-v2-dsvm-tls-barbican/ab14abc/ : SUCCESS in 45m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b998a94cd64a88917a25a7c55720b6f92f1bee3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-18 13:48:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"367f2e1ddbbaafb22f7fb23d095b55ca659d3ada","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-18 13:49:04.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"e5036e3d7cabf8000a5db080772d8d229a2d0bc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:22:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8552b326be24ecc80dfc33cdc537a29 : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8f884331a964989bd566dbf65de6f2c : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed4e17a9988a4de49a91dd2ca880fdda : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d1939f644cd4fa083b8529308349c2c : SUCCESS in 5m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0fd6ff25e90d4a22942e85aa4620b1f9 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61a0ad4e886644f5b8d2dd46753d316c : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d81b616a0b4a79bf21bfc82ac2ec7b : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1a629c9984347a7a29be6bbc15d8b03 : SUCCESS in 5m 49s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/83fbb9a7a8ed46d9aefc494bf8ad54ee : SUCCESS in 1h 23m 21s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/81c661f63cf040579d9dfe0ff0e77556 : SUCCESS in 1h 26m 00s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/056cfc69535644a5a2d41737e192d02a : SUCCESS in 1h 28m 16s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/28ab9d07657b43bebfd52d0f256cb7a3 : SUCCESS in 1h 39m 04s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a70471521a4e40a898b1b4bd46c6d0d2 : SUCCESS in 5m 21s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7b214a06acc5434dba9f03de19a0d613 : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/107902b3e28b44e0b0fa53a2622c5098 : SUCCESS in 5m 57s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5efe6ade15334066b405f2682634f4f9 : SUCCESS in 1h 05m 02s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f396e97047144651b8359e0873c28c53 : SUCCESS in 1h 05m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1a91d92def834d9fb7e78fb9b4474e16 : SUCCESS in 45m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d11f2244c3e64dde8d2545f5d79e8660 : SUCCESS in 45m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b346446cb6734e7e963ef9d5ba93f87e : SUCCESS in 1h 31m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4b5da8b93bec45deb91fa5886d369c90 : SUCCESS in 1h 26m 27s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/e681fdee038c4dc09a18118e151d1072 : SUCCESS in 1h 46m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a44caec39c6e45b5b3d0f3e7c68e1ff9 : FAILURE in 49m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c151334e6c94cfd84d63332674a2d39 : FAILURE in 40m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c7b8b9842af84321913ccd5e36d9c325 : FAILURE in 1h 31m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3b49dbdcbe2d483c87ae53a5a7d3615a : FAILURE in 1h 33m 08s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5b6df065975d48ac8b772e39d7985de3 : SUCCESS in 1h 17m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2b78b944c14745bfbe010f6622ddf5a4 : SUCCESS in 53m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"349ec37ac62410338f62c50cb39618a32a263ca5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:50:25.000000000","message":"Abandoned\n\nTrain Octavia extended the listener API to allow ingress from a list of CIDRs.\n\nReferences:\n- https://review.opendev.org/#/c/659626/\n- https://docs.openstack.org/api-ref/load-balancer/v2/#listeners","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9a259f0341040f81aa4c015d5cf3fac20c21e46","revisions":{"f837da04d240351ff4d4a6e36eac2025586cbfe4":{"kind":"REWORK","_number":1,"created":"2019-03-27 10:16:47.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/69/648069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/648069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/648069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/648069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/648069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/648069/1"}}},"commit":{"parents":[{"commit":"1b83d56ab0175183e7dd03b52939d8f109c54961","subject":"Update UPPER_CONSTRAINTS_FILE for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b83d56ab0175183e7dd03b52939d8f109c54961"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-09-14 07:39:39.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-03-27 10:16:47.000000000","tz":0},"subject":"Enabling SG customization on loadbalancer listerners","message":"Enabling SG customization on loadbalancer listerners\n\nBy changing the SG ownership to belong to the tenant creating the\nloadbalancer we allow the tenant to define more fine grain access\npolicies their loadbalancer, such as only enabling access to a given\nport/protocol from a given remote id or subnet CIDR -- instead of\nhaving to open that port/protocol from everywhere.\n\nAs this security group is only used for the amphora VM created for\nthe specific load balancer, and the user could already open access\nto it through listeners on any port, protocol, giving full access to\nmodify the SG will not entitle any extra security concern.\n\nNote this can be a temporal solution until Octavia enables this\nflexibility through their own API\n\nChange-Id: I2346c8a3f2fa4b4eb8a004b4aaab93986228c371\nStory: 2003686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f837da04d240351ff4d4a6e36eac2025586cbfe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f837da04d240351ff4d4a6e36eac2025586cbfe4"}]},"branch":"refs/heads/stable/stein"},"d9a259f0341040f81aa4c015d5cf3fac20c21e46":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-18 13:48:58.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/69/648069/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/648069/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/648069/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/648069/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/648069/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/648069/2"}}},"commit":{"parents":[{"commit":"f556b84cecf04a36d4a7369c629deb8ed3e774d9","subject":"Merge \"Fix catching driver exceptions\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f556b84cecf04a36d4a7369c629deb8ed3e774d9"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-09-14 07:39:39.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-18 13:48:58.000000000","tz":0},"subject":"Enabling SG customization on loadbalancer listerners","message":"Enabling SG customization on loadbalancer listerners\n\nBy changing the SG ownership to belong to the tenant creating the\nloadbalancer we allow the tenant to define more fine grain access\npolicies their loadbalancer, such as only enabling access to a given\nport/protocol from a given remote id or subnet CIDR -- instead of\nhaving to open that port/protocol from everywhere.\n\nAs this security group is only used for the amphora VM created for\nthe specific load balancer, and the user could already open access\nto it through listeners on any port, protocol, giving full access to\nmodify the SG will not entitle any extra security concern.\n\nNote this can be a temporal solution until Octavia enables this\nflexibility through their own API\n\nChange-Id: I2346c8a3f2fa4b4eb8a004b4aaab93986228c371\nStory: 2003686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9a259f0341040f81aa4c015d5cf3fac20c21e46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9a259f0341040f81aa4c015d5cf3fac20c21e46"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
