)]}'
{"id":"openstack%2Foctavia~649967","triplet_id":"openstack%2Foctavia~master~I05b31ba628bafeefec36cc5c000dae1aefb63d67","project":"openstack/octavia","branch":"master","topic":"centos_active_standby","hashtags":[],"change_id":"I05b31ba628bafeefec36cc5c000dae1aefb63d67","subject":"Fix IPv6 in Active/Standby topology on CentOS","status":"MERGED","created":"2019-04-04 11:24:12.000000000","updated":"2019-04-29 12:21:03.000000000","submitted":"2019-04-25 14:06:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"649967-1556201204459-332a2ac9","meta_rev_id":"5e6ae7713f5b611e9fbcdac39783a8adf032a907","_number":649967,"virtual_id_number":649967,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2019-04-04 14:27:52.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-04-25 14:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-04-25 11:50:51.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-04-24 22:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2019-04-25 11:50:51.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 22:16:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-04-25 11:50:51.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-04-25 14:06:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23a7d5e34e01ed90ed13c9589374bbfb80a90094","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-04-04 11:24:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c71c0331be9976fab643aaa3059bc5ed9c64b3ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 13:48:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/67/649967/1/check/openstack-tox-cover/6d8dbb3/cover/ : SUCCESS in 11m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/649967/1/check/openstack-tox-lower-constraints/2658093/ : SUCCESS in 8m 45s\n- openstack-tox-pep8 http://logs.openstack.org/67/649967/1/check/openstack-tox-pep8/dafe4b5/ : SUCCESS in 8m 28s\n- openstack-tox-py27 http://logs.openstack.org/67/649967/1/check/openstack-tox-py27/a30f6e3/ : SUCCESS in 6m 29s\n- openstack-tox-py35 http://logs.openstack.org/67/649967/1/check/openstack-tox-py35/c36c195/ : SUCCESS in 6m 30s\n- openstack-tox-py36 http://logs.openstack.org/67/649967/1/check/openstack-tox-py36/5795f67/ : SUCCESS in 7m 32s\n- openstack-tox-docs http://logs.openstack.org/67/649967/1/check/openstack-tox-docs/e376c16/html/ : SUCCESS in 7m 09s\n- build-openstack-releasenotes http://logs.openstack.org/67/649967/1/check/build-openstack-releasenotes/1fb9617/html/ : SUCCESS in 7m 14s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/67/649967/1/check/neutron-lbaasv2-dsvm-api/a0bd386/ : SUCCESS in 1h 20m 11s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/67/649967/1/check/neutron-lbaasv2-dsvm-py3x-api/d86061a/ : SUCCESS in 1h 16m 32s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/67/649967/1/check/neutron-lbaasv2-dsvm-scenario/fa0f553/ : SUCCESS in 1h 22m 04s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/67/649967/1/check/neutron-lbaasv2-dsvm-py3x-scenario/84086f2/ : FAILURE in 1h 23m 53s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/67/649967/1/check/openstack-tox-functional/e593667/ : SUCCESS in 9m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/67/649967/1/check/openstack-tox-functional-py35/3d71b99/ : SUCCESS in 8m 25s\n- openstack-tox-functional-py36 http://logs.openstack.org/67/649967/1/check/openstack-tox-functional-py36/32f617f/ : SUCCESS in 7m 25s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/67/649967/1/check/octavia-v1-dsvm-scenario/16de785/ : FAILURE in 1h 03m 27s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/67/649967/1/check/octavia-v1-dsvm-py3x-scenario/950af53/ : SUCCESS in 1h 10m 00s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-noop-api/bd9d5dc/ : SUCCESS in 42m 52s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-noop-py2-api/54a941c/ : SUCCESS in 39m 36s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-scenario/768f557/ : SUCCESS in 1h 38m 08s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-py2-scenario/053a53a/ : SUCCESS in 1h 30m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-py2-scenario-centos-7/826010b/ : FAILURE in 2h 04m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/a7a0a32/ : SUCCESS in 1h 30m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-scenario-two-node/5ebe9cf/ : SUCCESS in 1h 43m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-py2-scenario-two-node/fed3ede/ : FAILURE in 1h 28m 09s (non-voting)\n- octavia-grenade http://logs.openstack.org/67/649967/1/check/octavia-grenade/246bbc1/ : SUCCESS in 1h 14m 33s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-tls-barbican/6f600aa/ : FAILURE in 13m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3e080093f4147e6fc0504765ed4d04a35b2652b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-04-04 14:27:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ebf9a311a68e891f026de48881b438877b508079","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 16:55:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/67/649967/1/check/openstack-tox-cover/dbf89d4/cover/ : SUCCESS in 10m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/649967/1/check/openstack-tox-lower-constraints/9fb2690/ : SUCCESS in 9m 32s\n- openstack-tox-pep8 http://logs.openstack.org/67/649967/1/check/openstack-tox-pep8/6f3ef87/ : SUCCESS in 8m 01s\n- openstack-tox-py27 http://logs.openstack.org/67/649967/1/check/openstack-tox-py27/abd7521/ : SUCCESS in 7m 22s\n- openstack-tox-py35 http://logs.openstack.org/67/649967/1/check/openstack-tox-py35/12d8e30/ : SUCCESS in 7m 08s\n- openstack-tox-py36 http://logs.openstack.org/67/649967/1/check/openstack-tox-py36/b740d8b/ : SUCCESS in 7m 11s\n- openstack-tox-docs http://logs.openstack.org/67/649967/1/check/openstack-tox-docs/8ef3b06/html/ : SUCCESS in 7m 56s\n- build-openstack-releasenotes http://logs.openstack.org/67/649967/1/check/build-openstack-releasenotes/ba39855/html/ : SUCCESS in 5m 47s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/67/649967/1/check/neutron-lbaasv2-dsvm-api/aa558ca/ : SUCCESS in 1h 20m 36s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/67/649967/1/check/neutron-lbaasv2-dsvm-py3x-api/c2c4c5a/ : SUCCESS in 1h 23m 28s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/67/649967/1/check/neutron-lbaasv2-dsvm-scenario/0b34cda/ : FAILURE in 2h 03m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/67/649967/1/check/neutron-lbaasv2-dsvm-py3x-scenario/ad6cb07/ : SUCCESS in 1h 25m 41s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/67/649967/1/check/openstack-tox-functional/ed1988d/ : SUCCESS in 9m 04s\n- openstack-tox-functional-py35 http://logs.openstack.org/67/649967/1/check/openstack-tox-functional-py35/d26be2d/ : SUCCESS in 8m 57s\n- openstack-tox-functional-py36 http://logs.openstack.org/67/649967/1/check/openstack-tox-functional-py36/3b4cfee/ : SUCCESS in 10m 13s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/67/649967/1/check/octavia-v1-dsvm-scenario/968599c/ : SUCCESS in 1h 06m 03s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/67/649967/1/check/octavia-v1-dsvm-py3x-scenario/d71e768/ : SUCCESS in 1h 04m 29s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-noop-api/b05911d/ : SUCCESS in 45m 17s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-noop-py2-api/ef063df/ : SUCCESS in 47m 40s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-scenario/771f875/ : SUCCESS in 1h 41m 15s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-py2-scenario/9de74c2/ : SUCCESS in 1h 26m 12s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-py2-scenario-centos-7/c353100/ : SUCCESS in 2h 10m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/aaf6b98/ : SUCCESS in 1h 25m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-scenario-two-node/faf7c44/ : FAILURE in 1h 32m 52s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-py2-scenario-two-node/e793446/ : FAILURE in 1h 27m 03s (non-voting)\n- octavia-grenade http://logs.openstack.org/67/649967/1/check/octavia-grenade/089ae82/ : SUCCESS in 1h 20m 06s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/67/649967/1/check/octavia-v2-dsvm-tls-barbican/eeda2a8/ : FAILURE in 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5216cc3bb72e841e3d52c56d7b6944164bad599","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-04-24 22:16:59.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"7a3c65c8a940f34d48a2f692ee95468b196bf554","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-04-25 11:50:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20fb373535accc32783409ad2b9aecb56cfac218","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 11:51:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adb1a64976ab1bfae5fb6bddc2a8992e187eff7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 14:06:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/67/649967/1/gate/openstack-tox-lower-constraints/8064413/ : SUCCESS in 7m 34s\n- openstack-tox-pep8 http://logs.openstack.org/67/649967/1/gate/openstack-tox-pep8/cfee210/ : SUCCESS in 6m 17s\n- openstack-tox-py27 http://logs.openstack.org/67/649967/1/gate/openstack-tox-py27/481322d/ : SUCCESS in 5m 38s\n- openstack-tox-py35 http://logs.openstack.org/67/649967/1/gate/openstack-tox-py35/53544bd/ : SUCCESS in 5m 29s\n- openstack-tox-py36 http://logs.openstack.org/67/649967/1/gate/openstack-tox-py36/b1a643f/ : SUCCESS in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/67/649967/1/gate/openstack-tox-docs/62a30b4/html/ : SUCCESS in 6m 18s\n- build-openstack-releasenotes http://logs.openstack.org/67/649967/1/gate/build-openstack-releasenotes/a836873/html/ : SUCCESS in 5m 28s\n- openstack-tox-functional http://logs.openstack.org/67/649967/1/gate/openstack-tox-functional/2ab78be/ : SUCCESS in 6m 49s\n- openstack-tox-functional-py35 http://logs.openstack.org/67/649967/1/gate/openstack-tox-functional-py35/124dd84/ : SUCCESS in 8m 01s\n- openstack-tox-functional-py36 http://logs.openstack.org/67/649967/1/gate/openstack-tox-functional-py36/5b0239b/ : SUCCESS in 7m 13s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/67/649967/1/gate/octavia-v1-dsvm-scenario/4cf1237/ : SUCCESS in 1h 02m 06s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/67/649967/1/gate/octavia-v1-dsvm-py3x-scenario/3a93b4a/ : SUCCESS in 1h 09m 25s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/67/649967/1/gate/neutron-lbaasv2-dsvm-api/690f768/ : SUCCESS in 1h 18m 13s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/67/649967/1/gate/neutron-lbaasv2-dsvm-py3x-api/3c88aea/ : SUCCESS in 1h 24m 59s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/67/649967/1/gate/octavia-v2-dsvm-noop-api/964cff5/ : SUCCESS in 43m 25s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/67/649967/1/gate/octavia-v2-dsvm-noop-py2-api/ac36daf/ : SUCCESS in 42m 26s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/67/649967/1/gate/octavia-v2-dsvm-scenario/b4bdd27/ : SUCCESS in 1h 34m 48s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/67/649967/1/gate/octavia-v2-dsvm-py2-scenario/972ced4/ : SUCCESS in 1h 37m 32s\n- octavia-grenade http://logs.openstack.org/67/649967/1/gate/octavia-grenade/2480f15/ : SUCCESS in 1h 16m 42s","accounts_in_message":[],"_revision_number":1},{"id":"556e770070e59735e80e24d72513ca19a32cdce4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 14:06:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"11799bc34fa41acf77f4cbbb7fa51b0299482962","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-04-29 12:20:14.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 7dc3cb975ffd050ebcafd30956dae3d174b50c80","accounts_in_message":[],"_revision_number":1},{"id":"2a45ee1d4ef2470293593b8425544367a23faf3e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-04-29 12:20:47.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit cedb553b144688b0de6440e0e1a5eed0ff838b51","accounts_in_message":[],"_revision_number":1},{"id":"3173e61ad9126019b80fe05a489719b243df08e8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-04-29 12:21:03.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit de4eeff89a1c5ad7b4bf7338b1a13657cc0f56a7","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"951afb9a0a94413f8d452fa5516cf4dc412968a7","revisions":{"951afb9a0a94413f8d452fa5516cf4dc412968a7":{"kind":"REWORK","_number":1,"created":"2019-04-04 11:24:12.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/67/649967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/649967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/649967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/649967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/649967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/649967/1"}}},"commit":{"parents":[{"commit":"a728bc000f65e431dc57fef61d41e5ba63d72b02","subject":"Merge \"Fix setting of VIP QoS policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a728bc000f65e431dc57fef61d41e5ba63d72b02"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-04-02 12:24:53.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-04-04 11:23:56.000000000","tz":120},"subject":"Fix IPv6 in Active/Standby topology on CentOS","message":"Fix IPv6 in Active/Standby topology on CentOS\n\nAvoid duplicated IPv6 VIP addresses on CentOS in Active/Standby mode,\nKeepalived will set/unset the VIP address.\n\nStory: 2005365\nTask: 30340\n\nChange-Id: I05b31ba628bafeefec36cc5c000dae1aefb63d67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/951afb9a0a94413f8d452fa5516cf4dc412968a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/951afb9a0a94413f8d452fa5516cf4dc412968a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
