)]}'
{"id":"openstack%2Foctavia~650896","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I848b7fc69b977fcb39f8a07e2ea5fc7bd37b5c7a","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"I848b7fc69b977fcb39f8a07e2ea5fc7bd37b5c7a","subject":"Fix amp failover where failover already failed","status":"ABANDONED","created":"2019-04-08 13:45:23.000000000","updated":"2019-10-30 16:41:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"034dd27bd2268628846298bbf9dff123500ceae2","_number":650896,"virtual_id_number":650896,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-1,"date":"2019-08-22 17:58:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":-1,"date":"2019-08-22 16:05:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 16:05:07.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-22 17:58:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3689126731b18cc9c7680a75656eeee62265c6ad","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-08 13:45:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49411b12a05411cea18254bc3eca8981f326196a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-04-08 14:54:47.000000000","message":"Patch Set 1:\n\nThis patch depends on Amphora.updated_at which was introduced in Rocky and is non-backportable as it alters database schema and API. Unless we find another way to circumvent the need for such attribute, we cannot backport this patch.\n\nTypeError: \u0027updated_at\u0027 is an invalid keyword argument for Amphora","accounts_in_message":[],"_revision_number":1},{"id":"e886c4a6838648804e3446bb8bb761319d7eee30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 15:44:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/96/650896/1/check/openstack-tox-pep8/bf5f042/ : SUCCESS in 17m 04s\n- openstack-tox-py27 http://logs.openstack.org/96/650896/1/check/openstack-tox-py27/8deaa87/ : FAILURE in 8m 01s\n- openstack-tox-py35 http://logs.openstack.org/96/650896/1/check/openstack-tox-py35/af217d5/ : FAILURE in 8m 35s\n- openstack-tox-docs http://logs.openstack.org/96/650896/1/check/openstack-tox-docs/a207d14/html/ : SUCCESS in 7m 55s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/96/650896/1/check/neutron-lbaasv2-dsvm-api/3a7a96a/ : SUCCESS in 1h 23m 38s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/96/650896/1/check/neutron-lbaasv2-dsvm-py3x-api/07d0f64/ : SUCCESS in 1h 20m 32s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/96/650896/1/check/neutron-lbaasv2-dsvm-scenario/b836f65/ : SUCCESS in 1h 24m 30s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/96/650896/1/check/neutron-lbaasv2-dsvm-py3x-scenario/605f21f/ : SUCCESS in 1h 25m 08s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/96/650896/1/check/openstack-tox-cover/1f28f5a/ : FAILURE in 7m 28s\n- openstack-tox-functional http://logs.openstack.org/96/650896/1/check/openstack-tox-functional/1104089/ : SUCCESS in 8m 32s\n- openstack-tox-functional-py35 http://logs.openstack.org/96/650896/1/check/openstack-tox-functional-py35/ee1dc51/ : SUCCESS in 10m 27s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/96/650896/1/check/octavia-v1-dsvm-scenario/b0ea56e/ : SUCCESS in 1h 05m 16s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/96/650896/1/check/octavia-v1-dsvm-py3x-scenario/b20a695/ : SUCCESS in 1h 06m 46s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/96/650896/1/check/octavia-v1-dsvm-scenario-multinode/edcdf76/ : SUCCESS in 1h 15m 02s\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/96/650896/1/check/octavia-v1-dsvm-py3x-scenario-multinode/78565dd/ : SUCCESS in 1h 08m 01s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/96/650896/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/7a908ac/ : SUCCESS in 1h 08m 40s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/96/650896/1/check/octavia-v2-dsvm-noop-api/ead5298/ : SUCCESS in 45m 06s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/96/650896/1/check/octavia-v2-dsvm-noop-py35-api/9f5226a/ : SUCCESS in 43m 50s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/96/650896/1/check/octavia-v2-dsvm-scenario/a4f1068/ : SUCCESS in 1h 29m 09s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/96/650896/1/check/octavia-v2-dsvm-py35-scenario/e98588d/ : SUCCESS in 1h 32m 54s","accounts_in_message":[],"_revision_number":1},{"id":"aec9a31035da1c4e909093bd6037be03cdfa43bd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-01 19:58:50.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ab313cd5b12a9015ec020dddf8a33a512f16328d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-22 16:04:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0b6c58c0356db47f8273aa8ae90d05547606eb52","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-22 16:05:07.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"6af4d620c6276cad8d11cfc8c1632aafbaa99a9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 17:58:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f9ccf6c3ee44b4f9dd4dc5bd12f2375 : SUCCESS in 10m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6fe76d3bba744626b08bd745a6b1b79d : FAILURE in 6m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c8f13d53ac504d90b4e17a6ca97581f1 : FAILURE in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3cedfc4081c45aa9e9f5b57f894ca7d : SUCCESS in 7m 41s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/dce302d1745843ddba7947a0c985f7cf : SUCCESS in 1h 16m 45s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ebb3130bad054751ab8fff2a86bfb6db : SUCCESS in 1h 19m 28s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9068e954bcd428bb97a3082959ff6fd : SUCCESS in 1h 24m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fdc2d6ab948c42c585c65b45771a62f7 : SUCCESS in 1h 18m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0e94410b0e64f3db61a2164651f0379 : FAILURE in 6m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d5339bb30e114dd1b4f207535fb36a43 : SUCCESS in 11m 03s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2cf7b08fc5214b1c985e511bb4937515 : SUCCESS in 10m 01s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5ce594c0b2841b39b2f3e48a4784ffb : SUCCESS in 1h 01m 27s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/61ecefce322d49618854d23473ede66d : SUCCESS in 56m 10s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/7280da4cf83640cc9fd90b33e2347cf4 : SUCCESS in 1h 15m 50s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/61ba797e250b47578b608ac7a12cb738 : SUCCESS in 1h 23m 05s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/609152cc0e7146139141fa67b4dd4be9 : SUCCESS in 1h 00m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/032727a865434c50bc6e4598d6e96a93 : SUCCESS in 50m 58s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/8ce1e860fdca4d8a9c0299e7ac952798 : SUCCESS in 45m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/992269c814ca4f8ca23102a42377970a : SUCCESS in 1h 46m 48s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/cf951236b2be4dfdb5dcc84cd2d48ea2 : SUCCESS in 1h 22m 16s","accounts_in_message":[],"_revision_number":2},{"id":"034dd27bd2268628846298bbf9dff123500ceae2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-30 16:41:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ba36f04271cd3beffecc8e8f507695b079068933","revisions":{"4b1f06698f7aa25585c663b736725b6c113af2d1":{"kind":"REWORK","_number":1,"created":"2019-04-08 13:45:23.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/96/650896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/650896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/650896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/650896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/650896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/650896/1"}}},"commit":{"parents":[{"commit":"dfb471cbec8f2bcf2da45a0eef1ea15690dc24b8","subject":"Merge \"Fix setting of VIP QoS policy\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfb471cbec8f2bcf2da45a0eef1ea15690dc24b8"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-03-01 15:46:56.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-04-08 13:45:08.000000000","tz":120},"subject":"Fix amp failover where failover already failed","message":"Fix amp failover where failover already failed\n\nIf a failover ran on an amphora and was unsuccessful and reverted, it\nwould mark the amp status \"DELETED\" and un-busy the health record.\nIt would then be picked up on the next failover check, start failing\nover, and break early since it appeared to be \"unallocated\".\n\nAlso, housekeeping can now clean up expired amphora records based on the\namphora\u0027s updated_at time instead of the health record\u0027s time, which\nmeans the records won\u0027t be immediately cleaned up anymore after they go\nthrough failover flows.\n\nConflicts:\n    octavia/controller/worker/controller_worker.py\n    octavia/controller/worker/flows/amphora_flows.py\n    octavia/tests/unit/controller/worker/flows/test_amphora_flows.py\n    octavia/tests/unit/controller/worker/test_controller_worker.py\n    tools/create_flow_docs.py\n\nChange-Id: I848b7fc69b977fcb39f8a07e2ea5fc7bd37b5c7a\n(cherry picked from commit 96cce3ed74c63e5756e216a8e48c60de09b027c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b1f06698f7aa25585c663b736725b6c113af2d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b1f06698f7aa25585c663b736725b6c113af2d1"}]},"branch":"refs/heads/stable/queens"},"ba36f04271cd3beffecc8e8f507695b079068933":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-22 16:04:32.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/96/650896/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/650896/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/650896/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/650896/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/650896/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/650896/2"}}},"commit":{"parents":[{"commit":"7eb83accd543535065c150c62219ad0567d715b2","subject":"Update tox.ini for new upper constraints strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7eb83accd543535065c150c62219ad0567d715b2"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2018-03-01 15:46:56.000000000","tz":0},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-22 16:04:32.000000000","tz":0},"subject":"Fix amp failover where failover already failed","message":"Fix amp failover where failover already failed\n\nIf a failover ran on an amphora and was unsuccessful and reverted, it\nwould mark the amp status \"DELETED\" and un-busy the health record.\nIt would then be picked up on the next failover check, start failing\nover, and break early since it appeared to be \"unallocated\".\n\nAlso, housekeeping can now clean up expired amphora records based on the\namphora\u0027s updated_at time instead of the health record\u0027s time, which\nmeans the records won\u0027t be immediately cleaned up anymore after they go\nthrough failover flows.\n\nConflicts:\n    octavia/controller/worker/controller_worker.py\n    octavia/controller/worker/flows/amphora_flows.py\n    octavia/tests/unit/controller/worker/flows/test_amphora_flows.py\n    octavia/tests/unit/controller/worker/test_controller_worker.py\n    tools/create_flow_docs.py\n\nChange-Id: I848b7fc69b977fcb39f8a07e2ea5fc7bd37b5c7a\n(cherry picked from commit 96cce3ed74c63e5756e216a8e48c60de09b027c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba36f04271cd3beffecc8e8f507695b079068933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba36f04271cd3beffecc8e8f507695b079068933"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
