)]}'
{"id":"openstack%2Foctavia~651427","triplet_id":"openstack%2Foctavia~master~I4fcf369e0d34326e787aab9af42fccc1a5760d2e","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I4fcf369e0d34326e787aab9af42fccc1a5760d2e","subject":"Added mixed worker/thread support to health manager","status":"ABANDONED","created":"2019-04-10 02:17:03.000000000","updated":"2019-09-30 06:40:06.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"65ea9637bef5287ccab47b8f6bbbe20ab388fe95","_number":651427,"virtual_id_number":651427,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-05-12 06:02:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 06:02:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c1be7bca6bff41ae01fde89da93fc0ca02d953f9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-10 02:17:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62444974c7278d0c97a48b289e77518627c2f02e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-10 03:48:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cc674ee9969c6f7edf6646f44b8a9beea96a7d5e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-10 04:05:07.000000000","message":"Patch Set 2: Workflow-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ee16295792d2fa9253feca352130567010860b58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 05:54:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/27/651427/2/check/openstack-tox-cover/e8899cb/cover/ : SUCCESS in 8m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/651427/2/check/openstack-tox-lower-constraints/09d6e07/ : SUCCESS in 7m 29s\n- openstack-tox-pep8 http://logs.openstack.org/27/651427/2/check/openstack-tox-pep8/0898c9f/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/27/651427/2/check/openstack-tox-py27/da12e81/ : SUCCESS in 5m 03s\n- openstack-tox-py35 http://logs.openstack.org/27/651427/2/check/openstack-tox-py35/7d498cb/ : SUCCESS in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/27/651427/2/check/openstack-tox-py36/8ee7a2d/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/27/651427/2/check/openstack-tox-docs/9778ad6/html/ : SUCCESS in 6m 10s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/27/651427/2/check/neutron-lbaasv2-dsvm-api/c4677cc/ : SUCCESS in 1h 24m 03s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/27/651427/2/check/neutron-lbaasv2-dsvm-py3x-api/6ab7cea/ : SUCCESS in 1h 18m 25s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/27/651427/2/check/neutron-lbaasv2-dsvm-scenario/53eb647/ : SUCCESS in 1h 26m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/27/651427/2/check/neutron-lbaasv2-dsvm-py3x-scenario/52ece4f/ : SUCCESS in 1h 29m 18s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/27/651427/2/check/openstack-tox-functional/fca2f2c/ : SUCCESS in 8m 56s\n- openstack-tox-functional-py35 http://logs.openstack.org/27/651427/2/check/openstack-tox-functional-py35/ad77487/ : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 http://logs.openstack.org/27/651427/2/check/openstack-tox-functional-py36/43df893/ : SUCCESS in 7m 11s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/27/651427/2/check/octavia-v1-dsvm-scenario/3db98a1/ : SUCCESS in 1h 09m 29s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/27/651427/2/check/octavia-v1-dsvm-py3x-scenario/a7ac221/ : SUCCESS in 1h 04m 33s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/27/651427/2/check/octavia-v2-dsvm-noop-api/18eda00/ : SUCCESS in 46m 54s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/27/651427/2/check/octavia-v2-dsvm-noop-py2-api/81fbc8a/ : SUCCESS in 54m 00s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/27/651427/2/check/octavia-v2-dsvm-scenario/0f0e211/ : SUCCESS in 1h 25m 03s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/27/651427/2/check/octavia-v2-dsvm-py2-scenario/73d1abd/ : SUCCESS in 1h 51m 40s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/27/651427/2/check/octavia-v2-dsvm-py2-scenario-centos-7/2613e38/ : SUCCESS in 2h 02m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/27/651427/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/8e79eef/ : SUCCESS in 1h 34m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/27/651427/2/check/octavia-v2-dsvm-scenario-two-node/28367c4/ : SUCCESS in 1h 53m 22s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/27/651427/2/check/octavia-v2-dsvm-py2-scenario-two-node/02fa3c5/ : FAILURE in 1h 55m 38s (non-voting)\n- octavia-grenade http://logs.openstack.org/27/651427/2/check/octavia-grenade/397e9c5/ : SUCCESS in 1h 11m 34s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/27/651427/2/check/octavia-v2-dsvm-tls-barbican/907f7e0/ : SUCCESS in 54m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6bcd37be42bd5d7f767f4d2f107f52ed6f7b2383","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-15 01:59:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e945f20d8e764d628b5efe793ab0e1c417303637","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-15 02:00:06.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"785e77d53e5eca6bbafba6c2243ac8c85ca3ce79","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-15 02:02:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a37d9107d2a54bfbbf7ff6125deb57d88b7f9670","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-15 02:03:05.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"57e933d4c625a996ff920be457811e65153bbe4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 04:13:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/27/651427/4/check/openstack-tox-cover/853a005/cover/ : SUCCESS in 8m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/651427/4/check/openstack-tox-lower-constraints/8920469/ : SUCCESS in 7m 26s\n- openstack-tox-pep8 http://logs.openstack.org/27/651427/4/check/openstack-tox-pep8/18de031/ : FAILURE in 6m 07s\n- openstack-tox-py27 http://logs.openstack.org/27/651427/4/check/openstack-tox-py27/236d09b/ : SUCCESS in 6m 10s\n- openstack-tox-py35 http://logs.openstack.org/27/651427/4/check/openstack-tox-py35/ff7d3fa/ : SUCCESS in 6m 07s\n- openstack-tox-py36 http://logs.openstack.org/27/651427/4/check/openstack-tox-py36/8949aa2/ : SUCCESS in 6m 09s\n- openstack-tox-docs http://logs.openstack.org/27/651427/4/check/openstack-tox-docs/efddf22/html/ : SUCCESS in 6m 38s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/27/651427/4/check/neutron-lbaasv2-dsvm-api/4995fd9/ : SUCCESS in 1h 13m 52s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/27/651427/4/check/neutron-lbaasv2-dsvm-py3x-api/4a33fa9/ : SUCCESS in 1h 13m 38s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/27/651427/4/check/neutron-lbaasv2-dsvm-scenario/94b4d4b/ : SUCCESS in 1h 19m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/27/651427/4/check/neutron-lbaasv2-dsvm-py3x-scenario/5a3312f/ : SUCCESS in 1h 19m 06s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/27/651427/4/check/openstack-tox-functional/978580f/ : SUCCESS in 7m 09s\n- openstack-tox-functional-py35 http://logs.openstack.org/27/651427/4/check/openstack-tox-functional-py35/106e155/ : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 http://logs.openstack.org/27/651427/4/check/openstack-tox-functional-py36/6aa7f2d/ : SUCCESS in 6m 55s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/27/651427/4/check/octavia-v1-dsvm-scenario/898a245/ : SUCCESS in 59m 21s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/27/651427/4/check/octavia-v1-dsvm-py3x-scenario/ca5b90d/ : SUCCESS in 59m 28s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/27/651427/4/check/octavia-v2-dsvm-noop-api/e35b666/ : SUCCESS in 40m 33s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/27/651427/4/check/octavia-v2-dsvm-noop-py2-api/057f109/ : SUCCESS in 39m 11s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/27/651427/4/check/octavia-v2-dsvm-scenario/2389668/ : SUCCESS in 1h 24m 48s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/27/651427/4/check/octavia-v2-dsvm-py2-scenario/51052bb/ : SUCCESS in 1h 18m 33s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/27/651427/4/check/octavia-v2-dsvm-py2-scenario-centos-7/3b024ef/ : FAILURE in 2h 07m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/27/651427/4/check/octavia-v2-dsvm-scenario-ubuntu-bionic/dd91d54/ : SUCCESS in 1h 30m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/27/651427/4/check/octavia-v2-dsvm-scenario-two-node/2765888/ : FAILURE in 1h 47m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/27/651427/4/check/octavia-v2-dsvm-py2-scenario-two-node/93c1540/ : SUCCESS in 1h 38m 47s (non-voting)\n- octavia-grenade http://logs.openstack.org/27/651427/4/check/octavia-grenade/1759a66/ : SUCCESS in 1h 07m 27s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/27/651427/4/check/octavia-v2-dsvm-tls-barbican/4fabee2/ : SUCCESS in 49m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a0a4b6a62be0bbaea138a2ad39aa66318ca94410","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-15 04:47:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b0e3fa90573231d25d2ce5cb58e3879d8652659f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 07:09:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/27/651427/5/check/openstack-tox-cover/9757923/cover/ : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/651427/5/check/openstack-tox-lower-constraints/ec7eb41/ : SUCCESS in 8m 33s\n- openstack-tox-pep8 http://logs.openstack.org/27/651427/5/check/openstack-tox-pep8/78fab3b/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/27/651427/5/check/openstack-tox-py27/46b15ad/ : SUCCESS in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/27/651427/5/check/openstack-tox-py35/4e3c191/ : SUCCESS in 5m 52s\n- openstack-tox-py36 http://logs.openstack.org/27/651427/5/check/openstack-tox-py36/38f99ee/ : SUCCESS in 5m 46s\n- openstack-tox-docs http://logs.openstack.org/27/651427/5/check/openstack-tox-docs/c60b7c1/html/ : SUCCESS in 6m 25s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/27/651427/5/check/neutron-lbaasv2-dsvm-api/0d9d579/ : SUCCESS in 1h 20m 57s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/27/651427/5/check/neutron-lbaasv2-dsvm-py3x-api/f5285a5/ : SUCCESS in 1h 12m 37s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/27/651427/5/check/neutron-lbaasv2-dsvm-scenario/89fc9b7/ : SUCCESS in 1h 24m 33s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/27/651427/5/check/neutron-lbaasv2-dsvm-py3x-scenario/7953f1b/ : FAILURE in 43m 52s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/27/651427/5/check/openstack-tox-functional/5cd6998/ : SUCCESS in 6m 54s\n- openstack-tox-functional-py35 http://logs.openstack.org/27/651427/5/check/openstack-tox-functional-py35/886fcb9/ : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/27/651427/5/check/openstack-tox-functional-py36/34921d7/ : SUCCESS in 7m 27s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/27/651427/5/check/octavia-v1-dsvm-scenario/8a9ca20/ : FAILURE in 54m 55s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/27/651427/5/check/octavia-v1-dsvm-py3x-scenario/ab9975a/ : SUCCESS in 56m 22s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-noop-api/f135fca/ : SUCCESS in 46m 16s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-noop-py2-api/ed5ea35/ : FAILURE in 37m 38s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-scenario/08686f3/ : SUCCESS in 1h 25m 52s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-py2-scenario/57567e1/ : SUCCESS in 1h 33m 20s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-py2-scenario-centos-7/82c8d5e/ : FAILURE in 2h 02m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/6a813a3/ : SUCCESS in 1h 32m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-scenario-two-node/ff4504d/ : FAILURE in 1h 48m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-py2-scenario-two-node/5f34756/ : FAILURE in 26m 04s (non-voting)\n- octavia-grenade http://logs.openstack.org/27/651427/5/check/octavia-grenade/73d4b36/ : SUCCESS in 58m 03s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-tls-barbican/0f1f77c/ : SUCCESS in 55m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"998965b327337be4d043e53c6fc7d210e3e2fc63","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-04-15 07:34:00.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"43dcc55f9775fb0d10d49bcb8aa1cedc4f5ed120","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 09:42:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/27/651427/5/check/openstack-tox-cover/70e80cc/cover/ : SUCCESS in 10m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/651427/5/check/openstack-tox-lower-constraints/104c60a/ : SUCCESS in 8m 37s\n- openstack-tox-pep8 http://logs.openstack.org/27/651427/5/check/openstack-tox-pep8/aaa635f/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/27/651427/5/check/openstack-tox-py27/859abcc/ : SUCCESS in 5m 27s\n- openstack-tox-py35 http://logs.openstack.org/27/651427/5/check/openstack-tox-py35/35517e2/ : SUCCESS in 5m 52s\n- openstack-tox-py36 http://logs.openstack.org/27/651427/5/check/openstack-tox-py36/ebb8174/ : SUCCESS in 5m 32s\n- openstack-tox-docs http://logs.openstack.org/27/651427/5/check/openstack-tox-docs/6103009/html/ : SUCCESS in 6m 33s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/27/651427/5/check/neutron-lbaasv2-dsvm-api/c7e4d83/ : SUCCESS in 1h 23m 30s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/27/651427/5/check/neutron-lbaasv2-dsvm-py3x-api/8e7d4d3/ : SUCCESS in 1h 22m 25s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/27/651427/5/check/neutron-lbaasv2-dsvm-scenario/2d190aa/ : SUCCESS in 1h 30m 07s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/27/651427/5/check/neutron-lbaasv2-dsvm-py3x-scenario/88a3599/ : SUCCESS in 1h 25m 57s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/27/651427/5/check/openstack-tox-functional/cae7afd/ : SUCCESS in 6m 41s\n- openstack-tox-functional-py35 http://logs.openstack.org/27/651427/5/check/openstack-tox-functional-py35/24c37cb/ : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 http://logs.openstack.org/27/651427/5/check/openstack-tox-functional-py36/480c489/ : SUCCESS in 7m 20s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/27/651427/5/check/octavia-v1-dsvm-scenario/3fa06d6/ : SUCCESS in 1h 03m 49s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/27/651427/5/check/octavia-v1-dsvm-py3x-scenario/69328f1/ : SUCCESS in 1h 03m 14s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-noop-api/69bc246/ : SUCCESS in 42m 46s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-noop-py2-api/7ecae2c/ : SUCCESS in 46m 45s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-scenario/2dbfbc8/ : SUCCESS in 1h 35m 31s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-py2-scenario/9f6f423/ : SUCCESS in 1h 37m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-py2-scenario-centos-7/27a388a/ : FAILURE in 2h 02m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-scenario-ubuntu-bionic/7446cd7/ : SUCCESS in 1h 48m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-scenario-two-node/9b6bbd7/ : FAILURE in 1h 30m 28s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-py2-scenario-two-node/f61b4ba/ : FAILURE in 1h 25m 12s (non-voting)\n- octavia-grenade http://logs.openstack.org/27/651427/5/check/octavia-grenade/e866fa8/ : SUCCESS in 1h 20m 05s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/27/651427/5/check/octavia-v2-dsvm-tls-barbican/a2cac6a/ : SUCCESS in 1h 00m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"52b01e2da5d2690763dbef61d7a8c363cabf0858","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-12 03:58:12.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"29da8e05f0e9706a08810c6e2fb4cc9a4a8adce9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 06:02:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/27/651427/6/check/openstack-tox-cover/b5b8e8a/cover/ : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/651427/6/check/openstack-tox-lower-constraints/d3d588a/ : SUCCESS in 8m 58s\n- openstack-tox-pep8 http://logs.openstack.org/27/651427/6/check/openstack-tox-pep8/e0ebe59/ : SUCCESS in 5m 58s\n- openstack-tox-py27 http://logs.openstack.org/27/651427/6/check/openstack-tox-py27/af8a954/ : SUCCESS in 4m 45s\n- openstack-tox-py36 http://logs.openstack.org/27/651427/6/check/openstack-tox-py36/03d3d9c/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/27/651427/6/check/openstack-tox-docs/1e7543a/html/ : SUCCESS in 7m 31s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/27/651427/6/check/neutron-lbaasv2-dsvm-api/e3d5daf/ : SUCCESS in 1h 17m 52s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/27/651427/6/check/neutron-lbaasv2-dsvm-py3x-api/e204982/ : SUCCESS in 1h 13m 58s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/27/651427/6/check/neutron-lbaasv2-dsvm-scenario/a5f0ad2/ : SUCCESS in 1h 19m 41s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/27/651427/6/check/neutron-lbaasv2-dsvm-py3x-scenario/9c439e5/ : SUCCESS in 1h 20m 00s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/27/651427/6/check/openstack-tox-functional/6ffcdbb/ : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 http://logs.openstack.org/27/651427/6/check/openstack-tox-functional-py36/31b83ba/ : SUCCESS in 7m 52s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/27/651427/6/check/octavia-v1-dsvm-scenario/7623cbc/ : SUCCESS in 57m 37s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/27/651427/6/check/octavia-v1-dsvm-py3x-scenario/44be5a9/ : SUCCESS in 55m 10s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/27/651427/6/check/octavia-v2-dsvm-noop-api/cbe0472/ : SUCCESS in 40m 24s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/27/651427/6/check/octavia-v2-dsvm-noop-py2-api/bd79c92/ : SUCCESS in 42m 58s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/27/651427/6/check/octavia-v2-dsvm-scenario/071b19d/ : SUCCESS in 1h 17m 48s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/27/651427/6/check/octavia-v2-dsvm-py2-scenario/b35ed58/ : SUCCESS in 1h 30m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/27/651427/6/check/octavia-v2-dsvm-py2-scenario-centos-7/50d1053/ : FAILURE in 1h 50m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/27/651427/6/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c99346b/ : SUCCESS in 1h 38m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/27/651427/6/check/octavia-v2-dsvm-scenario-two-node/b363eef/ : SUCCESS in 1h 08m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/27/651427/6/check/octavia-v2-dsvm-py2-scenario-two-node/8bf92c2/ : FAILURE in 1h 24m 03s (non-voting)\n- octavia-grenade http://logs.openstack.org/27/651427/6/check/octavia-grenade/9d29fce/ : SUCCESS in 1h 08m 36s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/27/651427/6/check/octavia-v2-dsvm-tls-barbican/c89d36d/ : SUCCESS in 47m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"65ea9637bef5287ccab47b8f6bbbe20ab388fe95","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-30 06:40:06.000000000","message":"Abandoned\n\nGonna drop this. I still think it would be useful to mix processes and threads, but don\u0027t have time to push this.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c14db726ce1ae195eafcb4295b8e36bef97e81ba","revisions":{"91bd7bc57c87cc262ec9c3749019ba1e649e79c7":{"kind":"REWORK","_number":1,"created":"2019-04-10 02:17:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/651427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/651427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/651427/1"}}},"commit":{"parents":[{"commit":"6e321eb2182b259eb55d0539f52d41b0ebad1f67","subject":"Merge \"Fix the amphora base port coming up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e321eb2182b259eb55d0539f52d41b0ebad1f67"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-10 02:13:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-10 02:13:52.000000000","tz":-420},"subject":"[WIP] Add thread support to health manager","message":"[WIP] Add thread support to health manager\n\nChange-Id: I4fcf369e0d34326e787aab9af42fccc1a5760d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91bd7bc57c87cc262ec9c3749019ba1e649e79c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91bd7bc57c87cc262ec9c3749019ba1e649e79c7"}]},"branch":"refs/heads/master"},"404bed50516af432309e0525786f4866833c6d35":{"kind":"REWORK","_number":2,"created":"2019-04-10 03:48:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/651427/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/651427/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/651427/2"}}},"commit":{"parents":[{"commit":"6e321eb2182b259eb55d0539f52d41b0ebad1f67","subject":"Merge \"Fix the amphora base port coming up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e321eb2182b259eb55d0539f52d41b0ebad1f67"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-10 02:13:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-10 03:45:42.000000000","tz":-420},"subject":"[WIP] Add thread support to health manager","message":"[WIP] Add thread support to health manager\n\nChange-Id: I4fcf369e0d34326e787aab9af42fccc1a5760d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/404bed50516af432309e0525786f4866833c6d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/404bed50516af432309e0525786f4866833c6d35"}]},"branch":"refs/heads/master"},"50a6707e86bbd6d2096f6f725854c9be9a469651":{"kind":"REWORK","_number":3,"created":"2019-04-15 01:59:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/651427/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/651427/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/651427/3"}}},"commit":{"parents":[{"commit":"6e321eb2182b259eb55d0539f52d41b0ebad1f67","subject":"Merge \"Fix the amphora base port coming up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e321eb2182b259eb55d0539f52d41b0ebad1f67"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-10 02:13:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-15 01:59:15.000000000","tz":-420},"subject":"Added mixed worker/thread support to health manager","message":"Added mixed worker/thread support to health manager\n\nThis patch implements support for a mix of workers\nand threads to give the operators more fine grained\ncontrol when tuning the health manager.\n\nIn the past, the health manager ran on threads, but in\nan effort to by-pass GIL related issues these were\nconverted into workers. I believe that the sweet\nspot really is a compromise of the two.\n\nThe negative side of this implementation is that\neach worker will now handle both health and stats\nupdates (on separate threads), in the previous\nimplementation these workloads were isolated from\neach other, but similar to the previous implementation\nyou can allocate more resources to either one.\n\nChange-Id: I4fcf369e0d34326e787aab9af42fccc1a5760d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50a6707e86bbd6d2096f6f725854c9be9a469651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50a6707e86bbd6d2096f6f725854c9be9a469651"}]},"branch":"refs/heads/master"},"d7b8286bcfcd59f3f4f3501c69bc59013bd2efea":{"kind":"REWORK","_number":4,"created":"2019-04-15 02:02:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/651427/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/651427/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/651427/4"}}},"commit":{"parents":[{"commit":"6e321eb2182b259eb55d0539f52d41b0ebad1f67","subject":"Merge \"Fix the amphora base port coming up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e321eb2182b259eb55d0539f52d41b0ebad1f67"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-10 02:13:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-15 02:02:44.000000000","tz":-420},"subject":"Added mixed worker/thread support to health manager","message":"Added mixed worker/thread support to health manager\n\nThis patch implements support for a mix of workers\nand threads to give the operators more fine grained\ncontrol when tuning the health manager.\n\nIn the past, the health manager ran on threads, but in\nan effort to by-pass GIL related issues these were\nconverted into workers. I believe that the sweet\nspot really is a compromise of the two.\n\nThe negative side of this implementation is that\neach worker will now handle both health and stats\nupdates (on separate threads), in the previous\nimplementation these workloads were isolated from\neach other, but similar to the previous implementation\nyou can allocate more resources to either one.\n\nChange-Id: I4fcf369e0d34326e787aab9af42fccc1a5760d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7b8286bcfcd59f3f4f3501c69bc59013bd2efea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7b8286bcfcd59f3f4f3501c69bc59013bd2efea"}]},"branch":"refs/heads/master"},"9a8771969e88b82769eecbc898534037f6acde53":{"kind":"REWORK","_number":5,"created":"2019-04-15 04:47:08.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/651427/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/651427/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/651427/5"}}},"commit":{"parents":[{"commit":"6e321eb2182b259eb55d0539f52d41b0ebad1f67","subject":"Merge \"Fix the amphora base port coming up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e321eb2182b259eb55d0539f52d41b0ebad1f67"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-10 02:13:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-15 04:46:51.000000000","tz":-420},"subject":"Added mixed worker/thread support to health manager","message":"Added mixed worker/thread support to health manager\n\nThis patch implements support for a mix of workers\nand threads to give the operators more fine grained\ncontrol when tuning the health manager.\n\nIn the past, the health manager ran on threads, but in\nan effort to by-pass GIL related issues these were\nconverted into workers. I believe that the sweet\nspot really is a compromise of the two.\n\nThe negative side of this implementation is that\neach worker will now handle both health and stats\nupdates (on separate threads), in the previous\nimplementation these workloads were isolated from\neach other, but similar to the previous implementation\nyou can allocate more resources to either one.\n\nChange-Id: I4fcf369e0d34326e787aab9af42fccc1a5760d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a8771969e88b82769eecbc898534037f6acde53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a8771969e88b82769eecbc898534037f6acde53"}]},"branch":"refs/heads/master"},"c14db726ce1ae195eafcb4295b8e36bef97e81ba":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-05-12 03:58:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/651427/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/651427/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/651427/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/651427/6"}}},"commit":{"parents":[{"commit":"1a87298ac774d921709ab9cbe5016414db65783e","subject":"Bandit test exclusions syntax change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a87298ac774d921709ab9cbe5016414db65783e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-04-10 02:13:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-12 03:58:12.000000000","tz":0},"subject":"Added mixed worker/thread support to health manager","message":"Added mixed worker/thread support to health manager\n\nThis patch implements support for a mix of workers\nand threads to give the operators more fine grained\ncontrol when tuning the health manager.\n\nIn the past, the health manager ran on threads, but in\nan effort to by-pass GIL related issues these were\nconverted into workers. I believe that the sweet\nspot really is a compromise of the two.\n\nThe negative side of this implementation is that\neach worker will now handle both health and stats\nupdates (on separate threads), in the previous\nimplementation these workloads were isolated from\neach other, but similar to the previous implementation\nyou can allocate more resources to either one.\n\nChange-Id: I4fcf369e0d34326e787aab9af42fccc1a5760d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c14db726ce1ae195eafcb4295b8e36bef97e81ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c14db726ce1ae195eafcb4295b8e36bef97e81ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
